)]}'
{"id":"openstack%2Fkolla~918599","triplet_id":"openstack%2Fkolla~stable%2F2024.1~Ibf4d24657f246a191a4b4c1ab95b47d6873e468a","project":"openstack/kolla","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-05-13 22:11:56.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-05-13 18:18:29.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-05-13 22:11:56.000000000","reason":"Change was submitted"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-05-13 13:25:45.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"34411":{"account":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"last_update":"2024-05-13 22:11:56.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-05-13 22:11:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibf4d24657f246a191a4b4c1ab95b47d6873e468a","subject":"[2024.1 only] switch rpm distros to Cloud SIG Caracal repos","status":"MERGED","created":"2024-05-08 11:58:24.000000000","updated":"2024-05-15 21:11:00.000000000","submitted":"2024-05-13 22:11:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918599","meta_rev_id":"2abd3e04cb84f40e79b15240ae496899a29f0cf8","_number":918599,"virtual_id_number":918599,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-13 22:11:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-05-13 16:18:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-05-15 21:11:00.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-05-13 13:25:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-05-13 16:18:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2024-05-13 13:24:30.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-05-13 13:24:30.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-08 12:51:37.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-05-08 13:09:09.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-05-08 14:13:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-08 14:52:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-11 05:24:18.000000000","updated_by":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"reviewer":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"state":"REVIEWER"},{"updated":"2024-05-13 08:14:30.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-05-13 08:49:11.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"3b5ab56e74e541cc4e88e6312d4f7307776f0fb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-08 11:58:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"309ee14aa5071971829923079d66c6807ed38a5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-08 12:22:51.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b15e4b1a61e27b3b11b188f253cd1c3cbd0ec93d","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-05-08 12:51:37.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fc057d9abc02455f8d47513e357e29fe36369eed","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-05-08 13:09:09.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f0e456fb703ed41abeeb8655a4ca8d8b4aa367ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 14:13:04.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb845ef2581948af80bbd0799907d716\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7d8bcaa4c5df4134a5309a97cc084396 : FAILURE in 5m 00s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/590386788bd7491ca52b749e36f30907 : SUCCESS in 22m 09s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dad8c8e3f2984f9a90546c6ab5cd3e0a : SUCCESS in 1h 48m 22s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dd2d16c47a6c47ec8e7e34d24f7f0c9e : SUCCESS in 2m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a3f29c3c733242469074d7bd3f20faa6 : SUCCESS in 3m 48s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/93778b0e70b14c98b7d01b03748ec785 : FAILURE in 4m 37s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b6e785646bbf4ce5a00bb31ea8f683a3 : SUCCESS in 1h 44m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9535f96e49a39f9d79de1e11083361bf11341f01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 14:52:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc9d623c2c5c4c0bb2ab484143c3691e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1825668a5fb54c729449383b00635779 : FAILURE in 7m 02s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2807b1458c0c464e98372a52e63c6ba3 : FAILURE in 13m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/061c26382cc44285b2aa6a685a34b7a7 : SUCCESS in 18m 04s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/50d5ce4399884196b7bc1d3befe99679 : SUCCESS in 26m 53s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/38e62ea4775e4c969a526db46fca38e3 : SUCCESS in 46m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/533e95c75b684a38aa37b3e4978b0030 : SUCCESS in 1h 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05c25d0017f245b1b334b413061022d6 : SUCCESS in 2m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65577d24f9154c799d370ae0ec98cd44 : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ecfea361f9734565ad5cbc37b67af4c3 : SUCCESS in 3m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48e1673d131b4f3ca5133c5ea6c28817 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a2429617d4148f890f5e4dd414b9336 : SUCCESS in 4m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/37dc3a11f94945b8877dfb1581b60a1f : FAILURE in 4m 55s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e4c8e74a27ec414f9a0d1be7fbf98239 : FAILURE in 7m 20s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4556bc3816a447cbb784cb549ec6564f : FAILURE in 8m 11s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6c9b59dc16354d2f8ded2558e02c3abd : FAILURE in 8m 48s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/c7e01728ad394104a3bf77ae495e84c1 : FAILURE in 12m 46s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ea7312bf28714d958b7405d94aec4963 : FAILURE in 9m 59s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/cd4c25a7f8fc4a6a95e27f43a25f1235 : FAILURE in 9m 30s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/7839494d1ebc498e9063ce31ad17a8b0 : FAILURE in 10m 24s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/c1873369e28a4e3b921a2c7015df72de : FAILURE in 12m 32s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/97c8dd906b6146c290ee9f93d8c7481b : FAILURE in 10m 38s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/918517e096924261b17e05bc50d6ef4a : FAILURE in 12m 17s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/169446c7050a45dc96e74a9c2bb020ce : FAILURE in 13m 11s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/91f2bb72d056421dafc72495e4425801 : FAILURE in 8m 40s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8e792627200445db9317f1f6bb06d94c : SUCCESS in 31m 46s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/287d7c1592644704b27de0bede3ef5db : SUCCESS in 1h 01m 06s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/699c8aead3514ef9a4f4fbb245c0a47c : SUCCESS in 52m 32s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f6071f1f4b214a2ca738061337a2a1d7 : SUCCESS in 1h 19m 24s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/1d848d96caef4d25983297b1af5c4724 : SUCCESS in 1h 26m 10s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f7f83a0c80c24cccaf86d08102052149 : SUCCESS in 2h 22m 55s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/62a371ed409949548bf41285e58ba540 : SUCCESS in 1h 08m 02s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/7b440a162d254ef1969920b3f731debd : SUCCESS in 1h 11m 03s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/386387f312a9433c954bbc325efc69f0 : FAILURE in 23m 21s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/466b97453e43403f97c61b2a7e58554f : SUCCESS in 48m 58s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/8a8760b75744457580540aef3ede0fa5 : FAILURE in 49m 29s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/2944025f798d4db3bb436cbf7e5cf55e : SUCCESS in 1h 12m 14s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/4fb10a9a4122455d92cc17a3284d21c7 : SUCCESS in 26m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fdb6e07f12b92b26aea0b665654e03819bee5a80","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-10 04:20:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0d34c44ffd20976c8114762d75c3432cbc4800eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-10 04:44:19.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6464db2d7df1eb101669e1a09a3368f731fb70f5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-10 05:15:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"febe59b9270030197273f9c098f6f873be6594a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 06:44:38.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8e556b4b260498bbaf6fb99ca107c96\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ddc5b711502c493e9f32d0171147277a : SUCCESS in 34m 55s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c800a85914544f138f282f4a21b400b2 : SUCCESS in 1h 00m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cb21241b5b1841aa87884a34ea25483d : SUCCESS in 19m 23s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/510a11f260b24fd4b793c14bbfa73f0f : SUCCESS in 22m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/260560ebfe0c4c03948511d339f64296 : SUCCESS in 1h 02m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d3e1f7d52fb84656bbe57ad88c3851c3 : SUCCESS in 1h 23m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4757a54bc9f24b019ce0d9e76f7ad1d6 : SUCCESS in 2m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/926bfcc950e54de59ecefde4a710a462 : SUCCESS in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b5a989753bb4ab89a46e85258548199 : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecfde8cb08a5485fb82816023ed9cdcf : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bb30a529b3341ebb434462627d84035 : SUCCESS in 4m 40s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/da82713e5f1c4527877b92b9ff932f40 : FAILURE in 20m 36s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a3ca54242adc4179917c0d5a702c0f49 : FAILURE in 27m 26s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/202775df25c746709eac96541a1ddc72 : SUCCESS in 1h 06m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7c9efbc6a2a448838639d8e894887133 : SUCCESS in 1h 29m 26s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/e7720434c4624537bc1790e64d184bbb : SUCCESS in 1h 45m 49s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f436389460544e85a6d65cd7b3251633 : FAILURE in 1h 21m 25s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/7644c5f695e3409d9722bd3f4a45ebfd : SUCCESS in 1h 22m 16s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/41942bee10834caab1f87ace28d9f587 : SUCCESS in 1h 10m 25s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/5bb56a70f56c468d8b0eb09e77949e15 : SUCCESS in 27m 29s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/0b2fb38346bc4422be904662a7fab1f2 : SUCCESS in 1h 02m 33s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/b5626027c959404f87a14bc1822c9c0c : FAILURE in 1h 02m 51s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/d343dc088ce942fab7a11274bcf4cd4e : SUCCESS in 1h 23m 28s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/713885978d9b4d6c86b789818aead923 : SUCCESS in 27m 10s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a57defd49b104354bb42ccfe84cefbce : SUCCESS in 33m 10s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4ade77a4f4d84880bdfc1e9eb779861a : SUCCESS in 46m 28s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4dc63513edc0427c95eaa9c68ab96b07 : SUCCESS in 49m 23s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/bb768a4938f74decaa727a4eac059eea : SUCCESS in 1h 10m 36s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/13afd68e82af4454999e60406f5277ce : SUCCESS in 1h 17m 33s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/76791500722c49ac8fe11a808f38f94a : SUCCESS in 1h 53m 58s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/ea3a0b0db214475697fd07359cf5e2a6 : SUCCESS in 1h 06m 51s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/1ee1d79ae5e64d5198e6731d689881c4 : SUCCESS in 1h 07m 16s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/04f5fbbe2ad24cd39113c5601019f123 : SUCCESS in 22m 22s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/9260b19f8ef64e06bb3a3ea5a9bbe35f : SUCCESS in 50m 19s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/e96743ec046e48d09f5a95a4bb05cdce : FAILURE in 52m 35s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/f5656e6ab4cc47f9bdc9c5e8d4e90f6d : SUCCESS in 55m 25s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/be7420e3ca08499793884ff7d259b6f9 : SUCCESS in 30m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"61975082e97a03b91cb60ab789e4a1aab2e501f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 07:16:24.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7eaf9afdbc6b4c9fb61fe3b13f6079b4\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7c813e81250647248c94d9ae51c12868 : SUCCESS in 22m 33s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/06545e8cfd314cf3a6f8e2bda4bafd76 : SUCCESS in 22m 02s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f15b6aa3d3264a168f06cbe6afe288c7 : SUCCESS in 1h 46m 44s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e62f0f6b6fa74fb98293212006f86a23 : SUCCESS in 2m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/02494c3ff26b48e49be0802f5acb81c6 : SUCCESS in 4m 10s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0716df2a53a140019beba6b2ef537d80 : SUCCESS in 22m 36s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e987f2578f4b4129a4689e0ff1ea23f4 : SUCCESS in 2h 30m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f963546e040e001c9e97d746184114a6878ff0ac","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-05-10 11:09:16.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"cfcf7c494fdd9eb9e8996cdd9345378c65879505","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-11 04:53:42.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"1b644b99eefc683143ded075faefe1342c237ae0","author":{"_account_id":34411,"name":"Neil Hanlon","email":"neil@shrug.pw","username":"nhanlon"},"date":"2024-05-11 05:24:18.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"c3779298d543f59e2e0b73f61bf1c33e50d7de13","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 07:07:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3cb6d80d79934690a1d3472c98746329\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/fe817373c34e444888e3c559980f8045 : SUCCESS in 20m 38s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b17838dd4e8b4c57a20ff08b5d7fb972 : SUCCESS in 20m 36s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e74f087bf2804bf59f0eec999531d471 : SUCCESS in 1h 02m 36s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a10fa4def29c4c41a57c878f045fa97a : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2564093d32194d11bf9b4a771683a7f2 : SUCCESS in 3m 36s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/af40b51a0dfb495fb1a70650ec729217 : SUCCESS in 2h 11m 35s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d33ec9dbc9e945be94705382460f3d90 : SUCCESS in 20m 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"279ec9cb85cb0fa5aee862e396dfd35c41f90997","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 07:37:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/518303cc566444f286d4dd04f02de20e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6860b27579b740a7b614371a43c62d97 : SUCCESS in 31m 38s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c5cfc735bcf246f4a81ad10b8f78608c : SUCCESS in 56m 28s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7e247402b9d443ac8b516b5144a60409 : SUCCESS in 18m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b332fc81d2934048b15b7f6a19b20e3e : SUCCESS in 26m 16s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2be00fcb01f648a3b987ef899ab429d5 : SUCCESS in 47m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/10720ebefda9445b86b9082e66a72426 : SUCCESS in 1h 16m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f9541b857914686866bf6b6c2e8170c : SUCCESS in 3m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f63aa3f5f444c33a2d5f8c68eedd823 : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2a9a8a01548749c4bebd6bf91f9c62af : SUCCESS in 3m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67ea78bb31ea42c4b0f13b83e013dc5f : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0fb4c6eb4a1451ab5997214370d4913 : SUCCESS in 5m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/81a834f0f1d24a68bb63b7d5cb267638 : SUCCESS in 24m 45s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/115544520ca24953aa3f4e176f1f37cd : SUCCESS in 28m 22s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1aa50fb137354b079d704404767784ce : SUCCESS in 1h 04m 21s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f5157ac5f89b4db6b22b3e7851158540 : SUCCESS in 1h 33m 32s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/32292994e2004ff4854048bb29a3f49a : SUCCESS in 1h 46m 17s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8879ffa7e04f440db9259f4a0ce40994 : SUCCESS in 2h 30m 21s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/3c25b5540c254514b9078e2976b225a7 : SUCCESS in 1h 16m 53s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/c4c9adcd27af4838a71c7634f988967a : SUCCESS in 1h 28m 58s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/80bb69225804486684db7856890149a2 : SUCCESS in 27m 14s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/9ba69472b864493ab2caabca60b42645 : SUCCESS in 1h 02m 58s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/d02c4aeba3d841c6bf9cd13186dca3a8 : FAILURE in 1h 01m 32s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/dfde1a8192ff49ebad02dd101f44a1f0 : SUCCESS in 1h 05m 41s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/48996873c4ea46979b0ad52aba3bdd3c : SUCCESS in 26m 00s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b73bb0336e864c589ae37b77d05615da : SUCCESS in 14m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0a5d170e45dd4ad7b4c151f3658e1d04 : SUCCESS in 1h 02m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7b7b1e6fe9dd48a1871276772b2914e6 : SUCCESS in 52m 13s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2d98cbb39d9f473ea667f6c01b994ec0 : SUCCESS in 1h 09m 09s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/417de96b48e3461fb0ce71031a5d86e3 : SUCCESS in 1h 28m 58s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/daba07604869479b94c6d042939885d7 : FAILURE in 2h 23m 09s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/44762000451146cda0781637087fa43f : SUCCESS in 1h 05m 24s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/e8747a1eb2f243d19479e6e36f67c4b5 : SUCCESS in 54m 56s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/ab13b1ebc395447fb7e8febd5f0dcb68 : SUCCESS in 21m 44s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/d11d29820da847ab8a9adba81d7ac842 : SUCCESS in 48m 52s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/3c858c5e405b40acb02c06076404c387 : FAILURE in 49m 37s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/39b322b37ccc44cf90fb00f110ad7240 : SUCCESS in 1h 03m 06s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a55de732b19b4541bdeddeca068159b8 : SUCCESS in 20m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2db67dbcd690ed8259d91026445868881f8d892d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-05-13 08:14:30.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"970e147c8bfa4796b8885f2cd053a4fe5f7a1b68","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-13 08:49:11.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"74b88202cc3f39bd173108f5d98ae2a1b164ff14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-13 08:51:06.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"1496246c9c11e8df63ea9610b6db24bb5d438746","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-13 08:51:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8d54f382ca6b4ea5e27056bb46a94d6e127b20a0","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-05-13 08:54:33.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"43e61abf2cdc6f7aae7bf14e71e05b1e15650310","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-05-13 08:55:05.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8f9cfe7544e692d59c81a1b7a3e6e94b6a1d9661","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-05-13 09:17:38.000000000","message":"Patch Set 5: Backport-Candidate-1 Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"38bd94d8343004a839c91b19e69207666f62112e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 10:58:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0db2c55232d644c0956d452271c396eb\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c7d4a12fca9d4af18d0d6dc4cc5479bf : FAILURE in 5m 45s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6cdf525e9a784c54a4ccc1f54899ecf7 : SUCCESS in 2h 05m 15s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6ddd68939d77478dbe2aadf9081cc775 : SUCCESS in 1h 17m 17s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/49883eee90c346fb8091b6fb09494dad : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/14a102390f1b4ccca9f1271b664ba46f : SUCCESS in 3m 51s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/6f5dcf0e544a4061806a7b7c7657afa2 : FAILURE in 11m 37s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b276e7887dcd4689adc53063f418c8a0 : SUCCESS in 16m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f7efbb0826dd131540267d83b9ccf893f8505380","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 11:11:22.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a8c6ae68177646be9e8269f7bbbc0a04\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9653997a8a9940e99b15f148039f56a1 : FAILURE in 7m 14s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a69966653efd41e7ac97af45f8f9a3de : FAILURE in 11m 33s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/29b930a7f58a4aadb69beb0a5e252614 : SUCCESS in 21m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8e8a6e7568fd4365823ecfa6caeeb58c : SUCCESS in 31m 59s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0d580b4d09b94a5580d6259a25857ad4 : SUCCESS in 49m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6b5ff197038849c399cc23a086e2e3eb : SUCCESS in 1h 34m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58699abc2f144b5682ba2e8ca219d135 : SUCCESS in 2m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1e3a7efdd964fa8b6ea4825172b37b2 : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3545c0cbc2454fc28896c058cb5621ee : SUCCESS in 2m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b022863cdc0f4429acb59b5f0a4f2415 : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/522024aed70647ac8e657ff191c8f4b8 : SUCCESS in 6m 56s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6012442b34f44d55a373244ec47ebdd1 : FAILURE in 5m 10s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7a983307b6ba45e5a7a157f179c16933 : FAILURE in 5m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/02f9d2de41e24503b90813ce7f787652 : FAILURE in 11m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5a24328cf29748e1b5bad7832e0a30b8 : FAILURE in 8m 30s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/5921482abf474620b804fdda09def266 : FAILURE in 9m 49s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/17c4716acb95477881ee2aed97f4c79b : FAILURE in 10m 37s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/b09960e7c30444518746dca89a42f2e5 : FAILURE in 11m 48s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/1f8d1e17c5f34acfab78e9e61c20f821 : FAILURE in 10m 46s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/3edb3ce2354342eb91dc2f7754a64ebf : FAILURE in 9m 32s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/9bade79ae26247b4939b61e75ed42820 : FAILURE in 10m 02s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/ec5ecfdc90414d75a406034373147407 : FAILURE in 8m 43s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/279132dc6b7747aca9b7a7b4beee933f : FAILURE in 10m 34s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e7f273ab9e87418bb61624873badb296 : FAILURE in 10m 50s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7dbc69602f2c4f89aeda1de1f8762a86 : SUCCESS in 24m 21s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0e8124b3a61c4a72bc177b00d9ab2f88 : SUCCESS in 42m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/71d69cf6a4ac4aeeb745e40c294cf31a : SUCCESS in 53m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3fe0a32a499040ad9b50f54cb7d8a21e : SUCCESS in 1h 27m 08s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/6d529ee8b70d4b6680de0c95115171ab : SUCCESS in 1h 15m 22s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ff6e66cb2d35431e946215ab928f4fa7 : SUCCESS in 2h 17m 05s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/28ec32fa245f4537aeee69772e0b656c : SUCCESS in 52m 52s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/b28449c36d8c4ecd947ad86ed63d3a7b : SUCCESS in 1h 05m 28s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/d3910fe035204af886349255cbfdf4b2 : SUCCESS in 21m 21s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/863bae6279fc46ddaf873aa758d21559 : SUCCESS in 49m 28s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/7d79a06507e548e38a737776afaf41b3 : FAILURE in 49m 58s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/96d287a41e334af4a81f1129303b839c : SUCCESS in 1h 04m 44s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/89fbedd9a0e44fb79f737b26e8be6296 : SUCCESS in 31m 32s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0250bfdabbc29f108eb9236a63f309a7559f7279","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-13 12:10:49.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"6ab3e20271c3e60dfefaaa8abca431ad50301d37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-13 12:12:07.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":7},{"id":"e70b1ff98f988747e2d8018b1fd3a30d6110d738","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-13 13:24:30.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":8},{"id":"c21597fcf573608c10d07918cb4cc8ec2b04d9c4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-13 13:25:20.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"eaa2f9bdf81c8d97095cedb873f6f90d40bea117","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-05-13 13:25:45.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"42a9d2fc2138c322d7e5d37cc2ceff273ef2a3b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 15:52:52.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ab79fcf12af45b3af155f925f57c1b2\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/5857adf54f26435ba8b3ef4ac1b3ae52 : SUCCESS in 23m 24s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f7d6df1b613a413899af74e1b1fdf43b : SUCCESS in 2h 26m 03s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/474a4d1fb1a342b2bafb5be665037044 : SUCCESS in 1h 08m 24s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fe3fdfdf6bf2499a827ad8ddba49ba89 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1929e3c6936c4df7b9b62d95d79b60ec : SUCCESS in 4m 28s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/dd3fa9c2ac3d4793ad48eb6104ca2314 : SUCCESS in 45m 29s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/838d03decd7e4e4d938851b01720b093 : SUCCESS in 2h 23m 45s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a9356404493ea6afd4a3fd5017246c3f9932667d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 16:07:02.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23daff6a502e434086c1c3cd68220887\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d49513c774e5408cacc86de2c8d0fd8b : SUCCESS in 20m 28s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/967666c7643a4f66921c1bd5b3c8c40b : SUCCESS in 1h 04m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a581b4d119414593b63112150e038b5b : SUCCESS in 17m 11s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1ace389b33774d57b825a6ede6b94f5e : SUCCESS in 28m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f8962ef73ecd4bd88ba6d86bd0902a00 : SUCCESS in 57m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0918380e401a461aa7a9b5fab29ede83 : SUCCESS in 1h 33m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eebd1446bead42bf93fc4e028811e1f5 : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1406687455c948ceb4b31ee35859a32a : SUCCESS in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/524693c0a59f4678aa4fb2c405c5c627 : SUCCESS in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aaf57ed37f164e77b4f8fc592657bdd3 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a342ec28c074645a803242642faa4a5 : SUCCESS in 4m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0f4d5d041d1b4a93bdbff54864c70d08 : SUCCESS in 23m 16s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a7c2092662a94d90be4018958eea0b8b : SUCCESS in 30m 17s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5e4640b039204930ade57dfefab7e1ae : SUCCESS in 1h 06m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/68e248c2aee14018bddf3b28c99c2a18 : SUCCESS in 1h 35m 36s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/eab93c458e254228a242e15172031b9c : SUCCESS in 1h 54m 22s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/21531d41aa604dc9bd7837ddf204d925 : SUCCESS in 2h 33m 04s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/acceccf3880f4171b5ef0291e5bd7024 : SUCCESS in 1h 10m 25s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/7a6e391ec2c240c6966ce97029e0ac10 : SUCCESS in 1h 19m 37s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/bb6e15cce84f4bb9998452faace14bd5 : SUCCESS in 27m 10s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/f7b763b70fd3414eae004ce6e2a3e64b : SUCCESS in 1h 03m 14s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/0f07816994784c3da7056035100be862 : FAILURE in 58m 32s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/960f17d59ac34b81a01ca8965bf8027e : SUCCESS in 1h 29m 27s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/733659a949414d11b2fc540b97828894 : SUCCESS in 28m 43s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5a171ca8e2734e41ad156419fd821fcb : SUCCESS in 22m 50s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/13cfb772d6fa4571a0ed2f84bf9d72c5 : SUCCESS in 43m 22s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/41f4ff2f4cea4543a0119e79ffdb56ed : SUCCESS in 52m 11s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f0b049179cee4f90818d15f0354b9eb1 : SUCCESS in 1h 13m 35s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/0bd42ad48c954e6190e4dc4ee4645208 : SUCCESS in 1h 21m 58s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/afe5ec5f6a0045a4992132b8f3ec463a : SUCCESS in 2h 24m 41s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/d9ec064fc5ec4a01858662f036ab0b07 : SUCCESS in 1h 07m 25s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/2ceb502b900643539ce61c8dcab96a09 : SUCCESS in 1h 02m 48s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/cf3d6d15ecd949d7a43e404a1af12c2f : SUCCESS in 24m 17s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/302398a5d0564979ba12d8735ef1369a : SUCCESS in 46m 02s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/de386831083e46d3be1a3fc0551bb237 : FAILURE in 49m 32s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/56f4a3037abd405ba0ebe50ceae3e0b9 : SUCCESS in 1h 05m 07s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/7477a251c29f4e0f92320d74b1274439 : SUCCESS in 28m 48s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"621378f6bfddc9f17ef949701e555def8677075f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-13 16:18:58.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"be98b9675f9d70e4a458484dacdb5ab252cecbe5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 16:19:19.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"eb76fcfa17af4fdb9df4991c981ef15c7bb0d085","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 17:49:38.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1b6e0cba3afe43009963b4dc4e2d6658\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/bcc9bc4d7a644093b8a955633adad7db : SUCCESS in 20m 10s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/aee986cdd3d44e698ef4989809b35a49 : SUCCESS in 29m 00s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b022e8d198394508a49df6e525210f05 : SUCCESS in 59m 30s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5abe7bc00fb44e3fae991bf0e2918444 : FAILURE in 9m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d98058de55b849f287f09f2b235936f4 : SUCCESS in 2m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a8156aca501444c8c4142c96928c105 : SUCCESS in 2m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d005880777fa4df5b0f9765de2131ca4 : SUCCESS in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ca78dc2bcd24dacbf84b3de1ccff26c : SUCCESS in 4m 52s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c21a3452109c4821860ccadaf08f02bc : SUCCESS in 22m 58s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5c09e31265174a4aa7e089a38454b65a : SUCCESS in 25m 30s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fabd4f82c96b46af864ed2eaf055429c : SUCCESS in 1h 00m 43s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1e2f135226d34f2f9c928230e3aebc85 : SUCCESS in 1h 29m 10s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7998a2a292644bbeb284906f7b56b177 : SUCCESS in 27m 47s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/87633d2199fd47628a5980f53db5da75 : SUCCESS in 49m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/30abee5ef3b14e32b9f289374c0402b8 : SUCCESS in 55m 41s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ee1680ddf1a34e9c93440232ff312c1c : SUCCESS in 1h 09m 35s","accounts_in_message":[],"_revision_number":8},{"id":"2e57d432be36bf2ecfad9793b4692e7084052067","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-13 18:18:29.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d38298a4785406c7684bd6d3777711c4ecd1d7db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 20:33:41.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d61ba33f03b2471ea0e140e7206c0baf\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/45957973bb6c42d6a3a533573d96ddc6 : SUCCESS in 25m 44s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/325a7ef7b2664ce5be5c2027b9f2f056 : SUCCESS in 1h 03m 43s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6783061f49f14edc8783a60c8507d1f5 : SUCCESS in 21m 05s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3f50f8a8e5e44d878600cb99281678d6 : SUCCESS in 22m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bdb27c817e7747b08897a4ff0d23bb43 : SUCCESS in 1h 07m 48s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4f5d66bf6bb7434ba279d7d14150d4ae : SUCCESS in 1h 39m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1211f01105514f8d8824732cdab50d45 : SUCCESS in 2m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d8e7e49cfbf40abbd95faeed0c5db53 : SUCCESS in 2m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/98ff12e3348a46e985765525fba1be35 : SUCCESS in 2m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61716b322e6a4401be7f9f48f37ab7b6 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd9a8dccc6d141969765fba9268dab7f : SUCCESS in 4m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c4e282fe2b6843df8d8e5a711062a686 : SUCCESS in 20m 50s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c934754063eb4916a8d1e82fdf5d93ab : SUCCESS in 23m 11s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0d45c894672a43f191f0209535619ed2 : SUCCESS in 1h 00m 47s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cb65207bdcc34b4899015a25cd5e5c77 : SUCCESS in 1h 30m 06s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/dc18c69672f14d8babbc1d5cae85c8e2 : SUCCESS in 1h 31m 20s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d6ccb7bcb8634863858a0942c85d2f79 : SUCCESS in 2h 06m 45s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/e66e074ee31548ffa28a30b07523c6e0 : SUCCESS in 1h 06m 02s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/6ca114a3584b49b6b55d835f479874c7 : SUCCESS in 1h 14m 12s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/e9d064473185413f81518cc819c73fd8 : SUCCESS in 22m 06s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/1f05b1d1da0d4af38085730fd6063494 : SUCCESS in 51m 05s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/7120a60ccd354dda93fc826cb93c200b : SUCCESS in 1h 13m 18s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/38331f59f0274e21958660fc50a1d61c : SUCCESS in 1h 27m 18s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/9b30d01e94c443e38d039dc26e0140bb : SUCCESS in 29m 05s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2be1b6b97dc74f5da6d8e601f2633e36 : SUCCESS in 26m 18s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e9f7d785e05e404790930dc4e9ebf012 : SUCCESS in 40m 08s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5a411d83c06443bea942b23cf18ce78a : SUCCESS in 43m 26s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/054292a3c8bf48ab9d84718cc36258c0 : SUCCESS in 1h 24m 06s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/5c0a54a2feea4c1bb7d2e3003c643850 : SUCCESS in 1h 28m 40s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8a32115532804a6e95d9df7bd23f7248 : SUCCESS in 1h 54m 33s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/4ac979e9e2a146a08bbaf10d81ccf0cd : SUCCESS in 1h 06m 38s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/6fd65122f470488b884e74fa3edce2d5 : SUCCESS in 54m 09s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/bdab44248b1241df84d12670cd3ea293 : FAILURE in 21m 18s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/39649d45cf0a459fa872e560b6a71f44 : SUCCESS in 40m 51s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/403774046609491bba221fd3090fd630 : SUCCESS in 58m 00s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/d6a106561c7e4f329d9448e0a45ac884 : SUCCESS in 1h 09m 55s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d2036a5d2cc640da9253fb93f488a4a0 : SUCCESS in 30m 44s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"f71cc528a9ac925907ab67561c1c5a317301d563","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 20:34:02.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"c09183c92329ce455ee88ec2934acb2bdb963a3a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 21:16:42.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/860d527b0ad64f1483c7017eec752f65\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ef4867a806474bf1ab5037d5bb55770b : SUCCESS in 21m 39s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bc5c0a9402dc4514aca910436270de44 : SUCCESS in 2h 16m 24s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/62e5b9d450244c3481b4930eeb56bed5 : SUCCESS in 1h 10m 14s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3efeb05f8fe24acb9e8683c53631e95e : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/32e7f0765fd547a69d92f4e9bbb654f8 : SUCCESS in 2m 59s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b821cdd2cbaf425a9754d111f499c0f9 : SUCCESS in 2h 56m 01s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c741105eb628486bad70546bb222e823 : SUCCESS in 15m 46s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"5577aab72c47badb59fcc1f1f024b06b0768d07f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 22:11:56.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd420a56588e43df849a41e0162720f4\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/130ef0b28a074a6293683fc8d732f2d1 : SUCCESS in 15m 16s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d7725e165f4a4b9a892867dc2759bbca : SUCCESS in 19m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5a446de1454d45edb9eac86504395d7c : SUCCESS in 1h 04m 45s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/01ea2fc9a6584bf281559a80ca156963 : SUCCESS in 1h 33m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43d7e7cf2a5f466d8f218cbe57cb7b0b : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae1f097aced346a2a59dea5f77942e5c : SUCCESS in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a46314268834e7198ba7a0adaa47b0b : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae51f5db78944fa6b2007759841fe3f5 : SUCCESS in 4m 22s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1cc6ebbd65824d35850d1eec99dffb9e : SUCCESS in 18m 29s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/47caa4abde0943fa803d9a69818b4f58 : SUCCESS in 30m 04s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9db621d915fc40468c16d550e0d120f8 : SUCCESS in 1h 04m 16s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c19793c78c10423fba1db328aeae77c8 : SUCCESS in 1h 15m 56s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5885cdc24db84f16928d08d5d18bc48f : SUCCESS in 22m 33s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6f1d2366959741a5a6648554e8cfc650 : SUCCESS in 40m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4ee3f3f28e0a430b9cf58259e5a36b9d : SUCCESS in 43m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/478ea5b7a0fc47328acc095fa5da5883 : SUCCESS in 1h 27m 51s","accounts_in_message":[],"_revision_number":8},{"id":"d1ca684531735adcd2b5c5ec0242f735df483728","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 22:11:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"d03c694f2bf89de4737f250f3e4edc486359c888","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-13 22:12:52.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a740c889b3d74fc3bb7dfd28b317a45b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11e1a39d27344cae86a797c6145d776e : SUCCESS in 39s","accounts_in_message":[],"_revision_number":8},{"id":"2abd3e04cb84f40e79b15240ae496899a29f0cf8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-05-15 21:11:00.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"adaef5768c527677e5fc62e3173c9af7699e16bf","revisions":{"e70d8ae4662dc4734453c79631ece983a5a2c9e3":{"kind":"REWORK","_number":1,"created":"2024-05-08 11:58:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/99/918599/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/918599/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/918599/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/918599/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/918599/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/918599/1"}}},"commit":{"parents":[{"commit":"d48e7ee9da54117fe428c5aab438bad41420acb2","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d48e7ee9da54117fe428c5aab438bad41420acb2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-08 11:58:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-08 11:58:22.000000000","tz":120},"subject":"[2024.1 only] switch rpm distros to Cloud SIG Caracal repos","message":"[2024.1 only] switch rpm distros to Cloud SIG Caracal repos\n\nChange-Id: Ibf4d24657f246a191a4b4c1ab95b47d6873e468a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e70d8ae4662dc4734453c79631ece983a5a2c9e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e70d8ae4662dc4734453c79631ece983a5a2c9e3"}]},"branch":"refs/heads/stable/2024.1"},"4a149a4f57eb42264ae320590e4e9c255c8caf9d":{"kind":"REWORK","_number":2,"created":"2024-05-08 12:22:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/99/918599/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/918599/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/918599/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/918599/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/918599/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/918599/2"}}},"commit":{"parents":[{"commit":"d48e7ee9da54117fe428c5aab438bad41420acb2","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d48e7ee9da54117fe428c5aab438bad41420acb2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-08 11:58:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-08 12:22:51.000000000","tz":0},"subject":"[2024.1 only] switch rpm distros to Cloud SIG Caracal repos","message":"[2024.1 only] switch rpm distros to Cloud SIG Caracal repos\n\nChange-Id: Ibf4d24657f246a191a4b4c1ab95b47d6873e468a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a149a4f57eb42264ae320590e4e9c255c8caf9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a149a4f57eb42264ae320590e4e9c255c8caf9d"}]},"branch":"refs/heads/stable/2024.1"},"3639e2d3f58c03b21a2e4c4d04a645b9c7f872dd":{"kind":"REWORK","_number":3,"created":"2024-05-10 04:44:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/99/918599/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/918599/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/918599/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/918599/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/918599/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/918599/3"}}},"commit":{"parents":[{"commit":"d48e7ee9da54117fe428c5aab438bad41420acb2","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d48e7ee9da54117fe428c5aab438bad41420acb2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-08 11:58:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-10 04:44:02.000000000","tz":0},"subject":"[2024.1 only] switch rpm distros to Cloud SIG Caracal repos","message":"[2024.1 only] switch rpm distros to Cloud SIG Caracal repos\n\nChange-Id: Ibf4d24657f246a191a4b4c1ab95b47d6873e468a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3639e2d3f58c03b21a2e4c4d04a645b9c7f872dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3639e2d3f58c03b21a2e4c4d04a645b9c7f872dd"}]},"branch":"refs/heads/stable/2024.1"},"950123684615b18084cf3f47fb6d0277ac0739e8":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-05-11 04:53:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/99/918599/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/918599/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/918599/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/918599/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/918599/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/918599/4"}}},"commit":{"parents":[{"commit":"84f036a54b2550f392931c8bd2553a92954ab077","subject":"collectd: Mark unbuildable on rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84f036a54b2550f392931c8bd2553a92954ab077"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-08 11:58:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-11 04:53:42.000000000","tz":0},"subject":"[2024.1 only] switch rpm distros to Cloud SIG Caracal repos","message":"[2024.1 only] switch rpm distros to Cloud SIG Caracal repos\n\nChange-Id: Ibf4d24657f246a191a4b4c1ab95b47d6873e468a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/950123684615b18084cf3f47fb6d0277ac0739e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/950123684615b18084cf3f47fb6d0277ac0739e8"}]},"branch":"refs/heads/stable/2024.1","description":"Rebase"},"17b52066215f3cecd24c2497a6c457ece0ee99cb":{"kind":"REWORK","_number":5,"created":"2024-05-13 08:51:06.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/99/918599/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/918599/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/918599/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/918599/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/918599/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/918599/5"}}},"commit":{"parents":[{"commit":"84f036a54b2550f392931c8bd2553a92954ab077","subject":"collectd: Mark unbuildable on rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84f036a54b2550f392931c8bd2553a92954ab077"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-08 11:58:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-13 08:51:06.000000000","tz":0},"subject":"[2024.1 only] switch rpm distros to Cloud SIG Caracal repos","message":"[2024.1 only] switch rpm distros to Cloud SIG Caracal repos\n\nChange-Id: Ibf4d24657f246a191a4b4c1ab95b47d6873e468a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17b52066215f3cecd24c2497a6c457ece0ee99cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/17b52066215f3cecd24c2497a6c457ece0ee99cb"}]},"branch":"refs/heads/stable/2024.1"},"77c29b853782a78a96547d1ed3d70c46dcc28cc3":{"kind":"REWORK","_number":6,"created":"2024-05-13 12:10:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/99/918599/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/918599/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/918599/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/918599/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/918599/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/918599/6"}}},"commit":{"parents":[{"commit":"84f036a54b2550f392931c8bd2553a92954ab077","subject":"collectd: Mark unbuildable on rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84f036a54b2550f392931c8bd2553a92954ab077"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-08 11:58:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-13 12:10:49.000000000","tz":0},"subject":"[2024.1 only] switch rpm distros to Cloud SIG Caracal repos","message":"[2024.1 only] switch rpm distros to Cloud SIG Caracal repos\n\nChange-Id: Ibf4d24657f246a191a4b4c1ab95b47d6873e468a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77c29b853782a78a96547d1ed3d70c46dcc28cc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/77c29b853782a78a96547d1ed3d70c46dcc28cc3"}]},"branch":"refs/heads/stable/2024.1"},"8d0aac057f206e244b5b21a07361de236c8f73df":{"kind":"REWORK","_number":7,"created":"2024-05-13 12:12:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/99/918599/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/918599/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/918599/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/918599/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/918599/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/918599/7"}}},"commit":{"parents":[{"commit":"84f036a54b2550f392931c8bd2553a92954ab077","subject":"collectd: Mark unbuildable on rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84f036a54b2550f392931c8bd2553a92954ab077"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-08 11:58:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-13 12:12:07.000000000","tz":0},"subject":"[2024.1 only] switch rpm distros to Cloud SIG Caracal repos","message":"[2024.1 only] switch rpm distros to Cloud SIG Caracal repos\n\nChange-Id: Ibf4d24657f246a191a4b4c1ab95b47d6873e468a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d0aac057f206e244b5b21a07361de236c8f73df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8d0aac057f206e244b5b21a07361de236c8f73df"}]},"branch":"refs/heads/stable/2024.1"},"adaef5768c527677e5fc62e3173c9af7699e16bf":{"kind":"REWORK","_number":8,"created":"2024-05-13 13:24:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/99/918599/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/918599/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/918599/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/918599/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/918599/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/918599/8"}}},"commit":{"parents":[{"commit":"84f036a54b2550f392931c8bd2553a92954ab077","subject":"collectd: Mark unbuildable on rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/84f036a54b2550f392931c8bd2553a92954ab077"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-08 11:58:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-13 13:24:30.000000000","tz":0},"subject":"[2024.1 only] switch rpm distros to Cloud SIG Caracal repos","message":"[2024.1 only] switch rpm distros to Cloud SIG Caracal repos\n\nChange-Id: Ibf4d24657f246a191a4b4c1ab95b47d6873e468a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/adaef5768c527677e5fc62e3173c9af7699e16bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/adaef5768c527677e5fc62e3173c9af7699e16bf"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
