)]}'
{"id":"openstack%2Fkolla~918652","triplet_id":"openstack%2Fkolla~stable%2F2023.2~Ic0929becd44f25a9fb8d6d789b8169f41c95a226","project":"openstack/kolla","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-05-11 04:46:44.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"Ic0929becd44f25a9fb8d6d789b8169f41c95a226","subject":"collectd: Mark unbuildable on rocky","status":"MERGED","created":"2024-05-10 14:42:08.000000000","updated":"2024-05-11 06:36:47.000000000","submitted":"2024-05-11 06:35:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918652","cherry_pick_of_change":918962,"cherry_pick_of_patch_set":1,"meta_rev_id":"92481a840664faee52784ca85b7251fe51a1354a","_number":918652,"virtual_id_number":918652,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-11 06:35:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-05-11 04:46:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-05-10 21:08:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-05-10 20:26:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-05-11 04:46:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-10 16:12:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-10 20:26:22.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-05-10 21:08:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"f84524f4785518c1ea7b647472645ec12655627e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-10 14:42:08.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"42598cce9737210ba4b2ee24f81c04424aac321c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 16:12:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9679ceb258fa40e5999862e4c9ac5ed8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b2e708cee5774798abd8596d7f0a15fc : SUCCESS in 24m 13s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f78765718e1442049b3e909bc00966ee : SUCCESS in 1h 02m 08s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6bb9690e2d514b0eace339c998a6f0c5 : SUCCESS in 19m 47s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3e76d62f703f4e33944e08d0582fe1d6 : SUCCESS in 29m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/71b960703fc84457b6ce46f1e17e8504 : SUCCESS in 1h 06m 13s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/55e2ffff3b3c4f258247ff1235dc782a : SUCCESS in 1h 15m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce6436be778744d19b3de41d700e5037 : SUCCESS in 2m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7d51f0428f694360bdab9d8b76b33184 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53d57de81e53454191992b06329838a2 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f554d92859a8411a81a985a3df4a4d3c : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b17025ce31b34637ad622bd5d2e62371 : SUCCESS in 3m 08s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74e7cc9acf9f4323830b6fd7f00504d3 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc5ec7f3cb3342978bdb412d7ff2f4e5 : SUCCESS in 6m 16s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c8de32369ba44696945bd6d6577d494b : SUCCESS in 26m 48s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1e63dffd53494254be8652f96eae12fe : SUCCESS in 31m 04s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e9a7a6ff19da433aab757238c1152680 : SUCCESS in 1h 09m 52s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3bfcc234fdbe4effa2b12027196e2449 : SUCCESS in 1h 27m 21s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/459e276cdd4c4f9fa42bfc5214636a81 : SUCCESS in 25m 39s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b5fdc844aab542b9b558ddcaec867a6f : SUCCESS in 1h 07m 50s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/206a03be15584b61a4cfdcc6b8b5aaf1 : SUCCESS in 54m 20s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/de1eb7949fb8496586304a8b96b981ee : SUCCESS in 1h 18m 02s","accounts_in_message":[],"_revision_number":1},{"id":"ef3e19567728acbcb9b728bb33db7e64be6a27cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 17:39:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39093b5434cc40a1b330ab00dbe32faf\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a35ff9c3c3b544ff879e85ed1bade130 : SUCCESS in 2h 47m 38s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/da228cedbb2b418aae60c167aedadc83 : SUCCESS in 32m 39s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/23c8cd42bd8a435e92baba89a853b265 : SUCCESS in 1h 57m 42s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3bd84012da6c47a6971f49a52cf6e76a : SUCCESS in 2m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a7444d48cdd44068a3ca4be2f407c277 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8f3324756c324d118cba47faa574f67b : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1e2a5919d9fa479d9b3945075f21479a : SUCCESS in 6m 26s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/dbba696e99304e349a7b25c2d97d11eb : SUCCESS in 34m 03s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0af32b1503584baf90e1d7964e3740dc : SUCCESS in 2h 54m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d343edd9402d111fcf3ac480e13296c7e5f81dbe","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-05-10 20:26:22.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a6fe1f02cf020cb47e9a0f4cc94fd10d949fc049","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-05-10 21:08:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b117b6b3f2d0b5bc88dde9bec1e427feeaf06682","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-11 04:46:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f6e2de9d166ff5e07f9a7243e26a0e4efa25b5cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 04:47:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"48588b0e7e73724955fdc26408a86e3573ed659f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 06:35:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b81d11ffd5e4206b6e131f7bdaf38de\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c9534c76f1dd43fab80553f60ba6b560 : SUCCESS in 18m 17s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3ea0d6857de1439ab39b2719c7fe6626 : SUCCESS in 24m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/35e2b6b3dcb3466a8a310a390d549742 : SUCCESS in 52m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4487393742b542ddb422823fd4e2479c : SUCCESS in 1h 14m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d9f92b7227e47c1bac778032baad1c7 : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2e65eca477c48f0920aca3461d256aa : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ceaa24a8cba4df79b77bff2f4bc2ad7 : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a3e1a675afc44592808baedee4a2440b : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8a64a8689ca44e39dc435115dd0b9e0 : SUCCESS in 4m 24s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e2964e4aafe540d6ae68f6da76e71432 : SUCCESS in 23m 43s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/99b5af688c2245f4a6be68889a4b7aaf : SUCCESS in 33m 01s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9b5ef3da53064539b6e40496b24df48b : SUCCESS in 1h 08m 30s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/67474525a2d24070bd7a0800e91a606c : SUCCESS in 1h 34m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3827e5a9b76e4ae2a755e1e40da6cf1a : SUCCESS in 19m 09s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4cec13957efa4c5ab94bdcfebb2bff42 : SUCCESS in 1h 07m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/435a6d97319a4cb8ad3651b335b5c53e : SUCCESS in 54m 40s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0fc641d3ad5f4e2cab21a100d2f915b7 : SUCCESS in 1h 16m 30s","accounts_in_message":[],"_revision_number":1},{"id":"4bfb2a5e22fa7eb68fe5ee50f5284e55e8310777","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 06:35:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"92481a840664faee52784ca85b7251fe51a1354a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 06:36:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0923ed197a0451287b015600c6e7e87\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bf9d2160cdf4bb390f2ea2ef9b75622 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ba0aebd6730d7ee09e84260d4218d4dc2a86235c","revisions":{"ba0aebd6730d7ee09e84260d4218d4dc2a86235c":{"kind":"REWORK","_number":1,"created":"2024-05-10 14:42:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/918652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/918652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/918652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/918652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/918652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/918652/1"}}},"commit":{"parents":[{"commit":"fde0ef25a797495458cd9fdf7ef52446bf5591d4","subject":"Merge \"Nova: fix swtpm and swtpm-tools missing from deb installs\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fde0ef25a797495458cd9fdf7ef52446bf5591d4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-10 04:29:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-10 14:42:08.000000000","tz":0},"subject":"collectd: Mark unbuildable on rocky","message":"collectd: Mark unbuildable on rocky\n\nSince Rocky 9.4 release - delivered libpqos is 5, not 4 - and\nCentOS Stream 9 OpsTools packages have not been rebuilt.\n\nRelated-Bug: #2065362\nChange-Id: Ic0929becd44f25a9fb8d6d789b8169f41c95a226\n(cherry picked from commit 25f357ecc81bc2ad25a1e1b2f549d6a4b755b638)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ba0aebd6730d7ee09e84260d4218d4dc2a86235c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ba0aebd6730d7ee09e84260d4218d4dc2a86235c"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
