)]}'
{"id":"openstack%2Fkolla~918653","triplet_id":"openstack%2Fkolla~stable%2F2023.1~Ic0929becd44f25a9fb8d6d789b8169f41c95a226","project":"openstack/kolla","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-05-11 04:46:25.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"Ic0929becd44f25a9fb8d6d789b8169f41c95a226","subject":"collectd: Mark unbuildable on rocky","status":"MERGED","created":"2024-05-10 14:42:44.000000000","updated":"2024-05-11 06:36:48.000000000","submitted":"2024-05-11 06:35:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"918653","cherry_pick_of_change":918962,"cherry_pick_of_patch_set":1,"meta_rev_id":"6cfd88137bdb37b58bed67899dd32a6fa8482603","_number":918653,"virtual_id_number":918653,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-11 06:35:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-05-11 04:46:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-05-10 21:07:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-05-10 20:26:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-05-11 04:46:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-10 16:13:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-10 20:26:27.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-05-10 21:07:48.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"f827d2c258b02a05b36f57b0ebdfdcc5da2be604","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-10 14:42:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"946d975f7b137697b01865dae7dd53b23e8fafa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 16:13:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4827eed592e4f03b0ca3b05e3a6cd4f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b02b44f449ca4a6fb0bc268621b63d73 : SUCCESS in 25m 18s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f8704c0e7f9348a79781c9cd616aadaf : SUCCESS in 48m 08s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/435501a5a5664113ab275f6e9757993d : SUCCESS in 21m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f70cad610b8044fdb98e6616531d19b6 : SUCCESS in 2m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88f42091a79c4f3e92f94dca0a3ddce8 : SUCCESS in 2m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b4167ed7adb47208902e05262715246 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e55b4c8d51c94c5bbc1b565da116009f : SUCCESS in 4m 52s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9c1c1e623a224a919d07d5115daed150 : SUCCESS in 21m 17s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/bf067db3fa064803b4a6968596aa123e : SUCCESS in 1h 00m 49s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4d00fa6f3e25494ca86d5a367603e294 : SUCCESS in 1h 24m 56s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2d71f48467504aa9b643c7671733eaf8 : SUCCESS in 23m 53s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9a99bba901a141ed9f16f86bff95da80 : SUCCESS in 55m 13s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/88855b015a5b4fcc95eb58d63cc68388 : SUCCESS in 1h 15m 56s","accounts_in_message":[],"_revision_number":1},{"id":"5397c97a60ca75097a8dfda5ba4107dc37e8fd89","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 17:48:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b225fd67f8704736a91927c4f4470cdf\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3c303c88afa347e799daec80582b6146 : SUCCESS in 2h 49m 32s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/41c281976a524e1287e80e68efaa4abb : SUCCESS in 33m 42s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/24e220a0169e432d853345f88f3c4b66 : SUCCESS in 5m 47s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0f4ba0d41ab44c29b70e07a425b26ba6 : SUCCESS in 32m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c68b853bd7174a8e88bd19ee2f3a4f5b : TIMED_OUT in 3h 03m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"79f75d4f9fd66f4e1f22a4c69d1744a0d44533d3","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-05-10 20:26:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6d206bdb1e3439dc4853aaf8b563570aa74d5ad9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-05-10 21:07:48.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"fb32c882ee504884ece63a84b6bdfbf22f19d6c4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-11 04:46:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c91246599bfc484c35b25f66b66f9a8bc49c0d9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 04:46:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"79501b633e1cdfa140ee28ec1d72e7ce4e49f8a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 06:35:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/756150330eea4d53a6fcf32a4f313a0d\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/68a41a315192412abb4d83ccfe2fbc40 : SUCCESS in 24m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d707d8a77974de3a7bb0822b01ce405 : SUCCESS in 2m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70ef2bfc65a442e4b889eb785c16b7ab : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39e1369e3a524b8f95d1692b27e4581e : SUCCESS in 4m 35s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/74e841e087af41078ae45178b927792f : SUCCESS in 25m 19s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/10450ee983cb4dda8290941e91df59b3 : SUCCESS in 1h 04m 07s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2850a9c733f44556a9ad1e54ae83ee2c : SUCCESS in 1h 43m 21s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3714ae506cd842949d7c4bc1e96ec855 : SUCCESS in 36m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/120751133a914d3282e7a8363494ed34 : SUCCESS in 51m 39s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a339b0558ef04b0ab067a4b89386f7b6 : SUCCESS in 1h 07m 46s","accounts_in_message":[],"_revision_number":1},{"id":"7ba9e2fdd9c602cfe409adaa865593b7bb4aea51","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 06:35:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6cfd88137bdb37b58bed67899dd32a6fa8482603","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 06:36:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e939440c5b8341bfbfa44ff76f3881db\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c693d860025740b992a6be71fec01b00 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d455ec28357e463a4719a388764d89a0e7b2787f","revisions":{"d455ec28357e463a4719a388764d89a0e7b2787f":{"kind":"REWORK","_number":1,"created":"2024-05-10 14:42:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/53/918653/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/918653/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/918653/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/918653/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/918653/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/918653/1"}}},"commit":{"parents":[{"commit":"264bba39ac20e21ffd09446ea3e46c15e6dd80d6","subject":"Merge \"Nova: fix swtpm and swtpm-tools missing from deb installs\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/264bba39ac20e21ffd09446ea3e46c15e6dd80d6"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-10 04:29:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-10 14:42:44.000000000","tz":0},"subject":"collectd: Mark unbuildable on rocky","message":"collectd: Mark unbuildable on rocky\n\nSince Rocky 9.4 release - delivered libpqos is 5, not 4 - and\nCentOS Stream 9 OpsTools packages have not been rebuilt.\n\nRelated-Bug: #2065362\nChange-Id: Ic0929becd44f25a9fb8d6d789b8169f41c95a226\n(cherry picked from commit 25f357ecc81bc2ad25a1e1b2f549d6a4b755b638)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d455ec28357e463a4719a388764d89a0e7b2787f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d455ec28357e463a4719a388764d89a0e7b2787f"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
