)]}'
{"id":"openstack%2Fkolla~919938","triplet_id":"openstack%2Fkolla~master~Iceb16a5cf2ebc64effe30bbc9144144cdffebab0","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-04-23 17:55:51.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-04-19 14:58:59.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2025-04-28 09:38:35.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-04-28 09:38:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iceb16a5cf2ebc64effe30bbc9144144cdffebab0","subject":"opensearch: remove example certs and securityadmin_demo.sh","status":"MERGED","created":"2024-05-17 10:53:36.000000000","updated":"2025-04-28 09:39:17.000000000","submitted":"2025-04-28 09:38:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919938","meta_rev_id":"5e8e34cb69d805b7f456f3f9d20a11af79fe113e","_number":919938,"virtual_id_number":919938,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-28 09:38:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-04-23 17:55:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-04-19 14:58:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-18 08:43:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-04-28 07:57:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2025-04-28 07:57:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":1,"date":"2025-04-23 17:55:51.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-18 08:43:51.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-17 13:26:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-17 13:56:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-13 12:48:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-09-30 17:07:33.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2024-12-09 16:17:22.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"9875916470fe8d18b9228aeaafbe349a5bdcc869","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-17 10:53:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a6c668b9d40a1e3119c4c5aaa3184a7d6033190","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-17 10:55:09.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"d066d53e08d05782e455cf3e46683ee575ae4995","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 13:26:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43ff87ac74e2435384d526b1ffbf5095\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/818e33d3e22140b09d5b0e0c2c853877 : SUCCESS in 22m 58s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4abcae00b51a40b69ffa630dec33a883 : SUCCESS in 20m 56s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9d452c9b142f44898c5b36268f9da2fc : SUCCESS in 1h 18m 09s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c46d2686e6b249c4bad7d51e9bfac050 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c668e7d89c304941a959b480bdc6ae4e : SUCCESS in 2m 37s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/61e314e2e2374fbd921a01985a231e05 : SUCCESS in 2h 30m 44s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/19c79e413d4f458a9679857410dc9be8 : SUCCESS in 17m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"58643601f3751e623a5d1bea01fefcc352823df4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 13:56:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b53dfcd7b4f049dcbd7fc25383c43a93\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b8ce636e56484d17be30d581c878f447 : SUCCESS in 32m 13s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/cc322ad50deb4a6b893fd2d94a260584 : FAILURE in 1h 47m 04s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/97aeccbfbfe24f9881496e5bdab845c3 : SUCCESS in 19m 09s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/eb958837e7e64ec2850c1c43befa3f86 : SUCCESS in 28m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a28eb7519632494d9ed4614292f8dd56 : SUCCESS in 1h 06m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/684373538e8e4bbb99eabcde2dea67c3 : SUCCESS in 1h 31m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/562555baa5f74b9d8a4d7bbf2da627f4 : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13db68eb22794ed29c7a61b76b444cf9 : SUCCESS in 2m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b5f8bbb0fce24380a37bfcf88d622666 : SUCCESS in 3m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ed2cbe1e846646d49f22eb7d7ea038f2 : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/179fb85182db49ed8c3840dbb95449f8 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e08a3a2fcd14815b16b332d3ba3d624 : SUCCESS in 4m 16s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7ab523ae3f19499599a22fe6d4dc6982 : SUCCESS in 24m 56s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/714ed3180463404e94a502d211634663 : SUCCESS in 32m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5f5eb31dee4e4d2d9e84d9663f894927 : SUCCESS in 1h 07m 41s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/aca689fcf6cf4c40881f13bbdbf29b67 : SUCCESS in 1h 37m 50s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/81cbb897bd804be083b943269fd97b69 : SUCCESS in 41m 11s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4ad0a4a572e94e67ab0b5253653057aa : SUCCESS in 32m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/30eb4ea50afd45929611a719aa7fab81 : TIMED_OUT in 3h 01m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/97871e3d30384d639b47c0841d1048c8 : SUCCESS in 45m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/25b1059bcbaa4da198cca810ba4ae0bf : SUCCESS in 1h 25m 14s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5858ed2d35d749a19904fa58362948a8 : SUCCESS in 37m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1515e47ffff53eeb62b0fd9f1180d80ee1a462dc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-20 16:05:53.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"018915e9738a512aa01956852efe4554b1d73dc0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 16:57:56.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/124fb65b885240cf9f3576b5f76165b4\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1ccf3802b2c8442fb0a87d21265b2ba4 : SUCCESS in 20m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a1765c75ed4b46f2aee406b368842c90 : FAILURE in 7m 53s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2ec9b5a1f29244e2a0d87bbb75079d4e : FAILURE in 10m 55s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/80a760db30cf48e8a595488a411055a0 : SUCCESS in 2m 43s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8c5de774573d4088bdfaf2473192668f : SUCCESS in 5m 28s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/58ad6da81daa4ec193b998d1f46775d9 : SUCCESS in 51m 29s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/389bd96427974c008f89e0edc58a686c : FAILURE in 8m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"644cede908ff78bc2e1f06de92f560665ac8a8f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 18:28:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d4851c1fd214670bf98a75f968c59f7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/37f70665dc724bccb85f1d62a371b676 : SUCCESS in 25m 50s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0d1ba7d2d8504d4f8ad116349c367ab0 : SUCCESS in 1h 05m 05s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/da8d06b3d8ff4b7e9146a7820ed668c0 : SUCCESS in 22m 43s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1b25c755a5464867a0d43724f00b9e52 : FAILURE in 33m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/72697192687940a38409290664dcfbba : FAILURE in 11m 17s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5395f655c04547c6ad8a4d3a4ce5faf5 : FAILURE in 17m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2c00eacacdd470fa627c8f4e8f4b0ed : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44fed21757b04da099d58d59cc1380fd : SUCCESS in 3m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d9782026ec094be89678fa8be81d4493 : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c44069c566a74937ab967cb42daf55f6 : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd3930f6365e4a1fbf531bb6e7759c80 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57175a5f2b7a442999c71a9688c76889 : SUCCESS in 5m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a1c052b2bf8047d48e7c6cf7d3fc8348 : SUCCESS in 26m 17s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ed33111a33df4c23a26d946a43b8cf82 : SUCCESS in 33m 04s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/17ff784c3a5f44b4b049875cc1d229f2 : SUCCESS in 1h 04m 25s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/958c07ed6a2d4ea1ae5488e938a7c7ab : SUCCESS in 1h 38m 21s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/adf2859ef5aa4169b562ea1cead55546 : SUCCESS in 35m 30s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b2b79ab372f74b21aeb3bba582ebbc3e : FAILURE in 9m 52s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4d830eefb33649c78583ca2db31c2f50 : FAILURE in 24m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/aab2724fa94b4e118c5045f52d76fdaa : FAILURE in 12m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/953bcc3f803844b29f79038a52a7f9c8 : FAILURE in 11m 54s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f1296c6c2db440e398be3e931be563f8 : FAILURE in 11m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5d0d9bca7c5ad02c00603d7beb968619370f3eef","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-21 09:09:27.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"b69103f1863a194dd8ec51d0d47f992a7906ad29","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-21 09:09:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dbc3e8be7b8a235a3e7eedae66b14aa084f4728b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-21 10:25:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53a4060179b04f68bc2557ea9a556bc4\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/689e9801ebc5410dab60cff188cb253a : SUCCESS in 24m 58s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3f29cfc29ec747c2845b2effd81abb7b : SUCCESS in 22m 54s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/441eeab490554571aec7a5f69257dbb5 : SUCCESS in 1h 14m 23s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ab31e3616e48449caf54be0f1e5b99ca : SUCCESS in 3m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/61a0e0adb7834a9591332c7857c125bc : SUCCESS in 3m 20s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8115ea8360ed4c2eb6b717ac53184fe6 : SUCCESS in 50m 10s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0826bd652c094d81a0104ec86076fa46 : SUCCESS in 21m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5669817aa58e7863776004911509ee9763f4b7ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-21 11:22:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e4c9e321ea44f95bc5440be4e700298\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/07c7b1edacde4ed99f737584f1aa6137 : SUCCESS in 25m 12s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/952f12df5bdf41e398592cc8c7cde7cf : SUCCESS in 1h 03m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b118cb266217432390972ad8b05c167c : SUCCESS in 18m 50s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/161918e2813d4c1280f5cdbad94b6cf2 : SUCCESS in 29m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/95bf4b6c304341059ce9ab96268a693b : SUCCESS in 1h 02m 47s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ec0f3837fb7b4a43830cb2cd0ffa103a : SUCCESS in 1h 28m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/771d9ac5c3d24bd38f27155118538107 : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1c852108c5f43b9af9264873de344cb : SUCCESS in 2m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/660085172a1049639f3bf2c584c96850 : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e3fa041330d24e8ab0ac5d450bec8384 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a8658b7e7324e24a737e75fa7be1587 : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e6846d9ed734930b4d78e60c9703f52 : SUCCESS in 5m 16s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9acffa0c77cc43caac8bd18ac3332b70 : SUCCESS in 26m 54s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/f626ab8cfb1944599aab934fb567ab72 : SUCCESS in 26m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c3da73889a5a4792b0e4b774e286c67d : SUCCESS in 1h 04m 25s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e7edd3efea744037a4a6560075a863c4 : SUCCESS in 1h 36m 14s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/618a80df5e884e9abc9e159a8118081a : SUCCESS in 31m 51s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f035387efc8f40e49c4eb12bb20f2caa : SUCCESS in 17m 52s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d2c8b3d1abed413db512954401b4c488 : SUCCESS in 50m 07s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/09028c3020f445bdbb1ed8725aafa200 : SUCCESS in 48m 57s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6688688564d748288baa4f86f1c4e1c1 : SUCCESS in 1h 31m 53s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/844d6d389c96462e8f1321d7ec7bbcef : SUCCESS in 26m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5780e8cf792b5fda942b8a9a0ddbcd04d371f7e1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-27 05:19:24.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"3c79162ad5975b52fe158a4f64cf8318fdf9da5f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-08-13 12:48:04.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"6ce68bd725f81bc54c22d36cbdffb713cda2541f","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-30 17:07:33.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7bfc53369bd73a29f47564e7636587d7179143bf","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-12-09 16:17:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"f547c12d0b9afd11f05e03541f8b41cc4594a7b9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-04-18 08:39:14.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nThe following files contain Git conflicts:\n* docker/opensearch/opensearch/Dockerfile.j2\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"688a913b76fd8fadbc7f701ecb116b412069ff70","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-04-18 08:43:03.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":4},{"id":"96d9b1ab4a1ca8e09475ce82b68186b8903a2d42","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-04-18 08:43:19.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"95b35a32ba75302332cc73cdcdcfb466a9c50254","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-04-18 08:43:51.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"abd2bce5402e957c48465eaaf0e1a8475ade7e6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-18 09:51:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2269b34f1c424d8690c8170f45e80f2f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a93373284c9740cbb1d6491980c1eca8 : SUCCESS in 30m 05s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/42128aeb7fe44cdc9a97edd56a0c8e5d : SUCCESS in 35m 23s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b447e738d9144fa4b5e0d28b267d97b2 : SUCCESS in 1h 06m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b49655428b9e4e819cc31ddda0f0eb75 : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0b4756452d684dc69ab1ac93304a026b : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/78c5bf0e13c54b37abd2f4133a063119 : FAILURE in 7m 43s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/09280d9ae2c947b991ef7ec153af80e4 : SUCCESS in 40m 59s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e4b040de2e3240aebf6618135329629b : SUCCESS in 42m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4b0a70917ac8add1a0d25e84dac168d6cb998cfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-18 10:24:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d98c343a89f04d26b658889bc9bdad1e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/877c12278151424ca06648bce5f222fa : SUCCESS in 25m 39s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0478238e42634ba6ae13f137ff27fea0 : SUCCESS in 1h 10m 48s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/42b635bd38d247f4acb685fd8cdcaaef : SUCCESS in 20m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7b5a40d332b4401a846dca458101cce9 : SUCCESS in 28m 27s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/51adf4449dbc4cad8479f28b6a8dc6bf : SUCCESS in 1h 08m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fff6c67d65344078ab055a6de142a7ee : SUCCESS in 1h 25m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b3e42f20c1041e7b5b4e8364971a7e3 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0aeb3daeb30d4a33984d174e9630f0a3 : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e9ca177df0b44b46899760e000cb3c7a : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0cfa782744794a23b37394a6523df4ec : FAILURE in 5m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5002105fbc84471bbc4a9cd9836847d : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af49c72c633a450ea67f0dcbabd2ba0d : SUCCESS in 5m 54s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/7e37eb7bbd544adcb83abf31ad858046 : SUCCESS in 2m 40s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/485e62b820134895b0f96ef02340d629 : SUCCESS in 22m 28s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d1d9956b2c9b4ae6b7b244a2f9b42d21 : SUCCESS in 1h 18m 34s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/dab3f3dc56024adab9a13f8b94fe5676 : SUCCESS in 59m 56s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/856439886844436db648feb741d92f84 : SUCCESS in 1h 35m 59s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/fbe784411f8645158e793cbda9cd9a50 : SUCCESS in 15m 42s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/939dcf1de0ce4b7783f68b908064d03a : SUCCESS in 16m 21s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3267370a952e4e128621b96c7ecf5d8e : SUCCESS in 19m 00s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/903f49dfcd494cc39988e408d2030858 : SUCCESS in 30m 16s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/748e5b51fd9449d78cf5ebf144b81483 : SUCCESS in 1h 19m 12s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/19dbc190434042c6a5d60d43b01a1b8c : SUCCESS in 21m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2e0e611d3abfe0db0a43d19da4a41c86659a92b5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-04-19 14:58:59.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"da0bded1cfc7eed39bcb7978bc0e6b7c6361109d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-23 17:55:51.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"489ef193b9f7ee9f0390d0b64d3ad8c06246872b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-28 07:57:15.000000000","message":"Patch Set 4: -Workflow Backport-Candidate-2","accounts_in_message":[],"_revision_number":4},{"id":"a69c208b74ced83be711cc31960bf30cfcfaddc7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-28 07:57:19.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"022904f1f6897014f134cee719f8bdff02d23c6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 07:57:28.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"407ca821b4ab7f72cbb5a59d3237e2f552fdefa6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 09:38:35.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98369c8ee1ae483eb457597dd9b2754d\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0262dd1d887c41f4a0a7336516f23266 : SUCCESS in 19m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/34851a9ce0a2474a90c47d9aa76ea197 : SUCCESS in 26m 05s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2f2780b1f7434dba9552e58d6b876c50 : SUCCESS in 1h 01m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5d68c6fbb3904d74a5828f409ff8b515 : SUCCESS in 1h 30m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef8834a7812241bcaf6dfe00ee776cf5 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4eca6b396b3149968ad32e897bfb5a66 : SUCCESS in 2m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6c956a17c949415ab4931904fb728140 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/100365ef89b8430c8c0b88c3ae2a760a : SUCCESS in 5m 57s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/9a6f9300e0394929be98c0557d970fe6 : SUCCESS in 2m 25s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2d0d475812eb481c8cb5a8c7af0a9d47 : SUCCESS in 24m 05s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/71ba4cd63cf444e0afc4b1ba5b004e26 : SUCCESS in 1h 25m 53s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/bc7662a6853647749ae80c4e161b4c24 : SUCCESS in 1h 07m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c83f55399d6f4b51afb59ba76279c406 : SUCCESS in 1h 36m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0f7c9502a3ce4f8e8cc670c0c38808df : SUCCESS in 23m 35s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/582fe92f3ac242c88e3a8f72f8589eae : SUCCESS in 37m 16s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/95c31fcf222143b591feb44dad4bae7c : SUCCESS in 58m 38s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/fb817052efc1412aa8106f2d24471964 : SUCCESS in 1h 33m 31s","accounts_in_message":[],"_revision_number":4},{"id":"f95a6df0a9885cf3cb94a27041d6cf39f0d000cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 09:38:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"5e8e34cb69d805b7f456f3f9d20a11af79fe113e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 09:39:17.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a2aa1dd85b5487daae6230977157548\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ade5cdb0eadf48a0bd482813b06d1dd4 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3eb606a36ed421235863b638ef57ef5b26630fad","revisions":{"a13158e56cdbd6a5bb7be01660d8467ebffc3a05":{"kind":"REWORK","_number":1,"created":"2024-05-17 10:53:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/919938/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/919938/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/919938/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/919938/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/919938/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/919938/1"}}},"commit":{"parents":[{"commit":"25f357ecc81bc2ad25a1e1b2f549d6a4b755b638","subject":"collectd: Mark unbuildable on rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25f357ecc81bc2ad25a1e1b2f549d6a4b755b638"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-17 10:53:15.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-17 10:53:33.000000000","tz":120},"subject":"opensearch: remove example certs and securityadmin_demo.sh","message":"opensearch: remove example certs and securityadmin_demo.sh\n\nChange-Id: Iceb16a5cf2ebc64effe30bbc9144144cdffebab0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a13158e56cdbd6a5bb7be01660d8467ebffc3a05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a13158e56cdbd6a5bb7be01660d8467ebffc3a05"}]},"branch":"refs/heads/master"},"687a0186e0e643e9d101be94f6c8f1a0d17e0573":{"kind":"REWORK","_number":2,"created":"2024-06-20 16:05:53.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/919938/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/919938/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/919938/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/919938/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/919938/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/919938/2"}}},"commit":{"parents":[{"commit":"25f357ecc81bc2ad25a1e1b2f549d6a4b755b638","subject":"collectd: Mark unbuildable on rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25f357ecc81bc2ad25a1e1b2f549d6a4b755b638"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-17 10:53:15.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-20 16:05:53.000000000","tz":0},"subject":"opensearch: remove example certs and securityadmin_demo.sh","message":"opensearch: remove example certs and securityadmin_demo.sh\n\nChange-Id: Iceb16a5cf2ebc64effe30bbc9144144cdffebab0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/687a0186e0e643e9d101be94f6c8f1a0d17e0573"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/687a0186e0e643e9d101be94f6c8f1a0d17e0573"}]},"branch":"refs/heads/master"},"cee097ec6c3969fa88d93b990efbb1e778953a51":{"kind":"REWORK","_number":3,"created":"2025-04-18 08:39:14.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/919938/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/919938/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/919938/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/919938/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/919938/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/919938/3"}}},"commit":{"parents":[{"commit":"a0af54161e1734ab150dd5d7aea22947eadf0b91","subject":"Merge \"rabbitmq: Bump to 4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0af54161e1734ab150dd5d7aea22947eadf0b91"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-17 10:53:15.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-04-18 08:39:14.000000000","tz":0},"subject":"opensearch: remove example certs and securityadmin_demo.sh","message":"opensearch: remove example certs and securityadmin_demo.sh\n\nChange-Id: Iceb16a5cf2ebc64effe30bbc9144144cdffebab0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cee097ec6c3969fa88d93b990efbb1e778953a51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cee097ec6c3969fa88d93b990efbb1e778953a51"}]},"branch":"refs/heads/master","description":"Rebase"},"3eb606a36ed421235863b638ef57ef5b26630fad":{"kind":"REWORK","_number":4,"created":"2025-04-18 08:43:03.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/38/919938/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/919938/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/919938/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/919938/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/919938/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/919938/4"}}},"commit":{"parents":[{"commit":"a0af54161e1734ab150dd5d7aea22947eadf0b91","subject":"Merge \"rabbitmq: Bump to 4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0af54161e1734ab150dd5d7aea22947eadf0b91"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-17 10:53:15.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-04-18 08:43:03.000000000","tz":0},"subject":"opensearch: remove example certs and securityadmin_demo.sh","message":"opensearch: remove example certs and securityadmin_demo.sh\n\nChange-Id: Iceb16a5cf2ebc64effe30bbc9144144cdffebab0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3eb606a36ed421235863b638ef57ef5b26630fad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3eb606a36ed421235863b638ef57ef5b26630fad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
