)]}'
{"id":"openstack%2Fkolla~920037","triplet_id":"openstack%2Fkolla~master~I64e9442545d98364585e3c632ae9142cd6b4f9bf","project":"openstack/kolla","branch":"master","topic":"bug/2066171","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-06-11 07:54:55.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-07-09 14:49:31.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2024-07-09 14:49:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I64e9442545d98364585e3c632ae9142cd6b4f9bf","subject":"Pass rabbitmq apt preferences into kolla-toolbox","status":"MERGED","created":"2024-05-20 13:36:36.000000000","updated":"2024-07-09 14:50:39.000000000","submitted":"2024-07-09 14:49:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920037-bug/2066171","meta_rev_id":"463918a0e0b62323dcde9aeed92fa506e1d0a7a7","_number":920037,"virtual_id_number":920037,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-09 14:49:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-07-09 08:42:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-06-11 07:54:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-09 08:53:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-07-09 12:53:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-07-09 12:53:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-07-09 08:42:43.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-20 15:26:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-20 17:46:25.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2024-05-24 10:33:16.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-07-09 08:34:50.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2024-07-09 08:42:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-07-09 08:53:41.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"af5f62edc9409017fcce22a3b2daa05940e65900","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-05-20 13:36:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b039b6fec90d81bfdd848e4887ea59c9e1ce24ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-05-20 13:45:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b52fb82d67139b6518a9174816a7acc67d1d2466","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 15:26:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0b361ed3ec145a28b93c5ff84bf76cc\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f5208303f83d42b1b6b1f56ab5895d0d : SUCCESS in 22m 51s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/258d4f8270cb4feb975fe715bac5a94f : SUCCESS in 1h 02m 31s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/66b29188b6af4b29b9f6a05afaee0ee9 : SUCCESS in 19m 27s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6198291751b14f7593c372159ea2c51c : SUCCESS in 28m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/da44a88811a648b7b1712d85ba940dab : SUCCESS in 1h 02m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/deed1f0bb8074329b0321ba329a0969c : SUCCESS in 1h 31m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/991f692dbad0439fadf4c142bf0ec51b : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5d32bb10346437c8699b7d254b86f94 : SUCCESS in 2m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bbe775c09baa4069b62bc70a13d53eac : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/df0ad4fc71e6414db3dfd5c8f1c20b4b : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77f19c127fd24926a6b3227cb9679ed8 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a036ad797ff4984947302a96cd6033c : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/70831c1b6006463db998a9b1b902e3fb : SUCCESS in 3m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5d411a2add5b48d59791046100a34686 : SUCCESS in 24m 57s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/dee96b2810534e4e8e4afab2cfb24f97 : SUCCESS in 21m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/50522b5a0b32401c842e6362436878cc : SUCCESS in 1h 02m 51s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c36c9c3f3b434215b73ca5b4c0285f1c : SUCCESS in 1h 32m 29s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4b05851f31e64060b7173f7cdbc48d81 : SUCCESS in 19m 36s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/262664d225624382a03975ff0b552f9c : SUCCESS in 55m 27s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0b6571c9b5dd4621a0e903647f62af03 : SUCCESS in 51m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/dae2d64af6f04fccb0f579560bc0e136 : SUCCESS in 1h 19m 27s","accounts_in_message":[],"_revision_number":2},{"id":"1cb1c9705c98eecb3815e13ff5194a86d1c5ea33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 16:33:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2d126edbede447b99d3be84c983e700\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/72a3e748eefa4641900ff4c16eddd0e3 : SUCCESS in 2h 31m 51s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7b62252f5a1b48e991fcaf3e164725eb : SUCCESS in 24m 47s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5a7839e0390443a4a72ebb263425ef81 : SUCCESS in 2h 10m 46s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a330e003e56341d0ae5444f3343947f1 : SUCCESS in 3m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/304a770c124341cd9531d6e7c96c2e1d : SUCCESS in 5m 08s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/978a37c9527e41b89f32dd38d3354381 : SUCCESS in 1h 10m 45s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7663501a50004819872c4a25a3cafe36 : SUCCESS in 2h 45m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ae8964e93c790e0463c39e641b0837aff3f4a942","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-05-20 17:46:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4414e0303819ca92cc2666595f51d63377abbb50","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-05-20 17:47:31.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"e8775713e3595b2284160859caf8602e2fbce957","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-05-21 08:21:03.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"23065e3c064e75a65fa6af7ab5573cf5f8a736dc","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-05-21 08:21:15.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"2006bcf2324e104b3e375ebc4e00901862984aaa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 08:31:16.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55331f4a1fb0402e914c5f683e825e33\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c968c2402a6b442789ce1745e3e0064b : FAILURE in 4m 09s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1cb25ceb28be48999917b53f7751d5b0 : FAILURE in 4m 48s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d15a5156b0194d9682c16c3ab518d30b : FAILURE in 6m 20s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4bd8939cad3143739a91f97d1be05216 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/131db9d18a88430693cddf8b3493b137 : SUCCESS in 2m 50s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a485f84356204969abf6909f281688d4 : FAILURE in 7m 40s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e116cb35f44848d7aa6e7b9d39b20f30 : FAILURE in 3m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d644e68239f89d407b663eb8a5025d11e7a26480","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-21 09:22:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b514b65a34b748e293294101052f89b8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a11f6408b9f74a3eb382de0c34bce862 : FAILURE in 4m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c700fb34bca344c486ed68f9d82dcffc : FAILURE in 8m 53s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b343a31137cd4b5cb536192c96573b28 : FAILURE in 4m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/34afd17eaf044938baf28c5bedd404b0 : SUCCESS in 30m 13s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f977961be3b0400780725ee9795d8e04 : FAILURE in 6m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/264d3bb96d85404d8eeb9c13b3ab8c92 : FAILURE in 11m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb4a31b9653e46cbbbc30cd97dd144ec : SUCCESS in 2m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/38b598ae859b4493aa3b2860a80a36ef : SUCCESS in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/edda4b37439946288e538b962b7cd659 : SUCCESS in 3m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b02c90aa92f34b068678af3b50063ccb : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59c066810b8b4d2dab8459979dd80104 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/375221ab42374bb7a4d5835b5de27729 : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2b63f621173462ca2af7efeb9d5315a : SUCCESS in 3m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8e8282514b874b0a8967e97799b2b75a : FAILURE in 4m 08s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/867f474ec8074c3bbafe26059b3a9a1d : SUCCESS in 35m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/085052e5730c443192a7495e65e60932 : FAILURE in 7m 32s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1cde4b616d854e6699b246d9b2a6f14c : FAILURE in 10m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3d14482580314f78809bfc45a4c7d6fa : FAILURE in 3m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/60814525366946fe970c64acf7c210b8 : SUCCESS in 58m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6de39606736b4eb4bd8afb2ab840266d : FAILURE in 7m 36s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cbb261137e8b4ef5896c6a7008cae1f8 : FAILURE in 8m 37s","accounts_in_message":[],"_revision_number":3},{"id":"08e2f26ebfb9244419c358ed1ca251287d4026b3","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-05-24 10:33:16.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f632e1a407608e93c419d1761c884cac2d81211b","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-06-04 10:21:36.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"621ad5f29fc58e16718497fe7f89500a7d166b56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-06-04 10:21:43.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2d5cad26151e3a10976a2d0bf8d2d7b0f6c6b77e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 12:17:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bd40df4bd7e4ef8acddbf6e31e2a2b1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/71137ba41ce3426b91ce4e4656fc59f8 : SUCCESS in 31m 39s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6f35e0fe6ea946a18bd104244e39374b : SUCCESS in 1h 11m 49s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/213744bd6c9043018772a960b12305b3 : SUCCESS in 23m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/cb41a3cb2e33482ab65eaaf3e7dce5c2 : SUCCESS in 29m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b4978f647b6d4e23b7be01b15a08b01d : SUCCESS in 1h 07m 44s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b1e26a5f3001460db7d2d2bd7a5c312d : SUCCESS in 1h 31m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10f147baee954b8ba54d767ebeacbdf5 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e6c6c56e492470490e6dc4d7847f92c : SUCCESS in 2m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a4a8a10846a6478b83302f008ec93e54 : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a3f1abd7286449d934431fd08722ac5 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4db566af7a014cf58b67d38c14d576ba : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3692b61f1c6540a7a63b7b4235855d3c : SUCCESS in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61dbf10e848c425091c92479524b6d1e : SUCCESS in 3m 26s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/14655dda354a41539bb4e47cc86a70cc : SUCCESS in 26m 28s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a4815d13d9af40e48c8e1c0416735503 : SUCCESS in 33m 49s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5184c107b27f44c08fe385bcc6dff9a1 : SUCCESS in 1h 03m 05s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/416c26ce2d944f98805fdb1f72c61cfb : SUCCESS in 1h 50m 16s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fdcf16a263f245aaa85f72e0603efb6f : SUCCESS in 15m 34s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/02a1b0aa01eb4bf182527232e578b04b : SUCCESS in 1h 05m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ceeaa7125c5a49d7bcb4d823f92753bc : SUCCESS in 49m 43s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ebbb22aeb17843e69d4824aa963319e4 : SUCCESS in 1h 18m 50s","accounts_in_message":[],"_revision_number":4},{"id":"2673f29a95bf6e526abf62d759f9ce5e5db36581","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 12:36:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0340d1113094c44855a928aad697718\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3cd5cb113431483c8c2793dee354e5a4 : SUCCESS in 19m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/744d4c268b5c4c3eb684381de1995844 : SUCCESS in 21m 08s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/686bdb7bf9624154b25ecd3236f4224e : SUCCESS in 1h 01m 04s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c0009e3f0e134bf093a2df92b431b657 : SUCCESS in 2m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f9a68a25ce0f4a1cb18095bfe64dda6b : SUCCESS in 4m 58s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7243c159c3ff4e2bbe57dfa03558405a : SUCCESS in 48m 05s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9e97fc7c6d4b4dde985b8b6dc82b9d8f : SUCCESS in 2h 13m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2368c56112349a467132bb82ad53f7fc02fff3c2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-06-05 10:22:06.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ff347027db76e17d7c527ffbe0c39a035bbf263c","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-06-10 14:44:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"15e2893944c43662d35d119c97eec0ed39a6fd46","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-06-11 07:54:55.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2f13946dd8e3f9aceebea5fc32fe60d23d6ddbd1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-09 08:42:43.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"4c75265990597133f5f872362e7ff4957f7d620b","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-07-09 08:53:41.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"b6d8a2b38938fbac0d9542096c1f3461f45b189d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-07-09 12:53:35.000000000","message":"Patch Set 4: Backport-Candidate+1 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b5641f0c9c408d5267eadc9dae4b95da198977a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 12:53:57.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7cfb6331e23e059a22b5e216b7bff8357614cf96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 14:49:31.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad5bcd503ee847f48226d3acf81756f1\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/506257f1070e481ba6fba1f06ae28821 : SUCCESS in 19m 15s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/61908f87c3ee44ffabc0645a2e20e402 : SUCCESS in 29m 26s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e65ec4a423bb4244b07dbb22cf28eae8 : SUCCESS in 56m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/890c04e07e754607a12960dbcd658868 : SUCCESS in 1h 38m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/914196a94fd447c6a31d90f30eb758e3 : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/508849dcd3f24c1ea20f3d2daee122b8 : SUCCESS in 2m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b601ba26efb4447a9a22e6d8b612f5cf : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8737901aec1c45128dfecf5a5dd39340 : SUCCESS in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69a06eb64be7421dbb92d27e64b20034 : SUCCESS in 4m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c504bf349aab4a2ca6b44c045ccd13d9 : SUCCESS in 21m 28s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/11788b85c1b044b4b5fee0665f7e68ef : SUCCESS in 35m 24s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1f7cf3702d254743ab4e110504088e23 : SUCCESS in 1h 16m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1feff74e665b42a2b55d8465e4b9d1ba : SUCCESS in 1h 28m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/202d9308f2c54f13b3a4da85270200d3 : SUCCESS in 17m 59s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/acac89353ef346d5a4aa8b0d99a5e45f : SUCCESS in 48m 37s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3d3b0cb50a30421b8ce32a7a3757f0ce : SUCCESS in 57m 21s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/89bcfcd2b007480ba46f17c7aa64ce28 : SUCCESS in 1h 20m 04s","accounts_in_message":[],"_revision_number":4},{"id":"8e21b6848f4e96751afb1b3dcb3a5d10d0fcc0ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 14:49:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"463918a0e0b62323dcde9aeed92fa506e1d0a7a7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 14:50:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd013f65eaa24054a25c4890f4f851e1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31cf5ac3763c496d94c87c2ca6b449dd : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ca1694243a84da6b754351150e3985d : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"006ccee101b3944eeb22299226d45520850dba0e","revisions":{"738d51211d6af561768978476b74ae113ef23e54":{"kind":"REWORK","_number":1,"created":"2024-05-20 13:36:36.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/37/920037/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/37/920037/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/920037/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/920037/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/920037/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/37/920037/1"}}},"commit":{"parents":[{"commit":"25f357ecc81bc2ad25a1e1b2f549d6a4b755b638","subject":"collectd: Mark unbuildable on rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25f357ecc81bc2ad25a1e1b2f549d6a4b755b638"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-05-20 13:29:01.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-05-20 13:36:19.000000000","tz":60},"subject":"Pass rabbitmq apt preferences into kolla-toolbox","message":"Pass rabbitmq apt preferences into kolla-toolbox\n\nIf the latest and targetted rabbitmq-server depencencies mismatch, kolla-toolbox will fail to build.\n\n```\nrabbitmq-server : Depends: erlang-base (\u003e\u003d 1:26.0) but 1:25.3.2.12-1rmq1ppa1~ubuntu22.04.1 is to be installed\n```\n\nCloses-Bug: #2066171\nChange-Id: I64e9442545d98364585e3c632ae9142cd6b4f9bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/738d51211d6af561768978476b74ae113ef23e54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/738d51211d6af561768978476b74ae113ef23e54"}]},"branch":"refs/heads/master"},"1cb107856e6be92e9df624a5e0fdb2a1b8a46ce2":{"kind":"REWORK","_number":2,"created":"2024-05-20 13:45:52.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/37/920037/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/37/920037/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/920037/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/920037/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/920037/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/37/920037/2"}}},"commit":{"parents":[{"commit":"25f357ecc81bc2ad25a1e1b2f549d6a4b755b638","subject":"collectd: Mark unbuildable on rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25f357ecc81bc2ad25a1e1b2f549d6a4b755b638"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-05-20 13:29:01.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-05-20 13:45:40.000000000","tz":60},"subject":"Pass rabbitmq apt preferences into kolla-toolbox","message":"Pass rabbitmq apt preferences into kolla-toolbox\n\nIf the latest and targetted rabbitmq-server depencencies mismatch, kolla-toolbox will fail to build.\n\n```\nrabbitmq-server : Depends: erlang-base (\u003e\u003d 1:26.0) but 1:25.3.2.12-1rmq1ppa1~ubuntu22.04.1 is to be installed\n```\n\nCloses-Bug: #2066171\nChange-Id: I64e9442545d98364585e3c632ae9142cd6b4f9bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1cb107856e6be92e9df624a5e0fdb2a1b8a46ce2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1cb107856e6be92e9df624a5e0fdb2a1b8a46ce2"}]},"branch":"refs/heads/master"},"a29088ec3d3e5bea5b050aba65dede946a8b6931":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-05-21 08:21:03.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/37/920037/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/37/920037/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/920037/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/920037/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/920037/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/37/920037/3"}}},"commit":{"parents":[{"commit":"25f357ecc81bc2ad25a1e1b2f549d6a4b755b638","subject":"collectd: Mark unbuildable on rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25f357ecc81bc2ad25a1e1b2f549d6a4b755b638"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-05-20 13:29:01.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-05-21 08:21:03.000000000","tz":0},"subject":"Pass rabbitmq apt preferences into kolla-toolbox","message":"Pass rabbitmq apt preferences into kolla-toolbox\n\nIf the latest and targeted rabbitmq-server dependencies mismatch,\nkolla-toolbox will fail to build.\n\n```\nrabbitmq-server : Depends: erlang-base (\u003e\u003d 1:26.0) but 1:25.3.2.12-1rmq1ppa1~ubuntu22.04.1 is to be installed\n```\n\nCloses-Bug: #2066171\nChange-Id: I64e9442545d98364585e3c632ae9142cd6b4f9bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a29088ec3d3e5bea5b050aba65dede946a8b6931"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a29088ec3d3e5bea5b050aba65dede946a8b6931"}]},"branch":"refs/heads/master","description":"Edit commit message"},"006ccee101b3944eeb22299226d45520850dba0e":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-06-04 10:21:43.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/37/920037/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/37/920037/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/920037/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/920037/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/920037/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/37/920037/4"}}},"commit":{"parents":[{"commit":"25f357ecc81bc2ad25a1e1b2f549d6a4b755b638","subject":"collectd: Mark unbuildable on rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25f357ecc81bc2ad25a1e1b2f549d6a4b755b638"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-05-20 13:29:01.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-06-04 10:21:43.000000000","tz":0},"subject":"Pass rabbitmq apt preferences into kolla-toolbox","message":"Pass rabbitmq apt preferences into kolla-toolbox\n\nIf the latest and targeted rabbitmq-server dependencies mismatch,\nkolla-toolbox will fail to build.\n\nCloses-Bug: #2066171\nChange-Id: I64e9442545d98364585e3c632ae9142cd6b4f9bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/006ccee101b3944eeb22299226d45520850dba0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/006ccee101b3944eeb22299226d45520850dba0e"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
