)]}'
{"id":"openstack%2Fkolla~920165","triplet_id":"openstack%2Fkolla~master~I50db1ec402e51c576ba3d70ec2df810fb05daa02","project":"openstack/kolla","branch":"master","topic":"ci_pin_requests","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-05-23 08:21:54.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-05-28 09:35:17.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}}},"hashtags":[],"change_id":"I50db1ec402e51c576ba3d70ec2df810fb05daa02","subject":"CI: Fix breakage due to requests bug","status":"ABANDONED","created":"2024-05-22 08:43:04.000000000","updated":"2024-05-28 09:35:55.000000000","total_comment_count":8,"unresolved_comment_count":3,"has_review_started":true,"meta_rev_id":"407f1a15a6f3778d424bdfdb6b97b88ef17cbb19","_number":920165,"virtual_id_number":920165,"owner":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-05-22 09:36:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2024-05-22 08:57:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":-1,"date":"2024-05-28 09:35:17.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-22 08:57:19.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-05-22 09:36:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9863c6124b574bad8b95a3b0ffca135eb59ef981","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-05-22 08:43:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f0a7adbdd0828f5628e7b6dc22446d3d7fdc4412","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-05-22 08:46:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7abed666d61fd129afecff70ff18365583dce383","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-05-22 08:51:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6c91a9b0e5275bf720916b0e116daa4d69e02c58","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-22 08:57:19.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"3e81ee198b7992d0bc0cd6b21809d13bab25e75d","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-05-22 09:01:28.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"880b8ee703d1d75b6512245f755e76af5b543c38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 09:36:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b5845fc6ec44d15878ae36c839ac13a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/bfae811d3a244137b426efd603595315 : FAILURE in 4m 15s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6ac83634424c428f9cba0dc24d8bfd11 : FAILURE in 7m 30s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cf1eaa9a287a42658c1b6be3b201a6d4 : SUCCESS in 20m 02s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/05857115740d4614807ad2d1539b57d8 : SUCCESS in 28m 37s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5b229bd6fef345c093a65a04f5dd3d49 : FAILURE in 6m 49s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/676fb8f668f24880a99045fea43dadef : FAILURE in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18678f19ea094570bbb49ad29440f7ff : SUCCESS in 2m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ae98d269c8043e5b30fd95fee72192e : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3b36f477c453485d9dedc24c0c4d0e05 : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eae0aa2f5c5e426383fd44bbb8490688 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d386313f0f3490ca43c2e3830a173a0 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7921700f2f5643cc9e81b06539675472 : SUCCESS in 4m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2e3674e45cad47ed9e53985fb33292a0 : FAILURE in 20m 33s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/cb2c2fc8ddbf4ff5a142a7d207fbc1ef : FAILURE in 7m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/21b583857f0344f184767dba7c25b9a5 : FAILURE in 8m 04s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/13bdec6ec8324e12b7d8041d8ffd51a8 : FAILURE in 8m 02s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/12ed7d9089da4c31a471ae8f26c9ac19 : FAILURE in 8m 49s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e69a02bc52f044a8963f64950cfdbecb : FAILURE in 8m 32s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/a380f2d64f2f4ccc9476435125628ea8 : FAILURE in 8m 03s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/836745c61c044379b0ed386863db3a11 : FAILURE in 9m 26s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/01e3e22145e84149ac779d17a549f9fc : FAILURE in 9m 25s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/0a6bc911e1d24722aa89e479ebc814bd : FAILURE in 8m 02s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/6d7d9846765f49ef8468bd6b286eee43 : FAILURE in 9m 25s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5eee567f9716462eb10b4b8bd77a6ad3 : SUCCESS in 22m 20s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/01a75c110517488a9a74221d9a4ef9f2 : SUCCESS in 43m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7b02c0584397429f9b8c1feb26c8d0ec : FAILURE in 7m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d3c23104c46348eda23eb72956db50ca : FAILURE in 5m 41s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/05955b95d1684a7d9e3cdc4067926d08 : FAILURE in 7m 22s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/da327568e336470d8b425dc47cfffb7e : FAILURE in 8m 33s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/09f8672ead944b9a8da2672198d476e2 : FAILURE in 6m 00s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/028c2e68ac5149e1a898900f3e9409df : FAILURE in 8m 50s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/8a5dcd9d6a774980a70278c8f552d4ad : FAILURE in 11m 33s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/d466765a9bca4348bfb2a55d6f78c493 : FAILURE in 7m 17s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/184c759d7fb947c697c10a2167973cfd : FAILURE in 8m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7b5b5432dbfcccf90c72a710c12bf43d23216ffc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 11:59:18.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1101661192a48f59e54d15687d3a91a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8efd2a3234654d04a91d33f969b30587 : SUCCESS in 20m 26s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/eef3062c86a54af496e5598b9cc16a44 : SUCCESS in 2h 29m 25s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/596e095306b0465a86f669deecb53184 : FAILURE in 6m 52s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d1bfb6c0bf7a48c7a19a40cefefaa6c0 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fc97d794e7714162b498429396861c0c : SUCCESS in 2m 20s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1a0a515dd5df4f9d924c8eae97c2969b : TIMED_OUT in 3h 05m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4054113bc40a4d7bbce7039dc1fb6d0c : SUCCESS in 17m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8fd7353b64850062580726d6742b27153b02b407","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-23 08:21:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1d7a718f2d1c18b44ef04d853424dea567a27439","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-23 08:22:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"98bb9fb4d94fb7e151a6a62f8453d38c6ffc0259","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-23 08:24:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3528eabdead640f47b68ace4143d13e360297110","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-05-28 09:35:17.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"407f1a15a6f3778d424bdfdb6b97b88ef17cbb19","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-05-28 09:35:55.000000000","message":"Abandoned\n\nsuperseeded by https://review.opendev.org/c/openstack/ansible-collection-kolla/+/920172 and others.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"20cf8223562db2b798857bedd499fee4eb0f6343","revisions":{"b9361cef5ef2af589d655f68beed15604669dfbe":{"kind":"REWORK","_number":1,"created":"2024-05-22 08:43:04.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/65/920165/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/65/920165/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/920165/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/920165/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/920165/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/65/920165/1"}}},"commit":{"parents":[{"commit":"25f357ecc81bc2ad25a1e1b2f549d6a4b755b638","subject":"collectd: Mark unbuildable on rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25f357ecc81bc2ad25a1e1b2f549d6a4b755b638"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-05-22 08:42:20.000000000","tz":120},"committer":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-05-22 08:42:20.000000000","tz":120},"subject":"CI: Fix breakage due to requests bug","message":"CI: Fix breakage due to requests bug\n\nmore information: https://github.com/docker/docker-py/issues/3256\n\nChange-Id: I50db1ec402e51c576ba3d70ec2df810fb05daa02\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b9361cef5ef2af589d655f68beed15604669dfbe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b9361cef5ef2af589d655f68beed15604669dfbe"}]},"branch":"refs/heads/master"},"20cf8223562db2b798857bedd499fee4eb0f6343":{"kind":"REWORK","_number":2,"created":"2024-05-22 08:51:09.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/65/920165/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/65/920165/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/920165/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/920165/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/920165/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/65/920165/2"}}},"commit":{"parents":[{"commit":"25f357ecc81bc2ad25a1e1b2f549d6a4b755b638","subject":"collectd: Mark unbuildable on rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25f357ecc81bc2ad25a1e1b2f549d6a4b755b638"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-05-22 08:42:20.000000000","tz":120},"committer":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-05-22 08:51:01.000000000","tz":120},"subject":"CI: Fix breakage due to requests bug","message":"CI: Fix breakage due to requests bug\n\nmore information: https://github.com/docker/docker-py/issues/3256\n\nChange-Id: I50db1ec402e51c576ba3d70ec2df810fb05daa02\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20cf8223562db2b798857bedd499fee4eb0f6343"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20cf8223562db2b798857bedd499fee4eb0f6343"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
