)]}'
{"id":"openstack%2Fkolla~920381","triplet_id":"openstack%2Fkolla~master~I3b878204c3ccfd1dc32a8387884ec67cbc2469b7","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2024-05-28 08:29:07.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-05-28 08:29:07.000000000","reason":"Change was abandoned"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-05-28 08:29:07.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3b878204c3ccfd1dc32a8387884ec67cbc2469b7","subject":"Add packaging as an explicit dependency","status":"ABANDONED","created":"2024-05-24 09:30:13.000000000","updated":"2024-05-28 08:29:07.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a270a2b8061f22f2daafab37e84074d9cbab2494","_number":920381,"virtual_id_number":920381,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-05-27 20:52:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-24 09:41:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-24 10:14:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-24 10:42:20.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-05-27 19:06:17.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"}],"messages":[{"id":"972b93a1c537ddf96407841541d29da31832fe5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-24 09:30:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5a866c71fb45c0c41137f4c15c48e2743e5238b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 09:41:16.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74536670834b43ad8bbbbeaef804149e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/cf6ff8df98b14c14941b946d1fe3e259 : FAILURE in 4m 12s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/307be19894f14ec3aa34e9e37d7bdbb0 : FAILURE in 3m 10s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4d73bf17a7694d558f3aa6da9a0af2f0 : FAILURE in 7m 13s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ed0c77861a9d4940947e27cadcdc604c : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/38654afc07164703ae5a4f0ccb93c3e5 : SUCCESS in 2m 51s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b79edc6668d14aeb966d284708d7984e : FAILURE in 8m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/75535dd8e95f4201b202259265a7cecf : FAILURE in 4m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bb5580b5a886bf33c7706411b2b0558fe2a3c392","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 10:14:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0971c4afd99e49cfb8db9bbf84ee7ac4\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a3501d2a849c4116aa978fe1ab660abf : FAILURE in 4m 16s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8d5629e6fde6419194b08cf006cd13e8 : FAILURE in 6m 48s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a5831b87610d44ee845a12f2c252929c : FAILURE in 3m 15s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/92e77501991d4aeaa040f7ad6937e140 : SUCCESS in 22m 58s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4d387d8bccce4ae6b67ad9ff7f162c89 : FAILURE in 7m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f876e662b16f4bbbabb72b1352067e7b : FAILURE in 19m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ab19dc23f99477cb038bb5012f0301f : FAILURE in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9669acebc3d44847b1df6be57e61bd9a : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b623715a3bbc407390c2dba33ca23d98 : SUCCESS in 2m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c87b3240f913437e92eb7957ac9fee3e : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e86142224c3947aab9fac8b85c2fc09e : SUCCESS in 2m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d3cf7a89590a41079abb3a18dc4b40ea : FAILURE in 1m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e9fa74cef5a434db918d484ec911fad : SUCCESS in 3m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8c28fe771b8479b81c08314176a1066 : SUCCESS in 3m 23s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/30624196f15c4c32acc96fa6f33b0945 : FAILURE in 4m 26s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e21d9313a9164aa888983aff4f9c31d0 : SUCCESS in 35m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/098467b1e710488d9326b770f34674bf : FAILURE in 8m 17s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1e5ef9a0955b4be9b17650db3282c61a : FAILURE in 7m 33s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6ee412c4314c492291774018b68e8e88 : FAILURE in 3m 31s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/109cc1bc01a1434eb7321ad7ef4f6893 : SUCCESS in 43m 00s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4195a91a878843f6a54d7545986e6143 : FAILURE in 6m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1e3b961b2f884fdea94609e55e839e69 : FAILURE in 5m 59s","accounts_in_message":[],"_revision_number":1},{"id":"af628c0ddc920319300c432c7b65613b0dc48734","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-05-27 19:04:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"08fa19828fa060a4d581d0311fda1795b7e57ffb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-05-27 19:06:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"829b6139e5e6940147103b24113faafe31d86377","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-27 20:10:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26eb313b56f94cff9f43bcba593c21d9\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/cc5bbe5bdf144fedaf51096bb580c02c : FAILURE in 5m 06s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/72ae3378ea2549ba84473a169023c687 : FAILURE in 3m 16s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/53cde5dc534c466c9a9f028929967b01 : SUCCESS in 1h 05m 28s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/99bfb1053eeb43d487a18887eabe21ec : SUCCESS in 2m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/662b63a1dda04aff986c93f0445ce634 : SUCCESS in 2m 43s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/95f06d1097f74792ba1b39eaddae393c : FAILURE in 5m 00s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e629568d01f24f538ef30f812440c736 : FAILURE in 6m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7646dd3b9c3ed191a74df3f69b40ff1c458ab407","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-27 20:52:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/09b74de215aa44cab1dad3673c958db0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/041cf98f06b74efda7e307f95bfed065 : FAILURE in 3m 52s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8807100b294949b29549b8e9126e8b9c : SUCCESS in 53m 21s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/445b274af8e14c0dbdb319151657dd6f : FAILURE in 3m 29s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/519cc0bf6d234040a24b6445177c8af1 : SUCCESS in 22m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/edbbd16422234bb3a6865fdb6ae83112 : SUCCESS in 50m 53s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9d559c6a7352493d8d750def43ee8814 : SUCCESS in 1h 25m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ac40fdb9a1149409203f7a2eb697437 : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/837e492794244710b6a54b71c30628c3 : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e126c6ff8ae0480fadb4154289c8362d : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/34a6dc11108846bcb80858eb588e2516 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e3fbf6fd2664f97b346742131bfba38 : SUCCESS in 5m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/162709ac5c254cf581fe0cdbcf647afa : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eec8fc5ec13d4e6c948b20483eea1413 : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2c19e39e30764b9a876e889712d3b8c3 : SUCCESS in 4m 46s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/60bcb903914c4057939112111ca9c4ab : FAILURE in 3m 50s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/46cd1aecdfc549bb859bfeb8ef14bd46 : SUCCESS in 30m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/725932e6c6b945229205a064ab82578d : SUCCESS in 1h 05m 30s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d077addc35814588b569e7efe7c91a31 : SUCCESS in 1h 39m 21s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7330d58f36564127810d7413f45ee706 : FAILURE in 3m 11s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6a76f136eac4492abffd923497bb37f9 : SUCCESS in 43m 50s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/793a4222f3b8449ea925374867fbbde5 : SUCCESS in 51m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3251df34d77940258b1184f1cc745606 : SUCCESS in 1h 06m 58s","accounts_in_message":[],"_revision_number":2},{"id":"a270a2b8061f22f2daafab37e84074d9cbab2494","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2024-05-28 08:29:07.000000000","message":"Abandoned\n\nAbandoning in favour of: https://review.opendev.org/c/openstack/kolla/+/920279","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c0046472e2c595a8a344c4a86f92ea91c552892a","revisions":{"28625d300b0d10a4076c4eb17607047eb7e6dd73":{"kind":"REWORK","_number":1,"created":"2024-05-24 09:30:13.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/81/920381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/920381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/920381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/920381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/920381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/920381/1"}}},"commit":{"parents":[{"commit":"25f357ecc81bc2ad25a1e1b2f549d6a4b755b638","subject":"collectd: Mark unbuildable on rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25f357ecc81bc2ad25a1e1b2f549d6a4b755b638"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-05-24 09:28:03.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-05-24 09:28:03.000000000","tz":60},"subject":"Add packaging as an explicit dependency","message":"Add packaging as an explicit dependency\n\nWe are using this package here:\n\nhttps://github.com/openstack/kolla/blob/25f357ecc81bc2ad25a1e1b2f549d6a4b755b638/kolla/image/build.py#L114-L122\n\nBut we don\u0027t explicitly install it.\n\nChange-Id: I3b878204c3ccfd1dc32a8387884ec67cbc2469b7\nCloses-Bug: #2067042\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28625d300b0d10a4076c4eb17607047eb7e6dd73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/28625d300b0d10a4076c4eb17607047eb7e6dd73"}]},"branch":"refs/heads/master"},"c0046472e2c595a8a344c4a86f92ea91c552892a":{"kind":"REWORK","_number":2,"created":"2024-05-27 19:04:08.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/81/920381/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/920381/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/920381/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/920381/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/920381/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/920381/2"}}},"commit":{"parents":[{"commit":"91f91e1cc28942870e0fd01f039872c24d62c99b","subject":"CI: pin codespell to \u003c2.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91f91e1cc28942870e0fd01f039872c24d62c99b"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2024-05-24 09:28:03.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-05-27 19:01:19.000000000","tz":120},"subject":"Add packaging as an explicit dependency","message":"Add packaging as an explicit dependency\n\nWe are using this package here:\n\nhttps://github.com/openstack/kolla/blob/25f357ecc81bc2ad25a1e1b2f549d6a4b755b638/kolla/image/build.py#L114-L122\n\nBut we don\u0027t explicitly install it.\n\nThe requirements-check job forces us to pick a lower bound. Pick 22.0\nwhich is the first release officially supporting Python 3.11 [1].\n\n[1] https://github.com/pypa/packaging/releases/tag/22.0\n\nChange-Id: I3b878204c3ccfd1dc32a8387884ec67cbc2469b7\nCloses-Bug: #2067042\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c0046472e2c595a8a344c4a86f92ea91c552892a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c0046472e2c595a8a344c4a86f92ea91c552892a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
