)]}'
{"id":"openstack%2Fkolla~920405","triplet_id":"openstack%2Fkolla~master~I3995fa8124f9ec549fb48aa16d801898d33c8ac2","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2024-11-22 10:35:56.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3995fa8124f9ec549fb48aa16d801898d33c8ac2","subject":"Restore Apache Zookeeper service","status":"ABANDONED","created":"2024-05-24 16:40:22.000000000","updated":"2024-11-22 10:35:56.000000000","total_comment_count":8,"unresolved_comment_count":5,"has_review_started":true,"meta_rev_id":"031e5d69b4da4788227191f5942e12f4ca6fed6a","_number":920405,"virtual_id_number":920405,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-05-31 14:21:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"disliked":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2024-10-22 11:41:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-08-13 12:44:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-24 17:53:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-24 18:07:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-13 12:44:28.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-10-22 11:41:18.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"e3beb85c039caca7691b62bbd4f222957fc50533","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2024-05-24 16:40:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bde64cbd1b68b93351ab9fa67a757ce1c0b0f265","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 17:53:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f75c0a76bb314a6e8f4f82b88f6b5202\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/50a8973cbaa240e3bdf9646e65a92649 : FAILURE in 4m 56s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3363f7031ed848fba5d41d5f590e19c8 : FAILURE in 4m 09s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a719df674ffc4fc0878e4a1a48dc1490 : SUCCESS in 1h 11m 55s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/41b9eda6a9e14abb8cd17eb49877a583 : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0b049419b3944d6fa143f21b4dfc7780 : SUCCESS in 5m 10s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f782c8c923a848cea6daaec465a0e1e3 : FAILURE in 5m 28s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/fb0123850e1246a4837cde1c7e9b798d : FAILURE in 6m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bc466ae7600ec20ef471aec0b6dcd12276775647","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 18:07:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/404f69f278ae42c594d0ed51628ad823\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a9fa9f8e6bdb4b4eb16d108ab86fb59f : FAILURE in 4m 08s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c74a06c273d940cc8361ca05d516c4ab : SUCCESS in 49m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c4b14c26304f471fafb6dae8f829bf79 : FAILURE in 3m 07s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8785a9951b4c4273954977e0a2f82eb7 : SUCCESS in 19m 19s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/87d70dd7e2b443c997ceb7f1360aadbc : SUCCESS in 1h 03m 57s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/dbccbd837c0c4fd8b52f39970aa5943a : SUCCESS in 1h 07m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6b0fa24256c46bda522510bc50594cb : FAILURE in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88c099ca030a4f958aad6ec7c341e02d : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb5e60264a654c24bca892320b180918 : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/deacc5d917dd4cc1b48a8c436a41c6db : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5645eef26bc495f9b335725dd0e419e : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d094b57db58d4572838398d83246a195 : SUCCESS in 5m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e28a5c7c385444759349c61bbd6999f2 : FAILURE in 4m 33s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3abdc6310a174f2fa1ddfd5ff787fe96 : SUCCESS in 34m 03s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7e905d3cce2f42369737e28ebaa2eab2 : SUCCESS in 1h 02m 40s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e0070148185b41bab68c0e5c3f1cb9ce : SUCCESS in 1h 26m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c0d8a07e170f48fd9fffa9b48058a2ac : FAILURE in 4m 21s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2780a606b23e469da9b27c4159f170c5 : SUCCESS in 57m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d5abca9b168c4d3abb20e50f116d61de : SUCCESS in 51m 07s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6d0dbe4a3c7c4beeb314758d6105abf1 : SUCCESS in 1h 16m 24s","accounts_in_message":[],"_revision_number":1},{"id":"09ff14770bf2194ffc87018faf29e23db0f1afea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2024-05-31 10:00:34.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"037e10fb2ab50feb477af081901042985f7ff600","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 11:48:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9d7587e9f534914909cf4ef4ac4d067\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/df2d77cea23846d3a3cae415209413ee : SUCCESS in 25m 10s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5a68eab41bf2443eae686b963b7fdfe3 : SUCCESS in 1h 09m 27s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0d7283298c104bfcb310374693073977 : SUCCESS in 24m 14s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7c4c2e64d6c7487b8735e28af9988907 : SUCCESS in 20m 58s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a58e497732b14e87b23e7829cbe5fe5a : SUCCESS in 1h 06m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0fa94967aa944842814c7a7cba394a76 : SUCCESS in 1h 39m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae7bebe43ef44eeba5a79513e53056ef : SUCCESS in 2m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb1157668daa495d8ff2d866f912c56e : SUCCESS in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5e72d5d03599443a9d91fb96d3596983 : SUCCESS in 2m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/42832435940c4abaa0d730f52421510f : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20e6a2ea84314887af3806ca3208b4dd : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a00860acecc144a492ae10c25a23edab : SUCCESS in 4m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/53f80d2653b942d296c818dd4e3c247c : SUCCESS in 24m 33s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b419f8e622b645bcb083ebb5d3d29a9f : SUCCESS in 34m 57s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/20c889731a484b2aa30248418d568155 : SUCCESS in 1h 13m 13s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e895cd1c3fac41f0b233d1cb7c6d0994 : SUCCESS in 1h 33m 34s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/62386658a33742c0af3dc87cb90caf37 : SUCCESS in 22m 59s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b15a06361ad249bc90c25c100ef44cb5 : SUCCESS in 42m 35s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6588766063724172b8401bb30a5db682 : SUCCESS in 55m 43s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/70e0f51d706a46dfba537fa41915beb0 : SUCCESS in 1h 20m 33s","accounts_in_message":[],"_revision_number":2},{"id":"9f3ec3ecd69996873fe473457e5897fcea3636e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2024-05-31 12:30:09.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"acf49f38116e731b0aa51c4ade278a85ce3a4e70","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2024-05-31 12:41:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"49d2d027fa68b6281791b090e788b8b9810e77ae","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2024-05-31 12:49:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1ee8269c7d46b33cad588270a632291a1496117f","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2024-05-31 13:15:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"12abd53f7a9491733782996b61010b6c7c3311af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 14:21:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b7a0d7c9a21422c9a4dc252108d8df5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0d1ac4b57a534955b9b6babc6290a957 : SUCCESS in 28m 53s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5457a54f2c254d39aa54e7791cac6a08 : SUCCESS in 1h 04m 41s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4d92fc7d29b045e38034613eefcd9367 : SUCCESS in 21m 27s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fb45a18b158349f9a6f4c4c45e8d3d76 : SUCCESS in 30m 41s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/cc4a8768152c457ca265423f3550f63a : SUCCESS in 1h 05m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/31655a6f74544a1492598f79731e183f : SUCCESS in 1h 16m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c15338921fa447ec8053db619a447df9 : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bbf28c46a3743b898cc688d0a62a6fb : SUCCESS in 2m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7e9ed7850b45431db64c37d279631138 : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3c2befde959d4f69a6c10446f4188e9e : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e0071c8514b40eaa81556649b6f2687 : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fe3b74f150d474caab99e28ab3928ed : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/044ae546a4ad4d55bf3d54b31a5ed456 : SUCCESS in 3m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c89cee7924ec46e893df83ad582121e6 : SUCCESS in 22m 47s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/279a031186054c18a596658891c3ceef : SUCCESS in 35m 05s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/26db53101f6d40228b101111aa135065 : SUCCESS in 1h 07m 13s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/fcb6e45931a5437ea6fb6e35393b783d : SUCCESS in 1h 46m 02s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/67f53d2c72b3449d82f6fe220c675cdf : SUCCESS in 23m 04s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d5266ba2258a47ffb3e5efaf70709cae : SUCCESS in 43m 50s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9c85325884bf478daa9cc6dfc4351615 : SUCCESS in 44m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1d5b5e2732a241d4ac3801350bfd07fb : SUCCESS in 1h 08m 25s","accounts_in_message":[],"_revision_number":3},{"id":"85a12077162b91c0b1cf433fda294141d9606433","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 14:36:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34a4fe3a2f4848b7a94bf6543488789e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/022cee8761594e80ac641d9f690e7fc0 : SUCCESS in 22m 30s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/38b1ff9db62c4b9e8fe602d48f31b2f1 : SUCCESS in 20m 05s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3a8ad5c4c901494da9d4c791da721e7a : SUCCESS in 2h 03m 51s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cb7cbffabe454fe48428b9efa4b6d574 : RETRY_LIMIT in 1m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/aba32e135c5b4bbeaba076f8afb2a1cb : SUCCESS in 5m 41s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7e8bea20bc284fb3b4782e8150dcdb81 : SUCCESS in 54m 57s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8dbee1dd68144218bdeb5687c2947a0a : RETRY_LIMIT in 1m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bf641e5f73e2b84721450897158657f70bcf53e8","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-08-13 12:44:28.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"2b80bace3ff45759287ffddc906b24537ab59087","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-22 11:41:18.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ac360a64fd9968273443895df719263beea0d754","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-22 11:41:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bb265fc9daa60c5b40ddc1a327eb4cc307b628d1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-22 11:42:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"70a00fe6e64f998ad1081ea57b944d5394ab21eb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-22 11:43:06.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"031e5d69b4da4788227191f5942e12f4ca6fed6a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2024-11-22 10:35:56.000000000","message":"Abandoned\n\nThis was requested in the Dalmation PTG, but the decision changed at some point","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6ff63be72785a77499d0a1f2cad5690ed2143932","revisions":{"baa33b0b98edf1714a27d9005a46a804b820795c":{"kind":"REWORK","_number":1,"created":"2024-05-24 16:40:22.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/05/920405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/920405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/920405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/920405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/920405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/920405/1"}}},"commit":{"parents":[{"commit":"25f357ecc81bc2ad25a1e1b2f549d6a4b755b638","subject":"collectd: Mark unbuildable on rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25f357ecc81bc2ad25a1e1b2f549d6a4b755b638"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2024-05-24 12:22:22.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2024-05-24 16:39:14.000000000","tz":60},"subject":"WIP: Restore Zookeeper","message":"WIP: Restore Zookeeper\n\nBring back Zookeeper for use as a Tooz backend.\n\nThe reference Tooz driver is Zookeeper.\n\nThis partially reverts commit 52a53927f5d390886c3940e364fe9fa888e3bc98.\n\nChange-Id: I3995fa8124f9ec549fb48aa16d801898d33c8ac2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/baa33b0b98edf1714a27d9005a46a804b820795c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/baa33b0b98edf1714a27d9005a46a804b820795c"}]},"branch":"refs/heads/master"},"46e01c70b49977b4af48f493d05c33edea81317f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-05-31 10:00:34.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/05/920405/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/920405/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/920405/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/920405/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/920405/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/920405/2"}}},"commit":{"parents":[{"commit":"968711579fbe73316227bd2b1993c5656adad9f4","subject":"Drop packaging requirement and unpin docker-py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/968711579fbe73316227bd2b1993c5656adad9f4"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2024-05-24 12:22:22.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2024-05-31 10:00:34.000000000","tz":0},"subject":"WIP: Restore Zookeeper","message":"WIP: Restore Zookeeper\n\nBring back Zookeeper for use as a Tooz backend.\n\nThe reference Tooz driver is Zookeeper.\n\nThis partially reverts commit 52a53927f5d390886c3940e364fe9fa888e3bc98.\n\nChange-Id: I3995fa8124f9ec549fb48aa16d801898d33c8ac2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46e01c70b49977b4af48f493d05c33edea81317f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46e01c70b49977b4af48f493d05c33edea81317f"}]},"branch":"refs/heads/master","description":"Rebase"},"6ff63be72785a77499d0a1f2cad5690ed2143932":{"kind":"REWORK","_number":3,"created":"2024-05-31 12:30:09.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/05/920405/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/920405/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/920405/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/920405/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/920405/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/920405/3"}}},"commit":{"parents":[{"commit":"25f357ecc81bc2ad25a1e1b2f549d6a4b755b638","subject":"collectd: Mark unbuildable on rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/25f357ecc81bc2ad25a1e1b2f549d6a4b755b638"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2024-05-24 12:22:22.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2024-05-31 12:29:01.000000000","tz":60},"subject":"Restore Apache Zookeeper service","message":"Restore Apache Zookeeper service\n\nThe reference driver for Tooz, which is used as a coordination\nservice by some OpenStack services, is based on Zookeeper [1,2].\nWe therefore bring back Zookeeper for this purpose.\n\n[1] https://zookeeper.apache.org\n[2] https://docs.openstack.org/tooz/latest/user/drivers.html#zookeeper\n\nThis partially reverts commit 52a53927f5d390886c3940e364fe9fa888e3bc98.\n\nChange-Id: I3995fa8124f9ec549fb48aa16d801898d33c8ac2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ff63be72785a77499d0a1f2cad5690ed2143932"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ff63be72785a77499d0a1f2cad5690ed2143932"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
