)]}'
{"id":"openstack%2Fkolla~920628","triplet_id":"openstack%2Fkolla~stable%2F2024.1~I11d306be5e5bb3d5b05b50dd8ee64ac9424854a1","project":"openstack/kolla","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-05-28 23:39:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I11d306be5e5bb3d5b05b50dd8ee64ac9424854a1","subject":"Drop packaging requirement and unpin docker-py","status":"MERGED","created":"2024-05-28 12:13:52.000000000","updated":"2024-06-04 10:34:54.000000000","submitted":"2024-05-28 23:39:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"920628","cherry_pick_of_change":920279,"cherry_pick_of_patch_set":16,"meta_rev_id":"d8af81203722f1d9114b7b89c7adbc3f9e4f4159","_number":920628,"virtual_id_number":920628,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-28 23:39:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-05-28 14:16:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2024-05-28 22:05:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-05-28 22:05:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-06-04 10:34:54.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-28 15:17:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-28 15:23:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-28 22:05:24.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"9dc6e3261be6829cf4c932fd35d8874bd4bbf4f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-28 12:13:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a656b85232dee8e700e7521a5f1d5813a9e779ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-28 12:20:16.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c55639640a7e8fdeb9f07f14c68317bc1054fa70","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-28 14:16:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f46e383d7f1df6c7593ab51468fd79e6c0d69856","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 15:17:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bab64bc231984a08b059ba7706ec1322\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d083804b2171443bab049f85246d06b8 : SUCCESS in 2h 02m 58s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8970e48a920d48a7b23ccf4d81d27842 : SUCCESS in 2h 05m 34s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7d8dfbd9936146c8b368190544da5cad : SUCCESS in 1h 08m 02s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/93af27b204a64fc4b9f997cb06c22bd2 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6a4a720009644a0aacbb78a6a073d02e : SUCCESS in 3m 38s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5f6aa5a4093a425ba8c0dd2d0089d5c1 : SUCCESS in 2h 55m 16s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e66625da34bf400d8d65e6cefeb8a24e : SUCCESS in 21m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"110a68b98dca64ea4b2867300dfde7f182475467","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 15:23:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/877011bcd75f42e3abb0988532c5a8e7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9ba5ebc4da8e4eb1a587addf35e07efd : SUCCESS in 29m 32s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/eb5aa736016f4a1e86073e2d65ee7482 : SUCCESS in 54m 41s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5a0abf289d9c48c4b02eb01ea996f8b4 : SUCCESS in 18m 31s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4c83c1875f83452ca317edc2a54f2998 : SUCCESS in 27m 42s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/284fbbc614564a6b9542ec09e7b01e5e : SUCCESS in 1h 08m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7c7036a4a4854cc484b33cc67b08814d : SUCCESS in 1h 25m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04cd48b5cdbb40308cf04e955c6c77e0 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ff5a1cab5a34ba48df327d94318d7aa : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5fde53edc7e84a85b97ec8acbf93a220 : SUCCESS in 2m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42f5db8a50964271b02c5c278d1704c3 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95110d8d160a4acb95fc34f80cb81604 : SUCCESS in 4m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/77f5f42234dd49d58a581c9b696f177a : SUCCESS in 20m 33s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5bd11d95235f45f5be594e63382b8a96 : SUCCESS in 26m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0c964efe27154934b31459b1a533de0d : SUCCESS in 57m 14s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f3a077be6b2d4ce3a1137c8951d0d0b1 : SUCCESS in 1h 24m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/985c0e62c1634ba6b4438e7f3bb1d018 : SUCCESS in 27m 49s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/df50df2d055b47ba82801382ed38024f : TIMED_OUT in 3h 02m 28s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/cdf1130d18a64afe8f6eb75792822e36 : SUCCESS in 53m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/17a9100220a643b8b75e7bd586bef122 : SUCCESS in 1h 20m 56s","accounts_in_message":[],"_revision_number":2},{"id":"e01538a0a765dd43f45411eaa8b2bc20351e033a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-28 15:28:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c6739a24f766bcaeadaf274ae59c8238450bd50f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 16:52:04.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e3fc380cad844931a144c05b68c325ab\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1d23ab3a921647f18f4a04ff6631172b : SUCCESS in 27m 08s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/da3addd6569e4f24b5013dbb6cca1695 : SUCCESS in 1h 06m 40s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/83e5e98362004f31933c4737906fc075 : SUCCESS in 20m 13s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ce1f190e4bd44777bd27bcc8da2e452e : FAILURE in 30m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/adee73f5f2384ca892acf287a54dbac1 : SUCCESS in 1h 04m 52s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/52993d2a071e46828920306318dc24e6 : FAILURE in 22m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/170eb9305e3f4b5f883dbfbbdcd9eec0 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1edd62c143604021a0fee1c406b12188 : SUCCESS in 4m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0cdf8beeb99348a5a78c76370bad65ae : SUCCESS in 4m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b05d9bcc1c234ba69ca457268babf2e9 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/014cecc6416f4256b4632c555f6d5885 : SUCCESS in 5m 55s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/74a06baea9b64c6380b18410981fc3b3 : SUCCESS in 19m 21s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e03e4872282f48bdb38c50492e83e502 : SUCCESS in 24m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1be429a736af41d19e832cdd080723a9 : SUCCESS in 1h 02m 45s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/94b0195b38e748cab8bf74f750a6eeb4 : SUCCESS in 1h 22m 14s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/aef09eeeb8be4c178fd01cc41270ea26 : SUCCESS in 17m 36s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7ac064b740b3418bbe70d2b3638f317a : SUCCESS in 55m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/95bee29f594d4f12b8f775fd19448e85 : SUCCESS in 55m 29s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c3a79b8519ee4606a86d05e7a615d8e2 : SUCCESS in 1h 15m 41s","accounts_in_message":[],"_revision_number":2},{"id":"a5256cc3fea66da59e9e41cda3a061b4ecceea2d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-28 17:09:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4b741b7ab88221eaf62ff2d52942143c999065ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 18:34:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d994f2d1b62c4eccb222d7e65130ba97\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/68a9154d6119461da56e8888c9abe802 : SUCCESS in 26m 46s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/166223be02584180bd454af785ee5f54 : FAILURE in 2h 28m 08s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4af0375a95554cc1a8e0ee6b7c75d258 : SUCCESS in 1h 10m 36s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dc2b362d3eec4ed785fdc75a7bed75d4 : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ca5e68f457cc49cabb586585a0036fa0 : SUCCESS in 3m 22s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8c6087e6894d4bea946faf563dacf468 : SUCCESS in 3h 03m 11s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1bd816cd57894c19ba7ecf1361d46b69 : SUCCESS in 24m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"32aa2500f2821ead0508005cb37817e690b5658d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 19:07:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/941ceeedbc2c46f68292e6ac5337144b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/baa0566941f24803b7332016849f6081 : SUCCESS in 26m 45s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/58179a207e35405ea7bd3a5e748ab9da : SUCCESS in 50m 50s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9a8662a4897d4082b6a4b29799e52a78 : SUCCESS in 19m 47s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/637057ad1e60464f9731cff35ae13d70 : SUCCESS in 35m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1338054647164676915ca6c0188cb9bc : SUCCESS in 57m 36s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/980d440bb68044a38f7ecd3c1b4ae707 : SUCCESS in 1h 25m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/facfee206a2742b1a6aba4b4a4be7155 : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f05d43a9190b4d40beb5518abb64be57 : SUCCESS in 2m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c790d8d1128f45499e7f256ebd04c0d2 : SUCCESS in 4m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6dbcd1f9acab45eb80ee424e61739abc : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41b05e7ad66a4c3290ffef8d2d78dab2 : SUCCESS in 7m 57s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2b8487fab62d417187c95c5a120fc5af : SUCCESS in 18m 40s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e9cd19b0714242488cfec9345a4deb9e : SUCCESS in 25m 26s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ed9f9abe81e84d0482620c35074a5f2f : SUCCESS in 1h 04m 05s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ba3790a755f64a66b5790e343da46499 : SUCCESS in 1h 35m 28s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0f673cd1378641f48a884abb2259a8d7 : SUCCESS in 20m 56s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6ab193af8cc641bfb9e504bbebb4c5c0 : SUCCESS in 43m 08s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/402228c3d1794d28a74efbd00e233390 : SUCCESS in 48m 07s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5aca070a1c45444684a093a931f3da23 : SUCCESS in 1h 23m 09s","accounts_in_message":[],"_revision_number":2},{"id":"56dac4f5bb1f6d9724357de8ba526a47a3786605","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-05-28 22:05:24.000000000","message":"Patch Set 2: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c7dd2bd87ea956e11e010a7931635467cd213129","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 22:05:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d514b215708310466e472428414e6b4744ab9f9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 23:39:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2803508e88944be8bc14ee8879b4625\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ffabcd1a9e984708952440ef53ba0aec : SUCCESS in 21m 05s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/80550ff777a54ecbb7dbed31f30e8a5c : SUCCESS in 28m 26s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/423ce20a8bc14d8390098a752c0663e6 : SUCCESS in 55m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6fcff8b928cc47788b44522fbaab14e1 : SUCCESS in 1h 16m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb0b711a6a8c456aaeee9134211099be : SUCCESS in 2m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/886659fc932349fe9fb439a7fbfada6d : SUCCESS in 2m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/961662cb394544588bc74051e7b8d97b : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6dd5ed81962475fa953af2957991ba5 : SUCCESS in 3m 52s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9f049228e18648ada2564dc4d0d459fa : SUCCESS in 23m 42s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/52e494fff33a4e16aea635bf924b5fd3 : SUCCESS in 31m 05s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/56a7a3943cf34ed7a5be56fc8e12b7ec : SUCCESS in 1h 05m 32s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/54df46a6421148ae80bf9857750007a9 : SUCCESS in 1h 28m 48s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b18f0ebc9c724859838738dbdbe847f5 : SUCCESS in 25m 04s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a29bcc339ecd4d488f5e3d1c658b8ba5 : SUCCESS in 41m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0c0fb6c421b7445493ea3e74dd2bfab2 : SUCCESS in 50m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b25ed9f02bf8445782839fae602ed208 : SUCCESS in 1h 20m 00s","accounts_in_message":[],"_revision_number":2},{"id":"37f1bb6d5804277802d0ca0d3a3a7e45aa171810","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 23:39:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"846f194d841d6a18ce2d7c44bbb135be41e8f2a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 23:40:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8185317267664e988de5fa4f4a8e05b8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/151635411f40441a82bd00d558630da3 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2},{"id":"d8af81203722f1d9114b7b89c7adbc3f9e4f4159","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-06-04 10:34:54.000000000","message":"Patch Set 2: Backport-Candidate+1","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7efe8a37f05dcb2656b2abf8ce9204efeaafb82b","revisions":{"7bf7e599190f3829c7a709eead2eec84b8c07227":{"kind":"REWORK","_number":1,"created":"2024-05-28 12:13:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/920628/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/920628/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/920628/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/920628/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/920628/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/920628/1"}}},"commit":{"parents":[{"commit":"adaef5768c527677e5fc62e3173c9af7699e16bf","subject":"[2024.1 only] switch rpm distros to Cloud SIG Caracal repos","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/adaef5768c527677e5fc62e3173c9af7699e16bf"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-23 13:14:53.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-28 12:13:52.000000000","tz":0},"subject":"Drop packaging requirement and unpin docker-py","message":"Drop packaging requirement and unpin docker-py\n\nChange-Id: I11d306be5e5bb3d5b05b50dd8ee64ac9424854a1\n(cherry picked from commit 968711579fbe73316227bd2b1993c5656adad9f4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bf7e599190f3829c7a709eead2eec84b8c07227"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bf7e599190f3829c7a709eead2eec84b8c07227"}]},"branch":"refs/heads/stable/2024.1"},"7efe8a37f05dcb2656b2abf8ce9204efeaafb82b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-05-28 12:20:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/28/920628/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/920628/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/920628/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/920628/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/920628/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/920628/2"}}},"commit":{"parents":[{"commit":"eb23d9cc28dfb1f6c4e7d57fc4923eef61c72865","subject":"CI: pin codespell to \u003c2.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb23d9cc28dfb1f6c4e7d57fc4923eef61c72865"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-23 13:14:53.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-28 12:20:16.000000000","tz":0},"subject":"Drop packaging requirement and unpin docker-py","message":"Drop packaging requirement and unpin docker-py\n\nChange-Id: I11d306be5e5bb3d5b05b50dd8ee64ac9424854a1\n(cherry picked from commit 968711579fbe73316227bd2b1993c5656adad9f4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7efe8a37f05dcb2656b2abf8ce9204efeaafb82b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7efe8a37f05dcb2656b2abf8ce9204efeaafb82b"}]},"branch":"refs/heads/stable/2024.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
