)]}'
{"id":"openstack%2Fkolla~920684","triplet_id":"openstack%2Fkolla~master~I70981e9f7dcbd755b5df43d860b44c3feb8f90b1","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-12-17 10:50:41.000000000","reason":"Change was abandoned"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-12-17 10:50:36.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-12-17 10:50:41.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I70981e9f7dcbd755b5df43d860b44c3feb8f90b1","subject":"Add pyproject.toml to support pip 23.1","status":"ABANDONED","created":"2024-05-28 15:37:48.000000000","updated":"2024-12-17 10:50:41.000000000","total_comment_count":13,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"4d8168c6669ec6d4e26bbe1bf416ca12ee50ac60","_number":920684,"virtual_id_number":920684,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-08-30 15:25:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":1,"date":"2024-08-30 13:59:08.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":-1,"date":"2024-08-30 13:59:08.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-28 18:32:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-28 18:40:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-16 09:10:10.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-07-16 11:22:26.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"}],"messages":[{"id":"715c905e2f94111c94cd676a8767a21f668ef9a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-28 15:37:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6d784b270bbaffc7b2d98d4707947a4dd90fed7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-28 15:41:47.000000000","message":"Topic set to kolla-ubuntu-noble","accounts_in_message":[],"_revision_number":1},{"id":"9a381c37229252b55b19aa16b64e24256b9abf4b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 18:32:29.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/010deae780034b759be554f60f560964\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b1e845d9f1dc4f0a8871d5e0033683e4 : SUCCESS in 2h 12m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/36e80a8ed6a04b9595028ab20d8914a1 : FAILURE in 27m 44s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/68a33b8dc4154d7a94ab97e93496d6e1 : FAILURE in 18m 28s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5130aa03d2f54199900224f1dbfee799 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9cb3c5f41a0142a8b5247a617559aba9 : SUCCESS in 2m 54s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e2f46aeea9444bcf94865da406c5055c : SUCCESS in 2h 51m 43s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/fc5f78235efc46a180eeaf90dd00fc46 : FAILURE in 21m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a24146083ec84d068a52880061fd35914274c3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 18:40:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f600b68bb9ae43ef83081a79468ca23f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f3af1f17167c47fc80129f3f7bb279f3 : SUCCESS in 28m 33s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/000ffe589719407ba6652f5b163b022c : SUCCESS in 1h 07m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2c15b8e9f23f48e69fd2beafa76dd65f : FAILURE in 18m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/439b4ea9b9054f3486232c63d4c2de2a : FAILURE in 32m 10s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3ce9939fb34a4787ad6959086c56bf5e : FAILURE in 15m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9d5dd527cc1d46b894c235bf32633655 : SUCCESS in 1h 39m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae7e5411d2174bc2ade5169b13ac1588 : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de34c7d2f5cd49beb69ef710a8b5822a : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8e9d8620a4d24240a875d9e081ccfa0a : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bb2d62e75e60496ca308b0631d30ba02 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0a56656a6074d8f97f513d92906038c : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77b93c98ad534024bb83c8365e3f10f6 : SUCCESS in 5m 49s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/97859e8a392f45c2a98d6bd5c1db420d : SUCCESS in 19m 04s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/34ae1e903f3f44fa8dd81a92867689a6 : SUCCESS in 26m 34s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9772baae506c464ea5af0ecba6c41888 : SUCCESS in 1h 02m 44s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0a75f3506efc40609794ba7233d80b4f : SUCCESS in 1h 32m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/785c0fc18efa44909a4751854490d0db : TIMED_OUT in 3h 02m 05s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/90ba17e678b947df80bae8f29322fa40 : FAILURE in 44m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/413dc8cc35e046ca8a1898cd9352a130 : SUCCESS in 51m 04s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/60bffcc00d8b4e3297692ab611ef5181 : FAILURE in 12m 34s","accounts_in_message":[],"_revision_number":1},{"id":"80f1b889c16f4c9514adbad6bd7c59da86c5979f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-05-29 12:08:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ff1a22d26352f68452ce113caa248b174d55cfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 13:46:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a18999efe124a46bafd8d977293554f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ec76516379bc4473a660a0feb4098a6e : SUCCESS in 23m 59s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/20b9685d7c594c589ee3c7c27613356a : SUCCESS in 1h 09m 28s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/856e6333cdb74d958d8b75a1f6edf883 : SUCCESS in 18m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fcca39480f644b158074e34b3f519aaa : SUCCESS in 29m 10s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/87ac53a33424462da2526bc3b1af35a9 : SUCCESS in 1h 04m 49s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a4fb4a65457d41e4a419187c7adc591e : SUCCESS in 1h 32m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01efa8f9d46e49a0a42abf8327f2b961 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cfa4fe363ab84b99b93d63ddfc05bc7f : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9829b983e12b48538400ceda177049c2 : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5219b67fa9a644379f2e4194863e0672 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5d75074fd6f43d3a0d368890aa7b4a9 : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6acaadfd5104892b7ff6476df89f935 : SUCCESS in 5m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3c394c6e48c5481ca73327ac760d034e : SUCCESS in 29m 17s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b26c3e328d7f4b16884b1a948905ac4b : SUCCESS in 38m 04s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a9c4927520d5464ab3f80602e6301d94 : SUCCESS in 1h 06m 09s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/591228b048ca43cf96f84cab67d65b2a : SUCCESS in 1h 20m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2f60023ac2974090b513dc04cd65b847 : SUCCESS in 24m 59s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4912a26b16ad4565a472bd2533e70a41 : SUCCESS in 53m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c2ec2304002c44ba88a935117a991981 : SUCCESS in 54m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/26a4e1a64dc8457cbcda22c6cce4ec2e : SUCCESS in 1h 26m 24s","accounts_in_message":[],"_revision_number":1},{"id":"c9d0c750a02b286e8ddaa04127baabec76bd849a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 15:04:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3747b739cb3409dbafc94ab279ab8df\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/cad740e82c884d3eb50875db8c0f8944 : SUCCESS in 2h 52m 27s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3d68a6bb9f234af98f7512dc87fd330e : SUCCESS in 30m 25s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/91f02a2a83e647daa2d050ff2285e012 : SUCCESS in 2h 11m 54s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ec8d4d066c9941f7a2f270a811bd7e9d : RETRY_LIMIT in 1m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/530048e1800b4e7c975429ede1a24081 : SUCCESS in 3m 49s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/700d980cc560443ab5b5a958f078845e : SUCCESS in 1h 13m 56s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/fc2aa3b767c84647aef32413ec3aae74 : SUCCESS in 2h 32m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1cef7208e586f5482ad5c3d0e60f8a32fd686e5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-16 06:26:32.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1436485a6970868efa86c30aa598cd409d6b69b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 08:03:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/feef70f3ae6f4b468b4be93e86db46b8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e24c4fedd503439aade554baff9eb380 : SUCCESS in 26m 09s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1126eabb109241ef964267c31fe2bcf5 : SUCCESS in 50m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5b55b49ad8484916a965815326a26576 : SUCCESS in 18m 08s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f46a5d95819c4c2182cf335afab59951 : SUCCESS in 27m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8fc04c8a4074469a9608f8049660ce01 : SUCCESS in 55m 41s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0aa7ad3a2ee3435a8b8d6c0121fe12e4 : SUCCESS in 1h 36m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53959c26aba644a2a72c4a0b9352f835 : SUCCESS in 2m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/085818be5cbe4fc6a0ef07f4848c0ce7 : SUCCESS in 2m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac10a410f0c4484dbd556d9b7679c71d : SUCCESS in 3m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/46f931be53154289bc00dfaf939bb8d0 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef46eaaa00b34ae2b6a661339df2425c : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc834292ffcf439f824fe9ef167d119d : SUCCESS in 5m 55s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b10121f305a04582a2d8be116a1f8c58 : SUCCESS in 20m 54s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/8b912f5fa4574bbdb103eb4dd06cc9f0 : SUCCESS in 34m 45s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8458e5b955604ff4a7a2f8902aea5437 : SUCCESS in 1h 11m 28s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/79da808ad27c4eed8999a36ed0d7a244 : SUCCESS in 1h 25m 31s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/94523f42da26409a8dd203eb5cca5b7c : SUCCESS in 18m 58s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/64dec48d111b479fa5c34c74a75834e2 : SUCCESS in 1h 02m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/60ff24d7c23f449e92898e72d1509820 : SUCCESS in 1h 01m 36s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f43087bc237b4368af5f2622d838ece9 : SUCCESS in 1h 13m 43s","accounts_in_message":[],"_revision_number":2},{"id":"bb6f5afbbf8f81ea8421d33966d8aed1677f871f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-16 09:10:10.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1a618588f2af5c8910d1833f32f9352ef8f7d31c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 09:36:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a08aa005c8f741f4be517c18e0307a7c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/52ec7b68b1da4ffa9a5a7218e8b37465 : SUCCESS in 21m 53s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9c2050684af24edea5af9dc450e0067f : SUCCESS in 21m 29s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f770cc344bef40f6bc427857d9e0471c : SUCCESS in 1h 05m 21s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/75636a802163414aaab55f62f86ef8bc : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/83667b3fe06e402f93a47f3e1c9fa687 : SUCCESS in 3m 46s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5825fc1d825740298ec7ba35eb68da0c : TIMED_OUT in 3h 06m 32s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5d433ee57c954048be178a69997e3b28 : SUCCESS in 18m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cdd06cd24a97078addcff7f281465685adb44105","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-24 09:33:34.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"dc46c19c8b4ebb3f4184b1b7b716174139b3ffc1","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-07-29 14:11:36.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6c658f02b43e3479dfb83acdca82346541883246","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-09 09:57:00.000000000","message":"Topic kolla-ubuntu-noble removed","accounts_in_message":[],"_revision_number":2},{"id":"51a91d24fd8c5dd7e58e6033d1befaa089b454ce","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-09 09:57:18.000000000","message":"Abandoned\n\nNot required now, have no clue why other projects are adding this.","accounts_in_message":[],"_revision_number":2},{"id":"47ab1790fa3667972bd9d4b9074bc77c6c7ad9c2","tag":"autogenerated:gerrit:restore","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 06:52:43.000000000","message":"Restored","accounts_in_message":[],"_revision_number":2},{"id":"d3147630b51bb681cebf0437798e706b0556913a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 06:52:49.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"8ecb4534111dcbef7f6daa042eddd3847a42dde6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 06:53:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"038e7730029fc12e7111283750ddb25a3b10c396","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 08:15:43.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0aa88b0f4fb5494a8537a2aeb23ac1e8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ea52295d8c8247bdab50a04989259bbf : SUCCESS in 23m 13s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/69bed8be203540d3a2369076b7ecded5 : SUCCESS in 1h 02m 24s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/77a66352469f43a1a188d320f862da99 : SUCCESS in 12m 37s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0e41c7f6ad234d85900f0a1f1e65f63f : FAILURE in 22m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c72eac1db3b84e8a8ef6bd88f2a37559 : SUCCESS in 49m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ef12b8ebf39c448e883bee14fef0f71d : SUCCESS in 1h 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd4d76df4f044e92ba792fde2ee74aad : SUCCESS in 2m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62b155c605894ff2a3434307fbd15957 : SUCCESS in 2m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1344a5b3542447e7a3844d96322766c3 : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3fb11515484546e6813c08c5e850d4ab : SUCCESS in 2m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95f1cfc23dea41c28730e43e2d7f3d25 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e23ebaf87104840828afcbf9a55d76a : SUCCESS in 3m 50s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1050f16aa8004c1184f87a7682ac3a13 : SUCCESS in 29m 17s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ae73d8f85cae494e999b4b667c51920b : SUCCESS in 35m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/100bce3ed62f4e70bd47783cf446965e : SUCCESS in 57m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b4c539d131d64db59efcf61ec02f4888 : SUCCESS in 1h 22m 28s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a4cc4f8b5b1b4c8f99a22efc0146f90a : SUCCESS in 18m 04s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7330d5a22a8a4b49809d6e26bf7ca3ec : SUCCESS in 50m 26s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c89c8c30817a430daad62c2409e71afd : SUCCESS in 57m 26s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/69b3417ad42548c1a6e9404463ead0f5 : FAILURE in 1h 13m 23s","accounts_in_message":[],"_revision_number":3},{"id":"d07e37e13b7894c199b8c3ffb46e4d444fc45aec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 08:25:52.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bc8b95fe3284c68bfe7367ef4470d60\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/39ec34f996f04cd0889a95d4a01f1e24 : SUCCESS in 1h 31m 27s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6b059eaf1d6f4dc69d8a26d661533533 : FAILURE in 40m 12s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d94a5db3d0c44b8b8c8d47b6bc2249dc : SUCCESS in 2m 59s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ef8bb5eebefb43fe8ad255c7f8f9ba81 : SUCCESS in 3m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e62348529ff1090c8352154a0bd0c5a8802441fa","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-30 13:57:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b0f6154a39e2de8e492b7c591270527b6f12c6df","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-30 13:59:08.000000000","message":"Patch Set 3: Backport-Candidate-1 Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"715c7fd459f3df3b6b53098bb21f90bcdbf7590d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 15:25:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bd1848172e247c3975c3e7466e011d3\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/53c963518f964b5f815f62ecb227adc2 : SUCCESS in 22m 34s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/95aebc67f73648de820d7b9a32780e76 : SUCCESS in 1h 04m 11s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/15eadd0aa2424db48b1eca46ecb77750 : SUCCESS in 17m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f946996346d84bad840abc662cc1beba : SUCCESS in 33m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5d4ad73800f44a2f8ede4cf7fb3b0693 : SUCCESS in 1h 00m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3f40517074394dc4b0219f74460dacd3 : SUCCESS in 1h 20m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87c653449bde4552a0f0bb8609874651 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b5cad4cd32e40608e17d17f7f82d375 : SUCCESS in 2m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c685e2f87b34c049f23a4ae72b95164 : SUCCESS in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bbddced087c7490d8a2578b6ba80b05f : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c57bb4e921d74346b3b0810a4cc47549 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12a57c81e19f4755bcded89aa23d7e53 : SUCCESS in 4m 36s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cfcdd11d62f44c4e9be817ed6379ec8c : SUCCESS in 29m 35s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/df42930556b341b1935806b296974910 : SUCCESS in 37m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/db9d0e9cf1cf4d4e97196be8afb16fa5 : SUCCESS in 1h 06m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a191560f96d34bbd804160ff3877ef13 : SUCCESS in 1h 23m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0f9540b1d5e24a7e97f211b93378ed8a : SUCCESS in 19m 15s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c5fd8290bf1045f5b8b9ff4b4b9ce0a9 : SUCCESS in 51m 53s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8981fb863b2f43089561efe3e4043057 : SUCCESS in 54m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/dd232e8ca4a1404a8b7ce26443791160 : SUCCESS in 1h 14m 40s","accounts_in_message":[],"_revision_number":3},{"id":"c29c32b80d93f43d5cd1b8c6a50b10be73696ba1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 17:02:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a32ea916f7284deab5f81e4b25e9e19c\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/97c15c3658ee484389f48d7f32be2b61 : TIMED_OUT in 3h 03m 39s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ee29ec3b8ea94a67b8866133bda8bb80 : SUCCESS in 1h 55m 08s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c459fcb5c113417092e09c31cb2c87b6 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b8e16bb2027a46acb6cfcf74b53bf0dd : SUCCESS in 4m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"61704f7a2711b7d3b4797c0fa450a31fd5979dfe","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-17 10:36:42.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"999da3a83a54af31d1c68fa65a21cb63eeb5a865","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-17 10:50:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4d8168c6669ec6d4e26bbe1bf416ca12ee50ac60","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-17 10:50:41.000000000","message":"Abandoned\n\nnot needed now","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"79c615a3d14d8a9a625c1b9775f3ccb0f005b58b","revisions":{"fa0a2958a5612cc87632566fc54a2b88b8be9499":{"kind":"REWORK","_number":1,"created":"2024-05-28 15:37:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/84/920684/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/920684/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/920684/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/920684/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/920684/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/920684/1"}}},"commit":{"parents":[{"commit":"968711579fbe73316227bd2b1993c5656adad9f4","subject":"Drop packaging requirement and unpin docker-py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/968711579fbe73316227bd2b1993c5656adad9f4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-28 15:35:19.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-28 15:37:46.000000000","tz":120},"subject":"Add pyproject.toml to support pip 23.1","message":"Add pyproject.toml to support pip 23.1\n\npip 23.1 removed the \"setup.py install\" fallback for projects\nthat do not have pyproject.toml and now uses a pyproject.toml\nwhich is vendored in pip.\nTo address that, this change adds the minimal pyproject.toml\nto enable pbr to be properly used to build editable wheels.\n\nThis is required to support installing devstack on\ncentos stream 9 and related distros with GLOBAL_VENV\u003dTrue\nWithout this change the wsgi scripts are not generated in\neditable mode. i.e. pip install -e /opt/stack/keystone\n\nSee https://pip.pypa.io/en/stable/news/#v23-1\nand https://github.com/pypa/pip/issues/8368 for more\ndetails on the removal of the fallback support.\n\nsetuptools v64.0.0 is used to support editable installs\nvia its PEP-660 implmentation\nhttps://github.com/pypa/setuptools/pull/3488\n\nChange-Id: I70981e9f7dcbd755b5df43d860b44c3feb8f90b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa0a2958a5612cc87632566fc54a2b88b8be9499"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa0a2958a5612cc87632566fc54a2b88b8be9499"}]},"branch":"refs/heads/master"},"434578f7e3167c119650657f7ee15e2b9355f4a3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-16 06:26:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/84/920684/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/920684/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/920684/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/920684/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/920684/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/920684/2"}}},"commit":{"parents":[{"commit":"139851087e15e0aa0c47765768bcdc1da80246d6","subject":"Merge \"Pass rabbitmq apt preferences into kolla-toolbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/139851087e15e0aa0c47765768bcdc1da80246d6"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-28 15:35:19.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 06:26:32.000000000","tz":0},"subject":"Add pyproject.toml to support pip 23.1","message":"Add pyproject.toml to support pip 23.1\n\npip 23.1 removed the \"setup.py install\" fallback for projects\nthat do not have pyproject.toml and now uses a pyproject.toml\nwhich is vendored in pip.\nTo address that, this change adds the minimal pyproject.toml\nto enable pbr to be properly used to build editable wheels.\n\nThis is required to support installing devstack on\ncentos stream 9 and related distros with GLOBAL_VENV\u003dTrue\nWithout this change the wsgi scripts are not generated in\neditable mode. i.e. pip install -e /opt/stack/keystone\n\nSee https://pip.pypa.io/en/stable/news/#v23-1\nand https://github.com/pypa/pip/issues/8368 for more\ndetails on the removal of the fallback support.\n\nsetuptools v64.0.0 is used to support editable installs\nvia its PEP-660 implmentation\nhttps://github.com/pypa/setuptools/pull/3488\n\nChange-Id: I70981e9f7dcbd755b5df43d860b44c3feb8f90b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/434578f7e3167c119650657f7ee15e2b9355f4a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/434578f7e3167c119650657f7ee15e2b9355f4a3"}]},"branch":"refs/heads/master","description":"Rebase"},"79c615a3d14d8a9a625c1b9775f3ccb0f005b58b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-08-28 06:52:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/84/920684/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/920684/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/920684/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/920684/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/920684/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/920684/3"}}},"commit":{"parents":[{"commit":"ece67cd2aec82d661ba355ce6bc6e41bb88001b6","subject":"Check if archive is tarfile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ece67cd2aec82d661ba355ce6bc6e41bb88001b6"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-05-28 15:35:19.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 06:52:49.000000000","tz":0},"subject":"Add pyproject.toml to support pip 23.1","message":"Add pyproject.toml to support pip 23.1\n\npip 23.1 removed the \"setup.py install\" fallback for projects\nthat do not have pyproject.toml and now uses a pyproject.toml\nwhich is vendored in pip.\nTo address that, this change adds the minimal pyproject.toml\nto enable pbr to be properly used to build editable wheels.\n\nThis is required to support installing devstack on\ncentos stream 9 and related distros with GLOBAL_VENV\u003dTrue\nWithout this change the wsgi scripts are not generated in\neditable mode. i.e. pip install -e /opt/stack/keystone\n\nSee https://pip.pypa.io/en/stable/news/#v23-1\nand https://github.com/pypa/pip/issues/8368 for more\ndetails on the removal of the fallback support.\n\nsetuptools v64.0.0 is used to support editable installs\nvia its PEP-660 implmentation\nhttps://github.com/pypa/setuptools/pull/3488\n\nChange-Id: I70981e9f7dcbd755b5df43d860b44c3feb8f90b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/79c615a3d14d8a9a625c1b9775f3ccb0f005b58b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/79c615a3d14d8a9a625c1b9775f3ccb0f005b58b"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
