)]}'
{"id":"openstack%2Fkolla~921032","triplet_id":"openstack%2Fkolla~master~I1dda73b0bc9a9f57ec8fda3f2234449b04906410","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-06-11 14:29:44.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-06-11 14:29:44.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-06-11 11:52:48.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-06-11 14:29:44.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-06-11 14:29:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1dda73b0bc9a9f57ec8fda3f2234449b04906410","subject":"CI: Pin requests to \u003c2.32 for publish jobs","status":"MERGED","created":"2024-06-03 13:44:00.000000000","updated":"2024-12-11 07:41:28.000000000","submitted":"2024-06-11 14:29:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921032","meta_rev_id":"5398eaa658f445534bb719eec6e6d44ecc6dba3a","_number":921032,"virtual_id_number":921032,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-11 14:29:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-06-11 11:52:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-06-11 14:48:09.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-11 11:07:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-11 11:07:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-06-11 11:07:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-11 11:07:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-06-11 11:07:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-11 11:07:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-03 15:39:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-04 15:06:19.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2024-06-05 13:05:04.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-06-05 13:05:43.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-06-10 08:36:58.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-06-11 09:56:39.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-06-11 10:11:20.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"572d7cf09fe089f11e99970a33a7d9e7b28ac424","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-06-03 13:44:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34c74704e2d1c01f2c2565971b6522ab0666ae13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-06-03 13:46:18.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7e7e44e497d680ff404b5245465a7259aec26aff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 15:39:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff516a727a3d449793440ace6db236e6\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1ff93c75124b4c5a94c377f8cc0b6a57 : SUCCESS in 29m 44s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a7b01e3ffb044b0abcfd0ffd3e1de074 : FAILURE in 8m 58s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0bff83a0520847968ffc866553665ba6 : SUCCESS in 24m 18s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d73f813cdc66429dbbb6d33ab2ad1659 : SUCCESS in 30m 55s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a98385ce0a744b2b856f0ce436eaacaf : SUCCESS in 1h 03m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2cea2dd4ea8c41e2b520e6ae57311e32 : SUCCESS in 1h 29m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82895a15fa7342eca6ead99d65008a97 : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eae7264bc6bb4c96b85db6ceecd2d912 : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/650d61b3bb554bf58694a2c02440634e : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/76c5f44504ff4afc9d48860a13a3b9e5 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f177f7a5a68d4b2e8a2010549437ee2c : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/032e4d2043c54520b70b999dfc522d83 : SUCCESS in 4m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d5f419c0bf7b49e09d05c8a792f73280 : SUCCESS in 29m 00s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/f2841f8fd26840249ba4297bbb384574 : SUCCESS in 40m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/92303fa5dbab4a81b0c578d538074642 : SUCCESS in 1h 10m 38s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cebda94871314ef295fdf13d6f9f60d2 : SUCCESS in 1h 33m 53s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/996f3bef39c04818bd4c4ab737850d26 : SUCCESS in 17m 46s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5b58a83a78aa478c947b24074070f879 : SUCCESS in 42m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f7b5dfb281b74525a3797b4db829e27b : SUCCESS in 50m 11s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c09413d02a6b4a5e87069fad418ece8a : SUCCESS in 1h 09m 50s","accounts_in_message":[],"_revision_number":2},{"id":"0b61bf5bdffcd9af262e843baf977f7fa7a60860","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 16:26:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/621b15126e4b4f9d824293e42e3c4e0d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b149a174abce477b8d975260267f0a7a : SUCCESS in 2h 31m 21s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/55b6c722862a4194bc71d0f371daf655 : SUCCESS in 31m 46s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/01580c5066114581b7cb8dbed70cf0ca : SUCCESS in 2h 07m 50s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f737b11d461a4b3e83240d7cd4e63c2b : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4a69bba453a1435abd50e17ac9c93683 : SUCCESS in 5m 10s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f904d41bfaa34b26871383414da3a54d : SUCCESS in 1h 01m 08s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/844497834ea54ffab952841fce59ce2a : SUCCESS in 2h 38m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f5350e30c1261c121acea0fbca05efbdaefdbaa5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-06-04 13:14:55.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"06e772ef9d64f85b340554607fe7617104367a2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 14:53:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d8f56aaedf94bb8a447abe72b362de6\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7d7b25664dfd453289312b70363cf3b9 : SUCCESS in 40m 02s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4d136756694145179fd372b17da9ec6f : SUCCESS in 1h 19m 27s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/969fbc0d4f58487e824d4dab3da4e379 : SUCCESS in 20m 28s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2542191652e54fb6be06b3bc98b245b9 : SUCCESS in 30m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6d9bfc2bdbaf4fecb50679aa8d679eb2 : SUCCESS in 59m 27s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/60896fd58f2a427183aa8db1c0c704e2 : SUCCESS in 1h 32m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fbb92bcff464c85aca27e00bd02eca6 : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30ae12b7793e4114aebfe63d4aa53bca : SUCCESS in 2m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c4d96335d5914d26a3708b1965667c98 : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3ce6eb5ff7f141419a53c8c2eaacc858 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/816e527b2c96408b9ca1149a53c85f91 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6879d3720140454daccddf1aa4108498 : SUCCESS in 4m 22s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ca94016b91a44ed9aa6d59406c53d543 : SUCCESS in 22m 41s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/dc4a6bb9a53e491492444c9b50cda832 : SUCCESS in 24m 04s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1c3351c2b7014837804affc6aa92d799 : SUCCESS in 1h 04m 45s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ed9b2ee1974c4dd18823d30bb331c286 : SUCCESS in 1h 22m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e7cfa23f8b2e4e47a3314aa325d9f6ce : SUCCESS in 18m 10s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/464131dd858246a5a91a641a05690884 : SUCCESS in 1h 05m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/cfddfa18cbd0497e8160b710640a656b : SUCCESS in 48m 10s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ef2dbc763d624a748df6529e9ce42576 : SUCCESS in 1h 23m 11s","accounts_in_message":[],"_revision_number":3},{"id":"e8fb9dc6e3f0695752a48a2c50d9071760f6fb8e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 15:04:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/755f4cb6bec34a17ac93090806129961\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/90472f8d0db644d7ba7207d57dcdfcac : SUCCESS in 20m 31s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/698a76a7d78d4d788c1f8471e172fd14 : SUCCESS in 20m 33s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7790a5034afc40a9a1f7c3158f19852a : SUCCESS in 1h 01m 05s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b7ab2b9fb6124f9cbe238c54d7e13ac1 : SUCCESS in 2m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cc6f7c1c4d8a4c8382d4cd4d39142693 : SUCCESS in 3m 57s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/747784bd68064b658b81f0862797133d : SUCCESS in 47m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9f91cee2a25e4f0db5c99bf3f0624f5c : SUCCESS in 1h 47m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"81eedf77d418128ab1e25ba091bd68d38ea3bbed","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-04 15:06:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1e2c2374cbaea8ea7668260554625a49dfb0f973","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-06-10 07:16:34.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"483ff78fbe7f0fc999ef8e813de69b6072fffcb3","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-06-10 08:36:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"28eaec234592b7fdf7885ab2927f5455c1a7c6b6","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-06-10 09:00:30.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"e5cd4bcd2ba22cd1deb3bd57280c23d0ad301cae","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-06-10 14:21:18.000000000","message":"Patch Set 3: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fcea35cedf6beaf76431e823c26591f2f4eac702","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-06-11 09:51:25.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"3621850078ce264bc2828b84572adc2da64f8bf2","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-06-11 09:51:52.000000000","message":"Patch Set 4:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":4},{"id":"671bd137ae5d262dd16b345feb6dd1d34a5ebf52","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-06-11 09:56:29.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"66bcf1a3f040d5c4058641776acf60ca142b7f52","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-11 09:56:39.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b8155db2721a3c0b9c55c827070d6c05dcec5f69","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-11 09:56:43.000000000","message":"Patch Set 4: Review-Priority+2","accounts_in_message":[],"_revision_number":4},{"id":"785b745b7ffd6c92cff828f85d367cec7581833f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-11 09:56:49.000000000","message":"Patch Set 4: Backport-Candidate+2","accounts_in_message":[],"_revision_number":4},{"id":"cb467d07002ffd738131490cd3b0221ce387a533","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-06-11 09:59:03.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"06087d34a0b622f50fdca2bb8ab52cf5453fa370","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-06-11 10:11:20.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8f04e3d877c7dfc79c2b0f0a0b17a30bc6807c53","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-11 10:19:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4a7d5845145418dc308f7d2b0c89484bccfe1df0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-06-11 10:31:52.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"2a0e1cc0f09717720af422fd93fc801238cdbd19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-06-11 10:57:43.000000000","message":"Uploaded patch set 5.\n\nCopied Votes:\n* Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":5},{"id":"c167465f55b435a4401873e6a92ee86637d1e3f9","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-06-11 10:58:41.000000000","message":"Patch Set 5: Backport-Candidate+2 Code-Review+2 Review-Priority+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1fdf87ab02febcc50a14e58f0d3e015fe8d9d5b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-06-11 11:07:42.000000000","message":"Patch Set 6: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n* Review-Priority+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":6},{"id":"fb876374d4a3f459dc8df0c76c0a981d25d033dc","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-06-11 11:07:56.000000000","message":"Patch Set 6: Backport-Candidate+2 Code-Review+2 Review-Priority+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"744ce65d21e1b75ab450f2c420ad0aefc1cd9b8e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-06-11 11:52:48.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"6aaec86d99fdf36a79f4337b952b0c98c6ce5c40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 12:50:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/564872b93f3f4596997818390db12e55\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0ad4c810d68a4f3eb459f6fa4693593d : SUCCESS in 22m 57s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a7726b65954f41e996ae5f714a851a76 : SUCCESS in 1h 07m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d12a42cabc8444909f90f089f80caf89 : SUCCESS in 19m 22s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6151053e6554438a86de2d516a364de0 : SUCCESS in 25m 43s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/84028a2cae0948aa967241a917e40f31 : SUCCESS in 47m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/913facfb50034a768c8b578785fbb2d5 : SUCCESS in 1h 17m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03eca081f3ca433ab2b2c87156bdc2bb : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/733e9ea8e14145bfa0233aa2af2e7956 : SUCCESS in 3m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c60c2795d5fa47b490dc5003045d0d25 : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3602d08f48934132b92b86d1e355b158 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acac1409d7894c80bb4517c72d4b78aa : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e20dd410b2d4c45a8766a964cbc7d52 : SUCCESS in 4m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c5c4aa99dad24f51a8e7a2bcad08e679 : SUCCESS in 28m 39s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e40b8da42c3d40b4a1842172b584c4b3 : SUCCESS in 38m 09s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5a23d103f4474e768b77361500be9bb9 : SUCCESS in 1h 08m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7f27d71b42bc45b38a1548d4add4bd45 : SUCCESS in 1h 32m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6dd24399e95a4d3781ef53c67be5130b : SUCCESS in 22m 40s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/16bc082f65854dedb84ef2cfa241d4cf : SUCCESS in 1h 01m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2bb1322c1f36466398df611975c8e374 : SUCCESS in 53m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6b720d82bd1643da872bac19bb771833 : SUCCESS in 1h 19m 58s","accounts_in_message":[],"_revision_number":6},{"id":"1caa7c9afd37e4cc23b0999efc9761b103eef2a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 12:51:14.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"5201782eb5d70d156e967af6ac35d73ea95562b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 14:18:23.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47030deada364868a875a4aeb97cb065\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f1717668453f4cb98576da14c990a9f4 : SUCCESS in 22m 27s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7eae5d9e43f1406f82ad2b034c76f433 : SUCCESS in 19m 57s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3126b2b3c44d4602947894cb14d42c78 : SUCCESS in 1h 10m 03s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9f5f637e58cf4f378e4b73e83ce1a7a4 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d422fcd7db044b0d87d0828e1a45665c : SUCCESS in 2m 27s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/503196eb3c30469882bdf814bfdd4dea : TIMED_OUT in 3h 08m 19s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7b79c8797f25491f9ca400092d42c27d : SUCCESS in 20m 40s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a82e3f705f927ec96b7789dda5d819b0ca489058","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 14:29:43.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a8e930894174bc8b5a7b77a2fa21128\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0b8b5bc70c5449f894baf23c69a69ac0 : SUCCESS in 20m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0d9c634932d04a87adee69ff4b903aa5 : SUCCESS in 27m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0004bfb0813341b5bc05210aecfd2985 : SUCCESS in 1h 01m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/481beb2f561147559116442a6ead686b : SUCCESS in 1h 30m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/644aa8c4ea6f435bbdff6ebc5616aacf : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cab11c95ee5a4bc9a9a5f0612eddd5f7 : SUCCESS in 2m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c6a2beee93434509bd4d4004fe799590 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8feb2a5ba0e45d4b2a1171c8c333d44 : SUCCESS in 4m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/563432b92ab84873a55c21ba8786296c : SUCCESS in 27m 43s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3d2f7a40e1724d7d9f5fabaacb9c0076 : SUCCESS in 36m 11s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d3c8c392b804440e881b06e0403b0647 : SUCCESS in 1h 04m 06s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/de89a1d6af854899ac5e03d5916721be : SUCCESS in 1h 14m 28s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a5233c91dc20405ab5cfc4be2e829b72 : SUCCESS in 17m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/68cf696cf7a24793801d694d59e96e16 : SUCCESS in 1h 09m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/71bd17f5b31a46fd9be39d584c2ba94b : SUCCESS in 48m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c140066471144e90b48a13b23fd29fa2 : SUCCESS in 1h 03m 45s","accounts_in_message":[],"_revision_number":6},{"id":"3baf75015db385db12e77dd1dedaf77851172842","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 14:29:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"578db1efe80e77a534d29fe8e3eb5e4c9a8c9781","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 14:30:51.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcb89e0c40ad4a2db7b71b33f39c719d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bdb19cb54cf4ebbad88c039942c1852 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":6},{"id":"7c7b1a5e6bedc349eab8f6965a12759a4a12fdbe","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-06-11 14:48:09.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"5398eaa658f445534bb719eec6e6d44ecc6dba3a","tag":"autogenerated:gerrit:revert","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-11 07:41:28.000000000","message":"Created a revert of this change as Ife393e7f1b00fa37921e7822668d6a40dd467d5b","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"428f37124986019bedc897469f9b1f3470b62082","revisions":{"6ce4aaec97d5c2f9f734de2a5dd0f844ecdec05e":{"kind":"REWORK","_number":1,"created":"2024-06-03 13:44:00.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/32/921032/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/32/921032/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/921032/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/921032/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/921032/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/32/921032/1"}}},"commit":{"parents":[{"commit":"968711579fbe73316227bd2b1993c5656adad9f4","subject":"Drop packaging requirement and unpin docker-py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/968711579fbe73316227bd2b1993c5656adad9f4"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-06-03 13:42:45.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-06-03 13:43:59.000000000","tz":120},"subject":"CI: Pin requests to \u003c2.32 for docker sdk","message":"CI: Pin requests to \u003c2.32 for docker sdk\n\nChange-Id: I1dda73b0bc9a9f57ec8fda3f2234449b04906410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ce4aaec97d5c2f9f734de2a5dd0f844ecdec05e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ce4aaec97d5c2f9f734de2a5dd0f844ecdec05e"}]},"branch":"refs/heads/master"},"db63a734ccf7139f23e78be58b2b0044187edb0d":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-06-03 13:46:18.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/32/921032/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/32/921032/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/921032/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/921032/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/921032/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/32/921032/2"}}},"commit":{"parents":[{"commit":"968711579fbe73316227bd2b1993c5656adad9f4","subject":"Drop packaging requirement and unpin docker-py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/968711579fbe73316227bd2b1993c5656adad9f4"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-06-03 13:42:45.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-06-03 13:46:10.000000000","tz":120},"subject":"CI: Pin requests to \u003c2.32 for docker sdk","message":"CI: Pin requests to \u003c2.32 for docker sdk\n\nRelated-Bug: #2066364\nChange-Id: I1dda73b0bc9a9f57ec8fda3f2234449b04906410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db63a734ccf7139f23e78be58b2b0044187edb0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db63a734ccf7139f23e78be58b2b0044187edb0d"}]},"branch":"refs/heads/master"},"8580c502c6e83035bf6b40eba87ecc0f34ec7728":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-06-04 13:14:55.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/32/921032/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/32/921032/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/921032/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/921032/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/921032/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/32/921032/3"}}},"commit":{"parents":[{"commit":"968711579fbe73316227bd2b1993c5656adad9f4","subject":"Drop packaging requirement and unpin docker-py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/968711579fbe73316227bd2b1993c5656adad9f4"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-06-03 13:42:45.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-06-04 13:14:53.000000000","tz":120},"subject":"CI: Pin requests to \u003c2.32 for docker sdk","message":"CI: Pin requests to \u003c2.32 for docker sdk\n\nPeriodic jobs are failing because publish.yml [1] is using older\ncommunity.docker collection which does not support latest requests\nlibrary yet [2].\n\n[1] https://opendev.org/openstack/kolla/src/commit/968711579fbe73316227bd2b1993c5656adad9f4/tests/playbooks/publish.yml#L8\n[2] https://github.com/ansible-collections/community.docker/issues/868\n\nRelated-Bug: #2066364\nChange-Id: I1dda73b0bc9a9f57ec8fda3f2234449b04906410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8580c502c6e83035bf6b40eba87ecc0f34ec7728"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8580c502c6e83035bf6b40eba87ecc0f34ec7728"}]},"branch":"refs/heads/master"},"5b6988acd8963641e60ed45df3d98c2bfb06b158":{"kind":"REWORK","_number":4,"created":"2024-06-11 09:51:25.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/32/921032/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/32/921032/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/921032/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/921032/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/921032/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/32/921032/4"}}},"commit":{"parents":[{"commit":"968711579fbe73316227bd2b1993c5656adad9f4","subject":"Drop packaging requirement and unpin docker-py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/968711579fbe73316227bd2b1993c5656adad9f4"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-06-03 13:42:45.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-06-11 09:41:15.000000000","tz":120},"subject":"CI: Pin requests to \u003c2.32 for publish jobs","message":"CI: Pin requests to \u003c2.32 for publish jobs\n\nPeriodic jobs are failing because Zuul Ansible executor runs with\nAnsible 8 which uses community.docker collection that needs reqests\u003c2.32.\nRevert this patch when Zuul supports Ansible 10, and/or Ansible 9 gets\nfixed community.docker collection. [1]\n\n[1] https://github.com/ansible-collections/community.docker/issues/868\n\nRelated-Bug: #2066364\nChange-Id: I1dda73b0bc9a9f57ec8fda3f2234449b04906410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5b6988acd8963641e60ed45df3d98c2bfb06b158"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5b6988acd8963641e60ed45df3d98c2bfb06b158"}]},"branch":"refs/heads/master"},"4b00196410a6385cb3f17973f87c683b99ae4593":{"kind":"REWORK","_number":5,"created":"2024-06-11 10:57:43.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/32/921032/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/32/921032/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/921032/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/921032/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/921032/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/32/921032/5"}}},"commit":{"parents":[{"commit":"968711579fbe73316227bd2b1993c5656adad9f4","subject":"Drop packaging requirement and unpin docker-py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/968711579fbe73316227bd2b1993c5656adad9f4"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-06-03 13:42:45.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-06-11 10:57:34.000000000","tz":120},"subject":"CI: Pin requests to \u003c2.32 for publish jobs","message":"CI: Pin requests to \u003c2.32 for publish jobs\n\nPeriodic jobs are failing because Zuul Ansible executor runs with\nAnsible 8 which uses community.docker collection that needs reqests\u003c2.32.\nRevert this patch when Zuul supports Ansible 10, and/or Ansible 9 gets\nfixed community.docker collection. [1]\n\n[1] https://github.com/ansible-collections/community.docker/issues/868\n\nRelated-Bug: #2066364\nChange-Id: I1dda73b0bc9a9f57ec8fda3f2234449b04906410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b00196410a6385cb3f17973f87c683b99ae4593"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b00196410a6385cb3f17973f87c683b99ae4593"}]},"branch":"refs/heads/master"},"428f37124986019bedc897469f9b1f3470b62082":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2024-06-11 11:07:42.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/32/921032/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/32/921032/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/921032/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/921032/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/32/921032/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/32/921032/6"}}},"commit":{"parents":[{"commit":"968711579fbe73316227bd2b1993c5656adad9f4","subject":"Drop packaging requirement and unpin docker-py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/968711579fbe73316227bd2b1993c5656adad9f4"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-06-03 13:42:45.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-06-11 11:07:42.000000000","tz":0},"subject":"CI: Pin requests to \u003c2.32 for publish jobs","message":"CI: Pin requests to \u003c2.32 for publish jobs\n\nPeriodic jobs are failing because Zuul Ansible executor runs with\nAnsible 8 which uses community.docker collection that needs requests\u003c2.32.\nRevert this patch when Zuul supports Ansible 10, and/or Ansible 9 gets\nfixed community.docker collection. [1]\n\n[1] https://github.com/ansible-collections/community.docker/issues/868\n\nRelated-Bug: #2066364\nChange-Id: I1dda73b0bc9a9f57ec8fda3f2234449b04906410\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428f37124986019bedc897469f9b1f3470b62082"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428f37124986019bedc897469f9b1f3470b62082"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
