)]}'
{"id":"openstack%2Fkolla~921658","triplet_id":"openstack%2Fkolla~master~Ida297f5c6473a1b4f2e58b691a3dec6a5c3f078d","project":"openstack/kolla","branch":"master","topic":"bug/2068884","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-09-16 10:13:13.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-09-09 17:02:36.000000000","reason":"Change was abandoned"},"32398":{"account":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"last_update":"2024-10-01 08:52:13.000000000","reason":"\u003cGERRIT_ACCOUNT_32398\u003e replied on the change","reason_account":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"}}},"hashtags":[],"change_id":"Ida297f5c6473a1b4f2e58b691a3dec6a5c3f078d","subject":"Add new hardware prometheus exporter:","status":"ABANDONED","created":"2024-06-10 11:14:06.000000000","updated":"2025-09-09 17:02:36.000000000","total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9aa9c2c343b5f25a11876e6ee58932dc34a1c2f9","_number":921658,"virtual_id_number":921658,"owner":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-01 10:14:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-10 12:45:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-11 08:48:38.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2024-06-13 09:32:37.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-06-13 20:15:29.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2024-07-12 11:57:53.000000000","updated_by":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"reviewer":{"_account_id":34213,"name":"Grzegorz Bialas","display_name":"Grzegorz Bialas","email":"grzegorz@stackhpc.com","username":"gbialas"},"state":"CC"},{"updated":"2024-07-16 09:40:43.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"f05ee9d93759e90e676a22ac4632ace074036969","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-06-10 11:14:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62672270e5fd57e8fa695494f6599f28a3edeca5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 12:45:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c2c97c80e52d4fb0a64876b0bda2950f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/79458feac3c34f8fa3b1898670102754 : FAILURE in 25m 28s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/7254cfa9fac54e6e8ea2049216cd4707 : SUCCESS in 1h 02m 32s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a3bc7daa70c2439a9b9bdbc98efe9619 : FAILURE in 18m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/db5ba22b730349d7b2d052cd88c867e5 : FAILURE in 29m 41s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/142103e62a634ad98d50419a37050e59 : SUCCESS in 1h 01m 33s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3b47a201b04c45ec8873166616fcc7a2 : SUCCESS in 1h 17m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8d321c2b9734ebda741b0497b1d3e78 : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9bf350b1f63a43199dcecb70f523f9ef : SUCCESS in 2m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a3974dcc749b4f67815b6413df2f25f2 : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5ba3a3cbbd3f43a68714a0aa96e1d6a0 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/861b7964e9ec475899e6f8af0955eb44 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc0a300068014849a97190f589b62199 : SUCCESS in 3m 52s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e2644175a3ae4ab8a897bbd63a35408f : FAILURE in 25m 58s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6bb2e11368db4161a45c9ec766eb0c20 : FAILURE in 30m 49s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9f606f8bdb74406181b5051393be0e28 : SUCCESS in 1h 11m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/31b2e90dc910466e9e6ea6237fd00217 : SUCCESS in 1h 31m 13s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/45287679dabf4b78a397a9b51eabffff : FAILURE in 20m 53s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2e62d94ff394417ba4b7e2f2d2484165 : FAILURE in 15m 02s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a116252e9fa64d2fb4efe750b30f8b8b : FAILURE in 52m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/73834bdc9ea445e089320b9e1abcc6f5 : SUCCESS in 53m 10s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d8ea6adedc304dc68ec572bc79b03d7d : SUCCESS in 1h 18m 13s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/683933bf1fd544acb4711ddfe344f803 : FAILURE in 17m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2c8d2423e581c6b3a5f897175d000df87b146a36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 14:23:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8ce4b9ed20340838c9a4fc2083e31b7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7fa3aec05b68442fb94c84382281e2f3 : SUCCESS in 21m 36s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/14cf9cddbd7847bdbaee07986afec423 : TIMED_OUT in 3h 03m 40s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3ec77711185041118a90fa466a77dd39 : SUCCESS in 1h 11m 22s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e50a8ba426584070879f49b5497b0beb : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0b187b5712b44ef2b0b0b795d8597855 : SUCCESS in 3m 28s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b12519ef418a490db40714a54af75e15 : TIMED_OUT in 3h 06m 26s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a912206c07464dacb47cbe5f031b07f4 : SUCCESS in 21m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"56789c397c21508ec9c629bcc3b3ccdcfe681ba1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-11 08:48:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cffc8f8978d25dea13afc82eab712d36a34c6371","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-06-13 07:43:10.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"df9db3f85f41b6cb3242645da835822b5108b87c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-06-13 09:20:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"be7b6654b0a1b033b01bfc8271f8432a7c6fde7b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-13 09:32:37.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"9800b917f616fa7870fdabe0329323b969c927a0","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-06-13 09:53:58.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"494ce5b29876e07e932becd0845b8d91db799b1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-06-13 09:55:24.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"41909c9b991aac02762369bf0bc22887cf79c4ea","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-06-13 09:55:28.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":3},{"id":"bfbd8ee4f3a75775cc039aedd7f577171b0cac4d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-06-13 11:54:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2564054a734814332bd83df8cec11ba0682d83d3","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-06-13 11:56:49.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"cab232f2f2f9eada5a0ef0ba36939470fdbda896","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-06-13 11:58:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6e76dbce2c5dc75a53f0c9f1ac4dc36010787991","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 13:34:53.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/095362e6a37c437c9628e77e4fefa206\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/753465d2d5e343ce99c3f9ce82b352a3 : FAILURE in 28m 41s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ded1cf4b82f644148e5723360bd04216 : SUCCESS in 1h 02m 33s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/be40890f062948f396eb697ff9837d53 : FAILURE in 21m 27s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/165bf3672d4449a99087b90889bc2002 : FAILURE in 27m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b7d83181ad5f4dd89765c2e48ef5f35e : SUCCESS in 1h 05m 32s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d802d3cd786343459fb48a2419702d79 : SUCCESS in 1h 16m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a4f6794fa4c4a68be4d3c2a8a015466 : SUCCESS in 2m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea26535ea6af4cdc964d4a27dae5a7cf : SUCCESS in 2m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3132803e0645425ab7e68eac1fe25806 : SUCCESS in 3m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2e132bc69b9f4849a9c973f7f2e07560 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0581cd956982433c8099b19579440415 : SUCCESS in 2m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/738b9aafde4244d693959067de5e1ae1 : SUCCESS in 3m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32a86255bada4876a0eb7b77ef02d0b1 : SUCCESS in 3m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/be90049570bb4d2da344d495d46f00ce : FAILURE in 31m 02s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0e6475796aba433ca4dec5faf4b8bf1b : FAILURE in 37m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5de788963780407080abb6937a68284d : SUCCESS in 1h 16m 10s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c664324950564dbca1696d92b41ffe68 : SUCCESS in 1h 39m 42s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/62693e5834fb4a6d8bb7cda483ccc213 : FAILURE in 23m 50s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a4124ddffbd84a639391a5d4b40c48d2 : FAILURE in 19m 11s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ba544bc2aa9b406eb44373ae6a4fd9d0 : FAILURE in 1h 01m 27s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8a68cb4ae53044e89af9d15b4eeb618a : SUCCESS in 46m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0df908244e8e4bd2b65d33f0518ff735 : SUCCESS in 1h 17m 05s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/dc30801f35c24c18be69516d7accd923 : FAILURE in 19m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"433b1f9b79faf91d9de364695e91027cc118b77c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 14:45:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1134024c518c43e8b188d6ddf1109722\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/85770e0604294c609ce6e1041dcbcf0d : FAILURE in 35m 48s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/033ac5845efe4e539d2a49cf7b6dd7e4 : FAILURE in 42m 19s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/64b7352feb8a402c81a1a48788b80d7b : SUCCESS in 2h 02m 13s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b44ab59a8beb4cc9bb886570286b2e5d : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eb9f446ddfaf4862b3af8980328512d7 : SUCCESS in 5m 13s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/62c78b06a2c24ab6929f51ac57327acb : FAILURE in 1h 04m 05s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ea89826db8b34708ad265c91368f365b : FAILURE in 2h 47m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f2df77f701ff72770e900b84aae7a00c646b5100","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-06-13 20:01:43.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"be5fc8dbdb8a7ac4d258f3edbd2b1c9a88049fe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-06-13 20:03:04.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"c1564cbecf55c28c6e8f024cf7955e1c2fe21941","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 21:59:02.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c380149e9c74851882e7eca74a7eb06\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7c5bacd8df654daea15e9ec4a330c377 : SUCCESS in 24m 32s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f9f8175fed9c4ad6b6d76a821e9c23f7 : SUCCESS in 1h 05m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/52c5e7a20a9e47a4b72009bf21aa7e40 : SUCCESS in 22m 11s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2d6211ced17f48168e8cefe03ab94a4b : SUCCESS in 36m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d7acd0ca6d5a4c11823d90715a5679d9 : SUCCESS in 53m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a86aa75e12634dd5a6216b103204e7c7 : SUCCESS in 1h 29m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a159404fb1040fca19efc295b2f7dea : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6d968c0e182d46ff8bed963d36813242 : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/811f01f977134029a7493b267f40ad7f : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3b9acce16af343d4af6e142aff6a35e5 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4151026292b44af0b3dd2f60352c3773 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea9c5c765395419bae6a980797b4ec38 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21b6a6a59d5c4f0aaa559c55bfd5d500 : SUCCESS in 4m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4f6929a289c0491e8a59ecec6ee167e8 : SUCCESS in 26m 09s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1e38fcbc1c2944b4a6be08d818e21bfc : SUCCESS in 33m 04s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c56e4964ceb54625a264f9f66878f7ba : SUCCESS in 1h 02m 46s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6460a969becc43159b3b5aeee40b3945 : SUCCESS in 1h 33m 08s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/95ba331bd45a4bb5a10016aec4fdcfcd : SUCCESS in 36m 36s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/74a51c9a8b4f45fd9d1e72ca5e6c2b85 : SUCCESS in 19m 16s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c42c022d9a6f49e092186050d3b351bb : SUCCESS in 1h 11m 25s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/76478a67a697412db8b1ce153d4e2775 : SUCCESS in 51m 36s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5c347eefb75e4a5c887ad996d1761c3f : SUCCESS in 1h 23m 29s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/22e7954f1fca47579f1b1c898c46b2eb : SUCCESS in 29m 24s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"92e97c2860c8c106590d9ac886317ea66ce126f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 22:58:57.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9836da6bbe940c5b8e76a45fb7b7a9c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1f5734d9bb7c492ba4900a830f76ee7e : SUCCESS in 2h 49m 45s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/be4dbb76b90d43288ed10f21b80195b6 : SUCCESS in 22m 41s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3604453340324d1c99e2f768595d9190 : SUCCESS in 2h 14m 34s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4d19f1be64f047fab531ce79273cdc74 : SUCCESS in 2m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d50f2263d4e74c3aa9fa6d0bfbd207b2 : SUCCESS in 6m 28s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f5e50ef0c214426d8546ef364a74d31c : SUCCESS in 55m 53s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/bb6601dfae4d4ec9938a46592d5cadfa : SUCCESS in 2h 52m 47s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"800b7b2fb6289ac8c4ac797acae476262cabb6b6","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-07-16 09:40:43.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"1681f550aef3033c44e8a5c8df998026b3f1b8c4","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-07-16 14:22:46.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"79ae2ab9339eabad546b4c08a7dd98ec6e360ec7","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-07-18 09:34:08.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1b9240629f5c089393de6591121eb212348c19fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-07-18 09:34:55.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"f5fe5b52e9861cb0259d339a775b9bcd25403ff3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-07-18 09:36:22.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"6933b5ebf7a57f0386eb9eb1dd9545765c593127","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-07-18 09:37:05.000000000","message":"Uploaded patch set 9: New patch set was added with same tree, parent tree, and commit message as Patch Set 8.","accounts_in_message":[],"_revision_number":9},{"id":"068b76f636929d86bdcd1c41bdd98a7e4dcae3a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 11:36:18.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cb0afb361e942f8a8aca4aac11e5773\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/061852a62ba949298ecd1c944ad7bbb7 : SUCCESS in 24m 44s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/13cd3369b41e410f869bdfc922fcaac4 : SUCCESS in 1h 08m 08s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/45356ef18c474ef5ab17052013c2c958 : SUCCESS in 19m 57s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7311f49d9de14de590d6e1e55115f3c6 : SUCCESS in 28m 13s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/96ccf5126cbf47c7a3afc6a85c1585f1 : SUCCESS in 1h 02m 48s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6badd22f77a641fd95783f1a2dc451b3 : SUCCESS in 1h 23m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c427b7cda025495181edc27cbcf1ca7f : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9242ddb6b2044dbbc1a9deb9ed6c371 : SUCCESS in 2m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3f0373051754413a91ed4a0afbf9287c : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/97b9db27df3248c29d1a18b3b48a1888 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b422bd606f14e91a02ea507cbb2ca10 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f39161c782654770b562bbe573c404b0 : SUCCESS in 5m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64ec619ff18c4c3b92cdf6f8cadabdd9 : SUCCESS in 3m 55s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6a140e9415ff46ddbf6b13da06042275 : SUCCESS in 25m 54s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1a9eef6ae3c24780b0e44fb0a249823b : SUCCESS in 32m 28s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ae7b6545ff8441db81b68cda35380840 : SUCCESS in 1h 06m 05s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5a3d6e3fc5bb4a9085a9946784551dcb : SUCCESS in 1h 51m 23s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f290d102ae7a4dcca0335d99671c7679 : SUCCESS in 35m 30s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2df82b9b7f4c4a7ea866ac978b0df3e7 : SUCCESS in 20m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/439136888bef4d80b6788d31340b5f28 : SUCCESS in 48m 57s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/798ee7fe3c6f40d3aa9d7bee43d71463 : SUCCESS in 55m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/68d9f3d9f17b434ca8245e39bb00e725 : SUCCESS in 1h 12m 46s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c39cf9e49f5f45149c7adf1b64a1ae77 : SUCCESS in 30m 40s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"1cee303ac37dd08c4a3b92749b6c19c5cec46bf6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 12:42:50.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15e365969e0e469d8ff5b1c2ee6a633d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/95842c2bbd8045edac497db4b6d6091c : SUCCESS in 24m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6a07121bb7404b9ab167780dbf47cb42 : TIMED_OUT in 3h 04m 03s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ec40943444674dda9d51567dfbb806c9 : SUCCESS in 1h 09m 42s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cea59131c20448969737cfe4c7b62c42 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/52ddc85dc1d048f19d09708bcd456242 : SUCCESS in 4m 27s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1f15714ef03f488982bcbaf752468f1a : SUCCESS in 32m 54s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/439b9e96d1584328a7845bdba74fe628 : SUCCESS in 20m 37s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f272ee2748fc5930ee539ea34429155e17e8b072","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-07-21 07:41:15.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9be5a069fc2edced93e96e7d30ec7d1694d9a041","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-07-22 07:39:15.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"4176badeadc0c4a6e6d0709de6940fa013aef833","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-07-22 07:41:01.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"f332695c653f4632d6788e29467f289eadb4d403","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 09:30:11.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da5a2997cd2d4c03938c6e645fafc8cc\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e7bc10d4be7749f9956c2d04b237bb28 : SUCCESS in 24m 21s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/7e3f03b0025b4f3a9c6f4f4a900e6cc5 : SUCCESS in 1h 05m 56s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4dfb1d636be947c0bfd2a69f76a74227 : SUCCESS in 19m 21s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e9a2c036ee7a417facee30e4c65f8077 : SUCCESS in 29m 37s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e8e8d6699303495980a607ef6e8f59a6 : SUCCESS in 1h 11m 36s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4f627eb58dce4210b0bd3e704f9640c7 : SUCCESS in 1h 38m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68ec65d01cb442c29ebfca9283f7ef41 : SUCCESS in 2m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21a4aa66c0cb40e28ccfe97c3047d006 : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b6741245627e4034a53d273329e19e42 : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/13a700fa086d45608fe5c229707058fa : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b7cac9c455646b88e601fa57e3e7d0e : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48a6283ede53472e9d3512db39530ebd : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14181cfc70754ad786dba87fac66d49f : SUCCESS in 3m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/07254ff65d44428db5cadd81da20929a : SUCCESS in 23m 51s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c2761c55ff1f4197ab398429415612dd : SUCCESS in 29m 01s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/767c45a83f124823aa0d72fece6383d0 : SUCCESS in 59m 28s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/78bd5f917d0e4074a56a4133fa33ed21 : SUCCESS in 1h 38m 29s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/6dc3f54c460e41c786447e0b078ae303 : SUCCESS in 30m 35s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/916ea80938194aa788952d856fd31a3d : SUCCESS in 19m 13s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1f6c0a55dfbc4187b27a98a6906559b6 : SUCCESS in 41m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/749b1047b3c4451d822588a44039241b : SUCCESS in 54m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1c51b4ab649c4415a68551d5641062df : SUCCESS in 1h 22m 26s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/19cf0b19056b44c78352b4ac3d8bfb7e : SUCCESS in 27m 07s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"b5bfb7dee1f13833dcdf72c63464ea7888f8334c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 09:43:22.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf51001b53f842adbb8b04f9622a5d9e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f92945508e76426cba893e4885c5da45 : SUCCESS in 31m 44s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/273012aaf4ac42d08c6f569ff58c17c4 : SUCCESS in 25m 34s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6bec0bf70c254106bf69c04e9a2a4424 : SUCCESS in 1h 06m 26s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/63040ef1d6b045d28bde3a3e80010c93 : SUCCESS in 2m 47s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0abd0246e5ef4f69bae42a62342dd53c : SUCCESS in 3m 07s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1a399c849669427b8d7d852faddaebd1 : SUCCESS in 38m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8ab54795cdc64151b365ceb2c5f19b8f : SUCCESS in 2h 02m 08s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f17daf7eb3b48c94aa777b61c2924b60c876fd42","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-07-23 07:38:37.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"a99fff03e6d5b0c2fa1474f7e7080d61cbb75b42","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-07-29 14:59:12.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"f38e5aa1bfe9fcefebbe2321ca5a6d13ebaca2c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-09-16 08:33:52.000000000","message":"Patch Set 11: Patch Set 10 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"4813949740e1cf42e3157990a7736434164cc56a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 10:09:53.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a246d919dc424ff79ae8a35004fac8e5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b181cbf6800c43708acfb13acf22b205 : SUCCESS in 21m 26s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/93e067957f0d412480a88aae3eaa57b6 : FAILURE in 5m 50s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5c46725b5f2f4633a716ec005d08ea05 : SUCCESS in 16m 08s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/64946ef754de40ce922410b24cb98839 : SUCCESS in 22m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/fd2b433242a94e3eb556bc2d25edf52c : SUCCESS in 59m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3654da5317c4474d94e4a162c76a4293 : FAILURE in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e0527dc2dfa48a489cc3d198b4e0779 : SUCCESS in 2m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0a59eedd64e472981015f44371595b5 : SUCCESS in 2m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/016853e5ba7143898da800923da8903a : SUCCESS in 2m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6795f220c4e349ae8b6f1d693c50cfe9 : SUCCESS in 2m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08c7b934b49745cf806a255ac183a2f0 : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16b497b3dbfe4ce89f61392f5287905a : SUCCESS in 4m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/639bcf15bbbb4a479c823aee6a1b60af : SUCCESS in 4m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/eede8afa8ffe46c8bfcdf5b98a36ac53 : SUCCESS in 28m 31s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5bcb8fe1b1c64b1595f53d97203557d3 : SUCCESS in 37m 56s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d9b1c9c862574133aacf9c3a9f2ab229 : SUCCESS in 1h 03m 07s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3b8a126e6198432dbba7250a359001e2 : SUCCESS in 1h 25m 53s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f360b75dc00549e483a82e8d8069974d : SUCCESS in 32m 52s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7166202397844a8aa83ce5b4a685e62f : SUCCESS in 20m 36s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1a1267fbacfa4da389571efeb4a77a41 : SUCCESS in 50m 53s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/710418cf4e364a679a9b8cc6f06b57c4 : SUCCESS in 57m 04s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9c06f4e783754d44b7bbd38f36259ed7 : SUCCESS in 1h 19m 50s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/329825ac46e742138db77b803a4e041b : FAILURE in 5m 38s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"0c5d4121d1878ac1ad89db8c4543c1d3b70b56c2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-16 10:13:13.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"7ead49d581bbc0f172789588adbde20705d08de6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 11:38:52.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b10bfde431ce49eabef0b1527982d951\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7e7b84316bfb4e6a8defa2ac2f3c3b06 : TIMED_OUT in 3h 03m 12s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cf21c24a60f34e32ab2bbffaeebb8d09 : SUCCESS in 2h 05m 15s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c937448e5e684db5809f79ccd6879182 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1b9ee3a111d24046be584646629f761e : SUCCESS in 5m 12s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ef6548504d8d5a5c30b3180d74b6e3152b17d3e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 11:48:27.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8262cb96a7ae46ab8d64d37b2adf5661\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8989e79b46574b28a27f85fa145081ac : SUCCESS in 26m 18s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2d4201ac329a4ad89cb01194f45fef14 : SUCCESS in 1h 04m 30s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/af9c6bd74fe24abb84559574d045ca8c : SUCCESS in 15m 56s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d4b335308c344d93bb74308a0583e404 : SUCCESS in 24m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5c2ffca0f9044944af0aa952f9a00a6c : SUCCESS in 1h 02m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/25afb245394540cf9ec6919487cb3afd : SUCCESS in 1h 29m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fc84594c2da49408d9a4d66114e6a09 : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/070892c39ecf48f694c7b92068324a3c : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4b674b5978e843fa9f86cf0d6a07cd53 : SUCCESS in 2m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a320fefe874f494e88cdaf86a26bbd04 : SUCCESS in 3m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ab98857140f4f05b415fa8c6088a219 : SUCCESS in 2m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0c746eda9bc4b638b445d7aeffd2750 : SUCCESS in 4m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f53285a0287d47ce8dadb522d00bb1ac : SUCCESS in 3m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/21f51936662244869bf3360096e19aa5 : SUCCESS in 28m 01s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b3ef12be404f4671911d293372b1b6fa : SUCCESS in 44m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/03829e18fe704cf09646c0712a977a5d : SUCCESS in 1h 08m 37s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a196c7280a5a49ab84c2e2a430d4a468 : SUCCESS in 1h 25m 52s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c13fcb0aa7a14f1eb961db8de47eecb4 : SUCCESS in 31m 39s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/38761a6650b34a04851f33e2ee416caf : SUCCESS in 17m 54s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8d48af7620e845babafdcc5f0370032b : SUCCESS in 1h 08m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8ba687b54ad0461dbf11ddf8e98a977c : SUCCESS in 50m 10s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/03d81d883e85445eb4465ffcb967f717 : SUCCESS in 1h 14m 02s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b14242135d9b498e8d940c7ec0c8087c : SUCCESS in 22m 23s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"51ab89ca199fe941d72f3f1101f658a1bfbf4fe0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-10-01 08:50:13.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"105fcac4d1d52aa666d51d9841e767ec779d5137","author":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"date":"2024-10-01 08:52:13.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"aa6090c69321784c28694dde1a013bfc134ad048","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 10:14:58.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d8aff95504c46d4aca0a88faba4ef75\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/570a153acc8e4b42b78aaf6f1e45d180 : SUCCESS in 28m 55s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/71a6009f292d471c95dc77f030482028 : FAILURE in 6m 26s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/09aa3d15d1d045248429b8af19e9b93e : SUCCESS in 20m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c2f8420861c848c5a0951f63dca475e4 : SUCCESS in 33m 16s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/fe795e3acd2c437a81123e2e0a5c4bd1 : SUCCESS in 1h 01m 14s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5bababb95073495797405919d92d4040 : SUCCESS in 1h 21m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2e19ae543344392b4f4106d39e60b11 : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4da03f5fa5094dd48629bcac60fa8e07 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e404b2af98104236baaa058238183e42 : SUCCESS in 3m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/963750fe49d343338bbe3ae3d3e391df : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52113c10a738403296274c2e7276b037 : SUCCESS in 4m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02e2cf083c2f4013bd12983a705ebece : SUCCESS in 3m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e245dfc40eec4f0e94b653e264740bd7 : SUCCESS in 33m 33s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/561925ddabf94edba9c94410f2e03ac4 : SUCCESS in 36m 24s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f9bb8681a7f74db596df433371f78f49 : SUCCESS in 1h 11m 48s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/54d0c2fffcd54ac9a1816550b39622e7 : SUCCESS in 1h 15m 00s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/6c6273f811c94edf8458a959213f517f : SUCCESS in 34m 14s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8b1cc616d1394ccda83818125ce70b27 : SUCCESS in 22m 43s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/439e6e78b6cb445e80971887dc0ba238 : SUCCESS in 41m 26s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a0786bd55efc4adfac49dad9ec28357a : SUCCESS in 53m 10s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ad52ad864b464d4aab3d4c6e1c5d489c : SUCCESS in 1h 22m 11s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/03451d56a6fb4b6392779d643a30cc14 : SUCCESS in 26m 43s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"f993de55a1e4d81b63a12e59a7612033e9a047d0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 11:54:58.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/042a16771b5c4e289f3328883b20d8fa\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e9f1ae3838e845409dc1ca43fcf1c98c : SUCCESS in 22m 17s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e35667d50c8e40d2ad9a1e01a89c4183 : TIMED_OUT in 3h 02m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d465b9dfb9fd4fcaa0214b8b17a97f33 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/839df3b21981414a92fc66e9442a7f58 : SUCCESS in 8m 03s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"9aa9c2c343b5f25a11876e6ee58932dc34a1c2f9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-09 17:02:36.000000000","message":"Abandoned\n\nno updates for over a year","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"0d3f5dff5fb79b6ad990043f30b87c2156e2ad1c","revisions":{"aaa3cd6ab971e616a6176648a1fee08a71711b82":{"kind":"REWORK","_number":1,"created":"2024-06-10 11:14:06.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/58/921658/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/921658/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/921658/1"}}},"commit":{"parents":[{"commit":"968711579fbe73316227bd2b1993c5656adad9f4","subject":"Drop packaging requirement and unpin docker-py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/968711579fbe73316227bd2b1993c5656adad9f4"}]}],"author":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2024-06-10 11:13:52.000000000","tz":120},"committer":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2024-06-10 11:13:52.000000000","tz":120},"subject":"[WIP] - Add new hardware prometheus exporter:","message":"[WIP] - Add new hardware prometheus exporter:\n\n  * Add a BMC dedicated exporter that leverage BMCs redfish endpoints.\n\nCloses-Bug: #2068884\nChange-Id: Ida297f5c6473a1b4f2e58b691a3dec6a5c3f078d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aaa3cd6ab971e616a6176648a1fee08a71711b82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aaa3cd6ab971e616a6176648a1fee08a71711b82"}]},"branch":"refs/heads/master"},"15a970e69cd55687edfa1211574836f4c9ce3e1f":{"kind":"REWORK","_number":2,"created":"2024-06-13 09:20:43.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/58/921658/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/921658/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/921658/2"}}},"commit":{"parents":[{"commit":"968711579fbe73316227bd2b1993c5656adad9f4","subject":"Drop packaging requirement and unpin docker-py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/968711579fbe73316227bd2b1993c5656adad9f4"}]}],"author":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2024-06-13 09:20:29.000000000","tz":120},"committer":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2024-06-13 09:20:29.000000000","tz":120},"subject":"[WIP] - Add new hardware prometheus exporter:","message":"[WIP] - Add new hardware prometheus exporter:\n\n  * Add a BMC dedicated exporter that leverage BMCs redfish endpoints.\n\nCloses-Bug: #2068884\nChange-Id: Ida297f5c6473a1b4f2e58b691a3dec6a5c3f078d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/15a970e69cd55687edfa1211574836f4c9ce3e1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/15a970e69cd55687edfa1211574836f4c9ce3e1f"}]},"branch":"refs/heads/master"},"0991e4e7024738845f73f14077e21a9620413d4b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-06-13 09:55:24.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/58/921658/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/921658/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/921658/3"}}},"commit":{"parents":[{"commit":"968711579fbe73316227bd2b1993c5656adad9f4","subject":"Drop packaging requirement and unpin docker-py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/968711579fbe73316227bd2b1993c5656adad9f4"}]}],"author":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2024-06-13 09:20:29.000000000","tz":120},"committer":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2024-06-13 09:55:24.000000000","tz":0},"subject":"Add new hardware prometheus exporter:","message":"Add new hardware prometheus exporter:\n\n  * Add a BMC dedicated exporter that leverage BMCs redfish endpoints.\n\nCloses-Bug: #2068884\nChange-Id: Ida297f5c6473a1b4f2e58b691a3dec6a5c3f078d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0991e4e7024738845f73f14077e21a9620413d4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0991e4e7024738845f73f14077e21a9620413d4b"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9ecbd9763e442f9fb0e3ff0ee27f7a8fc11100bc":{"kind":"REWORK","_number":4,"created":"2024-06-13 11:54:35.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/58/921658/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/921658/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/921658/4"}}},"commit":{"parents":[{"commit":"968711579fbe73316227bd2b1993c5656adad9f4","subject":"Drop packaging requirement and unpin docker-py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/968711579fbe73316227bd2b1993c5656adad9f4"}]}],"author":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2024-06-13 11:54:19.000000000","tz":120},"committer":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2024-06-13 11:54:19.000000000","tz":120},"subject":"Add new hardware prometheus exporter:","message":"Add new hardware prometheus exporter:\n\n  * Add a BMC dedicated exporter that leverage BMCs redfish endpoints.\n\nCloses-Bug: #2068884\nChange-Id: Ida297f5c6473a1b4f2e58b691a3dec6a5c3f078d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ecbd9763e442f9fb0e3ff0ee27f7a8fc11100bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ecbd9763e442f9fb0e3ff0ee27f7a8fc11100bc"}]},"branch":"refs/heads/master"},"1a3ce83d03a7f9db09443831fa2591a035385219":{"kind":"REWORK","_number":5,"created":"2024-06-13 20:01:43.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/58/921658/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/921658/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/921658/5"}}},"commit":{"parents":[{"commit":"968711579fbe73316227bd2b1993c5656adad9f4","subject":"Drop packaging requirement and unpin docker-py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/968711579fbe73316227bd2b1993c5656adad9f4"}]}],"author":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2024-06-13 20:01:25.000000000","tz":120},"committer":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2024-06-13 20:01:25.000000000","tz":120},"subject":"Add new hardware prometheus exporter:","message":"Add new hardware prometheus exporter:\n\n  * Add a BMC dedicated exporter that leverage BMCs redfish endpoints.\n\nCloses-Bug: #2068884\nChange-Id: Ida297f5c6473a1b4f2e58b691a3dec6a5c3f078d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a3ce83d03a7f9db09443831fa2591a035385219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a3ce83d03a7f9db09443831fa2591a035385219"}]},"branch":"refs/heads/master"},"67cac21d1db430e4c475199168167bcd1d7a4f6e":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2024-06-13 20:03:04.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/58/921658/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/921658/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/921658/6"}}},"commit":{"parents":[{"commit":"968711579fbe73316227bd2b1993c5656adad9f4","subject":"Drop packaging requirement and unpin docker-py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/968711579fbe73316227bd2b1993c5656adad9f4"}]}],"author":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2024-06-13 20:01:25.000000000","tz":120},"committer":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2024-06-13 20:03:04.000000000","tz":0},"subject":"Add new hardware prometheus exporter:","message":"Add new hardware prometheus exporter:\n\n  * Add a BMC dedicated exporter that leverage BMCs IPMI endpoints.\n\nCloses-Bug: #2068884\nChange-Id: Ida297f5c6473a1b4f2e58b691a3dec6a5c3f078d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67cac21d1db430e4c475199168167bcd1d7a4f6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67cac21d1db430e4c475199168167bcd1d7a4f6e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"204d8f79fd932dbdb1f7bcd116813df888ff0200":{"kind":"REWORK","_number":7,"created":"2024-07-18 09:34:55.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/58/921658/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/921658/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/921658/7"}}},"commit":{"parents":[{"commit":"968711579fbe73316227bd2b1993c5656adad9f4","subject":"Drop packaging requirement and unpin docker-py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/968711579fbe73316227bd2b1993c5656adad9f4"}]}],"author":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2024-07-18 09:34:29.000000000","tz":120},"committer":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2024-07-18 09:34:29.000000000","tz":120},"subject":"Add new hardware prometheus exporter:","message":"Add new hardware prometheus exporter:\n\n  * Add a BMC dedicated exporter that leverage BMCs redfish endpoints.\n\nCloses-Bug: #2068884\nChange-Id: Ida297f5c6473a1b4f2e58b691a3dec6a5c3f078d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/204d8f79fd932dbdb1f7bcd116813df888ff0200"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/204d8f79fd932dbdb1f7bcd116813df888ff0200"}]},"branch":"refs/heads/master"},"33e5e8aa0d22955fef297ac5764428224f1f868f":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2024-07-18 09:36:22.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/58/921658/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/921658/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/921658/8"}}},"commit":{"parents":[{"commit":"968711579fbe73316227bd2b1993c5656adad9f4","subject":"Drop packaging requirement and unpin docker-py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/968711579fbe73316227bd2b1993c5656adad9f4"}]}],"author":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2024-07-18 09:34:29.000000000","tz":120},"committer":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2024-07-18 09:36:22.000000000","tz":0},"subject":"Add new hardware prometheus exporter:","message":"Add new hardware prometheus exporter:\n\n  * Add a BMC dedicated exporter that leverage BMCs IPMI endpoints.\n\nCloses-Bug: #2068884\nChange-Id: Ida297f5c6473a1b4f2e58b691a3dec6a5c3f078d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33e5e8aa0d22955fef297ac5764428224f1f868f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/33e5e8aa0d22955fef297ac5764428224f1f868f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"5ef94f04329d8982a6179861b5461cb448ce61ec":{"kind":"NO_CHANGE","_number":9,"created":"2024-07-18 09:37:05.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/58/921658/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/921658/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/921658/9"}}},"commit":{"parents":[{"commit":"968711579fbe73316227bd2b1993c5656adad9f4","subject":"Drop packaging requirement and unpin docker-py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/968711579fbe73316227bd2b1993c5656adad9f4"}]}],"author":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2024-07-18 09:36:43.000000000","tz":120},"committer":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2024-07-18 09:36:43.000000000","tz":120},"subject":"Add new hardware prometheus exporter:","message":"Add new hardware prometheus exporter:\n\n  * Add a BMC dedicated exporter that leverage BMCs IPMI endpoints.\n\nCloses-Bug: #2068884\nChange-Id: Ida297f5c6473a1b4f2e58b691a3dec6a5c3f078d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ef94f04329d8982a6179861b5461cb448ce61ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ef94f04329d8982a6179861b5461cb448ce61ec"}]},"branch":"refs/heads/master"},"9795a56f43c105c78afbad32d91f0e03788c37ea":{"kind":"REWORK","_number":10,"created":"2024-07-22 07:39:15.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/58/921658/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/921658/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/921658/10"}}},"commit":{"parents":[{"commit":"968711579fbe73316227bd2b1993c5656adad9f4","subject":"Drop packaging requirement and unpin docker-py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/968711579fbe73316227bd2b1993c5656adad9f4"}]}],"author":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2024-07-18 09:36:43.000000000","tz":120},"committer":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2024-07-22 07:39:15.000000000","tz":0},"subject":"Add new hardware prometheus exporter:","message":"Add new hardware prometheus exporter:\n\n  * Add a BMC dedicated exporter that leverage BMCs IPMI endpoints.\n\nCloses-Bug: #2068884\nChange-Id: Ida297f5c6473a1b4f2e58b691a3dec6a5c3f078d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9795a56f43c105c78afbad32d91f0e03788c37ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9795a56f43c105c78afbad32d91f0e03788c37ea"}]},"branch":"refs/heads/master"},"835a8fed768e87ff828fe3526bb9e8046b02607b":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2024-09-16 08:33:52.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/58/921658/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/921658/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/921658/11"}}},"commit":{"parents":[{"commit":"6c5a443efe4e267639913e827918b83551318cab","subject":"Merge \"rabbitmq: Switch to Cloudsmith mirrors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c5a443efe4e267639913e827918b83551318cab"}]}],"author":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2024-07-18 09:36:43.000000000","tz":120},"committer":{"name":"Gaël THEROND","email":"gael.therond@bitswalk.com","date":"2024-09-16 08:33:52.000000000","tz":0},"subject":"Add new hardware prometheus exporter:","message":"Add new hardware prometheus exporter:\n\n  * Add a BMC dedicated exporter that leverage BMCs IPMI endpoints.\n\nCloses-Bug: #2068884\nChange-Id: Ida297f5c6473a1b4f2e58b691a3dec6a5c3f078d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/835a8fed768e87ff828fe3526bb9e8046b02607b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/835a8fed768e87ff828fe3526bb9e8046b02607b"}]},"branch":"refs/heads/master","description":"Rebase"},"0d3f5dff5fb79b6ad990043f30b87c2156e2ad1c":{"kind":"REWORK","_number":12,"created":"2024-10-01 08:50:13.000000000","uploader":{"_account_id":32398,"name":"Gaël THEROND","display_name":"Fl1nt","email":"gael.therond@bitswalk.com","username":"Fl1nt"},"ref":"refs/changes/58/921658/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/58/921658/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/58/921658/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/58/921658/12"}}},"commit":{"parents":[{"commit":"968711579fbe73316227bd2b1993c5656adad9f4","subject":"Drop packaging requirement and unpin docker-py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/968711579fbe73316227bd2b1993c5656adad9f4"}]}],"author":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2024-10-01 08:50:00.000000000","tz":120},"committer":{"name":"Gaël THEROND (Fl1nt)","email":"gael.therond@bitswalk.com","date":"2024-10-01 08:50:00.000000000","tz":120},"subject":"Add new hardware prometheus exporter:","message":"Add new hardware prometheus exporter:\n\n  * Add a BMC dedicated exporter that leverage BMCs IPMI endpoints.\n\nCloses-Bug: #2068884\nChange-Id: Ida297f5c6473a1b4f2e58b691a3dec6a5c3f078d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d3f5dff5fb79b6ad990043f30b87c2156e2ad1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d3f5dff5fb79b6ad990043f30b87c2156e2ad1c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
