)]}'
{"id":"openstack%2Fkolla~921820","triplet_id":"openstack%2Fkolla~stable%2F2024.1~I1dda73b0bc9a9f57ec8fda3f2234449b04906410","project":"openstack/kolla","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-06-12 11:45:48.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I1dda73b0bc9a9f57ec8fda3f2234449b04906410","subject":"CI: Pin requests to \u003c2.32 for publish jobs","status":"MERGED","created":"2024-06-12 05:59:07.000000000","updated":"2024-06-12 13:25:35.000000000","submitted":"2024-06-12 13:24:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921820","cherry_pick_of_change":921032,"cherry_pick_of_patch_set":6,"meta_rev_id":"975bc0b646f3ba3c1b5797b345ed349923b2abaa","_number":921820,"virtual_id_number":921820,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-12 13:24:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-06-12 11:36:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-06-12 07:24:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2024-06-12 11:45:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2024-06-12 11:45:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":-1,"date":"2024-06-12 07:24:39.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-12 07:24:39.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-06-12 08:24:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-12 09:06:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-12 11:36:18.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"277a01c141a4c40fa6893cc8ad00f205149c4511","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-06-12 05:59:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"33010e0e7fb3501195cb30714dc8db42b405fa76","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-06-12 07:24:39.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"641641ab59055449778de162833535460d63cbb6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 08:24:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13d239aee4534aba860002cedb149f58\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d5d34e590c2045c4aaaa97bd7294c5d9 : SUCCESS in 33m 04s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1ae208da5b5a46199be78233857dde54 : SUCCESS in 24m 50s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cacdb9fd20ff486680146f8399f18c74 : SUCCESS in 1h 06m 00s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aecdeb1fd51a4f65b3692ac6eea030db : SUCCESS in 2m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e45c98c126e94f28829bed93d196f277 : SUCCESS in 3m 00s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/990bc9a3900d44408fd8012ffe446ce1 : SUCCESS in 52m 45s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/335d4916230c47738312dc6533bf02e3 : SUCCESS in 2h 22m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"232ca9fa99c413a15b646ec11ca8295675f7bd42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 09:06:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/972901b83bc548188c0b9932f09cdee9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f1706d1a5df144b0a88e7588f4757b24 : SUCCESS in 26m 56s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/de9f586b3ebf463e8f4d83184d25ac39 : SUCCESS in 1h 10m 00s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/949f8d4f69c047abb76cd7ed69d7f436 : SUCCESS in 19m 38s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f53fe391887c4756bf3cc1f21c8e7397 : SUCCESS in 30m 01s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a1cbd9cd367b4ce883fd43430d24f472 : SUCCESS in 1h 04m 47s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/769050f83b094d7f8ebb8ca302e9748c : SUCCESS in 1h 30m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/056708dc041e44c888075b5f1c2e0677 : SUCCESS in 2m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cdde4a1b4444159904958085d4051c8 : SUCCESS in 2m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/05ddd60787644eaba6255431ce4d835a : SUCCESS in 3m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f9460b844ca4e73975d0c04d52a7b1a : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa0f2cf51c77429992cdfca08158d012 : SUCCESS in 4m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ce98012ab7924928b4b7eb9d1b7537b2 : SUCCESS in 20m 12s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3955543808064318a5d83e6746d60bfe : SUCCESS in 33m 01s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/95fe1bf17bb54b3cbaf59252ee43342b : SUCCESS in 1h 11m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e1fc9d32dca64cba8bf3e7106d15c647 : SUCCESS in 1h 29m 17s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/65a3e9d2294c45909dfd5285061a2169 : SUCCESS in 16m 10s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/36ff119714ad4db18a447a41d7a0c0ea : TIMED_OUT in 3h 01m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4e72d77475c640009fb0e41679971f37 : SUCCESS in 54m 13s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b77f4d7883cb43ed818830ab3f44bcc3 : SUCCESS in 1h 17m 33s","accounts_in_message":[],"_revision_number":1},{"id":"f90f46a0c068a5fc8f8078ec5119953ff648eb41","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-06-12 09:38:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"690bb8feb85ae620dc46f56c9cd11423065d9d33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 11:19:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cce53d89f27e4e65ae88ef84b8e283d1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f129e32fd49642f099ba018f0bc429ea : SUCCESS in 21m 57s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/76c319dae8f645cfbbd0f18017557a66 : SUCCESS in 57m 43s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7fa286b87efd4f468ffbc4f9df19b118 : SUCCESS in 16m 59s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6f5378a548244222a55bbb70e4dfdf13 : SUCCESS in 30m 58s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8f164219d19a4b9fb139d567ff0bd221 : SUCCESS in 1h 07m 11s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4bf6049ea4cb4b4ba6ea532a636b834b : SUCCESS in 1h 34m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7ef80b761ad4cbaad2bc147e76f8ef5 : SUCCESS in 2m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14b8a4cf1465478dbb467a583159a63c : SUCCESS in 3m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e465d4c877e5409598b6129ad1ba1a5b : SUCCESS in 2m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f9e464f96c14d4bb708c313e42b0c96 : SUCCESS in 2m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2244d5062f2246b69de74a5480344ecc : SUCCESS in 4m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/569b00cd0ad147258388b47e57a0afbd : SUCCESS in 22m 38s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/edb01f71969c4ee7b9e588b3728d788b : SUCCESS in 27m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8f18d426dc5d42c39ec586d9a0923cd0 : SUCCESS in 58m 45s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e605a1622daf40e38aff472f23779c9e : SUCCESS in 1h 22m 11s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9a212a6fa91f4c7390a49314045df27e : SUCCESS in 15m 50s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/57857832cd75448d8986c6f2d5bf5628 : SUCCESS in 1h 00m 56s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/91afa969e0fa4ed9a6e10aef467fbbfe : SUCCESS in 51m 21s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1306f9f25eda4a1783b785cc26c88223 : SUCCESS in 1h 12m 14s","accounts_in_message":[],"_revision_number":1},{"id":"65f01dec4e0536067a9a9673f12e4fe15ad52a9f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-12 11:36:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5bdeb4597a301a8f4cbe26f922e9473b8038e2f1","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-06-12 11:45:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a74abd73ca32af3b01224ed4283b436901b024d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 11:46:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"73246c99b33c35edaf4ccb0b1e70fa108ea605fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 12:31:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/985720fbae9641cabf50ee32433ec7b7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/79ef9b5e3ea94c5898d8cd75ca54a071 : SUCCESS in 21m 07s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/753e1213ec7b42299a863db123f3f873 : SUCCESS in 16m 02s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0bb84abd232b4d9c8d0a42d08be062b6 : SUCCESS in 1h 12m 48s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/347cf5df16cb435c8a3c4f0f0cc86204 : SUCCESS in 2m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d1f6da2e85414667b49aee9d315d0def : SUCCESS in 2m 24s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/44983dae6c904a6d90e6fe0b779b2dcf : SUCCESS in 47m 48s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d9451bc6ceee463d9b9b9ff624f7c630 : SUCCESS in 2h 51m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"44d4dd85888f1b14b346748a68b5f54fb757a0ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 13:24:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0442503a6de3488496a0e43a68a988c3\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/431a4786eff840e5b842057c331a62a7 : SUCCESS in 19m 13s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0b045b26c3b24577af8dc36c270cc19f : SUCCESS in 26m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a681ebfcff844a1f8e08dd000bd770bf : SUCCESS in 1h 04m 45s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8ba2ce2e7fbe464ebf3e74a8bb2b4de5 : SUCCESS in 1h 33m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0637eb67c664d25a0b03c41a6549d97 : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94ec87a3f27347eab9ee6f76526f8915 : SUCCESS in 4m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c85b4ed155c24ddba34ad9aa61da7c3d : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/995848f83b3e43d5bf4f839b6bef7e54 : SUCCESS in 4m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a3f7371c111c4975a785581ac026fc5d : SUCCESS in 26m 05s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3374f255f67049039f3c9ef4f32f717f : SUCCESS in 31m 43s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b860e51e17c745048333085f9548472d : SUCCESS in 1h 11m 10s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/660d0c5a96454245a74512b287f6dec4 : SUCCESS in 1h 36m 33s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6d6b2105600f41e0b338e95cf6cd3e5f : SUCCESS in 16m 32s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b05513c8793447389165035fe56a2070 : SUCCESS in 54m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/12064ea4a70845ffa47d00ed6f1b8ae0 : SUCCESS in 49m 10s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8d5b5cd95d934e7d958f912de17f6707 : SUCCESS in 1h 12m 59s","accounts_in_message":[],"_revision_number":1},{"id":"7cc4041ede0fa1f65a854728dce101bee4ad5bbd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 13:24:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"975bc0b646f3ba3c1b5797b345ed349923b2abaa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 13:25:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77b9448a3c584defb4a8d7abbc8dad43\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90a1a5409fe9478aa8928b82277f3c28 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bbe297d5b9a0833139dd79a269139e59973cfed9","revisions":{"bbe297d5b9a0833139dd79a269139e59973cfed9":{"kind":"REWORK","_number":1,"created":"2024-06-12 05:59:07.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/20/921820/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/20/921820/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/921820/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/921820/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/20/921820/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/20/921820/1"}}},"commit":{"parents":[{"commit":"7efe8a37f05dcb2656b2abf8ce9204efeaafb82b","subject":"Drop packaging requirement and unpin docker-py","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7efe8a37f05dcb2656b2abf8ce9204efeaafb82b"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-06-03 13:42:45.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-06-12 05:59:07.000000000","tz":0},"subject":"CI: Pin requests to \u003c2.32 for publish jobs","message":"CI: Pin requests to \u003c2.32 for publish jobs\n\nPeriodic jobs are failing because Zuul Ansible executor runs with\nAnsible 8 which uses community.docker collection that needs requests\u003c2.32.\nRevert this patch when Zuul supports Ansible 10, and/or Ansible 9 gets\nfixed community.docker collection. [1]\n\n[1] https://github.com/ansible-collections/community.docker/issues/868\n\nRelated-Bug: #2066364\nChange-Id: I1dda73b0bc9a9f57ec8fda3f2234449b04906410\n(cherry picked from commit 428f37124986019bedc897469f9b1f3470b62082)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bbe297d5b9a0833139dd79a269139e59973cfed9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bbe297d5b9a0833139dd79a269139e59973cfed9"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
