)]}'
{"id":"openstack%2Fkolla~922066","triplet_id":"openstack%2Fkolla~master~I9fdccbe3bde07a7f9a1e6ac82b1096a024e1b0ee","project":"openstack/kolla","branch":"master","topic":"fix/apple_silicon_arch","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-07-16 11:12:11.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"34311":{"account":{"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},"last_update":"2024-07-16 14:39:21.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-07-16 11:07:54.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I9fdccbe3bde07a7f9a1e6ac82b1096a024e1b0ee","subject":"fix: apple silicon architecture selection","status":"MERGED","created":"2024-06-16 16:49:31.000000000","updated":"2024-07-16 14:40:35.000000000","submitted":"2024-07-16 14:39:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922066-fix/apple_silicon_arch","meta_rev_id":"c311892e93eedd7771da175d02ed12e346a275a1","_number":922066,"virtual_id_number":922066,"owner":{"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-16 14:39:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-07-16 11:12:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-07-16 11:07:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-07-16 11:12:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-07-16 11:12:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-07-16 11:07:54.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-16 18:30:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-17 08:33:02.000000000","updated_by":{"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-07-16 09:44:13.000000000","updated_by":{"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"0ca7692f371ef62171cf6b99e7987781b4998534","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},"date":"2024-06-16 16:49:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfd45c28de6935c41ddc6fb83ba45adc5a5a8f1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-16 18:30:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98e946119e3f41bf95f9585f815ac4ae\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f4b797c83dea465d878dd4a39c92a87b : SUCCESS in 24m 10s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3f3b7e66678d4d71be2320decc1054c4 : SUCCESS in 1h 01m 35s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/57b96a9961324c1682f9fb5fb4101630 : SUCCESS in 18m 03s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ddafdda5ced74868a9bbf347865dc430 : SUCCESS in 28m 46s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/32be89b12de04217a6ab53fa3a7e76d3 : SUCCESS in 57m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/babf721a4fdd48e3a6d64c5f2f26a8c1 : SUCCESS in 1h 27m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23622c858c3a4c899b36d8649e967f54 : SUCCESS in 2m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b723f35a0910465b905953bf2a921fe7 : SUCCESS in 2m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/967cfb5c2bb84a689dc78bcc281cf199 : SUCCESS in 2m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/43e9d542f4294865a976c48c116df4db : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ab204d8ecba41a783900111a64a2bc6 : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60278c0dcc6a4f43bd92f33961de88ce : SUCCESS in 3m 57s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/299d96a06a0147cc95fbaf3caa932557 : SUCCESS in 28m 15s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9c681c79e53d4014a2820a8969623fcf : SUCCESS in 40m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7e55beaf686d4750a55e93ac1cb4f0cb : SUCCESS in 1h 03m 10s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3c82b5e2a13947a5b65c225cb3159720 : SUCCESS in 1h 33m 49s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/50aba73e049544fa8bb5a9081e7ec722 : SUCCESS in 19m 32s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/287a7410c30e41d78453715988d93eb3 : SUCCESS in 57m 08s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c2983395bd9e4af39b3b52aefbd7e583 : SUCCESS in 50m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8822eb0e5c6d4b28909f8bf43a338431 : SUCCESS in 1h 19m 25s","accounts_in_message":[],"_revision_number":1},{"id":"a42cd48ebcf2fd5a04c1a817e966723bb1452107","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-16 19:08:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0729d2bc0971412a94913533b9495a58\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/4c45a80cd22f4b9a953bf860518913fc : SUCCESS in 25m 29s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c292d115994a41ca82aee7c43d68abb9 : SUCCESS in 18m 20s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/754a8165cca644658904796f7a831140 : SUCCESS in 54m 42s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6599c5d4cd1345eeafb2f0b87f4ee5ee : SUCCESS in 2m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e85551b082b740a6a604e0f0a6923af1 : SUCCESS in 4m 44s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/229b02eb5a3e47a8a45c0d8552005002 : SUCCESS in 58m 47s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e4aaf6e2fbf44eda80b6da0ef963b335 : SUCCESS in 2h 16m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0ca5b7e92aa3a56d8ce6c7e8d4df855719f72583","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-07-16 09:52:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7f6e0bf2e9da5efad1f9518617adcfa7df5fbc00","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-16 10:37:58.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"add28790fc9373679757dff7bcd4b9f5506240f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},"date":"2024-07-16 11:02:52.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"c1bdb4b53dd6afa980e5914273ab84b6092cd37d","author":{"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},"date":"2024-07-16 11:03:46.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a9f743dfd46016f84bb6768569d758f844784520","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-07-16 11:07:54.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"b0dd738230247078eb51bae208a11a75e4ff8aa5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-16 11:12:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5c4e9d635f5f70eb8f89009872e205d70afa50d0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-16 11:12:19.000000000","message":"Patch Set 2: Backport-Candidate+2","accounts_in_message":[],"_revision_number":2},{"id":"f55f1489d6045ebd78041cb0992fd6ade084650d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 12:51:23.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f402e4250d34ff4831ae56e6f8f759b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/68e75c29e01c432ca66c9ce9a5ec0fac : SUCCESS in 30m 04s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/fd526879d0894820ab2059e401d1901a : SUCCESS in 1h 05m 02s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fa3a8c3e382a4c0ba80ae3858a17880c : SUCCESS in 16m 27s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f5d40548254c4b6ea6b12c1fa398f038 : SUCCESS in 28m 19s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/34bcaae7296940b79356bb03562d6106 : SUCCESS in 1h 00m 04s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1d3366c108974dcc9ece2117a8ee57ce : SUCCESS in 1h 41m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80e45424c7c84e6193c6274f8487e998 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93e3b7ca150c435cac3465c42b3947db : SUCCESS in 3m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/123b6167f18841dda27f072221580234 : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d0590b32959648859e19253a691a15ad : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf641c7d5fc94c4ab3d8c167ac296661 : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2de8678137ca45cc882e9eb5bd26d396 : SUCCESS in 4m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2b6d14234e2948fcb4a50e77d8dd2200 : SUCCESS in 29m 35s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7288babd0b7645e38c7120f0fcbf3aa9 : SUCCESS in 34m 00s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e20595d531f648ca98bafde92b5c649b : SUCCESS in 1h 11m 16s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6b86f5dc74d74719a8431de7a819de2c : SUCCESS in 1h 38m 02s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/12cdbb54f3f84b8a9862749f2251cbfa : SUCCESS in 25m 58s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b83de1288cf846cea92115118db7149b : SUCCESS in 54m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ae274b932ba84ca3883df98a6d534e37 : SUCCESS in 1h 01m 35s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2295c1dff9074bc5aab4bb959fcb56e9 : SUCCESS in 1h 22m 35s","accounts_in_message":[],"_revision_number":2},{"id":"4f2bba0ad9f055939660edb9f84ac969d31b7b95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 12:51:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b9ce676046952a7e98cc9a5eb5605430e6dd9c74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 13:44:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/242fb5738ce24543b4f27edd64181d18\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7bd0c24edeb14edc87150e1c99893a4e : SUCCESS in 2h 38m 53s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8c6a724f1d9c4165b8e59b601374d844 : SUCCESS in 30m 48s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2e9a594c48814d78b921624948cb071a : SUCCESS in 2h 10m 23s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1546a6bd73fe48bba84cd01795f58bc8 : SUCCESS in 3m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e5a73af3fdad4811a005c313532ff5c3 : SUCCESS in 4m 11s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a69aec3e5c444d01855f6ffab8b8fd67 : SUCCESS in 32m 29s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3b02ba765a41424f97d02e3e5f139cc0 : SUCCESS in 29m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f565a6bed1b1d5fab53421881ac022b8619f3a5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 14:39:21.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b26321343144cdca4172165acc2811b\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/07fbf09e099849698860e27fa86bc928 : SUCCESS in 20m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d2370c5f3e604cddb4d13c41c69205db : SUCCESS in 27m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e905480d616343a08f35b091e2ffa4f4 : SUCCESS in 51m 32s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ec82965532c74a7dbc01512c73fdc597 : SUCCESS in 1h 24m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e244b81054fd485ab42e4ba1309d6604 : SUCCESS in 2m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ddb29470622f47459538491d77d3cfd2 : SUCCESS in 2m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2c737725c052464b844f339042f74b63 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a470b7b8507b46ad80892f9fc5ac7588 : SUCCESS in 4m 15s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a1b2946e01304087a8eadf05e90fbf1c : SUCCESS in 27m 58s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/36ef626e6113491d8d93a54644f1c1b6 : SUCCESS in 32m 49s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ff9eb1c9be134e1cbcdb43f21dff2e60 : SUCCESS in 1h 08m 20s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1330d5913f584f11822c2d2dec7c3f6a : SUCCESS in 1h 41m 16s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b248c78341a640e78293ec07a941fb97 : SUCCESS in 21m 28s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/75a9a104a5cb4ab3affbc3ce2601875e : SUCCESS in 50m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7878f62a59394b75ab36fcbe0622a40c : SUCCESS in 54m 24s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/44a9d2073a264ddf98cd9d3e5d9aaa17 : SUCCESS in 1h 24m 50s","accounts_in_message":[],"_revision_number":2},{"id":"7367d52d0dd27f1993f2d7e0580da21f12c051ea","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 14:39:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"c311892e93eedd7771da175d02ed12e346a275a1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 14:40:35.000000000","message":"Patch Set 2:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/134f5730c3eb407db3fd6fd90aec2c01\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58e3cf4a89fc4564a62907457b97aaf3 : FAILURE in 56s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8b98145f23676121002ea654d2806b50e402edec","revisions":{"c8e1c31861499606ec9f9d68cfbdc16f9715201a":{"kind":"REWORK","_number":1,"created":"2024-06-16 16:49:31.000000000","uploader":{"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},"ref":"refs/changes/66/922066/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/922066/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/922066/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/922066/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/922066/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/922066/1"}}},"commit":{"parents":[{"commit":"428f37124986019bedc897469f9b1f3470b62082","subject":"CI: Pin requests to \u003c2.32 for publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428f37124986019bedc897469f9b1f3470b62082"}]}],"author":{"name":"Tadas Sutkaitis","email":"tadasas@gmail.com","date":"2024-06-16 16:49:16.000000000","tz":180},"committer":{"name":"Tadas Sutkaitis","email":"tadasas@gmail.com","date":"2024-06-16 16:49:16.000000000","tz":180},"subject":"fix: apple silicon architecture selection","message":"fix: apple silicon architecture selection\n\nChange-Id: I9fdccbe3bde07a7f9a1e6ac82b1096a024e1b0ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8e1c31861499606ec9f9d68cfbdc16f9715201a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8e1c31861499606ec9f9d68cfbdc16f9715201a"}]},"branch":"refs/heads/master"},"8b98145f23676121002ea654d2806b50e402edec":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-07-16 11:02:52.000000000","uploader":{"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},"ref":"refs/changes/66/922066/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/66/922066/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/922066/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/922066/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/66/922066/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/66/922066/2"}}},"commit":{"parents":[{"commit":"428f37124986019bedc897469f9b1f3470b62082","subject":"CI: Pin requests to \u003c2.32 for publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428f37124986019bedc897469f9b1f3470b62082"}]}],"author":{"name":"Tadas Sutkaitis","email":"tadasas@gmail.com","date":"2024-06-16 16:49:16.000000000","tz":180},"committer":{"name":"Tadas Sutkaitis","email":"tadasas@gmail.com","date":"2024-07-16 11:02:52.000000000","tz":0},"subject":"fix: apple silicon architecture selection","message":"fix: apple silicon architecture selection\n\nCloses-Bug: #2073256\n\nChange-Id: I9fdccbe3bde07a7f9a1e6ac82b1096a024e1b0ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b98145f23676121002ea654d2806b50e402edec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8b98145f23676121002ea654d2806b50e402edec"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d2"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
