)]}'
{"id":"openstack%2Fkolla~922148","triplet_id":"openstack%2Fkolla~master~Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-10-02 08:57:54.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-10-02 11:07:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65","subject":"CI: Move image building to a role","status":"MERGED","created":"2024-06-17 19:49:09.000000000","updated":"2024-10-02 11:10:39.000000000","submitted":"2024-10-02 11:07:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922148","meta_rev_id":"214520723a1e1e69e7860070faa1e009cdde5a40","_number":922148,"virtual_id_number":922148,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-02 11:07:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-10-02 08:07:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-10-02 08:10:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-10-02 08:05:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-10-02 08:07:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2024-10-02 08:07:43.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-17 20:57:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-17 21:33:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-02 08:05:35.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-10-02 08:10:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"aa5a0ca4eba7acf09d477ca1823ddeadcecc65f5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-17 19:49:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4eb582c768896750a060ce73ecb8ff69c63e271","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-17 19:51:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"12e20f96f0080b26b16282f7bf03c1b68ea42c17","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 20:57:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4470aaea8ce349a0b872f11594d56bbb\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2fa554cc3b524327a354797eb928e638 : FAILURE in 3m 17s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4e46c4f320eb402a9643ee99bf1836fa : FAILURE in 2m 30s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5c2b9956c1374911ad6e3e4bae044c68 : SUCCESS in 1h 04m 16s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a9d91a9e064b40eca57634d2294d5656 : SUCCESS in 2m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/407276955b6e40a1b0ccf1f9a7cc3170 : SUCCESS in 3m 12s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1022426417974ea18d509e0c04b453ca : FAILURE in 7m 27s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/766686f7f68245968f8c159e61171b03 : FAILURE in 4m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"13f6b1b6d0f503e7b596523060e36b6e39388207","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 21:33:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/404a74cb518a4c96802f15592c8e7151\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1ddacf4382824c3abe2d0659bec2e2fa : FAILURE in 4m 27s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/02aaef9847944e73bb43670d700dc3a7 : SUCCESS in 1h 07m 00s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/80219460a6434963b80696166c17049c : FAILURE in 2m 42s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/027e170212814a08acb43268530617f4 : FAILURE in 3m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/05c5b67d0c134fa69fcb99efffae08a5 : SUCCESS in 1h 04m 28s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/75b36ccf79df491d875385901f422e6f : SUCCESS in 1h 22m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1de4c6ced684a499ba6e3639d935e10 : FAILURE in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8748610d3344db48266b01bdbde3bb1 : SUCCESS in 2m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/189739a780634364a58f9f8e27f27bd8 : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b1b5e7a1091045809300d662a07e5147 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3c6da1610b94e67b6249116c2ecebd2 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e758df1a5fdf42888ad404b864bbdce1 : SUCCESS in 4m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3954c0d4e819468c8050138745a6f8e2 : FAILURE in 5m 26s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7a701b81969644dfbf8ac1962d33589e : FAILURE in 6m 43s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a4bc12f2d96245ddaaca94fdaf26e934 : SUCCESS in 1h 06m 06s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/fe6f8cd958074eb3994722db3e73d6cc : SUCCESS in 1h 36m 33s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9191a1cad1784b6e8384651821ac4018 : FAILURE in 3m 16s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/aa36cdcb28ed4ff6b7074522d7bb02c8 : FAILURE in 2m 27s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ec3a57a8ced74be99335daf737151885 : SUCCESS in 50m 39s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/df0b7082b462495187bad43a9bd97650 : SUCCESS in 1h 28m 14s","accounts_in_message":[],"_revision_number":2},{"id":"d70be0d7c1d7b7477ab08d1c322fab863e60ea24","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-18 05:57:53.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"90e5760b987d3786dd6e3fe0c4cb97acbbd0419b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-18 06:01:05.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"8271f36a040dc1b97706fd7a5ba25b9c662a3e9c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-18 06:04:00.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"19242e4811c4fb7170c4da03c3dfbff35aa2068f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-18 06:07:13.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"4fe555ffcecf8d3dad7d6414d91a633a5770b6cc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-18 07:34:42.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"5f2fce62561d56ae80fd4b5be954077993137188","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-18 07:41:04.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"8c543fcab225921e3a7a4d865fcf16ce87a6032b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-18 07:48:03.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"33263134fd070437d1d0b6b0290d1f2d0d2acdaf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-18 07:55:30.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"12558ac5e0b1dbd8f388e60849fbc6e15e679f62","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-18 08:51:13.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"f51159b70caa369ac51dbd487c40ea998f7c3a30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 09:54:44.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0651aab9fd34fb790b1bd5e6b82bbef\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3df78b6474cf4b1caf0834a288bfe1e0 : FAILURE in 6m 09s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/946fb40cd7f544679bfb335de2ff096b : FAILURE in 4m 12s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d19fe97965234281a50416c5ba18b752 : SUCCESS in 1h 02m 17s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/22e76762fdf34522ac22ace042bdebe1 : SUCCESS in 3m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/26d0853e81ac48f4b3cf4a26b6dc35a1 : SUCCESS in 3m 23s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/78e0e2f8e8884b2a9f5d45727b08c686 : FAILURE in 8m 12s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/77650153f7da4c559be5d0fd864eab8b : FAILURE in 5m 04s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e883571197ea48f87739a1c85d51c527f448e99f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 11:13:42.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1dac53912a544bd5bf9968513244b890\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/91fc4219a1de461fbaf4d6dbcc0bf0bf : FAILURE in 3m 50s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/33a2c79d721e4073885368eab56d9d0d : SUCCESS in 1h 02m 00s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/42b05fc561594d9685ba1bdf866932aa : FAILURE in 3m 12s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/29cd8669e742484785c868cc7269ce74 : FAILURE in 3m 08s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1c4457435c374abf8e15dba7778d2e85 : SUCCESS in 58m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ab3075ff70b24ab6a7edbc8cdcc9aa5f : SUCCESS in 1h 41m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55050bd69ef34e8986a2c86cdd835cc8 : FAILURE in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b2d6f69fbe14feb9cf1d4318566909c : SUCCESS in 3m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8c1150efbf034b8199fd5a07a0b28496 : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c09117d7a0c44eceb157eee96bb65200 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4263c7e14af14644a073424964a4f4c5 : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab09557956594a1ab711e068478cad19 : SUCCESS in 4m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b4c15d189f0b4b089a3cffcfb5d5a4da : FAILURE in 8m 00s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/8153f9de4dd64ace88101a4b963e05d5 : FAILURE in 6m 34s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/94e395da93f24e71b24d1a32e0dc264d : SUCCESS in 1h 17m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3d5185851b504de4b639578869dd035d : FAILURE in 6m 49s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c288604292c74a70be899731334e68ea : FAILURE in 2m 53s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0e8eab0572f94b41927bc63bffa6c40a : FAILURE in 2m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/587f67edef7c40bda6c748d2dc7d384c : SUCCESS in 49m 26s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/62f83cc8a6c74ec7abaf1725385c9885 : FAILURE in 5m 18s","accounts_in_message":[],"_revision_number":11},{"id":"8ba218d8fff420e594707f4ae6a59574b24de809","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-18 13:24:34.000000000","message":"Patch Set 12: Published edit on patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"60716c8566887037af02a489cea93df392acf38c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-18 13:35:42.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"448f8bd9a2a5e21b7dae6822e1c17f5f0d61976b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-18 13:37:04.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"a6bafb95b48f4e17ab63ccaac6bc795d8cb2c233","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-18 13:38:01.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"3daf3f10c8434162edbac13a161ac3212e3dd23a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 15:18:58.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f170c2f535334d73823ea6fad7dfd09d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d69a39ac02f44079bd301b714dfff2e3 : FAILURE in 7m 29s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ceb0b00337834b6a979536405d2c578e : RETRY_LIMIT in 3m 01s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8539c1f9d2054badb91848d16985ec25 : SUCCESS in 1h 38m 52s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d50805633470465b8b7601cd42b3a2ee : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cbd65e903ae14add977ceb6be0ac573c : SUCCESS in 3m 07s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a4dcaf197f98457bb12c15f9bef226b7 : FAILURE in 12m 39s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e72c64698d1b43c387df0be2753e8d7f : FAILURE in 3m 52s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"8eb1a6edb8606f71712211bf195744a348ae3f4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 15:31:10.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/98f8fb921636459b86bb8b2023fb097a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a16aa9c655f045b6a4939a8bb90b3024 : FAILURE in 4m 36s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1c3acba6be7840028b5d69280dc2ba55 : SUCCESS in 49m 12s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/155a289deebe4ce7adf551bbf0c18c35 : RETRY_LIMIT in 3m 33s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e15d919b7a9c4d448ffaae41e9e7d46b : FAILURE in 2m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/13b54583f1cd4c1c8eba40dd61117c67 : SUCCESS in 1h 05m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0766439348df4166a2c799b4123ad781 : SUCCESS in 1h 39m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e285a1ef21f4c188601787e75bed8e0 : FAILURE in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c1df4f90609545d59b05416bd052931b : SUCCESS in 3m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/58736b601abc4df89850353601a9b0ff : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/edf4cdfac5884119ab0f37375b9ddaf1 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8dae25957244b708136afa93b464cdf : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6411ceac27274965a45b4d1cbe4732fd : SUCCESS in 3m 42s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4531f2738a6440babac35d2ac727b725 : FAILURE in 5m 46s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/62ab2f74b68f4f028b6b228deb0689d5 : FAILURE in 7m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/216cb56641454458b421a147e26b0494 : SUCCESS in 1h 00m 55s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0207c56b27c14a019b366913946fcd0b : SUCCESS in 1h 46m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2e5bf26bf52842f9b897d36952df3092 : FAILURE in 2m 36s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/37de7675079141f29e54a73d6fc683fe : FAILURE in 2m 26s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/674e300bd411436fbe1a1284f9e7e885 : SUCCESS in 1h 01m 49s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2cedfd52f6f3421f9affc4748c4aee66 : SUCCESS in 1h 16m 00s","accounts_in_message":[],"_revision_number":15},{"id":"e52adbf24c24926da61ef9ddc3b8af5e2c0d10f9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-06-18 15:32:22.000000000","message":"Patch Set 16: Published edit on patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"a53ba4117df7780c16563df6ba793f2b60c703db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 16:31:47.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27e275d8db4b4bfa9b7bdda06dbe89a5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/03478b29b1d44824a1b2a5c70788b0f8 : FAILURE in 3m 14s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e0a22e998e1046adaf4ef8227525caef : RETRY_LIMIT in 1m 49s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d066c3f04d7242e385e55795eaf462cf : SUCCESS in 58m 10s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c77a4d57c0bd49bc88c2f3bbc62e00b0 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e726d834bca04a319a2757f14deaf74f : SUCCESS in 2m 45s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/be03176a4c864bca9d4d6bba0c47b21c : FAILURE in 10m 28s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4d618b95abbb4d939054ee4ea7507f3c : FAILURE in 2m 37s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"bc895cb5b1200360e89cc631d767b728cb3b484f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 17:08:05.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9d8bd9471a744cbb2670f073df90560\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/175328fa2b254a1183cc19485967ef18 : FAILURE in 3m 44s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/888ee64ef6aa4beeb2ef82521facdbbb : SUCCESS in 1h 14m 23s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/89a8a5d3112e4832855b42b6a52faae9 : RETRY_LIMIT in 2m 46s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/205bd790dde840eeaf0674a6bcb5c676 : FAILURE in 3m 38s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9d0f9c67c9eb4e75b76e4479e0e2d7d8 : SUCCESS in 57m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7779095dacf04acd879d6bb53de683a7 : SUCCESS in 1h 15m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea6417088c5f4165af65a360d8796733 : FAILURE in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd33e32f492347bf9d7805cdd53ac6ce : SUCCESS in 2m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d307c4e776e44c4b9b929728d933aa40 : SUCCESS in 2m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9d17f5201923442a8988536a804bbcdd : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9bb566affc948ea88c15956b5ab3120 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f463f78952444b5b85d2e1f216320063 : SUCCESS in 5m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/17433c50ae4a4756a7a0351948ba1844 : FAILURE in 6m 39s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d8ec3bdb75d14be79b615a9016310668 : FAILURE in 8m 40s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a37f2aa079fd49f69cb6a07d77291903 : SUCCESS in 1h 09m 46s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5a4fe4a4ec984688b1301b9735392fa7 : SUCCESS in 1h 30m 41s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/09446c26309543d4abc1c3a7b0d434f1 : FAILURE in 3m 16s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c67cfca702e9473c8f5a75be4d39b559 : FAILURE in 3m 01s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0cbb4fceb3fa45379651bb85be9fdc09 : SUCCESS in 44m 58s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b9a2a02eaf4f4cc38c41248f89059fdb : SUCCESS in 1h 12m 09s","accounts_in_message":[],"_revision_number":16},{"id":"7c9903bb50c577325471e930ac4c8ade9c193374","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-27 20:23:50.000000000","message":"Patch Set 17: Patch Set 16 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"342eba249613d8b3629e5e0915dc128c83c8b489","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 21:54:04.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/686f51d82c34492cb56a146d0b335699\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2fd65235512749d6a45645cb1099fdec : FAILURE in 3m 34s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/92c62318cfc549a6b269349f319c5256 : SUCCESS in 1h 03m 40s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1b551d782041488d94ad45d071c20f57 : RETRY_LIMIT in 1m 31s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/53e224d0a20a421f80e1c3a05ef4d6a3 : FAILURE in 2m 39s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e11282c219994b5e937f459ea0fe082e : SUCCESS in 55m 45s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6a6317e966be40e686224d2835961939 : SUCCESS in 1h 11m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60e36f89db3c4188ab2a1c7356e06edc : FAILURE in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa0f4157e1b44bd6b5af9a9a811cb28c : SUCCESS in 2m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8bc5bd7ce1014f3cb270d865ece9a855 : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b992259abd3e4ca1bdb29400f0200646 : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4439e03292fb4220b26d38c8e66f7395 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d646427d2e0347d2b6bed268b8796618 : SUCCESS in 4m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0d15e519b23c4ffab4c36a7d8f52b42d : FAILURE in 8m 30s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/4181f80177e840e0af160ff3c4c20727 : FAILURE in 6m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a5d92fa3d0f042cdb6ee7ae84750a015 : SUCCESS in 57m 38s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c3c1c861d5ce45ae845976914c368549 : SUCCESS in 1h 28m 28s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/58cda9ddc7d344109d59e043566be290 : FAILURE in 2m 38s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6044accf721a4f2b83cc24037ae71529 : FAILURE in 2m 41s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4e9dfb3c66664b02a3f5fabe28970065 : SUCCESS in 46m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4a2285575eb24fe6aa7a324de7e12d81 : SUCCESS in 1h 05m 52s","accounts_in_message":[],"_revision_number":17},{"id":"5268036930740a39066344e07b827e68c67707e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 01:06:52.000000000","message":"Patch Set 17:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0052b67a115f4d94b13ab07a62b2819a\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c696e1e293ca47b08f4cf902c7d10ab7 : RETRY_LIMIT in 3m 01s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a78a9ee506164a6f88961d4f7be63313 : FAILURE in 21m 05s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d868ebe5fcb4859905fb56b962c87cb : SUCCESS in 7m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d60faebcaca04adfb31182a8f80eae34 : SUCCESS in 6m 20s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"abe75cc6948d8f9cab4009d5d25039456bf81a70","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 05:58:49.000000000","message":"Patch Set 18: Published edit on patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"6feec8127ed044f34d8c6c33ed589b0038b9902c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 05:59:16.000000000","message":"Patch Set 19: Published edit on patch set 18.","accounts_in_message":[],"_revision_number":19},{"id":"8114f09cbeff53834364806e9476b4ca81474524","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 06:05:05.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"beef0ff7f559ee596121a2fd2dbd2d61c4a6d7c6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 06:06:29.000000000","message":"Patch Set 21: Published edit on patch set 20.","accounts_in_message":[],"_revision_number":21},{"id":"13f85d2c3f9b4ae2c874bbf2c69b0e443a13282a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 06:07:08.000000000","message":"Patch Set 22: Published edit on patch set 21.","accounts_in_message":[],"_revision_number":22},{"id":"f77f01e10cda9cd0a8b5cade8c1bd43f1186a99d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 06:07:19.000000000","message":"Patch Set 23: Published edit on patch set 22.","accounts_in_message":[],"_revision_number":23},{"id":"5a597bd98a63bfd5a787e229966fb336e7e6e1d2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 06:11:31.000000000","message":"Patch Set 24: Published edit on patch set 23.","accounts_in_message":[],"_revision_number":24},{"id":"bccdc9ae76a31f0a9fdde35c6bbb51ea889941be","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 06:11:35.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":24},{"id":"575d8fd1a45475c505e43e13c65b98139994ed35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 06:11:56.000000000","message":"Patch Set 25: Published edit on patch set 24.","accounts_in_message":[],"_revision_number":25},{"id":"ceb4e9dad54805adc3baaf1422d123ffc22fd5d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 06:23:57.000000000","message":"Patch Set 26: Published edit on patch set 25.","accounts_in_message":[],"_revision_number":26},{"id":"139653ad08ed6f5b7884c3d54d4b24e884fa6077","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 06:30:35.000000000","message":"Patch Set 27: Published edit on patch set 26.","accounts_in_message":[],"_revision_number":27},{"id":"c9e736b2e4d825ef0b19df2a644bc09595254e91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 06:32:14.000000000","message":"Patch Set 28: Published edit on patch set 27.","accounts_in_message":[],"_revision_number":28},{"id":"5ed93fcb28a0c60d61a6a9e7afbfbada2f0f4f87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 06:38:26.000000000","message":"Patch Set 29: Published edit on patch set 28.","accounts_in_message":[],"_revision_number":29},{"id":"bd6e34f90ef55560bde192f423686affdd663d43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 06:47:10.000000000","message":"Patch Set 30: Published edit on patch set 29.","accounts_in_message":[],"_revision_number":30},{"id":"b541aaca9833fb816c044badb90df0a6b01932c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 06:51:51.000000000","message":"Patch Set 31: Published edit on patch set 30.","accounts_in_message":[],"_revision_number":31},{"id":"8de0e3b9d37d207930c37cfc0d35413be2d68883","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 06:52:19.000000000","message":"Patch Set 32: Published edit on patch set 31.","accounts_in_message":[],"_revision_number":32},{"id":"4e734251126e6bb03a04e855ae35fe7b439ef1af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 06:55:09.000000000","message":"Patch Set 33: Published edit on patch set 32.","accounts_in_message":[],"_revision_number":33},{"id":"7005af9f99ad6b1d9a00798cbda1ab79067abd0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 07:05:16.000000000","message":"Patch Set 34: Published edit on patch set 33.","accounts_in_message":[],"_revision_number":34},{"id":"68d4bd281903b46cc1badf7cf4f0a76468f631f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 07:16:24.000000000","message":"Patch Set 35: Published edit on patch set 34.","accounts_in_message":[],"_revision_number":35},{"id":"5414400321424941e706da7f3f153024cc552f22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 07:24:59.000000000","message":"Patch Set 36: Published edit on patch set 35.","accounts_in_message":[],"_revision_number":36},{"id":"c8dc5e9330deaed79bbcf81b5880d27c1b05a5c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 07:40:34.000000000","message":"Patch Set 37: Published edit on patch set 36.","accounts_in_message":[],"_revision_number":37},{"id":"499df79a015b0f0f098c238c88ec1f36b31c1755","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 07:54:11.000000000","message":"Patch Set 38: Published edit on patch set 37.","accounts_in_message":[],"_revision_number":38},{"id":"1275d1bdc97835e2604a80225d48c0d088100000","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 09:50:19.000000000","message":"Patch Set 38: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/969e4b9ba89d43b38188f894e52da269\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d4cc42d730d743c6ad428d1096235e34 : FAILURE in 4m 39s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d0b654a480904803a46d4688982c87a8 : SUCCESS in 57m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/38e0683ccac24c4ab89bc26edad75e3b : FAILURE in 3m 30s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1ecead4f41984aa2841941210fe169f2 : FAILURE in 3m 27s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6adfb99214cb4313ae225fd5c4aeedb8 : SUCCESS in 1h 02m 33s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/40ce2c3dd2ff4b449057bfbe4f6f904b : SUCCESS in 1h 37m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d35a7aefe6a448384e9396acb9d2970 : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e3f6f5f45e240b98912bcd44f76d9c0 : SUCCESS in 2m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e0c29e1e46b43aeb46c3a55b379dc03 : SUCCESS in 2m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1ee1fdb48f1f4d138cd75dfa3478cff2 : SUCCESS in 3m 17s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db71c8ff2893481493ea26cb0b1231d0 : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21070658482247d6a764287ce7c46013 : SUCCESS in 4m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ec880fa8d1e942e6b68f0758bb7c90b2 : FAILURE in 7m 58s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ec9ace16e43a4010aeb24d60c485bcc8 : FAILURE in 6m 45s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1389822f110f4b99b90edb02405c38ff : SUCCESS in 1h 07m 14s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/30d7045effdf4407aeda074c9b0eb7fb : SUCCESS in 1h 35m 11s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ea92e77afe3d408c9157b9bb618badfc : FAILURE in 3m 12s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ecb2484323424f66a5ab1cf5ab51a8bd : FAILURE in 2m 52s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/bdac430420094a6b93d86a4ddfd65c52 : SUCCESS in 55m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d6bbbbc9a3304a9db85194a2b5b6fb3a : SUCCESS in 1h 21m 08s","accounts_in_message":[],"_revision_number":38},{"id":"e645d954d7ee3ae59058e960f7540f22aee8210e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 10:27:32.000000000","message":"Patch Set 38:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ccf1249d40e43899e77d15d35da5674\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cfdda2bdf0d54c108b9ab2eb73e19fc9 : FAILURE in 3m 46s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fd38be46da7a4f18a9ee42f56ee0accc : SUCCESS in 1h 33m 02s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bc159cf7c6754bc99fe6ebceee40b595 : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/47035588c9154f0aafc185fdb72dcfdf : SUCCESS in 3m 25s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"95b153b5399bedbb606e47804cf57f681591617f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 10:29:45.000000000","message":"Patch Set 39: Published edit on patch set 38.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":39},{"id":"a4fbf53855ceee8b2b74feef90e25951b141d5e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 11:13:33.000000000","message":"Patch Set 40: Published edit on patch set 39.","accounts_in_message":[],"_revision_number":40},{"id":"33ba6233e5dcb44f4eace0986a1e47cadef8eeeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 11:28:13.000000000","message":"Patch Set 41: Published edit on patch set 40.","accounts_in_message":[],"_revision_number":41},{"id":"474a3a55b5a375cd6a118ff9c000c0bfb33d1f1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 13:05:00.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cd40c094a5a4bc8a681a18ea907f6c5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c6fb72d4b58047219ca6916e13413dbb : SUCCESS in 32m 05s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b860f475a6134d9db31ac3d24274300b : SUCCESS in 1h 14m 29s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0d3ca2e9157a4da3bb8e7746e2754ee3 : SUCCESS in 20m 27s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/83233f3473154b10b5ba5ec1d48c2bc1 : SUCCESS in 31m 35s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3e6fffb8fea949948e811a0c1069c445 : SUCCESS in 1h 00m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/22091e2ac99d4d3d9a01ac8ee2de20b1 : SUCCESS in 1h 23m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89c153b690c4427fb9a8d89c77c671a7 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3403119569de404e8070296e15443987 : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0ce76661fab3412ba88d1783295704d7 : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6aec204d4a174644ae5b3dd7a18f0eeb : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3ec422bb052f4a0699faeb0d4522d892 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d36388c0c90447e98ddb6a9ec928df6d : SUCCESS in 13m 36s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b44d2f7399ee4796bf1ec5a5b5506815 : SUCCESS in 24m 36s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/799dd420da6e4aefbf7263333cfa4d2e : SUCCESS in 35m 14s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c262047642be4870a3d191f8733ff1b9 : SUCCESS in 1h 06m 03s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e3c684e1ddf4498c958a4606e7e2ce67 : SUCCESS in 1h 31m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5469a18d39934a57bb54b997950b91b7 : SUCCESS in 18m 10s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7c31e3b61ae64362b9af658817adc0f1 : SUCCESS in 52m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/bbc136910c904ea2b35a4cda772e1b87 : SUCCESS in 50m 36s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3e0b6c1d4aaf45a7962a25c7c7b49eb8 : SUCCESS in 1h 13m 36s","accounts_in_message":[],"_revision_number":41},{"id":"82b653f2127ad3356295a8651ecfcccba1f3e042","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 14:17:36.000000000","message":"Patch Set 41:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/211bd97ca09c429b92ee8a9ac708d249\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/aad7d8d53f9b4b409a5256d5e6557080 : SUCCESS in 2h 23m 54s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3c370f081d074c6ca96561e72173f30d : FAILURE in 1h 43m 30s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4df6815827cc42878c646b8a7dd0fc14 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4bb1e368c9044fe485b03ce8df2a6e3e : SUCCESS in 4m 28s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"33510f5284948a21fe8a2c07c81ad89df649b3a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-02 07:37:29.000000000","message":"Patch Set 42: Patch Set 41 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":42},{"id":"0765e2f670a4d7d2fe8d7b990578ffeaef59b617","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-10-02 08:05:35.000000000","message":"Patch Set 42: Code-Review+2","accounts_in_message":[],"_revision_number":42},{"id":"f484b6680baf29372edfe4f756e40447ee9c7c9b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-02 08:07:37.000000000","message":"Patch Set 42: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"6fe5582a96a351d01ad9bb6b74818a2cff4e7814","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-02 08:07:43.000000000","message":"Patch Set 42: Backport-Candidate-2","accounts_in_message":[],"_revision_number":42},{"id":"bc113b2308fb5ce746e7f95ab39d17f7e3614b58","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-02 08:10:12.000000000","message":"Patch Set 42: Code-Review+1","accounts_in_message":[],"_revision_number":42},{"id":"e681203dd47fbe1d1f3a5dbe5519c5e721ddb6aa","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-02 08:11:03.000000000","message":"Patch Set 42:\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"0b3cbf7f9bfd51a73039ba63d8499679cc3a3d17","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-02 08:57:54.000000000","message":"Patch Set 42:\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"82d008acd4661e8a13575696098870672e071316","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 09:21:24.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02f5ec2061144adeac20059e87b08fd7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d86bd6de63ce4fd984fe3a6e3c481b19 : SUCCESS in 27m 02s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f512b205366843e685b6b369ccc3d5d2 : FAILURE in 5m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/41a455efd8d54095bbfa3e5ae53bbc33 : SUCCESS in 25m 12s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3b1694c1dcb44f70b766e120c5edef1d : SUCCESS in 33m 11s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6295a53a94c44d41a716a3caf4be6601 : SUCCESS in 58m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/444ed4c85e044b2d94be0db7b8a81243 : SUCCESS in 1h 35m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b01fe662ea174edd8816a4a7915fcbea : SUCCESS in 2m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82d341816ddc41b1bcad49e4c70ef457 : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3c6f2c84038b40e2b818cf1b34043cde : SUCCESS in 5m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9532216c70b945df961c30f017ad700c : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b4627ce13a94026a783f058a4f1a00a : SUCCESS in 5m 01s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/440e5aca2732480896858b31b9144e2e : SUCCESS in 32m 20s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/01be0f0110314db4a878eaf60a59edc2 : SUCCESS in 43m 07s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f4659b07554a4923ab5462c0a3ed9c3a : SUCCESS in 1h 05m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9b6a3e08e05844a588b7fcacfd3b53c4 : SUCCESS in 1h 31m 53s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f43ad309dad14a4db1c3b3e8567b83c6 : SUCCESS in 21m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/88b4829f45fe424a9c76e78055b12bd3 : SUCCESS in 27m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4d77bd66fe924f3596f714cca3d6065d : SUCCESS in 54m 07s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c47ab580be5d4466ad66252222aedd2e : SUCCESS in 1h 24m 04s","accounts_in_message":[],"_revision_number":42},{"id":"8f4501ccbcb8dc821da7410ebd8f01e8da21c8d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 09:21:52.000000000","message":"Patch Set 42: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":42},{"id":"de9b1a26dee925fbf29c0dcf89636f3b16f53c25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 09:38:39.000000000","message":"Patch Set 42:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86c49afff6e24914b59d2e985e3843f7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/cbb1762686f546dbaa489f646fc14f8d : SUCCESS in 30m 17s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7420f9b8240142a09117676c28863d2f : SUCCESS in 22m 01s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b804d6e7f9a247a090de4303576a2c90 : SUCCESS in 1h 59m 01s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c5d55d1d71274df782d6383051b1c838 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0c6303fe005141c0be16492d62bdfb37 : SUCCESS in 5m 04s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1b48c708ff46446dae9bf05986566593 : SUCCESS in 24m 14s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0f5bcc14500e41ebb3742ea2ee779283 : SUCCESS in 22m 33s (non-voting)","accounts_in_message":[],"_revision_number":42},{"id":"c27d175297c4976d902e0594d15c6916a18ccd68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 11:07:45.000000000","message":"Patch Set 42: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e948e238329401c80227754610908a5\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f07d41459d944bbcbe91f55e4c9d7056 : SUCCESS in 15m 59s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6b819b1f5cc44b99a843cec425419013 : SUCCESS in 31m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/93ebe35498a54eaaaedd4702c09d793c : SUCCESS in 56m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2b6c618d7d2742b4a685d87d304bbbeb : SUCCESS in 1h 36m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eac35b766d134339826a02aa281eb16a : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11a85ebf801340d7a7b58a611fe79bcd : SUCCESS in 9m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c2325a0dfd1c48c5814e9303f6b9b81d : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/843720c4938245f1b1853c1843883560 : SUCCESS in 4m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b8eadc1068f24dde91fd8e4cfcf30a1e : SUCCESS in 27m 06s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ad54f72ab34145edb9013c59188d662a : SUCCESS in 34m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b7536555401648199077c06a1b43ba78 : SUCCESS in 1h 11m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/197b63eb51ee411599445ecb9365fc46 : SUCCESS in 1h 33m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/56e1e587ee6d4004a99b7ccfe9e61bbd : SUCCESS in 23m 31s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/bb2769ebd2004ed588e4fa61743f5e24 : SUCCESS in 27m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3ac4ac4a92d14cae9aa10dcc563f1ac3 : SUCCESS in 55m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8f55818425d2420aa1216c4de2f83bec : SUCCESS in 1h 32m 45s","accounts_in_message":[],"_revision_number":42},{"id":"54ec7beaa579beb5a75777c82e3788012ea81708","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 11:07:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":42},{"id":"214520723a1e1e69e7860070faa1e009cdde5a40","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 11:10:39.000000000","message":"Patch Set 42:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b5439d2631f47fcac4b83e5407ac8a6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23a4c670a34846779bb653bea73a33a7 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":42}],"current_revision_number":42,"current_revision":"20cdc32537275ce4845eae62ce9a837868a0f800","revisions":{"4a968b532b2970a4e57782860a614d26afb46f02":{"kind":"REWORK","_number":1,"created":"2024-06-17 19:49:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/1"}}},"commit":{"parents":[{"commit":"428f37124986019bedc897469f9b1f3470b62082","subject":"CI: Pin requests to \u003c2.32 for publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428f37124986019bedc897469f9b1f3470b62082"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:49:06.000000000","tz":120},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a968b532b2970a4e57782860a614d26afb46f02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4a968b532b2970a4e57782860a614d26afb46f02"}]},"branch":"refs/heads/master"},"09ebdde45050e227a0659594a3609e3fe293eea7":{"kind":"REWORK","_number":2,"created":"2024-06-17 19:51:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/2"}}},"commit":{"parents":[{"commit":"428f37124986019bedc897469f9b1f3470b62082","subject":"CI: Pin requests to \u003c2.32 for publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428f37124986019bedc897469f9b1f3470b62082"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:51:36.000000000","tz":120},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09ebdde45050e227a0659594a3609e3fe293eea7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09ebdde45050e227a0659594a3609e3fe293eea7"}]},"branch":"refs/heads/master"},"4f0ca3a1c950982e48219c9af81bccdd4504954c":{"kind":"REWORK","_number":3,"created":"2024-06-18 05:57:53.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/3"}}},"commit":{"parents":[{"commit":"428f37124986019bedc897469f9b1f3470b62082","subject":"CI: Pin requests to \u003c2.32 for publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428f37124986019bedc897469f9b1f3470b62082"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-18 05:57:53.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4f0ca3a1c950982e48219c9af81bccdd4504954c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4f0ca3a1c950982e48219c9af81bccdd4504954c"}]},"branch":"refs/heads/master"},"29604ad52ea8b6393959565155f3794a2f83fcdb":{"kind":"REWORK","_number":4,"created":"2024-06-18 06:01:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/4"}}},"commit":{"parents":[{"commit":"428f37124986019bedc897469f9b1f3470b62082","subject":"CI: Pin requests to \u003c2.32 for publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428f37124986019bedc897469f9b1f3470b62082"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-18 06:01:05.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29604ad52ea8b6393959565155f3794a2f83fcdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29604ad52ea8b6393959565155f3794a2f83fcdb"}]},"branch":"refs/heads/master"},"7cfa79848f3fc61280fed8988b31f6b2ef8d88a7":{"kind":"REWORK","_number":5,"created":"2024-06-18 06:04:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/5"}}},"commit":{"parents":[{"commit":"428f37124986019bedc897469f9b1f3470b62082","subject":"CI: Pin requests to \u003c2.32 for publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428f37124986019bedc897469f9b1f3470b62082"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-18 06:04:00.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cfa79848f3fc61280fed8988b31f6b2ef8d88a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cfa79848f3fc61280fed8988b31f6b2ef8d88a7"}]},"branch":"refs/heads/master"},"8af468162b7180f4ec941ff76c80631c25ce3aa8":{"kind":"REWORK","_number":6,"created":"2024-06-18 06:07:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/6"}}},"commit":{"parents":[{"commit":"428f37124986019bedc897469f9b1f3470b62082","subject":"CI: Pin requests to \u003c2.32 for publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428f37124986019bedc897469f9b1f3470b62082"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-18 06:07:13.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8af468162b7180f4ec941ff76c80631c25ce3aa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8af468162b7180f4ec941ff76c80631c25ce3aa8"}]},"branch":"refs/heads/master"},"7cbf7b4dfa2d130942f551e5eddb28bc22c06e8b":{"kind":"REWORK","_number":7,"created":"2024-06-18 07:34:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/7"}}},"commit":{"parents":[{"commit":"428f37124986019bedc897469f9b1f3470b62082","subject":"CI: Pin requests to \u003c2.32 for publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428f37124986019bedc897469f9b1f3470b62082"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-18 07:34:42.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cbf7b4dfa2d130942f551e5eddb28bc22c06e8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cbf7b4dfa2d130942f551e5eddb28bc22c06e8b"}]},"branch":"refs/heads/master"},"bfa4f16441d47780a0c8e3c469fc59b7b9c8f78b":{"kind":"REWORK","_number":8,"created":"2024-06-18 07:41:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/8"}}},"commit":{"parents":[{"commit":"428f37124986019bedc897469f9b1f3470b62082","subject":"CI: Pin requests to \u003c2.32 for publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428f37124986019bedc897469f9b1f3470b62082"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-18 07:41:04.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bfa4f16441d47780a0c8e3c469fc59b7b9c8f78b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bfa4f16441d47780a0c8e3c469fc59b7b9c8f78b"}]},"branch":"refs/heads/master"},"abd8a6a4e8ef596daf7b71387fa2ae722800c175":{"kind":"REWORK","_number":9,"created":"2024-06-18 07:48:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/9"}}},"commit":{"parents":[{"commit":"428f37124986019bedc897469f9b1f3470b62082","subject":"CI: Pin requests to \u003c2.32 for publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428f37124986019bedc897469f9b1f3470b62082"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-18 07:48:02.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abd8a6a4e8ef596daf7b71387fa2ae722800c175"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/abd8a6a4e8ef596daf7b71387fa2ae722800c175"}]},"branch":"refs/heads/master"},"4b2dac2bc5c225f67606078531acc5e395427b29":{"kind":"REWORK","_number":10,"created":"2024-06-18 07:55:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/10"}}},"commit":{"parents":[{"commit":"428f37124986019bedc897469f9b1f3470b62082","subject":"CI: Pin requests to \u003c2.32 for publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428f37124986019bedc897469f9b1f3470b62082"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-18 07:55:30.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b2dac2bc5c225f67606078531acc5e395427b29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b2dac2bc5c225f67606078531acc5e395427b29"}]},"branch":"refs/heads/master"},"98224e05e6cac0b2d793e4c0e586e05afdb01488":{"kind":"REWORK","_number":11,"created":"2024-06-18 08:51:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/11"}}},"commit":{"parents":[{"commit":"428f37124986019bedc897469f9b1f3470b62082","subject":"CI: Pin requests to \u003c2.32 for publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428f37124986019bedc897469f9b1f3470b62082"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-18 08:51:13.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98224e05e6cac0b2d793e4c0e586e05afdb01488"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98224e05e6cac0b2d793e4c0e586e05afdb01488"}]},"branch":"refs/heads/master"},"127471017d8c52737389d5f5d459ca4184a5d9b5":{"kind":"REWORK","_number":12,"created":"2024-06-18 13:24:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/12"}}},"commit":{"parents":[{"commit":"428f37124986019bedc897469f9b1f3470b62082","subject":"CI: Pin requests to \u003c2.32 for publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428f37124986019bedc897469f9b1f3470b62082"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-18 13:24:34.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/127471017d8c52737389d5f5d459ca4184a5d9b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/127471017d8c52737389d5f5d459ca4184a5d9b5"}]},"branch":"refs/heads/master"},"78e81e586d9940860d51d6c558986eac2f36b169":{"kind":"REWORK","_number":13,"created":"2024-06-18 13:35:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/13"}}},"commit":{"parents":[{"commit":"428f37124986019bedc897469f9b1f3470b62082","subject":"CI: Pin requests to \u003c2.32 for publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428f37124986019bedc897469f9b1f3470b62082"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-18 13:35:42.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78e81e586d9940860d51d6c558986eac2f36b169"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78e81e586d9940860d51d6c558986eac2f36b169"}]},"branch":"refs/heads/master"},"687fa39fab9848ab051b414b6f28afb508c1587f":{"kind":"REWORK","_number":14,"created":"2024-06-18 13:37:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/14"}}},"commit":{"parents":[{"commit":"428f37124986019bedc897469f9b1f3470b62082","subject":"CI: Pin requests to \u003c2.32 for publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428f37124986019bedc897469f9b1f3470b62082"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-18 13:37:03.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/687fa39fab9848ab051b414b6f28afb508c1587f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/687fa39fab9848ab051b414b6f28afb508c1587f"}]},"branch":"refs/heads/master"},"c2de2178b2a9e1aa15f11262c51f1121ee817683":{"kind":"REWORK","_number":15,"created":"2024-06-18 13:38:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/15"}}},"commit":{"parents":[{"commit":"428f37124986019bedc897469f9b1f3470b62082","subject":"CI: Pin requests to \u003c2.32 for publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428f37124986019bedc897469f9b1f3470b62082"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-18 13:38:01.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c2de2178b2a9e1aa15f11262c51f1121ee817683"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c2de2178b2a9e1aa15f11262c51f1121ee817683"}]},"branch":"refs/heads/master"},"1806580412c7b0ef1fd5f44feb981b11cd9a2ac0":{"kind":"REWORK","_number":16,"created":"2024-06-18 15:32:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/16"}}},"commit":{"parents":[{"commit":"428f37124986019bedc897469f9b1f3470b62082","subject":"CI: Pin requests to \u003c2.32 for publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/428f37124986019bedc897469f9b1f3470b62082"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-18 15:32:22.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1806580412c7b0ef1fd5f44feb981b11cd9a2ac0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1806580412c7b0ef1fd5f44feb981b11cd9a2ac0"}]},"branch":"refs/heads/master"},"0c4ff1ab3fc5241656d3ad8e8873c28793a8c0cb":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2024-08-27 20:23:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/17"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-27 20:23:50.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c4ff1ab3fc5241656d3ad8e8873c28793a8c0cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0c4ff1ab3fc5241656d3ad8e8873c28793a8c0cb"}]},"branch":"refs/heads/master","description":"Rebase"},"6a9acefb45f3aca51884b5056eec87d42c3fa18f":{"kind":"REWORK","_number":18,"created":"2024-08-28 05:58:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/18"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 05:58:04.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a9acefb45f3aca51884b5056eec87d42c3fa18f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a9acefb45f3aca51884b5056eec87d42c3fa18f"}]},"branch":"refs/heads/master"},"58eb319e58182aa30e1edbd2dbafa37a922da18b":{"kind":"REWORK","_number":19,"created":"2024-08-28 05:59:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/19"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 05:59:16.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58eb319e58182aa30e1edbd2dbafa37a922da18b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58eb319e58182aa30e1edbd2dbafa37a922da18b"}]},"branch":"refs/heads/master"},"db17b20dfa634162b814cf0594aed2093d2a3388":{"kind":"REWORK","_number":20,"created":"2024-08-28 06:05:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/20"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 06:05:05.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db17b20dfa634162b814cf0594aed2093d2a3388"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db17b20dfa634162b814cf0594aed2093d2a3388"}]},"branch":"refs/heads/master"},"3b37af14c32f52aa47baeaa5bb086088c3d79552":{"kind":"REWORK","_number":21,"created":"2024-08-28 06:06:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/21"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 06:06:29.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3b37af14c32f52aa47baeaa5bb086088c3d79552"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3b37af14c32f52aa47baeaa5bb086088c3d79552"}]},"branch":"refs/heads/master"},"16432fe3c2a794dd80e19d86e56743d9d2565357":{"kind":"REWORK","_number":22,"created":"2024-08-28 06:07:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/22"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 06:07:08.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/16432fe3c2a794dd80e19d86e56743d9d2565357"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/16432fe3c2a794dd80e19d86e56743d9d2565357"}]},"branch":"refs/heads/master"},"aabec712fd1f6510879c74bd1df896e6ad2e522d":{"kind":"REWORK","_number":23,"created":"2024-08-28 06:07:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/23"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 06:07:19.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aabec712fd1f6510879c74bd1df896e6ad2e522d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aabec712fd1f6510879c74bd1df896e6ad2e522d"}]},"branch":"refs/heads/master"},"0ba4d46e6573a64eb20a7b725c2269e778d42c1f":{"kind":"REWORK","_number":24,"created":"2024-08-28 06:11:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/24"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 06:11:31.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ba4d46e6573a64eb20a7b725c2269e778d42c1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0ba4d46e6573a64eb20a7b725c2269e778d42c1f"}]},"branch":"refs/heads/master"},"8c2f3bb4b6d008706809ab32a2d9c0ccb0c82aca":{"kind":"REWORK","_number":25,"created":"2024-08-28 06:11:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/25"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 06:11:56.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c2f3bb4b6d008706809ab32a2d9c0ccb0c82aca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c2f3bb4b6d008706809ab32a2d9c0ccb0c82aca"}]},"branch":"refs/heads/master"},"31f903a6ae819af183d598fd9fb58cd2e96703b8":{"kind":"REWORK","_number":26,"created":"2024-08-28 06:23:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/26"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 06:23:57.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31f903a6ae819af183d598fd9fb58cd2e96703b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/31f903a6ae819af183d598fd9fb58cd2e96703b8"}]},"branch":"refs/heads/master"},"d5541a7f1cb07ce85ab4dc0a40d62dfefa8218ad":{"kind":"REWORK","_number":27,"created":"2024-08-28 06:30:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/27"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 06:30:35.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5541a7f1cb07ce85ab4dc0a40d62dfefa8218ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5541a7f1cb07ce85ab4dc0a40d62dfefa8218ad"}]},"branch":"refs/heads/master"},"78c5cc0b3799eaf0252e732435eb3b517ba3393a":{"kind":"REWORK","_number":28,"created":"2024-08-28 06:32:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/28"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 06:32:13.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78c5cc0b3799eaf0252e732435eb3b517ba3393a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78c5cc0b3799eaf0252e732435eb3b517ba3393a"}]},"branch":"refs/heads/master"},"32f4ec47deb9cd443ee9dcd18ccc2ac4249c2ac5":{"kind":"REWORK","_number":29,"created":"2024-08-28 06:38:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/29"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 06:38:26.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32f4ec47deb9cd443ee9dcd18ccc2ac4249c2ac5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32f4ec47deb9cd443ee9dcd18ccc2ac4249c2ac5"}]},"branch":"refs/heads/master"},"ee7b7c77b9bdea66c895d20fdeb1566fd3f69d8b":{"kind":"REWORK","_number":30,"created":"2024-08-28 06:47:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/30"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 06:47:10.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee7b7c77b9bdea66c895d20fdeb1566fd3f69d8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee7b7c77b9bdea66c895d20fdeb1566fd3f69d8b"}]},"branch":"refs/heads/master"},"a7f745e2a7311ddba823a6c87ba0a4e7625ba008":{"kind":"REWORK","_number":31,"created":"2024-08-28 06:51:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/31"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 06:51:51.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7f745e2a7311ddba823a6c87ba0a4e7625ba008"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7f745e2a7311ddba823a6c87ba0a4e7625ba008"}]},"branch":"refs/heads/master"},"a5d16d7715b99604d3103cfdebb61c672cccbca0":{"kind":"REWORK","_number":32,"created":"2024-08-28 06:52:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/32"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 06:52:19.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a5d16d7715b99604d3103cfdebb61c672cccbca0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a5d16d7715b99604d3103cfdebb61c672cccbca0"}]},"branch":"refs/heads/master"},"b583f63ec4d49205714f2a544cd828af787edb2d":{"kind":"REWORK","_number":33,"created":"2024-08-28 06:55:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/33"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 06:55:09.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b583f63ec4d49205714f2a544cd828af787edb2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b583f63ec4d49205714f2a544cd828af787edb2d"}]},"branch":"refs/heads/master"},"fc42b293dc98cdd8583b9009ce9fd302a1a01750":{"kind":"REWORK","_number":34,"created":"2024-08-28 07:05:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/34"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 07:05:16.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc42b293dc98cdd8583b9009ce9fd302a1a01750"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fc42b293dc98cdd8583b9009ce9fd302a1a01750"}]},"branch":"refs/heads/master"},"d9ff1f81165ce5f1e5d27fb3ab836fbe2e318e12":{"kind":"REWORK","_number":35,"created":"2024-08-28 07:16:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/35"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 07:16:24.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9ff1f81165ce5f1e5d27fb3ab836fbe2e318e12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9ff1f81165ce5f1e5d27fb3ab836fbe2e318e12"}]},"branch":"refs/heads/master"},"22d322fe93bf5e5d836fd3342dda59a6f1979806":{"kind":"REWORK","_number":36,"created":"2024-08-28 07:24:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/36"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 07:24:59.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22d322fe93bf5e5d836fd3342dda59a6f1979806"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22d322fe93bf5e5d836fd3342dda59a6f1979806"}]},"branch":"refs/heads/master"},"fe61cdfb583f59a61e9468daa3494b35aac5fc76":{"kind":"REWORK","_number":37,"created":"2024-08-28 07:40:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/37"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 07:40:34.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe61cdfb583f59a61e9468daa3494b35aac5fc76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe61cdfb583f59a61e9468daa3494b35aac5fc76"}]},"branch":"refs/heads/master"},"a0c58a5e2f10b7df00fb697e387ef096e969f60a":{"kind":"REWORK","_number":38,"created":"2024-08-28 07:54:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/38"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 07:54:11.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0c58a5e2f10b7df00fb697e387ef096e969f60a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0c58a5e2f10b7df00fb697e387ef096e969f60a"}]},"branch":"refs/heads/master"},"5bbbf57b36e52c9326e89795d12913fc280f84f4":{"kind":"REWORK","_number":39,"created":"2024-08-28 10:29:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/39"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 10:29:45.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5bbbf57b36e52c9326e89795d12913fc280f84f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5bbbf57b36e52c9326e89795d12913fc280f84f4"}]},"branch":"refs/heads/master"},"1da66f6b8830ac77c18d2c03a8b58dbc680ac142":{"kind":"REWORK","_number":40,"created":"2024-08-28 11:13:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/40"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 11:13:32.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1da66f6b8830ac77c18d2c03a8b58dbc680ac142"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1da66f6b8830ac77c18d2c03a8b58dbc680ac142"}]},"branch":"refs/heads/master"},"5cb3dd90dd6b13d0b7593c4db70e2f0602ce77f7":{"kind":"REWORK","_number":41,"created":"2024-08-28 11:28:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/41"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 11:28:13.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5cb3dd90dd6b13d0b7593c4db70e2f0602ce77f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5cb3dd90dd6b13d0b7593c4db70e2f0602ce77f7"}]},"branch":"refs/heads/master"},"20cdc32537275ce4845eae62ce9a837868a0f800":{"kind":"TRIVIAL_REBASE","_number":42,"created":"2024-10-02 07:37:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/48/922148/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/48/922148/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/48/922148/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/48/922148/42"}}},"commit":{"parents":[{"commit":"9b10b7d23579c0d06aa4b75b456f42d4116951f9","subject":"CI: Fix aarch64 publish jobs on x86","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b10b7d23579c0d06aa4b75b456f42d4116951f9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-06-17 19:48:27.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-02 07:37:29.000000000","tz":0},"subject":"CI: Move image building to a role","message":"CI: Move image building to a role\n\nThis role can be then used in kolla-ansible instead of setup_gate.sh\n\nChange-Id: Ibf04bd90a15e7d3949214d15a8b8e25f2050bb65\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20cdc32537275ce4845eae62ce9a837868a0f800"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/20cdc32537275ce4845eae62ce9a837868a0f800"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
