)]}'
{"id":"openstack%2Fkolla~923142","triplet_id":"openstack%2Fkolla~master~Ie6d3fe06dea59d2a0ee7cc0fb6f85a85c549c16d","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-07-12 09:08:59.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie6d3fe06dea59d2a0ee7cc0fb6f85a85c549c16d","subject":"[CI] [DNM] Test openstack-base CI build","status":"ABANDONED","created":"2024-07-01 09:33:59.000000000","updated":"2024-07-12 09:08:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fa1c49044441b5783bd90d734ef44b6b9f9b3a2d","_number":923142,"virtual_id_number":923142,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-07-01 15:15:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-01 11:53:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-01 15:15:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5e21b49765f7123a42c51e4efa446632fc98ff16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-07-01 09:33:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"14818a17b8d5b3e10c10a9638e2bd5a2d2e4214a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-07-01 09:54:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d61019b51ad419f11602f224d71e37557b099264","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-07-01 10:14:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1ededebfbdcddb27d29f66093a6413c65af31a23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-07-01 11:53:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9d8514f4e55060397c58257d27bbe07d872e80d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-01 11:53:58.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8fec31cc4494f9babcb7dd268a32e9a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/89e59b498cbe4b16b0732675eac46bf8 : FAILURE in 15m 45s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d5627e454c8b4f23aab2a53f9e8ba02b : FAILURE in 13m 01s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ad5705c599134ea08207c76801c184ad : FAILURE in 24m 56s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/33cb35ebff50400ebc6b0fa3fdeec75a : SUCCESS in 2m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fc5000bab6c54d9f8e76e8b125592a26 : SUCCESS in 5m 33s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1783ef6b6d9c44d6830b42ea55d26760 : FAILURE in 1h 36m 35s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a78bfa4bed654a968a5f2f884523b5c7 : FAILURE in 14m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6ccbc9b1ad7debb8f83b19ac34bdc73c163bdea3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-07-01 12:21:59.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"005661ce4da9f9cab04c1c04be3ec523a3d02d8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-01 15:15:08.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/12253bda1be3427094078c8213fce245\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/22843204e4d34ec8b5aa75468c36b8de : SUCCESS in 24m 57s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9d1d6465c4d34a709c29822cf61f2245 : SUCCESS in 1h 05m 41s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/12c60ee1ecf44f42b465f13c8eac5968 : FAILURE in 18m 12s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/51942fad6b674dcd919eb18a111b6b4f : FAILURE in 27m 36s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3a25fe71e9ac42a786c7be95c4f5f4e9 : FAILURE in 15m 02s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1a2f9bdee5ac451a824ec94945b2293f : FAILURE in 25m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba701a49b11943b1b5bb36099b4db29d : SUCCESS in 2m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f666c5f53cb746d3ae4797e3994dbdc7 : SUCCESS in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b564e08f0c714165b80cd5599a181e10 : SUCCESS in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0f0db403c89444e883ebe2c782652cd0 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0509ee29ca8947808d2ebe2c39a6fa91 : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67db33121bc34f079832f9064d1d3c80 : SUCCESS in 4m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0610b0f4cb25441d8d25b460e59fae24 : SUCCESS in 29m 21s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2e06df1b2cb04cae9291910181defec6 : SUCCESS in 34m 01s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1c5965ee0d2b4225a99a37beb416438d : SUCCESS in 54m 30s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b381992eb91d439284f6c08e886c2b1f : SUCCESS in 1h 38m 49s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/b4534c73ec914e409d0e06bd26d39d9c : SUCCESS in 1h 30m 05s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/37294b34c21047e3a30377643dd9a52a : SUCCESS in 2h 39m 04s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/cf1015fa79794a6da607966a67f8e6a3 : SUCCESS in 1h 21m 03s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/19230262f95c40cab46fd7079a792274 : SUCCESS in 1h 20m 44s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/f333cc34239f4e70ad76c46f86c7dee1 : SUCCESS in 27m 46s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/c416061d9cdf4aabb400bb9fbe0b5552 : SUCCESS in 1h 00m 36s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/17fcf56b901d48208b81cd460661295b : SUCCESS in 1h 10m 57s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/b161752cf1f74adaa4c6953f2a7da5ff : SUCCESS in 1h 30m 16s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/19c01f9812a74d5e9fe615fd36d870a5 : SUCCESS in 30m 56s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/763cc651789345188122e26c5f5f2054 : FAILURE in 3m 57s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/faf4fd0ac05b40f19fa1f4aae67302b1 : FAILURE in 4m 41s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8fd8b394d54e4bb28d33eed479cbee8e : FAILURE in 7m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a7ec59c98bcb43f0bfd677518fd3c9b8 : FAILURE in 7m 28s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/9d16b7e679a14d8c8a6efb8699c0d923 : FAILURE in 8m 30s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2d49b94d3e2e4a74a21d2bb07d9fdd94 : FAILURE in 9m 22s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/72c5aacb8f7241cf86465416529d7865 : FAILURE in 7m 20s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/902cc07a512e445787756b90e065a664 : FAILURE in 10m 05s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/e60ddb454241449fa00a140c504610ca : FAILURE in 7m 34s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/ecadace8917a4210b0aa78491d0bdb91 : FAILURE in 8m 08s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/f12fa3248f5f4e12b0f2d567a4eca4ea : FAILURE in 8m 38s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/986b6137646c498a8d3a01942757e475 : FAILURE in 9m 09s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ce8e24de67664895a8be4b648c0cb5f0 : FAILURE in 8m 01s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8110d7e0817113d4eb84a8e08ac48c92e56e841e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-01 15:28:28.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ca6cad3ffa846feae67e37d6513b43a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0a88c94c9f4d4bc78f00525223d0234d : TIMED_OUT in 3h 04m 11s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/27b6612eab8b44839956192c32b9a08d : FAILURE in 12m 50s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e6a12cab01844067b0db0e7a49c76502 : FAILURE in 56m 56s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4e8a66b134a241f19176c4657b80f200 : SUCCESS in 2m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/01a6286f6461427c8e344ada3e59f5c2 : SUCCESS in 4m 16s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d5c6c9b7c934418084fede86ecd13276 : FAILURE in 22m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4b00f081ab55486790e733400fa29cba : FAILURE in 3m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fa1c49044441b5783bd90d734ef44b6b9f9b3a2d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-07-12 09:08:59.000000000","message":"Abandoned\n\nnot needed","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"15dacd5077e0bc43df911a929d4d675de535e770","revisions":{"d455a250bcb7d9ea4f590dba41fa869d3ab7afd4":{"kind":"REWORK","_number":1,"created":"2024-07-01 09:33:59.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/42/923142/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/923142/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/923142/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/923142/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/923142/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/923142/1"}}},"commit":{"parents":[{"commit":"603f58cac55dd3081bcb5b7104e902f216089646","subject":"Merge \"Change logic for plugins in grafana image build\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/603f58cac55dd3081bcb5b7104e902f216089646"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-07-01 09:33:35.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-07-01 09:33:35.000000000","tz":120},"subject":"[CI] [DNM] Test openstack-base CI build","message":"[CI] [DNM] Test openstack-base CI build\n\nChange-Id: Ie6d3fe06dea59d2a0ee7cc0fb6f85a85c549c16d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d455a250bcb7d9ea4f590dba41fa869d3ab7afd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d455a250bcb7d9ea4f590dba41fa869d3ab7afd4"}]},"branch":"refs/heads/master"},"448e91bf3c823d4f25ae9dc3cf76b628fe0bed9a":{"kind":"REWORK","_number":2,"created":"2024-07-01 09:54:44.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/42/923142/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/923142/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/923142/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/923142/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/923142/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/923142/2"}}},"commit":{"parents":[{"commit":"603f58cac55dd3081bcb5b7104e902f216089646","subject":"Merge \"Change logic for plugins in grafana image build\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/603f58cac55dd3081bcb5b7104e902f216089646"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-07-01 09:33:35.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-07-01 09:54:30.000000000","tz":120},"subject":"[CI] [DNM] Test openstack-base CI build","message":"[CI] [DNM] Test openstack-base CI build\n\nChange-Id: Ie6d3fe06dea59d2a0ee7cc0fb6f85a85c549c16d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/448e91bf3c823d4f25ae9dc3cf76b628fe0bed9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/448e91bf3c823d4f25ae9dc3cf76b628fe0bed9a"}]},"branch":"refs/heads/master"},"a3868e4b3124a5f2f633889a8c572044485c9fb6":{"kind":"REWORK","_number":3,"created":"2024-07-01 10:14:59.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/42/923142/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/923142/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/923142/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/923142/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/923142/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/923142/3"}}},"commit":{"parents":[{"commit":"603f58cac55dd3081bcb5b7104e902f216089646","subject":"Merge \"Change logic for plugins in grafana image build\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/603f58cac55dd3081bcb5b7104e902f216089646"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-07-01 09:33:35.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-07-01 10:14:52.000000000","tz":120},"subject":"[CI] [DNM] Test openstack-base CI build","message":"[CI] [DNM] Test openstack-base CI build\n\nChange-Id: Ie6d3fe06dea59d2a0ee7cc0fb6f85a85c549c16d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3868e4b3124a5f2f633889a8c572044485c9fb6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3868e4b3124a5f2f633889a8c572044485c9fb6"}]},"branch":"refs/heads/master"},"64d986cc700c41ea4ebdf35fb7dc4ff46ca62746":{"kind":"REWORK","_number":4,"created":"2024-07-01 11:53:53.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/42/923142/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/923142/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/923142/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/923142/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/923142/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/923142/4"}}},"commit":{"parents":[{"commit":"603f58cac55dd3081bcb5b7104e902f216089646","subject":"Merge \"Change logic for plugins in grafana image build\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/603f58cac55dd3081bcb5b7104e902f216089646"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-07-01 09:33:35.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-07-01 11:53:46.000000000","tz":120},"subject":"[CI] [DNM] Test openstack-base CI build","message":"[CI] [DNM] Test openstack-base CI build\n\nChange-Id: Ie6d3fe06dea59d2a0ee7cc0fb6f85a85c549c16d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64d986cc700c41ea4ebdf35fb7dc4ff46ca62746"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64d986cc700c41ea4ebdf35fb7dc4ff46ca62746"}]},"branch":"refs/heads/master"},"15dacd5077e0bc43df911a929d4d675de535e770":{"kind":"REWORK","_number":5,"created":"2024-07-01 12:21:59.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/42/923142/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/923142/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/923142/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/923142/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/923142/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/923142/5"}}},"commit":{"parents":[{"commit":"603f58cac55dd3081bcb5b7104e902f216089646","subject":"Merge \"Change logic for plugins in grafana image build\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/603f58cac55dd3081bcb5b7104e902f216089646"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-07-01 09:33:35.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-07-01 12:13:54.000000000","tz":120},"subject":"[CI] [DNM] Test openstack-base CI build","message":"[CI] [DNM] Test openstack-base CI build\n\nChange-Id: Ie6d3fe06dea59d2a0ee7cc0fb6f85a85c549c16d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/15dacd5077e0bc43df911a929d4d675de535e770"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/15dacd5077e0bc43df911a929d4d675de535e770"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
