)]}'
{"id":"openstack%2Fkolla~923942","triplet_id":"openstack%2Fkolla~stable%2F2023.1~I64e9442545d98364585e3c632ae9142cd6b4f9bf","project":"openstack/kolla","branch":"stable/2023.1","topic":"bug/2066171-stable/2023.1","attention_set":{},"removed_from_attention_set":{"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2024-07-12 08:17:14.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I64e9442545d98364585e3c632ae9142cd6b4f9bf","subject":"Pass rabbitmq apt preferences into kolla-toolbox","status":"ABANDONED","created":"2024-07-11 10:12:58.000000000","updated":"2024-07-12 08:17:14.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":920037,"cherry_pick_of_patch_set":4,"meta_rev_id":"1fe9d1a755547ab5dc57c301f9847744aad88234","_number":923942,"virtual_id_number":923942,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-11 17:21:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-11 10:54:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-11 11:35:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07f28dec89ef596cfcf8ea6f9dde0b0b28821ba3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-07-11 10:12:58.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2723ccb0ae151e17cd54585217b41bb9ffb05610","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 10:54:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c30bbc34c9d24af2a8fc74410cec823a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/016a7b1423ff45d3b6f6d8cf1522224a : RETRY_LIMIT Host unreachable in 5m 07s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/df3f12ed6f204ab6896bdf3252f7c1af : RETRY_LIMIT in 5m 02s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/edab40b571ef49e486949178cae89089 : SUCCESS in 5m 01s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1fab74acf647431b90eee6af8146463a : RETRY_LIMIT in 6m 14s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/496fb90120424149a73132a0708e7921 : RETRY_LIMIT in 3m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6f9b5b5dac3d74f8bd6938b223af26861748bf15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 11:35:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a4726f0fa284e3e9fd9093a311af269\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/01263c661a5e4f87a915c73524ba02bf : RETRY_LIMIT in 2m 37s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/706dddaa4696408da4b9e3a477d8eece : FAILURE in 8m 20s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/004b7e5812864a6097ca3d6fc45e2283 : RETRY_LIMIT Host unreachable in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/501a80be4f984c13b359ea896e9df956 : SUCCESS in 2m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac9f9c4c609d451ba3c5ea99ed3b6f0d : SUCCESS in 3m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61c8ff26f0e444dda4abff542ea95475 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9baf49b81efe4afa8d47a6b53dbf96b4 : SUCCESS in 3m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bad1d8386b9d4f35abf0224c89520b54 : SUCCESS in 3m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b04d9c8fc613445eba0e00ddc0e185ae : RETRY_LIMIT in 3m 34s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/549803712ffd4238b20f400bb9ac1978 : FAILURE in 9m 00s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8893d20621d94861b66fa6d0677597d2 : FAILURE in 8m 07s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8bd09b3dfd0944d1bc9d0a80e2d9c0da : RETRY_LIMIT in 2m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/dd65d20db3af484c8bad324b4536005f : FAILURE in 7m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/400fbbebe547401a9f737fc63c457cce : FAILURE in 7m 18s","accounts_in_message":[],"_revision_number":1},{"id":"99c272184a9526c7583d6b85afddd484f6e3fe36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-07-11 11:58:25.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bb7d8953416034aed19ed2f1814ff7a4f4273b2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 13:46:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/22a6d01a0f474f1ab34832e68f72b969\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8eb2fe8a6ace4f6c964c5e3d7af1a91d : SUCCESS in 27m 59s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/dcdd703e38f44ddc9d94b1d2ba0df75b : SUCCESS in 1h 02m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8c0cf7ba22614658b76ced2df9791874 : FAILURE in 19m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ee029c36a8e40c898af14b57aff2037 : SUCCESS in 2m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf4a5b32156b4accb04a1abead739f32 : SUCCESS in 3m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5754d470b21d4f17b19fabae8d178553 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b6d7bb112714957926f05ca18c7eaf6 : SUCCESS in 4m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15228a0615074af0bdabb2827f8cbaea : SUCCESS in 3m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d5da3326f23f4ad8a2ac239a7e547214 : SUCCESS in 24m 43s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c8a4818b6eac4e9987b667f4f0488a28 : SUCCESS in 1h 03m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/780dd2e10f514b6eb0021fd313837320 : SUCCESS in 1h 47m 24s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f8e54250b43e4f019a8097264fd393ce : FAILURE in 23m 46s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/74c439a8728e497fa14d02da30df3d43 : FAILURE in 14m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a6dc175e6c524ccd9e42dfded50a4c40 : FAILURE in 14m 19s","accounts_in_message":[],"_revision_number":2},{"id":"2a5a0b4fb93dc479bf20f54bec42915b425ba203","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 15:32:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b67013db46704f70a957a182bdfe97f6\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/fa757b330b4a4f41a2a2b53f932a0b48 : TIMED_OUT in 3h 04m 54s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bef4e5566b5742ddb83fa804df57c0b3 : FAILURE in 18m 57s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5a5b34fbc1074b409c6c4c5cc296dfde : SUCCESS in 6m 37s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/74f307cb580a4045af24568a31178e97 : TIMED_OUT in 3h 04m 06s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/069c4ab376bd4802a02ef8176533749c : TIMED_OUT in 3h 02m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"eead23209f84e36c704b7555eefc122b0a56a7eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-07-11 15:53:02.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"053b966d999272408a33b0e9f605032362d998e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 17:21:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80af37cb8dfc429aa4a7bec3d652ea91\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/21f1be1d238d47e88e1158859f0b6c9f : SUCCESS in 26m 03s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/02b31c61f4f74efcbf0387614b61b6c8 : SUCCESS in 58m 24s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/616082ffe336497f9ebf3eb59dddd3bf : SUCCESS in 17m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64d2f89ac8e04480a249d7aed4c73be5 : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62abc7a249aa4320b75d4c1e6d962502 : SUCCESS in 2m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/409d862e9c7a45f285feeeb5606c69ed : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b018e8ad55804b2699dcf1d5493ec3d5 : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ccade3b0bdc444f0863c76410697df71 : SUCCESS in 3m 26s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/07dde2cc838e48f8b31119e627674737 : SUCCESS in 25m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c608da385b8046b389e4d63ce72715fa : SUCCESS in 1h 07m 28s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1848b382096e462395ce8f64a76158df : SUCCESS in 1h 23m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b84866d017094ebbba690bfd43f6fbf6 : SUCCESS in 30m 04s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0dc1f750934f4031a637b7bcb7c73645 : SUCCESS in 1h 00m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/27bd9322d3684a3ba545d695648c0294 : SUCCESS in 1h 25m 02s","accounts_in_message":[],"_revision_number":3},{"id":"a3999e3d0d84624741df685d651608b9ddfd43bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 18:52:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31cefbcb302b476d8106582ae7556651\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/4e765fc0572e49ffbdd626a8d46e82a7 : FAILURE in 20m 43s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4286603bab044d1b8b560daea2f61ccd : SUCCESS in 2h 51m 35s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5da2e552f3184248b57ec5efbdfd71e3 : SUCCESS in 2m 43s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7c8e6f72a8424dcdb268d1df592c6452 : SUCCESS in 2h 56m 12s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5fd9a181686e4eed8f1caa0f58515960 : FAILURE in 15m 28s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1fe9d1a755547ab5dc57c301f9847744aad88234","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-07-12 08:17:14.000000000","message":"Abandoned\n\nabandonded: merged into this patch: https://review.opendev.org/c/openstack/kolla/+/918973 as fixes are required together","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"43252a53a75f71c601f6b5061767e1e67d5b6a1c","revisions":{"35e5c89b40ea81731106927ce614acdedc33c0d2":{"kind":"REWORK","_number":1,"created":"2024-07-11 10:12:58.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/42/923942/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/923942/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/923942/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/923942/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/923942/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/923942/1"}}},"commit":{"parents":[{"commit":"46989bd53c4d6964faaa40b6d0e4c974bbdb2ff1","subject":"CI: fix publish docker credential var","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/46989bd53c4d6964faaa40b6d0e4c974bbdb2ff1"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-05-20 13:29:01.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-07-11 10:12:58.000000000","tz":0},"subject":"Pass rabbitmq apt preferences into kolla-toolbox","message":"Pass rabbitmq apt preferences into kolla-toolbox\n\nIf the latest and targeted rabbitmq-server dependencies mismatch,\nkolla-toolbox will fail to build.\n\nCloses-Bug: #2066171\nChange-Id: I64e9442545d98364585e3c632ae9142cd6b4f9bf\n(cherry picked from commit 006ccee101b3944eeb22299226d45520850dba0e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35e5c89b40ea81731106927ce614acdedc33c0d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35e5c89b40ea81731106927ce614acdedc33c0d2"}]},"branch":"refs/heads/stable/2023.1"},"b586769e0cc7ad20f2e39fdfc3dd5d5c9a6ce3a8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-11 11:58:25.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/42/923942/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/923942/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/923942/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/923942/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/923942/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/923942/2"}}},"commit":{"parents":[{"commit":"095fe099f82380ebd68a9244f920e5306b37912b","subject":"Move apt_prefs pins to respective images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/095fe099f82380ebd68a9244f920e5306b37912b"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-05-20 13:29:01.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-07-11 11:57:57.000000000","tz":60},"subject":"Pass rabbitmq apt preferences into kolla-toolbox","message":"Pass rabbitmq apt preferences into kolla-toolbox\n\nIf the latest and targeted rabbitmq-server dependencies mismatch,\nkolla-toolbox will fail to build.\n\nCloses-Bug: #2066171\nChange-Id: I64e9442545d98364585e3c632ae9142cd6b4f9bf\n(cherry picked from commit 006ccee101b3944eeb22299226d45520850dba0e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b586769e0cc7ad20f2e39fdfc3dd5d5c9a6ce3a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b586769e0cc7ad20f2e39fdfc3dd5d5c9a6ce3a8"}]},"branch":"refs/heads/stable/2023.1"},"43252a53a75f71c601f6b5061767e1e67d5b6a1c":{"kind":"REWORK","_number":3,"created":"2024-07-11 15:53:02.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/42/923942/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/923942/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/923942/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/923942/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/923942/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/923942/3"}}},"commit":{"parents":[{"commit":"aeb4f59d4e86350c291169eb4df5de4ed0556056","subject":"Move apt_prefs pins to respective images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aeb4f59d4e86350c291169eb4df5de4ed0556056"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-05-20 13:29:01.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-07-11 15:52:33.000000000","tz":60},"subject":"Pass rabbitmq apt preferences into kolla-toolbox","message":"Pass rabbitmq apt preferences into kolla-toolbox\n\nIf the latest and targeted rabbitmq-server dependencies mismatch,\nkolla-toolbox will fail to build.\n\nCloses-Bug: #2066171\nChange-Id: I64e9442545d98364585e3c632ae9142cd6b4f9bf\n(cherry picked from commit 006ccee101b3944eeb22299226d45520850dba0e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/43252a53a75f71c601f6b5061767e1e67d5b6a1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/43252a53a75f71c601f6b5061767e1e67d5b6a1c"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
