)]}'
{"id":"openstack%2Fkolla~924211","triplet_id":"openstack%2Fkolla~master~I1bb815629ee38e215c51b314a2168dc5b689792c","project":"openstack/kolla","branch":"master","topic":"fix/missing_kerberos_packages_for_keystone","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-07-16 12:01:23.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"34311":{"account":{"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},"last_update":"2024-10-02 09:47:11.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-07-16 20:13:17.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I1bb815629ee38e215c51b314a2168dc5b689792c","subject":"fix: add missing packages for krb5-conf","status":"ABANDONED","created":"2024-07-16 10:55:48.000000000","updated":"2024-10-02 09:47:11.000000000","total_comment_count":7,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"bf8a9119a2abaeda0ec5c375ce6be15e2ae48ca8","_number":924211,"virtual_id_number":924211,"owner":{"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-16 12:41:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2024-07-16 11:03:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2024-07-16 20:13:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-16 10:57:51.000000000","updated_by":{"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-07-16 12:41:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-16 13:14:04.000000000","updated_by":{"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"86f61377f7c031800e2792a89d3e5330a05200f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},"date":"2024-07-16 10:55:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c945773b27c3735d0417e9bf963cbffeea85a28","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-16 11:03:26.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"39cc49c619d6e662cec2dc0b569baf60346a4f66","author":{"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},"date":"2024-07-16 11:21:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3d230385cd1b4cbb5c6ef0d9e7f7a2a3f1f97c2e","author":{"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},"date":"2024-07-16 11:50:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b832fe99b477e263bca18b6a71c564b39e281856","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-16 12:01:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e7547d08046f0daa0d2c1852a872ea3ae271f6a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 12:41:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5feeb15d9aa7406e80b0c17a19d8ae3c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/fa753621b1d049aebdc576a29604c7c4 : SUCCESS in 26m 53s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9f1d9f60aba14eea86eee450a0cc40fa : SUCCESS in 1h 16m 25s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/47f4aef3361542c18b84c2b962371967 : SUCCESS in 17m 58s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d71fa4d4d241484a976bcbc15a95433c : SUCCESS in 26m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/66ddd1f322e847b899e7301f97379da6 : SUCCESS in 1h 01m 41s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ab5f46af28c049cbb734849a51314a2d : SUCCESS in 1h 27m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b3e0780596d47698fea08dfd7f735b9 : SUCCESS in 2m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd9595325ad540d8817f6431d5e061f4 : SUCCESS in 2m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd6737bece364c039067e8d88f6c06c3 : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b334813445c84abbaca8677def0b5ec4 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bcfe890fef0d44bb811d00acfbfce608 : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f50e7ef30ad1479bb4bf9f4b023b8466 : SUCCESS in 4m 22s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4fbc6081d8c54cfbb64d4cf6119ee34f : SUCCESS in 22m 34s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/8b217ac0cd804b1399657f733bd127f2 : SUCCESS in 39m 21s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/250f9775b7af4896aad034eae75103e0 : SUCCESS in 1h 11m 28s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/24ff196132da4da2a6478aa7229d4806 : SUCCESS in 1h 33m 29s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/35d3569d20b543998b8ce2d0d59cbbfc : SUCCESS in 16m 58s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/650606c26a8f41bdafabbb722bda3ed7 : SUCCESS in 55m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/fb6eb8c21f364ab8a52615eacfe3cde2 : SUCCESS in 48m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ce1311b797d44cfba6e414841fd059bd : SUCCESS in 1h 14m 58s","accounts_in_message":[],"_revision_number":1},{"id":"f64a49aa4222ac66a6fe9ce463594143eff00847","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 13:32:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31af1e37a5454754b1bedfcf3d42239b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6789b442ce0c4fca8aa7ea59b63c0037 : SUCCESS in 2h 33m 33s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/161e22d4ae184126a00b0171ef68514b : SUCCESS in 30m 11s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9d327cdbd9ea4a34bf258d9ce886daee : SUCCESS in 2h 07m 08s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4a32e99776054691b0c1b04c56ecc011 : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/acc4e62e90dd4072a07287fab5b6b16e : SUCCESS in 4m 32s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/93e20aaf8094428485e9435361f08a9f : SUCCESS in 33m 30s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/01ba354a10b148728a1be61a60594c04 : SUCCESS in 2h 34m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e519b53abb1ef7bec19a65045d99e19aeea2941b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-07-16 16:42:48.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f5f3664747a2eb04aa394aff3ed5a8f8b71fc379","author":{"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},"date":"2024-07-16 18:58:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"016e765444e9f2072e6cb5355cb7d0ec72bb5360","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-07-16 20:13:17.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bf8a9119a2abaeda0ec5c375ce6be15e2ae48ca8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-02 09:47:11.000000000","message":"Abandoned\n\nIt has been recently fixed by https://review.opendev.org/c/openstack/kolla/+/927775","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2f1b076c6bfe770b27a69c75f133a9c7dbf29c7f","revisions":{"2f1b076c6bfe770b27a69c75f133a9c7dbf29c7f":{"kind":"REWORK","_number":1,"created":"2024-07-16 10:55:48.000000000","uploader":{"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},"ref":"refs/changes/11/924211/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/11/924211/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/924211/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/924211/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/11/924211/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/11/924211/1"}}},"commit":{"parents":[{"commit":"139851087e15e0aa0c47765768bcdc1da80246d6","subject":"Merge \"Pass rabbitmq apt preferences into kolla-toolbox\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/139851087e15e0aa0c47765768bcdc1da80246d6"}]}],"author":{"name":"Tadas Sutkaitis","email":"tadasas@gmail.com","date":"2024-07-16 10:55:20.000000000","tz":180},"committer":{"name":"Tadas Sutkaitis","email":"tadasas@gmail.com","date":"2024-07-16 10:55:20.000000000","tz":180},"subject":"fix: add missing packages for krb5-conf","message":"fix: add missing packages for krb5-conf\n\nCloses-Bug: #2073253\nChange-Id: I1bb815629ee38e215c51b314a2168dc5b689792c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2f1b076c6bfe770b27a69c75f133a9c7dbf29c7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2f1b076c6bfe770b27a69c75f133a9c7dbf29c7f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
