)]}'
{"id":"openstack%2Fkolla~924245","triplet_id":"openstack%2Fkolla~master~I000e2f2aacb011618a0b3c7b41fe286b3680614f","project":"openstack/kolla","branch":"master","topic":"kolla_ansible_2_17","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-07-24 21:50:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I000e2f2aacb011618a0b3c7b41fe286b3680614f","subject":"toolbox: Bump ansible-core to 2.17","status":"MERGED","created":"2024-07-16 15:54:53.000000000","updated":"2024-07-24 21:51:56.000000000","submitted":"2024-07-24 21:50:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924245-kolla_ansible_2_17","meta_rev_id":"c5e3e528d3132f7fd16452fb8e1ef75cc9b012a8","_number":924245,"virtual_id_number":924245,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-24 21:50:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-07-22 15:44:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2024-07-24 10:08:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-07-24 10:08:15.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2024-07-22 15:44:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-07-17 09:26:11.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-16 17:30:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-22 15:44:35.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-07-24 10:08:15.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"ce300697956792d78eda55420ae52274fa62f670","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-16 15:54:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8bfae1f501173b10b48ebb1ad3e0f460397c490f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 17:30:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1d33287a6f342b5bec4af30740973b8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2991357dba1b41ff92c4538fd3ee818c : SUCCESS in 27m 38s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/20eb03e69a8a476d890ca9511bdc6939 : SUCCESS in 1h 07m 46s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e3b850c175854c6e8ca51a623770801b : SUCCESS in 22m 55s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/32b81dbf4888483da85865ac0bd59601 : SUCCESS in 29m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/156c6eccadc147f98d8039bbeca16b81 : SUCCESS in 1h 11m 13s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f162ade02f56479d95b712dad07f822d : SUCCESS in 1h 28m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3e3dd71352d4c3896cfebfa761f5abf : SUCCESS in 3m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5d2d6befb0242be80dcbdb9a1d8bd30 : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1816e722fff74f9a8a7f04b5a8ea44c0 : SUCCESS in 4m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7ae99cbe10934d9f95ff4c1799f9c25c : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/737d987d23b944c5b12fcccf2273ef9a : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1826d313e9904cd78ce824de8bc9cb85 : SUCCESS in 5m 05s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b554a0f0d7c84aa7b7527a59c05c16cf : SUCCESS in 25m 12s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2e816ab6c899475bade64a8e2fb6338a : SUCCESS in 32m 56s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a2f85645acb148789418d2e24b5619fd : SUCCESS in 1h 13m 54s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/478739ccc4f444e1ba0b422e833b2f55 : SUCCESS in 1h 27m 17s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/444ac8108447478cab6d8394b95b5f96 : SUCCESS in 21m 26s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/dff593cf30f24eb4a420f555f332261b : SUCCESS in 49m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4ba512cf63d0483dac38014b5e02f2b3 : SUCCESS in 53m 04s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b1e4327274c642edb90d774f3e2acb98 : SUCCESS in 1h 24m 45s","accounts_in_message":[],"_revision_number":1},{"id":"e817408591ef66e74ff65e6792bdad40a90a4fd9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 19:12:02.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e8cc25e447d44858f0a50ca31564633\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a6db0f412a2e4c0cba756b70b8226715 : SUCCESS in 49m 10s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c564a67d1126472f95a156d2d2bc0ba9 : SUCCESS in 46m 14s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/aea7535c6b0b4e61a6fef7c283686a50 : TIMED_OUT in 3h 03m 29s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d9f824f5517546cfa1515b8ad3295e69 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b0548539183142d6be4a79f47128807a : SUCCESS in 5m 42s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/98e7fe2653bb4191a85df3a2075d20b0 : TIMED_OUT in 3h 06m 02s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/264986ac95ad42358327d376b812a66b : SUCCESS in 37m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b119cfc4c241a2c51bf702ccd4b153a3cc39fd8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-17 09:26:11.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"f385f66d286b4c73584ec4bbe0a511e697522039","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-22 15:44:35.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b53be9211ad6439cc8711f4e77c5b1c465cc4a4e","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-07-24 10:08:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e436930cc1baa065a279e8b174d8fe28f12ec5b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 10:08:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0a4b0e28f443c200de0d57971d1b96120f22ffb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 11:45:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/02a8799b68df44198c91bdcd0eb3d0ea\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b37466961b824a0d9d094c68ca4d4af4 : SUCCESS in 16m 13s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3accce51127a4ea8bf4c416bba5f3ab4 : SUCCESS in 28m 57s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/eab5c40c73fd4cc38813ea317cb61f24 : SUCCESS in 1h 08m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/13d2664a26294f4baedf877054026c69 : SUCCESS in 1h 27m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b02f4f474e654925a549809adca1429e : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/681e3f0be21941b9b0faed289b45557b : SUCCESS in 2m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f18e0a5df5e744b4805ad9c03f295c5e : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6188ddc21bc45c9b5bb6f0d1bd58ab7 : SUCCESS in 4m 24s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6989390c57a34aaeb008aaa35a65fb6e : SUCCESS in 27m 18s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d1a21bcfbb634232968f0093de32a6de : SUCCESS in 34m 34s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8e3392c4fdbf4b8da62f9f874c1ef190 : SUCCESS in 1h 13m 50s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d6306a4a085f461a9aaa2a0bdccd88d9 : SUCCESS in 1h 35m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d79f32b3485940af9f5d4b86a4e7894a : SUCCESS in 22m 43s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5d0725dc1a2542328529c6df58f9dc82 : SUCCESS in 53m 22s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7cf7fca118a746fd88c5a16eb1c3ecc3 : FAILURE in 43m 45s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/915056797c0d4587a254a98ba53c488e : SUCCESS in 1h 21m 41s","accounts_in_message":[],"_revision_number":1},{"id":"2edcf31facb89224204fead480e9a21200f86031","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-24 12:24:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"daeb421f9c84d81aab1a23236c969692de76ce93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 14:28:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55cfe6469c6c4406997bca706143a4a9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0a56066088534f3d9a597ff618a063d2 : SUCCESS in 22m 41s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2dc66bb4bf704bde814d08f5d86f817b : SUCCESS in 1h 01m 44s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ca6504b9d1194a0fa28f25017cefd41f : SUCCESS in 20m 43s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6bf6258a84b5450f9492ef9e59e6934a : SUCCESS in 22m 34s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4c9f5aad8fa04f4b89915ee5ad71dc98 : SUCCESS in 59m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2273cf6c020747bf9863ec5c42aee4b4 : SUCCESS in 1h 19m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6eeb9b07270e4e5b9c3151c5f0a312d1 : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/066a7e2ae30f44aab1ddb376e749af0b : SUCCESS in 2m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36a3d95321644f2fb580135a03c03e83 : SUCCESS in 2m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b65f69d772bf4b1f9c7c80716110529a : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ece7d58234bf4ced9606a4edcb6e0126 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/797db77fe689486b815f3d01013cb31a : SUCCESS in 5m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f99d87828abf4bd69f4ccd9318c65430 : SUCCESS in 26m 10s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/59d423ecaac64c638dc0fc3509f8418b : SUCCESS in 48m 27s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f67e71a1072a414f90b9f58c007ccc00 : SUCCESS in 1h 09m 40s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4008d835c1b1411cbb632b3d88a62fa6 : SUCCESS in 1h 56m 41s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f101bdf9b659475ebf796c1c0c3c5023 : SUCCESS in 17m 44s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a2c496af2c954622a6fa68e741424cb0 : SUCCESS in 1h 00m 29s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e7b0eb4965e94005873ede1608b2c248 : SUCCESS in 52m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/38e72455116e40a79709c653e753fc68 : SUCCESS in 1h 13m 01s","accounts_in_message":[],"_revision_number":1},{"id":"f14ac521dc9412e3c44abb152aca525e81bf3e9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 14:28:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1ef0c59453c065bc3d0a97eb07b13c6088a6615f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 15:00:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a07914f7856046918f4243f84f531c6f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6fd74ce22d3840d5bba33246e227028b : FAILURE in 21m 22s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7c45d3e5338547ffa59ea82a6e3933d2 : SUCCESS in 17m 58s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f5951ccb9d924dc4ad2bf5f389557d7e : SUCCESS in 1h 19m 18s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fa90a06b5e184c68ba4aa8e3ef51f136 : SUCCESS in 2m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5cd4d4f7febd41989553976db4cdea96 : SUCCESS in 3m 58s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/976b2b458eec43b3b9db8bfd4b0bb2b1 : SUCCESS in 27m 41s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f53cee9bab0f4c8da0948571808050ad : SUCCESS in 2h 33m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"40fba98172fc25fbaf692d8cda1408d122cfb7eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 16:04:34.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cfadf5e75e09425f87362540e91a1c1e\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/82ac7994192e4189bf93ef48783e0334 : SUCCESS in 19m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7a2dba6095f24cff8a003833e2e3da9b : SUCCESS in 32m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0eb4b35d15434a2c950bf9ad191cb2a5 : SUCCESS in 1h 06m 37s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/010707ac636249898f8bb5b808d9aecf : FAILURE in 1h 27m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b0759703a5e475f97207e9d8c047987 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/443c70fbf0ab4a32afc1799bd15c9a30 : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9a937f5b4cb74cd4891064de76b81598 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/831dc674f3984369b0bd9bc5f2ee30ad : SUCCESS in 4m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3413cb1c7926453590a2a6331da459b3 : SUCCESS in 27m 45s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c7b70ac123c147108ab89d7aa50fd7d3 : SUCCESS in 32m 46s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/de19522facdb48ce8937a4de714961e2 : SUCCESS in 1h 10m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1fee39740dd44e4ea8e8aec816f16e8f : SUCCESS in 1h 34m 19s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/73e18edcdc6d4ebca2bce6f6a805e3f2 : SUCCESS in 18m 59s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1483354c726f43fbb0f8e7010c3d85fa : SUCCESS in 44m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6db2a2030d53491292f5859f0a7dcab3 : SUCCESS in 53m 19s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b67a10f513a248e788be8b8c637ce270 : SUCCESS in 1h 25m 19s","accounts_in_message":[],"_revision_number":1},{"id":"36df5d000769c589bf880280961429e179216cdb","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-24 18:14:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"99ed864f7aa517d966cd00d6b8a8c128c90a6fd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 20:03:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be6fb59ad8d644ff980495fae290f66a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/45dca63010d24310a29b36143ec2942f : SUCCESS in 25m 04s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d4879d35bed24a8abaf65a0b9326d581 : SUCCESS in 1h 08m 42s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b524edf26f9a408dbf89e31ec5ed0cc8 : SUCCESS in 19m 09s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5350efb1085c4cae8fb40bfc53bd7a54 : SUCCESS in 29m 42s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/015e2bde64814cbf9efefe91868a5065 : SUCCESS in 55m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/08411b1e42274d119c1c87fda161504d : SUCCESS in 1h 19m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95980342d4fc458ca8762bedb7b22936 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/650c01756f6443d29026109e42cc444e : SUCCESS in 3m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/da8b79aa79bd473fa39d0b6bb103e3c2 : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/abace9fc4a6744d2bbe16d91284104cc : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02d86acadad0450d9baf3b43cebd06d4 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dde1c380d3ac49ed80e32a04b0a6f346 : SUCCESS in 4m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9d3e54aaa7da4abf88a0ea6cb8d0e711 : SUCCESS in 27m 21s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/bb96df7ce7904717a3f0909b0f1c33b1 : SUCCESS in 37m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/bc00bf5f4399436c8ba695097c249624 : SUCCESS in 1h 08m 11s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/03c0ae7d654a402fac8a15884559a4f9 : SUCCESS in 1h 43m 33s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a290df6f88ab4fe0881d60b9dcf9e363 : SUCCESS in 23m 05s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/32e4345a75d94742a9ef0d2402cd6acd : SUCCESS in 57m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0817ed9e8e2143428d15ae29c6c2569f : SUCCESS in 50m 29s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5de6f23481604172b7021105d80bf0a7 : SUCCESS in 1h 30m 38s","accounts_in_message":[],"_revision_number":1},{"id":"01cb2265bd0136776fba1119dc57bcc7d27f18f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 20:04:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9503dba73ecfb16607fcae3a6494cf904418f4d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 20:36:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/025e7cf197404d5ebf14c507cf40f9dd\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/eb51cb97a64a4b97a9857e6e9e848639 : SUCCESS in 20m 02s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/08772abe7dce45f0a725f222e41d45de : SUCCESS in 21m 09s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0b46e4644bb84e5488b16f7309bb8ed1 : SUCCESS in 1h 08m 45s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a4f04a8e8aa2414999dc24890963cb54 : SUCCESS in 2m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d39b9980d82346f28aa3503968fa431b : SUCCESS in 4m 50s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/6e4e1289795b42858bb3597306935741 : SUCCESS in 28m 03s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/321250dc3cd14e9eb220bd0215327bbd : SUCCESS in 2h 19m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9c9b35e063fe7fd2a2a2173b715bd95b44c0b81b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 21:50:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a43290a2814495ab33fa7203c270c80\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/15f8e88022c04ea386b2936f2dff4b78 : SUCCESS in 14m 05s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ead30dd0d25e49c7b3d55e6d32fce0a1 : SUCCESS in 20m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b14faefe74ba4e388b299c8c287e4d1c : SUCCESS in 53m 40s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/749449a6be124e56aff282525827591b : SUCCESS in 1h 12m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b670ac707fde4257bcdc76588579cfb4 : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0aabc63b70fc400fad87dda85a00f941 : SUCCESS in 2m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/03de847cf7774334b45d87ac8b68e332 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d47936c5b374ed9bfe80e3fe20f8b40 : SUCCESS in 4m 10s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7ce9f2ae61eb403b8cf6b1707132af7d : SUCCESS in 24m 25s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5a91358d45d143ecbf336cab4f653d52 : SUCCESS in 30m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/cc934b8ff0364fe1b9c579d44f97e30e : SUCCESS in 1h 03m 53s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6883ca5cb9914ec7994240e02fd3db29 : SUCCESS in 1h 41m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/50dbc2cb0e0747a4bda368153a4877dc : SUCCESS in 17m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/24aebfa912534674aca968de9ac0411e : SUCCESS in 1h 13m 53s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/546416e3358544e38e9800e382ef5eec : SUCCESS in 56m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/248938f6560a486b90dccd0a42c8fd39 : SUCCESS in 1h 23m 03s","accounts_in_message":[],"_revision_number":1},{"id":"2f654238a512a0e199c3b35ca113208638c516f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 21:50:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c5e3e528d3132f7fd16452fb8e1ef75cc9b012a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 21:51:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0265a31e82f4bfb8ed4090b1f717697\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7ac4d00f4c34694bb75337c28cded13 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","revisions":{"e5d19ffca5f5a795424f332dea1c1f123e7077d5":{"kind":"REWORK","_number":1,"created":"2024-07-16 15:54:53.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/45/924245/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/45/924245/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/924245/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/924245/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/924245/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/45/924245/1"}}},"commit":{"parents":[{"commit":"2f28e30912bd06350a4325d2606235197347458d","subject":"Merge \"fix: apple silicon architecture selection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2f28e30912bd06350a4325d2606235197347458d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:54:07.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:54:50.000000000","tz":120},"subject":"toolbox: Bump ansible-core to 2.17","message":"toolbox: Bump ansible-core to 2.17\n\nAnd also python3.11 to python3.12 on rpm distro side\n\nChange-Id: I000e2f2aacb011618a0b3c7b41fe286b3680614f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
