)]}'
{"id":"openstack%2Fkolla~924246","triplet_id":"openstack%2Fkolla~master~Ia2721dd7459a11324726910c99ea4c357047631e","project":"openstack/kolla","branch":"master","topic":"kolla_ansible_2_17","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-08-23 11:33:07.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-08-23 17:09:51.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-08-23 17:09:51.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-08-23 13:57:59.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Ia2721dd7459a11324726910c99ea4c357047631e","subject":"bifrost: bump Ansible version to 8 (rpm) and 9 (deb)","status":"MERGED","created":"2024-07-16 15:56:07.000000000","updated":"2024-08-24 15:53:01.000000000","submitted":"2024-08-23 17:09:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924246-kolla_ansible_2_17","meta_rev_id":"c270d6f2668da8d3d1ea4356825b9455e44d5f04","_number":924246,"virtual_id_number":924246,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-23 17:09:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-08-23 13:57:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-23 13:57:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-23 13:57:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2024-08-23 13:57:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-08-23 13:57:31.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-16 20:31:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-13 09:22:53.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2024-08-20 10:24:31.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-08-20 10:29:24.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2024-08-20 10:29:30.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2024-08-23 07:46:37.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"76b1dfa32fbee621c9e09dd7f70f31e349a1d248","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-16 15:56:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a85fd7448861e2514588fb85e5f9cba8e923dc84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-16 16:49:02.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"400b4908144fecb95950892a1e276f36aee6342d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 20:31:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/36af4f9c008e447baf21c32e31d1391e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d717c2e1ef3f4dcab8aa35bcc91bccce : RETRY_LIMIT Host unreachable in 24m 37s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/63ad0a5aeb9348e19cadefe7bf0116c9 : SUCCESS in 1h 01m 26s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/314e1a50fd7544e19c0d463e02ee43dc : FAILURE in 19m 53s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5e67f0045de341a89da0a1363002033d : FAILURE in 28m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/12e5dc867f634f5ab3d6b8e081ae3353 : SUCCESS in 55m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/16db486c054a43c68a91171161fbb943 : SUCCESS in 1h 20m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e91689b93414f2caaae610e6f0af650 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a667344090c4de5bda2dd8b16beb2fc : SUCCESS in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa5fa5d6e30843adb194cacf946e0689 : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fef8a8a498784c3fb80fa120527f923e : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8034eea23fff4c8b882b001319b2a888 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1c05f113f2f49a69c6decd471b51ebe : SUCCESS in 4m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/106fc5f743a446bc84f45656b23f5f73 : FAILURE in 26m 01s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b9ef84abe08246c3ab58d8825c9f304c : FAILURE in 29m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b807ed46975c40beb037e2f2fc370b46 : SUCCESS in 1h 04m 53s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f3b592cae22547449d28640b7ad8049f : SUCCESS in 1h 52m 17s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/d049ceaca0c045ef933de3d7b2c2aaa2 : FAILURE in 10m 26s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7eb1b9c0e8394be2b02c792cbb0e0411 : RETRY_LIMIT Host unreachable in 20m 04s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/679df940372c4cc1b28499806c3f556d : FAILURE in 51m 41s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/dfb0742607454889a16b4e1ce3e83553 : SUCCESS in 50m 26s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/eaaf8dea15d44c76aada1f06f4b70324 : SUCCESS in 1h 16m 50s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/5a92ab05027648e8879cd487f095c657 : FAILURE in 9m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"679088c381b614fc71f3de931aec4d8df4da1181","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 22:43:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93031298ebca4a6589834747c07a1a68\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2f08691ce06042cf94dbd83a11c0759c : SUCCESS in 27m 49s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1a20b747d6e8401288148f5137dbe530 : TIMED_OUT in 3h 02m 21s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/629b92f88f444b2e9b2c0a521f4cb8e6 : SUCCESS in 1h 11m 21s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e23173f334f34277bcf43b3ffa668bb6 : SUCCESS in 3m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/30b1421dbf944efea06c0c3b4dc6fbc1 : SUCCESS in 3m 21s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/11781ee8e46848d0a815593ee8c8a962 : SUCCESS in 42m 13s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ce10d7ba61964ada815d5c125626efe7 : SUCCESS in 35m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"99ba458b59fbafd9555a1e3512ae5e06297ab2f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-17 06:53:09.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"20b582cf3102b35ba5d0be9d49a9a66fe9f50d65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-17 06:56:28.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"e10f64b0fea1a7b10698c0a321d4f92a8b0a5ea2","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-17 06:56:41.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"cb0396fe078646583670b384bc9be34c8aa463bd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-17 06:58:04.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"0ea8b8da6eb5404331f2af665df7729eea96abed","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-17 07:56:10.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"52a15306ae56088e022f74d380e1554b0a891186","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 09:51:21.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bcca330cb102459895a5d9cfdc5da3e1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6c836c4127df41cc9fcef287043f45c9 : SUCCESS in 26m 07s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d4bae86e36864574888089cc4bfaa0a8 : SUCCESS in 1h 02m 46s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c45d37eb4c89484e8bb153def9b8797c : SUCCESS in 21m 20s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4cb2e7a54c844fac9563e1b5fe713e09 : SUCCESS in 28m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/de3b7eb07eac46bc82f01d0dc8524f62 : SUCCESS in 1h 10m 29s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/04d341ebe5f24ba499d63d55152d782b : SUCCESS in 1h 26m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5c8bfde01144e38bab3d1e44b546815 : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84fc41596f014be5b99fc0e30706b73e : SUCCESS in 2m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3de23571508e4e17988a964e8bb8175e : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b32f0fa9e33e46d9af07f5e11d3a63d0 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af08525e7f8a49018376d3a025d964c3 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cd9d3b392b943cfb5bf0a5f19ad306a : SUCCESS in 5m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f23b668b7c0c4cc0b8cf0235bcf693c6 : FAILURE in 34m 25s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d4b42963178e4600b487b91d4113a7d0 : FAILURE in 40m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a8cd321b5a694a41bc8dd07e69c5b399 : SUCCESS in 1h 05m 42s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4281b826d55849748f3b2d7df525e963 : SUCCESS in 1h 33m 03s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/037f5253078a4e06bdc17f5c13221cf9 : FAILURE in 26m 59s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ea8ea083a3ce493b9b4c4dd4e21edea3 : SUCCESS in 18m 48s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ec2f32744c2142f6adcdcd9ece686bd5 : SUCCESS in 50m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8ffc9802356f4835b71e990af0d1271d : SUCCESS in 57m 39s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/23d3dd62180e4f47928cd5f40ad3e682 : SUCCESS in 1h 18m 37s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/81272f74f7a44309a5510f04c7e4f1b4 : SUCCESS in 25m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d3f1d10644722edb53dd28ba956d6ba10309f2d0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-17 10:02:04.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"497e1fe8d01e49d8028292ec5cab0289457e2ee2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-17 10:05:48.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"95a0f9c706d206c1ac9b52f399d6ad03c3810c34","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-17 10:57:25.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"6d662a3e1e7f42e05436fa17e15c078069f9f9e6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-17 11:32:11.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"216f0db1f4f22c75ff5a7e430e0045f3bcbcca28","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-17 11:59:28.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"19dc0e4f3a6f1a9a82fec17f7eef7769bb84587a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 13:36:23.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0157283fc1ca476a8eab1e235ab44088\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2cad619c6cde4fe8b2e781d5e68b1d70 : FAILURE in 31m 01s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9c8d6e73ed594aa697c30a29196ad919 : SUCCESS in 57m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8c3b8f79a38744c7b7ba530f4498ac9b : SUCCESS in 16m 04s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f49d07cff89346408cc45d7175cf506c : SUCCESS in 19m 44s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3f315fc133a84839a771c64df5930555 : SUCCESS in 58m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/25b8151e44d549a1a7cf9c38bfbeaf10 : SUCCESS in 1h 29m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be890638311f48d69d2f59d1f8bbd241 : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9df5f8910e1f4fc99b3bd72379adbc75 : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9926e21626a34f14ab42830da2c14ad6 : SUCCESS in 2m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/04ae9dc2e5354d8cbb24924e194c0316 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afc42ce663b34fb2a945ad01fc4ce4ab : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cb39d86c60a42629ffef93cd3d8538a : SUCCESS in 4m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4e788f535cb842a78f280bc9486c9125 : FAILURE in 35m 25s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e96e1957371b4ad4ba805ac2b77b2309 : FAILURE in 43m 41s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/01dca8e082344b78b3059762009728d1 : SUCCESS in 1h 04m 34s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/45b0f102497249db83cb01e6ad25dd16 : SUCCESS in 1h 31m 37s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/9d019db3e2064fe48fb4f6a1d69e42ba : FAILURE in 27m 14s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8449ffb778504c028642d42c285e4c30 : SUCCESS in 18m 39s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e9a2cf4b244d46fda6b789ef4d64cdc0 : SUCCESS in 45m 24s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ffbc0bc2eb2d403484f1f56d0bb32810 : SUCCESS in 50m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1bcf87d26a304e1f94ae1dde73bce767 : SUCCESS in 1h 15m 14s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/ae34599bb13d4306a264a3fe84870021 : SUCCESS in 25m 49s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"b9f7333de9a41e1a94134b128de95566731b459a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 15:06:50.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b1534a922f947f393a248de7b901648\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/84829fb84f224c8e8e68b32270bd855b : TIMED_OUT in 3h 04m 38s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/135f01d95f664122b94a9c3459bd48ae : SUCCESS in 34m 03s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7cad614a991d4445b425249e653ae921 : TIMED_OUT in 3h 02m 18s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5516cd82e65a457cb2d730f3754409d9 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7c60e3b0419a44db9b2ae9832ac7db8a : SUCCESS in 4m 19s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e392ada17ea94ac89a5838a486a4c65f : SUCCESS in 32m 48s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9a7ed692a29e460db908a467ecf98991 : TIMED_OUT in 3h 03m 57s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d87bddd2e5a8ea0e36f5722e6f802c36f421fd67","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-17 18:00:30.000000000","message":"Patch Set 12: Published edit on patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"d54c3cba5222225e6cc9b18809ea93ab2cbf2b53","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-17 18:35:30.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"50e7c6dad3dbda44d584dce0d3f5ebf3adf00f9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 20:16:58.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dbd096b9d0b5412abfa07b6bf554c242\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0119bbeb45cc4724a3878ea929122941 : SUCCESS in 26m 43s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/bce9d2de1afb41daac5abf4015e5c559 : SUCCESS in 51m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e5a67fed825444e0bba7d6490f7b774f : SUCCESS in 23m 55s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/eaae4f1ec3494f09b523428d942b03e2 : SUCCESS in 27m 38s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/707276e7f888404ab239b6cb55d56770 : SUCCESS in 1h 00m 49s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8bad6fe4fd7841d788077ddfeb25d942 : SUCCESS in 1h 30m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5728fb8d253a4566b8bd8316e44f16d4 : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3fefc1da58d42298129323c3662e46f : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f27f20d3b8348329365236593dcacc8 : SUCCESS in 3m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d0bdd9ebb0d6420c8a3016073dbf381e : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/955980945ae74e39a3d046cc985738c5 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6b77a111e964f6ea808f5a937f2f08e : SUCCESS in 5m 22s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9f14d2c5faed42eca953b97293405c1c : FAILURE in 30m 27s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/42b19efa21c44e83abc7ab89e610a4c5 : FAILURE in 45m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/071e280b9cea4a0cbd19ca391ed27777 : SUCCESS in 56m 42s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4cb3ac6e712c48a0b3710fb0c3c68647 : SUCCESS in 1h 28m 56s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/4ea03579f86742fb925e6b5e64b767f5 : FAILURE in 21m 12s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0d1807a8fee547c18c99ac469b1d0efd : SUCCESS in 20m 54s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/091aa90f00df4da4bc18ba543111f180 : SUCCESS in 58m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/07345d120a174b22b8b94162cbaf864c : SUCCESS in 1h 00m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b539c04f7ef14f458f3065afaae73030 : SUCCESS in 1h 04m 37s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/70a2bb88e90b49aeafc908cd7c556640 : SUCCESS in 24m 21s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"195686ea0a9efacd861bd26c0bcf0d7253c67ccc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-17 20:31:03.000000000","message":"Patch Set 14: Published edit on patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"34c7367eb174e5656327005dfb99e7cffd56d3f9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-17 20:31:31.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"4ff3f2e18790cbea8a52095ddce866a408411bb9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-17 20:31:48.000000000","message":"Patch Set 16: Published edit on patch set 15.","accounts_in_message":[],"_revision_number":16},{"id":"32bbe3469b26833015216e8c1604c28f514ec71d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 22:15:21.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ef23b0239794436a8a7b7adcc95a67d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b9dab0d98c2349a8b1bbc195fb2cbde0 : SUCCESS in 50m 08s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ec42545a1fba484ea3a9b364c204dfb5 : SUCCESS in 47m 28s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/06a715c56cbd4da281c61c5d7c1a8f13 : SUCCESS in 1h 32m 18s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/664858d032ef4b4197d6594d5be3fa1d : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c848b476ea0249238b2ee373228be5bc : SUCCESS in 3m 26s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e80e7f75d4bb449483be24377a29d09b : SUCCESS in 50m 27s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/fadda9642a36469e8b6a067472891422 : SUCCESS in 44m 41s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"8026b6be68a0ed3c20798470d40172b19edabd73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 22:51:13.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/de747c3e34b642198d62115b029a7d77\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/fd5b406796d24ec7a2ec7fdd35e4b779 : FAILURE in 25m 46s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a8410c07e95b4a80955f773c1ccb6946 : SUCCESS in 1h 18m 05s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b516ed9d9e974fef8c7238174d09e894 : FAILURE in 21m 14s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e8c7096d15174ce8b7c3d9cdd636b4eb : FAILURE in 31m 26s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d9f68250858343bea8d9608bf32f31a6 : SUCCESS in 1h 05m 17s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1585c638f8f1492599135bd3315d7a72 : SUCCESS in 1h 35m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a280c9b66a7469e9292e5aafa723f9d : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33177a2fe5e94d0a9a64ed5c978eb2c1 : SUCCESS in 2m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d40f6913322f406d89d81640067f3ba7 : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5ec4c187ab4445868f977e2256d40cc7 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03b39b2a8a5b4c25a0756a0c5d86bca1 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b98c9b00328043149010bbf2703572a7 : SUCCESS in 4m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f23aaab6a9304380b1cf3cbb10e8cf89 : FAILURE in 27m 15s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/80a13513c226416d9664862061a47042 : FAILURE in 35m 34s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a0ced8ff47064e378ef21f7df02f1e5c : SUCCESS in 1h 05m 44s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/59cc7a5168814b279d0d75e87dfab0dd : SUCCESS in 1h 45m 14s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/377026b01fd242fba0618c7c76726ae4 : FAILURE in 16m 40s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/943c9f1e7f0545baa7a246f9b5ee5647 : FAILURE in 17m 56s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/aae90496506b4de080ee291332618493 : FAILURE in 52m 29s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a6a8f747a81c4e6d9cd7d8dda153ae90 : SUCCESS in 1h 07m 49s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b3e8eca608b6427a9c3b6ae7c2807aab : SUCCESS in 1h 19m 20s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/226f3c0fc38b4e7d9a80ef943f5151e8 : FAILURE in 13m 46s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"940aca2d3671b357da61b84ef043db340ba7ed6f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-18 05:13:52.000000000","message":"Patch Set 17: Published edit on patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"c4ebd664c5a290a8d92370699119ed343f71d2df","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-18 05:17:14.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"b28df83fac555d6fd74261d51b51b4739f417753","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 07:03:31.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6646d43e178c43a892e57c6990741a77\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a27643e60d914bf9aa70b9112abd6710 : FAILURE in 18m 06s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e9a0a55ab2c7470bb2de02cd90f210d4 : SUCCESS in 1h 07m 56s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1942df208dfa4c838f3c4e864f67cb62 : FAILURE in 13m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2eea8bbdfd594589bf35d88be854f952 : FAILURE in 19m 43s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/dd8ab3c4617d42d4b0fe911a05ad6ee9 : SUCCESS in 51m 33s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3f9cf5062acc43ddafbedbf02ae65093 : SUCCESS in 1h 14m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25ac2312909943efbb3bcff63062cee0 : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8441638f10254a7397f802a759a9fd01 : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d5e77ff74d364f6ea5fc664ff9642a7c : SUCCESS in 2m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/655c85b7ba67431f90ed404600fc2edb : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f8b5a72a75c4c489eea2203c50d2d10 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00535e56f0954996a369a22be6296b36 : SUCCESS in 4m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/412e8f8bdb1543299823d14a46b9b96c : FAILURE in 28m 10s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/24fa727ef2c34c6a9b1634a8a3025a30 : FAILURE in 40m 34s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/baef96a21b884f7fb60692d9174f2f2c : SUCCESS in 1h 07m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/fabe57ba7ef54809859a389d7414796c : SUCCESS in 1h 39m 15s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/0fd211ceaf0b41e281e3c8689dcd6f35 : FAILURE in 15m 31s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5b1127df88c74e149f9f6f585afc3f1f : FAILURE in 15m 06s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/13071a64a7914604bc26c815c11f670a : FAILURE in 41m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e6314aa01bdd408fa9988ef860161796 : SUCCESS in 44m 32s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f46fef5312894a5f85eadef4cfca1183 : SUCCESS in 1h 06m 18s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/89dddbb480ce44ba82058b6edad72da5 : FAILURE in 8m 23s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"02f62415ff59249c64c18bff2962ad687136cb4e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-18 07:07:13.000000000","message":"Patch Set 19: Published edit on patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"73120e98f944dbfda9804b75204c6a90584870f2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-18 08:13:50.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"54551ba858e4b16a42e1e512fe0648e7765bf564","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 10:07:01.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f6fc43bcf534e6ea8a92d4ec1288b47\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5524c2a4f0654a178949d4c03247f11b : FAILURE in 25m 39s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/11ffd61809744cf7951853440770634b : SUCCESS in 1h 09m 46s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b3a9fb7dbd86465389123ed36b634be3 : SUCCESS in 19m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/cb0542b316a847569bc199b6320229a9 : SUCCESS in 22m 50s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a139ab26cb324e968436e419100b8f9f : SUCCESS in 1h 05m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/dc87d16f27b7407e9b32654965ca3891 : SUCCESS in 1h 16m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/179560baf05d4e119f2f52420a727060 : SUCCESS in 2m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ea559f43dc446e88cce109807362775 : SUCCESS in 2m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f446482902e44ba4b295bff122c23b85 : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4c614ac216694c72ab0b463cee8f56eb : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7565ca8089ad4c8bbff21580f12d55dd : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b1e2454a5914d75a23d78250168754c : SUCCESS in 4m 09s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/450c34967ed44eae9a7176f171839c00 : FAILURE in 38m 46s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/48cf9f0d7a0243eb8094241488f062cc : FAILURE in 43m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e7007c4d8a8e4e8fb946bb76fc952b27 : SUCCESS in 1h 08m 10s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/614e1e9da1424f7b8065f42d19590d66 : SUCCESS in 1h 45m 11s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/78f00f37ae53430ea0e4217fea18bb39 : FAILURE in 30m 00s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/bc942358ef764b0cb886d3cc99eb9c52 : SUCCESS in 19m 29s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3feee7c37bf5406f86894fbbcfd2a9c4 : SUCCESS in 1h 03m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/08bfb026c0c346c083a91ca0701dc553 : SUCCESS in 52m 59s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/49508f8ddeaa4bad86a4ac0e4d41725c : SUCCESS in 1h 24m 35s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/abdc216f90d14177b19dd6a4c18b6e25 : FAILURE in 19m 45s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"b222f73f3a001ef6f92784e676ffe6c9ccb96b43","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 11:20:50.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c704d52c3ce8460586604ce6261e7b3c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/78aaba86d3f04d20abf3b8dd67e03d0e : SUCCESS in 36m 32s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/04d8d7e7b70a4bd09a50c06e1e8485e6 : SUCCESS in 36m 45s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/34afb30279764be9b16ea86dd0dd7e22 : SUCCESS in 1h 18m 46s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/059651adeeeb4e26926586a6b4728f1c : SUCCESS in 2m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a25eff4065504121ada58b5413a3609c : SUCCESS in 4m 19s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/fa077339e0a8436ea3ddc1a914eeb1d6 : TIMED_OUT in 3h 04m 46s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/af640273a66d4546a9e834e87ca98f42 : TIMED_OUT in 3h 03m 03s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"66806eb7981426bb85257323d3b987b5453f928f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-22 14:06:48.000000000","message":"Patch Set 21: Published edit on patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"61b9776b2efbe12fb20ecacb138c5fa13fb05b7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 15:45:53.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/68eb951ffb0944ce83e1576739b01e84\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9b7cccb91afd4d888af18cb19ed0effd : FAILURE in 31m 44s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/beba4c5a827b49938d05a1f31299c5bd : SUCCESS in 1h 10m 13s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8c27a3b1fa194d69ab5ef6615218530d : SUCCESS in 22m 04s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f3f4efbb24b64a6bb83ef8402a91d856 : SUCCESS in 28m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8c4963850e3a48d38438533eb1096413 : SUCCESS in 1h 03m 02s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/967484b90e0b40bb91c6d2d6836488ee : SUCCESS in 1h 31m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c28a379d4eee464491680bdd47972641 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d23bd36db9c48b499999e3d94d39400 : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8d4336cb162047a7a9a253df8d293cab : SUCCESS in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b40208d1707c49edbc3c09a704671b00 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e50184742414d6f981aaf727b052364 : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd200168e3fd432bbc6006ac4aca3be9 : SUCCESS in 5m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9bf9a080698940a0ac718b2ef6872f23 : FAILURE in 31m 39s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b39b9c93f459409585a56394ca7c3a7e : FAILURE in 45m 52s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/789382ca08ae4023aba87973989a94bf : SUCCESS in 1h 13m 52s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e6a13cee45f245db9927ecb0afb69c27 : SUCCESS in 1h 33m 36s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/6fbb4ad0759c41538011ad3bb908377d : FAILURE in 28m 50s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1663754a514f4d60ae806c64a6d931f9 : SUCCESS in 23m 43s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/08eefbed30e442fe9418862be68c0b1e : SUCCESS in 52m 58s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/51ea2a16179c45bfbc4874a5d3a9f411 : SUCCESS in 52m 36s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e5d64b7ac93c434c9684945bb9a6bf50 : SUCCESS in 1h 18m 47s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/34c13042c6ac4a74b0b6d37d13c3545b : FAILURE in 19m 09s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"9d6e4b9e5d07a5e9fcf34ea7c44ac0c1b2b7af42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 16:44:19.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de561faaa90544f2b6f00068a7fad14c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/67130eba4eee4851b2950d64c50cce3c : SUCCESS in 27m 44s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bac91b2028d14a92ac5851695114c081 : SUCCESS in 27m 07s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c2b649e535f84ffc91c675a6c6fe9e1c : SUCCESS in 1h 19m 15s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/196d476774df46bcb159bd80d888f20d : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cd007a3b3c5e420ca75af28fc837208b : SUCCESS in 3m 51s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a6e14d4e6888417882a6e58e07104850 : SUCCESS in 2h 34m 35s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b68ed96265d246fd817d0651d5dbf357 : SUCCESS in 25m 04s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"88aff84988681dd0c211b876253116c831c85250","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-23 07:29:37.000000000","message":"Patch Set 22: Published edit on patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"b52165024f059c779506497e910c137a2185fed8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 09:14:53.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca93baea56db4d15871ffc5415233f78\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b075035a33ce412eac8ba9b4e6c46a76 : FAILURE in 23m 47s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a5c31b5bf48e4d4a8958e62adc241de9 : SUCCESS in 1h 12m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/42638631ffe042b2853dd50e5eef0dfc : FAILURE in 16m 28s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8d1bbc7a86624dc3be56768cc60a269c : FAILURE in 23m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/32ceb98abb104b4c83fb0b903a3775d3 : FAILURE in 7m 47s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fe1f7c78ad5548cdb3b525caf84e4b50 : SUCCESS in 1h 29m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e17a7986f2794cae839c9170a3c75d9e : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/edd109a3f20245df8ba56ff15291c748 : SUCCESS in 3m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/56a9f5bf8085491baed8ba136a24fe8d : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/978dd4ed7ad349f6b9660d0c93e2c221 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb3b6d1e5e864d139640c76da919034b : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de9022c3295d4782902521c48f635bbf : SUCCESS in 5m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7fb82c7c17854787b77b35134b05f54a : FAILURE in 22m 09s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/fd77c4e0b2ec409aa1eabe5aa8206077 : FAILURE in 30m 59s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/58192053009743a0aff2ad2dfd61140f : SUCCESS in 1h 06m 39s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/aed7f0b0f04647f9a338f4d28589e344 : SUCCESS in 1h 43m 09s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/23ad33f3a37d4f609340eafed7a0c798 : FAILURE in 15m 50s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/266f16c91c924b4baff90f5f1a6822e7 : FAILURE in 20m 10s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/907ceb9f9f1446d19b0d2569620cdfec : FAILURE in 52m 45s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/71036764ffdf4c00a4e33ee64787e8ca : SUCCESS in 45m 59s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6278dfc7f48a499580a85ee199efabeb : SUCCESS in 1h 33m 42s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/f8623642efb548f6aaabfecba980ea5d : FAILURE in 11m 07s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"e6dfea640afc3a4fbf659dc71ebf2e4fb261db32","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-23 09:35:37.000000000","message":"Patch Set 23: Published edit on patch set 22.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"4ce43c29f587d058bd8ea6e0e33fa29452cd5fe8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-23 09:36:01.000000000","message":"Patch Set 24: Published edit on patch set 23.","accounts_in_message":[],"_revision_number":24},{"id":"e021f3c71e0f174d38270c4970ef5e10e643a35b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 11:31:51.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0ff63628dbd4a48aaa56861ca7aef89\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b940e85feb3e41fab6da62532e2b3887 : FAILURE in 26m 48s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/165c8602e966476ab76e603727e37b64 : SUCCESS in 1h 02m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/17a8833bd15849ae8da2227d6e78699e : SUCCESS in 18m 57s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8cfde87c43cf4ed39e14fdd76f24e16c : SUCCESS in 23m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7aa5d18ba6d7422ea30ee5ad617b2f29 : SUCCESS in 59m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6b012c857ef143e8b8022d2a4646d536 : SUCCESS in 1h 27m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cba4dab1b3f47efb5615db9c4133204 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/daffcb3bceee43adae576f9183af1388 : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b4a50ee7e9594431a76f2221e9cdd3cd : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6150fcae68f54c8799d696834c2bf99e : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb23245ffd1f46d4b8f61117b51995c7 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2380ec3939a44743a98a27354f25bc5e : SUCCESS in 6m 53s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0c7240aa0348489d8a3b1d955ad2c416 : FAILURE in 24m 00s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0d617532955e43c2805cae98cca43634 : FAILURE in 34m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/bec118cf52e64a33860d3e531d8c5d2a : SUCCESS in 1h 07m 30s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/121191cbe24d47f289fbdaa2083555aa : SUCCESS in 1h 39m 17s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/884118dc9c6f4e6e918b2e6f251411d7 : FAILURE in 14m 06s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/14fa37fc65d84a32b8778f4346170f0d : SUCCESS in 25m 59s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5cbc42aec1bc4f528d84e8f6b3981fec : SUCCESS in 57m 56s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5fec757f8933465f941702d86e214a14 : SUCCESS in 51m 33s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5276de19f2e142018db6802510bb225f : SUCCESS in 1h 11m 42s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/3f40f35190f149b6bc4fd969c096c864 : SUCCESS in 28m 44s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"eb8f854e337290dabe06555af3e427a4baa80ec1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-23 11:47:36.000000000","message":"Patch Set 25: Published edit on patch set 24.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"66796af75b71b7c361214ded7d22a03c1ae7016c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 13:28:18.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/25b7b76f072e4f6d8379ce159a6efbe5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6d40925da38d4497aa07cb574c2309ab : FAILURE in 27m 24s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/af244667e5e54b9f9038cbccb1003738 : SUCCESS in 1h 07m 51s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b1ada692046c421ba7730e52f9132c55 : SUCCESS in 20m 15s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6e108dd5ba15400ca471025d4471aae1 : SUCCESS in 29m 16s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1db66eb150f745d89dd31b3294afcea8 : SUCCESS in 1h 12m 58s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7da087e5a97943908203232d605bf359 : SUCCESS in 1h 33m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f72e20ec3e34cc188c400812e4d648d : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/936d51b03f6945ad88041d7f14f7ec93 : SUCCESS in 3m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7cef739855894182b7ffe4f2c519ae00 : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/64cefe91dd8a4c53b36cf917f7b0a135 : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/697d0f742c654ed1a8439b71be3f0bbc : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ffbff0aee4f14ba08244f22efa20b1d1 : SUCCESS in 4m 36s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/094911f1d52e4b70b73fe0abf64bedcb : FAILURE in 29m 53s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/850ef176acc64a78b7c760601ce7cddd : FAILURE in 31m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/570e5b66ab7148c0aa13ad14331fe02c : SUCCESS in 1h 02m 25s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/df8d5c7a747c4c71bf849d407091d800 : SUCCESS in 1h 26m 45s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/05ecc7fc644043fe926137bd7a0645da : FAILURE in 15m 30s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/44e8e399d43d48e3b611b9895a5d63d3 : SUCCESS in 32m 06s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/11b5cf3bc3f64012bc2e0b83a1b96f9a : SUCCESS in 56m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7a42d34033ea428fbf739e2d32f03979 : SUCCESS in 55m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/feae8b52126643e29b8ba03872e971b7 : FAILURE in 1h 21m 45s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/6757dca4a2f345b4afcabf2e367da09d : SUCCESS in 26m 08s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"742b7ca2ba69996c2638041a2952c99828ba26db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 14:54:08.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50c4aa85749041bcaa1e8f0bb1892ece\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3bf5549acf524099972c97055da48f83 : SUCCESS in 25m 01s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d192fa7320674fa1902f657d7b994acf : SUCCESS in 19m 07s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1a06e70ecd9548d9a45f1f9fb92355c9 : SUCCESS in 1h 07m 46s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2187f92bcab64b97b9bc2491184b3ff0 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/672d75762d964debb076de728d7d5af9 : SUCCESS in 2m 57s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d6eec7f4ecc941dd937310e4191872d4 : TIMED_OUT in 3h 04m 06s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/29ca6d16f5884e0a832b0ce19b5b7c9c : SUCCESS in 16m 10s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"9a50e66fb06ba432c35d82f7ac5c51aa6f7e6925","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-23 19:15:11.000000000","message":"Patch Set 26: Published edit on patch set 25.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":26},{"id":"0f8d1ba76be3c22c9a85288fc8f93a98e0861d0f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-23 19:58:46.000000000","message":"Patch Set 27: Published edit on patch set 26.","accounts_in_message":[],"_revision_number":27},{"id":"264f83ccb47cee75642cf42867ede87100df53e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 21:38:40.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a3cd2afce864f6484b5f3c55953f5a1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/98feaa25ae6a4476b1344ca07c63e3ef : SUCCESS in 24m 56s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/86f2064182824d598d4698ddcf02b336 : SUCCESS in 1h 11m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f0c5f26b7c16491c815337ecd919c5ab : SUCCESS in 20m 15s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/848b232e0d1c4ac69cc2caa5aabadf0a : SUCCESS in 28m 08s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c83eba28f9874ff0aa171d89b07942f0 : SUCCESS in 55m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2cefe640f29c444fb34f70378c4b2c51 : SUCCESS in 1h 14m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0549948d48464fb3beba08a7ea222bfd : SUCCESS in 2m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e2a875793f04149befaf980ea6ce8b0 : SUCCESS in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/35fabf112c77454d8e8d3fbce595b24d : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/482e70abf12a4802b73a5dd9f86dbda3 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3687ca3134f426584a54cd0100a3d31 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a55b292636ae4f30b06c88df4c6ea7af : SUCCESS in 3m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/fdcf404b3c32443cb5bba2dd1cb16232 : FAILURE in 48m 58s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c8c8cef2e0034bca9883d023fbece7be : FAILURE in 53m 45s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/cb7fff586dcd4242bac91d7b0580f5a4 : SUCCESS in 1h 16m 47s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/409f273da9fc40749b38c863e3826238 : SUCCESS in 1h 34m 40s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/f4332a29293143b0b5c2425048143340 : FAILURE in 48m 30s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/67f7a8c840c246af856fa35e40a82bb7 : SUCCESS in 18m 16s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/95f9d0d6d3454e63921124fca546a28c : SUCCESS in 48m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/af58046107ea46cda52cfa20219812bb : SUCCESS in 58m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a1c72b2c93c04198acfc5218513129db : SUCCESS in 1h 26m 25s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/068a8437e9ba459e9b31a80ad6148008 : SUCCESS in 28m 29s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"8447107e305d5a75f5e862c5441efff77875ba8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 22:43:57.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e58ce9ef5784e319462031d4d620e85\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c353e20072f94054a02a17f5470b6977 : SUCCESS in 2h 42m 51s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ddd0643d0ad84c40b49660b9490f362d : SUCCESS in 2h 37m 32s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f04d729c8d5d4cd1a0ef1e4a76dd4008 : SUCCESS in 2h 07m 46s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f15a77e5a6a3420ab5ad7c71ce464b43 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e1d7fe0ecd94495cba2df095e9096e24 : SUCCESS in 2m 56s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0c7566a0a08b485696ceaed4ea0b81e4 : SUCCESS in 29m 37s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c6cea7607322460e802afa6e294843f8 : SUCCESS in 2h 35m 06s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"43c793bf475392163f84d33da7fb3a45237fd322","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-24 05:54:56.000000000","message":"Patch Set 28: Published edit on patch set 27.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":28},{"id":"b6a2c40b8c764aa52cd1ad58751bc379ae466b20","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-24 05:55:09.000000000","message":"Patch Set 29: Published edit on patch set 28.","accounts_in_message":[],"_revision_number":29},{"id":"ed3d82e1c7caf55ac94d9f34a6517a8ad4d699fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 07:51:15.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/22bcdafe28c748d3ad32b33ef39e7ded\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/350abb544c854d01b18bd789ae7aaae1 : FAILURE in 22m 34s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c567e2f30d594eca8bf84501797d6ec3 : SUCCESS in 50m 53s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9fc60e4d15074468bfedb3053403468b : SUCCESS in 13m 39s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/167f5d9792a244f3ba0a830af4b0c845 : SUCCESS in 27m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ce9afc1363b048c7bcceeb871aaebe81 : SUCCESS in 1h 03m 57s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/73bc5448eb3645339770e32f819658bc : SUCCESS in 1h 10m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08432be4b1e9463c84678acf3a397c5d : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7717b105267d4ca2a0c890484a912293 : SUCCESS in 2m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/338636486a4543e896b5d21311251f95 : SUCCESS in 2m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7f7a09daef814277a3d5c974397b04f6 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fb2d51dd5b24265860021754eb174d2 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26bf2d5b910c4be3866c4a334e83cee0 : SUCCESS in 3m 42s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1d9e75c70732497cbce7766d85bffbfb : FAILURE in 40m 42s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a8a0229b35d040259cf8404bd2ce3f31 : FAILURE in 45m 05s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1cb4ea5cc74f4d8ba948aeb5eb8fb968 : SUCCESS in 1h 07m 53s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/20940edd5d9f41a6b2858fcbca94bc8e : SUCCESS in 1h 45m 52s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/31be200925974ed6b3acfea3892eb5a3 : FAILURE in 27m 29s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1cffcd1d0d7a45daa11ce1ab98373afb : SUCCESS in 17m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ac89a301b3da4b12b8808104045d48a0 : SUCCESS in 44m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1cc66bef7d6e42a88ea3b79b8682a338 : SUCCESS in 46m 34s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1c061b1b2e384e168de6020cf0cc9536 : SUCCESS in 1h 09m 12s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/eeecb4596778479d9369b4b1c0f76f4d : SUCCESS in 23m 30s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"3eff4f63e7cd7f55a8f695e10daee49ab6cdde27","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 08:16:13.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca952af0e5c6493cb9bb796cb53fa680\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6875a2e89a4b40e2ba25131bc452d4e3 : SUCCESS in 24m 15s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/36f8eac7eae94f799cb487fe9306e823 : SUCCESS in 2h 01m 05s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1400113dfc614794b881835b5a9c9155 : SUCCESS in 57m 02s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/90295a1783d2463da21896244551fec0 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/49c6cbc5dc344723a159e3617514d725 : SUCCESS in 2m 38s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/2cdf0b577e1248f6867fec4ac7e065b3 : SUCCESS in 2h 18m 07s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/86407678f250476a87102d5c0e9704c9 : SUCCESS in 16m 35s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"ebf3747b960baa5c8961db382704da56390af10f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-24 08:57:52.000000000","message":"Patch Set 30: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":30},{"id":"94d072755c0de81ba658138490274e48ebcc5410","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 09:15:09.000000000","message":"Patch Set 30:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/924828 is needed.","accounts_in_message":[],"_revision_number":30},{"id":"c1c3a40f549c55fdbe2233814e6117ac2608c1a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 09:15:15.000000000","message":"Patch Set 30: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/924828 is needed.","accounts_in_message":[],"_revision_number":30},{"id":"d9661b7fcbd9b64732c1d47c7eecb75313900e1d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-24 09:23:32.000000000","message":"Patch Set 30:\n\n(1 comment)","accounts_in_message":[],"_revision_number":30},{"id":"773729db35aadb5ee1467a9b9ccb1441d81711d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 11:04:02.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5e5d3791796471da4d6ed8415e90df7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d6f06dd6a7a54149874608737a920269 : SUCCESS in 27m 26s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/41fd32248b5542a091ceca8a8cb51e11 : SUCCESS in 1h 05m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ee4079a7c5f04d11ad2d67c03b174fa6 : SUCCESS in 17m 33s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/562d7fc682c94315bc54a2360217df18 : SUCCESS in 28m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b190db3a89b44f319baf36055da8e120 : SUCCESS in 59m 47s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7b9abe14022f4d2ca5d5a100ce0018e3 : SUCCESS in 1h 32m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/912d03ece1dd4b3982dbbd842a0ae7ad : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba72a162621443f0a97f3b892d053553 : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c79a67a42bd04965abe43d9e29c75787 : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/67e1e46ad04c4b1e84ca552ada176b9f : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bef436d75c424e47b70da0b025dc8df9 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/402914e301f749d5bee1114919b8464e : SUCCESS in 4m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/addb8c9ffb7f4cf3a967d8433db95318 : SUCCESS in 27m 00s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e557622c675e417c9ed59730cbdf2dfe : SUCCESS in 33m 52s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2c1f3c5769d6411fb8cded8a900bd25a : SUCCESS in 1h 03m 11s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5e655e568056497d8cd47dedbc27ff97 : SUCCESS in 1h 29m 10s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/1b301ed514ec43849a0f365b3aaef1b3 : FAILURE in 27m 39s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d7ee3efbd16649ea9094db2c3fdfdd16 : SUCCESS in 17m 47s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d23bbc57b1ad480599b631ca93a6981f : SUCCESS in 52m 29s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/13c3ff9470f84e3890b73ce5907ed684 : SUCCESS in 59m 11s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cc4656fb6458435d820dcf008a669075 : SUCCESS in 1h 15m 51s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/bb5bd64ca79b42b3a8b6196ba8a089d6 : SUCCESS in 28m 12s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"3546099afe96f8414ee42d2857f8e7a0a9b7602a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 12:31:29.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5986bdce0c444686bfe0999f09601f73\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8d238db001704e4ca5762cdae079d715 : SUCCESS in 24m 16s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cfc4da9d2c4043319c3129bb40620886 : TIMED_OUT in 3h 03m 11s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e2e74af904b34d7e9f3061420504051f : SUCCESS in 1h 13m 39s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8cfc619431a34f12a6d7fb791058070c : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b50745db1f1948ca9d8b70c47ccac9de : SUCCESS in 2m 55s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d5150846de0243028f2a5a08625ad8f4 : TIMED_OUT in 3h 04m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/13561ff1a4ca4de3bf1d18d7846631c3 : SUCCESS in 19m 32s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"bc2fd1eea14ee8712847274386629e46d3f1358f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-07 08:13:55.000000000","message":"Patch Set 31: Patch Set 30 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":31},{"id":"d957d6d719450d0aec80e384b4511ed6f7c15c62","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-07 08:13:58.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":31},{"id":"ec0a1cf110f41a093b29fceebd4c56a2b687d7fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 09:10:04.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d671f8dd1160485e813e00102123583e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/bb04b2f8d1a84a80aa43c474236582e2 : SUCCESS in 25m 10s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1511642e9380471d9f88e1d2ddeee435 : FAILURE in 46m 31s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f7a63bfea4d64ad1b0c39c3b8aec6c62 : SUCCESS in 18m 17s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d3e668e67f904f7bbcfdf14c5c1d9a57 : SUCCESS in 27m 58s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/96d8435554f847b0a0966279f650724d : FAILURE in 42m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/65e5af0fac9846a9ae71309c6ff0ad1e : FAILURE in 44m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a92570280043469f98611e95f76dcb27 : SUCCESS in 2m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92d4f8e7d24a48798fb1a71de9c13e7c : SUCCESS in 2m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/206fcd3e90434eeaac6175f99847a56f : SUCCESS in 2m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e1b76be8de04a9c9ef86873dda67e07 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ca948e91d3743969be7bd5b5dbb770a : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74ca04955bbd4ed6911c613cf357242d : SUCCESS in 3m 46s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f335fc930e7f4a52b1ca25c279586217 : SUCCESS in 28m 11s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1854ec7f379f4ea197fc35b9f751e872 : SUCCESS in 33m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c1f8070f335f4cf69ae36414bdcefe05 : FAILURE in 47m 50s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f03dc96031e84321b02d4f89de3def9d : FAILURE in 43m 14s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/801579ff0394424dbf024cb340a187d0 : FAILURE in 26m 07s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/96519ee5ddda433e91d3348639cf2324 : SUCCESS in 23m 31s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a23f5a8cf9b047e6bf0d45cdeeb22ada : SUCCESS in 55m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/08b31bf016fa42eaa2e5bd0a9eab0c31 : FAILURE in 37m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e9cc47f5022942779f9a070ecb52dfb0 : FAILURE in 33m 30s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/d3eef2ff31e94f42bf3bac320692627d : SUCCESS in 23m 43s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"b490d3a3bcca093ca565ba629b1edcac57df615d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 11:40:34.000000000","message":"Patch Set 31:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25960ee66e7f4e5e9f851562c1fb6318\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c2bd92d5b7104c90b83d14dfcbc59945 : TIMED_OUT in 3h 03m 38s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/713534a1ebe848c793a37ce28a75b0a6 : TIMED_OUT in 3h 02m 42s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f459d575b6314d72b4c93954dd782da0 : FAILURE in 1h 40m 31s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/75065ca9319f4751bf455a99703464a5 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a269048bfdd94f1b9bb099cfe36c6f32 : SUCCESS in 4m 00s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/bb2719ce988148c5a7cfca751ca4f0cb : TIMED_OUT in 3h 04m 46s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4b7634c36f494288b6a90d76038fc7f1 : TIMED_OUT in 3h 02m 58s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"46794766d4fcf6363033bf22b3eaab0b9b743faf","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-08 11:34:02.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"2b9da3bc130a219f41de5cff4dc7771bee3cf427","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-08 13:14:47.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbd75ae1f09e4d329be09d70657931cb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a2c86fc74a3944e9af7bb19589c84d96 : SUCCESS in 23m 37s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/efc46cc7576743ffb3d7ccf06d1725ad : SUCCESS in 1h 02m 54s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/32613c04a0db495b8d4659385431bfc1 : SUCCESS in 19m 27s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/739ea6431f9e4eb39b1a000dfe5aea0b : SUCCESS in 24m 05s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/87ee387ac20f4fc08d6c40adac65257f : SUCCESS in 55m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0c64522c49de478899f47fbe26fb30eb : SUCCESS in 1h 35m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97e746124b31421896e8884ef074f982 : SUCCESS in 2m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/779da07e94374afca59d371db8efc801 : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea9e6666a31a4f6ea474638f0e021ec8 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8704bd50da7d437aa673f33aa4b11d97 : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1fb658bbf57d4f6c98205d57d437cf9e : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/171acdb37bd64de089520711e0549049 : SUCCESS in 3m 52s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/273127c5fcc14c21ae7e512026e472ef : SUCCESS in 30m 11s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ed5f17e45abc4fa98944f5ae96e18752 : SUCCESS in 40m 45s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/27594dd72edf43df878c2d374b3665fd : SUCCESS in 1h 02m 51s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e3cc6a683265424191b8480b763c1632 : SUCCESS in 1h 34m 24s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/a21deb001f964fa7bc6a0dadfd9e34ac : FAILURE in 31m 32s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e8bab545fe354a608e686d7594866159 : SUCCESS in 23m 29s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2df2e336d37d46639f192d044033420b : SUCCESS in 54m 01s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/90d624d3a7424f62b6bf4f34d70c3d85 : SUCCESS in 52m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6fb9ad0f78b14127bbbe7e9353b954a6 : SUCCESS in 1h 26m 17s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/30ffc8f8597a43479c08f20a24138d78 : SUCCESS in 25m 10s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"b12bfb00d6c75e6a1e494155610d4735aac67a31","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-08 13:33:08.000000000","message":"Patch Set 31: Review-Priority+1","accounts_in_message":[],"_revision_number":31},{"id":"66f86f331b734445d99ab63c5f2f9bf3d43210d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-08 16:46:55.000000000","message":"Patch Set 31:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31fe8da5214542bfb267d0ed30622409\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ebfe3281a65e48faa36c79861ba3db47 : TIMED_OUT in 3h 03m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bb6659bfe7674510a05cb25e422c3e40 : TIMED_OUT in 3h 04m 57s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c63101a64fbd40da8f27a52bab1b88eb : TIMED_OUT in 3h 02m 27s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3033073f6bba498abcba099341348ee8 : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b063e1eac0474b75afd2af25b8898e40 : SUCCESS in 4m 45s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8c0729aefee340b38fd9321710443664 : TIMED_OUT in 3h 05m 47s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b441f9eb2c8744adbabc5dc2d1efa1b0 : TIMED_OUT in 3h 03m 07s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"b9c07054f4eb1103454db326f87a93f29229037b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-09 08:01:17.000000000","message":"Patch Set 31: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"4beead5175c7180cc0cbfbc401d04818e69b169f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-19 07:17:23.000000000","message":"Patch Set 32: Published edit on patch set 31.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":32},{"id":"aba3c276f644df9cb4c18d064b259a4d1b7aa769","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 08:42:55.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/526fbbd446aa4928bb1bc308a28237b6\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5757d7cee48a4a1e9db7a80beeb62e63 : SUCCESS in 25m 08s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/37fbd0f1db8f4b5caa61b221b53e1ae5 : SUCCESS in 1h 02m 27s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fd434414db2c48acb8b625de588c7127 : SUCCESS in 18m 47s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3c88720ecff7432c812b53c951613b24 : SUCCESS in 27m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/785bd37a99aa447dbe17d5895ac1ac13 : SUCCESS in 56m 28s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e56dbfb4c3574c6d91dfbe160e6cfd3c : SUCCESS in 1h 20m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8602152ef1e429fbbeb323144b3a2d5 : SUCCESS in 2m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f555869d54cf41ae8762a904626726de : SUCCESS in 2m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4dccd740ff8346ce9985fb658cf12946 : SUCCESS in 2m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/11e533291a244164a1ed77d91f572a56 : SUCCESS in 2m 41s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fa99f0fc42f4cae803a6a33cd5fc31f : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1fbc01b843a49dbb0890bc7a69ca715 : SUCCESS in 4m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1ac5a440b4874950aace0dc9cb6dafb7 : SUCCESS in 29m 25s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/67fe0ca130f844a397a7c914b585e247 : SUCCESS in 36m 13s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ffa22cb63df345d18b996961e9307143 : SUCCESS in 1h 07m 18s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/95b4745ddc3e4ab0a50ad9b60e366928 : SUCCESS in 1h 22m 02s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/06b13ad678ea479b9b8759118b8be223 : SUCCESS in 33m 27s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/630ed9dc2c4045eeb5afaee4e16f92ef : SUCCESS in 19m 00s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/283125ae657a4170887bf60ffd5fae4a : SUCCESS in 48m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3ee7500d27604b568a699421363af6ea : SUCCESS in 49m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/58f011523d0a464cba5a1065883b98de : SUCCESS in 1h 14m 26s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/1e948cec222f4771bf5bcb51c5654f51 : SUCCESS in 20m 57s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"91593924da1e28b708368f304128e8ff2bbb373e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 10:24:51.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e12202d3d99432eb1945d990d815954\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0f9f9b72594941b98c092c6f2c6fbecd : TIMED_OUT in 3h 05m 04s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c7c8736fe54c44378067e56709cf0628 : TIMED_OUT in 3h 03m 20s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fb9b1bc6f1264dcdbbca37c4cc051410 : SUCCESS in 1h 49m 18s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1a269c740107489a8392ebe22b4f370a : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0d336e46a04144c1820bfc25632154e8 : SUCCESS in 3m 52s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b7a91c01e74f4751bc3e3f8a5fa4e883 : SUCCESS in 2h 36m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d4655c003ea94fac8134a765bc8bf44d : FAILURE in 1h 56m 20s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"13532168fec2a2b483fa0b05b77c4e00f5c2b25a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-19 11:14:16.000000000","message":"Patch Set 33: Published edit on patch set 32.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":33},{"id":"1b67af2dc1c2f60f24a2197b9ea4c25491a02ddc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 12:48:52.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f23528494baf4c6cb9166b6cdbd95ac7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/26e443e9886f48399089b6e51351bec2 : SUCCESS in 29m 40s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2b9bd975283f4c2ba3685981a5672cbb : SUCCESS in 1h 09m 05s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2600dad3b459452aac0d44d46f20c532 : SUCCESS in 22m 01s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2bc01780fbbb4bbdb6f02dfd9af9f3a7 : SUCCESS in 23m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6385ddfedb494dc7885c53a842fbc188 : SUCCESS in 56m 52s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0e7a31354124459992d525decf2fb561 : SUCCESS in 1h 28m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/777392a59f9a4341aee8868ca1d239e2 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/672f2b57e7914a33b1d0309bda05deaa : SUCCESS in 3m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa6b88fbaa3f4b8086cf4e4463c575c5 : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6d14084b3081444ca5d22bf952a3c202 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02bcad382a824b97a8f89d52cbdcdb17 : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66520ab36fd243dcba9617bc90e97279 : SUCCESS in 5m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/535587470aa44bba9371096f44c7778e : SUCCESS in 26m 07s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e1c605fbc37640e08982a420c518cef1 : SUCCESS in 34m 18s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/78c1fb8cd6b4412ba5d23f8c2bd4f407 : SUCCESS in 59m 43s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4cb0d677983e48ca9b488a24725b1e7f : SUCCESS in 1h 21m 30s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/7fa90ff88f35456bafa57aaccf013969 : SUCCESS in 32m 58s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/876958b8f68e4cacb1aa80fe4c153c46 : SUCCESS in 24m 22s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8052740e275d4613a17b8987db8082ec : SUCCESS in 52m 46s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3a5f2d6cf2074525b81976a1a0c2cc85 : SUCCESS in 49m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c95e06b29d994dfb872734536fa58551 : SUCCESS in 1h 12m 05s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/356d0b7c04c14b14ae9e043896b3801d : SUCCESS in 20m 53s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"9ebd69480e7fff0d556cfe6dcc74e53b92842b9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-19 12:55:49.000000000","message":"Patch Set 34: Published edit on patch set 33.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":34},{"id":"92a82ffbd1371ba0d14e664c33767284b78fcf75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-19 12:56:04.000000000","message":"Patch Set 35: Published edit on patch set 34.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":35},{"id":"f58353d955cd9000ed50c53adce2cd89f556eb5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-19 12:56:15.000000000","message":"Patch Set 36: Published edit on patch set 35.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":36},{"id":"10b23039ed17870094f8178fbeeb4d0bea62f6ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-19 12:56:33.000000000","message":"Patch Set 37: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":37},{"id":"eaac89f8527b07f44c462be4ca5f22f6cdbd7c45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-19 12:58:57.000000000","message":"Patch Set 38: Published edit on patch set 37.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":38},{"id":"d9635229b19f502a32ff6e6c920db3845c8052a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 14:32:54.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d98d23e447ae4c0cad3cbf6a33119aa6\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9a8bdd09a1894920bbeff958b1c54ca4 : SUCCESS in 19m 43s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/de92d75218c945538d000254cb914600 : SUCCESS in 1h 01m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f7abecc302764c38973e94a57a4317b4 : SUCCESS in 16m 18s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5f38c80238f44d80948c1f90e74d2715 : SUCCESS in 29m 19s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/aef51d24562342049966d7e479015b6f : SUCCESS in 59m 06s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e5ee0134a90346079bde58bb6dae8d86 : SUCCESS in 1h 26m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3800d37f9e14c1da25f40221b75dfff : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd874a03801e4e328fd8fade30a9c6b3 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea1c45f9b9d94930b6e7f442ec1195a7 : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a03ead6fb7843b99d4961750ee9f459 : SUCCESS in 3m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4249fd5400304d889f9e9304d07046c0 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fbd4700ae8940f286cbce47c18e7ba9 : SUCCESS in 5m 25s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7deccb210b12473285ef55b01de4583b : SUCCESS in 22m 30s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d5606fe27f714c528d8ae942a7788262 : SUCCESS in 37m 43s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c2640bd8944943e8a38f70c202526900 : SUCCESS in 1h 04m 55s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b800b4d0173148369a3051a9aec67428 : SUCCESS in 1h 28m 42s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/50a61beda6b44e01ba1a549d5b6e6f3c : SUCCESS in 33m 09s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c882c57ce0ca4e528fb13625e893b47b : SUCCESS in 17m 52s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/27be2ac1de6a49128bff0eaf4e317b3a : SUCCESS in 55m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8d9fb7b191c84aae9448f0e0c0b20a25 : SUCCESS in 54m 20s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c37cb8b47bba4368a0740c7765ea8cdd : SUCCESS in 1h 06m 57s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/aa84bdec6b3a45b7828cfb501d9aa3b5 : SUCCESS in 22m 03s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"127364d9e26e3c298bd813eb98bee8db2e43c758","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-19 14:46:52.000000000","message":"Patch Set 39: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":39},{"id":"d7d13980fbe9eaa3c7e8017635fd89529d6e93a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-19 15:17:56.000000000","message":"Patch Set 40: Patch Set 39 was rebased\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":40},{"id":"78ac4edf71ae5886cd88fe82d6ab430d213552c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 16:48:44.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d7390b5794f4f5c850e81108969dfdb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0055bcb5c1534fb69ef1a3ae39fcfc05 : SUCCESS in 20m 44s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1cf95b278da746188aee371074237555 : SUCCESS in 1h 07m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/962ecdda12424eca87504a387deb3e14 : SUCCESS in 17m 16s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/63f289efaece4e3eaa39ae101b461c66 : SUCCESS in 26m 42s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6eaaf49bad7f4a9b97fa9a164e9659c3 : SUCCESS in 59m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/75792ab6f06342aa9584ee70524d5f44 : SUCCESS in 1h 25m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/714255133c814c4a89d61cfeb42fa43b : SUCCESS in 2m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1abf66ffd9d4afaa2d7b11eba380196 : SUCCESS in 2m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/312410e50b2c4bf68cd65094459c8b4a : SUCCESS in 2m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/09a4ead7889f47d7816b29f1911dffe7 : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcc11d1f109a4fda87c43e2031333f8b : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e425716a13fc454ab37c972aace6fe71 : SUCCESS in 3m 50s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/813204cec5394116b9b35fe00bd9d5c2 : SUCCESS in 26m 05s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b61a76f989114722aaac98ff4df7cba4 : SUCCESS in 33m 49s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/43ea5085b1454cd3b60ad5c567a3ffbf : SUCCESS in 1h 02m 53s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a0c8bd013eef4511b19d6c8d0eb9b9f5 : SUCCESS in 1h 21m 13s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/8e5da148ee8742748e3f355321357d4d : SUCCESS in 37m 37s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4c8b5b52766d41728f4a88baeffd1a05 : SUCCESS in 36m 59s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/36806d5a772b4c14856157564f3797c0 : SUCCESS in 53m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d9de199c0b924ad58ace7cfd0ef11407 : SUCCESS in 45m 58s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/894aace0f67745d082c0a57bea565f68 : SUCCESS in 1h 17m 43s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/6629a44bd85b4332a925a3b66b5f203a : SUCCESS in 22m 49s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"ba98344bf43edceb1db78f40a355cded87894cdc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 22:31:41.000000000","message":"Patch Set 40:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23b178c370c14356bde8c60c66092db7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9d8c1bd83a9f41c4a446da8737f3ab85 : TIMED_OUT in 3h 04m 53s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/45d50ffd0a7b4e59aaef6b3af01fca46 : TIMED_OUT in 3h 02m 59s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f15781e1598045778640717d0dc2543c : TIMED_OUT in 3h 02m 56s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d0bd00eaa1ef4e12b205ad4a30e89d5a : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2d6d85a511384381a2d4bacf65843089 : SUCCESS in 4m 02s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e143257140c14cf889254c81b6a05228 : TIMED_OUT in 3h 04m 46s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e5d1b00bb1844da7871ec71168aa8bb9 : TIMED_OUT in 3h 02m 12s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"f8023a22294aee46eb3c131550fbad97be8ec34f","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-08-20 10:24:31.000000000","message":"Patch Set 40: Code-Review+2","accounts_in_message":[],"_revision_number":40},{"id":"e9a3ddf71e4570635c4824b5b0ffa79762c9f651","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-08-20 10:29:24.000000000","message":"Patch Set 40:\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"f5db5d790935fe2327d3d85c8657882f49790012","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-08-20 10:29:30.000000000","message":"Patch Set 40: Code-Review-1","accounts_in_message":[],"_revision_number":40},{"id":"0b4b3c733a01f8401f9cf25ddf1e16ac384059cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-22 20:24:16.000000000","message":"Uploaded patch set 41.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":41},{"id":"1c4e6e38c65a3dae537b00ef0b22db819048741e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-22 20:24:30.000000000","message":"Patch Set 40: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"4d8cfd3b32efb532b2e24e4858dc31291be1c88f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 22:07:50.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f460f94386b497bb78baf6f04dba995\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1e25ba696f054455958ce201b19ed52b : SUCCESS in 37m 38s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9fa0d10904774e5ab0aa34f6f7fecfd7 : SUCCESS in 1h 04m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/84fa6b6b04a84a77877167d333d2d4c7 : SUCCESS in 16m 37s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ede194f23dc24e5788c398f4cd4cd6c1 : SUCCESS in 32m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e9319a3c573a4af6a6a669e16b4e32c8 : SUCCESS in 55m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/88760a3ef8cc4ea8a4698dd6cef0560c : SUCCESS in 1h 23m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93938a5a3f134c9dbc60202601beaabc : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c36a5f61a20e46ea9f1bf6319c087ddb : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/61f485015d5d44889b34551cc5ad93a4 : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c25d03162cba4a06a9091dac576e5c67 : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc7a952800d946ca8214dd077f391eac : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a54d07d52ef4a62bc74879d21e3c49f : SUCCESS in 4m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60d9fc2834764baa995c806bb8ff8ac3 : SUCCESS in 5m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ff16f4cde63b4916b867119df03c9064 : SUCCESS in 25m 22s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d9d587fc1ca94e048f8d0b985b3e2c17 : SUCCESS in 35m 53s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/31d3b297a5954932975155d9441905fe : SUCCESS in 1h 05m 25s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1043835b045c4aa4bec003c487019a28 : SUCCESS in 1h 29m 41s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/bcbbdbe17eed47eda837f9ff35f9ae9b : SUCCESS in 38m 44s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1b9807b00f514ecbac916fc06762ace3 : SUCCESS in 16m 54s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/bbf424aa07d743da8cf2da0a1ac14862 : SUCCESS in 1h 17m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4fbef0b565d74680b3c6b6571cdd1d99 : SUCCESS in 50m 38s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a1633a51a7e8471bba3e2285e1da1689 : SUCCESS in 1h 23m 46s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/1caed0b9885b4152b66a29dfddb017cb : SUCCESS in 23m 15s (non-voting)","accounts_in_message":[],"_revision_number":41},{"id":"55a849042731256f746f041323118e5d38f5fe6f","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-08-23 07:12:29.000000000","message":"Patch Set 41: Code-Review+2","accounts_in_message":[],"_revision_number":41},{"id":"dd156fbe6bd1c57555ee245700f8d67a110a05fd","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-23 07:46:37.000000000","message":"Patch Set 41: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"1fb52e52a014cd651dc7b728d0c893e3c21b563a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-23 07:52:07.000000000","message":"Patch Set 42: Published edit on patch set 41.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":42},{"id":"a5f2970f8547daf2ce7087e17dd541d3ed19db1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-23 07:52:42.000000000","message":"Patch Set 43: Published edit on patch set 42.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":43},{"id":"8b8bcbe6a169df67cf25370c84c98217c37a1da6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-23 07:55:43.000000000","message":"Patch Set 43: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":43},{"id":"f2306b1b48cc71bcbabd8f111c3e73806b232fb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 09:35:33.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f40f42c9a1f4bb2b44f13531e73ac01\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8940ab7e66404c03b7e2dec17913e2dc : SUCCESS in 22m 14s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c145f36a8df4485f89d13a2297d91aa4 : SUCCESS in 1h 00m 37s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7fa6bc8331614d6d95a42a38fc2cad81 : SUCCESS in 15m 10s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/90eecc9cb96b47b084d431bab1e33124 : SUCCESS in 21m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8b8c4e7afb684e9cab79ed26d3a25c22 : SUCCESS in 59m 11s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/db2bf493586a4004bcb9ce9fa27c1559 : SUCCESS in 1h 28m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9965ea6dda9e429d96c8413d8300dbc5 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8fdeabfc9674ab8be11e76c528c987b : SUCCESS in 4m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/688909c41117410b8cf177b6b06a141f : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/50b39154715a4270a80539134c1bdbd0 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dded8e036082410c8db3ed96445b581b : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0059c1dae3614b49b6f2f1b5eba27d4f : SUCCESS in 4m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6aa2807ea571474dbb0fa7f0c84b1555 : SUCCESS in 3m 35s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0475b074421248448a7d30e310cc8430 : SUCCESS in 28m 06s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a1ff392db8964bb2a71eb941d796c522 : SUCCESS in 33m 30s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8beea0ce20f64ef7836d2b6c8adf1537 : SUCCESS in 1h 02m 53s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3ca5d4cc93a942ea85c4762d30e59711 : SUCCESS in 1h 29m 41s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/c7168dbba65f44959f538d6a4ddd7c23 : SUCCESS in 33m 08s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e1d3e8def1dc4003b700bd3c6952db13 : SUCCESS in 24m 45s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7ec3f687f9d4492294ac9253c3985db3 : SUCCESS in 54m 08s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d8d6123bf229447c98124a402e68afbb : SUCCESS in 51m 45s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1826eba0420f401488f225d2793341ed : SUCCESS in 1h 10m 23s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/8c09983cd334427f962486d1e5cb536c : SUCCESS in 24m 30s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"2715dbad78e00cf42678a2b05bdda60c28a11ca7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-23 11:33:07.000000000","message":"Patch Set 43: Backport-Candidate-1 Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":43},{"id":"25e98188b7f68272dcec27e55ab8a4976ff6dc22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-23 13:57:31.000000000","message":"Patch Set 44: Patch Set 43 was rebased\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":44},{"id":"ea4794865997ca6a2f6f26a32cbbb9f234e44fd3","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-08-23 13:57:59.000000000","message":"Patch Set 44: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":44},{"id":"749ac6628b8546807ff56a3c6a298de116684f7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 15:46:41.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89c2bbf1c28e4bc5acea0c6eff8e6d59\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0ecf0e9817354935a597898d98e9dee4 : SUCCESS in 27m 04s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2a68ef5236be44a5a06aa661ef6de055 : SUCCESS in 1h 10m 03s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/16bfc3f9ab77419895e4e357c7865569 : SUCCESS in 21m 28s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e7d6129412a540d5a78e926cb7b31a93 : SUCCESS in 42m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/cb7ecf824bae466e8fdff42d6cee7b70 : SUCCESS in 1h 02m 04s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5c8c9bdedcf448de8a1c86dd83c10b27 : SUCCESS in 1h 34m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c44edede42e4fe9854560a74533f68f : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60c9b93692f946228b5b11820a76e2ba : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/85af441735084f0d9245ff8ba05cc480 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0731e57244fc4c4bbe502299220cf069 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6910d8df47fb4783af6063a10e6e01b3 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4388e6f4dbf4b9eb7f8b4c9c72d9fc2 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/db8e47e3729045839119d27f55ae4daf : SUCCESS in 3m 46s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1f8d1a5a287245a283f6462fae645136 : SUCCESS in 27m 53s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/acbe1c17501d41ab9adadab407b9853a : SUCCESS in 48m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9fcaad4b88dc4289a97585e7c39c81ae : SUCCESS in 59m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5c57c075d59d40c9bb48d0d108dc4683 : SUCCESS in 1h 26m 30s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/2ce900fcdbcc461abd729a6842c8bb5c : SUCCESS in 43m 31s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b3f3f3bbd486410789457c67a2f36b62 : SUCCESS in 31m 54s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/945e31e542be4c478dd0dd71522ec39d : SUCCESS in 1h 00m 25s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1a02ee36b981423f9dcd54e5ca0cb693 : SUCCESS in 47m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b10cd63edb1c4c0c8cc136462e7ef7f6 : SUCCESS in 1h 16m 01s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/75b389c9fd7f4cbab6573e3eb49a5efc : SUCCESS in 22m 50s (non-voting)","accounts_in_message":[],"_revision_number":44},{"id":"f6e9194fe476bad363b975fb36913c269bc7a85c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 15:47:15.000000000","message":"Patch Set 44: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":44},{"id":"b29e7ca8f2c1199b27d3e019f58bc20a8c53df16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 17:09:51.000000000","message":"Patch Set 44: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/098f69046a894d2a88729e492b12e0cb\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a5e52c5a0687434f830e9d19103bd698 : SUCCESS in 20m 47s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fb60ed0d582c4ffb83e7c336d79a8855 : SUCCESS in 28m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a6a3651444a547c8aec1d38537ea7a0e : SUCCESS in 1h 02m 57s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0ec604176205407bacc3b6fa33e8aa40 : SUCCESS in 1h 08m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a44e4e89112e4f9eaf34abb6f1f7fa67 : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74b4116907ad494987d96bef39c9245a : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6864c58b41c84e56944809db377224cd : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe4528a60232483d82ef2b1427a86d42 : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ed75d9015bb4d9aa1fdc57a2cea9fd0 : SUCCESS in 3m 39s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4a4b0a66364240eca7e34569f0a96a14 : SUCCESS in 29m 06s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1160de7b584d4f45b39e4e9e7e173abd : SUCCESS in 33m 30s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2fe34d4958b442299b3a782fc9fe1029 : SUCCESS in 1h 09m 42s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d9e7629606a64b13b954b34385c73b8d : SUCCESS in 1h 21m 09s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4556272fe28f491a93038c84eaf90bcf : SUCCESS in 15m 36s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/de76880b93c64930af3545a89f6a0a87 : SUCCESS in 1h 06m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9e0282b733f44937a651d33743273ae3 : SUCCESS in 55m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/71480fdbea034008a340a9120990d7d9 : SUCCESS in 1h 15m 34s","accounts_in_message":[],"_revision_number":44},{"id":"413f26e7e22f72e99e40f5ef09ec7a55cb656155","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 17:09:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":44},{"id":"9a3b3a70cc7ac89251dbfc4a9cefb47867893b8a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 17:10:48.000000000","message":"Patch Set 44:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24e6dab5f8df46e4ac0b2f9e5bb68d63\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9699512baceb4b99bb901328efbf2aa5 : SUCCESS in 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a293cc850064e90b6325ec4e64e9efa : SUCCESS in 37s","accounts_in_message":[],"_revision_number":44},{"id":"c270d6f2668da8d3d1ea4356825b9455e44d5f04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-24 15:53:01.000000000","message":"Patch Set 44:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/299144d8e0214e22b04fba8b9aa86faa\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ad23487bd2ee4ecdbfed6761d4995507 : TIMED_OUT in 3h 07m 24s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/60efac9e346443cbbf13fd797eecce0b : TIMED_OUT in 3h 03m 36s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6f5e104ae4e04fee8d435b115a6c986f : TIMED_OUT in 3h 02m 56s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/37a6216ef6a8468aa8a52170e76d5202 : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/30ca485a2e844b53982167ec853ad282 : SUCCESS in 4m 07s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d346c3b87877460f8e0dd14d2dcb95db : TIMED_OUT in 3h 05m 46s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f9cb4c9be1e743669cfa5562f5953e3a : FAILURE in 2h 54m 53s (non-voting)","accounts_in_message":[],"_revision_number":44}],"current_revision_number":44,"current_revision":"cd26b72658cca7413b1f359985314a12ec194583","revisions":{"3ecf02c86ac69c3ebf05ca42ee1c79720d86a66a":{"kind":"REWORK","_number":1,"created":"2024-07-16 15:56:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/1"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"subject":"bifrost: Switch to ansible-core and bump up to 2.17","message":"bifrost: Switch to ansible-core and bump up to 2.17\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ecf02c86ac69c3ebf05ca42ee1c79720d86a66a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ecf02c86ac69c3ebf05ca42ee1c79720d86a66a"}]},"branch":"refs/heads/master"},"eaa71562512978f4396fc7f2a0a73e4982821cae":{"kind":"REWORK","_number":2,"created":"2024-07-16 16:49:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/2"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 16:49:01.000000000","tz":0},"subject":"bifrost: Switch to ansible-core and bump up to 2.17","message":"bifrost: Switch to ansible-core and bump up to 2.17\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eaa71562512978f4396fc7f2a0a73e4982821cae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eaa71562512978f4396fc7f2a0a73e4982821cae"}]},"branch":"refs/heads/master"},"487f55ac1c9b98b6db4ed59e141bafbec8a9c5bb":{"kind":"REWORK","_number":3,"created":"2024-07-17 06:53:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/3"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-17 06:53:09.000000000","tz":0},"subject":"bifrost: Switch to ansible-core and bump up to 2.17","message":"bifrost: Switch to ansible-core and bump up to 2.17\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/487f55ac1c9b98b6db4ed59e141bafbec8a9c5bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/487f55ac1c9b98b6db4ed59e141bafbec8a9c5bb"}]},"branch":"refs/heads/master"},"f643ed7c86ea406d8306bae7bf3973bd3b292733":{"kind":"REWORK","_number":4,"created":"2024-07-17 06:56:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/4"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-17 06:56:28.000000000","tz":0},"subject":"bifrost: Switch to ansible-core and bump up to 2.17","message":"bifrost: Switch to ansible-core and bump up to 2.17\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f643ed7c86ea406d8306bae7bf3973bd3b292733"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f643ed7c86ea406d8306bae7bf3973bd3b292733"}]},"branch":"refs/heads/master"},"56ed75d6dffbb4ff3a2e4acbbf8699649beee72a":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2024-07-17 06:58:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/5"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-17 06:58:04.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56ed75d6dffbb4ff3a2e4acbbf8699649beee72a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56ed75d6dffbb4ff3a2e4acbbf8699649beee72a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c9a1773bd73ba27344d68c12b32f634cc28049be":{"kind":"REWORK","_number":6,"created":"2024-07-17 07:56:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/6"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-17 07:56:10.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c9a1773bd73ba27344d68c12b32f634cc28049be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c9a1773bd73ba27344d68c12b32f634cc28049be"}]},"branch":"refs/heads/master"},"87d435b25d7012432a63b3c2e439b84e7e2da806":{"kind":"REWORK","_number":7,"created":"2024-07-17 10:02:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/7"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-17 10:02:04.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87d435b25d7012432a63b3c2e439b84e7e2da806"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87d435b25d7012432a63b3c2e439b84e7e2da806"}]},"branch":"refs/heads/master"},"04112398dd225800f87f783463deeb53eeb4b123":{"kind":"REWORK","_number":8,"created":"2024-07-17 10:05:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/8"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-17 10:05:48.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/04112398dd225800f87f783463deeb53eeb4b123"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/04112398dd225800f87f783463deeb53eeb4b123"}]},"branch":"refs/heads/master"},"442b008f2b2d6f807dba022518a8ddebbacea932":{"kind":"REWORK","_number":9,"created":"2024-07-17 10:57:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/9"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-17 10:57:25.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/442b008f2b2d6f807dba022518a8ddebbacea932"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/442b008f2b2d6f807dba022518a8ddebbacea932"}]},"branch":"refs/heads/master"},"e0b407c12f92673b0c68c78f74077228d1bd62ff":{"kind":"REWORK","_number":10,"created":"2024-07-17 11:32:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/10"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-17 11:32:11.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0b407c12f92673b0c68c78f74077228d1bd62ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0b407c12f92673b0c68c78f74077228d1bd62ff"}]},"branch":"refs/heads/master"},"f85209370f5afbf6bc8c8bfb603b863aa1ba0ea3":{"kind":"REWORK","_number":11,"created":"2024-07-17 11:59:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/11"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-17 11:59:28.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f85209370f5afbf6bc8c8bfb603b863aa1ba0ea3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f85209370f5afbf6bc8c8bfb603b863aa1ba0ea3"}]},"branch":"refs/heads/master"},"45910d7da52b3266e6965a38d9a43e3afe5ddb71":{"kind":"REWORK","_number":12,"created":"2024-07-17 18:00:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/12"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-17 18:00:30.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45910d7da52b3266e6965a38d9a43e3afe5ddb71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45910d7da52b3266e6965a38d9a43e3afe5ddb71"}]},"branch":"refs/heads/master"},"d0daa63746029bb0e8a610b75cbc7403fe17f643":{"kind":"REWORK","_number":13,"created":"2024-07-17 18:35:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/13"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-17 18:35:30.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d0daa63746029bb0e8a610b75cbc7403fe17f643"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d0daa63746029bb0e8a610b75cbc7403fe17f643"}]},"branch":"refs/heads/master"},"391d81a3d54d3d1094e76b52a9d53ca32d9bfa40":{"kind":"REWORK","_number":14,"created":"2024-07-17 20:31:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/14"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-17 20:31:03.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/391d81a3d54d3d1094e76b52a9d53ca32d9bfa40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/391d81a3d54d3d1094e76b52a9d53ca32d9bfa40"}]},"branch":"refs/heads/master"},"7014c3f9765228b9b952bba5f2949b06febf717d":{"kind":"REWORK","_number":15,"created":"2024-07-17 20:31:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/15"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-17 20:31:31.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7014c3f9765228b9b952bba5f2949b06febf717d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7014c3f9765228b9b952bba5f2949b06febf717d"}]},"branch":"refs/heads/master"},"9c6331be589f6cbb26b35839d4ded8473abd9ca7":{"kind":"REWORK","_number":16,"created":"2024-07-17 20:31:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/16"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-17 20:31:48.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c6331be589f6cbb26b35839d4ded8473abd9ca7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c6331be589f6cbb26b35839d4ded8473abd9ca7"}]},"branch":"refs/heads/master"},"0e7d9b078ad1a89be06d8b5588fbca1b531430b8":{"kind":"REWORK","_number":17,"created":"2024-07-18 05:13:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/17"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-18 05:13:50.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e7d9b078ad1a89be06d8b5588fbca1b531430b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0e7d9b078ad1a89be06d8b5588fbca1b531430b8"}]},"branch":"refs/heads/master"},"647cfcb46fc598e704320ca5e6d3113dfbd3c6a4":{"kind":"REWORK","_number":18,"created":"2024-07-18 05:17:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/18"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-18 05:17:14.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/647cfcb46fc598e704320ca5e6d3113dfbd3c6a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/647cfcb46fc598e704320ca5e6d3113dfbd3c6a4"}]},"branch":"refs/heads/master"},"a43fd9d31c2a33d12bc29a2c72f9abcfb2749e8c":{"kind":"REWORK","_number":19,"created":"2024-07-18 07:07:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/19"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-18 07:07:12.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a43fd9d31c2a33d12bc29a2c72f9abcfb2749e8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a43fd9d31c2a33d12bc29a2c72f9abcfb2749e8c"}]},"branch":"refs/heads/master"},"55089dc3218769faf880f39cb8c4178f32531d72":{"kind":"REWORK","_number":20,"created":"2024-07-18 08:13:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/20"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-18 08:13:50.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/55089dc3218769faf880f39cb8c4178f32531d72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/55089dc3218769faf880f39cb8c4178f32531d72"}]},"branch":"refs/heads/master"},"d407298247857e8540da59ddebba5c52f1f599ba":{"kind":"REWORK","_number":21,"created":"2024-07-22 14:06:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/21"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-22 14:06:48.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d407298247857e8540da59ddebba5c52f1f599ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d407298247857e8540da59ddebba5c52f1f599ba"}]},"branch":"refs/heads/master"},"c2fabb65d652804b284c53820fda3ef3f02dc39b":{"kind":"REWORK","_number":22,"created":"2024-07-23 07:29:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/22"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-23 07:29:37.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c2fabb65d652804b284c53820fda3ef3f02dc39b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c2fabb65d652804b284c53820fda3ef3f02dc39b"}]},"branch":"refs/heads/master"},"046c00c18b2ca28ccff2865f7f4ede6477b21823":{"kind":"REWORK","_number":23,"created":"2024-07-23 09:35:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/23"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-23 09:35:37.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/046c00c18b2ca28ccff2865f7f4ede6477b21823"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/046c00c18b2ca28ccff2865f7f4ede6477b21823"}]},"branch":"refs/heads/master"},"c64f3368f3281bab39e1f3aba51b05c4df52185a":{"kind":"REWORK","_number":24,"created":"2024-07-23 09:36:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/24"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-23 09:36:01.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c64f3368f3281bab39e1f3aba51b05c4df52185a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c64f3368f3281bab39e1f3aba51b05c4df52185a"}]},"branch":"refs/heads/master"},"a1ffd2c73ddd5985a9bc78aee316d0d2176ec5e6":{"kind":"REWORK","_number":25,"created":"2024-07-23 11:47:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/25"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-23 11:47:36.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1ffd2c73ddd5985a9bc78aee316d0d2176ec5e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a1ffd2c73ddd5985a9bc78aee316d0d2176ec5e6"}]},"branch":"refs/heads/master"},"6d4b75987b11852f00c9c5711152c1adfd35e0e0":{"kind":"REWORK","_number":26,"created":"2024-07-23 19:15:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/26"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-23 19:15:11.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d4b75987b11852f00c9c5711152c1adfd35e0e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d4b75987b11852f00c9c5711152c1adfd35e0e0"}]},"branch":"refs/heads/master"},"74a5dc8e36f78c33a011a6cb06666066d280b69c":{"kind":"REWORK","_number":27,"created":"2024-07-23 19:58:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/27"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-23 19:58:46.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/74a5dc8e36f78c33a011a6cb06666066d280b69c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/74a5dc8e36f78c33a011a6cb06666066d280b69c"}]},"branch":"refs/heads/master"},"d205e1c0a981da77ee148946ad77b650e2114fe1":{"kind":"REWORK","_number":28,"created":"2024-07-24 05:54:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/28"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-24 05:54:54.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d205e1c0a981da77ee148946ad77b650e2114fe1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d205e1c0a981da77ee148946ad77b650e2114fe1"}]},"branch":"refs/heads/master"},"b0a1d551cf2ac9f9c42ed7cd667ba2ec5889df88":{"kind":"REWORK","_number":29,"created":"2024-07-24 05:55:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/29"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-24 05:55:09.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0a1d551cf2ac9f9c42ed7cd667ba2ec5889df88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0a1d551cf2ac9f9c42ed7cd667ba2ec5889df88"}]},"branch":"refs/heads/master"},"24bad4218827baae9b8c555fdda78528bc9de264":{"kind":"NO_CODE_CHANGE","_number":30,"created":"2024-07-24 08:57:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/30"}}},"commit":{"parents":[{"commit":"e5d19ffca5f5a795424f332dea1c1f123e7077d5","subject":"toolbox: Bump ansible-core to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5d19ffca5f5a795424f332dea1c1f123e7077d5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-24 08:57:52.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nDepends-On: https://review.opendev.org/c/openstack/bifrost/+/924828\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/24bad4218827baae9b8c555fdda78528bc9de264"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/24bad4218827baae9b8c555fdda78528bc9de264"}]},"branch":"refs/heads/master","description":"Edit commit message"},"10af0bfe87bd86c3e4c6eb0740c7f284155c91b8":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2024-08-07 08:13:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/31"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-07 08:13:55.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nDepends-On: https://review.opendev.org/c/openstack/bifrost/+/924828\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10af0bfe87bd86c3e4c6eb0740c7f284155c91b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/10af0bfe87bd86c3e4c6eb0740c7f284155c91b8"}]},"branch":"refs/heads/master","description":"Rebase"},"775c645dd446134c6a2536d9313d8e4698e85d63":{"kind":"REWORK","_number":32,"created":"2024-08-19 07:17:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/32"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-19 07:17:23.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nDepends-On: https://review.opendev.org/c/openstack/bifrost/+/924828\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/775c645dd446134c6a2536d9313d8e4698e85d63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/775c645dd446134c6a2536d9313d8e4698e85d63"}]},"branch":"refs/heads/master"},"df24c7250c0bb2fd7f7d2322588d731c690984fa":{"kind":"REWORK","_number":33,"created":"2024-08-19 11:14:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/33"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-19 11:14:16.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nDepends-On: https://review.opendev.org/c/openstack/bifrost/+/924828\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/df24c7250c0bb2fd7f7d2322588d731c690984fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/df24c7250c0bb2fd7f7d2322588d731c690984fa"}]},"branch":"refs/heads/master"},"4793eb18e3f986e1f231573f939b4ca8b1372ca4":{"kind":"REWORK","_number":34,"created":"2024-08-19 12:55:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/34"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-19 12:55:49.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nDepends-On: https://review.opendev.org/c/openstack/bifrost/+/924828\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4793eb18e3f986e1f231573f939b4ca8b1372ca4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4793eb18e3f986e1f231573f939b4ca8b1372ca4"}]},"branch":"refs/heads/master"},"b5ba2d6c65d08cd9f4140097106a2868c17b3c05":{"kind":"REWORK","_number":35,"created":"2024-08-19 12:56:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/35"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-19 12:56:03.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nDepends-On: https://review.opendev.org/c/openstack/bifrost/+/924828\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5ba2d6c65d08cd9f4140097106a2868c17b3c05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5ba2d6c65d08cd9f4140097106a2868c17b3c05"}]},"branch":"refs/heads/master"},"9f3e2b3ee7c22b27a95ae59bd719a076bf3bfd33":{"kind":"REWORK","_number":36,"created":"2024-08-19 12:56:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/36"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-19 12:56:14.000000000","tz":0},"subject":"bifrost: Fix ansible and deps installation","message":"bifrost: Fix ansible and deps installation\n\nDepends-On: https://review.opendev.org/c/openstack/bifrost/+/924828\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f3e2b3ee7c22b27a95ae59bd719a076bf3bfd33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9f3e2b3ee7c22b27a95ae59bd719a076bf3bfd33"}]},"branch":"refs/heads/master"},"4512961e06369272bd501de582283a4560baddb7":{"kind":"NO_CODE_CHANGE","_number":37,"created":"2024-08-19 12:56:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/37"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-19 12:56:33.000000000","tz":0},"subject":"bifrost: bump Ansible version to 8/9","message":"bifrost: bump Ansible version to 8/9\n\nDepends-On: https://review.opendev.org/c/openstack/bifrost/+/924828\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4512961e06369272bd501de582283a4560baddb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4512961e06369272bd501de582283a4560baddb7"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b05e57d6418f154613c9729c64cf87c6ccae3c2e":{"kind":"REWORK","_number":38,"created":"2024-08-19 12:58:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/38"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-19 12:58:57.000000000","tz":0},"subject":"bifrost: bump Ansible version to 8/9","message":"bifrost: bump Ansible version to 8/9\n\nThis is used by bifrost in setup script [1].\n\n[1]: https://opendev.org/openstack/bifrost/src/commit/1c95336154c9ed3995bb63c4b4d16f344290b796/scripts/env-setup.sh#L9\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b05e57d6418f154613c9729c64cf87c6ccae3c2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b05e57d6418f154613c9729c64cf87c6ccae3c2e"}]},"branch":"refs/heads/master"},"4aa5c538dbadfe528f96e22a0b63e718bfc3d871":{"kind":"NO_CODE_CHANGE","_number":39,"created":"2024-08-19 14:46:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/39"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-19 14:46:52.000000000","tz":0},"subject":"bifrost: bump Ansible version to 8/9 (rpm) and 9/10 (deb)","message":"bifrost: bump Ansible version to 8/9 (rpm) and 9/10 (deb)\n\nThis is used by bifrost in setup script [1].\n\n[1]: https://opendev.org/openstack/bifrost/src/commit/1c95336154c9ed3995bb63c4b4d16f344290b796/scripts/env-setup.sh#L9\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4aa5c538dbadfe528f96e22a0b63e718bfc3d871"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4aa5c538dbadfe528f96e22a0b63e718bfc3d871"}]},"branch":"refs/heads/master","description":"Edit commit message"},"05ffe6741d7710f319980b2542ba7b5ae6037f1e":{"kind":"TRIVIAL_REBASE","_number":40,"created":"2024-08-19 15:17:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/40"}}},"commit":{"parents":[{"commit":"a3cc3c76ead59b9239aaa41971138898c38ca397","subject":"nova: drop dmidecode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3cc3c76ead59b9239aaa41971138898c38ca397"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-19 15:17:56.000000000","tz":0},"subject":"bifrost: bump Ansible version to 8/9 (rpm) and 9/10 (deb)","message":"bifrost: bump Ansible version to 8/9 (rpm) and 9/10 (deb)\n\nThis is used by bifrost in setup script [1].\n\n[1]: https://opendev.org/openstack/bifrost/src/commit/1c95336154c9ed3995bb63c4b4d16f344290b796/scripts/env-setup.sh#L9\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/05ffe6741d7710f319980b2542ba7b5ae6037f1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/05ffe6741d7710f319980b2542ba7b5ae6037f1e"}]},"branch":"refs/heads/master","description":"Rebase"},"6b6351d25fcd29b53ca751a4f02e6bbcc83377eb":{"kind":"REWORK","_number":41,"created":"2024-08-22 20:24:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/41"}}},"commit":{"parents":[{"commit":"a3cc3c76ead59b9239aaa41971138898c38ca397","subject":"nova: drop dmidecode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3cc3c76ead59b9239aaa41971138898c38ca397"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-22 20:23:55.000000000","tz":120},"subject":"bifrost: bump Ansible version to 8/9 (rpm) and 9/10 (deb)","message":"bifrost: bump Ansible version to 8/9 (rpm) and 9/10 (deb)\n\nThis is used by bifrost in setup script [1].\n\n[1]: https://opendev.org/openstack/bifrost/src/commit/1c95336154c9ed3995bb63c4b4d16f344290b796/scripts/env-setup.sh#L9\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b6351d25fcd29b53ca751a4f02e6bbcc83377eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6b6351d25fcd29b53ca751a4f02e6bbcc83377eb"}]},"branch":"refs/heads/master"},"395d9d21965a27380231d5b574b0c79bf2c0c672":{"kind":"REWORK","_number":42,"created":"2024-08-23 07:52:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/42"}}},"commit":{"parents":[{"commit":"a3cc3c76ead59b9239aaa41971138898c38ca397","subject":"nova: drop dmidecode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3cc3c76ead59b9239aaa41971138898c38ca397"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 07:52:07.000000000","tz":0},"subject":"bifrost: bump Ansible version to 8/9 (rpm) and 9/10 (deb)","message":"bifrost: bump Ansible version to 8/9 (rpm) and 9/10 (deb)\n\nThis is used by bifrost in setup script [1].\n\n[1]: https://opendev.org/openstack/bifrost/src/commit/1c95336154c9ed3995bb63c4b4d16f344290b796/scripts/env-setup.sh#L9\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/395d9d21965a27380231d5b574b0c79bf2c0c672"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/395d9d21965a27380231d5b574b0c79bf2c0c672"}]},"branch":"refs/heads/master"},"c7ade49f21d08d1d2cac89866ccf2074994e867d":{"kind":"REWORK","_number":43,"created":"2024-08-23 07:52:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/43"}}},"commit":{"parents":[{"commit":"a3cc3c76ead59b9239aaa41971138898c38ca397","subject":"nova: drop dmidecode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a3cc3c76ead59b9239aaa41971138898c38ca397"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 07:52:42.000000000","tz":0},"subject":"bifrost: bump Ansible version to 8 (rpm) and 9 (deb)","message":"bifrost: bump Ansible version to 8 (rpm) and 9 (deb)\n\nThis is used by bifrost in setup script [1].\n\n[1]: https://opendev.org/openstack/bifrost/src/commit/1c95336154c9ed3995bb63c4b4d16f344290b796/scripts/env-setup.sh#L9\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c7ade49f21d08d1d2cac89866ccf2074994e867d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c7ade49f21d08d1d2cac89866ccf2074994e867d"}]},"branch":"refs/heads/master"},"cd26b72658cca7413b1f359985314a12ec194583":{"kind":"NO_CHANGE","_number":44,"created":"2024-08-23 13:57:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/46/924246/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/46/924246/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/46/924246/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/46/924246/44"}}},"commit":{"parents":[{"commit":"29fc6a8bc27e9a588a19d16e1a33a63a5e6c5c5a","subject":"nova: drop dmidecode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/29fc6a8bc27e9a588a19d16e1a33a63a5e6c5c5a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-07-16 15:55:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 13:57:31.000000000","tz":0},"subject":"bifrost: bump Ansible version to 8 (rpm) and 9 (deb)","message":"bifrost: bump Ansible version to 8 (rpm) and 9 (deb)\n\nThis is used by bifrost in setup script [1].\n\n[1]: https://opendev.org/openstack/bifrost/src/commit/1c95336154c9ed3995bb63c4b4d16f344290b796/scripts/env-setup.sh#L9\n\nChange-Id: Ia2721dd7459a11324726910c99ea4c357047631e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd26b72658cca7413b1f359985314a12ec194583"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cd26b72658cca7413b1f359985314a12ec194583"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
