)]}'
{"id":"openstack%2Fkolla~924249","triplet_id":"openstack%2Fkolla~stable%2F2024.1~I9fdccbe3bde07a7f9a1e6ac82b1096a024e1b0ee","project":"openstack/kolla","branch":"stable/2024.1","topic":"fix/apple_silicon_arch-stable/2024.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-07-17 10:07:51.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-07-17 11:55:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9fdccbe3bde07a7f9a1e6ac82b1096a024e1b0ee","subject":"fix: apple silicon architecture selection","status":"MERGED","created":"2024-07-16 16:18:35.000000000","updated":"2024-07-17 11:56:21.000000000","submitted":"2024-07-17 11:55:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924249-fix/apple_silicon_arch-stable/2024.1","cherry_pick_of_change":922066,"cherry_pick_of_patch_set":2,"meta_rev_id":"efcb6d5c7903843c7a186e52621a9c717064c433","_number":924249,"virtual_id_number":924249,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-17 11:55:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-07-17 10:07:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-07-16 16:40:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},{"value":1,"date":"2024-07-16 16:39:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-07-17 10:07:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-16 16:18:35.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},"state":"CC"},{"updated":"2024-07-16 16:39:26.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-07-16 16:40:45.000000000","updated_by":{"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},"reviewer":{"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},"state":"REVIEWER"},{"updated":"2024-07-16 17:47:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"16ecac24b26346140c90cf2f654a1fdbe642bc11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-07-16 16:18:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ac6a78420512515e2748565ba4a5f3a09b57a652","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-07-16 16:39:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"42b6d4f8810c955c39c9ace54eab3cd3ed194688","author":{"_account_id":34311,"name":"Tadas Sutkaitis","display_name":"Tadas Sutkaitis","email":"tadasas@gmail.com","username":"t42x"},"date":"2024-07-16 16:40:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a21c05fb4917d12dfeed4455357e2d4a5c123d33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 17:47:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0f81d046d0c4fe893584dd4737c8ad0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/62ee7bf063794d0189ea3c9b3c10f1c0 : SUCCESS in 24m 29s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1fdff6a88de8468083b45070d1ad7c99 : SUCCESS in 1h 04m 26s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/44b5872489dd41ff8d9ee78b876201a2 : SUCCESS in 17m 32s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1466fcefcd774e248d44eb84c2151fd2 : SUCCESS in 29m 53s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9e28ec78693f434496d35f9a9a727d44 : SUCCESS in 57m 21s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c989a36e1af441da8ec7625ab83df573 : SUCCESS in 1h 24m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/252fdd7f32d54979b966ff13b4372811 : SUCCESS in 2m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11298d59a1c54e3b9f51ac256443ffa9 : SUCCESS in 2m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3044747427544b149cc0365fe8625b38 : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccbc180c84ed4705ad25d18248e681e7 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95b897c6630548f0922a64bcbb17b6b8 : SUCCESS in 4m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b05cb2041b474847ae9059df454d2a12 : SUCCESS in 25m 27s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/01d6dc16b77d40d593e4cced05a04451 : SUCCESS in 30m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5be7c4729b80427893ee121f1b5511f0 : SUCCESS in 1h 07m 20s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4e1d85604a8a49aab1a493cb4ac7d11a : SUCCESS in 1h 21m 39s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/846d0d7849994bc5aae7d7de3190444b : SUCCESS in 20m 43s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3b24464d48344cc29ac9ae078b442167 : SUCCESS in 54m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/412c4e7c4993484b87b622af8dcf717b : SUCCESS in 45m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6c3402a3bf4f40d7ad5b9bd35da0068c : SUCCESS in 1h 15m 00s","accounts_in_message":[],"_revision_number":1},{"id":"0861edd6faad038ed80891be11123bbcc9986226","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 22:44:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92384700aace44daa6c9cd54fcc60bb5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b4b5f220b99345558eb8647dfcee301e : SUCCESS in 29m 59s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/133e861dc93544e29b18eb8473bcbcda : SUCCESS in 20m 54s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dec11695a50946c9927ce443374b481d : SUCCESS in 1h 28m 13s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/92672b89fd874868943a0bbd507529c7 : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9eb92f1e2f084cde985559b26c5a68b0 : SUCCESS in 3m 41s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/cfc31f7e5892499a8e03e165a5754fe7 : SUCCESS in 25m 11s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3b18e1702f0543b59bfbdea699fcf984 : TIMED_OUT in 3h 03m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8e91346da7b28657ef8424599b8907b54d9e10fa","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-17 10:07:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93515250cad3a9303876d1ff65dca4cd1dc845dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 10:08:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"92c88a022801c1812057d81ffb67687c1a4e15cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 11:55:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34f9cb531f5c47449e85d530de2397e5\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c0aaf7565b2b4ac59a0a065411cb9fbd : SUCCESS in 17m 41s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b7349ccd3a674a99ad2442554f3bb3ba : SUCCESS in 23m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/650016e455f5411f8587e9a8d0ced31a : SUCCESS in 59m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6ff898b32ade47fbabd4ce2390c5c337 : SUCCESS in 1h 45m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7817efa50f064d6aab4006f9b6a2fa1f : SUCCESS in 2m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f7f1ae2162e4d5eac2201eb5c358669 : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5785a3541ae4411cbbe3332e00dd15d1 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/137936e212f8455c84759cfe29d43ad8 : SUCCESS in 4m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/df1ac13f80ed4e3f8e877239bd06a1ef : SUCCESS in 27m 50s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c3d4e18fd87b48b58fcceeb044d0d8a5 : SUCCESS in 33m 21s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/932f72c8c49943f986feb71df4987f5d : SUCCESS in 1h 05m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/34406f3d133c4e88afc4a50e1fac991a : SUCCESS in 1h 36m 49s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1b9792e5ae3c42978d6dc83c4247d7da : SUCCESS in 25m 53s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0477d973da124192982c4b507f004bae : SUCCESS in 56m 56s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f24fcf18b83b4aaf897939505b6e9a9a : SUCCESS in 50m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/759f8691da044ae7bc09174a3e784a94 : SUCCESS in 1h 21m 42s","accounts_in_message":[],"_revision_number":1},{"id":"fad364806ce6c1f19bfd47553c340b05acb2788e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 11:55:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"efcb6d5c7903843c7a186e52621a9c717064c433","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 11:56:21.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c830c28af01e4a87a18821e7b07e9cdd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e21573a64d064c4180380f1fed4aeae5 : FAILURE in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cc66d0a53412024d534b8aee4b323f3fc581e9b0","revisions":{"cc66d0a53412024d534b8aee4b323f3fc581e9b0":{"kind":"REWORK","_number":1,"created":"2024-07-16 16:18:35.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/49/924249/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/924249/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/924249/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/924249/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/924249/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/924249/1"}}},"commit":{"parents":[{"commit":"4936a6ff98bef93ff0912150f8114bd97834ba81","subject":"add kayobe to release management doc\u0027s Launchpad admin section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4936a6ff98bef93ff0912150f8114bd97834ba81"}]}],"author":{"name":"Tadas Sutkaitis","email":"tadasas@gmail.com","date":"2024-06-16 16:49:16.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-07-16 16:18:35.000000000","tz":0},"subject":"fix: apple silicon architecture selection","message":"fix: apple silicon architecture selection\n\nCloses-Bug: #2073256\n\nChange-Id: I9fdccbe3bde07a7f9a1e6ac82b1096a024e1b0ee\n(cherry picked from commit 8b98145f23676121002ea654d2806b50e402edec)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc66d0a53412024d534b8aee4b323f3fc581e9b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc66d0a53412024d534b8aee4b323f3fc581e9b0"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
