)]}'
{"id":"openstack%2Fkolla~924302","triplet_id":"openstack%2Fkolla~master~I552b9fbc8fa0b49d3e0e3212dd11962b02837b9f","project":"openstack/kolla","branch":"master","topic":"924302","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-12-20 08:59:57.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2026-01-28 14:22:14.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-11-06 08:52:36.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"37355":{"account":{"_account_id":37355,"name":"Pablo Colson","display_name":"Pablo Colson","email":"colson.pablo@gmail.com","username":"pabloclsn"},"last_update":"2026-04-02 02:36:26.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2026-04-02 02:36:26.000000000","reason":"Change was submitted"},"34380":{"account":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"last_update":"2026-04-02 02:36:26.000000000","reason":"Change was submitted"},"34476":{"account":{"_account_id":34476,"name":"Marc Vorwerk","email":"marc+ubuntu@marc-vorwerk.de","username":"marc-vorwerk"},"last_update":"2026-04-02 02:36:26.000000000","reason":"Change was submitted"},"37170":{"account":{"_account_id":37170,"name":"Enrico Kern","email":"enrico.kern@stackxperts.com"},"last_update":"2026-04-02 02:36:26.000000000","reason":"Change was submitted"},"37171":{"account":{"_account_id":37171,"name":"pmil","display_name":"pmil","email":"yurus1@o2.pl","username":"pmil"},"last_update":"2026-04-02 02:36:26.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2026-03-25 14:20:46.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"37306":{"account":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"last_update":"2026-04-01 22:09:22.000000000","reason":"\u003cGERRIT_ACCOUNT_37306\u003e replied on the change","reason_account":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"}},"30523":{"account":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"last_update":"2026-04-02 02:36:26.000000000","reason":"Change was submitted"},"37725":{"account":{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},"last_update":"2026-04-02 02:36:26.000000000","reason":"Change was submitted"},"37855":{"account":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"last_update":"2026-04-02 02:36:26.000000000","reason":"Change was submitted"}},"hashtags":["neutron-ovn-vpn-agent"],"change_id":"I552b9fbc8fa0b49d3e0e3212dd11962b02837b9f","subject":"Add Dockerfile for neutron-ovn-vpn-agent","status":"MERGED","created":"2024-07-17 07:51:39.000000000","updated":"2026-04-02 06:22:53.000000000","submitted":"2026-04-02 02:36:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":42,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924302-924302","meta_rev_id":"d17876c7c0d00e8d86c5e1e03a1ed22acbd67268","_number":924302,"virtual_id_number":924302,"owner":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":37355,"name":"Pablo Colson","display_name":"Pablo Colson","email":"colson.pablo@gmail.com","username":"pabloclsn"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-04-02 02:36:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34476,"name":"Marc Vorwerk","email":"marc+ubuntu@marc-vorwerk.de","username":"marc-vorwerk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2026-03-25 14:19:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2026-03-25 14:20:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":37355,"name":"Pablo Colson","display_name":"Pablo Colson","email":"colson.pablo@gmail.com","username":"pabloclsn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34476,"name":"Marc Vorwerk","email":"marc+ubuntu@marc-vorwerk.de","username":"marc-vorwerk"},{"value":2,"date":"2026-03-31 11:12:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":37355,"name":"Pablo Colson","display_name":"Pablo Colson","email":"colson.pablo@gmail.com","username":"pabloclsn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34476,"name":"Marc Vorwerk","email":"marc+ubuntu@marc-vorwerk.de","username":"marc-vorwerk"},{"value":1,"date":"2026-03-31 11:12:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2026-03-25 14:19:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":-1,"date":"2026-03-25 11:36:29.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":37355,"name":"Pablo Colson","display_name":"Pablo Colson","email":"colson.pablo@gmail.com","username":"pabloclsn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34476,"name":"Marc Vorwerk","email":"marc+ubuntu@marc-vorwerk.de","username":"marc-vorwerk"},{"value":-1,"date":"2026-03-31 11:12:17.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2026-03-25 14:19:44.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":37355,"name":"Pablo Colson","display_name":"Pablo Colson","email":"colson.pablo@gmail.com","username":"pabloclsn"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34476,"name":"Marc Vorwerk","email":"marc+ubuntu@marc-vorwerk.de","username":"marc-vorwerk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":34476,"name":"Marc Vorwerk","email":"marc+ubuntu@marc-vorwerk.de","username":"marc-vorwerk"},{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},{"_account_id":37355,"name":"Pablo Colson","display_name":"Pablo Colson","email":"colson.pablo@gmail.com","username":"pabloclsn"},{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"_account_id":37170,"name":"Enrico Kern","email":"enrico.kern@stackxperts.com"},{"_account_id":37171,"name":"pmil","display_name":"pmil","email":"yurus1@o2.pl","username":"pmil"},{"_account_id":38366,"name":"Maximilian Sesterhenn","display_name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@wiit.cloud","username":"msesterhenn","status":"@WIIT AG"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-17 07:54:45.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-07-17 08:23:44.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-07-17 10:42:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-22 08:36:28.000000000","updated_by":{"_account_id":37170,"name":"Enrico Kern","email":"enrico.kern@stackxperts.com"},"reviewer":{"_account_id":37170,"name":"Enrico Kern","email":"enrico.kern@stackxperts.com"},"state":"CC"},{"updated":"2024-07-22 11:52:23.000000000","updated_by":{"_account_id":37171,"name":"pmil","display_name":"pmil","email":"yurus1@o2.pl","username":"pmil"},"reviewer":{"_account_id":37171,"name":"pmil","display_name":"pmil","email":"yurus1@o2.pl","username":"pmil"},"state":"CC"},{"updated":"2024-07-22 16:06:12.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-08-13 12:33:05.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-08-21 12:40:29.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2024-10-29 09:41:22.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2024-10-29 11:18:33.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2024-11-05 12:26:23.000000000","updated_by":{"_account_id":37355,"name":"Pablo Colson","display_name":"Pablo Colson","email":"colson.pablo@gmail.com","username":"pabloclsn"},"reviewer":{"_account_id":37355,"name":"Pablo Colson","display_name":"Pablo Colson","email":"colson.pablo@gmail.com","username":"pabloclsn"},"state":"REVIEWER"},{"updated":"2024-11-05 12:27:05.000000000","updated_by":{"_account_id":37355,"name":"Pablo Colson","display_name":"Pablo Colson","email":"colson.pablo@gmail.com","username":"pabloclsn"},"reviewer":{"_account_id":37355,"name":"Pablo Colson","display_name":"Pablo Colson","email":"colson.pablo@gmail.com","username":"pabloclsn"},"state":"REMOVED"},{"updated":"2024-11-06 07:36:21.000000000","updated_by":{"_account_id":34476,"name":"Marc Vorwerk","email":"marc+ubuntu@marc-vorwerk.de","username":"marc-vorwerk"},"reviewer":{"_account_id":34476,"name":"Marc Vorwerk","email":"marc+ubuntu@marc-vorwerk.de","username":"marc-vorwerk"},"state":"REVIEWER"},{"updated":"2025-01-28 17:58:11.000000000","updated_by":{"_account_id":37355,"name":"Pablo Colson","display_name":"Pablo Colson","email":"colson.pablo@gmail.com","username":"pabloclsn"},"reviewer":{"_account_id":37355,"name":"Pablo Colson","display_name":"Pablo Colson","email":"colson.pablo@gmail.com","username":"pabloclsn"},"state":"CC"},{"updated":"2025-08-27 10:59:46.000000000","updated_by":{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},"reviewer":{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},"state":"REVIEWER"},{"updated":"2025-09-09 13:18:54.000000000","updated_by":{"_account_id":38366,"name":"Maximilian Sesterhenn","display_name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@wiit.cloud","username":"msesterhenn","status":"@WIIT AG"},"reviewer":{"_account_id":38366,"name":"Maximilian Sesterhenn","display_name":"Maximilian Sesterhenn","email":"maximilian.sesterhenn@wiit.cloud","username":"msesterhenn","status":"@WIIT AG"},"state":"CC"},{"updated":"2025-10-01 11:45:04.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"CC"},{"updated":"2025-11-21 09:13:58.000000000","updated_by":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"reviewer":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"state":"REVIEWER"},{"updated":"2026-02-02 16:44:18.000000000","updated_by":{"_account_id":37355,"name":"Pablo Colson","display_name":"Pablo Colson","email":"colson.pablo@gmail.com","username":"pabloclsn"},"reviewer":{"_account_id":37355,"name":"Pablo Colson","display_name":"Pablo Colson","email":"colson.pablo@gmail.com","username":"pabloclsn"},"state":"REVIEWER"},{"updated":"2026-03-21 17:19:01.000000000","updated_by":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"reviewer":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"state":"REVIEWER"},{"updated":"2026-03-31 11:12:17.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"bb0065bdc67e6b312caffcb09a1bdce177a32a73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"date":"2024-07-17 07:51:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"11ac651a8152a6e8fead990471c24a0666d93d61","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-17 08:23:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"025032b1bf8871bb6a5a587ac8f93a684f30da4a","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-07-17 08:36:58.000000000","message":"Patch Set 1: Backport-Candidate-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c652ce24fc347c9ef1ead87764eec3237f567482","author":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"date":"2024-07-17 08:43:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f85fb1d315ac3a00ec9b7af7c3e2e785790f9853","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"date":"2024-07-17 08:55:36.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"bce10f526063a861dc6ea0c10773025ab7d32690","author":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"date":"2024-07-17 08:56:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"461ef54de3a425552341e3b5df9848663b909920","author":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"date":"2024-07-17 09:00:58.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"cec521e7afabc0aabd7a45452fc888f77b9b2e33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 10:42:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c0f384ef8ce4186880860a2c19a67cd\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/834f0ac7b8164b049e9a0e5864ef6dbd : SUCCESS in 27m 30s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c3ff9fa6320a4d2ba66d5f05bd93f106 : SUCCESS in 1h 05m 57s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2e93d14d8d1e48f68d525f7f641358c9 : SUCCESS in 16m 34s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6ddb7f1bb21d49038508d399c766e97e : SUCCESS in 25m 41s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2d768e592f424bfeb7f03dbc9d56649e : SUCCESS in 59m 40s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4002cea6376b48c2bb515b71315d8cd1 : SUCCESS in 1h 22m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2954101867b44cd9a1fd109477d670a6 : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e969318dd824eda87030ad2e6fbe761 : SUCCESS in 2m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/45544595368541d29829393a65befd2d : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b401c59b3a8d4c8583218c68179c1250 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec964d4eabde472d8ee7fadfed78d08f : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66a060ba31ab41eb966ea3926a7fc7f3 : SUCCESS in 5m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ad11680ba1a4e5da474c62af6ceead1 : SUCCESS in 4m 26s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ea3520d0f0f143ce9a3322c7596cd2a7 : SUCCESS in 25m 34s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/dd1e19b0ad544ec4b7c41baf5b60461e : SUCCESS in 33m 22s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5bfeec4ee842485da721f28f8f4c9670 : SUCCESS in 1h 07m 26s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7b1151d5564940a1b2d79a45c34f045b : SUCCESS in 1h 36m 01s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/5d54e2e144f3457d86bf06530cc2b4c6 : FAILURE in 57m 49s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/14ab97d3a765400b9a553352976868f2 : SUCCESS in 1h 33m 03s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f397c955a5e5493681babfedbcaed3e4 : SUCCESS in 19m 10s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6edf914259704bc8899c99cffc8b9655 : SUCCESS in 1h 16m 46s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f35917df9587406a8d5b128951f1350d : SUCCESS in 52m 38s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8ff80d56c3d8413c9612f4f12ccc0c03 : SUCCESS in 1h 22m 33s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/0cec70e9b0a34555915dd55ed35a8ff6 : FAILURE in 47m 21s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/596c984917c74142b83f6618abf24f76 : SUCCESS in 1h 01m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a1b177e043f174065d908a49a86d56267724ae83","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-07-17 10:46:03.000000000","message":"Patch Set 2: Backport-Candidate-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dd24635a0a470687b26a1a7b1ee11f353e06e1a7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-17 10:59:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2d14e615c4858c99cb37df7cfb4822471c210688","author":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"date":"2024-07-17 11:23:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ea8d56db962b4c3c9bfa346bc72324b1055da191","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 12:02:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96bdaf6d523b4a288e3c0f39cf8d5986\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0854a21a5a21413092884130ea6f8df9 : SUCCESS in 29m 39s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/67a758de42784a118fd73a4e5a8b4473 : TIMED_OUT in 3h 03m 43s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/31a5118d655c4bd9b9ce90b295d8e898 : SUCCESS in 1h 12m 57s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1ac8cdc55691409aa86d61de18b699a2 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3a7d82250e934b83a3bb29a56286027b : SUCCESS in 2m 38s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f50a64880a8a4b9d91a6073185292336 : TIMED_OUT in 3h 04m 12s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4602ec0f440f4dc0958cc54607c04267 : SUCCESS in 25m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d29aa42de0d2aaa7e560844d6b17b303453ee274","author":{"_account_id":37170,"name":"Enrico Kern","email":"enrico.kern@stackxperts.com"},"date":"2024-07-22 08:36:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"49d24160b398a4cea33a6d87f9b407b40d5cbd9a","author":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"date":"2024-07-22 09:06:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4960b39ecc768fe6df1d294ccb3ce577a8399457","author":{"_account_id":37171,"name":"pmil","display_name":"pmil","email":"yurus1@o2.pl","username":"pmil"},"date":"2024-07-22 11:52:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4a7f190d07e9b4ee687cd6c2d3554c8c3c90d883","author":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"date":"2024-07-22 12:14:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"edd87a1765466f243b769e864e459f7c67f350d5","author":{"_account_id":37171,"name":"pmil","display_name":"pmil","email":"yurus1@o2.pl","username":"pmil"},"date":"2024-07-22 12:58:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a22dd64eef4a2f9db4d39bb3b0fe4ab1f1584cc2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-22 16:06:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"57c6c47191c82b5fe92d4afb34693dd742b20fd1","author":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"date":"2024-07-22 17:58:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"95098f4568b765292629ad421f95008a69931034","author":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"date":"2024-07-24 07:20:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"97c5df9475f4f5e52b5b8271c3c10cfe4bda3e78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-10-23 15:44:46.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"24376f7afda03cf67b9e148bb9efd2c9d140f783","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 17:21:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec6a6e85fa734232803f613163e5a7b7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d34ad417019547da9da035b0a093c149 : SUCCESS in 33m 07s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b3cbc35d05954758ba2ec48753da6d44 : FAILURE in 6m 01s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/04e5348768dc4411b6e09d3c9854b8fe : SUCCESS in 22m 55s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b2e8beb42afc4f68a83074d1d68f1e2c : SUCCESS in 23m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/df20eda30aa448c5917dbab7b20b841b : SUCCESS in 1h 08m 27s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/89123374dfb64c44ba7cf3bf2730bdfc : FAILURE in 1h 25m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/327fa9c6e9964814813ad01f6aaa09a6 : SUCCESS in 2m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2cb15929762d46629e257e442ec6afa7 : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d10c015bdb284d7bbf774d88de3b847d : SUCCESS in 4m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28f31a5246694a0f95f69dcc88dbbfd7 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb1f1293be4d46d7b70a2782d3a2689a : SUCCESS in 3m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ef88d588c194697ba0e5f6b4931e214 : SUCCESS in 3m 33s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3bbecdb2a5384a32bd0887d4581cd606 : SUCCESS in 27m 54s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b9ab0f1493664e1b9bb018fc577f4878 : SUCCESS in 34m 54s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/946437c81cd6487086efb58d8118a435 : SUCCESS in 1h 09m 43s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ebbadc049db14e65b75c15334fb921ed : SUCCESS in 1h 29m 59s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/a01e608aba8c4aef83014abb2a622b59 : SUCCESS in 1h 05m 23s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/19b2d1a9681b426d85623c1f16d8c3ad : SUCCESS in 1h 18m 18s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/de149db1dc80476c8f502acc4784b4a5 : SUCCESS in 22m 36s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/99f4010adce7493e8be7079bba01b6e4 : SUCCESS in 25m 19s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2e788b5d68d646a7897ad91ead35f3a9 : SUCCESS in 1h 04m 21s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/87a1cba596be4d1888dbc01b10b2c5e7 : FAILURE in 1h 27m 01s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/d20ea1489d0447319a535b42bf16af6b : SUCCESS in 1h 09m 51s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/1222b815bef94a369be4157fd1ca4463 : FAILURE in 49m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d10ae677c3d9aa9ab94e06b36785d9f449561514","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 18:51:21.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f49d013d4f44eb09cb0731537e86f35\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d46b582d62464e4eb016a59fb3d7e29b : SUCCESS in 2h 03m 32s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/46ba8b0ab274457e9a0db75406b77171 : SUCCESS in 2h 40m 11s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b2938048ebdd4a8b9de69026fbb6068a : SUCCESS in 2h 19m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/43d9278655104b3a88b68494c0515f84 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eb921ddfccbf4237b6af7d5808878721 : SUCCESS in 5m 26s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/448eeacf94064b74aefbd034047b9e8c : SUCCESS in 2h 11m 34s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f115d3b638074a1ca39ca3dc6bc75841 : TIMED_OUT in 3h 01m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ef833bd36b9df193c95b15e8b952a4950232c8f9","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-10-29 09:41:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5e49e54be43e6f883442f31a949288f575569fed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-10-29 09:41:43.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c61d0578c5de5e12d529f6d6959c1863dd974cbc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-29 09:56:28.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cb4215c94777c2048e94dcde9d3935a2ee69cb82","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-10-29 10:09:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"da3456dad8027dd797d57e1dbd4b624e7a902248","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-29 10:35:19.000000000","message":"Patch Set 4: Code-Review+1 Review-Priority-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"7aa37f581542f2e9ada5db220d5e69cbf8f7f8aa","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-10-29 10:54:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"13b5527c2da7c9d53f45094cc6cd6918073857cf","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-29 11:06:50.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"be48fe7c2118b09b1b845bed849e1c4cc27d34ca","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-10-29 11:18:33.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"e1f7359d7ed931a04b250d5197332aa1a1c3f232","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 11:40:59.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcfd10cc86fd4563aa117e449258dd92\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f743fdacd34b41758f3fa922720b93c7 : SUCCESS in 22m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ed3a29cd69154d44bacff48a7ba4110d : SUCCESS in 1h 07m 31s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/96a90c06742d4453b25c1716eba15f55 : SUCCESS in 21m 38s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/cfc03488a0b3437c83a5df2d05ecb375 : SUCCESS in 30m 19s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/cf22c7b7258647d59f90caeeb6d15d22 : SUCCESS in 59m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ac5eb44fb6dc4188936ff84b1e159a65 : SUCCESS in 1h 34m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25d0377cf527420baae5c2662da37275 : SUCCESS in 2m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e24a13db470e436188fab9175e5d6054 : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/536ba521ffb548178e63182460e8d9a2 : SUCCESS in 3m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ece4aa41a2145dda9ee766630cfcafa : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/900ab17d1f1744129a655dbcb7772c1f : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0070778ece74491bf89d4f358c3ed79 : SUCCESS in 3m 46s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1d3a351032e44b5abb2ad95fcdd15eb3 : SUCCESS in 26m 19s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/87bf3afaa63a45d4b1a163d02ee81193 : SUCCESS in 28m 13s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6dbd6900e717412d8424701b15113985 : SUCCESS in 1h 07m 26s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1609b31d32ed4aeabf7b0442b1703623 : SUCCESS in 1h 29m 12s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/4d16e31d66e747ad9bd7425bfba56698 : FAILURE in 6m 57s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/860f90687e6e4718a01018c153881ab2 : FAILURE in 1h 08m 09s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e0190e13be534a56937604d458b176b0 : SUCCESS in 17m 15s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ca79331d22284de5828baefd37ee9147 : SUCCESS in 29m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/84bbdba00faa4b54a315c7b76aee4d64 : SUCCESS in 1h 07m 50s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/87789d68681d443582371d788aad6bfc : SUCCESS in 1h 18m 48s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/a470fd57470c4379ba622447a73f9d28 : FAILURE in 4m 36s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/814e385908c24534a5253129876e90a1 : SUCCESS in 1h 20m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3d3b76ed40af003c6a3249acdd35cfce48d998ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 13:12:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d38c2fa80e1b4824bce68c6f2eb066dc\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8ea9ca93251a4736a2aeb2efd2d5564e : TIMED_OUT in 3h 01m 26s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f39f046535a14fc88977ab15407bec68 : TIMED_OUT in 3h 01m 37s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6bd6aa2992e141bea21420eeac538d90 : SUCCESS in 2h 54m 08s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a1f3e03dcb694b1e8aac508bc4efba64 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/272a0fea2ede4c7daaae0fd5773a7925 : SUCCESS in 6m 19s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/760a3d27c86e45c48131b59c254f3e1f : SUCCESS in 2h 30m 43s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/000af2264ad64bc6978f28dfafd7c887 : FAILURE in 22m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9b448a3807d0fbae8549eb581f018ccf05904357","author":{"_account_id":37355,"name":"Pablo Colson","display_name":"Pablo Colson","email":"colson.pablo@gmail.com","username":"pabloclsn"},"date":"2024-11-05 12:26:23.000000000","message":"Patch Set 4: Backport-Candidate+1","accounts_in_message":[],"_revision_number":4},{"id":"757d529602c3b67a95ec2a0e95c5bcf94c2e01ef","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":37355,"name":"Pablo Colson","display_name":"Pablo Colson","email":"colson.pablo@gmail.com","username":"pabloclsn"},"date":"2024-11-05 12:27:05.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_37355\u003e with the following votes:\n\n* Backport-Candidate+1 by \u003cGERRIT_ACCOUNT_37355\u003e\n","accounts_in_message":[{"_account_id":37355,"name":"Pablo Colson","display_name":"Pablo Colson","email":"colson.pablo@gmail.com","username":"pabloclsn"}],"_revision_number":4},{"id":"77ad19f5c387efec6a61cbea8c3d594a38f4f73c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-05 12:39:22.000000000","message":"Patch Set 4: Backport-Candidate-2","accounts_in_message":[],"_revision_number":4},{"id":"0294bbca16d3f7f510e6f629888121e05faa5ead","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-05 12:40:17.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dfc8cda1634602c9bbfff62ada631718151b2a8f","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-11-05 13:59:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8a82d8ae9799dacc347145ff231588c8d8bd1e5e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-05 17:37:47.000000000","message":"Patch Set 4: -Code-Review -Review-Priority","accounts_in_message":[],"_revision_number":4},{"id":"024141f35ad29b82c49103b35d2daf33adc4ace7","author":{"_account_id":34476,"name":"Marc Vorwerk","email":"marc+ubuntu@marc-vorwerk.de","username":"marc-vorwerk"},"date":"2024-11-06 07:36:21.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"995fd166f788e4bc82b491a4f0bb5c37b223c2be","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-06 07:45:19.000000000","message":"Patch Set 4: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9d6a40206e9199a7f6501365dc5e3a96d2c8b827","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-11-06 08:52:36.000000000","message":"Patch Set 4: Backport-Candidate-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e657366e59f4ff2971367d58529207738a951f61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},"date":"2025-08-27 10:59:46.000000000","message":"Uploaded patch set 5.\n\nCopied Votes:\n* Backport-Candidate-1, Backport-Candidate-2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a3ffc65f0369f6aaf91891d11e1f8c21003cdeda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 12:21:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d4a7c971a6344d18a23b57f60a14c7d\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/86c48029d3d5496885ad5ad8e2432398 : SUCCESS in 11m 11s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/1542a72d986f4623a1f58e494c0306f1 : SUCCESS in 1h 13m 26s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fb14abfd254c41ad96fad7e53cf89bc2 : SUCCESS in 9m 35s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3fd9bd608c8742caa27af04d38af36b0 : SUCCESS in 29m 37s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c78f0d491e4243c797396137628ddb1e : SUCCESS in 43m 43s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6972d4f3617e4cc2b953570ad89b3ee1 : SUCCESS in 1h 08m 23s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d41fb0ef35c84a15980f7e1bd36db5ab : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6033be6d58204a5298d2572186ff0383 : SUCCESS in 1m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3de3bdbe7c6944cfa775b1bef5849beb : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/300a59c8b09b4c1aacd38cdce96aee87 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/522b42c5f74c4e8b94788138867519ce : SUCCESS in 2m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa2bd96720574ad09b0b55fe83a30b91 : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/367c1e71de5647988b88eeeb5a5a65db : SUCCESS in 3m 21s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5ca200e64186473da5f9072420f42bff : SUCCESS in 4m 01s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/dd4f53756f98418eaacb43e07e8dfa41 : SUCCESS in 10m 17s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/cdbbec9096ef41d8802153b573445e2b : SUCCESS in 16m 01s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0cfdb4002be842899e934d8b5e56a9ea : SUCCESS in 51m 43s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c3179100827f4f6d89034a8ca02cce73 : SUCCESS in 1h 15m 09s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/8e830cf14ee04846958e6908cbc257e5 : SUCCESS in 52m 52s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/5164cf39a0f84be4bb9cb7647f80b988 : FAILURE in 58m 20s (non-voting)\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/d628190e702a4bedb9ded68c7fa9dae7 : SUCCESS in 1h 11m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"674c5df328613d2120b734c9d7db4ad5f833745f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-27 13:28:11.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"8216f9ce3a606d6150858f3c15ef4cb95886376b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-27 13:33:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf89f3d512064dce9731c89b7980a461\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/c708d5ea31d2410eba0d3e3d74071f0e : SUCCESS in 40m 38s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/c09c0b2b8de341aca733947fe7cab2c4 : SUCCESS in 2h 30m 39s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b2b71443fe7840dc92a1ca753e1077fa : SUCCESS in 1h 00m 33s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dc54564239ce4ae69d2d6c0140a7e06c : FAILURE in 1h 47m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/15942d02f30048fab6e0fec8f26265ee : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/742561cce03d43e58a72571c59580f03 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f2399641df9a496c8b5428d1b5c50078 : SUCCESS in 10m 32s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/05e454cf568d43e9a309479714def7a1 : SUCCESS in 1h 04m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6a8adf32106d996446357b6f5940d6080ded4dc8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-28 07:28:26.000000000","message":"Topic ovn-vpn-agent removed","accounts_in_message":[],"_revision_number":5},{"id":"2e83c0d9bd82b40600c1e99322bb2035b489a83d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-28 07:28:36.000000000","message":"Topic set to kolla_ovn_vpn_agent","accounts_in_message":[],"_revision_number":5},{"id":"7b74a8dbd720622555691ee47d67738b8d53a9ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},"date":"2025-09-18 08:29:35.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Backport-Candidate-1, Backport-Candidate-2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"89ac262051f0de96f344a1f89343f63579d046e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 09:59:07.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/acf97ff08aef4f7780c10478b1652f12\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/d0caf29640e6419cb79b427469e6feb0 : FAILURE in 6m 55s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/44d226ba7fd2472e959b2b7ea1957fd1 : FAILURE in 14m 09s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2774dadc16fe4700861bc937b41d3696 : FAILURE in 18m 14s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/84e321e240d84960ae862fa2bcf45ca1 : FAILURE in 31m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1f5f2be5997d4d08a085244cd0465831 : SUCCESS in 45m 37s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/aae5aa2b63004e9aa7da01230a5b75d7 : SUCCESS in 1h 04m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc7b604f13ec4ddeb0f592a5ceb6b360 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb81508f82b345c3a04073c2caf8acfd : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a2502f3cd5c94a19ab8a1fbecadc4f89 : SUCCESS in 3m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3b8182ca2ff44b62a9bf51c0efe94a64 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26c37ae666444b95b0ab3f25bfaf8d49 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef44aac6888744fd801129c9994115f9 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fd556005b4a481d89fa19634b7a9e5e : SUCCESS in 4m 18s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/905ec4a609054d4c88d2d9b0fd828a7f : SUCCESS in 3m 48s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3a19ed736eea4f71a5793aa530c9cfa9 : FAILURE in 18m 57s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/08f8c0c4ce51483c9e7a4b05931dcf1b : FAILURE in 33m 51s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/df96f00523204d41b9d2cd00fdc38906 : SUCCESS in 50m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b80f69858d4b4e998fb511d01ed8779b : SUCCESS in 1h 21m 39s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/50675238c3ce49599cad2067a419baec : SUCCESS in 51m 02s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/0d2e1d892ce44192a058994b250f0609 : SUCCESS in 1h 21m 47s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4f9773b2f22cff4a7abbfbc65b9a5cd3eebfdf79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-18 12:15:32.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/581da40c412346f3825acc30ec177101\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/3cb2e308bc47403c89fd25a67d0f42d9 : FAILURE in 28m 35s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/f97b832dbf1d4d1fbd555cc8149497f0 : FAILURE in 29m 17s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fe872cbc0bea48a3a2f642dc7ae6d4c8 : FAILURE in 34m 12s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/df502d85ffa64b17b7b451798ff50125 : SUCCESS in 1h 14m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3c2c8196cb894c519a0c1b514a9856ca : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e2aaa7646856404c88cdda3d14202e03 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4e534dfebf114154a16d403e4991e467 : SUCCESS in 9m 29s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/199fe42ae5234d09a62a0b0eb50ffb60 : FAILURE in 1h 06m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"e500897836fb396a0405446b9a674cb5f394f55c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},"date":"2025-09-19 15:04:19.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nCopied Votes:\n* Backport-Candidate-1, Backport-Candidate-2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"a9b97113b259bdd69561f79fbeec18fa26e57841","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 16:37:00.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6dfff6e5cf843c9aee4c96341ecf378\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/67be00e0731c47d09c8b0bf5a2d993c3 : FAILURE in 9m 14s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/e34ef568bd844d82919367ebabd744b1 : FAILURE in 11m 19s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/58ff0ebfa7f74c85acd2dd7d6e9c0007 : SUCCESS in 16m 44s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7a48d349c83f45df9cb0412ad538921d : SUCCESS in 29m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c9ecd156c0b24dfc88448efc66f540a6 : SUCCESS in 44m 39s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/358296af8c44427489ccbf574ff4c800 : SUCCESS in 1h 06m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d06bcfa52de54609ae12e94dc0956558 : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/40be0af5d209443aae550d5e3dbc22b6 : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e0f6b5bd60684884a554ff9e8314726c : SUCCESS in 5m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/446a229d301448d0bf43dd076f4ac95f : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/796bfb0c160d42de8f6cb78d8ba6d7a6 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cffe27aed4dd4d2e9a8964a1e45c1667 : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a61df2413394eee8062c59bde00bf6f : SUCCESS in 3m 48s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b325897fa98741428c43daeb19737e24 : SUCCESS in 3m 02s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/eef5099aa3954662a5ea182f23eb929a : SUCCESS in 18m 35s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3e147cf33c6048109c85786c00b468e2 : SUCCESS in 29m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/93ae227d8c1344479643774f1e78795c : SUCCESS in 49m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5f37e67bde8e4680a03713f7f2fc141f : SUCCESS in 1h 21m 44s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/a9d4d2b908e843c78f055bf6db91eb93 : SUCCESS in 50m 38s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/966190b7a1a049f4b727edbc4c49bcf3 : SUCCESS in 1h 25m 46s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ecef5fa841f751d33da5389cfd107eb5751d6026","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 17:17:58.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/027882455ec74f99a014069a60318c71\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/9b5d2eebff894609aead199105939c7a : FAILURE in 11m 26s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/f2e350c8b5424ec0b8a5f07021fc7456 : FAILURE in 21m 14s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b12899bcae124dd7b294b8800674c16f : SUCCESS in 23m 39s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1f36d98887e2408d9f00c48608871782 : SUCCESS in 51m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c71f1dcd32134543bb4e7c0ed24c0f1b : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/56a2064ecf504491b9264b91437704f5 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/55c3d5b9b66a4ab5aee4331b470b2863 : SUCCESS in 6m 29s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/78d85c55e2264e27bab9129c5385462e : SUCCESS in 52m 12s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"79ed6da883a6ed01ab7a5798d3c1b85afbee1708","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2025-10-01 11:45:04.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4679864922a01a93755c874ed3c545f4dcb60f15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},"date":"2025-10-30 17:41:48.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nCopied Votes:\n* Backport-Candidate-1, Backport-Candidate-2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"d6ee5c66b1b5f9375a2d68b7c4224ecadf1b1f9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 18:44:14.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a294b650c90044c4858045bd4ee5c315\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/a1b213a4a06f43969166af1e8220e1d4 : SUCCESS in 52m 03s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/86be00dab2c14150960431bd54747a8c : FAILURE in 21m 22s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/40b20991e9cc4b00958a9aa2c644d4cd : SUCCESS in 47m 33s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/9e2e650c863247deb40d59bd76ebc19e : SUCCESS in 22m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3b3aaa0804b34c7fa32f4bf1a45fee51 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/19473047564943109f2b4dfb515b1952 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8c6f820858a3476a931b540ddb869f15 : SUCCESS in 8m 08s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"64ca534119f8f515952e625322df906089b79bd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 19:12:57.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8533128e9c4b46d0aeb95c3b102b1bbd\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/49da2e102d34434da6fcbb02d1f6075d : SUCCESS in 18m 27s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/70012f3858c44bf6a4b4fdf9535dd56d : SUCCESS in 22m 28s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/b77734850ae44c8fb567db6d25462805 : SUCCESS in 33m 48s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/961dbf4c5fbd454483a2693a6b7a05e9 : SUCCESS in 27m 19s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/a9092a3bdc524614b1313dac20fb8624 : SUCCESS in 43m 37s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5792180c2a2147c8b553ff438d7ef069 : SUCCESS in 21m 43s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/14832dcd4939403e867e6d87ae579b50 : SUCCESS in 29m 13s\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/d5dbabb2dbe84b9da1c51a8b3c597226 : SUCCESS in 57m 08s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/0eee2ac643d242f8a4571b0056acba30 : SUCCESS in 1h 07m 25s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/267edc9314dd4fb18de1b39e48cbe292 : SUCCESS in 1h 30m 07s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/12bbab33573749959e9e445fa369881d : SUCCESS in 1h 28m 17s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c7a60dfaf84f4b68a81352d4f48d6c3e : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/573fbd071cc04504aff0c336f708e102 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8afb82839f364515bf0d42fa9719a38e : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b5d7e35acb2847a5849cd363cee56ab4 : SUCCESS in 3m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/426820d436e344bebbda414a37e2143c : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ca2b68f1c264ea89233911dbd826d61 : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4b66c81bbd84dd58ff93603484513d5 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":8},{"id":"c7261d34d7f91e79540c926d2f74dfe9ea548a0a","author":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"date":"2025-11-21 09:13:58.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"cf523daf4748f23a7b1adea90fa010c9d77f5421","author":{"_account_id":37355,"name":"Pablo Colson","display_name":"Pablo Colson","email":"colson.pablo@gmail.com","username":"pabloclsn"},"date":"2026-01-23 07:27:22.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"574200558c50694bea2cd80e885fa2a6d2a57aa6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-01-28 14:22:14.000000000","message":"Patch Set 8: Backport-Candidate-1 Review-Priority+1","accounts_in_message":[],"_revision_number":8},{"id":"ac5b820f7aaecfbcbabd5d38d526f79ebaf40220","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37355,"name":"Pablo Colson","display_name":"Pablo Colson","email":"colson.pablo@gmail.com","username":"pabloclsn"},"date":"2026-02-02 16:44:18.000000000","message":"Uploaded patch set 9.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"921a804e4f61a90c8f1cadced7be593a45a605ba","author":{"_account_id":37355,"name":"Pablo Colson","display_name":"Pablo Colson","email":"colson.pablo@gmail.com","username":"pabloclsn"},"date":"2026-02-02 16:44:51.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"389161a5b09a5044d205575abe5dc8c971748163","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 18:02:39.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d99ef36a6b1c4dfc9550c43face3e7dd\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/ee8d597878aa4be68243ef7a6cd44a44 : SUCCESS in 47m 50s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/c3c23580c7fb4cc68e8cc02ff1341d13 : SUCCESS in 52m 22s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/4cae95d06f1b432d9757dd7d673baabb : SUCCESS in 12m 30s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/c880cc1a515e4ab4b057ff24d0664029 : SUCCESS in 47m 35s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/ee9b3d778eb94f4bb5b6691be6c768c0 : SUCCESS in 51m 25s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d78d4704f39d4324a1def2a637163903 : SUCCESS in 1h 01m 26s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/7918e1b58e9a402891bb1ad8507c1774 : SUCCESS in 46m 15s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/bc9437ba642b4c86a6e4d7021f6ec0cb : FAILURE in 36m 05s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/655c6297db974ab8b7f16aa0ea74dc73 : SUCCESS in 57m 46s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/ef09e77d11ed4217b952505bb6421f35 : SUCCESS in 1h 15m 15s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/5a69cd6c5c364593a934dc42c176c51d : SUCCESS in 57m 56s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/36170c14ebc144c699402134dff1796f : SUCCESS in 1h 17m 25s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2fc028bb07e145e1ba041fa8acbe9ed9 : SUCCESS in 57m 58s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/28cda1fa2971461d9065687460b75e0e : SUCCESS in 1h 17m 30s\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/971a88f5ce3142ef9d6b0c0c7986dd51 : SUCCESS in 1h 02m 14s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/59b6d734247643b19fce1ed84a68caef : FAILURE in 42m 57s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/dab97434f2b845d392dabd2ef633a3ad : FAILURE in 35m 11s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/63f32bff8faa4c5bbd4c5abe0318d4c4 : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/931267c169574c7fbcb9edc008d4dd75 : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c251a10abf8c4c67bafb2f0eb00e964d : SUCCESS in 3m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f8c6e42a049b419792e04b0911c779db : SUCCESS in 5m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/353c53cfa9d84c1d99c232970a6a421d : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86f45bd44fce4c3fbc01b9a37697b66d : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83e512dfeef447c8852210d766f236ee : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":9},{"id":"3c6de444400c67a116af218633b748a36bcd23fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-02 19:26:37.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45590235e775438ebb73c1537dd0e854\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/e0f20f445dda41a3b6b089b33ff5fbeb : SUCCESS in 1h 33m 28s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/4f87aebc0ba443db85eebf2fc9b3e083 : SUCCESS in 1h 21m 14s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/69e2fac1a7cb4e2dad27654390e7abb9 : SUCCESS in 1h 31m 06s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/cb35e6f08b804608b7bca0d37d029812 : SUCCESS in 1h 32m 10s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/a74a711c6fb543cd80470660ad5f8be3 : FAILURE in 1h 31m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0e7582335dd64ea9b0859e49b77090f8 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f2b571cd26b64bff9df3a19ec9adb5f4 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2cf17cfa61254900ae5ee8b9627d7e5e : RETRY_LIMIT in 2m 37s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"9bebcef7fc733a97493f2a4415b4aaa692415a85","author":{"_account_id":37855,"name":"Amir Hossein Ahmadi","display_name":"Amir Hossein","email":"sahahmadi96@gmail.com","username":"amir58118"},"date":"2026-02-11 08:15:53.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"3d2e29cf402d256ff4dc59c895d804c29b154727","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-02-11 09:14:07.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"ed8ffddf6a666a84e8a9d237bbb54a5bd98fd445","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-03-21 17:19:01.000000000","message":"Uploaded patch set 10.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"37890cab434cd50a3fa5c5b6e388725f5e8854f9","author":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"date":"2026-03-21 18:01:22.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"4e103bd157c75b9534fde305d99d51f4444bd27a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-21 18:39:54.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88a84ee9701d42ba95f8600dbafb0dce\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/a321229a11c14d8da0cd1db4315b1bb6 : RETRY_LIMIT in 3m 35s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/a67a92703dba49b0b8ed8064af2b85ee : RETRY_LIMIT in 3m 53s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/1544472639794611b78fe04843281621 : RETRY_LIMIT in 4m 09s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/464b1ccfc0284a69b78f26fde382c692 : RETRY_LIMIT in 4m 24s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/a3473117a1614797b729e5ff98d9c3d2 : FAILURE in 1h 18m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/55232a5342524f3f99f6098ade364761 : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/94ea0a1c2de84abe92f5950d9b1ccc44 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/0498edb6bf364c91b146933a3e5646f1 : SUCCESS in 8m 48s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"157d1c78cf5de830f460b5790d0ebb3983af0027","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-03-21 18:53:57.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"baa0795f86ee076f2b149367d643e6591bb1a44f","author":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"date":"2026-03-21 19:07:59.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"1b1b30561f3639e92255212f778490f497067373","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-21 19:45:44.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca5d2c7fb7d74dfb8cc781c0817b29eb\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/0eb1c96a572044c1a3f0708ce2322967 : RETRY_LIMIT in 2m 25s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/2e6deb63da4640b8a9e5e5699b1fb873 : RETRY_LIMIT in 3m 52s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/25230b7326f64b26948a57c8fd7ee046 : SUCCESS in 26m 44s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/3d37e3ea70ff4e40952edaf540e8f681 : RETRY_LIMIT in 4m 50s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/43a3515c807e4cde828d2db9161b507c : SUCCESS in 34m 39s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a7d5156873a0429a8f933addad1e45fb : RETRY_LIMIT in 2m 47s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ef8ee8f2e22940b7a936ec2f7bf93302 : SUCCESS in 28m 28s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/b851e92c82c5460781ce226c573f16be : SUCCESS in 1h 17m 20s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/3819d8540c5e4d23bb486a94f7b3a053 : SUCCESS in 1h 20m 12s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/bb81ce2a24f14a11afe231d68bc82e83 : SUCCESS in 1h 44m 32s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/94c532666de04cdfb3dba73390fc5c1e : SUCCESS in 1h 26m 12s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/5a9627738a064cecb595dd4196f3fdfc : SUCCESS in 1h 53m 19s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/37b328339c59407a9131b5a6e0b5d007 : SUCCESS in 1h 23m 16s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/8e2babac6eea4b12856dab52c588b956 : SUCCESS in 2h 16m 17s\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/46df63464e7146b5932679e3f3709a77 : SUCCESS in 2h 17m 07s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/c2592abbf0424519a0a1d941d3288482 : SUCCESS in 1h 53m 15s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/c82d46a0561646a9acc76369e9803e69 : SUCCESS in 2h 22m 19s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f337b6af415f4800acf4d1590f86630b : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ddd55107fdd4e2aa6e6d792096e4f9d : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c98f632ff794b548370a95e6d3ca026 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/917c567ae32a4d7b9e6deaabae42100f : SUCCESS in 4m 18s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b4f4bdd703e54b33b2716047fac2dccd : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f544c9e77c2940b38cfb0894693c5cc7 : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1837f603e1841b8a48098bfb65c2ae2 : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5accc0591284241b5145e744d218cac : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":10},{"id":"7c23974c134b9cd399a3c407f1ca3b7c2bf0c74f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-03-25 11:36:29.000000000","message":"Uploaded patch set 11.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"b32bbd9acc6e7dd4bfa9249ec2ff04a782636fa5","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-03-25 11:38:48.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"33ddc895aa05ae0ffa4dda94aaee80c89be08d3d","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-03-25 12:37:05.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"a5c882f51e9ec9776ef1d1aa1063b1b0d7d31326","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 13:34:03.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/23959411af0344c8a3a5664812b28a8c\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/6405e00b902d4ed8abb949c7b9bfd205 : SUCCESS in 48m 33s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/3a3404937bab457d9b6fc3b28b73234c : SUCCESS in 37m 58s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/8926bcd58be54c168b1848adfe4989d5 : SUCCESS in 14m 02s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/134e6d45a59e4b7da55966c71d2e6403 : SUCCESS in 51m 05s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/e883030b349c40f5841f20c8dd005174 : SUCCESS in 37m 28s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a72ef23a5e104e95b60c9afd9bacfa3d : FAILURE in 47m 26s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a61649f264814f2b87d8242f1e09c2fa : SUCCESS in 15m 00s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/f6dda1a7352948b8af3f07680b7538bb : SUCCESS in 1h 10m 51s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/fe2037e536214b5c91589695d94afc75 : SUCCESS in 1h 11m 38s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/cc322f549c3c4c7d8c858fb24681e163 : SUCCESS in 1h 45m 11s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/93d5894704db4a49ab0924899bc23d16 : SUCCESS in 1h 18m 33s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/f320608c389345ff875c80f1be9fdf1b : SUCCESS in 1h 56m 14s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1d92e975fe4746c2ba4ff76c448ee438 : SUCCESS in 1h 24m 41s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/2d970ce71f494e5aba478233ac890ea3 : SUCCESS in 1h 53m 47s\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/bc18b2bad0d546fb9e749cc749924f4e : SUCCESS in 1h 19m 07s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/44292e525ee641b5a50af6f82bb2fd5a : FAILURE in 1h 30m 04s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/d9df08add634464f80e0a9f239ee0c7e : SUCCESS in 1h 54m 41s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/56c550acb09e4d26a8a4c80f0c6b322b : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8da1f828b4a34e5d8ed31f78d8b9268d : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff7338b44b2b4e0e92100275ec243f4c : SUCCESS in 3m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a2b7643ce3ce4f9d96357d7eafc0192a : SUCCESS in 2m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/37f1c695ae7940d69b0e481b2a38d617 : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f0debf8a241415ea38b8a8613d87228 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58c46b2d6fe7439ebbd7ed19b6538b51 : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c2f4f7704aeb4c30bd516b459db55ed0 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":11},{"id":"33a5023dc0c9e24aaae834bb864d9680c6cafe21","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-25 14:19:44.000000000","message":"Patch Set 11: Backport-Candidate-1 Code-Review+2 Review-Priority+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"ee2b994cf1306097d7bdbd1b372bb58c72eb7e67","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2026-03-25 14:20:46.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"5582d1a2b922a8bc8f2e785e86121465ee510fc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 16:17:41.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/71ace47a60cd4469a784cce37a3e7a57\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/739b85fbbf2f4c279405d427428309c0 : SUCCESS in 49m 18s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/9c2e4f689e504c50bfb8c38aa27973d6 : SUCCESS in 39m 51s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/4efba56c5f13444ebbcfccaccec05406 : SUCCESS in 23m 23s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/5a5866c41746446faa65fda3be03837c : SUCCESS in 45m 52s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/04ddee0ea00d4ed5a2a31b8946608ae2 : SUCCESS in 53m 51s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8a095ddaa37a434ba13383d843b25dc8 : SUCCESS in 39m 13s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e62f8fc584534635a43ad8d4367509db : SUCCESS in 30m 15s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/a7e5faa0ae7d4f85b53044069bd35aea : SUCCESS in 1h 08m 15s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/e28f39ba2de7494aafcd175195296df9 : SUCCESS in 1h 16m 47s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/78564e005b3d4d57b3159f89891d0b4e : SUCCESS in 1h 38m 36s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/ab432e1f3fbf45749cba15de023ce5c9 : SUCCESS in 1h 15m 08s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/caa4cd7c6f2f45d182fcc39871d0551c : SUCCESS in 1h 56m 59s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7fee5877fdfc4651a2be5f214deef796 : SUCCESS in 1h 27m 35s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/d4f20d8bbfb94204a56de5878bdf1813 : FAILURE in 35m 40s\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/bafae355873049729bff7c12f7ca9201 : SUCCESS in 1h 23m 50s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/b396717d6ddc45c3a193de3a210fecb5 : SUCCESS in 1h 50m 17s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/0e56f4447d8240ec98bca35d246acb90 : SUCCESS in 1h 55m 50s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/91448ac46b1d47e58a016480a9bf87af : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99546c124f0d4135a2e862a63dda37d7 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5de25b30d30a4e0fa67190789afe1d14 : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/717c038cbc60467394a08a3c4786431f : SUCCESS in 2m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b00c65b29ce14b12ae962835260deed2 : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0edbf6a69e794007981959d9578e17f0 : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6583edce44924f7abb73f04720dcc121 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f62362eaeb147ac80eb5fd392d8306c : SUCCESS in 3m 23s","accounts_in_message":[],"_revision_number":11},{"id":"cf9591306c0d41f13902cc2e6afed539fa6070b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 02:51:10.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27f19960c50040dba20fc16a56c0ad87\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/6452bd407f78405d916009e8b1549bc9 : SUCCESS in 1h 40m 56s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/a9e3556b09354ca3a67b28aea6ed2bc8 : SUCCESS in 1h 07m 51s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/879fda0414fc461f89b27be46916242c : SUCCESS in 1h 10m 28s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/785441bf4f7e4bc2aa09aa4b4f521fa6 : SUCCESS in 58m 27s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/b6333dbc1fe141f0a29f7637b44fedf6 : FAILURE in 58m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/211e4dcb9871481ea1ee735103c00ecf : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/95c65a67c2b543e6bfddce66b19ab985 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7d9b33e38d344bf88a6a17007c940d6b : SUCCESS in 7m 06s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"f669b51d0fd521da48f21ee373785764b18fee2f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-03-30 19:13:07.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"70fc143dce55c61b9e1c15d4edad8c5bc2118ba9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 20:46:02.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7975a85a7c2e4a8a95b9159fdcc28e56\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/b6680ef73bf0452ab6744f8cb337cfe5 : FAILURE in 6m 11s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/f1bd72ff8a3d4974a34606331c424e8b : FAILURE in 6m 09s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/c4d9e194ae5645a2ac98e99dc18c3bcc : FAILURE in 6m 06s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/7ddc04e5182a44fd909647a00c45df0c : FAILURE in 6m 45s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/71bd60295edc4821bc4f3308af87a4b6 : FAILURE in 9m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f213e34dc5064a4f9064c5942804b111 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b4aa6c10296348c7bde0ae059c90f10f : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/2797885fb9dd4ffeb410547ffaa48daa : SUCCESS in 6m 18s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"36d04fec044714f8d56bc511b24c2e3a07bc081e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-30 21:17:43.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88b7141966cb4a6b9d35fb272e970e20\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/d8abf594a06b447388d893ed320c9bc0 : SUCCESS in 40m 15s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/d29b8b97c66a4ee88264533667658a0e : SUCCESS in 54m 52s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/98e8ffe27cb946fc923b68497a2e2d84 : SUCCESS in 28m 21s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/2e5ef932b67b4d9585a0e246d2de4b7c : SUCCESS in 38m 57s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/c2e99e60c4db431cb9dd6480b9abc8ae : SUCCESS in 1h 06m 32s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/80fb43177b36435ca8fe42f460838a69 : SUCCESS in 1h 27m 13s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/efbe6a780e6c4c96825f54fdbb84cb48 : SUCCESS in 16m 06s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/5ede5f21e74a44dc99d1effdb0391a9d : SUCCESS in 1h 19m 47s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/f87995ea462a42fea9633da3832c6919 : SUCCESS in 1h 15m 13s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/fa7d07d46028427898b1ef661fc391f4 : SUCCESS in 1h 44m 16s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/7f211370843248369bd15aed08bfbebc : SUCCESS in 1h 19m 11s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/8c8b5ae7152f4408b049eab295cec95e : SUCCESS in 1h 53m 44s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0461e1fae39b4d8b8fe40ced97f97ffa : SUCCESS in 1h 28m 27s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/6d6db1e98995498986f8d700e4b873f7 : SUCCESS in 1h 58m 41s\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/62a62e62d49242b59dac456e73503d3b : SUCCESS in 1h 17m 27s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/1d9c1caec3be40ebbfdf91049652ea43 : SUCCESS in 2h 02m 16s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/96fdcc15ffdb4a6888e9f9d88862370a : SUCCESS in 2h 03m 23s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/cd78a0843657458dbbdb4c41ca20c8c0 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/667bb5338b644a7993b4f2f2699ef19d : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a3330b1e2d84de4a13b8def383d68ba : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4a931118f66b44938419939c4c583b0d : SUCCESS in 3m 34s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ce41039351e84d38b9edfb81157d541b : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d06211e657c44e7831df7342770d532 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b83cdafa01534975bed812228dd8d4cc : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c57d299ad255400eaa2c6c2a9e64a1e8 : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":11},{"id":"1da698dcdf4ab7fc93d884367114a1a0679c43f0","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-03-31 05:02:39.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"7b4abd047aa3d67699e068213a7f69705bff00fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 07:02:13.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32c60c672ab94bc5bb94e39bd8831d58\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/bb493319745648d88b57045c43f3a9ca : SUCCESS in 59m 01s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/edc5a1080e22450a8d49424d3500b79c : SUCCESS in 1h 01m 32s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/772c0d81a8504c15a300168897a81503 : SUCCESS in 57m 11s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/896074d4aac4417f80538552de180e0b : SUCCESS in 1h 17m 31s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/520192fab3644176b17479d988813aee : FAILURE in 1h 57m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9387a3c787aa420ab4a723561c136e09 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ad63a313aad14a45bd63656f42b2d727 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/d23a677452994501872bf197819d1a3d : SUCCESS in 6m 44s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"3bdac787456c7bb4f28a265e835d979031d75c0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 07:11:52.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa791286320d43f99f0a70e3abae01d4\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/82ba1d55122a4120ba17a2ed42815969 : SUCCESS in 44m 34s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/c0689c36abd84c9988b8bbe137b44d29 : SUCCESS in 33m 30s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/ebddf935b1ec42f4ab94cd1c4f189ce9 : SUCCESS in 31m 58s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/e027b81294e4419186e9de9d99679bc0 : SUCCESS in 51m 14s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/2640abce751647d498e561baa020ae5e : SUCCESS in 40m 18s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8672b161692748ac9526ca01d40ef70c : SUCCESS in 33m 49s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/e0627177160a44a698aacdb11dccc8e5 : SUCCESS in 25m 13s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/37e4a34d74ac474ea8eeeb2efae8d445 : SUCCESS in 1h 17m 05s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/8000c9c928994dfbaa17707129d7255c : SUCCESS in 1h 20m 01s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/637561b3e2864b8daa9e532496efd762 : SUCCESS in 1h 51m 40s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/cf165bdab7a047cdae899b5dbc1e6a96 : SUCCESS in 1h 27m 22s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/7162739d11f94f9a83d55017b9bffac0 : SUCCESS in 1h 57m 55s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/addc4fdc99f64938921f4cd71456a173 : SUCCESS in 1h 16m 35s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/b88adac079a148adb9479085a7db6a16 : SUCCESS in 2h 04m 36s\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/18338e616f8247afb97ed64dbe671e37 : SUCCESS in 1h 35m 37s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/5f59ebee629a43f996498f49c3cb3b44 : SUCCESS in 1h 53m 21s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/2aba83db976d4f5d90076580b208baf8 : SUCCESS in 2h 07m 43s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/446ce50fc4974798bd07389a07ad0f73 : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cd75debfbdb47f1a388a7388ff9a003 : SUCCESS in 3m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d685f2b1ff20426399e6b69d43888b6e : SUCCESS in 2m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b93a867ecdba40af8cbb2c13750497eb : SUCCESS in 3m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/48859f605edb400591fa75c72c00b53d : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50549b4a8a8d4fecb3b0ea31300d268e : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/871ce65c2c774ac1932e109d047f22f9 : SUCCESS in 4m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/107c4089a2be4fef9f1584b6b1cbe327 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":11},{"id":"111fe43cae4f1028a6bd35618e0a536c81278e0b","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-03-31 11:04:40.000000000","message":"Hashtag added: neutron-ovn-vpn-agent","accounts_in_message":[],"_revision_number":11},{"id":"e83d9b466ef53cd583d562faeb60ec84a97adc85","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-03-31 11:12:17.000000000","message":"Patch Set 11: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"4ef6d3cc5e9591055796aea7ba93f7075f257650","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 11:12:56.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"d0773966620b6fa4d46493c261b8e68c5e46cfb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 15:14:03.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/320d9a2dcced440e94584b44068b6a9e\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/30dd04f263a645d28b1c48a84ac0e0b2 : SUCCESS in 1h 20m 36s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/eeceac5ab412436eb4b326e0f96c5340 : SUCCESS in 26m 45s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/c78e952f63e84441ac4e7dd665d5b8f6 : SUCCESS in 1h 16m 26s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/c486294b033f4fd494a74774d859a885 : SUCCESS in 39m 36s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/3ef5e385a37e4db696fa4960f4e2a1f6 : SUCCESS in 59m 59s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/fa4298ac2e9c429595e45617f80612b3 : SUCCESS in 36m 06s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/89a1770b59d7434e9cbb3c8b01566977 : SUCCESS in 2h 19m 09s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/3f60769b01964a3fba936a6872f800c1 : POST_FAILURE in 1h 42m 18s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/6a41d446fa2747f8af0fbb1601e87b8c : SUCCESS in 56m 59s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/30c533ce487241d5b7e0f04b2421e794 : SUCCESS in 1h 18m 23s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ce9d9c5266e24deb9be8dd61e0e5a99d : SUCCESS in 1h 58m 47s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/a701f06de674479c888535126164aa9c : SUCCESS in 1h 19m 07s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1aa1348a547b42d58ee4c1d21d8c5d71 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3adc22b675f4431283c6a89f13caf265 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4acaab3a49b4ab583c709763fd9eaad : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ea5d76d0c6ff4d1d8165740f29e86670 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1dec334239349b2b86e5da138c11704 : SUCCESS in 8m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a940b9ffad8c433d8a9c0d8e6c7d40ed : SUCCESS in 4m 28s","accounts_in_message":[],"_revision_number":11},{"id":"d057cf8014ecc80a678fb1eeb29590abd87c0975","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2026-04-01 10:25:06.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"9fbc040a2d664e971e6cc09c1d5471b2f7864f69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 13:02:31.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ef8e640d6e9448180f40f1ecb5c7d15\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/1b83877dcd7d49b889410aa96108fafd : SUCCESS in 1h 16m 41s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/3733ca58ff0d4ac0bca7bc80d2b04770 : SUCCESS in 1h 24m 12s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/3a7aac7db48d44a0ba4ef3c2add209ec : SUCCESS in 17m 00s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/601b2403340745f3b5eadc5d05c2ca7d : SUCCESS in 1h 19m 00s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/f769f010d23a4f20a365ae52f5fa918a : SUCCESS in 1h 07m 12s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/29fc6d3ad2f14d2a9347a2f1dd18ce80 : SUCCESS in 1h 22m 15s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/73f561834da64e5e87d8850e600fe7d0 : SUCCESS in 32m 24s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/c4ac6bc801e540eea4c0b78755c0caa8 : SUCCESS in 1h 48m 14s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/2baca68bedc44211a353cd3d90bb9ecc : SUCCESS in 1h 23m 13s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/abccac1376a34669bfe2155cd1a9fcf4 : SUCCESS in 1h 42m 59s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/690e3aab00f743519135c6911691913f : SUCCESS in 1h 31m 23s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/b8c578c94e7c465c8f99510fdc544105 : SUCCESS in 2h 13m 08s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/910b903d5a9f49cb8e073de56ca572e6 : SUCCESS in 1h 42m 16s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/af7b35f1fae3465293030a681e7f6843 : SUCCESS in 2h 08m 53s\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/2c2ae64533b94c8cb10889e7f26c310c : SUCCESS in 1h 44m 11s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/335730ec27d343f091dbfb2e6e3b5815 : TIMED_OUT in 2h 33m 27s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/cf40f6dc479e4bae8ec910ba4fe30db0 : SUCCESS in 2h 12m 56s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/599a3fffed094c72ab4ebb7c60b92fd2 : SUCCESS in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/634ab9d28acd459aa3b6cde6887d2a4d : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e33dc2a84464c14b600e66bf10d52b6 : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5932e52250ae4f659dc92b25bcde2550 : SUCCESS in 2m 41s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/98798a8f946d4d9cb72d4d4bf62b05fa : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87da17ffe3e4499f91f222d66580be96 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f8176e6d77748539778fa41c39ea53e : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c48028c611d4d09a01c1d56d4e160b6 : SUCCESS in 2m 45s","accounts_in_message":[],"_revision_number":11},{"id":"9bb31e210ffffbedcba43aa7f913cc49ee174a4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 13:02:51.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"2a446fa2c0a4952a8468a53a5ad9a5c68bf865d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 14:16:28.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e480194831404443b90a0c1eca8cb7b2\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/c57f78f530ab402c9b6ec11259c00157 : SUCCESS in 1h 29m 30s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/e9c261186dee499d8848759ef3c717b0 : SUCCESS in 1h 23m 50s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/b585f6c890554703bf98f415c02e0e6c : SUCCESS in 1h 16m 36s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/9d612b8780b64f50b76a272578447b31 : SUCCESS in 1h 22m 34s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/f28adb6d39474a08ab9309ad3b62c37c : FAILURE in 1h 38m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6c8e650f0e7546f6868cc553e910843c : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7be5c83e80294bb3acf21c6a17d02e90 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/28b8d2cfd93c467ea8c4269147fac1aa : SUCCESS in 6m 09s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"9f9c4bf18c77a32735c5b16d3a2207b579b9565c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-01 15:26:38.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ac8f69001614a3db437c7daba7aa849\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/d52ccbd99f204a9993f934331091c283 : FAILURE in 1h 13m 52s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/3a87392776a84610ba5d16c29d8eaeff : SUCCESS in 22m 34s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/b10c01fc4c0340edabd6dbee773cf300 : SUCCESS in 1h 10m 53s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/8279b76da7f84d9e8aab573693dbf2d4 : FAILURE in 37m 22s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f388bc163a2f41cb97a61666995f6e03 : FAILURE in 1h 28m 08s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/ffd952c0662d49c78351b2be6af7fe2c : FAILURE in 32m 10s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/bdfcb53beba549cea9561843432a3731 : SUCCESS in 1h 33m 20s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/2fe20715f0bf4769a5f657f87497e6b0 : SUCCESS in 1h 53m 47s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/97d2998d2e7e4e0fad975330ab5d7e60 : SUCCESS in 1h 50m 44s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/a69a236ae1504dd2b3876b03dcde7b4a : SUCCESS in 2h 23m 00s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/4da60d348eb7484aa9663e469a8f0086 : SUCCESS in 1h 38m 56s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/e30a0faf8db84bd8abcb0dc89672706c : SUCCESS in 2h 17m 10s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/657423059e894074ad71b5ceac7910ed : SUCCESS in 3m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e1960d52bac46f8bbab996eb23842dc : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de2f4d834b114d938f71c52364941a78 : SUCCESS in 2m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eac41887bf77419aa7291cc5062a94e6 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edbb735f0ded47438c7a4c8e778163d8 : SUCCESS in 6m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ca7056115564d6eb0e3098ad5a89660 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":11},{"id":"e018848e9688e7760d67bea85fa61a4645a82020","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2026-04-01 22:09:22.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"06b2ea1da9551891937099bec2e680cf988661c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 00:32:09.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7462fb98d5848cf93c88431db768fdc\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/dc16c145326143e39c8d498924bf4b2d : SUCCESS in 1h 15m 10s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/6d2ca9329cfe4df4aef9d87c6f249ff7 : SUCCESS in 1h 07m 43s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/59ecfc54ff97434a82ef527e190daa34 : SUCCESS in 24m 16s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/3de357ce05564ce8b51f9feba9154673 : SUCCESS in 1h 13m 01s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/907c98c160c246ceaddc8ebf123eddc7 : SUCCESS in 1h 06m 00s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/30bdc8456abd428dbc123fe42953c145 : SUCCESS in 57m 05s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5c56931b90e44484b6dd39e042b00db6 : SUCCESS in 30m 25s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/8c26eb3415ef4295904dbb25d5ea33bb : SUCCESS in 1h 19m 44s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/7eb46471e4504a1d8d9117e5d1f628a7 : SUCCESS in 1h 25m 38s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/8cdc03e2f837467eb0fa01a728d8d84a : SUCCESS in 1h 44m 06s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/af285738848a4337896ac62400c2bf03 : SUCCESS in 1h 44m 21s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/5161a2cec1ff43948efac5917d8eef25 : SUCCESS in 2h 12m 59s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e6e8f3f2b4c74592a992acbc059ba2f1 : SUCCESS in 1h 40m 27s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/7f22e29b209749b5b52b03e6e12b8418 : SUCCESS in 1h 53m 23s\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/e4ee8b95454a449bab74483eb42952c3 : SUCCESS in 1h 43m 53s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/99391a0c581b46778209dc22f6032005 : SUCCESS in 2h 21m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/26b38801d62246e08edb7318b914a11e : FAILURE in 1h 38m 01s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e0f5d987f391457b977c9647bde3028e : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/518a71fecc9f4d4e8cefd5cf146ce01e : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ffffd6433d4b4138aae722b23daa5c14 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/64a84a22276b48048b0ba5088476c3da : SUCCESS in 2m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8b11e2f9c4984bb9ab837f5fce351010 : SUCCESS in 6m 13s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9880504b89142f09fe8f9bfb59e037e : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d3b6e3600d74c24b5b0372109d45780 : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/def9cd09f611492ca59702ae11361032 : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":11},{"id":"5ffd002a87ad1012d7df107b6e018090753b315a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 00:32:21.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"788456a6be4f36ed71b10705012d28ba4fcd8fd1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 02:36:24.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e211cae6d07740b1a25a3ea8aa9fcc93\n\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/7442a6226a144f37ab4bba50bf9632de : SUCCESS in 55m 57s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/5ebbb2a1a5634777bfbac1bb1f4d6c69 : SUCCESS in 12m 12s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/bb47d7d5dfa04b48b9383321c42d26f4 : SUCCESS in 1h 05m 55s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/92485e299d1d4c91b7510a38732806a0 : SUCCESS in 35m 36s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5832996ce8a9479bb184c09b2ae8d101 : SUCCESS in 56m 59s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/0aa288cca56e4070aa98db68b239b3a1 : SUCCESS in 29m 03s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/71632b8ff319432f861a07f298a3853e : SUCCESS in 1h 21m 22s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/25dd14065c99492eabb5f6296447499b : SUCCESS in 1h 45m 50s\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/fc1d728b9f78450c8f345334c702e2e8 : SUCCESS in 1h 33m 28s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/abd92e263478497f8ea1a2dd6c861f2d : SUCCESS in 2h 03m 09s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5c034e07695b4346bd7fa2c005528b95 : SUCCESS in 1h 28m 01s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/eaee5a2524f34206ac9951506d5ce9e2 : SUCCESS in 1h 52m 39s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a2a3344244a04120981095f467f5993a : SUCCESS in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10851de0fb5b420f8a9505c7787256e8 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f67cf2cf8b8420f93e8dbbf59c2925e : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fc076026f81844c1829bd142e4c5075e : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d46b01a8b8843779e9f5cc5b0419365 : SUCCESS in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a49a3ae598184f5b90f5fe79879bd0bd : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":11},{"id":"8f0578bac46641a3bc2445972b84d67e82d28504","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 02:36:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"6e4bead5c0ad519d6b8af16af935eda66d51ca13","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 02:38:11.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/402ba6a0bb6049cda2f5ab93388df8c8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96dcf55d03bf4e81a3d287d754a79ce7 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/606fe198146d4db38b3e2be4076d2d75 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":11},{"id":"d17876c7c0d00e8d86c5e1e03a1ed22acbd67268","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-02 06:22:53.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/541fa10c2d3e4a5ab78f34dbe8dc9127\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/bcdd208999014dc9a609ab18637603c8 : SUCCESS in 1h 27m 16s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/d3f5d0a4fd474f3d84c96bc1f56fd921 : SUCCESS in 1h 25m 28s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/712cafacec534315888429a37a67af39 : SUCCESS in 1h 25m 37s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/2566dfdddc9943379e3540fd735e23c7 : SUCCESS in 1h 32m 14s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/9fa3cb3df92d4aaf85d58a1b3325ac80 : FAILURE in 1h 24m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2585ef7c6dcf4682a9d3fe668dfafb2b : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9188a9811cdd43ad8012075776061ecc : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/a6d3aef12b1d4a66abcb9c645d54a139 : SUCCESS in 6m 07s (non-voting)","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"63024ee8a26f08f5f65fec3f58be6bbb690365b7","revisions":{"0215bb2da928a313c2bcdeadfe55fc68cf6b3b8a":{"kind":"REWORK","_number":1,"created":"2024-07-17 07:51:39.000000000","uploader":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"ref":"refs/changes/02/924302/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/924302/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/924302/1"}}},"commit":{"parents":[{"commit":"2f28e30912bd06350a4325d2606235197347458d","subject":"Merge \"fix: apple silicon architecture selection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2f28e30912bd06350a4325d2606235197347458d"}]}],"author":{"name":"Maximilian Stinsky","email":"maximilian.stinsky-damke@wiit.cloud","date":"2024-07-17 07:37:05.000000000","tz":120},"committer":{"name":"Maximilian Stinsky","email":"maximilian.stinsky-damke@wiit.cloud","date":"2024-07-17 07:37:05.000000000","tz":120},"subject":"Add Dockerfile for neutron-ovn-vpn-agent","message":"Add Dockerfile for neutron-ovn-vpn-agent\n\nThis patch adds support for the standalone neutron-ovn-vpn-agent.\nThe new agent was added in the 2024.1 and is required for vpnaas\nwhen using ovn.\n\n[1] https://review.opendev.org/c/openstack/neutron-vpnaas/+/765353\n\nRelated-Bug: #2048392\nChange-Id: I552b9fbc8fa0b49d3e0e3212dd11962b02837b9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0215bb2da928a313c2bcdeadfe55fc68cf6b3b8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0215bb2da928a313c2bcdeadfe55fc68cf6b3b8a"}]},"branch":"refs/heads/master"},"6bad36142c8652a7304abfc19472fd488561afb0":{"kind":"REWORK","_number":2,"created":"2024-07-17 08:55:36.000000000","uploader":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"ref":"refs/changes/02/924302/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/924302/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/924302/2"}}},"commit":{"parents":[{"commit":"2f28e30912bd06350a4325d2606235197347458d","subject":"Merge \"fix: apple silicon architecture selection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2f28e30912bd06350a4325d2606235197347458d"}]}],"author":{"name":"Maximilian Stinsky","email":"maximilian.stinsky-damke@wiit.cloud","date":"2024-07-17 07:37:05.000000000","tz":120},"committer":{"name":"Maximilian Stinsky","email":"maximilian.stinsky-damke@wiit.cloud","date":"2024-07-17 08:54:55.000000000","tz":120},"subject":"Add Dockerfile for neutron-ovn-vpn-agent","message":"Add Dockerfile for neutron-ovn-vpn-agent\n\nThis patch adds support for the standalone neutron-ovn-vpn-agent.\nThe new agent was added in the 2024.1 and is required for vpnaas\nwhen using ovn.\n\n[1] https://review.opendev.org/c/openstack/neutron-vpnaas/+/765353\n\nRelated-Bug: #2048392\nChange-Id: I552b9fbc8fa0b49d3e0e3212dd11962b02837b9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6bad36142c8652a7304abfc19472fd488561afb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6bad36142c8652a7304abfc19472fd488561afb0"}]},"branch":"refs/heads/master"},"e477cc5e55c598817ff3e1718b0578695bc3739d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-10-23 15:44:46.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/02/924302/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/924302/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/924302/3"}}},"commit":{"parents":[{"commit":"71d7edd61512cdd56f1cd2c04beec83c5ab72e23","subject":"Merge \"Fix missing tss user in Ubuntu/Debian nova containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/71d7edd61512cdd56f1cd2c04beec83c5ab72e23"}]}],"author":{"name":"Maximilian Stinsky","email":"maximilian.stinsky-damke@wiit.cloud","date":"2024-07-17 07:37:05.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-10-23 15:44:38.000000000","tz":120},"subject":"Add Dockerfile for neutron-ovn-vpn-agent","message":"Add Dockerfile for neutron-ovn-vpn-agent\n\nThis patch adds support for the standalone neutron-ovn-vpn-agent.\nThe new agent was added in the 2024.1 and is required for vpnaas\nwhen using ovn.\n\n[1] https://review.opendev.org/c/openstack/neutron-vpnaas/+/765353\n\nRelated-Bug: #2048392\nChange-Id: I552b9fbc8fa0b49d3e0e3212dd11962b02837b9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e477cc5e55c598817ff3e1718b0578695bc3739d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e477cc5e55c598817ff3e1718b0578695bc3739d"}]},"branch":"refs/heads/master"},"6ef5e08facea1303eca28f356f5b5101a1dfe3e9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-10-29 09:41:43.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/02/924302/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/924302/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/924302/4"}}},"commit":{"parents":[{"commit":"ee7fe59f7ba366bb3f203eb072ae2c9babc2175c","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee7fe59f7ba366bb3f203eb072ae2c9babc2175c"}]}],"author":{"name":"Maximilian Stinsky","email":"maximilian.stinsky-damke@wiit.cloud","date":"2024-07-17 07:37:05.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-10-29 09:41:43.000000000","tz":0},"subject":"Add Dockerfile for neutron-ovn-vpn-agent","message":"Add Dockerfile for neutron-ovn-vpn-agent\n\nThis patch adds support for the standalone neutron-ovn-vpn-agent.\nThe new agent was added in the 2024.1 and is required for vpnaas\nwhen using ovn.\n\n[1] https://review.opendev.org/c/openstack/neutron-vpnaas/+/765353\n\nRelated-Bug: #2048392\nChange-Id: I552b9fbc8fa0b49d3e0e3212dd11962b02837b9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ef5e08facea1303eca28f356f5b5101a1dfe3e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ef5e08facea1303eca28f356f5b5101a1dfe3e9"}]},"branch":"refs/heads/master","description":"Rebase"},"49406bbed2e441498c0402e863e9b735d0c65a56":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":5,"created":"2025-08-27 10:59:46.000000000","uploader":{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},"ref":"refs/changes/02/924302/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/924302/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/924302/5"}}},"commit":{"parents":[{"commit":"d3e3f6b4145bc887afcd550cda67388c0f82771f","subject":"Merge \"[release] Use Flamingo Ubuntu Cloud Archive\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3e3f6b4145bc887afcd550cda67388c0f82771f"}]}],"author":{"name":"Maximilian Stinsky","email":"maximilian.stinsky-damke@wiit.cloud","date":"2024-07-17 07:37:05.000000000","tz":120},"committer":{"name":"Joachim de Groot","email":"joachim.de.groot@menzel-it.net","date":"2025-08-27 10:58:32.000000000","tz":120},"subject":"Add Dockerfile for neutron-ovn-vpn-agent","message":"Add Dockerfile for neutron-ovn-vpn-agent\n\nThis patch adds support for the standalone neutron-ovn-vpn-agent.\nThe new agent was added in the 2024.1 and is required for vpnaas\nwhen using ovn.\n\n[1] https://review.opendev.org/c/openstack/neutron-vpnaas/+/765353\n\nRelated-Bug: #2048392\nChange-Id: I552b9fbc8fa0b49d3e0e3212dd11962b02837b9f\nSigned-off-by: Joachim de Groot \u003cjoachim.de.groot@menzel-it.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49406bbed2e441498c0402e863e9b735d0c65a56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49406bbed2e441498c0402e863e9b735d0c65a56"}]},"branch":"refs/heads/master"},"1f31185993f8f4b6f96266e2a32e3130005f3d66":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-09-18 08:29:35.000000000","uploader":{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},"ref":"refs/changes/02/924302/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/924302/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/924302/6"}}},"commit":{"parents":[{"commit":"4f9d8c0e1ced1e558f30b56b31f4fbeab0e78d09","subject":"gnocchi: Bump to 4.7.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4f9d8c0e1ced1e558f30b56b31f4fbeab0e78d09"}]}],"author":{"name":"Maximilian Stinsky","email":"maximilian.stinsky-damke@wiit.cloud","date":"2024-07-17 07:37:05.000000000","tz":120},"committer":{"name":"Joachim de Groot","email":"joachim.de.groot@menzel-it.net","date":"2025-09-18 08:29:35.000000000","tz":0},"subject":"Add Dockerfile for neutron-ovn-vpn-agent","message":"Add Dockerfile for neutron-ovn-vpn-agent\n\nThis patch adds support for the standalone neutron-ovn-vpn-agent.\nThe new agent was added in the 2024.1 and is required for vpnaas\nwhen using ovn.\n\n[1] https://review.opendev.org/c/openstack/neutron-vpnaas/+/765353\n\nRelated-Bug: #2048392\nChange-Id: I552b9fbc8fa0b49d3e0e3212dd11962b02837b9f\nSigned-off-by: Joachim de Groot \u003cjoachim.de.groot@menzel-it.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f31185993f8f4b6f96266e2a32e3130005f3d66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f31185993f8f4b6f96266e2a32e3130005f3d66"}]},"branch":"refs/heads/master","description":"Rebase"},"3b84bb80e049623089dd8340b7ee8d0218babb64":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-09-19 15:04:19.000000000","uploader":{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},"ref":"refs/changes/02/924302/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/924302/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/924302/7"}}},"commit":{"parents":[{"commit":"8ef672bdd71f798a2fe99a4aabf22dbb0da0f26e","subject":"Merge \"Drop ironic-inspector\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ef672bdd71f798a2fe99a4aabf22dbb0da0f26e"}]}],"author":{"name":"Maximilian Stinsky","email":"maximilian.stinsky-damke@wiit.cloud","date":"2024-07-17 07:37:05.000000000","tz":120},"committer":{"name":"Joachim de Groot","email":"joachim.de.groot@menzel-it.net","date":"2025-09-19 15:04:19.000000000","tz":0},"subject":"Add Dockerfile for neutron-ovn-vpn-agent","message":"Add Dockerfile for neutron-ovn-vpn-agent\n\nThis patch adds support for the standalone neutron-ovn-vpn-agent.\nThe new agent was added in the 2024.1 and is required for vpnaas\nwhen using ovn.\n\n[1] https://review.opendev.org/c/openstack/neutron-vpnaas/+/765353\n\nRelated-Bug: #2048392\nChange-Id: I552b9fbc8fa0b49d3e0e3212dd11962b02837b9f\nSigned-off-by: Joachim de Groot \u003cjoachim.de.groot@menzel-it.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3b84bb80e049623089dd8340b7ee8d0218babb64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3b84bb80e049623089dd8340b7ee8d0218babb64"}]},"branch":"refs/heads/master","description":"Rebase"},"6797ee1cf2fe12d6e7be7ce036e6c5adcdbd88b8":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-10-30 17:41:48.000000000","uploader":{"_account_id":37725,"name":"Joachim de Groot","email":"jdegroot@web.de","username":"jaydg"},"ref":"refs/changes/02/924302/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/924302/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/924302/8"}}},"commit":{"parents":[{"commit":"8c2c3c90f9f0ac28dfb37c29ae61f78d453821ba","subject":"CI: Fix missing variable in publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c2c3c90f9f0ac28dfb37c29ae61f78d453821ba"}]}],"author":{"name":"Maximilian Stinsky","email":"maximilian.stinsky-damke@wiit.cloud","date":"2024-07-17 07:37:05.000000000","tz":120},"committer":{"name":"Joachim de Groot","email":"joachim.de.groot@menzel-it.net","date":"2025-10-30 17:41:19.000000000","tz":60},"subject":"Add Dockerfile for neutron-ovn-vpn-agent","message":"Add Dockerfile for neutron-ovn-vpn-agent\n\nThis patch adds support for the standalone neutron-ovn-vpn-agent.\nThe new agent was added in the 2024.1 and is required for vpnaas\nwhen using ovn.\n\n[1] https://review.opendev.org/c/openstack/neutron-vpnaas/+/765353\n\nRelated-Bug: #2048392\nChange-Id: I552b9fbc8fa0b49d3e0e3212dd11962b02837b9f\nSigned-off-by: Joachim de Groot \u003cjoachim.de.groot@menzel-it.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6797ee1cf2fe12d6e7be7ce036e6c5adcdbd88b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6797ee1cf2fe12d6e7be7ce036e6c5adcdbd88b8"}]},"branch":"refs/heads/master"},"75190ac460e5ed6490b2a7343d754263d33c181a":{"kind":"REWORK","_number":9,"created":"2026-02-02 16:44:18.000000000","uploader":{"_account_id":37355,"name":"Pablo Colson","display_name":"Pablo Colson","email":"colson.pablo@gmail.com","username":"pabloclsn"},"ref":"refs/changes/02/924302/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/924302/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/924302/9"}}},"commit":{"parents":[{"commit":"8c2c3c90f9f0ac28dfb37c29ae61f78d453821ba","subject":"CI: Fix missing variable in publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c2c3c90f9f0ac28dfb37c29ae61f78d453821ba"}]}],"author":{"name":"Maximilian Stinsky","email":"maximilian.stinsky-damke@wiit.cloud","date":"2024-07-17 07:37:05.000000000","tz":120},"committer":{"name":"Pablo COLSON","email":"colson.pablo@gmail.com","date":"2026-02-02 16:44:01.000000000","tz":60},"subject":"Add Dockerfile for neutron-ovn-vpn-agent","message":"Add Dockerfile for neutron-ovn-vpn-agent\n\nThis patch adds support for the standalone neutron-ovn-vpn-agent.\nThe new agent was added in the 2024.1 and is required for vpnaas\nwhen using ovn.\n\n[1] https://review.opendev.org/c/openstack/neutron-vpnaas/+/765353\n\nRelated-Bug: #2048392\nChange-Id: I552b9fbc8fa0b49d3e0e3212dd11962b02837b9f\nSigned-off-by: Joachim de Groot \u003cjoachim.de.groot@menzel-it.net\u003e\nSigned-off-by: Pablo COLSON \u003ccolson.pablo@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75190ac460e5ed6490b2a7343d754263d33c181a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/75190ac460e5ed6490b2a7343d754263d33c181a"}]},"branch":"refs/heads/master"},"db63f720490eb6158bf38f7d684466d5709f8e83":{"kind":"REWORK","_number":10,"created":"2026-03-21 17:19:01.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/02/924302/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/924302/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/924302/10"}}},"commit":{"parents":[{"commit":"8c2c3c90f9f0ac28dfb37c29ae61f78d453821ba","subject":"CI: Fix missing variable in publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c2c3c90f9f0ac28dfb37c29ae61f78d453821ba"}]}],"author":{"name":"Maximilian Stinsky","email":"maximilian.stinsky-damke@wiit.cloud","date":"2024-07-17 07:37:05.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-03-21 17:18:58.000000000","tz":60},"subject":"Add Dockerfile for neutron-ovn-vpn-agent","message":"Add Dockerfile for neutron-ovn-vpn-agent\n\nThis patch adds support for the standalone neutron-ovn-vpn-agent.\nThe new agent was added in the 2024.1 and is required for vpnaas\nwhen using ovn.\n\n[1] https://review.opendev.org/c/openstack/neutron-vpnaas/+/765353\n\nRelated-Bug: #2048392\nChange-Id: I552b9fbc8fa0b49d3e0e3212dd11962b02837b9f\nSigned-off-by: Joachim de Groot \u003cjoachim.de.groot@menzel-it.net\u003e\nSigned-off-by: Pablo COLSON \u003ccolson.pablo@gmail.com\u003e\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db63f720490eb6158bf38f7d684466d5709f8e83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db63f720490eb6158bf38f7d684466d5709f8e83"}]},"branch":"refs/heads/master"},"63024ee8a26f08f5f65fec3f58be6bbb690365b7":{"kind":"REWORK","_number":11,"created":"2026-03-25 11:36:29.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/02/924302/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/02/924302/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/02/924302/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/02/924302/11"}}},"commit":{"parents":[{"commit":"8c2c3c90f9f0ac28dfb37c29ae61f78d453821ba","subject":"CI: Fix missing variable in publish jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c2c3c90f9f0ac28dfb37c29ae61f78d453821ba"}]}],"author":{"name":"Maximilian Stinsky","email":"maximilian.stinsky-damke@wiit.cloud","date":"2024-07-17 07:37:05.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2026-03-25 11:36:26.000000000","tz":60},"subject":"Add Dockerfile for neutron-ovn-vpn-agent","message":"Add Dockerfile for neutron-ovn-vpn-agent\n\nThis patch adds support for the standalone neutron-ovn-vpn-agent.\nThe new agent was added in the 2024.1 and is required for vpnaas\nwhen using ovn.\n\n[1] https://review.opendev.org/c/openstack/neutron-vpnaas/+/765353\n\nRelated-Bug: #2048392\nChange-Id: I552b9fbc8fa0b49d3e0e3212dd11962b02837b9f\nSigned-off-by: Joachim de Groot \u003cjoachim.de.groot@menzel-it.net\u003e\nSigned-off-by: Pablo COLSON \u003ccolson.pablo@gmail.com\u003e\nSigned-off-by: Piotr Milewski \u003cvurmil@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63024ee8a26f08f5f65fec3f58be6bbb690365b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63024ee8a26f08f5f65fec3f58be6bbb690365b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
