)]}'
{"id":"openstack%2Fkolla~924365","triplet_id":"openstack%2Fkolla~master~I001babae32867310a2d272aa77b32ca09611d257","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2026-06-07 19:49:43.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I001babae32867310a2d272aa77b32ca09611d257","subject":"Add script for distributed cron","status":"NEW","created":"2024-07-17 20:34:01.000000000","updated":"2026-08-10 14:55:14.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":25,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"aa93157d5cd0f98fa4293000252672ca40ce14e3","_number":924365,"virtual_id_number":924365,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-08-10 14:55:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2024-07-17 21:11:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-17 21:35:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7cdb1849baa77f437b9dab1a31847df235c83d61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-07-17 20:34:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5125f6c1091c711bb5183825aedbe2d0368e15d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 21:11:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70710b4fb53c4f46805ef7c92d816667\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/5c125db4f5574966b9383e6614ac6a8c : FAILURE in 7m 08s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1273fa8e830e4d7f86b3b262ab6cdd2e : FAILURE in 4m 27s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7a5151892b094c7caa56fbf92d2f32ca : FAILURE in 13m 36s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2d2799192a714aef9992ef2779ced691 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4fd3bfd1d82c4ad69f4e3f63b79bcc11 : SUCCESS in 4m 44s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f6c21fe791a2443bb8bc61d678520749 : FAILURE in 11m 42s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a5e4a3f81d8f479bbf24f460f2b9a7ec : FAILURE in 5m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"483574d6f1e976c0847ac52f0b8b1ca53f5060c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-17 21:35:55.000000000","message":"Patch Set 1: Verified-1\n\n(25 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/efa6191bc4ab462d97c43571ba7dfd37\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a253b5557f9d4c6d97ef921e4ac6c865 : FAILURE in 5m 16s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3a10eec46d77486787d2002f724ab8e5 : FAILURE in 9m 17s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e5f0b26f8ecd49f699e38d6df01f8dea : FAILURE in 4m 19s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6e4af38fc5f149eaa45d5206962a50ae : FAILURE in 3m 38s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3636c81282eb4929ac29a099709b8bbb : FAILURE in 6m 58s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4eea62efd0b143b182ac674e0d605bfa : FAILURE in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fe9c3bd6f2146f1ae53484f8359dc8e : FAILURE in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b0eb78b9cc245fc93c403abce33f01f : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f1a704f980ca4336a2c465b764665db6 : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8aa4c5bdf7494c2581c51ed8fcc85aa0 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f05ad0ed4574052aa888e31dc2b55d2 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54ef2d96207c4df0a38481970480b7ba : SUCCESS in 5m 41s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/17095634ee0e443e9e3066ad8c2fa323 : FAILURE in 7m 09s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5adccdef2b8f4e14b627dff8b9f0dc43 : FAILURE in 7m 37s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7a6eabc0e208457192e7fe465b3749ec : FAILURE in 10m 07s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1f9dacc59c9c4d9fb0cb5057757a1817 : FAILURE in 12m 33s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/8401cb9dc55a44bba81be5e8537f0cd4 : FAILURE in 9m 36s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/bff0c2e729054bbe99ff7caf5a277472 : FAILURE in 10m 35s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/5ad525cb8c994f4fa09a3f0479aae268 : FAILURE in 9m 07s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/51e5dd1506a449a29249287cbebeb9f4 : FAILURE in 13m 42s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/9b78536870b549b199e64d7fe47cf0b4 : FAILURE in 9m 17s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/dbea97aa1eaf456e9d99caf8cc1bb4b8 : FAILURE in 11m 12s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/4be2ad78e7a9415889e1cf6b867b0f9e : FAILURE in 9m 09s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/b0d43587ebba4aa3bbf1860a8af7cb4b : FAILURE in 11m 10s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a39c13a3508647768dc78b7b00bc0d73 : FAILURE in 7m 46s (non-voting)\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/8da20a07bc3f4459abcebc0dc27bd522 : FAILURE in 10m 30s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6b0df73caa064de291aaca8ed9a0bebb : FAILURE in 3m 18s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3ca782b91d914f6f89ab174694d4b623 : FAILURE in 3m 22s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5b0494035e25463191b9d0331f4671d5 : FAILURE in 7m 07s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d69cf78dd8144b9c8afec5a5dba89d5d : FAILURE in 7m 32s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/5215af9c3ae34759ae0ba2ec06036745 : FAILURE in 8m 26s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5802baad885340e9b8ea24f950c5dc02 : FAILURE in 10m 07s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/6faf533b7bb8481ca8dfaceaf814555c : FAILURE in 11m 01s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/00a645727f1a4a5883e339ed6be164cc : FAILURE in 9m 10s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/6d87652460644036a41f7d6326d90024 : FAILURE in 8m 13s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/562ee390a12943fca6bb45894ff69ed7 : FAILURE in 9m 04s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/3912d7922828458c9caa20bdbfc2a1f5 : FAILURE in 6m 26s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/fa970f45c0d4445aab3be15ffefefcb1 : FAILURE in 8m 58s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/89a54d5e56134f57853152e85f72005c : FAILURE in 8m 43s (non-voting)\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/7c5ca59cf68e49468e609d33a008910d : FAILURE in 6m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"84a6edb23126bc0610044cd544644bcfd9c49256","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-06-07 19:49:43.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"953dddf3fad227a1a0897e1d91293a1f960435b0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2026-08-10 12:20:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":2},{"id":"aba4ab6dd2331779d152cf6437c6e1e92a5763d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 13:45:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdf6a67017c34b298bf1bd269923da4b\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/10bf119661954e83ad099c6dd800ca80 : SUCCESS in 37m 39s (non-voting)\n- kolla-build-debian-trixie-arm64 https://zuul.opendev.org/t/openstack/build/f44dc46ea1f04d58902b9bf7597e473d : SUCCESS in 40m 15s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/bb70111b443446f6bc0a40929fe60838 : SUCCESS in 42m 32s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/55e7c522a4314f22a184831210c4b927 : SUCCESS in 40m 44s (non-voting)\n- kolla-ansible-debian-trixie-aarch64 https://zuul.opendev.org/t/openstack/build/297b24ecfb0d4de4bedbd0148c845f6a : FAILURE in 1h 23m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/71ec2f778cc14b7ca3d7b0a81b277999 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/adc3d27452e0488b8348cd3ce11f0ba7 : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/b0667919baad41f086f87ddebc49a299 : SUCCESS in 9m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"aa93157d5cd0f98fa4293000252672ca40ce14e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-10 14:55:14.000000000","message":"Patch Set 2: Verified-1\n\n(25 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/296a198531f04b818b1acf592ce743d8\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/1fe21ccefcb349e5bca65f9a81205a22 : SUCCESS in 56m 34s (non-voting)\n- kolla-build-debian-trixie https://zuul.opendev.org/t/openstack/build/2bf03f723bdd48758e4f238f9f9d40d6 : SUCCESS in 14m 46s\n- kolla-build-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/89c20d8e2ed24738991d4b28b01d50e7 : SUCCESS in 14m 34s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/1dc2b3ca5f114a338ba2ea9f1e196459 : SUCCESS in 17m 32s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/702910359dac43759dbaacfab4157ef7 : SUCCESS in 49m 55s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/573d9f5f814747f7b77e3d365fa6e78a : SUCCESS in 25m 41s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2e75672c724c49a490343b5d5c488ef1 : SUCCESS in 16m 03s\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/7102137e940f4505ab8518e8daf03abd : SUCCESS in 59m 50s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/f150d68ef4a34394be5d5040ff96797d : SUCCESS in 1h 27m 41s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/c84f7769af764ce5b0afb9a49e00ec25 : SUCCESS in 1h 00m 16s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/b6e8147675154f25a95f7352f81bc3e4 : SUCCESS in 1h 29m 02s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dad48b0b80e8409dbab231f919c271dc : SUCCESS in 1h 02m 22s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/ea904dcc45c540c1b9027a04d0d4a8fa : SUCCESS in 1h 36m 01s\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/4299cb69250046b8a9bfc7b5a988682a : FAILURE in 52m 08s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/e825cd5e84ce4136a85f24fa22603713 : FAILURE in 1h 30m 27s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/c9e129127cf9424493e0d2d65eb5e66c : SUCCESS in 2h 09m 07s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/5271453a4d5844c4bbc4b84eff686c6d : SUCCESS in 1h 30m 07s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/2c6a0a9a29c2486497dd7334bf132035 : SUCCESS in 2h 17m 30s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/73157975064c4352bcc7cfe8d4280175 : SUCCESS in 1h 21m 08s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/d55a1697077a4ef597ad8a7917849943 : SUCCESS in 1h 24m 03s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/518827f26680436d8f889656af9a4843 : SUCCESS in 1h 19m 48s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/96ce8167fb39400d9fa49bc0c81d6096 : SUCCESS in 21m 51s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/047f7971b4674fbab317be35d91829b6 : SUCCESS in 23m 32s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/09ec8abb7ee64586b6f9be0130572e12 : SUCCESS in 17m 07s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/7bd25201324e432a9c2dd43151d800c7 : FAILURE in 25m 00s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/1556cc341594487d8c4020de985177a8 : SUCCESS in 45m 57s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/5f3fad5dca3f4aa8be0e0372ad028fdb : SUCCESS in 1h 10m 49s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/049b4d46296e48a19436d09f1cf5515c : SUCCESS in 59m 55s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/364369c9d63c4076bf7a6f8967dc6c98 : SUCCESS in 1h 01m 32s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/c630d7a536234806abbf8c3bf2ec063a : TIMED_OUT in 2h 32m 32s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/5ac43fb8b472498e8d51498cfec9c618 : SUCCESS in 1h 35m 58s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/0910cdb36e2c488eb081cb54d1365720 : FAILURE in 42m 31s (non-voting)\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/32f0d90e9b2f4fcc91f89ed18fecfa8c : SUCCESS in 18m 10s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/ecbf1436a3784ebc9335895e3f52484a : SUCCESS in 44m 13s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b7fdac359e694178b7af3c6924741fee : SUCCESS in 43m 40s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/b65316e7d5e744b0902511f4a03ba9ac : SUCCESS in 27m 16s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/90e5863e7470450789ac581d0b0fb04b : SUCCESS in 34m 13s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/8863d72c5f984f13852863adf9843ffc : SUCCESS in 56m 47s\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/cca72eba0a56423e93e59cb8fbea22f7 : FAILURE in 51m 43s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/0c17be654fa2482cb10105df436105a8 : SUCCESS in 44m 26s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/395f8c31a23b4851ae3ae09355852a34 : SUCCESS in 1h 26m 17s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/36133582541e475598c7f578e575ae5d : SUCCESS in 2m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1785d789d3204027b548312ba2f45d6d : FAILURE in 1m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e356ae58f82742e1a217507b1ee1e353 : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d53c9a8b2a9e4feba02090e290e177f9 : SUCCESS in 3m 25s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/be5f6286adf44859b1779cc917f5ac53 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/099115cefc90415caadf4add87839872 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27c7823cb62940ee82b1a1ac1d3a2d56 : SUCCESS in 3m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d10b23677ed54f8492871fe6e5ae67b7 : SUCCESS in 9m 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e955de99c1a21f59dfc77191d6835bb3799fc725","revisions":{"9168d271f1ace34e518005969275d5ea3a818488":{"kind":"REWORK","_number":1,"created":"2024-07-17 20:34:01.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/65/924365/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/65/924365/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/924365/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/924365/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/924365/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/65/924365/1"}}},"commit":{"parents":[{"commit":"2f28e30912bd06350a4325d2606235197347458d","subject":"Merge \"fix: apple silicon architecture selection\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2f28e30912bd06350a4325d2606235197347458d"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-07-17 13:40:20.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-07-17 20:32:11.000000000","tz":120},"subject":"Add script for distributed cron","message":"Add script for distributed cron\n\nChange-Id: I001babae32867310a2d272aa77b32ca09611d257\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9168d271f1ace34e518005969275d5ea3a818488"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9168d271f1ace34e518005969275d5ea3a818488"}]},"branch":"refs/heads/master"},"e955de99c1a21f59dfc77191d6835bb3799fc725":{"kind":"REWORK","_number":2,"created":"2026-08-10 12:20:49.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/65/924365/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/65/924365/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/924365/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/924365/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/65/924365/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/65/924365/2"}}},"commit":{"parents":[{"commit":"5aabec47f057d9cb32aab520d666e51bd9ee37c6","subject":"Add cleanup_repos to revert build_only overrides","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5aabec47f057d9cb32aab520d666e51bd9ee37c6"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-07-17 13:40:20.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2026-08-10 12:20:41.000000000","tz":120},"subject":"Add script for distributed cron","message":"Add script for distributed cron\n\nChange-Id: I001babae32867310a2d272aa77b32ca09611d257\nSigned-off-by: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e955de99c1a21f59dfc77191d6835bb3799fc725"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e955de99c1a21f59dfc77191d6835bb3799fc725"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
