)]}'
{"id":"openstack%2Fkolla~924759","triplet_id":"openstack%2Fkolla~master~I5a9162a30b23cfabb64dabe5c12c226fa8627abd","project":"openstack/kolla","branch":"master","topic":"fix-openstack-base-symlink","attention_set":{},"removed_from_attention_set":{"35264":{"account":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"last_update":"2024-08-23 06:50:59.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-09-03 08:22:18.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-09-03 08:22:18.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-08-30 10:26:19.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-08-30 07:08:30.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-09-03 08:22:18.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-09-03 06:37:33.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},"36765":{"account":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"last_update":"2024-09-03 08:22:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5a9162a30b23cfabb64dabe5c12c226fa8627abd","subject":"Fix symlinking of openstack-base source","status":"MERGED","created":"2024-07-23 16:12:11.000000000","updated":"2024-09-03 08:23:24.000000000","submitted":"2024-09-03 08:22:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":56,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"924759-fix-openstack-base-symlink","meta_rev_id":"5dedb1011ea094782f0bf62355d9ec97f9a96f63","_number":924759,"virtual_id_number":924759,"owner":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-03 08:22:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"recommended":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-08-30 10:26:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-08-30 07:08:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2024-09-03 06:37:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-08-30 07:05:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":1,"date":"2024-08-30 07:03:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2024-09-03 06:37:33.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-08-30 07:08:30.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":1,"date":"2024-08-30 07:03:04.000000000","_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-08-29 23:38:29.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-23 16:13:10.000000000","updated_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2024-07-23 18:26:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-23 19:18:05.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-07-24 09:14:25.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2024-07-25 09:14:40.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2024-08-05 17:14:57.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-08-13 12:32:20.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-08-28 08:51:10.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2024-08-28 15:29:15.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-08-30 07:05:22.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"94ac1ab2efe1935fe21db477049e9965b50c348a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-07-23 16:12:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eef080911064ee2acf48eccf9eaf8531466f9dc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 18:26:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/231b206b5904460c953ab57105dbaa0d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8498ca1407dd4a8280d55c0205553cef : SUCCESS in 23m 40s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/7c12410500ad4a1c9723ad571462bc7b : FAILURE in 43m 27s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1a09b04c9f9543679a6be42ce4d489ba : SUCCESS in 19m 43s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3c204105a0214086a6437d221a45b9ee : SUCCESS in 29m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e196be591cb7450e849fc84b94a1f92e : SUCCESS in 1h 06m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8d1da2707d6f467d91ca1ef8d98cd641 : SUCCESS in 1h 21m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/701ca43eb39348b78aa5a959a5515133 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6e337494499485c932376c709fb6f62 : SUCCESS in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d3427636153143c684c1851d0dc61100 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/845935b362664e19b29011b3810912bd : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54b3df1080fd49b4bf62aea4ae44c77b : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca1b0c775a654e2d9fa790e155278d47 : SUCCESS in 4m 35s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a7e0036b54144dbb81b00937fafad083 : SUCCESS in 23m 54s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/708d45082b56474d9bc4ca7043b07e1d : SUCCESS in 39m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1a875ab4331a43f6b76ba49cebba75d3 : SUCCESS in 1h 09m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/42cd9bf7d4d248f5911fa63a18582327 : SUCCESS in 1h 25m 35s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/dc416d449c844ba985cb6ccbf2a4900d : FAILURE in 1h 29m 00s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/85fc5aa104a944299c361b0c8cddffc8 : FAILURE in 1h 26m 29s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/aeed0c0bbea54e7a8897ad8056a17eb2 : SUCCESS in 1h 03m 49s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/8435a7a205ac48b4b05dd26cf5ff2bee : FAILURE in 1h 08m 02s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/4cb81ffceae046638b1221f175ce81c6 : SUCCESS in 1h 06m 44s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/d2d4cccc1acc4bfeb2ad4bf34f90573b : FAILURE in 49m 46s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/5262cd09535242afa803ebc128384c75 : SUCCESS in 1h 32m 25s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/329752877234422f9b6b36a323bbe368 : SUCCESS in 26m 52s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/80d579a995b54e25b13b83da08f9cb70 : RETRY_LIMIT in 3m 16s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7a885e15d31a4f6f854bb3521a1db9ff : SUCCESS in 51m 54s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9546f4fd26a84823a9c26f610dd2a456 : SUCCESS in 1h 17m 10s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/5c87bfe93e7f43dbac64082d57dfa22c : SUCCESS in 1h 27m 36s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d7f6b893ffe0475492b458b53d0860ab : SUCCESS in 2h 11m 52s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/0f6e7a6b1f0645979a3f2d726608a44d : SUCCESS in 59m 46s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/0580e0b59cd541cda390302c070cbe56 : FAILURE in 48m 33s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/fcd6a40fe3ad4631bdfc2d551e7d7631 : SUCCESS in 52m 05s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/f9e46c87a5bf4867add8e5a2f7a8c92a : FAILURE in 40m 16s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/1f87a91622d94c159a62c15a6d0f1bab : SUCCESS in 1h 06m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2365e98d4c149279d0bbf7cc183a5e044808dbc6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 19:01:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f05c7d0265344ce8a244cb4daab55e81\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ccb43469afe6430cacb005ed2411ea37 : SUCCESS in 25m 23s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a7c54199a7864816a132bdc5e9df6d08 : SUCCESS in 22m 37s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c4fbb15c551247e2994e55be9ae91cfe : SUCCESS in 1h 18m 47s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d5c1f516c0924d1284d637dc5b6798bf : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9c9b12d30018441a8f776fab7ed9bb80 : SUCCESS in 5m 07s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c1899aaf7b0949618a9ce332168bc535 : SUCCESS in 30m 38s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0190198dfb8f4f65abfd1a3316633230 : SUCCESS in 2h 46m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5160f1bbba4c1a805063749ce05db3e19359bcdf","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-23 19:18:05.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1c9ea377d48ac0a7268adb2e0e1c3e483254d25f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-23 19:19:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d807ef8b3c672b0e682bc9cd798173136de7033c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-07-24 09:14:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e23a7d04ede3a04f6280cd8df7ccf39d021dede3","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-07-24 14:53:45.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2c655f4bafeb988a63e96802dd7d7e52a4420099","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-07-25 08:45:06.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7ba6e30e290fd599698c4bede68d80d21d252d1c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-07-25 09:14:40.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"282de741aca91be38e10a737208b9d3a78b3ec49","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-07-25 09:42:17.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ec961aaa500635a28a6186a56b241f767eca3a00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 10:56:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e200e4c29d54fa6960c85d12a9ef714\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5742f1f5f7744aa79dd848ed4a662e05 : SUCCESS in 24m 38s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ac78264efe844da8a9ab5f59de297dba : SUCCESS in 1h 13m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d0c41c0170e74b01998c34f6d9ab826e : FAILURE in 16m 30s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/dd5b2448028a4727ad79f3ad4c980418 : FAILURE in 23m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b5f8b9358ce04fb19cad5f313fdde12a : FAILURE in 15m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/430a22f4dabe459e9f7327c41ac509b5 : FAILURE in 15m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60cebd7581484bee847831864e01025c : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b2b8988b5aa428181be2095d3fb1514 : SUCCESS in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1c0858d5f479475182dcec4cf2b9d0c5 : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e9701602da634f34af66607879dbbb5b : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e23d688204347a7a4850a0a713579d4 : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2ca89a3bed14c95945b08f0aa81c665 : SUCCESS in 3m 57s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/584de8f202e54af79dc92fd06c6e0004 : FAILURE in 25m 24s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7cd9099dfe9244f5835ba1b375846765 : FAILURE in 29m 25s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8a8972e23f994626972e16b97a62031d : FAILURE in 20m 00s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/41fc672cf9ad4ab99403072829aad4f9 : FAILURE in 19m 00s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/df7b4270954347fa8fd10182bf14c0be : FAILURE in 22m 06s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ae4903f69e11437eadb162f699bef031 : FAILURE in 21m 30s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/a3f157879ca14e3ebb95ae9902343982 : FAILURE in 21m 53s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/c8ca601f51834c869fb61a8c9d6bdabc : FAILURE in 21m 53s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/db8aeaf5ac2f4b1e9c8e3d98affe138a : FAILURE in 23m 16s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/80994d7348f3451ebf5d5198c198e093 : FAILURE in 18m 56s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/11232b99a2a344d18925c60070a0c453 : FAILURE in 21m 25s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/774073c3808c4a3a81aeec3c058167bc : SUCCESS in 21m 53s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4e74b1e46d37403d8639cb68ed2c5c1b : SUCCESS in 50m 58s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e78980aa609f4bb189dfc16cb64aac9a : SUCCESS in 50m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9f99bbef1b0c4338a10044740cdd7495 : SUCCESS in 1h 15m 25s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/8c1254c6e0ad4973b9c37df230619a61 : SUCCESS in 1h 32m 58s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/167eacf91bfa43408bff8e764d6c4f2b : SUCCESS in 2h 10m 44s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/8c00a50703cf4c4ca19e54b1ced6cdf6 : SUCCESS in 1h 01m 13s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/0c6fb7cb66e9400aa6ecb11a708a4894 : FAILURE in 50m 58s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/6c006c80cb134d26b3676095954a5e03 : SUCCESS in 48m 34s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/d69f8d7b68584f078936e11d7a1b1bce : FAILURE in 50m 46s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/2396e08ca0704c4393773f1cc155f84c : SUCCESS in 1h 00m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bf37ce1b15ae4468c3e4d681443ec72320d299a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-07-25 11:16:58.000000000","message":"Uploaded patch set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"2970d7d594d53c58f7a305a918f030316f2c2c98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 13:06:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c2714377502444e9cde2fbd2a0a1960\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a8ca56b960bb4a33a8b50ad91ef5c28d : SUCCESS in 27m 43s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f3f3dfcc4a434fbb891ebaf57b9c9b65 : SUCCESS in 1h 11m 38s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b99fae6453ee4912805cd606741af7b2 : FAILURE in 22m 44s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d5c9ad0e8c414f1ab5a4aebd5658c88c : FAILURE in 29m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/cda6aef8cbbe40169a2f9b804ed351c8 : FAILURE in 15m 15s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/10cd8bf388ff4e168ed41ad93bd29159 : FAILURE in 19m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66f63978d7b44b71950e038c018ff7d3 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4900301bf4f74f00a4c8141a9bde790f : SUCCESS in 2m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/328c71d6a3104658a22ee32e2a6e7a21 : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/61d212291633450eb2260321af7d416a : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ab1bd34a7b44554ac9ea6655a2cd7a1 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/510905c2c3d9424a97fad64eb0f67361 : SUCCESS in 4m 09s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f12c6cb275ae4d909dc287e602d6402d : FAILURE in 29m 21s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/dff9d061d4d44712985dd10caf5a2622 : FAILURE in 37m 14s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/00937faf9ec146f9b86363be44de6f01 : FAILURE in 22m 28s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8d19682c86414b5da4b31bccbecc4a3d : FAILURE in 22m 03s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/558492c19f594dfe8d9561f81035c705 : FAILURE in 22m 14s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/6f4e0a3770904d33ae72e8d56b44b415 : FAILURE in 23m 22s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/5d99799dcd394ddbbb581cca4c6158f2 : FAILURE in 22m 08s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/996560b883904f1aa95940e2377f1970 : FAILURE in 22m 02s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/4f87c5954cd648dca08ef7099ac497ac : FAILURE in 24m 58s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/76b5bef9e9374301b42f2e20575bde29 : FAILURE in 20m 34s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/c4fbdd7bcb2c408c9f1e313e578c5947 : FAILURE in 20m 00s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c392dae11c9441e5a69c0c3058145d89 : SUCCESS in 21m 07s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/325fba8e5f414a0586afd51384f15e77 : SUCCESS in 56m 28s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/09511c71f8f04b1f9e0f84272bf3d6c8 : SUCCESS in 57m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5624eb6d66ef41b1bc649df9533f0dd1 : SUCCESS in 1h 20m 36s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/ae4530ff2f634002ba939e2fa0b29db1 : SUCCESS in 1h 37m 01s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/20d336296b994734a7948effb3c5b781 : FAILURE in 1h 12m 03s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/ec57bd0bb9cf434e88113086f83d4e49 : SUCCESS in 1h 03m 42s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/c4e9425cfafd4f2aa05e7f839b7d43f8 : FAILURE in 50m 06s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/5978d2820dc242afa6f7ca24ecd97bdb : SUCCESS in 45m 58s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/753bf290c8d14d7b98987bda93162a3a : FAILURE in 42m 35s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/adb7f7c3559c4683aaeb8059f0cf397c : SUCCESS in 1h 13m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b808910c0985ebe73e33d2cae70df8eab3a40a31","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 14:24:39.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46863ab0d5f4463a94a3f95b3b537500\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/dd34291d33524b318c8ab9284e9198fa : TIMED_OUT in 3h 05m 27s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/29b448f5be574c48a3271dee6cb61dd9 : FAILURE in 33m 51s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d4b935721b9f41fea72ef3ff8e7bb3c0 : FAILURE in 1h 07m 40s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e21daa12cd184b3c91d8e226df22aae3 : SUCCESS in 2m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d9fb91ba3eb74c25ac7cef69c9c3b96a : SUCCESS in 5m 30s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1b0653c29ffb4585b8e9bb19b301d3c1 : FAILURE in 41m 54s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/64c7be06fe2742c883d27e06efd2bba4 : FAILURE in 2h 59m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"94892f54cff6aefe9535056d48eaa44b6aaba0c1","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-07-25 15:41:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5c9754de68f5f16d668a486c6568b82bcc844241","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-07-25 17:27:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5b01fc170b6eecbb726c15e6889c637012d4e56c","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-07-26 10:19:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5007e7ff41b0cabeedcc5ee040e23ddc2750b773","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-07-26 10:25:13.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"8f6e5c30256215c63016358b138d45204cf301cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 12:35:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16896d92550e4a8983e0867be40be725\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b36e74c4155545618171f117b73dfa98 : SUCCESS in 27m 57s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3bef16a29ab5496092cdbc71e1d17ba7 : SUCCESS in 59m 48s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/89b101c7fa1f47d49e5da40eb78103e6 : SUCCESS in 19m 03s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/90e5092c57f749c29207e6b73e1330fc : SUCCESS in 26m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4fb43339a1104ac8b32b35121bc06530 : SUCCESS in 55m 21s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/de2d254f4edc464ea67a1870275f16f1 : SUCCESS in 1h 37m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5d53d55ef624e49b8e7b80d3fd7fe22 : SUCCESS in 2m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3785d04483f4812981b13cfb44d59bb : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0fa16366fb71438b879721462160c2c8 : SUCCESS in 2m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9f5baf97aec245ac9b062cae1829b188 : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8d37d50452834d8bb5d2036fcf1f171e : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec113cc962154a4ab801b7734901cf23 : SUCCESS in 4m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8357b29a21c44a7986193b1aeb6a4a10 : SUCCESS in 27m 32s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/8ca1714225644c6d915501e156a23244 : SUCCESS in 29m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3aab5c9dccc347de85acad260baa4c23 : SUCCESS in 1h 06m 49s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6db346b98b69428e811a9db3681a5947 : SUCCESS in 1h 32m 55s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/03fd80dfda6e44aa89174250aec5e170 : SUCCESS in 2h 03m 08s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/667366457ca742eab106828f385262b0 : FAILURE in 1h 42m 21s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/e5e3887e488a4731b182da4740026485 : SUCCESS in 1h 11m 35s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/51d931a70c854b1fa26773e99eb0216d : SUCCESS in 1h 21m 47s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/c892f1e6069e4bf19638d05248d8b5f3 : SUCCESS in 57m 08s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/3ddec9d26ca84accaa3de2ef326a6872 : FAILURE in 53m 11s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/915834b4d8b048f49c4fdf426772d51c : SUCCESS in 1h 26m 45s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5cb220af6f0248ab880875871cd4ab9b : SUCCESS in 26m 08s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/52b9db8812164cf6818431d8a27bd0d8 : SUCCESS in 51m 07s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a6beefd8365f44f9bafaea3bd64b8888 : SUCCESS in 58m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9528f7322ed342a58e37e941a5d2f850 : SUCCESS in 1h 25m 59s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/3a78db8be80c4f759dc8f438e84d747b : SUCCESS in 1h 29m 34s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b2206b673492423a9354c07478931727 : FAILURE in 1h 09m 20s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/ddc00f3deb73467299df1e004d613261 : SUCCESS in 58m 20s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/24411d74e909445d8f9a2a94875772fd : TIMED_OUT in 2h 03m 35s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/2b0d6b2b581241318197f5cb58a43582 : SUCCESS in 49m 11s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/e7dcb5a49dcb45628a55a8d29baf1c99 : FAILURE in 44m 36s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/0bca0a2d0f2847d2981373d5ddfa2b09 : SUCCESS in 1h 07m 51s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"42a7e1f0921f2b1f06ee95f9792252c18717f7fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 13:12:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24d501c50c074807843ea6806616eba4\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/5bb7a1a42ee447ab8dbeff1fca6ff894 : SUCCESS in 2h 44m 33s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/79486a4065b1456287f51c0e73f80c12 : SUCCESS in 20m 55s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b9df79d7cae744fd8ededa293ddb83ae : SUCCESS in 1h 42m 33s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d4f75c4bbc2143449e6fff40a9bfb4a0 : SUCCESS in 2m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1cb4a19ea72d418bb4f3a5574d0a8477 : SUCCESS in 4m 52s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/11084fef4761404d9878421669d9e6e3 : SUCCESS in 23m 12s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1577c197f1494503a0b3cd5c0bd86ed4 : FAILURE in 2h 24m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"68a0260d588427012e72192b705ef55279b7d02a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-08-05 14:27:20.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6aac931d59ad844528d873eb5d38669c65f30385","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"date":"2024-08-05 15:54:36.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"0afb12e35daa28a1d898ea4bfafb353c065a09b2","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-05 17:14:57.000000000","message":"Patch Set 5: Backport-Candidate+1 Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"224d2ccbdde93a429f2bb3da305c598159bc1947","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-08-05 17:46:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"dee4f8bab422ec73b7f809283f453febeab61dd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 18:11:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f38d57ef373f48298c2b2fc801e2dbe5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a398efb9b73d467e9f0c2a60812ce819 : SUCCESS in 28m 24s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/096331d53d954e4f94d592ea5ed98663 : SUCCESS in 1h 01m 47s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b0c1f57331ac4ae99d592514c7910a57 : SUCCESS in 16m 07s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/532914a48aaf4027b54859e328670f6a : SUCCESS in 27m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b0e6e5dbf5e44a629935cb163a5fb2a3 : SUCCESS in 52m 36s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6bdd57e14e484bfeb98d7bacb1f0a65b : SUCCESS in 1h 32m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8af6717111ab4e6ba9e115aff564a65e : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6179a4cd3f3c4020a92d1741c930a317 : SUCCESS in 2m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b77db0364e604a4eb657076239aec4f6 : SUCCESS in 3m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8cc054ceea884173b1724e09bda3c769 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4870ee5e6e9944b9a063d3dd3e7c2315 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b50f010023ed4ceab5a5babc69f8c884 : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf2e8caa618247238b9ce74bece9c1f4 : SUCCESS in 3m 41s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d4fcaade41da48089fe68f42ffd13513 : SUCCESS in 27m 06s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/076150d852c94b6abce246c05862c45e : SUCCESS in 34m 19s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/bb1796fdf6bc4fa59d318bb74aa7a328 : SUCCESS in 1h 11m 03s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/da150b09b1e644aeb570b3fad193c460 : SUCCESS in 1h 44m 44s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/ce36ab59eac0482ba27d466cc84232e6 : SUCCESS in 1h 50m 36s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f9017a46f765409f8d3cd05caa0e72d4 : SUCCESS in 2h 15m 58s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/c80ab34bffbe498fbdf778605cb9d9a4 : SUCCESS in 1h 20m 55s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/be7fc1d6e1884b62a361457665f33553 : FAILURE in 1h 00m 58s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/d976238957f642c3b90723308b27314b : SUCCESS in 54m 08s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/2d2b3dc4664b4e38925522919c241060 : FAILURE in 58m 43s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/92406b67693d48128601c73d5b19ada7 : SUCCESS in 1h 20m 45s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c25f7a6599d14c409e423badbc57a421 : SUCCESS in 20m 42s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/46f23287ff074a7997df77a44edc1997 : SUCCESS in 48m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b01df9a590144e6db33a9b7dd5807ac0 : SUCCESS in 49m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a36918cecebf461aa2fd9d7c250c944e : SUCCESS in 1h 20m 50s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/af3a0365a03944248a1b43273e16d26e : SUCCESS in 1h 19m 23s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/061dc56a08db4970ab7cbf7d272ef305 : SUCCESS in 2h 02m 08s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/61cadb4c29c4491aa9e49a3169eb8f75 : SUCCESS in 1h 02m 44s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/66fe50c2478140cea3cb2b9dc429d121 : SUCCESS in 1h 04m 59s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/b24fe4b1eb334f30a4c3739c05f6c2f6 : SUCCESS in 49m 12s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/5bdc9996da984119947def2796ddb1a3 : FAILURE in 46m 26s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/f4bcc2ab83ef438d8a1ab06911a0f671 : SUCCESS in 59m 07s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6b45d3f069392791233fd90cacfa3280c206bfa7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-05 19:54:32.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24885f30ff1c4be2900e2521fb2569d3\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ad279fb7aecb4d9eb0a03ce0e2d208ea : TIMED_OUT in 3h 02m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/96bb98e7f4e44233aa956af12d1981f4 : TIMED_OUT in 3h 02m 48s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f63aa67e85ad4152a2d6aa0c44586dd8 : FAILURE in 2h 20m 32s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/72f3a0787b3e46c9a6970f0b3735b670 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/09a5ac2ef2b34bb6b2e43d83b1e64d22 : SUCCESS in 5m 09s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/cf9ceb44972c4a4eb74d22ed1b2d7118 : TIMED_OUT in 3h 04m 00s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/56f00f7469574fd89c8f5a02c08a6b63 : TIMED_OUT in 3h 02m 41s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f29bbe52b12b852ab5664846057dabdde4d9e41e","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-06 15:14:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f0f675554d2a12bbaa632345179348dc91c31147","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-08-06 15:15:30.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b87475f84c5585f10a371f6209ae8427d73cdd26","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-08-21 02:02:05.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"f8c8ac09297b45f8a1652c95eb1309e6be1be9e1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-22 11:09:04.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"615c1f42f4f8fde5c4a6d8e935858cec95d55e39","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-08-23 05:11:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"17bf2395b28c72331d3ba73b484e6ca157afcbf7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-23 06:09:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6f9443389a948d57884aaa12e9a4606029b36601","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-23 06:12:26.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5d357a8351e14836354b4766adf7ec3a43b0e973","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-23 06:17:28.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c18d1949c1cae13cbd985167a6f56b962d147a8b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-23 06:26:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d493440c238ff3c23476d5fbcf79e16ff3d685d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-08-23 06:45:23.000000000","message":"Uploaded patch set 6.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"899a96897017e48cd948960b6feee2ea3c0cfd15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-08-23 06:50:59.000000000","message":"Uploaded patch set 7.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":7},{"id":"e3b1fb45301a285b87c288638314318b137db055","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-08-23 06:51:43.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"3345b628ffe15e4218cb9e8c4b6edaca10ba1352","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-08-23 06:59:11.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"248439538bc04194dfaca01b391818df678b4f4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 09:24:03.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2006bb784ea4bfd9ef5c00adf8468c2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d45091f3b013421fad5347514753c251 : SUCCESS in 23m 09s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3915db0bc2a0413f9072dbeeb419857d : SUCCESS in 59m 17s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/96313d1b6526490eb2bed35724ce0714 : SUCCESS in 17m 33s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e82878d5f3e74b198b456eab910e9a14 : SUCCESS in 24m 05s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/66c1979a0c6b47f8adbb7630f33db65e : SUCCESS in 58m 11s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1c87c273f9dc4471b3b31e606bbea679 : SUCCESS in 1h 23m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb20654dbadb4fd5b9315992d7ea4010 : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d559f606356a4661a24720546ee4221c : SUCCESS in 2m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b8d02e06eac049c3bb0314d3ec63077b : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1ac0fa07fd5d41728da32d845b903cad : SUCCESS in 2m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01a98f51da024ab6bb392030c0250541 : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af1257189a2d41189b4df10641e4e22a : SUCCESS in 4m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0bfacffb1ad48d2bb15dad60b532644 : SUCCESS in 4m 16s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/36a0d138612743bab9a0b9a635cbf1b1 : SUCCESS in 25m 43s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/91521b7c9a89422f8eb2482480533021 : SUCCESS in 34m 41s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1cfcade72d1f468c98d8d18c17d18c2d : SUCCESS in 1h 08m 29s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b8af647bce6e43138be7cbbfc83198d0 : SUCCESS in 1h 32m 20s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/dc06fc0fbe2d4ef6ad22e62e0b88cdc7 : FAILURE in 1h 16m 51s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1f75091d4d1b40edb90f94f117a241bf : SUCCESS in 2h 27m 07s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/768f6d24cd754005b24eb849a6fb673e : FAILURE in 31m 23s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/70d89d23fbcd4e12b768dc4ef4260f56 : SUCCESS in 1h 10m 40s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/49a570a280504fa0a8627f456368a3f3 : SUCCESS in 52m 40s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/95968554f7ef47c48957f1c65969d42a : SUCCESS in 1h 01m 27s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/113ba09f7eea41c0a4a6014a81dc7256 : SUCCESS in 1h 22m 34s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8802ed1b24ad4204a361a87366841e05 : SUCCESS in 27m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6d618cd0087246c2b3b35f757402c741 : SUCCESS in 1h 10m 08s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e1e5faaf9e1b4dd0b6f6daa6729a878d : SUCCESS in 49m 39s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/29a1099e9258448e806f8ff77a557ad0 : SUCCESS in 1h 13m 12s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/f63f56848b9843d29d64454c29127875 : SUCCESS in 1h 25m 45s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b491b582d00c4455b4cbbc4e92a41d05 : SUCCESS in 2h 13m 14s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/e5e8a00e9e2e424d8df9218c7613a2f5 : SUCCESS in 1h 02m 51s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/d69ed2b7f41b4f27ae65ac159af79f80 : SUCCESS in 1h 03m 00s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/945b8f5391d54b2ebabd7b38a5d07ce9 : SUCCESS in 47m 26s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/1b49488286de40668d6c1883499b4bd4 : SUCCESS in 54m 51s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/b94b33a319dd4141ac7ca1b3993c3276 : SUCCESS in 1h 06m 27s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"9b71cb3efedfd37d585e1682bc2b3ab2dd196429","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-23 15:02:47.000000000","message":"Patch Set 7: Backport-Candidate+1 Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"3afd0eeaa2375b18a198f855a89690d909607000","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 17:52:54.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/031be5acb0e94f4c8fa0b0ca8cd4d89f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/416723f06990486780f8f53441b0ed9c : TIMED_OUT in 3h 04m 43s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b5a60595fd8f4fbba55eb9e7653c1d25 : TIMED_OUT in 3h 03m 35s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9667750d9d034948af941878b0dc68a4 : SUCCESS in 2h 55m 11s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/947c30e3e4764ade8455d06047e0272e : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d19e1aff58804cfcb8d9b40900da9697 : SUCCESS in 5m 06s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1574121337be4f6d94e0a565a77e298a : TIMED_OUT in 3h 07m 18s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d0e386e507aa444085c76401635e24da : FAILURE in 2h 43m 00s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"97b804b6ed45780f2bd24eab3f489956aac7966b","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-08-28 07:16:38.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bf5e9241e6e78ee9d7adf62ccc03b451852d1c60","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 07:36:32.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"573d26f25c0cca550d51918db4ff5eb2fe13377b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 08:28:54.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"9532414b722547b949e5807368b5eebf20f573b2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 08:31:26.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"5c2d2c29d3e53f01b4242553d6ed25275d7b90ef","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-28 08:46:21.000000000","message":"Patch Set 8: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"85a4cf09acfbcc934bf47febe9c3714716286560","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 08:47:56.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"61ce6ebf06f4ce290f4f0891c92b21f0ad20712d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 09:35:49.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b973cac360d845ea8d42e21e7262395d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/097db54fe76b4802ade957ed376854c3 : FAILURE in 32m 06s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c87223efff6e43959c992b575f6d28fc : FAILURE in 21m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/53386e58f34d4059b0dadb2b4b8bb63f : FAILURE in 21m 34s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0278800187cb45afb41427f7f49d7942 : FAILURE in 27m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6d14e7f1dab0426d90ec2024f1b7df3a : FAILURE in 15m 24s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c9150dd8294e480e950dda990d81f9c8 : FAILURE in 16m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0fa19b544bc4c93b3657de015e9562e : SUCCESS in 2m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6ba9713344a4155bf3f7505d78288b4 : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e611f84c911b45d4a1fae23afbc6a992 : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4896d22841224b1ea1d1ea259774e510 : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01a6c205a4cd445da3ff7af1f9250005 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c278c285ee5f4c44b0bd9374081095f4 : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46fbd2f56be84cea94857f84aa6e7f33 : SUCCESS in 4m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ec004d3b8cf3425abbb4e5c0b37832a4 : FAILURE in 24m 05s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d1b32e7120f74e808c55b96b87d95001 : FAILURE in 32m 27s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d2c70122b2cd49c6bcd2bca1a3cb4364 : FAILURE in 21m 16s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/01fab533673342edab2dcef5ca170f4d : FAILURE in 19m 51s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/7575f5ef97574a79ba6c97e9d3426692 : FAILURE in 20m 38s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/eced9ced76ad4fbda7eb3512d95b97ef : FAILURE in 21m 44s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/0fdf743c2c3446b7ab452f1640cbefb0 : FAILURE in 19m 12s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/85ba3bcc95384ad2b8c012dc59a20551 : FAILURE in 20m 21s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/832a7e15a58849eea6b53059d0e6a22d : FAILURE in 21m 20s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/46488fe671ad4b258edbb3321041fd8b : FAILURE in 21m 09s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/55b16c0f88dc4d518d04fa212cfd83be : FAILURE in 19m 23s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f2f1f75b387f495a9a34c53993d7ad0b : FAILURE in 27m 47s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e6c53784b83a4328953fcb369b92b848 : FAILURE in 56m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4d46fbb110ac48ba82ddeccfe3af6772 : FAILURE in 15m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f4f62c854d5d47a9ac28730705387fcb : FAILURE in 14m 18s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/bdfbdb9662d84dc7a36acea520c6c809 : FAILURE in 16m 36s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e213fb3e3d03481c820607062fcb0775 : FAILURE in 15m 28s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/ecbea6189ebc4bce9ae5a9cbae724be1 : FAILURE in 16m 01s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/0ef458d15d554ad894cea7689eb20454 : FAILURE in 16m 35s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/c78b10bda04f4bfa893b9263f71afb7a : FAILURE in 37m 22s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/1d2b44686517435f936c0a8e1dc446f6 : FAILURE in 13m 32s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/13719496cb32446c91c4b445006ca415 : FAILURE in 20m 08s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b3777e42f736750517cc98a4f754f63af545615e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 09:51:29.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"afa997e3c39e77a92f9ce8cfe2b678f36b82a681","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 11:21:34.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":10},{"id":"268019b2ff147d0616b06073bbe2544bc52fdaf9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 12:38:11.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":11},{"id":"64d6c41d25a37365a0413c9603dcd6c62f8d2f60","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-28 15:29:15.000000000","message":"Patch Set 11: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"d918876e2fd106f1f13c3bff3c13c1f5dd1fb5dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 15:37:06.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54c5217f05d04f1591c05f23c3ffa59f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8a4bd82d29cf412eae12064ef148ff8b : SUCCESS in 26m 16s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a60bca2fb09e43d9b3b05c43ae7c8c67 : SUCCESS in 1h 10m 45s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/17a5fef6fd814d3b89e101246679c2c0 : SUCCESS in 21m 20s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c81713e89995487296dcb8f4aad6d723 : SUCCESS in 26m 27s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/57fab26eb289477bb08acd624f680eb9 : SUCCESS in 59m 16s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/aa3e249512f04fa58d19bfb81ac9a4d0 : SUCCESS in 1h 22m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba6ece8fb67d4a61b805a55dbc5b8ec2 : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c1033de055e4b159be6f4a479e2431c : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be6fdd6937f0498d9a2e641a367a1af0 : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bd3d68a638b24b098c06130e819d3b01 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41f3fa228c724921aae404e73a38be54 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67f42fc29af44f7fa13e8634b5630949 : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2130fefb22147f898d1542266586c95 : SUCCESS in 3m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b00d33acaa524f5e9d6dc1586ce80ebb : SUCCESS in 32m 35s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/060f8822ba49433ea8ff8cccccf1b027 : SUCCESS in 34m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ea82e08f9bfa42b59a1d45251edc268f : SUCCESS in 1h 08m 36s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f41115419562439097873dfbc202c48c : SUCCESS in 1h 16m 54s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/790d8f59bff84ea3a09e140999fdb1a6 : SUCCESS in 1h 45m 07s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/76b71ecff8cb4f2e903a8247bf3f4668 : FAILURE in 1h 47m 26s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/ada9454e730547ab8fce7c6d7d848dd1 : SUCCESS in 1h 22m 05s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/6a4f7116a3434908a5b9e9a49824142b : SUCCESS in 1h 22m 13s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/35c72c20099649a0ac59b693b8955a9b : SUCCESS in 58m 34s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/fc867502f3ce433499893cba693bfe0f : SUCCESS in 1h 04m 33s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/5c4bf5e733f044dda178cef4fedabfd3 : SUCCESS in 1h 28m 50s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/82ee891c1e484768891e9a683b8a3d0b : SUCCESS in 29m 58s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b77b3b728f1244c89cdcc6a6e33a5c7d : SUCCESS in 58m 16s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b53e4426a2844475be30fad3614777a5 : SUCCESS in 57m 07s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/25e95f7f188844879038617ede0e4cda : SUCCESS in 1h 04m 38s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/83339e0051114958ba62a9ddd2db646b : SUCCESS in 1h 25m 49s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1585f9aeef3a460fb05fef0624c24922 : SUCCESS in 2h 20m 40s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/a493655152434e23895cb430697fe12a : SUCCESS in 1h 07m 06s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/f501889527de4b26ae81be0f550611e6 : SUCCESS in 1h 19m 24s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/3804041dad5a46a1ac9c5bcc8b9f3820 : SUCCESS in 1h 03m 53s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/22eaa9b9a74f4c2784ae84f21a6c5594 : SUCCESS in 57m 58s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/d6d72b832c65440fb4be9049fb5115d5 : SUCCESS in 1h 21m 41s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"3e5dae7988a3486b735d1b6244508c52a5b4138b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-28 15:39:43.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"f39c9c558fde86b802d42d1ffc9ef899647db14f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 15:55:54.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"b155b9c44bb571a60264415e74832cd03aa3c030","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 15:56:15.000000000","message":"Patch Set 12: Published edit on patch set 11.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"f013545ce01d2599970fbc5b3a125d5d9c6028b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 15:58:59.000000000","message":"Patch Set 13: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":13},{"id":"6040bcc53d5441a07430d158bf5f44b4688c1bf8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 16:00:18.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"d8eff2f814b286f080a5451a840be3ce1186d607","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 18:22:42.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8eaf967d7304227b7a7062e11353562\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d6d5d5c3f26a4c34b6564b2c67bed013 : SUCCESS in 28m 52s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2746c340b8764c68a03eeaeab9984ce7 : SUCCESS in 1h 05m 59s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/01fba52c475f473a96b07aa489f0e76b : SUCCESS in 21m 31s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2c8a3830d57f4e8cb9a53212cf0d8385 : SUCCESS in 29m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e2d16cc70e5a42559e96bf38aa989566 : SUCCESS in 1h 04m 33s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/eba6f4fe973c45f795a2e46dd647c624 : SUCCESS in 1h 21m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd3aee7292dd46c7ba52ce56afa0e00d : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5c56ae6fe7649548437576873978a5a : SUCCESS in 2m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c5af06addb3442c695ca04ea720a66b4 : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/892aa3a8946a473e80737eca3aed6b09 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62e84c9c3bd2465abe050217fee1e767 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/050195304af447b999de4a014f7a4ef9 : SUCCESS in 4m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d8eb86714aa48429ad81f6724a4715d : SUCCESS in 3m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e2a4b538d3344bebabf6e2372fad2f7b : SUCCESS in 28m 46s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/72438d6f0307495aabc34d4c702a1641 : SUCCESS in 36m 55s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f581b29598f54cf1b9a0ffb2c030825f : SUCCESS in 1h 10m 58s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9287ebee37504d92a03e05bb652c787f : SUCCESS in 1h 17m 05s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/e0a0c2bcc12d4ec0ac09ae87837c4c69 : SUCCESS in 1h 36m 00s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/bb85522c634d4453866016761b21f696 : SUCCESS in 2h 11m 53s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/e62619f7b33f4b8ea4565a9e57a871d6 : SUCCESS in 1h 08m 07s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/faf218a28aac43d1aed97516de1ed1a3 : SUCCESS in 1h 24m 57s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/cda2bdc2d8ff4b9880e598fe9c5ceb0e : SUCCESS in 57m 00s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/2ba26714b8534f29bc699e6fa9eb8a5f : SUCCESS in 1h 14m 33s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/2e3e50c323b147db8d891f74ac75aa22 : SUCCESS in 1h 18m 33s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5e39eda0224f44319864ecf4d3ddcf0f : SUCCESS in 27m 23s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8b8d2bd87f554fe18f7f8aae2110b8f9 : SUCCESS in 56m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/79e2711065434182ac80e559d89e39cb : SUCCESS in 58m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/814067aaf36a416e9c637a4c8ceb2d83 : SUCCESS in 1h 16m 59s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/0a692acdc20547c69f73455ad984fe27 : SUCCESS in 1h 33m 36s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f0b5a4cdea344ad081248d663093fbfe : SUCCESS in 2h 08m 58s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/8c79ee9c654b43b48036f2e155a529a2 : SUCCESS in 1h 08m 00s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/ecd164f1f32c4a76b15ffee97347e31a : SUCCESS in 1h 08m 27s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/0b0dd31df3e743fea92c8e8754fefc1c : SUCCESS in 52m 20s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/f7e8f2d11775447296d74a9ee7dd6665 : SUCCESS in 1h 03m 06s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/8f7e731d6a144515807fa0f953ae58f3 : SUCCESS in 1h 06m 06s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"3b5ca9c35b2aabdb637b4b38221d7a275e5829bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 19:28:26.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1fdde1137da4b10aba4065963a8c2bf\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/eb09c899e20d49679c07655d9d410ed9 : TIMED_OUT in 3h 04m 28s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3429314a48d147019d3f430346ad7233 : SUCCESS in 1h 44m 38s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/daa22007a98d45b6845e27926e144ea8 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0168128cdf5f4d67a8e579e9033a9d14 : SUCCESS in 4m 00s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"0adb527216c052d2fb495aa10107dc8abb125059","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-08-29 02:39:40.000000000","message":"Patch Set 13: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":13},{"id":"c68b5f08b1ceced37471693de11fad1fa345783f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-29 06:21:43.000000000","message":"Patch Set 14: Patch Set 13 was rebased\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"7902b7c3b6dfe877cc0663fa7ef9d8d139e586a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 09:27:43.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a8eac2b85b74222947dd15d5b5c427b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/dfd000800cc74f8fb5296da7cf6fd93a : SUCCESS in 32m 19s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/94889acbb2834cb7b9c7e071017a9897 : SUCCESS in 1h 04m 58s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/afb157138339420ca756bf0036dbe218 : SUCCESS in 16m 02s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/19a84f0cf2a342be8ec2f2d9b599a527 : SUCCESS in 29m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/cd71873d43e34e06b1c805b188931b38 : SUCCESS in 1h 02m 10s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c22a67fd54d74a098b00308e89e4bc7e : SUCCESS in 1h 23m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f73aa048bbd4f8fb68e9ae509694305 : SUCCESS in 2m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8454fcea423444d3b768742a29f9384a : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e52eccc9e76b425b9a6eb59f0d14942a : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e0fdae85e6404d3fa82d55f5f8f4e4e7 : SUCCESS in 2m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52fbf91379da469d8d963700f689c617 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/202feadcff924a1097d5a98e7d6c2f5a : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eafe3556d18c40979ace3d312f7d1adf : SUCCESS in 3m 50s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/55f4772ecf12465f838a6e47c1afc3f7 : SUCCESS in 27m 01s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2207106c4b78447c8f7ef4bf9078a48f : SUCCESS in 37m 41s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/da6d6987092a482bad2693bb45b23eb8 : SUCCESS in 1h 04m 28s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6b9d352dd85c411195c09929ea856a40 : SUCCESS in 1h 22m 38s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/58ad07f4efe84410be9897f07bff966b : FAILURE in 1h 22m 20s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/fb1b38ffa25d455d80f40bf5314d0311 : SUCCESS in 2h 36m 17s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/b6f29bec369a4ce4b9674ff03bd686b5 : SUCCESS in 1h 17m 54s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/47823b372f994407b7b1d57d38d43330 : SUCCESS in 1h 18m 43s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/0f12f51289714687adb117a0775d4bf9 : SUCCESS in 56m 13s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/4c2dd9a8966045d5bb489955378fb444 : SUCCESS in 1h 22m 03s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/0e828a0bbc1e47ed97caaa8e0384ddf7 : SUCCESS in 1h 28m 48s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8668a0467fe64cad9d2a99b1e4681d5c : SUCCESS in 19m 07s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/408bac26518240f287f5f8e403d9bf0b : SUCCESS in 1h 02m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5e7af5038331472babb47bad18fd2060 : SUCCESS in 51m 42s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a91e803811b741b8a1e0f3170c0661c0 : SUCCESS in 1h 10m 14s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/b8d6bb80bf704fab8db1c45ba42c001e : SUCCESS in 1h 41m 17s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c68a98f0d8204bd683e589ae3350a37b : SUCCESS in 2h 24m 07s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/967046e32d96478abe3ae28f3aac4fe3 : SUCCESS in 1h 00m 07s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/6522fbfaa86843c2a8994d49680e8b10 : SUCCESS in 1h 06m 13s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/959cc24d70da483fb202151dd2443e64 : SUCCESS in 46m 54s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/8d7389de7bd04a5ebcb1d15a601142cd : SUCCESS in 55m 07s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/cf27eae5b4cf49eeabc2e4d0f8c9c6b2 : SUCCESS in 1h 16m 39s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"0989771000245a46f5909d1c1feb2f338b1aa623","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 10:15:16.000000000","message":"Patch Set 14:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a7aa669f501469392e9153e5751c458\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ec497ed61b434d4c85f8526ec40914c8 : TIMED_OUT in 3h 04m 02s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/01b9af98a359474f83f2a6b3647adf6e : FAILURE in 2h 58m 47s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d1af9cd15efd4b1ca1b7cdbf55c5e445 : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c4a40ace92cf4427acc401c114fadc90 : SUCCESS in 6m 26s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"a71a32d27532d3b21430f3dbda944b837cfe5415","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-29 23:37:27.000000000","message":"Patch Set 14: Review-Priority+1","accounts_in_message":[],"_revision_number":14},{"id":"b9bbb2441c3850003aa8bce8dd717295fb51a8d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-29 23:38:29.000000000","message":"Patch Set 15: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"6020428ea9f8b8c52f512c0a678a49c29aec4456","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 02:28:11.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a79d500102844df9edea293d90a752a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/87902f1ed0414e4ab907befd16f86373 : SUCCESS in 27m 13s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/84f5ac979fa2429f99818875a68eb0ec : SUCCESS in 1h 02m 48s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b851e10ae7354918818a385f45289d36 : SUCCESS in 16m 48s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1a9c0ecf3e7d478f85a4af409b88c67a : SUCCESS in 29m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/95da1ae32b7843f98c6728a77873d509 : SUCCESS in 1h 02m 06s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/bc5308ea32794fddbfde5b23510917cf : SUCCESS in 1h 23m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26b42d1ad76546b4b0ff4d676213543a : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1337a3f7a494bbaa81715b386812b23 : SUCCESS in 2m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1433c9b228ca488da92bff8d89472fd6 : SUCCESS in 3m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6c0a6bd6ae884d4291620586cfd0c1cb : SUCCESS in 3m 13s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e817e0459db0496f9efeaf4b655115d0 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47024f774e094e9e9685e5a92577a17e : SUCCESS in 3m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/556b4ad8651b4be3a61a3272ad74d4b3 : SUCCESS in 4m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c4b047a7f5504b1091e3edc66e0aeb29 : SUCCESS in 29m 53s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b0a13d528c314945af394256b9c09d04 : SUCCESS in 30m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7f035c5554f1437ea1633c6d05bd87e5 : SUCCESS in 1h 04m 40s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5c94c8cb55374fd2aa46457865dee141 : SUCCESS in 1h 20m 25s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/a116c12a713841e4b21df5908ae027b0 : SUCCESS in 1h 40m 36s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/01cccddb26964f3c8a756de3fb8e4b7b : SUCCESS in 2h 43m 57s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/7ccdc420c1c7462e887a0749415e23a4 : SUCCESS in 1h 07m 28s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/7a7d5160492c458b87ae64b9941cca52 : SUCCESS in 1h 20m 23s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/981ab03e70314a8c8c73b20d12718f75 : SUCCESS in 1h 02m 07s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/02cfc56f6411443a81a81cbae090d252 : SUCCESS in 1h 11m 24s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/fb08783d34864c909d9c69a667be22fc : SUCCESS in 1h 17m 55s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a50b42d8e0ac419d86dde876918e1b19 : SUCCESS in 18m 10s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7632c4d3be91423d8ad965be8daf6137 : SUCCESS in 53m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3cc0078d5f39440cb2219d95ad885cff : SUCCESS in 54m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5486a8a376934f53ac9d65dae28f0930 : SUCCESS in 1h 12m 10s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/21a4ca45bd924defac62628fd8cb6771 : SUCCESS in 1h 23m 46s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1bb6fc4f56ed4efd9c4cd1b07f543b0d : FAILURE in 1h 07m 35s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/7bd20d2c111f42639b9d9960aede978f : SUCCESS in 1h 13m 48s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/3af0a2e2d753430899b576da59c34690 : SUCCESS in 1h 01m 35s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/a70cb6caa7e24d509f9eddb857b366aa : SUCCESS in 42m 44s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/be843297a0b74c4ca184a19566661801 : SUCCESS in 53m 58s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/adca28db6c4643f9a72d38d4739f1da7 : SUCCESS in 58m 40s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"4faf06c49ac6442c1e8fde7368e9e0e69fc266d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 02:47:30.000000000","message":"Patch Set 15:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f8071aa1ca748e5a92a2794725c7c10\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/acda4453e7494828a8dc40e927c981a3 : TIMED_OUT in 3h 07m 07s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7801e50f7f4c4369a8a5107669d48928 : FAILURE in 1h 45m 46s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d9d73e1920642f3870028aafb89e9a1 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a0f533536d694889976897775308ceb9 : SUCCESS in 3m 44s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"2a1b05610a9176ebb65fc5b372fe0fcd677fd76f","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2024-08-30 07:03:04.000000000","message":"Patch Set 15: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"2fa13eb442ee53561575e8612ffdad09011cd717","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-08-30 07:05:22.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"8e5afa7e2cb2ad4587ba38f22a97a2a6cf7c258a","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-30 07:08:30.000000000","message":"Patch Set 15: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"6868c0121a5772bfb8a825802a9f5b43b5ba80c4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-08-30 10:26:19.000000000","message":"Patch Set 15: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"69e7bebd959c48f8f07050abe83ad01263e060d3","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-03 06:37:33.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"c51863085220cd7f7118fb25fba8092aac7a3183","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 06:37:54.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"7de0eecd888f223e72ac01f2d66d74e0b5d6ed9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 08:22:18.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/760871472b7f4e869e3135debfbe9310\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1a2d090df1ac40e28a3596ea114e307f : SUCCESS in 22m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3ffde8443df84fd39c83f74c96d09118 : SUCCESS in 28m 50s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/80c66190a73e43ccb68a98dc5956e4ce : SUCCESS in 1h 00m 16s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d3e14385a5354b8d98552ad38fbaaeaf : SUCCESS in 1h 25m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6457ff742a74d498793aa928ba08019 : SUCCESS in 2m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/84a03a61208a4c98bdbadcd3cdeaf9fd : SUCCESS in 2m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/650f5dcc752b4399a293ff6a9cad0f58 : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a262e1ca52434bbea8b3e451058ca116 : SUCCESS in 3m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc0de4dc4b714d57904a680764f3dcca : SUCCESS in 3m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/19f0eb7056ed438a81f29419f3658be6 : SUCCESS in 21m 10s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/08183ae9c88f4223a555a88c1deab5ef : SUCCESS in 30m 22s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/76a5a0ef5b6444ca992231425ad0f254 : SUCCESS in 1h 01m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ecb6d1da214f453fac1b0dc5d4e4257e : SUCCESS in 1h 20m 14s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ca5281cae1a34b42918b9b01e109e2a0 : SUCCESS in 15m 30s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/75a8bc989db24a5b93fb38c247af14aa : SUCCESS in 59m 24s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e9976c8e7fe64a229feaf2268c11b7d3 : SUCCESS in 59m 07s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ee7c15399feb4a329c6482e38b8fe853 : SUCCESS in 1h 17m 41s","accounts_in_message":[],"_revision_number":15},{"id":"ddea88f78232873ef4b9c50df7c633ca58a2e0ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 08:22:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15},{"id":"5dedb1011ea094782f0bf62355d9ec97f9a96f63","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 08:23:24.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9e115b550d748769d5554d9de0fc35f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be575b651cc04deb976868f27abb04a9 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e481013f79542909e6f7995b9a2147c : SUCCESS in 41s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"978a2864f923fa2e65afa13808571ea7738729e6","revisions":{"c766d135278fdc0c1f69ca0cc414245ad2f6169d":{"kind":"REWORK","_number":1,"created":"2024-07-23 16:12:11.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/59/924759/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/924759/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/924759/1"}}},"commit":{"parents":[{"commit":"ec64f299bf8c4c75ceee22033ee9fd4d8d5a2d4e","subject":"Fix codespell on master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec64f299bf8c4c75ceee22033ee9fd4d8d5a2d4e"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-07-23 15:58:09.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-07-23 16:12:10.000000000","tz":60},"subject":"Fix symlinking of openstack-base source","message":"Fix symlinking of openstack-base source\n\nOriginal symlinking works fine with default openstack-base.\n\nHowever, when user supplies the source of it, Kolla cannot\nguarantee that the source direcotry name contains ``requirements``.\n\nThis patch fixes the cases where ``ln`` command fails to find\ncorrect source by choosing the directory that is NOT ``plugins``.\n\nChange-Id: I5a9162a30b23cfabb64dabe5c12c226fa8627abd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c766d135278fdc0c1f69ca0cc414245ad2f6169d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c766d135278fdc0c1f69ca0cc414245ad2f6169d"}]},"branch":"refs/heads/master"},"5ca04eda60a4a2329d58ed6b2356a0eb1ec649cb":{"kind":"REWORK","_number":2,"created":"2024-07-25 08:45:06.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/59/924759/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/924759/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/924759/2"}}},"commit":{"parents":[{"commit":"ec64f299bf8c4c75ceee22033ee9fd4d8d5a2d4e","subject":"Fix codespell on master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec64f299bf8c4c75ceee22033ee9fd4d8d5a2d4e"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-07-23 15:58:09.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-07-25 08:40:20.000000000","tz":60},"subject":"Fix symlinking of openstack-base source","message":"Fix symlinking of openstack-base source\n\nOriginal symlinking works fine with default openstack-base.\n\nHowever, when user supplies git type source of it, we cannot\nguarantee that the source direcotry name contains ``requirements``.\n\nThis patch fixes it by moving plugins directory outside of\nopenstack-base directory, which ultimately allow us to use simpler\nsymlinking as openstack-base-archive would be the only directory\ninside.\n\nChange-Id: I5a9162a30b23cfabb64dabe5c12c226fa8627abd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ca04eda60a4a2329d58ed6b2356a0eb1ec649cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5ca04eda60a4a2329d58ed6b2356a0eb1ec649cb"}]},"branch":"refs/heads/master"},"2b10c0b37b62ee09cae9b4d072ad16b8f6f3ec03":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-07-25 11:16:58.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/59/924759/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/924759/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/924759/3"}}},"commit":{"parents":[{"commit":"ec64f299bf8c4c75ceee22033ee9fd4d8d5a2d4e","subject":"Fix codespell on master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec64f299bf8c4c75ceee22033ee9fd4d8d5a2d4e"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-07-23 15:58:09.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-07-25 11:12:28.000000000","tz":60},"subject":"Fix symlinking of openstack-base source","message":"Fix symlinking of openstack-base source\n\nOriginal symlinking works fine with a tarball source for the\nrequirements repo.\n\nHowever, when user supplies git type source of it, we cannot\nguarantee that the source direcotry name contains ``requirements``.\n\nThis patch fixes it by moving plugins directory outside of\nopenstack-base directory, which ultimately allow us to use simpler\nsymlinking as openstack-base-archive would be the only directory\ninside.\nThis also follows the pattern used in Dockerfile of other services.\n\nCloses-Bug: #2074075\nChange-Id: I5a9162a30b23cfabb64dabe5c12c226fa8627abd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b10c0b37b62ee09cae9b4d072ad16b8f6f3ec03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2b10c0b37b62ee09cae9b4d072ad16b8f6f3ec03"}]},"branch":"refs/heads/master"},"91d7b812ea9d7ef4a077210836e6542ca1e5173f":{"kind":"REWORK","_number":4,"created":"2024-07-26 10:25:13.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/59/924759/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/924759/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/924759/4"}}},"commit":{"parents":[{"commit":"ec64f299bf8c4c75ceee22033ee9fd4d8d5a2d4e","subject":"Fix codespell on master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec64f299bf8c4c75ceee22033ee9fd4d8d5a2d4e"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-07-23 15:58:09.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-07-26 10:24:56.000000000","tz":60},"subject":"Fix symlinking of openstack-base source","message":"Fix symlinking of openstack-base source\n\nOriginal symlinking works fine with a tarball source for the\nrequirements repo.\n\nHowever, when user supplies git type source of it, we cannot\nguarantee that the source direcotry name contains ``requirements``.\n\nThis patch fixes it by moving plugins directory outside of\nopenstack-base directory, which ultimately allow us to use simpler\nsymlinking as openstack-base-archive would be the only directory\ninside.\nThis also follows the pattern used in Dockerfile of other services.\n\nCloses-Bug: #2074075\nChange-Id: I5a9162a30b23cfabb64dabe5c12c226fa8627abd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91d7b812ea9d7ef4a077210836e6542ca1e5173f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91d7b812ea9d7ef4a077210836e6542ca1e5173f"}]},"branch":"refs/heads/master"},"374c555866150843d162d1059f40f2f5535e65cc":{"kind":"REWORK","_number":5,"created":"2024-08-05 15:54:36.000000000","uploader":{"_account_id":35264,"name":"Alex Welsh","email":"alex@stackhpc.com","username":"alex-welsh"},"ref":"refs/changes/59/924759/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/924759/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/924759/5"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-07-23 15:58:09.000000000","tz":60},"committer":{"name":"Alex-Welsh","email":"alex@stackhpc.com","date":"2024-08-05 15:54:29.000000000","tz":60},"subject":"Fix symlinking of openstack-base source","message":"Fix symlinking of openstack-base source\n\nOriginal symlinking works fine with a tarball source for the\nrequirements repo.\n\nHowever, when user supplies git type source of it, we cannot\nguarantee that the source directory name contains ``requirements``.\n\nThis patch fixes it by moving plugins directory outside of\nopenstack-base directory, which ultimately allow us to use simpler\nsymlinking as openstack-base-archive would be the only directory\ninside.\nThis also follows the pattern used in Dockerfile of other services.\n\nCloses-Bug: #2074075\nChange-Id: I5a9162a30b23cfabb64dabe5c12c226fa8627abd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/374c555866150843d162d1059f40f2f5535e65cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/374c555866150843d162d1059f40f2f5535e65cc"}]},"branch":"refs/heads/master"},"ae0875f2c82c62a4f1fee5c37beedfbe568422b4":{"kind":"REWORK","_number":6,"created":"2024-08-23 06:45:23.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/59/924759/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/924759/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/924759/6"}}},"commit":{"parents":[{"commit":"ec64f299bf8c4c75ceee22033ee9fd4d8d5a2d4e","subject":"Fix codespell on master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec64f299bf8c4c75ceee22033ee9fd4d8d5a2d4e"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-07-23 15:58:09.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-08-23 06:44:50.000000000","tz":60},"subject":"Fix symlinking of openstack-base source","message":"Fix symlinking of openstack-base source\n\nOriginal symlinking works fine with a tarball source for the\nrequirements repo.\n\nHowever, when user supplies git type source of it, we cannot\nguarantee that the source direcotry name contains ``requirements``.\n\nThis patch fixes it by moving plugins directory outside of\nopenstack-base directory, which ultimately allow us to use simpler\nsymlinking as openstack-base-archive would be the only directory\ninside.\n\nCloses-Bug: #2074075\nChange-Id: I5a9162a30b23cfabb64dabe5c12c226fa8627abd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae0875f2c82c62a4f1fee5c37beedfbe568422b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae0875f2c82c62a4f1fee5c37beedfbe568422b4"}]},"branch":"refs/heads/master"},"d264db5965a8294cf6413f8d456d0aabd02263da":{"kind":"REWORK","_number":7,"created":"2024-08-23 06:50:59.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/59/924759/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/924759/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/924759/7"}}},"commit":{"parents":[{"commit":"ec64f299bf8c4c75ceee22033ee9fd4d8d5a2d4e","subject":"Fix codespell on master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec64f299bf8c4c75ceee22033ee9fd4d8d5a2d4e"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-07-23 15:58:09.000000000","tz":60},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-08-23 06:50:39.000000000","tz":60},"subject":"Fix symlinking of openstack-base source","message":"Fix symlinking of openstack-base source\n\nOriginal symlinking works fine with a tarball source for the\nrequirements repo.\n\nHowever, when user supplies git type source of it, we cannot\nguarantee that the source direcotry name contains ``requirements``.\n\nThis patch fixes it by moving plugins directory outside of\nopenstack-base directory, which ultimately allow us to use simpler\nsymlinking as openstack-base-archive would be the only directory\ninside.\n\nCloses-Bug: #2074075\nChange-Id: I5a9162a30b23cfabb64dabe5c12c226fa8627abd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d264db5965a8294cf6413f8d456d0aabd02263da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d264db5965a8294cf6413f8d456d0aabd02263da"}]},"branch":"refs/heads/master"},"515d49fca876452d1ae70f70c198fbcce7f01860":{"kind":"REWORK","_number":8,"created":"2024-08-28 08:28:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/59/924759/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/924759/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/924759/8"}}},"commit":{"parents":[{"commit":"ec64f299bf8c4c75ceee22033ee9fd4d8d5a2d4e","subject":"Fix codespell on master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec64f299bf8c4c75ceee22033ee9fd4d8d5a2d4e"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-07-23 15:58:09.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 08:28:54.000000000","tz":0},"subject":"Fix symlinking of openstack-base source","message":"Fix symlinking of openstack-base source\n\nOriginal symlinking works fine with a tarball source for the\nrequirements repo.\n\nHowever, when user supplies git type source of it, we cannot\nguarantee that the source direcotry name contains ``requirements``.\n\nThis patch fixes it by moving plugins directory outside of\nopenstack-base directory, which ultimately allow us to use simpler\nsymlinking as openstack-base-archive would be the only directory\ninside.\n\nCloses-Bug: #2074075\nChange-Id: I5a9162a30b23cfabb64dabe5c12c226fa8627abd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/515d49fca876452d1ae70f70c198fbcce7f01860"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/515d49fca876452d1ae70f70c198fbcce7f01860"}]},"branch":"refs/heads/master"},"be8586760af7e6755ad0208070c8e2c06fcb50b9":{"kind":"REWORK","_number":9,"created":"2024-08-28 09:51:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/59/924759/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/924759/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/924759/9"}}},"commit":{"parents":[{"commit":"ec64f299bf8c4c75ceee22033ee9fd4d8d5a2d4e","subject":"Fix codespell on master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec64f299bf8c4c75ceee22033ee9fd4d8d5a2d4e"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-07-23 15:58:09.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 09:51:29.000000000","tz":0},"subject":"Fix symlinking of openstack-base source","message":"Fix symlinking of openstack-base source\n\nOriginal symlinking works fine with a tarball source for the\nrequirements repo.\n\nHowever, when user supplies git type source of it, we cannot\nguarantee that the source direcotry name contains ``requirements``.\n\nThis patch fixes it by moving plugins directory outside of\nopenstack-base directory, which ultimately allow us to use simpler\nsymlinking as openstack-base-archive would be the only directory\ninside.\n\nCloses-Bug: #2074075\nChange-Id: I5a9162a30b23cfabb64dabe5c12c226fa8627abd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be8586760af7e6755ad0208070c8e2c06fcb50b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/be8586760af7e6755ad0208070c8e2c06fcb50b9"}]},"branch":"refs/heads/master"},"9aa3813940413a6f19039fc66ef8c62faa128217":{"kind":"REWORK","_number":10,"created":"2024-08-28 11:21:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/59/924759/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/924759/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/924759/10"}}},"commit":{"parents":[{"commit":"ec64f299bf8c4c75ceee22033ee9fd4d8d5a2d4e","subject":"Fix codespell on master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec64f299bf8c4c75ceee22033ee9fd4d8d5a2d4e"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-07-23 15:58:09.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 11:21:33.000000000","tz":0},"subject":"Fix symlinking of openstack-base source","message":"Fix symlinking of openstack-base source\n\nOriginal symlinking works fine with a tarball source for the\nrequirements repo.\n\nHowever, when user supplies git type source of it, we cannot\nguarantee that the source direcotry name contains ``requirements``.\n\nThis patch fixes it by moving plugins directory outside of\nopenstack-base directory, which ultimately allow us to use simpler\nsymlinking as openstack-base-archive would be the only directory\ninside.\n\nCloses-Bug: #2074075\nChange-Id: I5a9162a30b23cfabb64dabe5c12c226fa8627abd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9aa3813940413a6f19039fc66ef8c62faa128217"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9aa3813940413a6f19039fc66ef8c62faa128217"}]},"branch":"refs/heads/master"},"65f6597358b75a005d76743a8e61033b3f3f651f":{"kind":"REWORK","_number":11,"created":"2024-08-28 12:38:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/59/924759/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/924759/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/924759/11"}}},"commit":{"parents":[{"commit":"ec64f299bf8c4c75ceee22033ee9fd4d8d5a2d4e","subject":"Fix codespell on master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec64f299bf8c4c75ceee22033ee9fd4d8d5a2d4e"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-07-23 15:58:09.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 12:38:11.000000000","tz":0},"subject":"Fix symlinking of openstack-base source","message":"Fix symlinking of openstack-base source\n\nOriginal symlinking works fine with a tarball source for the\nrequirements repo.\n\nHowever, when user supplies git type source of it, we cannot\nguarantee that the source direcotry name contains ``requirements``.\n\nThis patch fixes it by moving plugins directory outside of\nopenstack-base directory, which ultimately allow us to use simpler\nsymlinking as openstack-base-archive would be the only directory\ninside.\n\nCloses-Bug: #2074075\nChange-Id: I5a9162a30b23cfabb64dabe5c12c226fa8627abd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65f6597358b75a005d76743a8e61033b3f3f651f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65f6597358b75a005d76743a8e61033b3f3f651f"}]},"branch":"refs/heads/master"},"357d3beef9352c2493ab37102d609f0e7eaa22c3":{"kind":"REWORK","_number":12,"created":"2024-08-28 15:56:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/59/924759/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/924759/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/924759/12"}}},"commit":{"parents":[{"commit":"ec64f299bf8c4c75ceee22033ee9fd4d8d5a2d4e","subject":"Fix codespell on master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec64f299bf8c4c75ceee22033ee9fd4d8d5a2d4e"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-07-23 15:58:09.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 15:56:14.000000000","tz":0},"subject":"Fix symlinking of openstack-base source","message":"Fix symlinking of openstack-base source\n\nOriginal symlinking works fine with a tarball source for the\nrequirements repo.\n\nHowever, when user supplies git type source of it, we cannot\nguarantee that the source direcotry name contains ``requirements``.\n\nThis patch fixes it by moving plugins directory outside of\nopenstack-base directory, which ultimately allow us to use simpler\nsymlinking as openstack-base-archive would be the only directory\ninside.\n\nCloses-Bug: #2074075\nChange-Id: I5a9162a30b23cfabb64dabe5c12c226fa8627abd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/357d3beef9352c2493ab37102d609f0e7eaa22c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/357d3beef9352c2493ab37102d609f0e7eaa22c3"}]},"branch":"refs/heads/master"},"2c4ddc5f12e6043a69c72fe32589593e812a94dc":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2024-08-28 15:58:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/59/924759/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/924759/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/924759/13"}}},"commit":{"parents":[{"commit":"ec64f299bf8c4c75ceee22033ee9fd4d8d5a2d4e","subject":"Fix codespell on master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec64f299bf8c4c75ceee22033ee9fd4d8d5a2d4e"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-07-23 15:58:09.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 15:58:58.000000000","tz":0},"subject":"Fix symlinking of openstack-base source","message":"Fix symlinking of openstack-base source\n\nOriginal symlinking works fine with a tarball source for the\nrequirements repo.\n\nHowever, when user supplies git type source of it, we cannot\nguarantee that the source direcotry name contains ``requirements``.\n\nThis patch fixes it by moving plugins directory outside of\nopenstack-base directory, which ultimately allow us to use simpler\nsymlinking as openstack-base-archive would be the only directory\ninside.\n\nAlso patching keystone is required, because until now we only\ninstalled requests-kerberos when /plugins/ was not empty\n(so basically never).\n\nCloses-Bug: #2074075\nChange-Id: I5a9162a30b23cfabb64dabe5c12c226fa8627abd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c4ddc5f12e6043a69c72fe32589593e812a94dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c4ddc5f12e6043a69c72fe32589593e812a94dc"}]},"branch":"refs/heads/master","description":"Edit commit message"},"67f314b5563892743b23f3e48b0689ce39f5d015":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2024-08-29 06:21:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/59/924759/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/924759/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/924759/14"}}},"commit":{"parents":[{"commit":"082d90cca02b0e3935363c7ce2e7ae718bcbb789","subject":"keystone: Fix requests-kerberos installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/082d90cca02b0e3935363c7ce2e7ae718bcbb789"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-07-23 15:58:09.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-29 06:21:43.000000000","tz":0},"subject":"Fix symlinking of openstack-base source","message":"Fix symlinking of openstack-base source\n\nOriginal symlinking works fine with a tarball source for the\nrequirements repo.\n\nHowever, when user supplies git type source of it, we cannot\nguarantee that the source direcotry name contains ``requirements``.\n\nThis patch fixes it by moving plugins directory outside of\nopenstack-base directory, which ultimately allow us to use simpler\nsymlinking as openstack-base-archive would be the only directory\ninside.\n\nAlso patching keystone is required, because until now we only\ninstalled requests-kerberos when /plugins/ was not empty\n(so basically never).\n\nCloses-Bug: #2074075\nChange-Id: I5a9162a30b23cfabb64dabe5c12c226fa8627abd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67f314b5563892743b23f3e48b0689ce39f5d015"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67f314b5563892743b23f3e48b0689ce39f5d015"}]},"branch":"refs/heads/master","description":"Rebase"},"978a2864f923fa2e65afa13808571ea7738729e6":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2024-08-29 23:38:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/59/924759/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/924759/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/924759/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/924759/15"}}},"commit":{"parents":[{"commit":"082d90cca02b0e3935363c7ce2e7ae718bcbb789","subject":"keystone: Fix requests-kerberos installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/082d90cca02b0e3935363c7ce2e7ae718bcbb789"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-07-23 15:58:09.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-29 23:38:29.000000000","tz":0},"subject":"Fix symlinking of openstack-base source","message":"Fix symlinking of openstack-base source\n\nOriginal symlinking works fine with a tarball source for the\nrequirements repo.\n\nHowever, when user supplies git type source of it, we cannot\nguarantee that the source direcotry name contains ``requirements``.\n\nThis patch fixes it by moving plugins directory outside of\nopenstack-base directory, which ultimately allow us to use simpler\nsymlinking as openstack-base-archive would be the only directory\ninside.\n\nCloses-Bug: #2074075\nChange-Id: I5a9162a30b23cfabb64dabe5c12c226fa8627abd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/978a2864f923fa2e65afa13808571ea7738729e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/978a2864f923fa2e65afa13808571ea7738729e6"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
