)]}'
{"id":"openstack%2Fkolla~925827","triplet_id":"openstack%2Fkolla~master~I7dea99ebcd673def53b52677832ea9f20bf1b448","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-08-07 08:58:08.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"8726":{"account":{"_account_id":8726,"name":"Victor Morales","email":"chipahuac@hotmail.com","username":"electrocucaracha"},"last_update":"2025-09-09 16:59:40.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-09-09 16:59:40.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I7dea99ebcd673def53b52677832ea9f20bf1b448","subject":"Add rust dependencies required by bcrypt","status":"ABANDONED","created":"2024-08-06 23:33:19.000000000","updated":"2025-09-09 16:59:40.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"730b927f45e89d49953d481938f8af93af785a66","_number":925827,"virtual_id_number":925827,"owner":{"_account_id":8726,"name":"Victor Morales","email":"chipahuac@hotmail.com","username":"electrocucaracha"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-08-07 02:30:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2024-08-07 08:58:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-06 23:37:16.000000000","updated_by":{"_account_id":8726,"name":"Victor Morales","email":"chipahuac@hotmail.com","username":"electrocucaracha"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-08-06 23:37:16.000000000","updated_by":{"_account_id":8726,"name":"Victor Morales","email":"chipahuac@hotmail.com","username":"electrocucaracha"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-08-07 02:30:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e6c3a25478757619cd8ef49a8ade0ec416bb386f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8726,"name":"Victor Morales","email":"chipahuac@hotmail.com","username":"electrocucaracha"},"date":"2024-08-06 23:33:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"184f9b5cc382058bdc66aeaf906a6ca6473af56d","author":{"_account_id":8726,"name":"Victor Morales","email":"chipahuac@hotmail.com","username":"electrocucaracha"},"date":"2024-08-06 23:37:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1e6fd21f8156ad212aa8c97468642642507399c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 02:30:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b4ce3450ee8445a878fe9a9d8d319d4\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2be42cfa8bff4753bfea48916a2ff5d3 : SUCCESS in 27m 01s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/29945042c7dd4477bafb959b1c31369d : FAILURE in 40m 18s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cf10303650af4941bf416ff0ab381e26 : SUCCESS in 16m 34s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/24b6bcaaa394427ba306f9f2e15c94b4 : SUCCESS in 29m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a76edce9652e4b9e816b823b7d8173f6 : FAILURE in 35m 41s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/98777da439cc42d099083fac304d1e77 : FAILURE in 46m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/405b7aa2284f4dc2b762c5f029e63bbd : SUCCESS in 2m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a31ac87b57074a379007c4fcb74235c7 : SUCCESS in 2m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/eb3d0b076a7d48ea8256931f41a2f42b : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7c640c3128764bda87c417ad8c35ef5d : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19e4df8ed4574938896cc00fff6cf3bf : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c75f0054f5784300b27c85df16aef1e0 : SUCCESS in 4m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/92d071acfefc422da66b1db0cf5f53da : SUCCESS in 26m 17s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b15961a6a6394e958f3f5df7401eeb59 : SUCCESS in 38m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6eeb0e9fbd094a2b969dea9fe245177b : FAILURE in 47m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e2b244e4c42048bd9c6aee8b94c089c4 : FAILURE in 47m 59s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/6040386b8eff405d8253ca2be4388a64 : SUCCESS in 1h 55m 33s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/818cf71803f144c68c7607dbc0241bc7 : SUCCESS in 2h 46m 57s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/62620d3b699642f490b08328a0e813cb : FAILURE in 46m 46s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/689fe20b87ef42abafba1a2205dda02f : FAILURE in 57m 26s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/16bb3fdbcc61476d91f9f64b944fa754 : SUCCESS in 1h 00m 50s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/3ba890568bfd4727989baa16cd8b62dc : FAILURE in 50m 42s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/8b24b90445714745a270d961c8ac9917 : FAILURE in 57m 50s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d9770600bfbc421ab7b2e5059cae9063 : SUCCESS in 25m 23s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6e5de8b3f17d41ae8d8247ec0c6b1ec7 : SUCCESS in 51m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6e87c17a41764617a602c579025b65e2 : FAILURE in 35m 45s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1ad527e3b96a42f4b93793ee24ba5d54 : FAILURE in 36m 40s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/4789cb674fc64908817896e288e8600e : SUCCESS in 1h 18m 17s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8668d98097ea4925b7fe2c51fabb472d : SUCCESS in 1h 59m 02s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/ebc8e8cbb9e847a493aebd134ea9c98c : FAILURE in 34m 50s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/09490ec3703d4e8abde653a2616744ec : FAILURE in 46m 24s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/fe7f8b40f09e4a97856db110e9f5b4ab : SUCCESS in 41m 48s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/a77a6bc707404e0fb7be4074c6b3a211 : FAILURE in 34m 28s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/b1544d21960f4c778b938b112e8c3a60 : FAILURE in 43m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9d3e6df85cf3c2bf0755bf45b605a54762de0df4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 02:42:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24c3ec01dfa941ecb770b744af1c3d5a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/329f6d97603745af864b83d07a397ece : TIMED_OUT in 3h 05m 18s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4a261e49bed141ca8b9f46ae7e5c252f : TIMED_OUT in 3h 04m 00s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1ec33f0743b643dd89caf9ac2e7e7666 : FAILURE in 2h 08m 44s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d6a7aaa211b4f2799686ce5214c2256 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b604cb39178f407b9856687846614af6 : SUCCESS in 4m 21s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1c6ce045cf1b4fe495a6213c1c1016f1 : TIMED_OUT in 3h 05m 49s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f0de1e1a033b4861b1715c177a1b0c0e : TIMED_OUT in 3h 04m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9af708bafea7d35eefba58e3ba9df62f4576713c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-07 08:58:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"730b927f45e89d49953d481938f8af93af785a66","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-09 16:59:40.000000000","message":"Abandoned\n\nno updates for over a year","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1475e5939d310c9643dc0260515648b807faa82f","revisions":{"1475e5939d310c9643dc0260515648b807faa82f":{"kind":"REWORK","_number":1,"created":"2024-08-06 23:33:19.000000000","uploader":{"_account_id":8726,"name":"Victor Morales","email":"chipahuac@hotmail.com","username":"electrocucaracha"},"ref":"refs/changes/27/925827/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/27/925827/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/925827/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/925827/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/27/925827/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/27/925827/1"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Victor Morales","email":"chipahuac@hotmail.com","date":"2024-08-06 23:30:09.000000000","tz":-420},"committer":{"name":"Victor Morales","email":"chipahuac@hotmail.com","date":"2024-08-06 23:33:02.000000000","tz":-420},"subject":"Add rust dependencies required by bcrypt","message":"Add rust dependencies required by bcrypt\n\nApparently the python bcrypt package requires the rust compiler OS\npackages during the installation. This change provides those\ndependencies on the template for the openstack-base image template.\n\nChange-Id: I7dea99ebcd673def53b52677832ea9f20bf1b448\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1475e5939d310c9643dc0260515648b807faa82f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1475e5939d310c9643dc0260515648b807faa82f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
