)]}'
{"id":"openstack%2Fkolla~925979","triplet_id":"openstack%2Fkolla~master~Ided63ff1462f7554302c8e2958ac46cc370c8455","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-08-20 12:39:19.000000000","reason":"Change was abandoned"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-08-20 12:13:01.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-08-09 10:26:20.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-08-20 12:39:19.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ided63ff1462f7554302c8e2958ac46cc370c8455","subject":"horizon: Copy wsgi.py to /etc/openstack-dashboard/","status":"ABANDONED","created":"2024-08-08 12:50:27.000000000","updated":"2024-08-20 12:39:19.000000000","total_comment_count":16,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"ea3f1b59ab19d013640be7e10ddbff1034b92e0a","_number":925979,"virtual_id_number":925979,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-08-09 11:19:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":-1,"date":"2024-08-13 12:15:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-08-09 10:26:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":-1,"date":"2024-08-09 11:32:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2024-08-20 12:04:26.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":1,"date":"2024-08-09 09:44:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-08-09 10:26:20.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-08 14:23:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-09 08:17:32.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-08-09 08:37:30.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"CC"},{"updated":"2024-08-09 08:50:38.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-08-09 11:32:02.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2024-08-13 12:15:53.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"addf21baae52d1e52d807d38c3ca591d4ebf5915","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-08 12:50:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0bf8b0911184fe01120f267016b881a7247ba858","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-08 14:23:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c5da8b3d1f44a398899e8dc11840d9f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/fea7e79cca85454d9e7d5986baf5aae8 : SUCCESS in 28m 16s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/765572bb63a741fdb1696dbc014bc243 : SUCCESS in 1h 11m 13s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1784b4bf9c8b4062818216a921243ed3 : SUCCESS in 20m 16s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e15c137178284560a9e42e14712d01fd : SUCCESS in 28m 26s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/288ea0e305d649ad85282b61ac7a0014 : SUCCESS in 1h 04m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/35d688f144f1481bb11eec67c1c541a2 : SUCCESS in 1h 27m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f052142dfbc541d88226ca64cd766173 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a05e821cfa3c4049815aa83264f5264f : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/385e17b31b444fbaaf76bd1292693bd2 : SUCCESS in 3m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/032000650b9f47c88b2afa32bbf62c3b : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0f178fccbfe4542adba2df9e71991f4 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2f1f04c33284b1da9d09164711ac87b : SUCCESS in 4m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5758c45697ca46efa20166747aeafc1c : SUCCESS in 24m 20s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e41841b3eedb46c1aed811854cac4ff6 : SUCCESS in 37m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/bba9e2b0a1384289b5b9bc7ae6647512 : SUCCESS in 1h 03m 46s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/686c75b99fea4d09902a794c49f88be8 : SUCCESS in 1h 15m 37s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2031c0b687244eeb94062b4ae1e0ed8c : SUCCESS in 22m 54s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d009df6d5bfc4128837a13d61c16f0aa : SUCCESS in 48m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d32aae11d3a44d20b2979fe697661029 : SUCCESS in 52m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d8d0220442634458974c8fb65736d565 : SUCCESS in 1h 24m 50s","accounts_in_message":[],"_revision_number":1},{"id":"5866fd790c3b4d0f54f5f586cd07194563820001","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-08 15:40:53.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"37424f23ca7a3c91d28388f589d841db0e2ac2ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-08 16:29:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c9ac9a3434d40a5941552d9c31b5541\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/77d0835e841e48c3931a23f40512b8a4 : FAILURE in 26m 20s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ff867a4b182a4d20b401fa536454d98b : FAILURE in 21m 25s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e6a57addef2c44b8a37848d69cd466e3 : FAILURE in 19m 23s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d5c73ec9f26746f59c50b1a4eeba66bc : FAILURE in 20m 57s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/32187fbc74534e9a9a5292d8c368aa2d : FAILURE in 20m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/91a5cc5632574104a5f3de8c7fb67f11 : FAILURE in 16m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a43a7c3418ad42eeb278e0ee7c3a5130 : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da4a3d4f23f64af2918f74f389035751 : SUCCESS in 2m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d3c9484c4d2a42a6858c98078efe893c : SUCCESS in 2m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a55538ed669f4345bdc23f0000e59451 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f7f7a702aa34035a2706bad88ae9dc5 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f4b46009d364a94ba64b7a8cb805680 : SUCCESS in 6m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/01f48660b4ed4696aa72ec13b8c6b137 : FAILURE in 27m 44s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/36e5e68730dc435e82f80ac630f55029 : FAILURE in 33m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/dccfc0bba0064704bf0afd018fb6b6a5 : FAILURE in 21m 58s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7caeef4f52e94f18bdd23c87ef6075e9 : FAILURE in 20m 07s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d425da04553446a7b8d4655c3af42808 : FAILURE in 19m 31s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/02fe93106f5746919bc89aef12763c17 : FAILURE in 47m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ee178b79f5634109abf99cf491ca3651 : FAILURE in 13m 59s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9040b51d57f248c48681e969cfb7a211 : FAILURE in 14m 21s","accounts_in_message":[],"_revision_number":2},{"id":"36bcf5538b58c4bbfaa40198dddd3e3d97cf9b96","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-08 23:36:30.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/989a0ef3dd4142d583114b24025b6f24\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6a304351d00f46329c1f10a42c354bf4 : TIMED_OUT in 3h 04m 57s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5291aba6403b40fcb4c58e18b21c3c5b : TIMED_OUT in 3h 04m 48s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9ae2065f784f4a15af4296fdcfd8020a : FAILURE in 1h 50m 43s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b8da6a3bdff84bf1b5373cf3600e6293 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a0cb65278ba9409c848bcaa4494eb28d : SUCCESS in 4m 42s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ef61ccc4de0b4d81be357cea7f81008f : TIMED_OUT in 3h 04m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/89c88419bf304e1387f8fa55b0dc2e62 : TIMED_OUT in 3h 03m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"13c873d52be991a724e1a3b366735df74d4ca404","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-09 06:13:37.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"9e712a46c02e0d5e12dc4b36886f6507e82355a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 08:03:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/164beaa05d234f6a88c16e85e330e916\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e2bf5559a9c24edcbe0ff046e1b7f5f7 : SUCCESS in 22m 11s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/28fb3df19315418da21f63ad39cc8f65 : SUCCESS in 1h 08m 29s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1a81186683d44e43b464d4203243c32c : SUCCESS in 19m 30s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/480e5b563249473e9d7f460a97274120 : SUCCESS in 28m 11s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a88a1b6e303f4ca2a23276aedb7b4b97 : SUCCESS in 1h 06m 14s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1c1b93f2430343f8a2df758020e9f873 : SUCCESS in 1h 23m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecc5488281af401dac2c14dc343c1cfe : SUCCESS in 2m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f27580a616e4ca18512aacdb92003ed : SUCCESS in 3m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae18ab32a6fa40b998756b81b3cac5bc : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/174fc2c98c6e4d53ad0ea87f0602de6d : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1a338d92a67b4c9c96e15e09031c96c6 : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f637f8b5046443cb831d6c564917b136 : SUCCESS in 3m 58s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f6f313734a39486fa8b79b96ff6ef094 : SUCCESS in 25m 04s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/708f30d2167440b09ccff31cb03e2379 : SUCCESS in 31m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0c6b051324be40e482d6de643d31d39b : SUCCESS in 1h 05m 47s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/bf7a34cd82ce40eeb76bc6f48dac95ee : SUCCESS in 1h 40m 02s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ba542ce9f5904849be8db419e104c29d : SUCCESS in 29m 46s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/77d47f7e0e8b41d0bf422d14ad6432fb : SUCCESS in 1h 04m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2f060a5087af4be5bb18aa50b32603e1 : SUCCESS in 49m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ff12e0792aa5431b9f689e40f2a18eb1 : SUCCESS in 1h 19m 59s","accounts_in_message":[],"_revision_number":3},{"id":"001dbb6a4cfba20d44dbdbea9bac841b2698581a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-09 08:17:32.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"fbecb1ca33901f96eaa46ed4d473c9fbc05c2eef","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-09 08:23:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e04de00b7cb7bca3c72bb306ec2f545effc0f468","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-09 08:34:00.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"7e1a5e1675c793a9efbd25eca0d208a93fa9cae3","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-09 08:34:02.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"c72f79c15e22293235fc3f1b0ae3f92e0ae59e39","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-09 08:37:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"66a1226ec54cf30732e090001c48744a12149fe1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-09 08:45:58.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"1918528a9166fbb3a587d29149d2a86c3476ee05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-09 08:48:50.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"62431faf4aaab64638e13a284079af42819500c5","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-09 08:50:38.000000000","message":"Patch Set 5: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a0d6c4ba75076f790ab6eb0b3c8b0e16f4c65a34","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-09 09:13:57.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a68c8f53bffd8b9bc070b68583c80de934e0aaed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-09 09:42:48.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":6},{"id":"0fd6224f75302d6d8a60ae8186fa58dfc14410ab","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-09 09:44:48.000000000","message":"Patch Set 6: Backport-Candidate+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"201d8dac0fbdd338c3a5ea6c9c86f1e89d6eb72f","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-09 10:26:20.000000000","message":"Patch Set 6: Backport-Candidate+1 Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e5c226fcd3528dba437eb187f48fe5dad5d2b4d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 11:19:28.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f30bd5f3859a48879be2c58885e4dce1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/67be11bd85a54ab7a4e23b0961750213 : SUCCESS in 29m 47s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2c8277e7951944b097cc3f29ab4e4b2b : SUCCESS in 1h 12m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6ec6c4762632426d9837efa8aa58c081 : SUCCESS in 23m 09s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/72c27e0d667249b8bc763b63a5d1de9f : SUCCESS in 23m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/10e1165e42974b4ba51dd7290e7cac0e : SUCCESS in 1h 10m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fc92bfbc22be49229bd11c6230e7fc0b : SUCCESS in 1h 31m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d96a380033e4d0ea276fab9c4ed5aaa : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe31b2b0646c42218eb2e44dc61dc7a6 : SUCCESS in 2m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec60ceb611214b38911c6ba2da7501e5 : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/efb42f6e7d74404792fe6877de8b1cd3 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67572ae31c364f5eb16fa5e2ff291a56 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43ad752b0fa44ed8ad9601180c1460f4 : SUCCESS in 4m 33s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b3b9bcc8d06f47ca940c4e08fbda4fba : SUCCESS in 30m 57s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7bcc585d5eea4af4ae62a51da72dfc5e : SUCCESS in 35m 09s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3e645fbd712744b990cbdea3459b9ea0 : SUCCESS in 1h 11m 07s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3c35c42d2cd6498fb6107591916fc5eb : SUCCESS in 1h 27m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fc5884fb36d947afbc1e12f2fc2ec3d1 : SUCCESS in 22m 47s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/328a19d43e71467a8a0dfc6a26953d28 : SUCCESS in 1h 02m 26s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/55a768ca4caf4b51aba33bdf0377364a : SUCCESS in 48m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/193c3e01a1b64e1384d30b531a65e092 : SUCCESS in 1h 19m 51s","accounts_in_message":[],"_revision_number":6},{"id":"eb92816ff90c8081f451bc59a27e4d047ef469bd","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-08-09 11:32:02.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"320f92524af6388716cf305f4e2794b5252e3963","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 17:30:34.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37f1d774a738447386b78fdf0db62152\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6ae971ec0bbd437a8503e70c925b885e : TIMED_OUT in 3h 03m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1e3b392c5dc6475d93e15ba21ef0c610 : TIMED_OUT in 3h 03m 02s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/982fef4a485c4cf0ba47331b1fa61c89 : FAILURE in 2h 11m 37s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c0e622da99bf4dc093e0b223bc7eb1c8 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/57bc25066b6b4646bf9cd89cdfec707f : SUCCESS in 3m 56s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/9f9b99680abf4409942a9aa3a8531a7d : TIMED_OUT in 3h 03m 13s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/abc37870a1e24476a296ee737b569cbd : TIMED_OUT in 3h 02m 45s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f94566c376688885599293548b49914b5dd9a214","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-08-13 12:15:53.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6458c92dfdd0a9b7f5bad543069e05111cb840d3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-20 12:04:26.000000000","message":"Patch Set 6: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4b501737171408f48ec1b6c8d905edd1bc76aa18","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-08-20 12:08:06.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2f6d1657442bceab438e39729bf40b5ef2a8ccfe","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-20 12:13:01.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ea3f1b59ab19d013640be7e10ddbff1034b92e0a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-20 12:39:19.000000000","message":"Abandoned\n\nin favor of https://review.opendev.org/c/openstack/kolla-ansible/+/926632","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"6f9fd0a4de6d9ef09daa1d5fb233b24eaa3f589f","revisions":{"34732c63b7038f27b2883bb004029f81404aad47":{"kind":"REWORK","_number":1,"created":"2024-08-08 12:50:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/925979/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/925979/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/925979/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/925979/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/925979/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/925979/1"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-08 12:49:19.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-08 12:50:25.000000000","tz":120},"subject":"WIP: horizon: Copy wsgi.py to /etc/openstack-dashboard/","message":"WIP: horizon: Copy wsgi.py to /etc/openstack-dashboard/\n\nChange-Id: Ided63ff1462f7554302c8e2958ac46cc370c8455\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34732c63b7038f27b2883bb004029f81404aad47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34732c63b7038f27b2883bb004029f81404aad47"}]},"branch":"refs/heads/master"},"5dee3230b6ee49a62ec6bae071760068c7ff1385":{"kind":"REWORK","_number":2,"created":"2024-08-08 15:40:53.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/925979/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/925979/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/925979/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/925979/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/925979/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/925979/2"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-08 12:49:19.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-08 15:40:53.000000000","tz":0},"subject":"WIP: horizon: Copy wsgi.py to /etc/openstack-dashboard/","message":"WIP: horizon: Copy wsgi.py to /etc/openstack-dashboard/\n\nChange-Id: Ided63ff1462f7554302c8e2958ac46cc370c8455\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5dee3230b6ee49a62ec6bae071760068c7ff1385"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5dee3230b6ee49a62ec6bae071760068c7ff1385"}]},"branch":"refs/heads/master"},"1fc1e528c857cc2216f65c3ced57cfcd4f19f821":{"kind":"REWORK","_number":3,"created":"2024-08-09 06:13:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/925979/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/925979/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/925979/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/925979/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/925979/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/925979/3"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-08 12:49:19.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-09 06:13:37.000000000","tz":0},"subject":"WIP: horizon: Copy wsgi.py to /etc/openstack-dashboard/","message":"WIP: horizon: Copy wsgi.py to /etc/openstack-dashboard/\n\nChange-Id: Ided63ff1462f7554302c8e2958ac46cc370c8455\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1fc1e528c857cc2216f65c3ced57cfcd4f19f821"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1fc1e528c857cc2216f65c3ced57cfcd4f19f821"}]},"branch":"refs/heads/master"},"0f0b2768b9b3c1287286248829b07a77b6344abc":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-08-09 08:34:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/925979/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/925979/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/925979/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/925979/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/925979/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/925979/4"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-08 12:49:19.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-09 08:34:00.000000000","tz":0},"subject":"horizon: Copy wsgi.py to /etc/openstack-dashboard/","message":"horizon: Copy wsgi.py to /etc/openstack-dashboard/\n\nChange-Id: Ided63ff1462f7554302c8e2958ac46cc370c8455\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f0b2768b9b3c1287286248829b07a77b6344abc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f0b2768b9b3c1287286248829b07a77b6344abc"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e72402dd42f818a2fa00a16a677373c7cede0b50":{"kind":"REWORK","_number":5,"created":"2024-08-09 08:48:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/925979/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/925979/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/925979/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/925979/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/925979/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/925979/5"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-08 12:49:19.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-09 08:48:50.000000000","tz":0},"subject":"horizon: Copy wsgi.py to /etc/openstack-dashboard/","message":"horizon: Copy wsgi.py to /etc/openstack-dashboard/\n\nEvery time new Debian/Ubuntu/Rocky comes out - we need\nto do a dance with passing distro_python_version from kolla\nto kolla-ansible, updating kolla-ansible distro_python_version\nand then removing the additional variables in CI\nThis patch will allow smoother migrations to newer OSes with\nnewer Python versions.\n\nChange-Id: Ided63ff1462f7554302c8e2958ac46cc370c8455\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e72402dd42f818a2fa00a16a677373c7cede0b50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e72402dd42f818a2fa00a16a677373c7cede0b50"}]},"branch":"refs/heads/master"},"6f9fd0a4de6d9ef09daa1d5fb233b24eaa3f589f":{"kind":"REWORK","_number":6,"created":"2024-08-09 09:42:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/79/925979/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/925979/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/925979/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/925979/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/925979/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/925979/6"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-08 12:49:19.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-09 09:42:47.000000000","tz":0},"subject":"horizon: Copy wsgi.py to /etc/openstack-dashboard/","message":"horizon: Copy wsgi.py to /etc/openstack-dashboard/\n\nEvery time new Debian/Ubuntu/Rocky comes out - we need\nto do a dance with passing distro_python_version from kolla\nto kolla-ansible, updating kolla-ansible distro_python_version\nand then removing the additional variables in CI\nThis patch will allow smoother migrations to newer OSes with\nnewer Python versions.\n\nChange-Id: Ided63ff1462f7554302c8e2958ac46cc370c8455\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f9fd0a4de6d9ef09daa1d5fb233b24eaa3f589f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f9fd0a4de6d9ef09daa1d5fb233b24eaa3f589f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d2"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
