)]}'
{"id":"openstack%2Fkolla~926084","triplet_id":"openstack%2Fkolla~master~I6af9baf09a19a7dbcaf98585870aa44ff79398d3","project":"openstack/kolla","branch":"master","topic":"configure-apache-locale","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-08-21 13:34:34.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-08-21 15:10:03.000000000","reason":"Change was submitted"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-08-12 12:29:35.000000000","reason":"\u003cGERRIT_ACCOUNT_32553\u003e replied on the change","reason_account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-08-21 13:35:52.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I6af9baf09a19a7dbcaf98585870aa44ff79398d3","subject":"Configure Apache services to use correct locale","status":"MERGED","created":"2024-08-09 18:23:59.000000000","updated":"2024-08-21 15:11:18.000000000","submitted":"2024-08-21 15:10:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926084-configure-apache-locale","meta_rev_id":"e5927a726ea249fe5deaf2a814c400541d91e682","_number":926084,"virtual_id_number":926084,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-21 15:10:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-08-21 13:34:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-08-12 12:29:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2024-08-21 13:35:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2024-08-21 13:35:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-08-12 09:55:59.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-08-12 12:29:35.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2024-08-21 13:35:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-09 20:00:48.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-08-09 20:00:48.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-08-09 20:00:48.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-08-09 21:32:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"98ad8cea4de7ef5be2e8b37b3eb65becf34f7f73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-08-09 18:23:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d1c6f0a72aa78384b6f0b78d113d9b418d3f6f4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-08-09 18:30:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"088c382f7d1c397b48db860eadd221b75b79fa77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-08-09 18:36:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4bc55c9af1ea85c4259f60bf9b61b076b09348bf","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-08-09 20:00:48.000000000","message":"Patch Set 3: Backport-Candidate+1","accounts_in_message":[],"_revision_number":3},{"id":"65d9b16e7a17708db68b2e7b6cc35909776491fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 21:32:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/494e4801e20b418dbeb861c3b5b1eed3\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1bf1dee6ea824c5bb62c1105119a20f8 : SUCCESS in 28m 46s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2b7ed8b8d364486f822c4b5eb6c020f3 : SUCCESS in 1h 13m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7213071c700c4ad1ba9cedfc261e9d91 : SUCCESS in 19m 15s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c9ddadc4f43b4d3fabfa9d6311dded6b : SUCCESS in 24m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e43f469348b34dd2a9d2166dd1514b9a : SUCCESS in 57m 27s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e5cbda878b1c4baf9686760f829ae95d : SUCCESS in 1h 21m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73c8ec40ae26423e95d5db4f2de25863 : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c1ee96ee0c04ecfb44424a0696d73f8 : SUCCESS in 2m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7c18a0532a00459286e081faa8cb2aae : SUCCESS in 2m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/415e6ff6708f41bdb8da2150cff3241b : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c474c9d47ad74814b7d571b14e3ab0e2 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc09f2359982492c91b32a235fc3edb4 : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d54667ec3694c4e925e359e3dc11914 : SUCCESS in 3m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/dcea44d7cfdc4d4ba217d4f1f0120581 : SUCCESS in 26m 50s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/521a7484c5514d57beb933286bf927fe : SUCCESS in 37m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/23716c3b5bc74734a57028bb7d1bf2f6 : SUCCESS in 53m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b320420304a14f368994a0fdf29a718f : SUCCESS in 1h 34m 16s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/8c89b117f1354b06b2ddac787ec2b9ec : FAILURE in 1h 34m 40s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f0929318717741a497ba69b19f98d3fb : SUCCESS in 2h 32m 22s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/fca7be3c12e24ff49d58b1433f8bd059 : SUCCESS in 1h 20m 10s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/aaff3555a335426ebc7d5223916c7f39 : SUCCESS in 1h 11m 32s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/2e6fd70974e24e289d274f0c641cf682 : SUCCESS in 28m 08s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/e5b41637b9bf4280a9242d45069c784c : SUCCESS in 1h 04m 34s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/52d9cf3951414fcb9b73e74ed25beaf4 : SUCCESS in 1h 00m 13s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/29fe4ebe3399466ba16fd6165ee5a1f7 : SUCCESS in 1h 28m 46s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/2435a2d9d2e44087b749ebf03ed7f8f6 : SUCCESS in 35m 13s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a92cad77e819415eb51adf92d299de27 : SUCCESS in 28m 08s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/66bd390b2517410aae9130dd51cd384d : SUCCESS in 55m 01s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c854942c10444761998637557af3fe1e : SUCCESS in 53m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5df0238352394331bfc00ec724c14edd : SUCCESS in 1h 06m 19s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/b37bb1aba8a6445299f41d5c4106be7f : SUCCESS in 1h 21m 49s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/7d98e58404a8498e84c338c56fdef519 : SUCCESS in 2h 08m 15s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/24b4af4d9cb045beb10ca0030894d319 : SUCCESS in 1h 04m 46s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/2f459d2b45ea49789a8e22d079ff66d0 : SUCCESS in 1h 05m 40s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/9b005c6ff8584a89af3f0fd5f54a0fc4 : SUCCESS in 22m 32s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/4409c7337802494abe3228db59a2d0b3 : SUCCESS in 42m 36s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/2188f93ee50b4c3bbf66b0b3efa6c6fe : SUCCESS in 1h 02m 08s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/125ffc14abea4b1b91ae1372a4a12211 : SUCCESS in 1h 05m 30s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/6d21e73d318e4be1b4f6d5aaa2241e5f : SUCCESS in 24m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1f726f3124f4d87a024376de1926c870242c555b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 23:16:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4a04626a40148d09c1f96d163813d54\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1df19d37a062490f8e6a83e39c6c2e7b : TIMED_OUT in 3h 03m 51s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4d707e019a1d4cdca96486f1a21be851 : TIMED_OUT in 3h 02m 31s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bb344fc57754469e94f7d0846290addf : SUCCESS in 2h 38m 57s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/38e377e187f04c5abb92c61a9a87e381 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6938b733c1f44ee0a5a8e65afff80338 : SUCCESS in 4m 36s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/96cc0175c9bd489cb2f122d49484ac21 : TIMED_OUT in 3h 03m 29s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b71d8a43deb14a5fa83b56ea1765bcd9 : TIMED_OUT in 3h 02m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"45607817e26ebf74b89943cf01f6b35f2b1eb1bd","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-12 07:25:19.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d607958cf348d64609710d70393f817b9b323d2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-08-12 09:55:41.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"7b5040e3c6e9a13cc7ba1cf63f3f7ff4576f42d9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-08-12 09:55:59.000000000","message":"Patch Set 4: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2abadbb509716a59869026808d7785df57084a0e","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-12 12:29:35.000000000","message":"Patch Set 4: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"ec4e3d9ed069429af9ca58e65083194ff74dd174","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-12 12:40:50.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a35cd17908e4d5794de7aa4874cb7bc\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f3547051060d42569dfd8824baae4141 : SUCCESS in 34m 44s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3b54c91b00b64e03ac2152e1d0c20a52 : SUCCESS in 1h 03m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5e9266c1284f4f259114824e40464626 : SUCCESS in 26m 17s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0097053ca82c48a3be233bfdda68888d : SUCCESS in 28m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5bba22b0b5d74736a1e6b76e35630189 : SUCCESS in 1h 03m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/de1ac65df0ca4481ab9f7a8c25933723 : SUCCESS in 1h 33m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c25a3703b5d4b10a36a73d99606aa19 : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f700ef92320403e9fd350851f3bcbd1 : SUCCESS in 2m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/05a692d9ba0e498fa006f29550d371ed : SUCCESS in 2m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8cb7cd918b644bcbaece82303426ce09 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d69cc3c17e0b44ee99d0dfde4915c31e : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac553614e3534329a1f9d03314c89829 : SUCCESS in 3m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d5ac733025f74251b8296ec7028aa1a2 : SUCCESS in 3m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d7a548c8fcb04b8c8916407fab98d36b : SUCCESS in 26m 03s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/abf4e461c2b94b5fa90b16fe1e3b7c45 : SUCCESS in 33m 48s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c963bb9c69174b6fb5a9f37321117d6c : SUCCESS in 1h 00m 55s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9d125de4295f4201afe746a2be8952c0 : SUCCESS in 1h 37m 24s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/73e29a42eea142bf9b747948c2e34c01 : SUCCESS in 1h 55m 57s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/302c9c11bfb9444185598b64d6b1dc8a : SUCCESS in 2h 42m 39s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/ce976e68dd6041f58e6f918ecbfb91e6 : SUCCESS in 1h 19m 06s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/395c03825d8e4ed486126cadd5bfcb64 : SUCCESS in 1h 29m 20s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/00c801e72ef84cdcb4c6461f4a27a9e7 : SUCCESS in 31m 52s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/f1f867b9f4fc411eb033c489f13f9fc0 : SUCCESS in 1h 02m 12s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/f137066307034c21a538974fa42e491c : SUCCESS in 1h 05m 05s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/8ab18fb7729b4f97976ac871b5c1dbbd : SUCCESS in 1h 24m 18s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/81821f95455141bd8b4e5fa6d19f8d1e : SUCCESS in 34m 15s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2bbb50da7630434ca50b7827a2de61e3 : SUCCESS in 31m 09s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/142600d1c60340ec9a6c34970d61069e : SUCCESS in 59m 16s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4dde408fd49d4fe8aac2d958b5900e39 : SUCCESS in 51m 04s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4b51eb60d79141039a68f7cda765da0c : SUCCESS in 1h 26m 44s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/cdacd58a1d11492c9dc626aa84284fc3 : SUCCESS in 1h 35m 55s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/11658368417b4cdead15612ab493ae3a : SUCCESS in 2h 36m 09s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/6161192d555a4016a0d4fc6b939b9ae8 : SUCCESS in 55m 20s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/2ee84aebcc3c4c20900985a3df1bfcd0 : SUCCESS in 1h 06m 11s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/a44115b8ddcf44058668e2f90bf74711 : SUCCESS in 20m 14s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/7808ea0da78a4b4ea9efc43f047b2c72 : SUCCESS in 44m 10s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/98b5214688c14a29a7cc0fc9ae1f96c5 : SUCCESS in 52m 51s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/49eb1a6c7c4f4f67b52ea7c8fed9c61f : SUCCESS in 59m 50s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/75a17b8a436749c9991de24766446d95 : SUCCESS in 26m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"33baecf1de3402b12b079551dc0a0e7343d33ad9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-12 18:30:02.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f362ba949db4f11b40bcfa1158c3824\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1813f3af31734eba8bcdb28c4b244f45 : TIMED_OUT in 3h 04m 18s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d4ba26c4ec6e479ea34d1f7bcbe07d55 : TIMED_OUT in 3h 02m 37s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5de6109ac4dd4351837b119412dd627b : TIMED_OUT in 3h 03m 49s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c8fb322e9b6245beb07b68276b5fa75e : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7b28ec8534a241338477722f27844c59 : SUCCESS in 4m 42s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/dc8f4e271e784361b5e6bfaeae7d992b : TIMED_OUT in 3h 08m 48s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/799e4a22c5664e6ba45f70b2da7f2ce0 : TIMED_OUT in 3h 03m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bf0e9213a3817a9694fa141658a960fed8bfaa86","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-21 13:34:34.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"11dcabffa1d03fd8d41fdc74b39ef9f82267961f","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-08-21 13:35:52.000000000","message":"Patch Set 4: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"0369980eec8a817de65ee51de4585c1beb84e042","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 13:36:48.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"39afc75c69286d2cef55e61f009655b7c95c3d9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 15:10:03.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd6b55fb2adb470b9d9ae2793ee741ec\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/adf1143a6f2d4038b2b44c9b7fff5fbd : SUCCESS in 23m 09s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/54db908c4b6a48c59a9d6c4087fb9f66 : SUCCESS in 27m 21s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/cbb0c5949fd24ddd8ad56e5513f90970 : SUCCESS in 53m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/71549b7b0aa04df9aaa74045fec273fe : SUCCESS in 1h 26m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dd1558afa764eb2a3b550fd1f4e8434 : SUCCESS in 2m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e19a1102f77b4c6bb62ed38e3d044f4a : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/af440601ec5c4f2b85d04e9294d0a1cf : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/608250584bd6433a92666237eb4a30ce : SUCCESS in 4m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/143a86acd9d841d98d66f1224954a918 : SUCCESS in 3m 54s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/472cc5af743f410195bde6332e41c877 : SUCCESS in 27m 37s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d1849c4504a04df8bacb3bc53b083dd8 : SUCCESS in 36m 56s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/33766962fbc14483ab40704e2b4799b0 : SUCCESS in 1h 04m 03s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/138c7c9866434bff8746e019f22bbfce : SUCCESS in 1h 25m 16s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5fb9caece6fd4c73a5f3e151e61d6a18 : SUCCESS in 21m 53s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e31d960d8a4542dcaa803645450d1c31 : SUCCESS in 1h 10m 26s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6968e3d41c22460baf9c166626cd3cba : SUCCESS in 47m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f01cdfe6aa4e48b0b854675d5b0c0c97 : SUCCESS in 1h 16m 44s","accounts_in_message":[],"_revision_number":4},{"id":"a282104872ac77043e389936d531d7c433a76daa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 15:10:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"e5927a726ea249fe5deaf2a814c400541d91e682","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 15:11:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91df5477a8864b4fb28888a63c00ddff\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77bebc3142fa405c977050dcf37d5884 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a330016a1cc44e3f91436c587d6b26a7 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2caa64df67349b365b5ce94a0e9625d0b6ff87d1","revisions":{"65046c3057b563a1ebf75fa277b73c306cf07845":{"kind":"REWORK","_number":1,"created":"2024-08-09 18:23:59.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/84/926084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/926084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/926084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/926084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/926084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/926084/1"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-08-09 18:21:48.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-08-09 18:23:02.000000000","tz":180},"subject":"Configure Apache services to use correct locale","message":"Configure Apache services to use correct locale\n\nCloses-Bug: #2076453\nChange-Id: I6af9baf09a19a7dbcaf98585870aa44ff79398d3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65046c3057b563a1ebf75fa277b73c306cf07845"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65046c3057b563a1ebf75fa277b73c306cf07845"}]},"branch":"refs/heads/master"},"564c7a0e2a7243da9f9bd5e12ac608877de29623":{"kind":"REWORK","_number":2,"created":"2024-08-09 18:30:45.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/84/926084/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/926084/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/926084/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/926084/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/926084/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/926084/2"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-08-09 18:21:48.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-08-09 18:30:35.000000000","tz":180},"subject":"Configure Apache services to use correct locale","message":"Configure Apache services to use correct locale\n\nCloses-Bug: #2076453\nChange-Id: I6af9baf09a19a7dbcaf98585870aa44ff79398d3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/564c7a0e2a7243da9f9bd5e12ac608877de29623"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/564c7a0e2a7243da9f9bd5e12ac608877de29623"}]},"branch":"refs/heads/master"},"8a12a8efe510f1332b55f0ffc8412d858afd0a59":{"kind":"REWORK","_number":3,"created":"2024-08-09 18:36:45.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/84/926084/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/926084/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/926084/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/926084/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/926084/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/926084/3"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-08-09 18:21:48.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-08-09 18:36:37.000000000","tz":180},"subject":"Configure Apache services to use correct locale","message":"Configure Apache services to use correct locale\n\nCloses-Bug: #2076453\nChange-Id: I6af9baf09a19a7dbcaf98585870aa44ff79398d3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a12a8efe510f1332b55f0ffc8412d858afd0a59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8a12a8efe510f1332b55f0ffc8412d858afd0a59"}]},"branch":"refs/heads/master"},"2caa64df67349b365b5ce94a0e9625d0b6ff87d1":{"kind":"REWORK","_number":4,"created":"2024-08-12 09:55:41.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/84/926084/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/926084/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/926084/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/926084/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/926084/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/926084/4"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-08-09 18:21:48.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2024-08-12 09:55:35.000000000","tz":0},"subject":"Configure Apache services to use correct locale","message":"Configure Apache services to use correct locale\n\nCloses-Bug: #2076453\nChange-Id: I6af9baf09a19a7dbcaf98585870aa44ff79398d3\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2caa64df67349b365b5ce94a0e9625d0b6ff87d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2caa64df67349b365b5ce94a0e9625d0b6ff87d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
