)]}'
{"id":"openstack%2Fkolla~926145","triplet_id":"openstack%2Fkolla~master~Ie133d6efc2e4c00002b8f0724e0183eaab35d1c5","project":"openstack/kolla","branch":"master","topic":"slurp-rabbitmq","attention_set":{},"removed_from_attention_set":{"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2024-08-23 16:44:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie133d6efc2e4c00002b8f0724e0183eaab35d1c5","subject":"Add support for checking incoming RabbitMQ version","status":"MERGED","created":"2024-08-12 15:10:39.000000000","updated":"2024-08-28 11:05:50.000000000","submitted":"2024-08-23 16:44:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926145-slurp-rabbitmq","meta_rev_id":"225831b3bea8cb204afb50d373e7dfe5a1d8367f","_number":926145,"virtual_id_number":926145,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-23 16:44:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-08-22 15:28:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-08-13 12:01:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-08-23 14:14:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-08-23 14:14:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-08-12 15:29:42.000000000","_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-12 17:45:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-13 12:01:59.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-08-22 15:28:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-08-23 14:14:36.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"996b188813a2cc7611baf25b85dfaf7ab722a323","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-08-12 15:10:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c4de0d1787420cf9e67e03a97044a5dc1f0417f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-08-12 15:10:50.000000000","message":"Topic set to slurp-rabbitmq","accounts_in_message":[],"_revision_number":1},{"id":"6b9f573468faa1f926233ef994f90a2873fb935c","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-08-12 15:29:42.000000000","message":"Patch Set 1: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5e041c6f0abeb2ce4cc8a0105d01045569aa7944","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-12 17:45:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f0e3b14ad2c479fb109c86bbbb7fa00\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0aafba2712cd4beba3331913e91ab008 : SUCCESS in 31m 14s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ec3892db85044682a1e9c7be7add8df0 : SUCCESS in 1h 06m 24s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/934e93f867124f07a40aa4e2dc26d953 : SUCCESS in 18m 54s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/04842290b94f4b54b6db267bf47e9c02 : SUCCESS in 25m 38s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/03798500157e4decb9ae9da8d2daaeb0 : SUCCESS in 59m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6197439575154cd7b7fef8d089ab78d5 : SUCCESS in 1h 36m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8202a7494a74f02844963ee938a97ad : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0395e0782f56484ea715420411813a38 : SUCCESS in 2m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a00bc4f36df84b6b91fb69d88e7ad803 : SUCCESS in 2m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6295d1ed51eb4c0d9558c8561df93420 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e9ac544320b40a3aef4f25ec44fcc08 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/989548934d9a491bbe7f7d5c9722a360 : SUCCESS in 4m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e84c6a9b4c23435084bf3ade5d4c1709 : SUCCESS in 26m 24s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3367b0f511fc416bac0dfbc5e487ed56 : SUCCESS in 32m 26s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/908f7c01664246249dc8e1f34dc2d5e0 : SUCCESS in 1h 07m 10s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9628d3bb73e349ed8e829d6e17de7dfd : SUCCESS in 1h 42m 58s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/6a22fe707d544c3db6cd34255869b7a4 : SUCCESS in 2h 01m 28s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f16256239e3a4f539967897a75ca7ad0 : SUCCESS in 2h 28m 31s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0885cc3726284f3d989e3f9beaf33c25 : SUCCESS in 22m 40s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2fea1bc031b34d07ac14a8048fc8b7e5 : SUCCESS in 58m 08s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/983c05f6d96e40cab204bb80bd12956c : SUCCESS in 56m 20s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/01149edfa4dd4e4197ff57fac86de6b5 : SUCCESS in 1h 19m 56s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/75795bc46f004143a61f43fc476897d8 : SUCCESS in 1h 30m 34s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e295b5430103437d80c5f939971d9ad0 : SUCCESS in 2h 25m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"82888d01ee9d2ce462a42b6233fe9f1ed7ce179a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 00:55:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/214fa8b71cc045dfb5292db7ee37e2ad\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7bd207b4aa6e4426b8925c3bd2e1c130 : TIMED_OUT in 3h 03m 40s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6e104ddcf3f04803b8800ad068f96f38 : TIMED_OUT in 3h 03m 19s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/415d521f0fc743589615aa2a3b57dada : SUCCESS in 2h 36m 49s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4d198cc6735e43a4923320aa391b11fa : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b7b195986938461dbdb8dca5acaa6a6d : SUCCESS in 4m 48s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a6606e5956c94d0ba0e66aa6b75dad6c : TIMED_OUT in 3h 05m 26s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/fa521f7fbe4f4655b19093c53a03bcad : TIMED_OUT in 3h 03m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"088bcfe7141f7dec85cf60821d025466c862b8ba","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-08-13 12:01:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"703cd2c032e0301f25fa00f35459d08a34bbd480","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-22 15:28:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ad13d5b1b6737b850de2ef2118403a61e2551971","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-08-23 14:14:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c076498288ac4c73de6b78dc863603386dc1c982","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 14:15:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5cea65b839711018241fdbad306fa0ab0f75a456","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 16:44:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e02c35dc1dc400ab7faa229798d5f47\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/18abcdb3df9d4e009aa2dc49483dfc41 : SUCCESS in 21m 31s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0784a6e4d3db41c091bb6fa9099006ff : SUCCESS in 26m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/59656476fb27483b89a8096163623647 : SUCCESS in 1h 03m 44s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1fc34e60a0e94a40b06cf82814582dbe : SUCCESS in 1h 22m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c56761a4e5344c6bb6734b0c0ad2428e : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9784929c843245fe997db8df0b304168 : SUCCESS in 3m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/52bea6d9c1b8494c8ed5c2a89135bd36 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ede4c2bdf13d410db1eb61154ff91e23 : SUCCESS in 4m 54s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9252760d4f92419e9c08c196756fa25f : SUCCESS in 29m 03s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ed9bbca0f0be4172a4ae7ae17a0ca8f3 : SUCCESS in 35m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e01fb27d44f14ec9b19aa38aab4bad43 : SUCCESS in 58m 16s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6e195f35b6014356921c06a0666292ee : SUCCESS in 1h 16m 05s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8bfb0a3ab28c4dfd88fe66a74abe5e7f : SUCCESS in 25m 27s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5a7d80acdfaf4f139b4d54b60c1d3c19 : SUCCESS in 1h 02m 09s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/26c891cf7b1548319629eb5d445c392c : SUCCESS in 1h 00m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/09f37adebe38402dbe64cbe3b22cb289 : SUCCESS in 1h 13m 23s","accounts_in_message":[],"_revision_number":1},{"id":"03c2354deb20c7d83e0092a5fb0b6f8ccd3e7f79","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 16:44:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2134d1934e64cb323a8772a4299cad164d901aa1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 16:45:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/007f90e45aba4f56ab30379bab8edf42\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2ad61df09f344ecbf1114f21d10b7c0 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1},{"id":"225831b3bea8cb204afb50d373e7dfe5a1d8367f","tag":"autogenerated:gerrit:revert","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-08-28 11:05:50.000000000","message":"Created a revert of this change as Ic0f9521c66240f13ff5b7fc94480dbbb76498339","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"36a7cd053d5b13ecc9fad1866f244a733fdd58e5","revisions":{"36a7cd053d5b13ecc9fad1866f244a733fdd58e5":{"kind":"REWORK","_number":1,"created":"2024-08-12 15:10:39.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/45/926145/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/45/926145/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/926145/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/926145/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/45/926145/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/45/926145/1"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-08-12 15:08:43.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-08-12 15:08:43.000000000","tz":60},"subject":"Add support for checking incoming RabbitMQ version","message":"Add support for checking incoming RabbitMQ version\n\nWill be used to check if the current RabbitMQ version is high enough\nbefore upgrading to the new one.\n\nChange-Id: Ie133d6efc2e4c00002b8f0724e0183eaab35d1c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36a7cd053d5b13ecc9fad1866f244a733fdd58e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36a7cd053d5b13ecc9fad1866f244a733fdd58e5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
