)]}'
{"id":"openstack%2Fkolla~926528","triplet_id":"openstack%2Fkolla~master~I0de9928b2657ab5f31580325e866bff28f2b5510","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-09-06 10:01:49.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-09-06 10:53:54.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-09-06 12:24:52.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0de9928b2657ab5f31580325e866bff28f2b5510","subject":"Pin OpenSearch Dashboards to 2.15","status":"MERGED","created":"2024-08-19 11:06:39.000000000","updated":"2024-09-27 09:11:28.000000000","submitted":"2024-09-06 12:24:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926528","meta_rev_id":"c09a68914cedd482847c7b8d3cfc3bcdca7b4a0f","_number":926528,"virtual_id_number":926528,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-06 12:24:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-09-06 10:01:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-09-06 10:53:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-06 10:34:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-06 10:34:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-09-06 10:34:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-19 11:47:28.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2024-08-19 12:36:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-19 13:07:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-09-06 10:01:49.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-09-06 10:34:05.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"2398b2edff02b841b796daed69a12397bf3f015d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-08-19 11:06:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9509d107fb2f6ceb16451efa5ee3486f83e43451","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-19 11:47:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"780ce181995403f942d0388fa00d61a3880b92bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 12:36:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97a33b25ad65457f8fdc2108a238539f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/96f963642370440c9bbec44124cb884d : SUCCESS in 30m 26s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ed0ef1d60a9943e29b0450538121ccff : SUCCESS in 1h 03m 35s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ee67f5398b174e1692b361cd913e7bc8 : SUCCESS in 21m 31s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ed599ab7de374c16b04a45bedd0ee95b : SUCCESS in 26m 19s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f2b92c109ef646818ca891fdd5e454a1 : SUCCESS in 1h 00m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4ca6bf4603fb42f0aa6af7f3320b7ce3 : SUCCESS in 1h 29m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b21f4d476b446a39773a735929c7c5f : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce70c9be44c14337968bed791a0db83e : SUCCESS in 2m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9804dee952d04f9faaa86618f74b9ca9 : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d079174ae6414151b6e63332ffffde4f : SUCCESS in 3m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa85855be1ba44ee9f1a152b47271cb1 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcb8ca1be25a4f038204cd943c0c027c : SUCCESS in 4m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0b8f3d84d234517af8776ab5a0f473b : SUCCESS in 3m 29s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f54eca3eb0434c0f91f8b11b23e75b72 : SUCCESS in 27m 12s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/06252ecc5cdf49bba6abcd68e386227e : SUCCESS in 39m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9e030fede9444dc9acae3623e672ad44 : SUCCESS in 1h 06m 38s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d218500dc3d149ec9ea371ebe148cf46 : SUCCESS in 1h 28m 22s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e26bcaf01b4d441193da99658bc4d1db : SUCCESS in 36m 32s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9bb65f831f37494eaca6f0c29d95e950 : SUCCESS in 20m 31s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ca6079376bdb4fd18b4de55033849035 : SUCCESS in 1h 09m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6eddcc951fa84eb9a71f27950766d21c : SUCCESS in 46m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cbf9828ec703409498506cb1a541c938 : SUCCESS in 1h 11m 37s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/413b3133f9914c06b163c6737b045e17 : SUCCESS in 24m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6693dd825d7d230a73b5b5eb9edcb583f4d7c7be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 15:08:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da677093042d478e8974651cb6d830d7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c78a17f064bb4b83af44531fe72b1a7c : TIMED_OUT in 3h 05m 38s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e48e5845584a406497b934ca1f4d90c2 : TIMED_OUT in 3h 03m 04s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c108812029694cdcb0b1031b3da14475 : SUCCESS in 2h 24m 41s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f00547118319473c8e9170059e7575ac : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f00a0425d59345208b2373cbeea95fa3 : SUCCESS in 4m 23s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/aa92cdebd6dd4f6bb37fe89c77efb9dc : TIMED_OUT in 3h 05m 27s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6b142e92cace4cfc8af6b253f26f9504 : FAILURE in 2h 31m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e4a2d63ae7ecfce0926f43ae5956027664169d0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-08-20 13:23:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fac99d6b9b9ca7421dc50432af31d98abce33c7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-08-20 15:44:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"de4c8a591563ff214aaae3039f08bf5b3eda1958","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-08-20 15:44:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e715eb2e159eab3792cc64ccfd464cd77ee3404e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 17:17:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d405a819e7b4312b4ea0e1dd008ff94\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2646c8cf55424d22ada084e473109ffd : SUCCESS in 24m 37s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d31dd308e9ab4b16be249c9dd64c1507 : SUCCESS in 1h 05m 29s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4ba07ad5565e433eacfedb802a1682ae : FAILURE in 16m 43s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/eed36a4445644d7fa49d7c9944c7bc94 : FAILURE in 24m 08s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/921e206d56564d7ab5554b7b669447fb : SUCCESS in 53m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/110b183ac901418bb4bf1e66d970b502 : SUCCESS in 1h 17m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0911ccbc34b14987b161a853ee2da869 : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/558c9127154245bf8589db9b031debe6 : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6b1a1dedcda448018253a0ad502def5b : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e143c99b25334bbf91267378f8bb2a27 : SUCCESS in 2m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5e90a8027584648a7dbd2197c9e57c3 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/270599b18a0a442d8811c5e8bb35fc29 : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20b88baaf39f4d4d9c987cce5c0f6e59 : SUCCESS in 3m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e68fa2cd48f34f2280568a0cf58b8426 : SUCCESS in 28m 26s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/87caffe915b442fbb4c4827eec6ddc5b : SUCCESS in 37m 46s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6787039351bd4dc58d1d7d95b05ee864 : SUCCESS in 1h 10m 41s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b67f0fb43a4d4e89a13091a708e3b963 : SUCCESS in 1h 20m 26s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/faa33aaee6d94b2fa11fe06c9af3c086 : SUCCESS in 34m 39s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/20a97f2806994946901601ecce3a518a : FAILURE in 25m 09s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5b4c99d407b34f0e90482dbf68ee0830 : FAILURE in 52m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e4fc700137bd4af39ba7ea6aa403e510 : SUCCESS in 48m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0f939a3b54144a109fa825b1f67dc6ed : SUCCESS in 1h 25m 37s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/9e487d5c78a44e0690471a94d46d9c6b : FAILURE in 14m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f6e89acce4c80d1640c175113743667f773b4d00","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 21:19:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c09fa439ef544b01a4462f00a90b1694\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f8d7d3c8b5eb4493841d082d6e5263f0 : SUCCESS in 2h 41m 46s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/14920bb4d419415b86b4a490e34ddf6c : TIMED_OUT in 3h 02m 21s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/77a30a8eb6544a64b5e129136941a398 : SUCCESS in 2h 00m 09s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f444433fc75a40d697a88d0779c6163a : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5bc5169e7f5a42e5a8867b51f8384bd2 : SUCCESS in 4m 25s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/bb90dcb6654841a4a4d9f14516f7bea9 : SUCCESS in 3h 03m 04s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0bdb57db44ec47d38dafc60f0a5de05f : FAILURE in 2h 24m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e0e63eef1cab2b4b9d5577137a04ce4b3f413413","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-08-21 08:53:50.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"2852764cdfa9921708b31f315fcca1d89f0cd8de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 10:38:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33556608c75b4e40b63a60c7371e1735\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8bf0169f308342cb978b2c1ffe2314fa : SUCCESS in 25m 07s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6779b40e075840c1a642ec17ac51281e : SUCCESS in 1h 04m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ccc69d047bc247b69efc16d5de3a323e : SUCCESS in 18m 07s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b0db7a85b93f4d8e9b54768da6064c71 : SUCCESS in 26m 58s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/399e0b64e1f94cb886971fac6553c628 : SUCCESS in 55m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9617d5a496904f5faca6edddaf392d39 : SUCCESS in 1h 30m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bdc7f3517df40cebe6e9789c8ea24d7 : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85c3bebb09ae48f5b1631782ce6f305d : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/92ac670b83e546f6ae0226255c4472dd : SUCCESS in 2m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/88def99a063f46839b542c33b649d74f : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/db76e1dcfc5540afacd79dbd5e7f863a : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89706ed473414679a7b7c666a389c57a : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/feaf6eb0d9794ee59cbb729369a88c3a : SUCCESS in 4m 22s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/111868b6e4b14fa9949239643f1c9ef1 : SUCCESS in 30m 22s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e5a7b54cecdd4d97a7ebbca478e71a25 : SUCCESS in 37m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7f358e43c6c848bc8c5a9f571c658999 : SUCCESS in 59m 39s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ddc98fd5179d4a00b2af6ff69c0426dd : SUCCESS in 1h 37m 07s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c178e2546b5d469ebccf4a52054c60f0 : SUCCESS in 35m 13s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8be5e6b974674027a7077ea9d99bb14e : SUCCESS in 32m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0c0ccc7e0f2049b2803969cf4410c1cc : SUCCESS in 57m 20s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0d9ad7317fab46749a8ad602abf97b2f : SUCCESS in 51m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4ab208f5131a41b9a2e9900e25a28140 : SUCCESS in 1h 18m 50s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/59b2b1a34238466e8b6ad091d1c24923 : SUCCESS in 29m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b5ef2b6eea03858276ab5d1f28eed26fc180e5aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 12:35:13.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a622a2c5bd74761b8e9ba36f7fd7c45\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/5193cac73f59489982efdb4b9ccf129d : TIMED_OUT in 3h 03m 28s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6ee1e9182a1449608ccb78b1c4b14c82 : TIMED_OUT in 3h 04m 01s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/89cfbe4bfb15431083f2d310487334a9 : FAILURE in 2h 44m 11s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a2bf30e4b290494089d1941b999fe4f7 : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1149ef81b66a4105ac0ee8d6b569841d : SUCCESS in 7m 19s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/52085c72db22413e84c9f3e7f22c1b0e : TIMED_OUT in 3h 05m 08s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ce27629106c84610b2e8087d842b9d79 : TIMED_OUT in 3h 04m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1903e5e4d43365cf0f2fd580485c1b78a9474bd6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-06 10:01:49.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"510a5a199bbeb86b9065e880b6b73a0d9d83e2ed","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-09-06 10:34:05.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7a5cee0468f175e92d08d3f4b3376707c02739e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 10:34:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"39a3cbf8cbf781db08f2accb78609dd8a5ed1672","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-06 10:53:54.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"15c7dc55c8ea5c134347f6832a960b1603fcfb04","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 12:24:52.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b6f798a6d7442f490a758ee779d2923\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8e63c8d6f0bb4af79365dd1b668cb764 : SUCCESS in 17m 08s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6a26a29b45fe4ae0b94208f4b9336eda : SUCCESS in 26m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/63c601265c9b425a96187db3776693ba : SUCCESS in 56m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e65924ef27c44918b2038d27abd71c95 : SUCCESS in 1h 27m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cf32d3dbfed4375a872df5ddb06ee0b : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e604b5d34fd48e6b697f3c8b56e3bba : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e7bf4871b3824fc29dd7db8e89ac4f89 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8402958061e04b1aa19a9d89836ecb3b : SUCCESS in 6m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04370d8046b64a4f9045b280abc33461 : SUCCESS in 3m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/29f75d79cacd44ebbd5f4bef6541b6ef : SUCCESS in 29m 44s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d262615ab9bb44a3b3b2b74aa1b38d49 : SUCCESS in 33m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4cccfdf7694147ae9e3f4f4995eb2fb6 : SUCCESS in 1h 07m 29s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/41710cb730fc498e90ef2339fff6cb0e : SUCCESS in 1h 27m 37s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/40a27015d8b64ebba284ae3efe44c10b : SUCCESS in 18m 50s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f63b340583d545b4a6e403e06a36dab6 : SUCCESS in 55m 51s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4c70df45ff1e4bf4b6718e7e6c64a9a4 : SUCCESS in 54m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b4da848c06774d8dac171a93588c0c20 : SUCCESS in 1h 04m 10s","accounts_in_message":[],"_revision_number":3},{"id":"615ca995d710731e4db0e60edd4db12cd76a701e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 12:24:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"92ffbcfba2207a204ac19d4f95d2dacfbb344ae2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-06 12:25:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5533dc9e7ac485581a0df15dcbe2031\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c4831c4352a452eac445ccb2b9cdbd3 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f0b30e3ec394580b18d5678e7b09be9 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3},{"id":"c09a68914cedd482847c7b8d3cfc3bcdca7b4a0f","tag":"autogenerated:gerrit:revert","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-09-27 09:11:28.000000000","message":"Created a revert of this change as Icc810554dc877c9dedaf26609d59a93059a38ba7","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3c49991a86cb26dd0447fa748d1df9df5aed92bb","revisions":{"61890ae7ac2d1206d5a89904d8c7020da3e6980b":{"kind":"REWORK","_number":1,"created":"2024-08-19 11:06:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/926528/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/926528/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/926528/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/926528/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/926528/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/926528/1"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-08-19 10:29:06.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-08-19 11:06:01.000000000","tz":60},"subject":"Pin OpenSearch \u0026 Dashboards to 2.15","message":"Pin OpenSearch \u0026 Dashboards to 2.15\n\nOpenSearch Dashboards 2.16.0 discover page hangs after initial\ndeployment due to an upstream bug [1].\n\nThis change fixes the issue by pinning OpenSearch and OpenSearch\nDashboards to 2.15.\n\nNote that it is not strictly necessary to pin OpenSearch to fix this\nissue, but it seemed appropriate to keep them using the same versions.\n\n[1] https://github.com/opensearch-project/OpenSearch-Dashboards/issues/7626\n\nCloses-Bug: #2077319\nChange-Id: I0de9928b2657ab5f31580325e866bff28f2b5510\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61890ae7ac2d1206d5a89904d8c7020da3e6980b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61890ae7ac2d1206d5a89904d8c7020da3e6980b"}]},"branch":"refs/heads/master"},"eb630e6186c9bb6420127aeba892de7698bc14c7":{"kind":"REWORK","_number":2,"created":"2024-08-20 15:44:22.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/926528/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/926528/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/926528/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/926528/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/926528/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/926528/2"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-08-19 10:29:06.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-08-20 15:44:06.000000000","tz":60},"subject":"Pin OpenSearch Dashboards to 2.15","message":"Pin OpenSearch Dashboards to 2.15\n\nOpenSearch Dashboards 2.16.0 discover page hangs after initial\ndeployment due to an upstream bug [1].\n\nThis change fixes the issue by pinning OpenSearch Dashboards to 2.15.\n\nNote that it seemed safer not to pin OpenSearch to fix this issue, to\navoid breaking deployments that have already deployed 2.16.\n\n[1] https://github.com/opensearch-project/OpenSearch-Dashboards/issues/7626\n\nCloses-Bug: #2077319\nChange-Id: I0de9928b2657ab5f31580325e866bff28f2b5510\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb630e6186c9bb6420127aeba892de7698bc14c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb630e6186c9bb6420127aeba892de7698bc14c7"}]},"branch":"refs/heads/master"},"3c49991a86cb26dd0447fa748d1df9df5aed92bb":{"kind":"REWORK","_number":3,"created":"2024-08-21 08:53:50.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/28/926528/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/28/926528/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/926528/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/926528/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/28/926528/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/28/926528/3"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-08-19 10:29:06.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2024-08-21 08:53:28.000000000","tz":60},"subject":"Pin OpenSearch Dashboards to 2.15","message":"Pin OpenSearch Dashboards to 2.15\n\nOpenSearch Dashboards 2.16.0 discover page hangs after initial\ndeployment due to an upstream bug [1].\n\nThis change fixes the issue by pinning OpenSearch Dashboards to 2.15.\n\nNote that it seemed safer not to pin OpenSearch to fix this issue, to\navoid breaking deployments that have already deployed 2.16.\n\n[1] https://github.com/opensearch-project/OpenSearch-Dashboards/issues/7626\n\nCloses-Bug: #2077319\nChange-Id: I0de9928b2657ab5f31580325e866bff28f2b5510\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c49991a86cb26dd0447fa748d1df9df5aed92bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c49991a86cb26dd0447fa748d1df9df5aed92bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
