)]}'
{"id":"openstack%2Fkolla~926541","triplet_id":"openstack%2Fkolla~master~Ieb52291529eb0d94206b3f30195c4d2527c9dc52","project":"openstack/kolla","branch":"master","topic":"kolla-ubuntu-noble","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-08-23 13:57:10.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"Ieb52291529eb0d94206b3f30195c4d2527c9dc52","subject":"nova: Drop xen remnants","status":"MERGED","created":"2024-08-19 15:17:01.000000000","updated":"2024-08-24 14:42:25.000000000","submitted":"2024-08-23 16:59:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926541-kolla-ubuntu-noble","meta_rev_id":"c83132244a214dc01eedebfc527eeb853e8a6356","_number":926541,"virtual_id_number":926541,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-23 16:59:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-08-23 07:38:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2024-08-23 07:37:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-08-23 07:38:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":-1,"date":"2024-08-19 15:19:13.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-08-22 05:47:01.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-19 15:19:13.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-08-19 16:42:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-23 07:37:11.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-08-23 07:38:33.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"616065c4e697f062a81f4ad67c87ee1f6a9e5df7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-19 15:17:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"17caf8afa0c7d5c12b9a9e0553fe884537177603","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-19 15:19:13.000000000","message":"Patch Set 1: Backport-Candidate-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2f8b71741d3c72faa594df8153953d198e680c01","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-19 15:19:49.000000000","message":"Topic set to kolla-ubuntu-noble","accounts_in_message":[],"_revision_number":1},{"id":"4c2487e284e554d238686c8a1476f01656ce58ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 16:42:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/402ba58f9d1c49bcb9f0886995bad87c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/243ba36089054f49bc99b90d0aca761f : SUCCESS in 26m 08s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/09b1c6bed01e4a52948233bdb9aa6948 : SUCCESS in 1h 03m 09s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c3765f4f4b5844afa21e5309b78beb8e : SUCCESS in 20m 28s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/76c40daa618841f989f9ce03c208cdb0 : SUCCESS in 30m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a52894b8d0ac409f8b2c9d572e57a1ef : SUCCESS in 56m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1eaf37fe8a1244df851d70ba6056cee9 : SUCCESS in 1h 14m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3050c274b184af69177c1150af35278 : SUCCESS in 2m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd329920754f44aeb132974f972ea70a : SUCCESS in 2m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ef8391e260774ff990995daddcce462d : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f49e38773ea04e85b2a4a55fdbad9061 : SUCCESS in 2m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42ea4c4435c149fbaa6f8717204d2218 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f18fa815118f4485911cb1996dc66bbc : SUCCESS in 4m 03s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/03ac4cfc5af84b1c8dba6ac122b91fb7 : SUCCESS in 23m 49s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/86d5c9c490b343afa7161ef4ef1c605e : SUCCESS in 38m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/cd93c5847b36499695d04f54b2818e25 : SUCCESS in 1h 04m 17s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/96846e90c4064ab18af1b366ef4246ff : SUCCESS in 1h 18m 25s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/9347b3c930f349448203c898e450ae23 : SUCCESS in 55m 12s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d5f1d2dbed924ef5a595ef8cebedda4a : SUCCESS in 39m 50s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/52295ac7e73140da81f14cc055096115 : SUCCESS in 1h 01m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ba622ded68ad497aaf2f7d42210364e5 : SUCCESS in 57m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/faf6d7b6c68846c189b05119018874b2 : SUCCESS in 1h 12m 08s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/0cf6e8c566464791b92fc38b63edb4df : SUCCESS in 35m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b2743f1b16ecd2708d2caa9b7e5064b56c5aa8f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 19:27:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f69393bfdc72410b9ff83dc728e1fc67\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6910ce087fd545129b55b2946a704e75 : TIMED_OUT in 3h 10m 46s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/681dd6e40a72401aa747540396dfa649 : TIMED_OUT in 3h 03m 58s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e45ec44e050849b0b8c9aa52281b53c9 : SUCCESS in 2h 23m 22s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/57768af1ba874db2835fa64f576ece71 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/adaafb3c83f2499c889ad6d9c9f8983a : SUCCESS in 4m 30s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/2cc34a002aea4593bb86954745219bf8 : TIMED_OUT in 3h 05m 32s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/68e05a4a90df4b449269f7e357f4156e : TIMED_OUT in 3h 04m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"230ab6999b43c79aa92f8c0ee352351954f5e808","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-22 05:47:01.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"0befbbd048da648d1005c3b6c65c7ad2377b60e8","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-08-23 07:37:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5d98ed68de0371daa8f4c3a9d86e7115774ea0df","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-23 07:38:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a647df41dc9e40d3fc38ef1c766c26bb38e2ec51","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 07:38:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3341063440258970405f11d973c39aed748cffb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 09:37:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a55231f37c84d01a3c81ca4d0006037\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9139f965437a4f2383fa5454ecdd5f0f : SUCCESS in 19m 57s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/89d4e2065f0940c5a27c2d9d04c64ecb : SUCCESS in 28m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c367bcb1415645df85b6b6692c75fbcc : FAILURE in 6m 37s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4112b8033bbe4991aa145466c0f56fda : SUCCESS in 1h 25m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28f91b46cafe49688302926cbc1ec8af : SUCCESS in 2m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83243e765f80405ca7d4f30766cd43ff : SUCCESS in 2m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/219f4788fbfa4b9ea06007d128db4087 : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e0610939c6446338ca5e7a5889e6637 : SUCCESS in 4m 25s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e084bdb90efa4975b368dda440a69211 : SUCCESS in 28m 55s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/67016a4b9582403a838026c0ec0e85cf : SUCCESS in 37m 46s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c0c2e97eec444899a26bc24843e9fd19 : SUCCESS in 1h 04m 17s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/79a4f47ed0984d42af76521f449d1ecb : SUCCESS in 1h 53m 35s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ff0b15bd031245b89bcb1bba0dbdc076 : SUCCESS in 17m 30s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1831cfe46e444792807c664fd73077a7 : SUCCESS in 56m 29s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/82d2c3516d564d1b951db8f8242828a9 : SUCCESS in 47m 40s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8b9a9775c69c4ec0ab425ec50e76bb90 : SUCCESS in 1h 10m 24s","accounts_in_message":[],"_revision_number":1},{"id":"d3a4d7d65d6bfcdfd9ee213e24ac10d4d5269cba","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-08-23 09:44:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f2f2a91a2b49f0af5934c1a67398b6951c91773a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 11:14:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/baed0c956fd8425fae82a60a88670144\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6f7e3f1fd78e479681ecf9e258f7a0f0 : SUCCESS in 25m 44s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/54dcffb89c164644aa6d506bbe6999e2 : SUCCESS in 1h 06m 25s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a9b85260f6914387a2733b3d8cc46685 : SUCCESS in 19m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e719d8db4a174b18b977301b70a823a0 : SUCCESS in 28m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/97b16ab57bca4d2d8facb7c15687e25c : SUCCESS in 58m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/bed4e2bcdad044a4ba4efc6d576b37f8 : SUCCESS in 1h 24m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a36305606b844f1b234a746a040adb9 : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6724ed399d4d450fb3687ab1fc0ae0f7 : SUCCESS in 2m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff24365842074ca786df359792b9a785 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7cf83c50b8634e6fae0f66bf471a7431 : SUCCESS in 3m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87091b03a2834e08bc3a4585be0ecba0 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78a107536a524f7eb49d8c1caf18ec07 : SUCCESS in 4m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8cfa42e018624a4fb40e005b8b82fe8e : SUCCESS in 24m 19s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/fcbb89adaf63441098abed9bbf891eb5 : SUCCESS in 33m 28s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/508ed255171e4a29bb80c6d4ca7e84c5 : SUCCESS in 1h 04m 13s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/fe3ea5bd7956487095dc217ef63eae27 : SUCCESS in 1h 25m 13s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/aa83df494e9a4e7f99a6d15420d97a3a : SUCCESS in 51m 42s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f9f668d1c3824474b334f31ae290c246 : SUCCESS in 28m 34s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/bfb2f8294f524857b850bb657acf660e : SUCCESS in 52m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/52e676b051ac4cb99ba36199586ae673 : SUCCESS in 51m 54s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3fd39d242d594a8a81b1ee6eee4daae7 : SUCCESS in 1h 12m 49s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/e0a6d16c47564b888bffdfa431153607 : SUCCESS in 35m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa9260ce365171f9affcc1cee9d23e58a2832c54","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 11:14:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"bf9726ce4ba3c5ca6ac3e0efe3edc4074f0e4a45","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 12:48:38.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/58048a4417c44c5584cf8c886567ab0c\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c6c784b187f141bc98a0c6dc79d9906d : SUCCESS in 14m 37s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b7066cb56fbb49b19db56833feb82419 : SUCCESS in 26m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ae872b3d161a4b66b179c9b79ebb9b17 : SUCCESS in 56m 48s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3b2ba4b784794d08a86704a21bb78787 : FAILURE in 1h 18m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7757cecdcfc247d69cbdea0680fc1663 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7cabf780e65640eb8d3a8d477159440a : SUCCESS in 5m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0265f370f9cd431080ff8145afeb00f0 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2b3b55ed0454fb2a27797292670f370 : SUCCESS in 4m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f58f168e3b65431cadb992b7c0064338 : SUCCESS in 25m 12s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6449111f753b49dba1204ed0c92777e1 : SUCCESS in 31m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/11002af8d81445b4897ba21f04f2dd00 : SUCCESS in 1h 03m 45s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d9b0a8d800014cdc881b6fd095e7f098 : SUCCESS in 1h 28m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b4d8713a8af942a98c8b979a4d516cbb : SUCCESS in 15m 09s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/378547a4671e450e972b114f11ffd8d1 : SUCCESS in 53m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ae8ec6a8749b465585032fbbe3121733 : SUCCESS in 54m 54s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5cba3a794931439c87d59376570f1118 : SUCCESS in 1h 12m 49s","accounts_in_message":[],"_revision_number":1},{"id":"890e038ee6a0fa28e9c9e4acea851ceb91a211c1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-23 13:57:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3eff90dc31acb0b0234fb6d530d10a9218307c7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 15:33:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea1fcdf497b342b7ac61bea600ab2f42\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5db7e6ee9b6a4696be220d406391a5ea : SUCCESS in 27m 00s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/dfb22275af6c43a78b80c9ccfef89629 : SUCCESS in 59m 59s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fe6192f5b6cc470f94dd3490fc3a1cca : SUCCESS in 21m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c252716af06e4e56bd1874f505b6c610 : SUCCESS in 26m 13s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ed9e1ec0d7d64823924e17f32191fb8d : SUCCESS in 1h 10m 52s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/10ed5e8396a24df3ac00ee13d272027a : SUCCESS in 1h 24m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62a7ca10a4b74b6a93f4364b5db1c6a4 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bbded81263064fa18403bd25e121951c : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b9ddb27a06e04361aa549170bd406b1e : SUCCESS in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2be91ad812ff4c6085c96f95335ce921 : SUCCESS in 2m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6f0a1465bdf47b68da2e50869d4b45d : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e499dd36f3e44dcb1f329cf986a5378 : SUCCESS in 4m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b2f9c2805f6d43f8862a8b9f17f04038 : SUCCESS in 29m 12s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ef02a0cc89354e53b59014105a040e6d : SUCCESS in 32m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/54f7c915ceec4555815d561f7e70eabe : SUCCESS in 1h 08m 45s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0238ecd61e92442e9fb3bf0f33d7d379 : SUCCESS in 1h 23m 55s\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/8434a8db6ce742198c27102bada7dda1 : SUCCESS in 53m 33s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e2858d4bc5f648b499847a97cf92530d : SUCCESS in 15m 34s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/476bb80ce2e44ab7b542cfaefc803558 : SUCCESS in 59m 26s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/aa63a4de54e14054a38fbfe1bd8ec863 : SUCCESS in 59m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b4ca26d742534bfb978a4ad802dc0f33 : SUCCESS in 1h 18m 41s\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/de65b893a3684f37812a30e2671cd94d : FAILURE in 41m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a4a00c94c732de674597097fb6c4347dee6bc076","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 15:33:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7118c77171db94aa884b2f70f4dd848c28fa2e92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 16:59:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97e2ed99510c40e4a01bb70ca4bbb443\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4b4169f9e7c14726939ff48614190ae6 : SUCCESS in 14m 37s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8c48eec2420242bfacf6440f502164ca : SUCCESS in 25m 59s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a34983e237b44ce89791de7ba7f05fc5 : SUCCESS in 54m 33s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b8ed366fcd774633be66e9616353594c : SUCCESS in 1h 14m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c23a14472c946b88af7ebf49c193196 : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/375865461cf241c9b31ea6525c06aa8b : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8f5de7b2e4bf430998baaafe50c85c79 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af25282eafc440f08273cd85d82a75bc : SUCCESS in 4m 16s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1d3e65c9d73940d89c107bde7bcac4b2 : SUCCESS in 24m 32s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/cd3f3283f3e5476b8e00ac85c1b35a68 : SUCCESS in 31m 54s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/77b25baee336403f9e582a3c5643265a : SUCCESS in 1h 02m 40s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a9fbc500dba04822bd8602de29426055 : SUCCESS in 1h 15m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/85d132c3821f4d1484a6b3af3adf670f : SUCCESS in 19m 55s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a85282fde1bd4287af853dd3a55c7245 : SUCCESS in 48m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/75742393c79e458c9393d2867478e001 : SUCCESS in 58m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c5566f2a64694a529e4fa95475869baf : SUCCESS in 1h 06m 46s","accounts_in_message":[],"_revision_number":1},{"id":"9d37cd0fb848d314c8d685786fc69ca32e2f9863","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 16:59:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"55e4b646d1d74867e1a940584163dc9b02adccc3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 17:01:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40ec62a4ff3b46cdaf24ef0d801c1380\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcf6062b5a324f7cb6c379d29ee728f1 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1},{"id":"c83132244a214dc01eedebfc527eeb853e8a6356","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-24 14:42:25.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39ed1db298d14d3dad6bd3de9e25a6ff\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/58cbe7f3ba3542fdbf959b4bee8db6e9 : TIMED_OUT in 3h 05m 12s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9742beadf70a4d52be095413b7e7d9e3 : TIMED_OUT in 3h 02m 52s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/51961289005e495caa95e1df3f6cba67 : FAILURE in 2h 57m 09s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/26dbf951653f4c0b9e610b846ad406dd : SUCCESS in 9m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0209f1b7c5414f298578b31ceca9f37b : SUCCESS in 5m 00s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ad97de76356c42dface7b07a546f157b : TIMED_OUT in 3h 04m 59s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a2a04acceb8c4563804d871cdf87057f : TIMED_OUT in 3h 04m 42s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56de572119b37a64f42b5399f966ebc18f8be223","revisions":{"56de572119b37a64f42b5399f966ebc18f8be223":{"kind":"REWORK","_number":1,"created":"2024-08-19 15:17:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/41/926541/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/926541/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/926541/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/926541/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/926541/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/926541/1"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-19 15:14:17.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-19 15:16:52.000000000","tz":120},"subject":"nova: Drop xen remnants","message":"nova: Drop xen remnants\n\nXen support in Nova has been dropped cycles ago.\n\nChange-Id: Ieb52291529eb0d94206b3f30195c4d2527c9dc52\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56de572119b37a64f42b5399f966ebc18f8be223"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/56de572119b37a64f42b5399f966ebc18f8be223"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
