)]}'
{"id":"openstack%2Fkolla~926744","triplet_id":"openstack%2Fkolla~master~I891749598247a5e951ed4a873b16acfd0566f59c","project":"openstack/kolla","branch":"master","topic":"kolla-distro-python-version","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-09-19 11:10:26.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-09-19 11:52:54.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-09-23 10:01:57.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-09-23 10:01:57.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-09-23 10:01:57.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2024-09-23 10:01:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I891749598247a5e951ed4a873b16acfd0566f59c","subject":"Switch to Dynamic Python Version Detection","status":"MERGED","created":"2024-08-21 12:08:42.000000000","updated":"2024-09-23 10:02:59.000000000","submitted":"2024-09-23 10:01:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926744-kolla-distro-python-version","meta_rev_id":"71a87b43baa89b75aa236d0068fd88101cd4f10b","_number":926744,"virtual_id_number":926744,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-23 10:01:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-09-23 08:42:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-09-19 11:10:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-09-19 11:52:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-09-23 08:42:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2024-09-23 08:42:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-09-20 23:17:24.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-21 13:19:10.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-08-21 16:35:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-22 13:01:24.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-08-22 13:01:24.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"state":"REVIEWER"},{"updated":"2024-08-30 15:46:32.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-09-19 04:41:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-09-23 08:42:59.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"69b6e51df4d7de3493aecc008685f12b65d4dc9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-08-21 12:08:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"83ed3abeb73ae8bd38d10bb4953167dc53aa30d9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-08-21 12:18:38.000000000","message":"Topic set to kolla-distro-python-version","accounts_in_message":[],"_revision_number":1},{"id":"34b75e5748d08a7a2ca8e6d3b109b52ec725cdd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-08-21 12:55:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9de39d5897f420081bed063e1458b11da0976e34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-08-21 13:57:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6cb37db356aaa76ad027cc028ef6168cb856afce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 16:35:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83e7f7ec264c4741ac7385c996556d53\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d8c17121d99f4fb28188d17f79e59cfa : SUCCESS in 25m 13s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/edfdb431c49b4e7e971d1484696f1344 : SUCCESS in 1h 03m 57s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ca22a87c3ac141cf90ee55d4e83bb48d : SUCCESS in 20m 14s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f58f7f2fbe6b4892b8356dbd6ab99cf8 : SUCCESS in 29m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/87d68e2dadcf4c0a911bdcd797a277dd : SUCCESS in 1h 01m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9e6f4868cdf343d0999438f6ec718046 : SUCCESS in 1h 35m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/940f5464ab924174be1143f6811de81e : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e1fd7030e6194f3cbb5b800ebe2c1803 : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d40d80279a574a2494b8ab984e7cf9a0 : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/00bd195bfacb4ade8837a2d5a1e5bd8f : SUCCESS in 3m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9aabbc7a7914ee7892f020277c9e81e : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b76de8078a4846bca694f289ae221e12 : SUCCESS in 4m 56s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3dcdc53a662345358103af4a80014a8b : SUCCESS in 24m 38s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ce12ce2136844fd69096a3f43cd8cc85 : SUCCESS in 35m 53s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/09cab53963c146a694271a0c43a832e2 : SUCCESS in 58m 12s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7688b844aeac4e619cbc14bb0877c6ea : SUCCESS in 1h 19m 57s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/76ab8f33b59c40eb812570e3d1ae1fa2 : SUCCESS in 1h 51m 48s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a25d6e7e147b4ac18135bea2d87893d0 : SUCCESS in 2h 37m 09s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/fa07ac79d60242809020638c62d495e8 : SUCCESS in 1h 20m 37s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/be3c0ba019724445ab3b4e9126812c93 : SUCCESS in 1h 19m 52s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/f2bf121aeaf5466cba5fbf4c56589825 : SUCCESS in 25m 51s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/0f424b99f6aa466fa331ca5ac88af0db : SUCCESS in 1h 04m 39s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/af01aa1b5fba4bf2ab3ad0403ea72b81 : SUCCESS in 1h 04m 42s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/5f371819896f4bfdb86c4c536944197e : SUCCESS in 1h 20m 21s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/780763c229364b20a33e90d0fcbb0f7a : SUCCESS in 34m 10s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6b7ab7cf669d4c92a1808461558f1072 : SUCCESS in 21m 33s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5a112143bd0d417aa004fffc026d9558 : SUCCESS in 54m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/db4d9536a82d45c3bccb5c1eb8af3b39 : SUCCESS in 56m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3fec3e45f803435fac17c7f28926e696 : SUCCESS in 1h 16m 25s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/9ec0ac3409bf4268810725c03551b7ed : SUCCESS in 1h 36m 50s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/305511952a6b4fcfb07e5d3491cf88b2 : SUCCESS in 2h 14m 48s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/1ebb04b9f97942d3b240655bb07651c4 : SUCCESS in 1h 01m 20s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/cec0ff68adc94dfcad504e3e6fcc0a16 : SUCCESS in 1h 15m 05s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/ab9a1c067d0f4c8d99f0ec4af82b535c : SUCCESS in 21m 20s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/11810709a0f24134b0cd1400f17a6bf8 : SUCCESS in 48m 38s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/f5c352a53d7047bd9e4bec7929ecd87b : SUCCESS in 1h 01m 39s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/28ee573922f146a99c86a8d832c80a08 : SUCCESS in 1h 16m 02s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/89f55fea9cfd4eddbcfde00443ffa12f : SUCCESS in 33m 43s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"aef554e47e328a69c71eaf8115e9f3022bd061fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 08:48:24.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1e380aaa56b4de8a733e4d4a512a40d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/137406c95cc443f6804b1238c709a340 : TIMED_OUT in 3h 03m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ba32f3dc7f004c758d92f9b275bf5e1d : TIMED_OUT in 3h 03m 03s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/17c684a1b47048edb2f5ac8059731f62 : FAILURE in 2h 13m 51s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/65134145ee4c427591123eaa2a38c496 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5b1d944d65674c23958e1e9fc5050511 : SUCCESS in 4m 30s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/2adb70972f334174826673ca75614d38 : TIMED_OUT in 3h 04m 47s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/de8da57a4d2e47dba8b4dc53c4e0ae74 : TIMED_OUT in 3h 03m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c50e10c3aa0acefa2bb5e5dc10d4c221217fbd72","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-15 19:02:27.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"91922aca95d7876e269ef88584848ac33edc0f19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-18 14:38:35.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"c713701e26933b8e9f6a295b4d2dd972c1613b1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 18:36:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b11b6b0134cf410180a71a11b5b78159\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d8cfa2b390134247b2a64cd116debd9c : SUCCESS in 25m 51s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0eccbb5e426245489a4657629b424155 : SUCCESS in 1h 03m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9b89698f962d4db483a5f6904ff5dc93 : SUCCESS in 19m 21s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/83faccf384fa464fb1ae7d87c2ac2650 : SUCCESS in 30m 42s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/59210762d4514b20bab4779f866cae52 : SUCCESS in 1h 04m 14s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/28a967feb2ad459b8ed8cb1cadeac5e6 : SUCCESS in 1h 24m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bd7b99656bc4a4eb5e2e01cd4ceec61 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6312a460c79415da4250c2906a94cbf : SUCCESS in 2m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9eb39503ea4a4b37b0079e5594ecc18e : SUCCESS in 3m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/360896db9727452e8f5d1ebbf7c492ba : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/480f09e4f8d24692bc9c76a18d84795c : SUCCESS in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b1a1d7fbba74071a4da9fcd40bb275c : SUCCESS in 5m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ace1796857e84354b97490b66321f09b : SUCCESS in 30m 56s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/58765edbba2b428695d0f99214ad31a6 : SUCCESS in 39m 04s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/de9247daa8ba45309e4e02cf75588c2c : SUCCESS in 1h 10m 09s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4e24b819de174556a9c22969a4249240 : SUCCESS in 1h 26m 15s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/c1a03c905d96482684dcf138da751ed2 : SUCCESS in 1h 50m 41s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/40c4733c4c334be3b4d59e2ccc79323d : SUCCESS in 2h 31m 12s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/591c249c5a524279ad1311a869d092a6 : SUCCESS in 1h 05m 40s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/4059156c6f954961895229be31f93c67 : SUCCESS in 1h 24m 32s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/6defbd79c6dd4b4a9fb849796904553a : SUCCESS in 25m 03s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/32f3ef5224ea4d939d2b5cfbb4a57dfe : SUCCESS in 1h 01m 57s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/d3a6bcff38394fc88f6b9be97d7f1e57 : SUCCESS in 1h 11m 42s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/79f745832d2c48c0a911143cb958fbae : SUCCESS in 1h 22m 39s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/82a29c4fc135458396e3f4a1d5936e24 : SUCCESS in 35m 49s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cf76ae6960404d1a92b77bdfbc1d80fc : SUCCESS in 20m 10s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/fca50a3eb77c4d6ea5df300492ca0980 : SUCCESS in 1h 00m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/42709221a8324d9ea93506711514e9be : SUCCESS in 46m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/da0fdd74135d446fa4575d5a70445b55 : SUCCESS in 1h 19m 19s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/7f180f8d996e4167bf968ce0c588f8fd : SUCCESS in 1h 44m 30s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e8df469a5a454dc68030f0473fbf23b3 : FAILURE in 1h 06m 19s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/a41afd2cc014456d904c150eb1c50fdd : SUCCESS in 58m 36s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/8c3938c42cde421f8883dd0555c9b9f4 : SUCCESS in 1h 04m 00s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/162f5b92c1034e2e9c599720c4707a20 : SUCCESS in 20m 34s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/6eadd22283ea4e6dae6514ce88519797 : SUCCESS in 52m 33s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/6112fffff5b54fe98b8c05f4a41a5b3c : SUCCESS in 58m 48s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/9d7f486b58784057b82c8090b10963cb : SUCCESS in 1h 15m 52s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/699653061c964fb5ae402c7ed4cf9d4e : SUCCESS in 26m 16s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f4918949bd69c99ad2f24d1067912c4baeb479e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 00:52:34.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b04121578a44fda830a4d726a61e6a0\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a5c3275968a342658926810351127b5f : TIMED_OUT in 3h 03m 25s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f3242adfff4a4b89a16f10ef81e790a7 : SUCCESS in 2h 19m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8ff544b6fd5f4bc3bad0e7b2fd5fbb05 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fb07df4eaba74e4789447ae0ef321ff8 : SUCCESS in 5m 33s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"aeb4cd7064ce8fd6648131ef12bf2e3d93d507f1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-19 04:41:14.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b27610f418461a8b043abe4471b93df8a0078d2a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-19 07:50:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1a3b99943346ce9bcd7da1d18fa3f0cf57f3afa1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-19 09:46:15.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"3d0981297543e701f347508d1511b640ee23b114","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-19 09:53:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7030517516548af88a61821fd7fd67383538e4ce","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-19 10:20:33.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5ae018c99dbbdc5173465779fe751ed50eff8c66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-19 10:22:23.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":6},{"id":"dfd68948b621ef2585c9047a9f34f6f7614f6fa4","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-19 10:22:32.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7856a0b83114e057a0f056b8e686637733dc3d64","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-19 10:22:47.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"63ac5058008fc539ce5474a9e919dd78f1d6e1f6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-19 11:10:26.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"722b0ee0100b2c29a556db7ebf678709b8c9c944","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-19 11:52:54.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"80a61b64f09a2de7a4e90b5d2bf626e238fcb9ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 13:12:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8fa7a784a1943e9b537ae34f7cf7c2b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2bf9dd12bc0a4009a65d8eecea8a1e98 : SUCCESS in 27m 36s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5e2ab62776684f8b9697a703bb3d5ff2 : SUCCESS in 1h 06m 06s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ab70c6f2292a4e38a57981b05a3cfbbb : SUCCESS in 20m 57s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/765cebdd45f24543af5a457ac9ce4142 : SUCCESS in 31m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/10ae176796a547079baa1fda7c9d7850 : SUCCESS in 50m 45s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/44a6634d8e9e47a99b586d5826ce9710 : SUCCESS in 1h 36m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be48331b3ad54f349e9d55fe06044e53 : SUCCESS in 2m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/948a6932b10445a4acec172a0470abdf : SUCCESS in 2m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8fee9095abf746b2953ad3507b687b07 : SUCCESS in 5m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d429c66664a047b2be9a5819a8e98d38 : SUCCESS in 3m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6560d33dd1740628cc4138b3066d9db : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa796949e0294059b00b2ee3120189ec : SUCCESS in 3m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48bb0d8c81c64b35ba6f9676e9f32590 : SUCCESS in 2m 56s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3ab540ceb6934646bf1189a25dea7bd2 : SUCCESS in 29m 45s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6051fc7fd6fe432181ac670c57b93656 : SUCCESS in 35m 24s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e29976bcc1014ec69558dace8318c146 : SUCCESS in 1h 07m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f19a086fbaf842e38b86d03e517fa1e1 : SUCCESS in 1h 16m 44s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/34f4bf49a297493c885ac06ef19be856 : FAILURE in 1h 25m 31s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/03f58693ce9a4483a9f0ecb0756e95e2 : SUCCESS in 2h 43m 43s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/1a5adbc206b14c6393b0a9eafd4e0673 : SUCCESS in 1h 08m 02s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/e3d0374d37244f5497e542cd158f57d9 : FAILURE in 8m 56s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/7229bc088ea442cdb2b89aefd08567f6 : SUCCESS in 28m 11s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/6f1ca6b454b94b61a69864184357b12d : SUCCESS in 56m 06s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/20ac091e24b0409e9ea158d2572f27d0 : SUCCESS in 1h 10m 38s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/ab3d42d21e88496986e9825c13993a98 : SUCCESS in 1h 18m 37s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/bef2cf81920f49c58a2045b0b9aca944 : SUCCESS in 33m 08s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3a0e038c398c414c94fe31d52e32e771 : SUCCESS in 16m 58s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/44eb4e826bf04adb82be1536c357ba58 : SUCCESS in 48m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/130a32fcbfbf41abb8223417684912c0 : SUCCESS in 54m 20s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/71b1ce5f33614fc4a19a721ffcb1fa6d : SUCCESS in 1h 07m 38s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/366fbdc0fc6d4dd89a58e2fea7310d14 : SUCCESS in 1h 27m 34s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a751982b04fa492a89e6929b6460c77e : SUCCESS in 2h 18m 53s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/68973f3fa3d94e20acc262b9dfda91dc : SUCCESS in 1h 02m 30s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/0c370447f3a7408688d9ff78ec362c46 : FAILURE in 1h 20m 29s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/c048311f65034492b67ba70f0b9f771c : SUCCESS in 26m 05s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/968e2e86ce0747948d90b1a83ea3c436 : SUCCESS in 42m 44s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/77ede62aebff46039158d29dc7a452ca : SUCCESS in 58m 39s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/ea74b2bc60004386b5971db608c62482 : SUCCESS in 1h 02m 06s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/80499ac1196f4be1a6425429439b34a1 : SUCCESS in 27m 53s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f96886c39b103615933568bc6b4350822faaedbd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 15:50:12.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e53c176f2ca46fa9e9b5a6751322037\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0877c4c7e2864710bf597008515c4949 : TIMED_OUT in 3h 03m 50s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/69d3e7b1270c4c2086fbb074a86da824 : SUCCESS in 2h 43m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f40423378fe94940be6b612d75db07d5 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/32306250e44d41bca1c24baddf39bb76 : SUCCESS in 4m 52s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"706f415b8e38b70fb7816be548bdbf6a31aa0d27","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-20 23:17:24.000000000","message":"Patch Set 6: Review-Priority+1","accounts_in_message":[],"_revision_number":6},{"id":"d3d8cfbbf32c2c3cfcf99be70ec5dd3fbec6b374","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-09-23 08:42:59.000000000","message":"Patch Set 6: Backport-Candidate-2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"dc3b5518f3c176bc0b43af66e5e30bb58cd8535b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 08:43:34.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"b6fe5176dbee33e2881e1faec9395d6ff62a47f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 10:01:57.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5da0d0e6769a474b9fb852d082756b24\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c7183b57881547f8996d8aafbb0bdd02 : SUCCESS in 17m 17s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ced55a79443f4688b81ad67e7fc94f09 : SUCCESS in 25m 21s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/237cd656375e40a285fca5b7075ed166 : SUCCESS in 1h 11m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/da049ff71386472e852f28a49e7abad3 : SUCCESS in 1h 12m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e151234998344f3ade1084fb39b2cf9 : SUCCESS in 2m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3aac3359c37b4beea8fdacfa59cf5e8a : SUCCESS in 2m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1cfa9d21ff04e5f8bf99eee3232fb22 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21f20cbda91a48eeae00686084663edd : SUCCESS in 3m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9d5baa2e63b4c20b59e848b1b5e4a57 : SUCCESS in 3m 36s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/71d36ff4c72b488c9814dfa0f17a9ce9 : SUCCESS in 26m 36s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b83bef8dfddf4c51bf4cd791a12707ff : SUCCESS in 31m 53s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/468a0fbeed664ab7b902a1393b11262d : SUCCESS in 1h 06m 45s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/05dd6b8f2ef14f3390dd7a382b50ed52 : SUCCESS in 1h 10m 06s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/54d64888808e43089d95a4eb8c3ea780 : SUCCESS in 26m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3d272b09ed964f43aa3d52f042f22151 : SUCCESS in 52m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/35e18a7fa6684e0f87a18bc7b8cfdaee : SUCCESS in 49m 43s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/23f03eb7d96a4856814116392d3ce474 : SUCCESS in 1h 10m 35s","accounts_in_message":[],"_revision_number":6},{"id":"fc7358c6503e7d22e8db1cdfb880628c3e14f336","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 10:01:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"71a87b43baa89b75aa236d0068fd88101cd4f10b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 10:02:59.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/770292319a3b45ba9b3c8a134ad6fdc0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d9426e0ba8a40baa794d30857ad6c12 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d267356fdb2043bb9cdd9e6f1ac6ab9f : SUCCESS in 44s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"1cda1acc7fd22bc42917c723f14c6ee076098297","revisions":{"5989eb7d4dac4842c7dc12ed048b26376631b3b9":{"kind":"REWORK","_number":1,"created":"2024-08-21 12:08:42.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/44/926744/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/926744/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/926744/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/926744/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/926744/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/926744/1"}}},"commit":{"parents":[{"commit":"44acae38943185a6bcc6abfad24ba87839fce28a","subject":"Fix build of prometheus-ovn-exporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/44acae38943185a6bcc6abfad24ba87839fce28a"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-08-21 12:08:34.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-08-21 12:08:34.000000000","tz":120},"subject":"Remove KOLLA_DISTRO_PYTHON_VERSION","message":"Remove KOLLA_DISTRO_PYTHON_VERSION\n\nChange-Id: I891749598247a5e951ed4a873b16acfd0566f59c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5989eb7d4dac4842c7dc12ed048b26376631b3b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5989eb7d4dac4842c7dc12ed048b26376631b3b9"}]},"branch":"refs/heads/master"},"5505dfa5416bdc48bc2254c020593300fe2ac485":{"kind":"REWORK","_number":2,"created":"2024-08-21 12:55:24.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/44/926744/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/926744/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/926744/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/926744/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/926744/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/926744/2"}}},"commit":{"parents":[{"commit":"44acae38943185a6bcc6abfad24ba87839fce28a","subject":"Fix build of prometheus-ovn-exporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/44acae38943185a6bcc6abfad24ba87839fce28a"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-08-21 12:08:34.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-08-21 12:55:16.000000000","tz":120},"subject":"Remove KOLLA_DISTRO_PYTHON_VERSION","message":"Remove KOLLA_DISTRO_PYTHON_VERSION\n\nChange-Id: I891749598247a5e951ed4a873b16acfd0566f59c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5505dfa5416bdc48bc2254c020593300fe2ac485"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5505dfa5416bdc48bc2254c020593300fe2ac485"}]},"branch":"refs/heads/master"},"1260898d521c1694f738a9d5826c5a8c3aaaca3c":{"kind":"REWORK","_number":3,"created":"2024-08-21 13:57:46.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/44/926744/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/926744/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/926744/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/926744/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/926744/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/926744/3"}}},"commit":{"parents":[{"commit":"44acae38943185a6bcc6abfad24ba87839fce28a","subject":"Fix build of prometheus-ovn-exporter","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/44acae38943185a6bcc6abfad24ba87839fce28a"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-08-21 12:08:34.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-08-21 13:57:36.000000000","tz":120},"subject":"Remove KOLLA_DISTRO_PYTHON_VERSION","message":"Remove KOLLA_DISTRO_PYTHON_VERSION\n\nChange-Id: I891749598247a5e951ed4a873b16acfd0566f59c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1260898d521c1694f738a9d5826c5a8c3aaaca3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1260898d521c1694f738a9d5826c5a8c3aaaca3c"}]},"branch":"refs/heads/master"},"a53bcd7811303de3ff2d6afebf0b8e82ecf42670":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2024-09-18 14:38:35.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/44/926744/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/926744/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/926744/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/926744/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/926744/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/926744/4"}}},"commit":{"parents":[{"commit":"015fd8d42a42a49243970c703fd513d3d5b2ce74","subject":"Merge \"Fix cinder-backup: add missing sysfsutils,lsscsi package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/015fd8d42a42a49243970c703fd513d3d5b2ce74"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-08-21 12:08:34.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-18 14:24:54.000000000","tz":120},"subject":"Switch to Dynamic Python Version Detection","message":"Switch to Dynamic Python Version Detection\n\nThis commit removes the hardcoded KOLLA_DISTRO_PYTHON_VERSION\nand replaces it with dynamic Python version detection during\ncontainer images build. Dockerfiles and scripts now use general\npython3 references, simplifying maintenance and ensuring\ncompatibility with future Python versions.\n\nTest cases relying on distro_python_version were also cleaned up.\n\nChange-Id: I891749598247a5e951ed4a873b16acfd0566f59c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a53bcd7811303de3ff2d6afebf0b8e82ecf42670"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a53bcd7811303de3ff2d6afebf0b8e82ecf42670"}]},"branch":"refs/heads/master"},"71b82cecce2f4dc62da376cb03090d89cd0031fc":{"kind":"REWORK","_number":5,"created":"2024-09-19 09:46:15.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/44/926744/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/926744/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/926744/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/926744/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/926744/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/926744/5"}}},"commit":{"parents":[{"commit":"015fd8d42a42a49243970c703fd513d3d5b2ce74","subject":"Merge \"Fix cinder-backup: add missing sysfsutils,lsscsi package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/015fd8d42a42a49243970c703fd513d3d5b2ce74"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-08-21 12:08:34.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-19 09:46:04.000000000","tz":120},"subject":"Switch to Dynamic Python Version Detection","message":"Switch to Dynamic Python Version Detection\n\nThis commit removes the hardcoded KOLLA_DISTRO_PYTHON_VERSION\nand replaces it with dynamic Python version detection during\ncontainer images build. Dockerfiles and scripts now use general\npython3 references, simplifying maintenance and ensuring\ncompatibility with future Python versions.\n\nTest cases relying on distro_python_version were also cleaned up.\n\nChange-Id: I891749598247a5e951ed4a873b16acfd0566f59c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/71b82cecce2f4dc62da376cb03090d89cd0031fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/71b82cecce2f4dc62da376cb03090d89cd0031fc"}]},"branch":"refs/heads/master"},"1cda1acc7fd22bc42917c723f14c6ee076098297":{"kind":"REWORK","_number":6,"created":"2024-09-19 10:22:23.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/44/926744/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/926744/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/926744/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/926744/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/926744/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/926744/6"}}},"commit":{"parents":[{"commit":"015fd8d42a42a49243970c703fd513d3d5b2ce74","subject":"Merge \"Fix cinder-backup: add missing sysfsutils,lsscsi package\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/015fd8d42a42a49243970c703fd513d3d5b2ce74"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-08-21 12:08:34.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-19 10:22:16.000000000","tz":120},"subject":"Switch to Dynamic Python Version Detection","message":"Switch to Dynamic Python Version Detection\n\nThis commit removes the hardcoded KOLLA_DISTRO_PYTHON_VERSION\nand replaces it with dynamic Python version detection during\ncontainer images build. Dockerfiles and scripts now use general\npython3 references, simplifying maintenance and ensuring\ncompatibility with future Python versions.\n\nTest cases relying on distro_python_version were also cleaned up.\n\nChange-Id: I891749598247a5e951ed4a873b16acfd0566f59c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1cda1acc7fd22bc42917c723f14c6ee076098297"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1cda1acc7fd22bc42917c723f14c6ee076098297"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
