)]}'
{"id":"openstack%2Fkolla~926882","triplet_id":"openstack%2Fkolla~master~I5120f1874f25a9ca8326e6db8b12dee4c81feb56","project":"openstack/kolla","branch":"master","topic":"radek/copy-certs","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-12-02 13:49:09.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-12-02 15:27:11.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-12-02 08:09:38.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-12-02 17:46:03.000000000","reason":"Change was submitted"},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2024-12-02 17:46:03.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-12-02 17:46:03.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-12-02 17:46:03.000000000","reason":"Change was submitted"},"36702":{"account":{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"},"last_update":"2024-12-02 17:46:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5120f1874f25a9ca8326e6db8b12dee4c81feb56","subject":"Change copy-cacerts behaviour","status":"MERGED","created":"2024-08-22 08:31:33.000000000","updated":"2024-12-02 17:47:05.000000000","submitted":"2024-12-02 17:46:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":21,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926882-radek/copy-certs","meta_rev_id":"b257ed648f4d3c6f2a95a72688e222b01e4dfebf","_number":926882,"virtual_id_number":926882,"owner":{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-02 17:46:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-12-02 13:49:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-12-02 15:27:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-12-02 08:09:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-12-02 07:16:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2024-12-02 15:27:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2024-12-02 13:49:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-12-02 09:01:40.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-22 11:05:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-17 08:46:57.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-12-01 17:48:13.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2024-12-02 07:16:02.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2024-12-02 09:01:29.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-12-02 09:01:29.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-12-02 09:01:29.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-12-02 09:09:29.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-12-02 09:09:29.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"53b8de21875591389ae8f8c99bc303772ac38060","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"},"date":"2024-08-22 08:31:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ac81ac7b7aea05308b3d3c9d46276fda55eb8a8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 11:05:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a621759cbb54ec09321c2dd9c05c803\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/cbf87013fc59404aa582e4b6eb5319c6 : SUCCESS in 37m 14s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e4eafcdcbbaf40df971f98d82653d701 : FAILURE in 29m 25s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7b2dfabe144640c2bd11c1aa49f605ab : SUCCESS in 17m 08s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/554993e94bdb4d8c981fd7fd06ffa001 : SUCCESS in 23m 42s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3893d1f10e094e9cbe88d985d040e96d : FAILURE in 31m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f98f450f90f447da832dbf744a732eab : FAILURE in 1h 12m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69bea1a0359640aaa00a1004b83e7aa0 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aeb76c2ae5144117b9f8750e92a0522d : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ed43fcef87cb4159944e635725b5ace7 : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bacd4017c8314a2a8b68b83862bb906b : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c5e7170352341e3b1ed1ceb8e24d75a : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e74e0d3f95b946acb597959a35e33629 : SUCCESS in 4m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9f38aa9dcc6b4043aeee0c5b75576a67 : SUCCESS in 27m 49s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/8171481436564e8b83a1e9f485f9b97f : SUCCESS in 36m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/57590310e4114aada5ac3f06498f06fb : FAILURE in 29m 40s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ae57fc1b19c4446c963e116fb9e69439 : FAILURE in 1h 17m 36s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/8b9d02ff60fc4802ae5295c28586068a : SUCCESS in 1h 40m 31s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ab95bfb453294958b985cb6f48ffdb6c : SUCCESS in 2h 24m 49s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/1a1e6cdf5f974b7d9b6a1454313f63da : SUCCESS in 1h 20m 14s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/23c825a04c1340ebab90dca4b8fcc2ca : SUCCESS in 1h 19m 24s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/d014f48994e94a078fc76b7e8f68a8a4 : SUCCESS in 24m 37s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/38aa9f79195e4c18b07ae8f1bfe51895 : SUCCESS in 55m 24s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/4d4c7a4413584f7ab14a615638d7fcc2 : SUCCESS in 1h 15m 12s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/b6d957c8a3d7499fa03e6859f4994948 : SUCCESS in 1h 21m 50s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/2f6145a255f34a11b048226d221e3b3e : SUCCESS in 31m 10s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8c7adeb321444bbe86d6ebf2c1a0b311 : SUCCESS in 30m 38s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/12cd87eff1664c45998567ce54bedbac : SUCCESS in 53m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7016359686df4cbcbd157c6bac5f6158 : FAILURE in 22m 36s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d28532b8f41b4aabae832e159ae58758 : FAILURE in 1h 11m 34s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/79f764e4de184bf09222236bf3d3241b : SUCCESS in 1h 23m 37s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c821f45940eb4decba1c9c837a25937f : SUCCESS in 2h 24m 51s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/f241853a340b49d1bd562cd0df123d5e : SUCCESS in 1h 12m 44s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/a4738be4fffa42d0abcd7729014ba38c : SUCCESS in 1h 17m 00s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/32b93cb6e8dc4da9889d7b5d7272378f : SUCCESS in 19m 40s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/eacafb8609564292837178b146616ddf : SUCCESS in 49m 37s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/2c3e80a0ce5a4e4abcc3dc899b2ff55c : SUCCESS in 1h 10m 01s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/d27dcd7d1ddb49328d8285b1533566c3 : SUCCESS in 1h 10m 06s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/75aad33add52437ab17dafa78766530d : SUCCESS in 23m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b3d6e90882654603150e22dadecbce8e7375f79b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 02:42:20.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17707609e5654dd297f27292ec5a0003\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/389d46d8ad0f4729b84956131ce6e165 : TIMED_OUT in 3h 05m 12s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/242685dccb164897bf992e0eba7d57d0 : TIMED_OUT in 3h 03m 45s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bf71ef9de9a14be9a2bf07edd820dd9c : FAILURE in 1h 46m 04s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cfcee187b6d04922b1ee8a6a745d3be4 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4783554c181347a5a77b5aa0c5f1da2b : SUCCESS in 5m 16s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8ff7e59ba80048ff8912b06d7e7e086d : TIMED_OUT in 3h 05m 49s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/38210dfaeb074917a990e77d048ef6c0 : TIMED_OUT in 3h 03m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"80d0525fb89a0540067be2da2dc7cbd1ad058280","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"},"date":"2024-09-17 08:36:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"519cfd3ac50eb4b83c4a890295f57ff5553a0c24","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-17 09:00:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7c8342285c41f9d96a56dafd8e3eefa542ba1b4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 11:07:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8e2db9ab599f4286b0f59860cabb11e0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2cf7bf1185cb48d3a4243de297a522c0 : SUCCESS in 27m 20s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/aee1dc51d50c4ddd9df13c8963df0b44 : FAILURE in 31m 33s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/bf75e25ea0154a8992979302aefef8e2 : SUCCESS in 20m 08s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9c31b6807287456caa041eccbfb7473c : SUCCESS in 30m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3dd00926881c4534b58b1cfcdc193160 : FAILURE in 28m 43s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8457ed28be11495292b53793fddf117a : FAILURE in 1h 04m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71e41e8a8e2a489889bb4d84cec33eee : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a579c09e6e29443d91099b44d2e0aa7d : SUCCESS in 2m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/487daf934f1d4ff8b5243179ece5d2ae : SUCCESS in 2m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f14f54829a5640acad4a6bd0efe6c368 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d9c0e3e9b1541198f3fc985a3dff3f8 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b4b3d36372d47e0a360007278c6eb8f : SUCCESS in 4m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1642ee86dcb04cbab125b678bfde8a7a : SUCCESS in 27m 31s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0fa7366111c14d8e82f394670f0a0b27 : SUCCESS in 34m 22s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8b93c7a309754c20898f9dbfeed3bdb2 : FAILURE in 30m 21s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/97ea7d1f85bc4f1b803550f1f58f40c7 : FAILURE in 1h 00m 34s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/74e89ecf3e124e52a51e0074e03d3141 : SUCCESS in 1h 45m 27s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/9a42b1e08b904112aad3648b26bfa04b : FAILURE in 1h 57m 59s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/52268128c54e42618937685072723ee1 : FAILURE in 35m 41s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/d309ea3c6a5746f1b8ddfca333a62858 : SUCCESS in 1h 20m 59s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/cb1ed784dea44f26990464b3e8cb0da1 : FAILURE in 28m 34s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/9d99745537b34c508c5ab97bd1299d26 : SUCCESS in 1h 03m 17s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/b9fb52e92de74effa1afba1f2858e564 : SUCCESS in 1h 14m 28s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/813bcac53d014ba7a602d0feba69572d : SUCCESS in 1h 20m 05s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/940b4d22eb4a40df95e6377351272a61 : SUCCESS in 38m 17s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/48208e2f951342ba9a3b839b3a1d767e : SUCCESS in 16m 25s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3a15ddccd94b4bca92c58161995c4ed8 : SUCCESS in 1h 07m 13s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2fc7a74b025946cd81c4778c3b1e1db3 : FAILURE in 28m 55s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/700072a10d19436d801f2b2d67f06b6a : FAILURE in 1h 00m 04s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/19efc2b636ee4f50894abd6735bd9949 : SUCCESS in 1h 31m 17s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0404add6659a43b29ade56e5150dd326 : SUCCESS in 2h 16m 30s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/a8a6f6c8928f4145a8647b3641b37646 : SUCCESS in 59m 32s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/b91ed40ab115490d913b4bbd0da922f1 : SUCCESS in 59m 25s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/d776f07959484b1c9bb63f343291494f : SUCCESS in 22m 13s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/a18f879413624a7a8ab37df2796002ba : SUCCESS in 43m 10s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/7ecfdb5a99474ed7834822b377472a52 : SUCCESS in 1h 01m 23s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/620ab1615f3146b8b6e4d659d112c6b8 : SUCCESS in 1h 07m 15s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8d05063f230e4290948144bbeac22f60 : SUCCESS in 26m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fd8e35eba5d20b6d70cf5a6c5c3094a1492769b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 13:50:05.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/500d51cdc932490791c3fd3a66d64184\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f6163a07daf3474a92660a1724799bea : TIMED_OUT in 3h 04m 35s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d8ad1355a0ab44d192090581a7c07c5d : FAILURE in 1h 27m 16s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/669a061880044541ac6d9246ae107870 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f9358de4c355417e9a7b2d8f2f9a889d : SUCCESS in 5m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6b501b55e102daef7f61d5a916bf2455a5cbebf7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"},"date":"2024-09-17 20:33:10.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"89d84c6bca9006e9657277390b404753c55385ce","author":{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"},"date":"2024-09-17 20:34:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"737889e9d20f63e414bfa1dcc3781af08e63b35c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 22:59:08.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/33d528c6103742c4996405375299ab0c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a9a737f5bfe344abb90b301ecb02db37 : SUCCESS in 29m 40s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/222f20d4a3994a9eaec82d5c57dd8bdc : FAILURE in 27m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/210b791b21c44fe0965d766f84b4c111 : SUCCESS in 20m 08s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/34de7226333f4beeb9c123c22ac11e0b : FAILURE in 20m 50s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6f6c49f650644ec1b56cffb26fa7bece : FAILURE in 33m 45s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/aca4e2574d52440ab08c2cb9ceff851e : FAILURE in 1h 11m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7880fe9b73ff4bd0bdebf65741e79b3c : SUCCESS in 2m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0de2d7881de460bb5be230dd1d80ac5 : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/85dd343d58234774a9c960010580f714 : SUCCESS in 2m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8d05a2637f164879b71e0cb5090d7c15 : SUCCESS in 3m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38914bf2adb74ba589e436b24637f10a : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2fe286b9291492a8c607fadef70cd66 : SUCCESS in 4m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8ffa9e1f98ae49e5853234ee946849cc : SUCCESS in 27m 00s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6103de0603204d7eb8216c0463237c53 : SUCCESS in 34m 22s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4915809944db40e8a1d1322c5712745f : FAILURE in 29m 47s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/264c9822e8d44e5f90969511a9d3f5c5 : FAILURE in 1h 09m 21s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/1f5d0fba99e84e49beb072923978cecb : SUCCESS in 2h 01m 02s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a16b73314a28450b97e185370290bde9 : SUCCESS in 2h 20m 56s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/4ad36c5014d6444bb416f79a4fbb626d : SUCCESS in 1h 17m 13s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/224582dfa3ac4fe2ae5e4262adc242ae : SUCCESS in 1h 14m 07s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/3e844f0a3b8045b1bec0fbfbadc4b52c : SUCCESS in 25m 08s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/2ce7f1b187cc42728be0d0f8e5130158 : SUCCESS in 1h 00m 29s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/bf8369c4d3254232b453b056f1d04e1b : SUCCESS in 1h 09m 18s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/61b9e5d3793145129540d9790fc191b9 : SUCCESS in 1h 14m 23s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d8ffd7a5e9834c9fbb53771ed982f638 : SUCCESS in 30m 07s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6fcbaf7e53e5418e8ddda5c9305bd0f6 : SUCCESS in 17m 01s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6cd0b35ac4bc4b67b387ad47f8d50f8b : SUCCESS in 1h 09m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0f15cd9c699e4b71a098c0b6fbd63b76 : FAILURE in 22m 49s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1d2b968ad6e549178d14d055348f97cc : FAILURE in 53m 33s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/7b8763f0147c45a6a2bb9c35f754ff27 : SUCCESS in 1h 29m 49s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/76e390e2b06843529ca25d8779276c99 : SUCCESS in 2h 05m 02s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/ca458e2fe72b46a6a4dc8db3462161b2 : SUCCESS in 1h 06m 14s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/432ca5313d95464fb974803aee5082ee : SUCCESS in 1h 13m 52s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/3007f46a3b5e4704a9a804f86a39f3d6 : SUCCESS in 24m 14s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/43aa30bc0084423ab255161ff334b2bb : SUCCESS in 47m 08s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/207dd970eccb42fcbb38344d470d4fd0 : SUCCESS in 54m 15s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/cab591391b1746faaccaaa25d9033506 : SUCCESS in 1h 04m 35s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3aa966e0100a4c0a8c6f9f5384bbfda5 : SUCCESS in 27m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bd67419b19e857ece2fc4e001e26aa8b3fc6c71f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 23:41:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/671b16b9433a4dc3bd6a3d38f188e2e3\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e6a77d4b17a24125b94b55e918dbcf7d : TIMED_OUT in 3h 03m 13s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/468bd3b2c4134557b9ecf0ba850e0795 : FAILURE in 1h 04m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c3610adab21f4d72b5a35710d061e3cd : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0d27f8f446a94722b6e45d67d2fe9b3a : SUCCESS in 3m 34s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e5587d4394332f272648ccd32014d76cad84ef97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"},"date":"2024-09-23 07:36:43.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"36f93cbf1621c69cad564241a50f2f1e3cdd43c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 10:01:24.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/61caef218a754b2bb21cd3fe1621fbb4\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/fb2b1cf26db94f6d90d037589757bfdf : SUCCESS in 25m 33s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/788a2e9607954ff1bc041259cd5bedb7 : FAILURE in 28m 08s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8113fef54cf0473bae85bc37480b8af6 : SUCCESS in 17m 49s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9df907597a2f4f0f8dc76a92468840b5 : SUCCESS in 30m 38s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/941f972ffb3e4994bef44d8bfcdda48c : FAILURE in 31m 33s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fb3145de55a84f0bb519932ca0d41c81 : FAILURE in 1h 04m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f05eb02daad48e3885332b7cb79cbf4 : SUCCESS in 2m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4faa3beb13648c6a8f9ad972fe43baa : SUCCESS in 2m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c9ce7a9b22b3405ca6ee6b679cbff269 : SUCCESS in 2m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/522ff41921184c74b95b51c89bfdd06f : SUCCESS in 2m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/894cacc24e9641dcb02e901f65dcb973 : SUCCESS in 2m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/002da3dd336a41b28f2d37d12cbcc753 : SUCCESS in 4m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/dd59d1052bc24682ae2b5b8a5ed4d425 : SUCCESS in 26m 45s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/25bcb6263d2f4b1aba40a40977826c24 : SUCCESS in 33m 05s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0bf59beba6c740a8afe7c0bf109c9032 : FAILURE in 30m 32s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6ae780f65a634cddaf46461b354fca5f : FAILURE in 1h 06m 43s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/dff0b0a10b87491387e06139dc301ed3 : SUCCESS in 1h 43m 22s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a28a20df67774841b346e87bca0b3e7f : SUCCESS in 2h 21m 19s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/16ab3c0a011a4ca78882453cf29ebc42 : SUCCESS in 1h 12m 31s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/6f7def4266064a4fab56c341624567c6 : SUCCESS in 1h 20m 30s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/5c65c8f3b27a40ff819a3cdcb2117568 : SUCCESS in 24m 02s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/98a9cef1b75b4cca9c7181c2d6e69267 : SUCCESS in 1h 00m 43s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/deddc22759c4436da3793fca6c2baf59 : SUCCESS in 1h 12m 59s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/482c88507ac94f04b45751c86fae2df3 : SUCCESS in 1h 20m 47s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a8e7e388dcd04b4e82562e00a5bf048f : SUCCESS in 27m 55s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/58de5e90317346d6992290330a2812ea : SUCCESS in 17m 02s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/98a5d18d15834a5bb3b1d1a76faf8482 : SUCCESS in 1h 06m 04s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e98b6b8c567f4e8fa48c08715bb79f8e : FAILURE in 20m 57s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2f0a839b86a44245aeec6c12c548f9cd : FAILURE in 53m 21s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/d904487ae4264e7a99485c36cc90d81f : SUCCESS in 1h 35m 01s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/62e3e11053f741008e2dc1aeb09ac49c : SUCCESS in 2h 13m 26s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/7ba93df52a5641788dcc633e01b8b80f : SUCCESS in 56m 36s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/dc15a65cdf524964ba7dd87a425c8486 : SUCCESS in 1h 00m 55s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/2cc3e9d708c4417c8fca878c9085f33b : SUCCESS in 20m 04s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/784aa63aef274825a56a8ab9f012bda6 : SUCCESS in 49m 37s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/ed1a94025ce14c6f96280e2b78870b57 : SUCCESS in 55m 35s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/b6ea82c5a9664c5ab3f65f92d40896fb : SUCCESS in 1h 04m 09s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0d22bba6d60d490cb0f81c19c8556b7a : SUCCESS in 24m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"65f5fa200d0f5e195ff00a8d4a52ffe8b8cfcadd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 10:11:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14778bae41d84534bc8a61df55e50ba0\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/45c37780d1c249e2898cd24c118c7fa4 : SUCCESS in 2h 33m 12s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2fe69972a3b741128da9899fc1fc0490 : FAILURE in 1h 01m 58s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ac4d3e09ec46452bae8f8562dc627e8a : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c6934fa79eef4a84a4593cb950c884e9 : SUCCESS in 3m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"dae4702aaa558110c742955ad1554d38360dd2b8","author":{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"},"date":"2024-09-23 16:48:24.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5b00c2da21f80e39106d1560a9e6a94678f4bde0","author":{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"},"date":"2024-09-24 07:18:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ccc45ca87aaf8e94cfeb638ec90c941d7b5321a7","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-01 17:48:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"14d24d8695474eb802a865354d59eb696b6ea067","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-01 21:11:38.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8286df56dce24a0e9acfa1561a92c573\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/87a67da08d214241bc3c5384bd5a0ea1 : SUCCESS in 24m 02s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/430e35f17a4f40b98f0da2e003ecd897 : TIMED_OUT in 2h 02m 26s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d2e76448429a4c2cadd089b766e69616 : SUCCESS in 27m 12s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fef17f10ffca4f7681da2138a22fc7b2 : SUCCESS in 28m 36s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/412636f2087b450bae86ea9aeb1a6311 : TIMED_OUT in 2h 01m 53s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a9b6470d60b94a72bd59b7576da332d0 : TIMED_OUT in 3h 01m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17a47429ce8c459f8d3ccdb5963e5178 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4af57ec5ad6b4e76a42766b3ca932057 : SUCCESS in 2m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fcdc97a0656c4b82baf965dcc783e574 : SUCCESS in 2m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5e124c99e304db88954e5fedb68ffef : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c46ef04f9644bc0831ef602c4301201 : SUCCESS in 3m 55s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e752b862eefb4e33a224ecde4582ba09 : SUCCESS in 29m 04s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/aa652664c52a4198a7efcf84866c776d : SUCCESS in 30m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3bea7aa229b9445387af7a9dd26f9290 : TIMED_OUT in 2h 03m 06s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/99ef098caf2e4bdfbed190f9f68d2c57 : TIMED_OUT in 3h 01m 54s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/537ed6a8ce0b423b863bda458e8b1ec4 : SUCCESS in 1h 46m 36s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/3b0cde7d77ff4a83b2a5e16acceccd26 : FAILURE in 2h 36m 03s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/ad83ab6a9ed9467e8c30912db662bd21 : SUCCESS in 56m 13s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/31c9c4e620074cddae5324c776bb954b : SUCCESS in 1h 17m 25s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/cb73764541eb4f59a5417dcb95803ddc : SUCCESS in 21m 23s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/d4f2892023144079ad22b0a953803ed7 : SUCCESS in 56m 59s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/045e1746b5be4ff987db8c6380b02f17 : SUCCESS in 54m 35s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/6eabfb7b1e384b9fa8fb012634c8a0b8 : SUCCESS in 1h 30m 02s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/7f06cf40741a47998cbc89cac0edbb50 : SUCCESS in 30m 58s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/120d88f626e546cbaf9ea4263279c17a : SUCCESS in 24m 20s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3b461e88cc774e14912a04cfc0d5a14c : SUCCESS in 29m 40s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3ac2dcef298b45c598a3a3eb8c5c6fee : TIMED_OUT in 2h 01m 10s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ab237b76a94548b5a6561e0aa8c9feff : TIMED_OUT in 3h 02m 33s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/9c6a1879fd31400fb6463a7ec3b7cee2 : SUCCESS in 1h 43m 23s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/cc9ee39c2c414eb7b249c1bf5a29eea9 : SUCCESS in 2h 33m 41s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/6f794bef01a541d7ba97f3f4e44cf442 : SUCCESS in 58m 23s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/3f6ed94a0002479ba478ec5cd32e2e9b : SUCCESS in 1h 22m 05s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/42edf5d7c352490e9ff8728bb1bb50c8 : SUCCESS in 22m 19s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/fe26992ef24e4cad94a9a2da015169d9 : SUCCESS in 1h 01m 00s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/58a5e1f048b9409d9b2ab47cdd780329 : FAILURE in 45m 17s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/a2fa714500934d788e0cd39a3f53307e : SUCCESS in 1h 02m 03s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d8ae888706cd40299dfe70ef4a5e1878 : SUCCESS in 27m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1fe80da2017fad3ab097d65b7ac13cc154c62167","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-01 21:12:21.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"da0b24905c6d39ef4a6b4553148d79c1a246795c","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-01 21:13:50.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"4c1cc53af5a631a43344fef4de646ff79256a7f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-01 21:58:39.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ccd7f1c0b674f2734bd246534322757b07a5f02d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-01 22:00:57.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5479d526fdd15af9dc50204bef183c37a936ac24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 00:40:11.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/843f0976f2c34cf884953b4095f5e24e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a9b3cee5cd6f4284950e2ca84eb5dc6b : SUCCESS in 23m 25s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ef07ea58d47e4372b0423724afee3ca9 : SUCCESS in 1h 02m 49s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d4d3027e92c141bda299f8d047c9d756 : SUCCESS in 20m 32s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/696c788e28024673abacbc644ce2ae1d : SUCCESS in 34m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0abc5df406ab4cc7a99fe1b8d9db16fc : SUCCESS in 58m 37s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/52c15e8ecead4471980c57a7ebd73f3a : SUCCESS in 1h 27m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/086f4410c1644f83beac12fa49286117 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3656d982718643a7a8bc952585bcd0c0 : SUCCESS in 3m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ce8f0deb814940f0977671331e8e8735 : SUCCESS in 2m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bafb1fff01c454fb4630eb31710e7ad : SUCCESS in 2m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f23f2956dd94f1c9c63b0f61b2fb79c : SUCCESS in 4m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ea3df1314c5846349e927016f0760d10 : SUCCESS in 28m 10s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/f61da76e90844585b9c1f2404b80795a : SUCCESS in 31m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7f6530e3152849b99f0ee22067db8067 : SUCCESS in 1h 05m 20s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/566aed6996c944a2b65dbadf1e3292a7 : SUCCESS in 2h 02m 41s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/0a1b67c604e940ec8196cf6dbd92dc24 : FAILURE in 1h 50m 03s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/31e2a261a2674e9fbb29590293cab947 : SUCCESS in 2h 15m 31s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/955ae68f58da4f9d93c8148816099bd2 : SUCCESS in 1h 13m 17s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/8ac40ff7f53049b88831f1a0aa6e006a : SUCCESS in 1h 17m 02s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/d0193061b932413f9577a408b6a927ee : SUCCESS in 30m 07s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/3bcfd1f28e584b2da605edd0c365bc41 : SUCCESS in 57m 11s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/a2131822cb954a2db3e632bca786d884 : SUCCESS in 1h 07m 36s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/60297a496d834348b95215051667f35c : SUCCESS in 1h 47m 08s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/cd73a6ef562c4dab9f6c026ebff4ae00 : RETRY_LIMIT in 2m 14s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fbbf01d30f9f4917b589b914b9d8f58d : SUCCESS in 30m 00s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c57ef3f89fbc4c4b97c4b5076244ba07 : SUCCESS in 1h 03m 27s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/934dd527215e40d48a70dca6cca291a2 : FAILURE in 1h 42m 20s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8a03ce8096424989b0d2bbebd4b4e148 : SUCCESS in 1h 16m 25s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/c210092cb5b445658d6e40179cf0f5c0 : SUCCESS in 1h 49m 46s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b2bba4f7f5024625b973855ca7bcf603 : SUCCESS in 2h 26m 54s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/b0aa61f64505481785f526cea8d0abac : SUCCESS in 1h 17m 27s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/ba5233260c9741be8a6e31db3ecf9466 : SUCCESS in 1h 19m 22s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/21480e251bca4a3d81a521dae1d3c200 : SUCCESS in 27m 32s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/c25a39a3d6ee47658b99ce43b9f5e745 : SUCCESS in 57m 08s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/add57b270fd5491db89aaca7c93365a8 : FAILURE in 55m 12s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/68666bb736294ca5b93923883006b22b : SUCCESS in 1h 17m 09s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/449de3f584824c1d9ecaf4fd40461e1a : SUCCESS in 40m 07s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8c76873166dcd675c061651b4328bdc8a8009e42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 01:27:50.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84dd0de53d3b41d287fa16d4d490c0ee\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b81232543c4a48bca702941a5f8d5d25 : TIMED_OUT in 3h 10m 08s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1d55352d366f41bb8766b2a172c3bd5e : SUCCESS in 1h 46m 37s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fb5443468a434909a93506080cde47d4 : SUCCESS in 1h 53m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/08163f4225544eb5940d9ad5466b00fe : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aad3db4c41444f19b4851d8d53cfcefd : SUCCESS in 5m 28s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ac8b0969c64643e19402bd2737ebd807 : TIMED_OUT in 3h 02m 18s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7ece3d00cdb84b7186ebf61f2ccd4780 : FAILURE in 23m 45s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0e5958e964f5c89031314cb069fef0bd7f6751ad","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-02 04:33:50.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"d55daadf378e5a0a487aa604e27f44033ef3eb2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-02 06:23:09.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"d9a9eb14a7d8b8ed5aca7991d8376ea76f8d3081","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-02 07:16:02.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"9af92a98f486043e821dfcc54e58ea4e3cc03123","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-12-02 08:09:38.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"78cbcb93c523eac851c0ba51a8af945fba4456f9","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-02 09:01:40.000000000","message":"Patch Set 8: Review-Priority+1","accounts_in_message":[],"_revision_number":8},{"id":"04a43d233b3259036d18b92b6517460014194399","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 09:06:18.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e340859fee24ec3983e5a922c165133\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/fc83fd642098416c9538d62e3448bfcc : SUCCESS in 18m 45s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9dec724c71584f58ad077906726aab6b : SUCCESS in 1h 10m 21s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/711a8441133746028365a47052ff87ac : SUCCESS in 15m 21s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4a0b7dd41f8b44bcb3caa4954c9b0173 : SUCCESS in 29m 58s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/75b21bf2f4244bc1a5967d064a9a52d2 : SUCCESS in 53m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/338495de930b453bb36a4e77b8f326e0 : SUCCESS in 1h 28m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f62d2559df34188b857be55b28b2725 : SUCCESS in 2m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/615608b4eea540549b57e66c4f77cf99 : SUCCESS in 2m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2655c6d5b1aa41f4a90b6c83fc35df76 : SUCCESS in 2m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d706bc082d2043ffaaa9ef44283f098d : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/694dc9bd170947259419b9a48fe8075f : SUCCESS in 2m 48s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4469177dccdc4447b41c53b700d8061b : SUCCESS in 25m 58s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/84281f279a1e499faea5246e74e1b08e : SUCCESS in 31m 27s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/92bfe7f259384bb589669e61b5a865de : SUCCESS in 1h 10m 49s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3a18147cea0f42c48c643ef3d4187d24 : SUCCESS in 1h 24m 46s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/96de015f5aa7403f9613afb50a7317a2 : SUCCESS in 1h 53m 08s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0f5b72aa31f344078da489f53c053a01 : SUCCESS in 2h 37m 18s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/13f78f5181f64207b84afe6cf471f447 : SUCCESS in 1h 11m 56s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/c867417d52f74815a9bba27ec51acf58 : SUCCESS in 1h 11m 38s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/2a52e292c7284490a147650c80556704 : SUCCESS in 25m 43s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/4b0285c02afc420492a2b0723b088b65 : SUCCESS in 50m 35s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/8cacaee1b875487d9914ce11d282abd5 : SUCCESS in 1h 02m 12s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/d35888f15d184b8ca43cf7b939be4d71 : SUCCESS in 1h 23m 47s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b65f8f4b81e24bb8983388cc3cd55270 : SUCCESS in 33m 25s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/437be4902684415fb2386b67cb14e96f : SUCCESS in 20m 16s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/62acee7403884dd398662788961858fd : SUCCESS in 27m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f66a0a9704f845a59f1f4bbfa0e35e7b : SUCCESS in 50m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7921306800a94ae48bb3248284609bc0 : SUCCESS in 1h 20m 26s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/058b9d803b794d7088e44840a16b8bc5 : SUCCESS in 1h 44m 01s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/4fb7563411574e0bae15b3d8bf1a7498 : SUCCESS in 2h 18m 17s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/9cd46ad46167437486e11ce8bf33c164 : SUCCESS in 58m 49s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/fc50007fe2984430b4dc58e30ff9fbb9 : SUCCESS in 1h 04m 42s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/becf4c665e7145c68c8d577902e9b259 : SUCCESS in 25m 34s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/2c2c2df9101f481aa99608133b025775 : SUCCESS in 1h 02m 22s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/93bbaad23d0f4f32bc2b7eeefb32d4de : FAILURE in 48m 36s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/08af36c7db004a149374088efd09db2a : SUCCESS in 1h 16m 46s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8f9aa451674d4d9bbf5a0bf3f63cae07 : SUCCESS in 30m 55s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"3088a514f1242f37bf7d97177858f8b6b4f78601","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-02 09:14:56.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"bff8c98342670f43de38fde3cf93324dc42e6f6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 09:31:17.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5001961b31a344abae807fc6b03c9a56\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/484d9247ac8d4fee8f13c9d19861de53 : SUCCESS in 1h 40m 22s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7c09702fb44d4429a0f6b2fda5ef3536 : SUCCESS in 2h 34m 59s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/247b8b72af0442dfada5c5008088fd4f : SUCCESS in 1h 37m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/42a64d618d484cf9838b9b248e93baac : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/43355062cc0441b499182142a3296b5c : SUCCESS in 4m 50s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7616ae6d916a4777987fa16497745369 : TIMED_OUT in 3h 02m 43s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b8b47a9b677c48aabbc3431a52aaece8 : TIMED_OUT in 3h 02m 18s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"c2a1ce47ca299001ffc4f8192871b23a63a583bb","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-02 09:55:40.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"dafb8a594c103bfca595da2d8dc13dfeee7c68f4","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-02 10:24:17.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"705bdf8e19f34dff1116c055c0dda515541d08a1","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-02 10:34:45.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"3bc4532016401861ec0269836ff26a9d9b936c10","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-02 10:37:30.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d37520ac229b97b59c3c872f4c929a7d580a71e4","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-02 10:43:56.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d39111cdfdc850e4a3dee2abca88389f77d35c10","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-02 10:53:10.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e7289c8d9d9f278e8ca1474cc7333a84813b976d","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-12-02 12:33:48.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"21fadb0e214528d3307283aa669abfe1e21e69f4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-12-02 13:49:09.000000000","message":"Patch Set 8: Backport-Candidate-1 Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"1c8a8601bdd1d98429b349ca4759df59e99df329","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-02 15:27:11.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"8188bd36760c304adfb5b32268d4ef722de06148","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 15:28:17.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"06dc0314fbb96df87e0e87b37ea7223a5afd51f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 17:46:03.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8169159489c49a4ad5ac5dc984a0b11\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d690d0f042ea489cb8f6c050abb845ce : SUCCESS in 22m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6fa803cf2e384723944be0677b192884 : SUCCESS in 26m 34s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9c1bcb8320514815a7abd06c099844e8 : SUCCESS in 1h 13m 44s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4d00b17467304aab9299a34c17a6c180 : SUCCESS in 1h 25m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a825994203b6407abc8c8e2cf6fc6989 : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4191a8d2250d464298e5f5655cd90181 : SUCCESS in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4a6743197d5148caa891874c7aa55f0f : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0107617ac23f44d2b1e48e89e75d713a : SUCCESS in 7m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5ea794566b0d441a8d323732124b1b42 : SUCCESS in 25m 36s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/05d5de5bead54bb79c885538b5330926 : SUCCESS in 27m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4da1d2f0a485418588746470f35c7ee8 : SUCCESS in 1h 13m 28s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2e741d8b1b32430bae0b403018befb8e : SUCCESS in 1h 42m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/017cfdff970243658f806a969635ed45 : SUCCESS in 27m 14s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3c2d7aef161a4a2ab768a9cebb87e0c1 : SUCCESS in 38m 56s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0d30acf1e58843e594c9da87133a0189 : SUCCESS in 1h 05m 26s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/360c25c1c05841458e8074b27bfed2d6 : SUCCESS in 1h 18m 33s","accounts_in_message":[],"_revision_number":8},{"id":"627242735c092991258fc0489ffbbe9d4992fb2a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 17:46:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"b257ed648f4d3c6f2a95a72688e222b01e4dfebf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 17:47:05.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e57b9b592e1f427fa0b26e4b5fed6764\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b266cd03f1714dd8b5e713da9a22cdf0 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"d22245c71125c8c1c053fa772a362f0446467d0b","revisions":{"770905395dca4fdcc700a93bf3589e14547fdf85":{"kind":"REWORK","_number":1,"created":"2024-08-22 08:31:33.000000000","uploader":{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"},"ref":"refs/changes/82/926882/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/926882/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/926882/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/926882/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/926882/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/926882/1"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Roman Krček","email":"roman.krcek@tietoevry.com","date":"2024-08-21 13:02:19.000000000","tz":0},"committer":{"name":"Roman Krček","email":"roman.krcek@tietoevry.com","date":"2024-08-22 08:30:32.000000000","tz":0},"subject":"Change copy-cacerts behaviour","message":"Change copy-cacerts behaviour\n\nSo far, the certificates were not part of config.json\nfiles for containers. With future patches that will remove\ncontainer restarts based on triggers from config and\nservice-copy-cert, it is important that all files that change\nduring config are specified in config.json so that\nkolla-set-configs --check can detect those changes and based\non that restart the container. This patch provides prerequisite\nfor future patch in kolla-ansible.\n\nChange-Id: I5120f1874f25a9ca8326e6db8b12dee4c81feb56\nSigned-off-by: Roman Krček \u003croman.krcek@tietoevry.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/770905395dca4fdcc700a93bf3589e14547fdf85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/770905395dca4fdcc700a93bf3589e14547fdf85"}]},"branch":"refs/heads/master"},"a61ac1cd0d7b258225a543d35e0bf89aa468e59f":{"kind":"REWORK","_number":2,"created":"2024-09-17 08:36:43.000000000","uploader":{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"},"ref":"refs/changes/82/926882/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/926882/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/926882/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/926882/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/926882/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/926882/2"}}},"commit":{"parents":[{"commit":"3c130f290ea5b1c5758803dc3c3cd303c41c8a82","subject":"Merge \"toolbox: Bump ansible-core to 2.17\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3c130f290ea5b1c5758803dc3c3cd303c41c8a82"}]}],"author":{"name":"Roman Krček","email":"roman.krcek@tietoevry.com","date":"2024-08-21 13:02:19.000000000","tz":0},"committer":{"name":"Roman Krček","email":"roman.krcek@tietoevry.com","date":"2024-09-17 08:36:20.000000000","tz":120},"subject":"Change copy-cacerts behaviour","message":"Change copy-cacerts behaviour\n\nSo far, the certificates were not part of config.json\nfiles for containers. With future patches that will remove\ncontainer restarts based on triggers from config and\nservice-copy-cert, it is important that all files that change\nduring config are specified in config.json so that\nkolla-set-configs --check can detect those changes and based\non that restart the container. This patch provides prerequisite\nfor future patch in kolla-ansible.\n\nChange-Id: I5120f1874f25a9ca8326e6db8b12dee4c81feb56\nSigned-off-by: Roman Krček \u003croman.krcek@tietoevry.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a61ac1cd0d7b258225a543d35e0bf89aa468e59f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a61ac1cd0d7b258225a543d35e0bf89aa468e59f"}]},"branch":"refs/heads/master"},"58bd1b9f6cde374269564c718911e6ca9d006edd":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-09-17 20:33:10.000000000","uploader":{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"},"ref":"refs/changes/82/926882/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/926882/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/926882/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/926882/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/926882/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/926882/3"}}},"commit":{"parents":[{"commit":"ea198170bacf5735eaca22306011a24c1ea65532","subject":"Fix state checking via kolla_set_configs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea198170bacf5735eaca22306011a24c1ea65532"}]}],"author":{"name":"Roman Krček","email":"roman.krcek@tietoevry.com","date":"2024-08-21 13:02:19.000000000","tz":0},"committer":{"name":"Roman Krček","email":"roman.krcek@tietoevry.com","date":"2024-09-17 20:33:10.000000000","tz":0},"subject":"Change copy-cacerts behaviour","message":"Change copy-cacerts behaviour\n\nSo far, the certificates were not part of config.json\nfiles for containers. With future patches that will remove\ncontainer restarts based on triggers from config and\nservice-copy-cert, it is important that all files that change\nduring config are specified in config.json so that\nkolla-set-configs --check can detect those changes and based\non that restart the container. This patch provides prerequisite\nfor future patch in kolla-ansible.\n\nChange-Id: I5120f1874f25a9ca8326e6db8b12dee4c81feb56\nSigned-off-by: Roman Krček \u003croman.krcek@tietoevry.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58bd1b9f6cde374269564c718911e6ca9d006edd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/58bd1b9f6cde374269564c718911e6ca9d006edd"}]},"branch":"refs/heads/master","description":"Rebase"},"2c8596cbae44bc22f0266b8b4214b31d4cecc062":{"kind":"REWORK","_number":4,"created":"2024-09-23 07:36:43.000000000","uploader":{"_account_id":36702,"name":"Roman Krcek","display_name":"Roman Krček","email":"roman.krcek@tietoevry.com","username":"r-krcek"},"ref":"refs/changes/82/926882/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/926882/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/926882/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/926882/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/926882/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/926882/4"}}},"commit":{"parents":[{"commit":"ea198170bacf5735eaca22306011a24c1ea65532","subject":"Fix state checking via kolla_set_configs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ea198170bacf5735eaca22306011a24c1ea65532"}]}],"author":{"name":"Roman Krček","email":"roman.krcek@tietoevry.com","date":"2024-08-21 13:02:19.000000000","tz":0},"committer":{"name":"Roman Krček","email":"roman.krcek@tietoevry.com","date":"2024-09-23 07:36:00.000000000","tz":120},"subject":"Change copy-cacerts behaviour","message":"Change copy-cacerts behaviour\n\nSo far, the certificates were not part of config.json\nfiles for containers. With future patches that will remove\ncontainer restarts based on triggers from config and\nservice-copy-cert, it is important that all files that change\nduring config are specified in config.json so that\nkolla-set-configs --check can detect those changes and based\non that restart the container. This patch provides prerequisite\nfor future patch in kolla-ansible.\n\nChange-Id: I5120f1874f25a9ca8326e6db8b12dee4c81feb56\nSigned-off-by: Roman Krček \u003croman.krcek@tietoevry.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c8596cbae44bc22f0266b8b4214b31d4cecc062"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c8596cbae44bc22f0266b8b4214b31d4cecc062"}]},"branch":"refs/heads/master"},"0794c69af7a611b57aab09b675d96f27b304cd74":{"kind":"REWORK","_number":5,"created":"2024-12-01 21:12:21.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/82/926882/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/926882/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/926882/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/926882/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/926882/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/926882/5"}}},"commit":{"parents":[{"commit":"23519e6d88826cb4f77a3e9b7a21eef02f313ae9","subject":"CI: Use debian/ubuntu images from quay mirror","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23519e6d88826cb4f77a3e9b7a21eef02f313ae9"}]}],"author":{"name":"Roman Krček","email":"roman.krcek@tietoevry.com","date":"2024-08-21 13:02:19.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-01 21:10:10.000000000","tz":60},"subject":"Change copy-cacerts behaviour","message":"Change copy-cacerts behaviour\n\nSo far, the certificates were not part of config.json\nfiles for containers. With future patches that will remove\ncontainer restarts based on triggers from config and\nservice-copy-cert, it is important that all files that change\nduring config are specified in config.json so that\nkolla-set-configs --check can detect those changes and based\non that restart the container. This patch provides prerequisite\nfor future patch in kolla-ansible.\n\nComments from kevko:\n\nThe script also takes into account whether review [1] is\nmerged or not, ensuring that it will function correctly in\nboth cases thanks to review [2], because using the state\nfile, we can effectively verify whether the config.json that\ncopies CA certs to /var/lib/kolla/share is being used or not.\n\nIf we didn’t handle it this way, we would have to rely on\nchecking whether the directory /var/lib/kolla/share exists or\nsome another magic, which is insufficient because various\nstates and combinations of Kolla image and Kolla-Ansible\nversions could result in certificates always being copied.\nThis method provides a clear and definitive distinction.\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/924651\n[2] https://review.opendev.org/c/openstack/kolla/+/915440\n\nChange-Id: I5120f1874f25a9ca8326e6db8b12dee4c81feb56\nSigned-off-by: Roman Krček \u003croman.krcek@tietoevry.com\u003e\nCo-Authored-By: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/924651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0794c69af7a611b57aab09b675d96f27b304cd74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0794c69af7a611b57aab09b675d96f27b304cd74"}]},"branch":"refs/heads/master"},"78d7f837500ee85d02e3638d58f443b12678dcba":{"kind":"REWORK","_number":6,"created":"2024-12-01 21:58:39.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/82/926882/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/926882/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/926882/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/926882/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/926882/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/926882/6"}}},"commit":{"parents":[{"commit":"23519e6d88826cb4f77a3e9b7a21eef02f313ae9","subject":"CI: Use debian/ubuntu images from quay mirror","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23519e6d88826cb4f77a3e9b7a21eef02f313ae9"}]}],"author":{"name":"Roman Krček","email":"roman.krcek@tietoevry.com","date":"2024-08-21 13:02:19.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-01 21:58:28.000000000","tz":60},"subject":"Change copy-cacerts behaviour","message":"Change copy-cacerts behaviour\n\nSo far, the certificates were not part of config.json\nfiles for containers. With future patches that will remove\ncontainer restarts based on triggers from config and\nservice-copy-cert, it is important that all files that change\nduring config are specified in config.json so that\nkolla-set-configs --check can detect those changes and based\non that restart the container. This patch provides prerequisite\nfor future patch in kolla-ansible.\n\nComments from kevko:\n\nThe script also takes into account whether review [1] is\nmerged or not, ensuring that it will function correctly in\nboth cases thanks to review [2], because using the state\nfile, we can effectively verify whether the config.json that\ncopies CA certs to /var/lib/kolla/share is being used or not.\n\nIf we didn’t handle it this way, we would have to rely on\nchecking whether the directory /var/lib/kolla/share exists or\nsome another magic, which is insufficient because various\nstates and combinations of Kolla image and Kolla-Ansible\nversions could result in certificates always being copied.\nThis method provides a clear and definitive distinction.\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/924651\n[2] https://review.opendev.org/c/openstack/kolla/+/915440\n\nChange-Id: I5120f1874f25a9ca8326e6db8b12dee4c81feb56\nSigned-off-by: Roman Krček \u003croman.krcek@tietoevry.com\u003e\nCo-Authored-By: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/924651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78d7f837500ee85d02e3638d58f443b12678dcba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/78d7f837500ee85d02e3638d58f443b12678dcba"}]},"branch":"refs/heads/master"},"5b37125a63ee64a32d204936fed0bb562b8aea07":{"kind":"REWORK","_number":7,"created":"2024-12-01 22:00:57.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/82/926882/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/926882/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/926882/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/926882/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/926882/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/926882/7"}}},"commit":{"parents":[{"commit":"23519e6d88826cb4f77a3e9b7a21eef02f313ae9","subject":"CI: Use debian/ubuntu images from quay mirror","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23519e6d88826cb4f77a3e9b7a21eef02f313ae9"}]}],"author":{"name":"Roman Krček","email":"roman.krcek@tietoevry.com","date":"2024-08-21 13:02:19.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-01 22:00:50.000000000","tz":60},"subject":"Change copy-cacerts behaviour","message":"Change copy-cacerts behaviour\n\nSo far, the certificates were not part of config.json\nfiles for containers. With future patches that will remove\ncontainer restarts based on triggers from config and\nservice-copy-cert, it is important that all files that change\nduring config are specified in config.json so that\nkolla-set-configs --check can detect those changes and based\non that restart the container. This patch provides prerequisite\nfor future patch in kolla-ansible.\n\nComments from kevko:\n\nThe script also takes into account whether review [1] is\nmerged or not, ensuring that it will function correctly in\nboth cases thanks to review [2], because using the state\nfile, we can effectively verify whether the config.json that\ncopies CA certs to /var/lib/kolla/share is being used or not.\n\nIf we didn’t handle it this way, we would have to rely on\nchecking whether the directory /var/lib/kolla/share exists or\nsome another magic, which is insufficient because various\nstates and combinations of Kolla image and Kolla-Ansible\nversions could result in certificates always being copied.\nThis method provides a clear and definitive distinction.\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/924651\n[2] https://review.opendev.org/c/openstack/kolla/+/915440\n\nChange-Id: I5120f1874f25a9ca8326e6db8b12dee4c81feb56\nSigned-off-by: Roman Krček \u003croman.krcek@tietoevry.com\u003e\nCo-Authored-By: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/924651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5b37125a63ee64a32d204936fed0bb562b8aea07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5b37125a63ee64a32d204936fed0bb562b8aea07"}]},"branch":"refs/heads/master"},"d22245c71125c8c1c053fa772a362f0446467d0b":{"kind":"REWORK","_number":8,"created":"2024-12-02 06:23:09.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/82/926882/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/82/926882/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/926882/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/926882/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/82/926882/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/82/926882/8"}}},"commit":{"parents":[{"commit":"23519e6d88826cb4f77a3e9b7a21eef02f313ae9","subject":"CI: Use debian/ubuntu images from quay mirror","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/23519e6d88826cb4f77a3e9b7a21eef02f313ae9"}]}],"author":{"name":"Roman Krček","email":"roman.krcek@tietoevry.com","date":"2024-08-21 13:02:19.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-12-02 05:54:21.000000000","tz":60},"subject":"Change copy-cacerts behaviour","message":"Change copy-cacerts behaviour\n\nSo far, the certificates were not part of config.json\nfiles for containers. With future patches that will remove\ncontainer restarts based on triggers from config and\nservice-copy-cert, it is important that all files that change\nduring config are specified in config.json so that\nkolla-set-configs --check can detect those changes and based\non that restart the container. This patch provides prerequisite\nfor future patch in kolla-ansible.\n\nComments from kevko:\n\nThe script also takes into account whether review [1] is\nmerged or not, ensuring that it will function correctly in\nboth cases thanks to review [2], because using the state\nfile, we can effectively verify whether the config.json that\ncopies CA certs to /var/lib/kolla/share is being used or not.\n\nIf we didn’t handle it this way, we would have to rely on\nchecking whether the directory /var/lib/kolla/share exists or\nsome another magic, which is insufficient because various\nstates and combinations of Kolla image and Kolla-Ansible\nversions could result in certificates always being copied.\nThis method provides a clear and definitive distinction.\n\n[1] https://review.opendev.org/c/openstack/kolla-ansible/+/924651\n[2] https://review.opendev.org/c/openstack/kolla/+/915440\n\nChange-Id: I5120f1874f25a9ca8326e6db8b12dee4c81feb56\nSigned-off-by: Roman Krček \u003croman.krcek@tietoevry.com\u003e\nCo-Authored-By: Michal Arbet \u003cmichal.arbet@ultimum.io\u003e\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/924651\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d22245c71125c8c1c053fa772a362f0446467d0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d22245c71125c8c1c053fa772a362f0446467d0b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
