)]}'
{"id":"openstack%2Fkolla~926901","triplet_id":"openstack%2Fkolla~master~I77bcbbb3644cbc2e526243e1ca9ff7cdf6835553","project":"openstack/kolla","branch":"master","topic":"rework_sources_py_and_prometheus_updates","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-08-18 12:48:53.000000000","reason":"Change was abandoned"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-10-07 09:58:01.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2025-08-18 12:48:53.000000000","reason":"Change was abandoned"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2025-08-18 12:48:53.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I77bcbbb3644cbc2e526243e1ca9ff7cdf6835553","subject":"replace sources.py with sources.yaml","status":"ABANDONED","created":"2024-08-22 11:17:28.000000000","updated":"2025-08-18 12:48:53.000000000","total_comment_count":18,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"9c35bc0ea4ad6f643de626388418717bab37836b","_number":926901,"virtual_id_number":926901,"owner":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-08-22 14:41:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"CC":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-22 11:28:08.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2024-08-22 11:36:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-30 18:01:07.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2024-10-07 09:43:02.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2025-03-21 13:05:29.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"}],"messages":[{"id":"893f504732a51080de4d1e55cc9a0c28270cbcaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-22 11:17:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"432dde4820985d0ce19e7b9e3b595d0d97f89f48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-22 11:19:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"eb5dbcd68e9b4aa62955db625a16308b5c28c8f1","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-08-22 11:28:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"66ad5ffaead9bda6ba87b03e4be741bce8704205","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 11:36:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4866dba6daa247ceb019c51337b39ffb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9e223d96f8ea4697ae50b019158db040 : FAILURE in 3m 42s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/88f025cc2d304df68532065428ac439c : FAILURE in 8m 04s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6e5e0e8b96b24494900adcbcc966c481 : FAILURE in 2m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/55af8854a4eb48bfb20a8f0dcaf2652e : FAILURE in 2m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/18b8a6239c7e41f4bb42f507ddd61f0f : FAILURE in 7m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ad3b94b916bf43c696a51e9da76c285c : FAILURE in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c0853a4c6844a43a62d64b574b8fe62 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6360372607484302bf5b67e7aad8cb3f : FAILURE in 2m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cbf7728847694ba391be9b1ef0da13e5 : FAILURE in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e3d39ac1d20e4a7f92bb955d82f61c46 : FAILURE in 3m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/590cd1e2e45c46b98726d207dfe2cac3 : FAILURE in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81e3cf69118c467e8423b3c97b170fef : SUCCESS in 3m 56s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f21eb27e875e45bb8cc8e7f02826e7cd : FAILURE in 7m 07s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5b2819879fe04c419be5287cbbe2ee99 : FAILURE in 7m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ccfa86be2f0b4ac19ee0e9bfe1345cce : FAILURE in 10m 00s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ad4dc56a15ea4ee79c73e68ba7784d1c : FAILURE in 12m 34s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1565820436f644c4b3fb6893003e02f3 : FAILURE in 3m 04s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/96ed91c0482d4da59279a56f4dc7a0a6 : FAILURE in 3m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e0805d83b8b747cfb1807a2554565f9b : FAILURE in 8m 30s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a7f70c44108e43bbad6d0c5727e29846 : FAILURE in 10m 02s","accounts_in_message":[],"_revision_number":2},{"id":"a4064e07e7e86f6baaa798435b441dc9dd6ad288","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-22 11:49:45.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"dcd80ace53cfc3f6f3bfe6ec8dcb872dff040e4e","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-22 11:50:43.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4e4810e388b6e339f0d00f35d4a9268c216907f7","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-08-22 11:55:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bb2948d3323845a97e29006c08361f32c2920b40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-22 11:59:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2d1c4f828eb00b48c61c33a4f6ce286250be027f","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-08-22 12:03:32.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e7374b6638b24c8c526e56452d3b7f2120a613c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 12:14:22.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/950a95fc58ed49a2b46e788a96f6fab0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4ace2abf09124cf6b7029900967a4cbc : FAILURE in 3m 42s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2ba6e9b0a0a04d6eab2f0083602e609b : FAILURE in 7m 20s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b7ff72a5c1d34267a51a4334bfac3244 : FAILURE in 2m 50s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/be9f6ec9d28441c6a6e3dcf3b157a177 : FAILURE in 3m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/88552c81671145baa0ac523dd4f9a611 : FAILURE in 6m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/10eef0bf6d22496785dc780291d00b88 : FAILURE in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a72e06c0ffec4d33ba3454a4ff645790 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ce8ae67a0df46f4abe80001a95edd72 : SUCCESS in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/912c531591d24da1ad3281369c8c7f50 : SUCCESS in 2m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aa51b3ed9167454591d66d0a232cf5f6 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfeac46dba784bdd84d6787630812d07 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f94f19edeae24202a92161d840dd700b : SUCCESS in 3m 54s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1c4f3a0270084004ac602c90a60c1b38 : FAILURE in 6m 14s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ff9ef37ef47e41aa90bbe0b67ae5a19d : FAILURE in 6m 36s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/98b1baaa01844f6b8e7ab455b5f93342 : FAILURE in 11m 11s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/799ad88019b047efb42a67cdbf0b4d46 : FAILURE in 8m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0a37c041a70842c7aa7ddae3618095d9 : FAILURE in 2m 25s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b889bd034a3447d48978ef2a594d3ed1 : FAILURE in 2m 41s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ffe4c620e4f6498eb0ed80f4bcde26f0 : FAILURE in 6m 30s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8160924b781f49ed8fd5f0b5f5498ef5 : FAILURE in 6m 55s","accounts_in_message":[],"_revision_number":4},{"id":"96f27ab8803be649a4ad178dc20234f52536bdcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-22 12:26:57.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"3d3081f69f916c2a269caa7205b8f3a131c78b6a","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-22 12:29:23.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"900bcea080f794b090b235ba48c508c70a83483a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 12:42:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/954adfd6938b4302b0a19c69ac06b509\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2a0a9693e11f453a9225484cab83796f : FAILURE in 3m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/40c40064d8eb4d02a79c52ed80c630db : FAILURE in 8m 01s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2963251752944d4daaa5f7b4b1077f36 : FAILURE in 2m 34s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/dcac7ca2fa694368b927a278541c2d1c : FAILURE in 2m 26s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9b4f57afd100467e8c9a46ccc3005192 : FAILURE in 6m 27s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c7d860ac927c4eef88673bffa8307230 : FAILURE in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de3aedf5e5a94054a4129dbadbaead13 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f2c18ceff4bc4428a5d3c2c9c22d4e50 : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c6e2b118ef4d47a590fe67d9266015ce : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3df75ec8e31d478ab6dd188ce1c58c68 : SUCCESS in 3m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b4ec6668bcc4d7589ccbb9317a72b3e : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1042063827ac48c0840898b921760d54 : SUCCESS in 4m 24s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e3a7d491e26347cd88bc4768d368add8 : FAILURE in 6m 04s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/55dc1f79d62045fa8a8d7cc675cce192 : FAILURE in 8m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d33d75a1e70a4dbdb094f57634082078 : FAILURE in 10m 11s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c741354216944b54a639a1e4ac2955b8 : FAILURE in 9m 02s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c2f6b82a52fe4f0ca4fb6943aa86bb11 : FAILURE in 2m 49s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5549759025ed400a9f1f473d5df82c8f : FAILURE in 3m 40s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/55bdb24ba30f4e729fc2249bb7f560fb : FAILURE in 6m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5e4a6f713c9b4ee1926969dced1422bd : FAILURE in 9m 31s","accounts_in_message":[],"_revision_number":5},{"id":"55696e55ec593307e10959777de29368767c4c69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-22 13:43:41.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"d953e348794b056310a78dbd8afa93064c93d788","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-22 13:45:14.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"44a3cb4b7558e299cea483fb19adfcdc924c1f63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 14:05:25.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9f3cd86d27440a7bb8ca13727a59454\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9cb2488f44744e0684ad55f7de4ccf4b : FAILURE in 3m 44s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/fe6ae5cf3fef466588a96e14d5123691 : FAILURE in 7m 57s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/94790527835c4404a4a002b81ab36fc9 : FAILURE in 2m 30s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4a6730be69a848bcb1219eaa4618d9c3 : FAILURE in 2m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2dd97a82e1a04d14ab07d931413b0c77 : FAILURE in 10m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/46c903efa7a64d298f77b44223ce1211 : FAILURE in 11m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/181b188f24d54c2b85ba6a5d3c943601 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b002ffd650424e838c8563f1693f53b5 : SUCCESS in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7bfa8c132e4d4fd4941645c2a7801967 : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/edc080d9c4cd467393fd4a8965518a27 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/034b63de49594067b0c8d79d86b673b2 : SUCCESS in 3m 02s\n- requirements-check https://zuul.opendev.org/t/openstack/build/342ed04695c04755bb74270af6eaad8d : FAILURE in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de320398fb5a413e837fabe804cff024 : SUCCESS in 4m 23s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/adc17a5c412d494fa18c9013addd65bc : FAILURE in 7m 50s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6bc0ca1a369342299459befb622914a6 : FAILURE in 5m 54s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/40b3e7859be94ae088789eff0bfce3f0 : FAILURE in 8m 49s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1f6d10ae42634dafbd44abc931a329d7 : FAILURE in 9m 59s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0bbea78dda984aae9dc123a1437c8f17 : FAILURE in 3m 06s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3211fe10855a4c8bbf0b2a9b87f79bab : FAILURE in 2m 27s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7189b134a1cd4abdbff04490b58db4c1 : FAILURE in 6m 35s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9b7a20631e7d4358a7da0184d1a88695 : FAILURE in 6m 45s","accounts_in_message":[],"_revision_number":7},{"id":"b0d2f01d62a61aa7d556c1e98a1ea5f967de0af1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-22 14:19:05.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"90239b410f08d3418ebcc04708e701f0aed4631d","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-08-22 14:20:11.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"bafa76ba6b4b49501ea8834951185cee1e265858","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 14:41:24.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f60c97e34df49b79fa77e739edea473\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1c089dd0d90946da960277a12b12be4c : FAILURE in 5m 15s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d0db01f7d1164793b2c28af627251480 : FAILURE in 11m 53s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a16128705fe343fd99d108df38ba9c23 : FAILURE in 5m 14s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/197b52c824a64a0ba6d647f0aedbb393 : FAILURE in 4m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7b846b7fca6646fca881ccef784c1c20 : FAILURE in 6m 19s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b040e715d6b748aeaa89ce5acf1d4da5 : FAILURE in 11m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b91add93c453439096e53a4ea54271e1 : SUCCESS in 2m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc38983bf6b24328909443e0663a9ded : SUCCESS in 2m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1bdf192138704488802663bab4f837cb : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3b46cefcb6dc47d89f86a8f5d4afa463 : SUCCESS in 2m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/260d1fd6cba64b3c803b5273ab27f0c1 : SUCCESS in 2m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c7f4be66827f4144b19ab602a050694e : FAILURE in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc75a8d4758a4b4a9f86104e111aaed3 : SUCCESS in 4m 02s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/bcee3b4b70e84f9d96cfcd5267cfd2a7 : FAILURE in 6m 49s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e7057495a420493c9971836a629e9b5d : FAILURE in 6m 56s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/582703f310b24f7aaaa89bca07c5acda : FAILURE in 9m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0abd9054541e475c88bdc5778a4e67cf : FAILURE in 9m 42s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/58ce4c2546d84402a8ab98a282eeac37 : FAILURE in 2m 34s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1ea2d0a132e94ff2ade05bf2f6494c72 : FAILURE in 2m 26s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0212e3d357ca4a05be49e3b8552bb4b9 : FAILURE in 7m 24s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8672622381fa4b08b11462d30b06c486 : FAILURE in 11m 33s","accounts_in_message":[],"_revision_number":8},{"id":"8be83c5004006866cf91d43215b2398bb18b7be1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 08:04:07.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b01692696e443ffa0124f7fa347c38a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/aa846f7a061548fc902295d72a99fea5 : FAILURE in 8m 01s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/33ab00651f354159b06736a7c0588ee5 : FAILURE in 4m 42s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cf4d3714f97b4ee0a5ef1625e16c92ca : FAILURE in 14m 03s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d1cf57bb1ba94b4da5f39440a3393a3e : SUCCESS in 9m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7c81de0e420047fa89a201efd46d7feb : SUCCESS in 4m 10s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ced8fe18523d483e9baee5a9d78845fe : FAILURE in 11m 04s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/dedcede6daa4496992983df61ee951c4 : FAILURE in 7m 40s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"09b6420e864724200263580835f45dcc2cffa0bd","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-08-30 18:01:07.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"f3ba6b0179ffd327e36ca36a5139d3907a687755","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-07 09:43:02.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d532f3bc255163a31bc1331595f3329858bb79cd","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-10-07 09:58:01.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"9c35bc0ea4ad6f643de626388418717bab37836b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-18 12:48:53.000000000","message":"Abandoned\n\nI don\u0027t think we want to go that route, especially that you can override sources with koala-build.conf. Please restore if you feel otherwise.","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"429d7364385dc49bfb4955ee11cf79713388ed57","revisions":{"37b15fff8af21f1e4287f6cab90c6f3c84793ca2":{"kind":"REWORK","_number":1,"created":"2024-08-22 11:17:28.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/01/926901/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/926901/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/926901/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/926901/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/926901/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/926901/1"}}},"commit":{"parents":[{"commit":"8f6bc332bfe84275836c9153ebab6902041ca83a","subject":"Merge \"Add exclude for permissions in kolla config.json file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f6bc332bfe84275836c9153ebab6902041ca83a"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-08-22 10:57:50.000000000","tz":120},"committer":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-08-22 11:15:29.000000000","tz":120},"subject":"replace sources.py with sources.json","message":"replace sources.py with sources.json\n\nit\u0027s way easier to load a json from\na file into a dict, change the necessary\nversion fields and dump it back onto disk.\n\nan example script for the prometheus dependencies\nis included.\n\nCloses-Bug: #2077609\n\nChange-Id: I77bcbbb3644cbc2e526243e1ca9ff7cdf6835553\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/37b15fff8af21f1e4287f6cab90c6f3c84793ca2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/37b15fff8af21f1e4287f6cab90c6f3c84793ca2"}]},"branch":"refs/heads/master"},"db9e278107a757cefcd61d36eba7c1ce272d8731":{"kind":"REWORK","_number":2,"created":"2024-08-22 11:19:46.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/01/926901/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/926901/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/926901/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/926901/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/926901/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/926901/2"}}},"commit":{"parents":[{"commit":"8f6bc332bfe84275836c9153ebab6902041ca83a","subject":"Merge \"Add exclude for permissions in kolla config.json file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f6bc332bfe84275836c9153ebab6902041ca83a"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-08-22 10:57:50.000000000","tz":120},"committer":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-08-22 11:19:16.000000000","tz":120},"subject":"replace sources.py with sources.json","message":"replace sources.py with sources.json\n\nit\u0027s way easier to load a json from\na file into a dict, change the necessary\nversion fields and dump it back onto disk.\n\nan example script for the prometheus dependencies\nis included.\n\nCloses-Bug: #2077609\n\nChange-Id: I77bcbbb3644cbc2e526243e1ca9ff7cdf6835553\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db9e278107a757cefcd61d36eba7c1ce272d8731"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db9e278107a757cefcd61d36eba7c1ce272d8731"}]},"branch":"refs/heads/master"},"7457e5089cdf0cc5a42c56fc82d76688e2b3b5ec":{"kind":"REWORK","_number":3,"created":"2024-08-22 11:49:45.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/01/926901/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/926901/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/926901/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/926901/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/926901/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/926901/3"}}},"commit":{"parents":[{"commit":"8f6bc332bfe84275836c9153ebab6902041ca83a","subject":"Merge \"Add exclude for permissions in kolla config.json file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f6bc332bfe84275836c9153ebab6902041ca83a"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-08-22 10:57:50.000000000","tz":120},"committer":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-08-22 11:49:17.000000000","tz":120},"subject":"replace sources.py with sources.json","message":"replace sources.py with sources.json\n\nit\u0027s way easier to load a json from\na file into a dict, change the necessary\nversion fields and dump it back onto disk.\n\nan example script for the prometheus dependencies\nis included.\n\nCloses-Bug: #2077609\n\nChange-Id: I77bcbbb3644cbc2e526243e1ca9ff7cdf6835553\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7457e5089cdf0cc5a42c56fc82d76688e2b3b5ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7457e5089cdf0cc5a42c56fc82d76688e2b3b5ec"}]},"branch":"refs/heads/master"},"35608ead1d8a8b2f61b615dd6088fc91d99d6c2b":{"kind":"REWORK","_number":4,"created":"2024-08-22 11:59:05.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/01/926901/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/926901/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/926901/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/926901/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/926901/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/926901/4"}}},"commit":{"parents":[{"commit":"8f6bc332bfe84275836c9153ebab6902041ca83a","subject":"Merge \"Add exclude for permissions in kolla config.json file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f6bc332bfe84275836c9153ebab6902041ca83a"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-08-22 10:57:50.000000000","tz":120},"committer":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-08-22 11:58:30.000000000","tz":120},"subject":"replace sources.py with sources.json","message":"replace sources.py with sources.json\n\nit\u0027s way easier to load a json from\na file into a dict, change the necessary\nversion fields and dump it back onto disk.\n\nan example script for the prometheus dependencies\nis included.\n\nCloses-Bug: #2077609\n\nChange-Id: I77bcbbb3644cbc2e526243e1ca9ff7cdf6835553\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35608ead1d8a8b2f61b615dd6088fc91d99d6c2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/35608ead1d8a8b2f61b615dd6088fc91d99d6c2b"}]},"branch":"refs/heads/master"},"5c620717c03432e80ddbe60b759c4b270baf451f":{"kind":"REWORK","_number":5,"created":"2024-08-22 12:26:57.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/01/926901/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/926901/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/926901/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/926901/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/926901/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/926901/5"}}},"commit":{"parents":[{"commit":"8f6bc332bfe84275836c9153ebab6902041ca83a","subject":"Merge \"Add exclude for permissions in kolla config.json file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f6bc332bfe84275836c9153ebab6902041ca83a"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-08-22 10:57:50.000000000","tz":120},"committer":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-08-22 12:23:51.000000000","tz":120},"subject":"replace sources.py with sources.json","message":"replace sources.py with sources.json\n\nit\u0027s way easier to load a json from\na file into a dict, change the necessary\nversion fields and dump it back onto disk.\n\nan example script for the prometheus dependencies\nis included.\n\nCloses-Bug: #2077609\n\nChange-Id: I77bcbbb3644cbc2e526243e1ca9ff7cdf6835553\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c620717c03432e80ddbe60b759c4b270baf451f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5c620717c03432e80ddbe60b759c4b270baf451f"}]},"branch":"refs/heads/master"},"830f7fd9f10c283bae5f8eb849e22dba2e671fda":{"kind":"REWORK","_number":6,"created":"2024-08-22 13:43:41.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/01/926901/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/926901/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/926901/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/926901/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/926901/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/926901/6"}}},"commit":{"parents":[{"commit":"8f6bc332bfe84275836c9153ebab6902041ca83a","subject":"Merge \"Add exclude for permissions in kolla config.json file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f6bc332bfe84275836c9153ebab6902041ca83a"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-08-22 10:57:50.000000000","tz":120},"committer":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-08-22 13:43:21.000000000","tz":120},"subject":"replace sources.py with sources.yaml","message":"replace sources.py with sources.yaml\n\nit\u0027s way easier to load a yaml from\na file into a dict, change the necessary\nversion fields and dump it back onto disk.\n\nan example script for the prometheus dependencies\nis included.\n\nCloses-Bug: #2077609\n\nChange-Id: I77bcbbb3644cbc2e526243e1ca9ff7cdf6835553\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/830f7fd9f10c283bae5f8eb849e22dba2e671fda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/830f7fd9f10c283bae5f8eb849e22dba2e671fda"}]},"branch":"refs/heads/master"},"38927991cd8dcde0a34a66e560f215af0f3254ed":{"kind":"REWORK","_number":7,"created":"2024-08-22 13:45:14.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/01/926901/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/926901/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/926901/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/926901/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/926901/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/926901/7"}}},"commit":{"parents":[{"commit":"8f6bc332bfe84275836c9153ebab6902041ca83a","subject":"Merge \"Add exclude for permissions in kolla config.json file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f6bc332bfe84275836c9153ebab6902041ca83a"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-08-22 10:57:50.000000000","tz":120},"committer":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-08-22 13:44:53.000000000","tz":120},"subject":"replace sources.py with sources.yaml","message":"replace sources.py with sources.yaml\n\nit\u0027s way easier to load a yaml from\na file into a dict, change the necessary\nversion fields and dump it back onto disk.\n\nan example script for the prometheus dependencies\nis included.\n\nCloses-Bug: #2077609\n\nChange-Id: I77bcbbb3644cbc2e526243e1ca9ff7cdf6835553\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38927991cd8dcde0a34a66e560f215af0f3254ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/38927991cd8dcde0a34a66e560f215af0f3254ed"}]},"branch":"refs/heads/master"},"429d7364385dc49bfb4955ee11cf79713388ed57":{"kind":"REWORK","_number":8,"created":"2024-08-22 14:19:05.000000000","uploader":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"ref":"refs/changes/01/926901/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/926901/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/926901/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/926901/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/926901/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/926901/8"}}},"commit":{"parents":[{"commit":"8f6bc332bfe84275836c9153ebab6902041ca83a","subject":"Merge \"Add exclude for permissions in kolla config.json file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f6bc332bfe84275836c9153ebab6902041ca83a"}]}],"author":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-08-22 10:57:50.000000000","tz":120},"committer":{"name":"Sven Kieske","email":"kieske@osism.tech","date":"2024-08-22 14:18:25.000000000","tz":120},"subject":"replace sources.py with sources.yaml","message":"replace sources.py with sources.yaml\n\nit\u0027s way easier to load a yaml from\na file into a dict, change the necessary\nversion fields and dump it back onto disk.\n\nan example script for the prometheus dependencies\nis included.\n\nCloses-Bug: #2077609\n\nChange-Id: I77bcbbb3644cbc2e526243e1ca9ff7cdf6835553\nSigned-off-by: Sven Kieske \u003ckieske@osism.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/429d7364385dc49bfb4955ee11cf79713388ed57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/429d7364385dc49bfb4955ee11cf79713388ed57"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
