)]}'
{"id":"openstack%2Fkolla~926981","triplet_id":"openstack%2Fkolla~master~Ifb41c0c48c4743e70e95fa7e1d329e91362e966c","project":"openstack/kolla","branch":"master","topic":"kolla_mariadb_11","attention_set":{},"removed_from_attention_set":{"35010":{"account":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"last_update":"2025-10-27 17:52:33.000000000","reason":"Change was submitted"},"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2025-10-27 17:52:33.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-10-27 17:09:41.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-10-23 13:08:29.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2025-10-27 17:52:33.000000000","reason":"Change was submitted"},"36765":{"account":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"last_update":"2025-10-27 17:52:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifb41c0c48c4743e70e95fa7e1d329e91362e966c","subject":"mariadb: Bump to 11.4","status":"MERGED","created":"2024-08-23 06:37:47.000000000","updated":"2025-10-27 17:53:47.000000000","submitted":"2025-10-27 17:52:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":47,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926981-kolla_mariadb_11","meta_rev_id":"b0df5d08d0305aa13e01d694c7d188b9293eefc9","_number":926981,"virtual_id_number":926981,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-27 17:52:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":2,"date":"2025-10-23 09:16:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2025-10-23 13:08:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-24 15:47:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":1,"date":"2025-10-27 17:11:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":-1,"date":"2025-10-27 17:10:23.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":-1,"date":"2025-10-14 15:08:28.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":2,"date":"2025-10-23 08:15:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-23 10:20:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-25 19:11:39.000000000","updated_by":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"reviewer":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"state":"REVIEWER"},{"updated":"2024-08-28 07:21:30.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-09-11 16:27:42.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-09-09 10:53:15.000000000","updated_by":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"reviewer":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"state":"REVIEWER"},{"updated":"2025-09-24 11:09:20.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2025-09-24 20:14:26.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-10-24 15:47:40.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"64e958db7c785ca0007fe339e289f11315573ea0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-23 06:37:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8a4ea5127dcab4345d973632aa28b867921a941","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-23 06:39:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"48595e3ffd7ef5800a206dc26ac417c00060b1f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-23 07:13:32.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"1623c76fee42fa4749d0ee547d6f72f283012a9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-23 08:16:07.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"16fa871dedf19719e190a12d9c72d7b8dbc5d858","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 10:20:08.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1b34260e88c47debcd7494be9326ae0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/12d8659376744ef180b71647ababa4f2 : SUCCESS in 25m 24s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/abb0b212a8d6499aa92ca7f29026e816 : FAILURE in 23m 12s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/95ded99ea8054274a36aba282f905bac : SUCCESS in 14m 59s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e1474378d48142f88f1db0656c0d1da1 : SUCCESS in 25m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/20680f7556b742f192564445051dd31b : FAILURE in 20m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8f4af47a8fc64ec28e93e50428774309 : FAILURE in 1h 14m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d791f48e1b843ba8a4891e6d235f3ba : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0561662e42f941b084243162da950b00 : SUCCESS in 2m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee786cd03310461ca2cd241fd3151219 : SUCCESS in 3m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/77f6d82a6c284586bffdb43d298515a7 : SUCCESS in 3m 06s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7bab27009224e2a8b09b51e2f28b847 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ee2490bb87c498b95398b5e40899e31 : SUCCESS in 4m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/986082278c08461aa9acb3382a3adef7 : SUCCESS in 3m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/545743e97bac47269e34c387f2b5ab30 : SUCCESS in 26m 57s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/abfa9a1233f448f4945ff92ff07471f0 : SUCCESS in 31m 38s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/faee31a5108d4352a16a9ab438440902 : FAILURE in 25m 50s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5eaa4b6bd3bd4f8589f5fb87ed758a8e : FAILURE in 1h 18m 49s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/55aca91da82241f59ce6ae1870dfab9a : FAILURE in 35m 18s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c82908d95bd44ed0b305352b122f5520 : FAILURE in 1h 55m 34s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/619744d6484349aa98f9c2c7cc85f8c2 : FAILURE in 22m 47s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/ecf97694a9504fbfa8674e0d0ce54243 : FAILURE in 35m 02s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/2742205b7cf34ef78ec93c2771d8d27a : FAILURE in 20m 06s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/4605dea8cedc4a6d92059afaa8b92b2f : FAILURE in 29m 18s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/49dfcc5629fa410eb9a7fbb111e42848 : FAILURE in 26m 32s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/4fcd10af19e94e2cac5d84a7f69860d9 : FAILURE in 39m 45s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/32e62dd3033247909e9072c80bb26aa9 : FAILURE in 23m 48s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7d35425bc0d441a99fca0567108911ab : SUCCESS in 14m 37s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/172ea4d009a14f5d826f789c0e12d719 : SUCCESS in 51m 37s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/71967cffe6bd4ee7a3b963c5671300bb : FAILURE in 21m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/73c48b2a737c4339ab352ef56ce086a8 : FAILURE in 1h 08m 28s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/05dd09b8bdf84066bc39d3410a5623d2 : FAILURE in 45m 29s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/41d2a6517c0343699b6cedd2bb63f50d : FAILURE in 1h 36m 30s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/fdff076451fa489891a9910accc178a4 : FAILURE in 23m 07s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/9a1e134adcbe49d4a905ad470379127b : FAILURE in 34m 35s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/711ea7e665534438a84d45bfc5780e31 : FAILURE in 30m 13s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/919b4eb0a1fe48c9b79cc4ec5b428dce : FAILURE in 21m 52s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/02684de18d0c4c03a3fd5a8ba840b07a : FAILURE in 19m 27s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/d2f871c9797f4a31a544e69f19c112e0 : FAILURE in 35m 28s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5d42dba2285c4213b0ac308f32fb0dd0 : FAILURE in 18m 04s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"64301a9fb59960096a2512fbf44de4b18e2ea1ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 13:44:38.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b68499872f104018befa5d2d6df0d54d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/477104d0826341a1862dd55ef25dad0c : TIMED_OUT in 3h 04m 52s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/794063bda022444fbcc0e719c104fefb : TIMED_OUT in 3h 05m 17s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d9333cc1b6ba489493cf6e8df5c3c561 : FAILURE in 1h 28m 53s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bfc8192a03b346c291700803e4140b87 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7ea7784e52f14ad98d95e18016ed120e : SUCCESS in 5m 11s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b12f35a473c842cf9dfa795aed3afe69 : TIMED_OUT in 3h 04m 10s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/cb521685cdb646b88c9c945808bef0d5 : FAILURE in 2h 54m 34s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"31f04803e16482272c1b46132bc36ab22e9a47c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-23 14:01:39.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"3828a3bd87794b7f3995f3b94c98e5e693f6ce31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-23 14:04:01.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"32d4b499715216e8665438b305f753e5e790d133","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-23 15:07:42.000000000","message":"Patch Set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"c9a52cb51f153b9bbc598dd78da5b84c255ed94e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 17:30:54.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1920fa9014c8419c991d7fffec840b1d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ad6ebba7070a4b808d2767b24aece126 : SUCCESS in 21m 52s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c5504b8ed7964eb9a0bfb3d130bf29f0 : FAILURE in 28m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/eee41ca9fe9a4c629d08d7505fc5ffe3 : SUCCESS in 19m 07s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b6b48e352c9d485b9d669dd50300d157 : SUCCESS in 30m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/248de38da0354d2ebf87eb30a5825a41 : FAILURE in 29m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c0d2414a45f74672aaaf0d5218bac146 : FAILURE in 1h 06m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecb64852f7a340c8b7506f5cc6554b2a : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39b4a5c2d9a9429aab27981df91d33e9 : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/051e727de2ec4cf68b25b73c035cef5d : SUCCESS in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1357584af28643d3935cc08d98d4c37a : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb4e6434ff3043ebb7df11720212919a : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63a50fe3edfd40da8d4355b37a81d006 : SUCCESS in 4m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/096aba1633724b1bb142d49d21ab72a3 : SUCCESS in 4m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/43da87264b2a460c81e02d73e1515dbe : SUCCESS in 28m 21s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/44df2464e8b240228ed6ef2fc945d87b : SUCCESS in 33m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/cc03f72df87c4a398bbb3530ba966737 : FAILURE in 32m 39s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/874bc3965633409a87875c3917898bd3 : FAILURE in 1h 05m 39s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/bb2ea079b4b8469596d63bbaf553e4b0 : FAILURE in 35m 48s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/134bc08291c4449ba5afc3ae6e65da73 : FAILURE in 1h 56m 38s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/3cad961f98674b5ea37595a47307fa74 : FAILURE in 26m 58s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/aa1bce05f7324396a573e1d88636f7e8 : FAILURE in 31m 03s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/7376b33ff8be45768045eebf6fc9d331 : FAILURE in 22m 49s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/5105335199df4cf5aed69226ddfb904d : FAILURE in 28m 37s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/6a7626d3bbd64212b252056a8188710b : FAILURE in 31m 40s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/61cafd6083cb48ceaecdf31a4ef4226f : FAILURE in 33m 29s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d3be4b4ac31a418ba7c58e5c5c459aa0 : FAILURE in 26m 30s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d1754299c44c400ead0566bacd46bcda : SUCCESS in 18m 26s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7e87116a21344fa1a24673768d6e85e0 : SUCCESS in 56m 57s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a7134e3a4dae45dc871f9cc031bfe53a : FAILURE in 23m 10s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d64a43cc74e04af2856ae5b28b94bb5a : FAILURE in 1h 00m 00s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/ead5ae6029e743169f1cbe28b199720b : FAILURE in 37m 24s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b2b03b4bb5e04e04a5d166eff96d5377 : FAILURE in 1h 43m 28s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/0faf5dae916b4cc885e83dd1fac0ea31 : FAILURE in 23m 37s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/9fc5ae8d9e5944878a1b08c4d65fcac7 : FAILURE in 24m 50s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/c3a2cd0800f04ec883935f97162ea455 : FAILURE in 20m 15s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/910777f934b742978dd2bc69ce09a00d : FAILURE in 27m 28s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/8e11b7a514684fc9934f4ac35b5bf709 : FAILURE in 19m 29s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/1fe2db27594d4df7a5b4a880d54f6d70 : FAILURE in 30m 47s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0e89e64dd18d4365846d896c29122411 : FAILURE in 21m 24s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"521da0191c36cde6570dc775c9a1ae181d28b5dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 20:16:35.000000000","message":"Patch Set 7:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/927039 is needed.","accounts_in_message":[],"_revision_number":7},{"id":"fb873cb05b3e2e3ba3cb885fb2221f31ed897eee","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-23 20:17:32.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ed812328a414528a8d3d661d1dd79763e44bd2c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 22:35:10.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/880035abdfaa4ea3a2db11ab3adf2c13\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/cd21de5efdb8484790e183eb95656f2f : SUCCESS in 22m 32s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8b8bcf6b7e814b1982095731c56f480e : FAILURE in 27m 28s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6a5cdcf0074941c0a2adf13c27acd9d0 : SUCCESS in 18m 59s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/aad6f28129e44003a4a3b6f8770343a0 : SUCCESS in 28m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/926261efe3654096ab8d15f2a41e9bb8 : FAILURE in 24m 40s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/04d46ce0529d4669a9e5ca54973dbb8a : FAILURE in 1h 04m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bc3562f4d1a41d6aa13230ac53650f6 : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6991afa8bf1c4a7fb5390524dd4e4eb1 : SUCCESS in 2m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a32eee83144343f7b58a5622b10e290b : SUCCESS in 3m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e53fb748c97a4293a23a61510096030d : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f3ba29e736c4c0e8c7f7f9d8c960b05 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed93f89baef9462abb769221415dfa98 : SUCCESS in 3m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/223430168b3149fcb4272ea153ce537d : SUCCESS in 4m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8e1e3c4b0b7f49719f1e3dd709a917b4 : SUCCESS in 23m 05s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3a70c8e690d848e2a0b09bb7960c844f : SUCCESS in 35m 19s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b7c562cfc62e48cc9f45a1e7e3c29a65 : FAILURE in 26m 48s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5eedd3556ddf46ebbe9121b822d0deb9 : FAILURE in 52m 41s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/816747b4e7374434865fe7b28d23435a : FAILURE in 35m 23s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a58bf602ab124531b6c6a4cb1a074116 : FAILURE in 1h 58m 05s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/c9ff6a91318c45bebbc74099f1a58665 : FAILURE in 26m 49s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/5b2a99fc81ad4e529b288f8d1e077643 : FAILURE in 29m 29s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/a11f43a9853f4a6db39f64108e030dc4 : FAILURE in 21m 38s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/06ef1f116852492d9264ba0edb5b81c1 : FAILURE in 28m 52s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/cc36e2f455d8444c8c938c1916ea34a5 : FAILURE in 24m 52s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/3c9ecc4fdff44d709828693d1337d164 : FAILURE in 29m 15s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/650dbcb247e446aea74b3a9f3f0f6405 : FAILURE in 22m 17s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ccbc0e3f5db44417bd7afe185318a41f : SUCCESS in 16m 56s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ccb292c77e4d4117b7712e98a5dcf459 : SUCCESS in 41m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/026595c561134508be5e7b93654803ed : FAILURE in 17m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/754c68f31f2d4ce6a45d788015277330 : FAILURE in 38m 53s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/1dfb483217d749b5b31ee7ff65b1cde7 : FAILURE in 31m 55s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/dbce839810b3488ebed10fc7f5644f0a : FAILURE in 1h 36m 18s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/4b3f22c2f7354a378895c477058833fd : FAILURE in 18m 16s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/e74f2e8e43c9481796aa60e6953aa4f3 : FAILURE in 28m 39s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/1aa302c7438b4a0b9349617d2a293328 : FAILURE in 14m 22s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/9cebffbe659540819e4887aaa9fa36c0 : FAILURE in 27m 44s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/a288cd7ba43a442a8555a2e03e4c2dce : FAILURE in 17m 35s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/26883f18f7ca4c57abd63b62d1e1c78d : FAILURE in 20m 30s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5e8644cb0bb04e5c9e4fdf227f9bfa4a : FAILURE in 21m 12s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"3734e1abc36fa707ce2a1e1fd5905a99da61fd65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-24 05:43:27.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"8599b162b74646b9b5383bb89643d6d1be2d7e6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-24 07:32:21.000000000","message":"Patch Set 8: Verified-1\n\n(13 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0097ffa2f194457a354f89241b5917f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ad8384aee6514469bfd5fd80c4670183 : SUCCESS in 26m 10s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3887c4a85cc2484b85fba3a8e46bf1b6 : SUCCESS in 48m 51s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f4aa027ca4f34b30ad2a7f66213a4443 : SUCCESS in 13m 04s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9c137e64a0e24a60a9a2b90f16a65a0f : SUCCESS in 20m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b6867004cd6a459cbcb6bf6149e2f9c3 : SUCCESS in 45m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/df748ab82c3b415399ff58ce4bffa23c : SUCCESS in 1h 06m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30f8e32888ed4501924fae570276b862 : FAILURE in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c24731551a15470cba9ed8b0ab59072c : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a60b686efca54d20a94637df6934231b : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4eda0de745204ec384863c183f647433 : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc64ffb375f343dea76bfd01b577edac : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e9a28f90f864655aa1aac1890797f1d : SUCCESS in 4m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dde0f9c414e242b6a3db4e7d23656855 : SUCCESS in 3m 22s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a28757761f95458ba52a63cd8041afae : SUCCESS in 19m 00s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d74f2568c400491cb6d00fbeff8bc802 : SUCCESS in 38m 53s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c593e37e98954c94a0c3596aa93d9bad : SUCCESS in 1h 06m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d85e1de5e604462097aa67fb2be090f6 : SUCCESS in 1h 40m 12s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/7db1358bcb6644c2b1736701d31b1240 : SUCCESS in 1h 41m 58s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/95c29c6bc1924d87976c765f585d82ec : FAILURE in 1h 30m 27s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/61bdd7de759c49a79f3fcc26fee47fb8 : SUCCESS in 1h 07m 49s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/62ac904d5e894e87916f55d3c02a6134 : SUCCESS in 1h 25m 08s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/dce0051d199f4e3490fb394c3bea7b29 : FAILURE in 23m 30s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/1669a253f28d45028276f34c273f94dc : SUCCESS in 59m 26s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/c100ffceac444c579afe6a81ccd499e2 : SUCCESS in 1h 10m 33s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/bf8db2b8c3194245b9a43206304454d9 : SUCCESS in 1h 15m 20s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/74f1e69be5764692928636dd9cfdfa97 : SUCCESS in 34m 18s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fc29b85b73814c63b43a5d477b2825a9 : SUCCESS in 18m 44s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/27f1fede8cfb4d228e7c5615f021c024 : SUCCESS in 45m 07s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b7a17311966a4465b2601b66f740c873 : SUCCESS in 42m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/791cb6fa0a4248019c361adf38b97d44 : SUCCESS in 58m 22s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/6e33e71e55ab49f488c1a571277f6180 : SUCCESS in 1h 09m 10s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0b78b266966845f89799b95d10687f46 : FAILURE in 1h 34m 32s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/72adeb180fd94ea4936bd9c8e4ce7fd0 : SUCCESS in 1h 01m 14s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/c768f821b7054b678a0505724043dfaf : SUCCESS in 1h 10m 34s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/a29f324c20794c8b829856473490c11f : FAILURE in 21m 05s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/8ef7634499a34f1a9b41bf88640f6f0e : SUCCESS in 46m 54s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/7e2f03d5b4c24dd1b41a808e0f2504fb : SUCCESS in 55m 07s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/78db6bfb468946a1aa7ed279daafd548 : SUCCESS in 1h 04m 01s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/39f868dc7a8d4c73b1eee33446d0537a : SUCCESS in 20m 17s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"b1ff4a7ea611e8f7344db8a5d5e00ece6a578c52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-24 18:22:55.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b10bdf168414931b35d43f231055afa\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/30ed7084463e4e0a8f1f3572705390e9 : TIMED_OUT in 3h 03m 52s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8c7561e44a5c4c44b13d26a79ab0d015 : TIMED_OUT in 3h 03m 36s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0ead4bb026274d9dbae23be5d2d8beb1 : FAILURE in 2h 25m 06s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/71a17e5c8abf47fead1fe016b0f51071 : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e58aaa12fd0e4c05881740bde504bf6d : SUCCESS in 4m 44s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/755938cfad6446a0b721dfa907221157 : SUCCESS in 2h 29m 14s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/72f147ae7f974d32923fab29c89fba3a : TIMED_OUT in 3h 03m 34s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"11a6aa62abcc2aa4440e705624d205ef4b12cad6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-25 17:32:28.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"4d17274bf3acd3cc93b949952bd63efc811d9d6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-25 17:33:19.000000000","message":"Patch Set 10: Commit message was updated.","accounts_in_message":[],"_revision_number":10},{"id":"86373deafb3c3476aa545044b0f139a5f9133888","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-25 18:33:42.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"be64ac8b092a0b0aa9ac8c87a5bd4a02f7a1c5c0","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2024-08-25 19:11:39.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"697afa92bd6417729465ba31c5786a84724f9455","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-25 20:35:21.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a807c612720447f98e47b7129ac16b9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/30d404ff906843399f16e6145613b24f : SUCCESS in 23m 57s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/956d3ce7e3bc47e1b020d7477e52065d : SUCCESS in 52m 42s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/be99bbfbfe3c4140923580cfece00516 : SUCCESS in 18m 20s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e0dd1112905a46b3959940f10b5177c2 : SUCCESS in 22m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/92e336a95e9c433b8e3a8fd422a6c27a : SUCCESS in 43m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/176917a3451c47e285eebcd87523a402 : SUCCESS in 1h 04m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff21976efa8f4693af9158c8e446291f : SUCCESS in 2m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72283fa3324e479ab05d2e703f45b355 : SUCCESS in 2m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae3a8a59716f4663ac5e801071089331 : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/206423bdd6e2468aba7043a9337f5e6e : SUCCESS in 2m 41s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/22ba83e5c7cb40fbbabbed267788ec4c : SUCCESS in 2m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72a746d3960d4a8688db1b8485399b1b : SUCCESS in 4m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7cbbd0295f542e0be40692cd7e95896 : SUCCESS in 4m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2ba9e0a1a551426c8627ffba1e8cce95 : SUCCESS in 28m 13s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e9d8e3216a564402be07dd0512c5f8f6 : SUCCESS in 33m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5ab6b1e0e37b4f729f59b8cf00626988 : SUCCESS in 1h 03m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0560959d9e4d48dea7b01842f716a94f : SUCCESS in 1h 16m 40s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/cbc318a30b8c491588af24f909957db5 : FAILURE in 1h 28m 02s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/7adc10c8368449f3b4c14382c57e4a7e : FAILURE in 1h 55m 26s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/82d563ec26a241df941091994025de82 : SUCCESS in 1h 19m 19s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/890043ce3c544eca8d0694ff1d08a7a0 : SUCCESS in 1h 17m 49s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/6fb0ee6f2392441482d95fd863ab40cf : FAILURE in 23m 55s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/12edc88c6d404d369d86db612d7d654d : SUCCESS in 54m 06s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/3ee220ebbda645789c5d9d8c46402d70 : SUCCESS in 1h 09m 29s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/5036153829b349f3ab3b8e1a212f6afa : SUCCESS in 1h 20m 13s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/dca505f00d024a269a33e55a6fa91d6f : SUCCESS in 36m 19s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/791eec8b49dd4504b80e3685f24cd93e : SUCCESS in 22m 17s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e233e0c34be8416bb19e6d02559a3486 : SUCCESS in 45m 41s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ae2a6972a2e94289a4cdc3d4477c1de3 : SUCCESS in 47m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b728756db12d42a4a2f46d269cd322a2 : SUCCESS in 1h 01m 19s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/2625e5d3f66f4c3bb33b47d7837283e4 : SUCCESS in 1h 13m 37s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2296428e3a7e4aceab3c34fe64c98684 : FAILURE in 1h 20m 02s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/e1e1392412874958a5bbeb8ed50e0926 : SUCCESS in 53m 41s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/12c4fc161f9345b5b03c6839abca09f0 : SUCCESS in 1h 03m 07s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/dbe31f4100ba445e953dad6bc8df0e56 : FAILURE in 20m 13s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/2ab2ecbef9304b5b97c1bdc2d96237ba : SUCCESS in 46m 43s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/fac8b170ce9f43c68a6b060c31d2725a : SUCCESS in 48m 13s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/83c664020d784dc8852cdbdc20fcb656 : SUCCESS in 58m 03s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/71d308fb4b1a426bb5f4aa6c6db02989 : SUCCESS in 29m 43s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"f8ae6eca04e185d92a1dfbc67cc802b0bc8a9216","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-25 21:39:00.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46d8800990ad4cdba086b2d6cb1f551b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a95da3061553481198816f31320e0a15 : SUCCESS in 2h 58m 08s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2e696e3452964fceaab9312e3d951287 : SUCCESS in 2h 59m 38s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5549789158354b22a681b47f99d34202 : FAILURE in 2h 02m 31s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a49fe0f0ffeb4244a9bb349ba7bfe531 : SUCCESS in 9m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/855c4936814147aebcd8cca0be3ac3a5 : SUCCESS in 4m 30s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7c669fad52f8453bb4cda20cfd4408b9 : TIMED_OUT in 3h 04m 52s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d4b68d0ba7f241838bd13096ed4d04e7 : FAILURE in 2h 09m 39s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"369832752e9f8a0fdfb66a155aee3429e941d3d8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-26 06:49:38.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"a59ca4d57d67708e246b94bba0c93b0e8611c92a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 08:45:35.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9060ffd68aae45fba943b8681bbbdff8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/abee12bf148c4ff1983d3cf74e42a1c4 : SUCCESS in 22m 04s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ad1663ebc3734073a33f2463341de1b4 : SUCCESS in 1h 01m 58s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/537c414b6aca4df7ab1b0e0d9e0ce56b : SUCCESS in 16m 20s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6934040bd7bd4e0f8ebff6a9d43dd018 : SUCCESS in 27m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7fe05b2b24ea410095aa7e44538092f7 : SUCCESS in 55m 44s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4c7d379fdbb14d5c82e1821026fec18e : SUCCESS in 1h 11m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8b4e2412b954f41990a85fc472f923a : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6538932c8f644ab96b1c9f2c810ceeb : SUCCESS in 2m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e68132a91b7e4fe5a2815b417479eac8 : SUCCESS in 2m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b1742dd2de2041bf98e2cdfe488a1264 : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7432346a2d8e43d0b3ff6f7deb03819d : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55c2091fc6364a9ea11cf1b4a147d38a : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/976cd241605446249ea56a4934e9cc04 : SUCCESS in 3m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/decdf05746484f2696ff74a7afb83709 : SUCCESS in 29m 46s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/66763ef9254f41e28a9e31b5b8eb0f0d : SUCCESS in 34m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/84aa0bebebe145498bbf5a7dd93f52ed : SUCCESS in 1h 05m 14s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d5cf0a7716ff496c900bdb12c4cf6100 : SUCCESS in 1h 14m 11s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/d927d65b0f7c430bbbbb882f6da72d69 : SUCCESS in 1h 53m 52s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b0e0d56027d643259792ff1b1d620429 : FAILURE in 1h 46m 52s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/eb229f029f3e43ecacf5f443eee656c7 : SUCCESS in 1h 09m 02s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/49fb313fd7994a21bc97f7e490a9fc18 : SUCCESS in 1h 13m 10s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/f0ac35a78b014617a8b9101c9395722a : FAILURE in 26m 26s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/b0aa8be6c6e244dea81e78418e3b9e24 : SUCCESS in 1h 08m 18s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/edd0c01ba31d4bdf9011a19e1fdee45a : SUCCESS in 1h 06m 12s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/3d5dcd3d00e941c8b1cf632992ccdd46 : SUCCESS in 1h 21m 22s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/270ac44b7a26432e96be62fd41703cd2 : SUCCESS in 34m 09s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6575394ccd854a75831e6f1e6164a14a : SUCCESS in 16m 47s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7315a3f7f5d44591bd542384be6e045d : SUCCESS in 1h 02m 06s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/59f652be77b44624b538f0282d65bfc5 : SUCCESS in 1h 00m 54s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a590ab72155c45248a2dfdf559e3d79a : SUCCESS in 1h 12m 06s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/5ed671804b8c4dd7bd410a5fd49d55cd : SUCCESS in 1h 37m 48s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/79eb946c1cb049e9a5ccd3a5b70eb562 : FAILURE in 1h 30m 57s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/dee9808238b342c3b17e275ddf355d78 : SUCCESS in 1h 11m 35s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/2c7addda11634e23b1cd1099dcd1bc28 : SUCCESS in 1h 02m 28s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/a6ef5874ab244222805870b4ac51121d : FAILURE in 20m 29s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/524b49bc257340bcaef604cb98dd83a3 : SUCCESS in 47m 08s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/974c3caf073c47beb5b7efb8b13c855b : SUCCESS in 52m 27s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/a9837a107755411dac44d924206738dc : SUCCESS in 1h 08m 24s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/fdaf809967b249f7a57411458466a855 : SUCCESS in 24m 19s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"051e30d8573c305e2bd315c039a4e5ca4e4b8de1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-26 12:14:55.000000000","message":"Patch Set 12: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"081a98c378182e4638b952c29de9cbc2d0d9e649","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 14:04:46.000000000","message":"Patch Set 12: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/927096 is needed.","accounts_in_message":[],"_revision_number":12},{"id":"309f5f4ed9a9998afb8f62e3611291c9865f67ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 14:04:48.000000000","message":"Patch Set 12:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/927096 is needed.","accounts_in_message":[],"_revision_number":12},{"id":"9be7a04adcf764061e4df38a9f517a48628ec62a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-26 14:05:09.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"183b510713c77b9de14cdf3393858ed11e030bda","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-26 15:12:57.000000000","message":"Topic set to kolla_mariadb_11","accounts_in_message":[],"_revision_number":12},{"id":"d3da9bce5ecea6ac644b5b5bfaabb408f7ecb5b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-26 15:13:09.000000000","message":"Patch Set 13: Patch Set 12 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"0b8d6f4ffc264edfa9e7cb3432d9ba016b7892c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 15:13:15.000000000","message":"Patch Set 12: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/927096 is needed.","accounts_in_message":[],"_revision_number":12},{"id":"075af7649cba318776fed09aeae5bc66db88d056","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-26 15:13:19.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"5f7a68a0873cce5ef8f69511de8b096aeff6c90b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 17:52:22.000000000","message":"Patch Set 14:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/927096 is needed.","accounts_in_message":[],"_revision_number":14},{"id":"ecf4e97e8f755b7afada55129a30ecd9c910f503","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 17:52:22.000000000","message":"Patch Set 14: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/927096 is needed.","accounts_in_message":[],"_revision_number":14},{"id":"03078bb249e83d080d87ce9ef163b82910b4b788","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-26 19:42:21.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"56ede350fd64eaaf3cf66c839c1b0feb519b63c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 21:47:43.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e924d1e56a5473fb95475e2db51e816\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3b313d1163e84ac9a038cd6614f72e30 : SUCCESS in 25m 59s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/73a4d55b3de940d9b515f23dd26d49fa : SUCCESS in 1h 08m 38s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e15634f22654493086d52e25430365f1 : SUCCESS in 21m 49s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/230159c9be494d20a88341e32e4fc814 : SUCCESS in 28m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1d27e047a8fa458490d0f13772440976 : SUCCESS in 1h 16m 49s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/90e5c56889e843ec8b9404f963cd3d25 : SUCCESS in 1h 12m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f080267c86a46a2adc1c0ce9093f6f6 : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c532ae4497ff44cb872310a448008470 : SUCCESS in 2m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9927a2d8cce94a7c9d25c84750ec3031 : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0c927024c0cb40c8a613c0c54a41ecab : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4bb7f86a50744ae5b63e0fde5fd5618f : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bffc1de99d7f498395f0206617903d89 : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02ebeada6cc74ff6966b7683e4bb62ea : SUCCESS in 3m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b2ec7700c99c4053bec2d12c82f676b3 : SUCCESS in 30m 08s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0d1be55fa0c347719c378edbbe725eb5 : SUCCESS in 38m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3b3fd14bef774d06af25d347e4e7b2e4 : SUCCESS in 1h 00m 53s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c449d982eae44e89a4858da2fa5ca306 : SUCCESS in 1h 26m 41s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/7784f7d190444adaaa3e569db8eae613 : SUCCESS in 1h 44m 19s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/dcbb99880bfa4db88e9572947ae6acb5 : FAILURE in 1h 37m 30s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/17725ac827334cbbb5b89659c81f09bb : SUCCESS in 1h 10m 40s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/bf81d91e87884e288c82ed96b1359d60 : SUCCESS in 1h 20m 25s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/e31089835f3247e688b674eb43c144ef : SUCCESS in 26m 52s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/588146486047427aadbf526f94d26b21 : SUCCESS in 58m 24s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/ee5c196e5a7241fbb3b5b76b60f74e4e : SUCCESS in 59m 22s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/11494ba2da4e40d984458baa65ec8e8a : SUCCESS in 1h 11m 32s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b60351ce1900440d85d1548cd9b4a273 : SUCCESS in 36m 55s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1b1e4d403ccd4dfe877a650ed71f47d1 : SUCCESS in 17m 29s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/33065ec8e5d44920ac9e0ec5b7757431 : SUCCESS in 1h 02m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6a169781181d4b45b93d57cad6c2a898 : SUCCESS in 57m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/817fda9ac1ea42d7aa3da790eaf961b6 : SUCCESS in 1h 11m 06s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/b8e0cda4b3dd4da8a175607c7d25ea3b : SUCCESS in 1h 37m 58s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/77952fe5a7b44c28b204cb5b2cc849b4 : FAILURE in 1h 33m 51s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/ce2bc1ae81274f0dae23fc76928dafe7 : SUCCESS in 1h 00m 52s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/b2825312b7f94059a0a0d4e25627a0dd : SUCCESS in 1h 02m 17s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/0c881f1282fa4e588e926dfaa102d21b : SUCCESS in 23m 21s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/65a841306d304282bbfc302ad9685f82 : SUCCESS in 53m 33s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/dd8e977d347b492182c11258c352df2f : SUCCESS in 50m 57s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/b100652eb1cd4741aaf15a1353633540 : SUCCESS in 1h 05m 39s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a06bcb1355ae42bd944a20c1c9ba1035 : SUCCESS in 22m 36s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"07eb363eb083c1dee9c7cc6112a2d464e75ebb12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 14:04:41.000000000","message":"Patch Set 14:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1475710a33de4600ab351b674d8665c7\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4de14057bc9b45d6930a7bd9db6be359 : TIMED_OUT in 3h 02m 27s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e6d654ae84054a939a7749a0e53cadd5 : FAILURE in 1h 51m 47s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3bdcad64da3447a38217a90be2bc4c2e : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/aedf6eb9095d4a11b2173c1ae1857f41 : SUCCESS in 4m 42s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"bf586bba5f444adb534b00cab5594252d916bebb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-27 14:47:38.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"1ed88c5c84c0bf12cbd97056b63e328b0b368fd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 17:06:25.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15ac9808c13b447aaea7e174a88b70d7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/752a88d98009472d891ca9f508773c80 : SUCCESS in 24m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/de5e8334aad6463eb2f1c82595cf2f48 : SUCCESS in 57m 52s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3a8a138c15cb4f808ca64ecdf383e05b : SUCCESS in 16m 49s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2da4cc3eb3554eb6aadffe5276d0ac8a : SUCCESS in 29m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1360552e526a472eb8fb1f276a597d3e : SUCCESS in 53m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/490f5819bcb84e1eb073af4e268c68b5 : SUCCESS in 1h 20m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36525846d42a404da2a6cfbbf774adf8 : SUCCESS in 2m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b44e8798cc74df68760173cc8c1a7cc : SUCCESS in 2m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/af982f30dea448f0be2d436ded777a17 : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7053e3d363914b5f93bc815864250f7c : SUCCESS in 3m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb4c8cdc60a549b4ba3ab1d55c3a9a00 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f51a778da4374681b361f244446da68e : SUCCESS in 4m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fff9093f3d64bab8cb4d36f6deed7a4 : SUCCESS in 3m 53s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ebcc545066984b22b44068a68b566352 : SUCCESS in 29m 21s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b785c4d7f295457cb9fa4a6be4690094 : SUCCESS in 35m 01s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c1d1089f871846c1a1b41e337a1a0c30 : SUCCESS in 1h 00m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/933342b8ad6f4f4f9f289d6ba9f6492f : SUCCESS in 1h 18m 48s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/dd1518f4afb64d83a720e00d6f829e72 : SUCCESS in 1h 47m 48s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a2b8312f334a43ff9d3b94df431ff603 : FAILURE in 1h 53m 01s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/72216348e7e9460dab3afb0c614f9bee : SUCCESS in 1h 11m 47s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/92e2796d51b245ada4903572a9b6e9d5 : SUCCESS in 1h 14m 27s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/b87d465052b44079a8c1cda92f29c4bd : SUCCESS in 26m 30s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/ee9ed8df58864935aff55a6b115c78a4 : SUCCESS in 54m 28s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/0475f7dea3424382bdab093cea112384 : SUCCESS in 1h 04m 21s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/b7f6ac37f7134aa58b92dc13ba1a05f8 : SUCCESS in 1h 25m 53s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/6c2c7746421045b39d56709d5e1e6a36 : SUCCESS in 36m 09s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/100242afafe44eab820efb3801c0a25a : SUCCESS in 18m 01s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/75a81e87c3e0473181691bc672550d08 : SUCCESS in 1h 12m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a4c5ae09b98344e68b14d4cc9adaf6ee : SUCCESS in 56m 51s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/665b0672bb4d41f595dc6a95790a8fb1 : SUCCESS in 1h 08m 48s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/9993271bb42d4c1698978bf77470e76b : SUCCESS in 1h 32m 58s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d48bc06a15cf49c1912cd623d89a609f : FAILURE in 1h 44m 31s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/434f0c674bce4317b6ac13c8bc15c1aa : SUCCESS in 56m 44s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/6aed328541174ddb951647f74664d1d0 : SUCCESS in 1h 00m 27s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/efb625ed86304205a793b65f289bc0df : SUCCESS in 26m 36s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/49afcba9f4b6474297df289942673df9 : SUCCESS in 44m 03s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/0db83eb87b7846cdaf89e91af8380d34 : SUCCESS in 57m 55s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/6228262565244d5a818c443e7df3ee88 : SUCCESS in 1h 14m 05s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/10613c90ca944fb986fef1203500ae34 : SUCCESS in 25m 08s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"e144382b04e869e360c4bcb0ea9242821b64e138","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 18:04:02.000000000","message":"Patch Set 14:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/927096 is needed.","accounts_in_message":[],"_revision_number":14},{"id":"23fca452a044631f4780708d5a32a3dc83f0f1ed","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-27 18:04:09.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"55a6bd4c316d9040d4c1533deb6bbcd30124bf9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 19:57:24.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b8f767b54c146b0964469768ad81d02\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5328c170ddf047e391597e093fa3d5f5 : SUCCESS in 28m 05s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/431ad51b3efe4a6487d1cd8f4af585ea : SUCCESS in 1h 00m 21s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4434c7f3b3a0470e9add30b5b25c6a3c : SUCCESS in 18m 23s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/aee2ff209bb242cc8cc53eb9bb047222 : SUCCESS in 28m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/87b3a73f757b42f1acac66742d86e737 : SUCCESS in 55m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7781278e6ab240609de347652e3c1aa1 : SUCCESS in 1h 30m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4894cc8bd1bf4949923a89af0342eb01 : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb6dd8c4b44741ccbe5a8d91e7151548 : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c16ff8666fd64c1eabf3d3ac56219448 : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/46099e013e7040fc96e6999791ad89b7 : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df3eed82e72f47d986f4183bf54c3b77 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c26ea7e2f8e04c4b9b7f58744ca1494b : SUCCESS in 4m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81c1b172d93d4f22a61e6f7a07b7617f : SUCCESS in 4m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/25f52b8c09c74061a6cb1bb01f3a6723 : SUCCESS in 26m 14s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/fc85bc51fa9a46238ca143758f9ac438 : SUCCESS in 37m 57s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/be3e60746d644ca3b5474df09a08a391 : SUCCESS in 1h 15m 55s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b2396e5984c74e2f901779295f7c89de : SUCCESS in 1h 27m 44s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/1eca006b997c40039d6d09ad81912ddb : SUCCESS in 1h 44m 24s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/03046e3dd4b544c395f317cc31de7b87 : FAILURE in 1h 45m 16s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/53cb460ee8c047b99a5e09caaa0ba455 : SUCCESS in 1h 21m 07s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/44f699abda5342af936dc76539cd06d4 : SUCCESS in 1h 31m 41s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/bf2bc8537b5d442c9ff309fc5318b6bd : FAILURE in 26m 46s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/e3c29815c93e4c4b8228bfa9abf3e01c : SUCCESS in 54m 52s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/2faeaa3b40014d8297a30d7d422dba18 : SUCCESS in 1h 13m 47s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/366648abef2c4afba504580fa337f887 : SUCCESS in 1h 18m 03s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ea8fcc359b2d48db8b3acfb08e2c5028 : SUCCESS in 32m 57s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2ed1eee30925449d88a7ebe012ee3c21 : SUCCESS in 23m 50s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b565c06281114e3e9d8de7bfb1e7bd69 : SUCCESS in 1h 02m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d9847e767a444b9990b5fce98d178b86 : SUCCESS in 48m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b60508f0a30a41f0a5e7bf62856a455a : SUCCESS in 1h 13m 28s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/06bbb18551124dc086fc23cd89138158 : SUCCESS in 1h 36m 38s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/80e8f931afe341b3b1d30f7073654927 : FAILURE in 1h 39m 43s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/f10d6a059be141469cc77791bc0b3896 : SUCCESS in 1h 02m 59s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/a5f6c9959468488c91532050ff4d20d6 : SUCCESS in 1h 09m 16s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/bf283e93e65b4e909952c16b41900b55 : FAILURE in 21m 58s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/9db321324c67479d9aad1ea02c29aeef : SUCCESS in 50m 16s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/66e3b7957415497d8706612a45f7d01f : SUCCESS in 1h 01m 47s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/b5f53df91cea4ee68dae4b9c15be891f : SUCCESS in 1h 02m 14s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/609d32c86afd40cfb6b2efabe0927bf8 : SUCCESS in 29m 46s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"04b37155ffb277749ca909794dc456e41d590f33","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2024-08-27 20:18:13.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"2949bbbef4bca3a53bc84089adc7a4cf6ed9e4da","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-09-02 08:48:36.000000000","message":"Patch Set 14: Backport-Candidate-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"550b15c353d92e501ffdb1a49fe67fef467236a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-08 08:10:40.000000000","message":"Patch Set 15: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"a7460283591df054505634efd5e5d3456f9f220f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-08 09:41:29.000000000","message":"Patch Set 15:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/928487 is needed.","accounts_in_message":[],"_revision_number":15},{"id":"0ac9bba6f4c6096fb499cbb14cace8757162b355","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-08 09:41:33.000000000","message":"Patch Set 15: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/928487 is needed.","accounts_in_message":[],"_revision_number":15},{"id":"43e24d18efba3506dde69aa84c8b433279884953","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-08 18:54:29.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"2e85a214161b7906fb6e9d3e489ef542b7096883","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-08 20:54:00.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0bd8a4ec40c48828149fa4f80a4c077\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/34af2ea01ea24abe8af5cb3e64e83ee4 : SUCCESS in 1h 57m 47s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/401ae0eea53a4ddc890f58a1740a007c : SUCCESS in 1h 20m 45s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/acd81561cb7e4c329df29be80d1c2cd1 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9e470355093e4217b8725f3bc2cf4b0f : SUCCESS in 4m 53s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"ce645e94043bfb73045e87043bb441e5187be42d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-08 21:14:24.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0f22ad17e874a0a8968d338ae7da828\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4c1105877da84938a7a78a0d30844ecd : SUCCESS in 24m 28s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3b6d164bb3f1444d8779889d8291f2c9 : SUCCESS in 50m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/520db31d0b0a4e3683abfeee5512083f : SUCCESS in 16m 59s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ed437215f43b428b9440c987d8105f05 : SUCCESS in 28m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/25b18d1eff5d4302a51e27fbfd0dabed : SUCCESS in 53m 32s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f3e27bbfd7ad47c8820febc7142505d9 : SUCCESS in 1h 23m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cd4cb30f6ee4801960ccf0bb5562821 : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16b0ba8ebe55453098c3efee693c91f4 : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/67fb7cf7b0d5409f84f6e3d782f9894b : SUCCESS in 2m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/643587c91d584cd59eba6e2a4463bc4c : SUCCESS in 2m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1efd544a5c844b809d6b6740573a57c1 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed5a8c1246b14db4bee334c7d3d9244b : SUCCESS in 4m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17d3b32e6241447aa6f6ff3625d73576 : SUCCESS in 3m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b7881cf721b44802a1559d4ddeb54140 : SUCCESS in 29m 46s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5abbaee6f9d4491ea62b181480112897 : SUCCESS in 30m 53s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5c8cd72f62004f2ca66729e2aef847f3 : SUCCESS in 1h 12m 09s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/20c72d5c86944dfcb8a07ed40ebdde12 : SUCCESS in 1h 24m 52s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/297b109c1ac14fe59949a81f612149f1 : SUCCESS in 1h 40m 36s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/867ec59ecf5f43ac9dd219fa7c5c648b : FAILURE in 2h 06m 29s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/8a868fcfde734fe493ddf23fa4cc44e0 : SUCCESS in 1h 19m 20s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/7297ab75eb934c889fbc0f9f2600c369 : SUCCESS in 1h 07m 20s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/f8684ef45b77408da59fff864fd5bb49 : FAILURE in 25m 02s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/e85c407f29484700817bee2f76a60e8b : SUCCESS in 54m 05s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/7fdf456ee0ca487ab5f497905a64b57e : SUCCESS in 54m 13s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/98fa6d19b1f64194ba4098e629f23398 : SUCCESS in 1h 09m 33s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/743e16803d754f0797970a89b7f94c30 : SUCCESS in 27m 17s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/03fe906c744143d19d0cd5dd34dbaef6 : SUCCESS in 21m 31s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5a30a7d8192a4dee84b47b4fe4642be8 : SUCCESS in 47m 50s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/80eecd8d2cf5412cae7a9179c03e59f8 : SUCCESS in 47m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d4d67785def147ee8d8b88ed02ed6989 : SUCCESS in 1h 04m 13s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/8ca7ea5d9d2749bea23a13126ac04090 : SUCCESS in 1h 14m 34s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/69afc3e9fc7d4d32b37b99e9b6d31c24 : FAILURE in 1h 18m 21s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/1637bf873e1f49f084694f201025ca64 : SUCCESS in 52m 01s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/b6700766024d4b1c9cb8d64402ae58b9 : SUCCESS in 58m 29s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/31db15a654d1416c9cca830b3e01ec1c : FAILURE in 19m 40s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/9ff82e318f1b4676b438093c13d7b027 : SUCCESS in 40m 23s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/7f0f663a4e09402399b684be6a2af15f : SUCCESS in 56m 11s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/10371e5cc5ec4ce78ec2d2e7517fa805 : SUCCESS in 57m 34s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/617117d7c74a401a9c109b482e773d29 : SUCCESS in 28m 54s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"81b4023cec43137422e5464314bbd62496355e62","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2024-09-08 21:43:12.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"a857be2f57db81505fb40bdcc049c4239deec55f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-09 05:33:50.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"7ed82e25ea1f66ba1230bf2b7fd3d0095d938aec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 07:11:19.000000000","message":"Patch Set 15:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/928487 is needed.","accounts_in_message":[],"_revision_number":15},{"id":"c5ce49a3bfb70a1c8d510fd0069f8289f1ed15dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 07:11:22.000000000","message":"Patch Set 15: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/928487 is needed.","accounts_in_message":[],"_revision_number":15},{"id":"04e2ba310ce8006ab9a73d3607ce63be5b97e509","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-09 07:13:29.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"2528bf40601dff55007c1005e6ac90727121ef4b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 08:10:47.000000000","message":"Patch Set 15:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/928487 is needed.","accounts_in_message":[],"_revision_number":15},{"id":"02adadeccba4dc6efbac477d6cbdf1d316733b37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 08:10:50.000000000","message":"Patch Set 15:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/928487 is needed.","accounts_in_message":[],"_revision_number":15},{"id":"fbdc94a1baae0bb7bafeb1d5dac59e5caa19e341","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-09 08:15:22.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"214a05ded2c6ca8745c408a61be5fba3facc05f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 09:56:39.000000000","message":"Patch Set 15:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/928487 is needed.","accounts_in_message":[],"_revision_number":15},{"id":"817d4127500d58e327fdf479301570fc73bba52b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 09:56:43.000000000","message":"Patch Set 15:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/928487 is needed.","accounts_in_message":[],"_revision_number":15},{"id":"6107d27468a8f799050ee467dd1bb8691e623233","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-09 09:57:53.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"3168e576aeff7d73719f3bd3d1b50f509fedfa2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 12:08:56.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5005e95abf5411dacd7e10307f0f5eb\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/948bf0ba35f042c5b26631107e823fe7 : SUCCESS in 26m 43s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/276c46ca0d76498d9cdcee204554b53b : SUCCESS in 1h 01m 44s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3d216ed4fa664d7cb3d322e1c9047c2a : SUCCESS in 18m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/70f1f633931f4108829f32f8e9ed8ab4 : SUCCESS in 33m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3cc35a7258024b9ea346512482188d68 : SUCCESS in 55m 19s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b6cc59dff62c4e39832e1a487331ab6e : SUCCESS in 1h 27m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/419825ba40824da6b96fcd1cc0bd0729 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e3c2b0e4bc1462f9a4989c2ca2f2f0d : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e37592a77d2c47e6bf7ee69fe90e69da : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/434cf028138b48a6b6967de036456bed : SUCCESS in 2m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32c3d1d5b3e94514825536012ee83ab5 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fe8de54eb6a4c51bded590020dda8fa : SUCCESS in 3m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd493104e4514501bb2e64a7ff8af2c6 : SUCCESS in 4m 24s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1444c50a1cc64ad3b4874568edc336f0 : SUCCESS in 27m 27s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/44d59fe7280243a09140d96536cc8c90 : SUCCESS in 35m 17s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/806f0957c3a54c858b9cdb02bb1c3456 : SUCCESS in 1h 11m 17s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/50175fda1b6f4a49b1252f089506f545 : SUCCESS in 1h 17m 41s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/964f17f2ef52488e9f90bcb77d1a3d21 : SUCCESS in 1h 38m 52s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c873fdeaf2654d3882b75dfb267e3972 : FAILURE in 1h 33m 36s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/2c6f16df16e243aaaafd7bd93a5fbadb : SUCCESS in 1h 14m 33s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/cd145d5df226454f8394f6a2b949c446 : SUCCESS in 1h 14m 39s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/bef2f2dac8c1407f87e913d4101218e0 : SUCCESS in 26m 43s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/a81b333ad3ad406e86530f8220cab352 : SUCCESS in 59m 13s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/2527616743924cb2bcd6e9d28a9b3ec6 : SUCCESS in 1h 06m 56s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/93f86913d5f0471b809cb798cabc4b5e : SUCCESS in 1h 15m 30s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8d84fcd79f8c4295a63080f703b6f419 : SUCCESS in 29m 48s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/63f61863bbaa4aacbf3152e04582feeb : SUCCESS in 18m 28s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c4d502d9f0294013a006e616f2bddb60 : SUCCESS in 1h 06m 20s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6d73890640974098ad43947a4aad549d : SUCCESS in 56m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/73c8e2e933b448f18c8fa25b6278ef4b : SUCCESS in 1h 18m 16s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/2f4577bc8bfb44268c4840b0fa03f29e : SUCCESS in 1h 32m 05s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ec0ed5097bf14c199471f1eeba3f6449 : FAILURE in 1h 39m 40s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/dc09fffb5c3e4310aea965cb0068a67f : SUCCESS in 1h 01m 12s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/caffb57f0d6c4a52ae218ae395a51de2 : SUCCESS in 59m 17s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/884ae0e65d9f4ee887a8f2e41b5284be : SUCCESS in 25m 36s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/86b5ffadf31d4ac9af2ffd09e0b23695 : SUCCESS in 50m 35s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/8bbff93ff6c644f8a08df60f84b99b23 : SUCCESS in 59m 37s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/b087aa6e646b48009087cc9fc1fbf465 : SUCCESS in 1h 05m 47s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c698835488e846fba3f5b63022c85fbe : SUCCESS in 23m 17s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"7247908bc06383e66b0fc0b39e1ea092162ae2a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 14:43:03.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39b83669f141423098f73a2cb4e7f94f\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2984c250f62b46eba164f7a44313d393 : TIMED_OUT in 3h 06m 42s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cd93738be7794c0babd17279bb62c03a : SUCCESS in 2h 00m 18s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4162d33f56614d90a3779d40d060e8d1 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/db8e41869a774364a1498d48e17ace07 : SUCCESS in 4m 36s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"4457eb390a864e7e78fc776c0d16267d8d64b08b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-09 14:50:17.000000000","message":"Patch Set 15: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"d71eb217597e3a4f6da8562646613d2c88b83e7f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-11 16:27:42.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"f2c884aaf7476db32a6fb7481582615f5e8515b4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-27 16:23:00.000000000","message":"Patch Set 16: Patch Set 15 was rebased\n\nThe following files contain Git conflicts:\n* docker/mariadb/mariadb-base/Dockerfile.j2\n* kolla/template/repos.yaml\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":16},{"id":"ef6c8dfaffaee2f1ae76db97f5a0f9eb3f310bbb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-27 16:24:01.000000000","message":"Patch Set 17: Published edit on patch set 16.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":17},{"id":"cbe08eba326bb29b9d1bd0d10c82f6a5d401ac19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 18:19:22.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/70073ee00ffb4c68a43fb5ea5b7888d9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4cd192dbf05e4368abdc01a103556f30 : SUCCESS in 27m 43s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/255712dccb5245f6ba45b43b7baccb31 : SUCCESS in 1h 05m 32s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ba04f8f6c9e54f61b0c6293f2f475f23 : SUCCESS in 27m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/85ef2ba559d645f49d8fc28771617a23 : SUCCESS in 30m 50s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ff1d3f989af341cdbaebfcdfa38c57f0 : SUCCESS in 1h 07m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a3ade65d6bac43cc8f3de7fe072f702d : SUCCESS in 1h 24m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8bacd3fb9e340e584693b1231b5dbe9 : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47f017ecb2c94c91ab6275f9481578d3 : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f77e826a57e64f57821bed8b389970fa : SUCCESS in 4m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd53d79676234d81b184beb2c7a3ef05 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2abf4045dac42b781f04c421cd158ef : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe238606a7444772bb3b9413bbadc46a : SUCCESS in 3m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1083d5266a7048c3ae8d331dfaf6dff5 : SUCCESS in 28m 29s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c7b558cbc0434dc3a0bb8393922b2c0f : SUCCESS in 36m 28s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2081f78e4b8a44839cb29dc59925cad5 : SUCCESS in 1h 10m 01s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e7f576b0f99f4bc8b6075e40a889d119 : SUCCESS in 1h 30m 50s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/39aca48b8fdd4f1a9b579db24d46eca9 : FAILURE in 38m 46s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/265441ec2c1149cb968b489690ae5213 : FAILURE in 1h 39m 19s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/d29341009d694830a6a815dc32bea5ee : FAILURE in 27m 07s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/e2eb520c38e4438995812fb35b79dc1c : FAILURE in 33m 17s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/8ad66c5204f444ab886408c5db00bfbb : FAILURE in 21m 39s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/950de2e802bf425e9a4578bb08782364 : FAILURE in 35m 40s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/ba9b4839351849b296c94bbae24db1b6 : FAILURE in 27m 38s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/b4f12606f478412aaf490c6d1d41d4c8 : FAILURE in 30m 11s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b24373f53fff455193efd3fef826aece : FAILURE in 26m 04s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7594f00ba1f64c8b97012271948a8cd9 : FAILURE in 4m 38s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3b05dcae510c47f3a96549e93bc215c7 : SUCCESS in 27m 54s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/42a4cda192f34e8da538237aeaba1e0b : SUCCESS in 1h 02m 13s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e8f0d4eb9c204f6b87dda18e07a81ab9 : SUCCESS in 1h 15m 51s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/60a819bdb7af4053be38d3b4b30cd314 : FAILURE in 44m 07s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f905180b8f5f42fa86fdaa86ff720e30 : FAILURE in 1h 27m 09s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/880830d2a4154989aac1f8b71d221642 : FAILURE in 9m 22s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/2bb0c81c68014ac9a2ac4f164c14672b : FAILURE in 10m 02s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/3e6b992f77bc48d0be09bc58563494f1 : FAILURE in 9m 43s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/c8f8f8aa79a5457c8b3722f1fc3efef3 : FAILURE in 12m 55s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/a3e5a8f4a9264465b2c309f0af2a9ce1 : FAILURE in 25m 28s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/3e30d502a5d9412abed3a6baf80e2a0a : FAILURE in 32m 34s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c06b55b95d504f46940c9dec82e31bfc : FAILURE in 22m 24s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"a9ebc2b35e6fabd4710f8476b8c7cdedf3d62076","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 03:57:59.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9ccac55b4074845b9558dbad2e5c420\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a8621cd58cbd4d4383bdc55661f38522 : SUCCESS in 2h 35m 39s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8e9ef91416f34b97a25fd1e3ffb3aba0 : TIMED_OUT in 3h 02m 04s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b30ff1d67f5f43c99dbc4da1d4f00ae4 : TIMED_OUT in 3h 02m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b812242a20284c9da2b66f57d22c9830 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3f7523c9d3a8410191d82784734f176e : SUCCESS in 7m 57s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f301ace71d7b48e18a7bd80ca9e7b111 : SUCCESS in 2h 14m 26s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/117477f3dbb549bda4cf9521079d4d42 : TIMED_OUT in 3h 01m 21s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"53a0b0205a75f1f1d6d5f972427452c8cfe5122f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-03 12:54:48.000000000","message":"Abandoned\n\nBumping to 11.8 instead","accounts_in_message":[],"_revision_number":17},{"id":"a4fa1cc44389ec87906145640d38485b63adb1ad","tag":"autogenerated:gerrit:restore","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-09 09:34:18.000000000","message":"Restored","accounts_in_message":[],"_revision_number":17},{"id":"b74b198b2f185fad73221039d34b9ffc64e64ae2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-09-09 10:53:15.000000000","message":"Uploaded patch set 18.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"762bcae5a604fedc356cabddfe76d16e6c0aaaf4","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-09 11:33:14.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":18},{"id":"6a4ce641f1b1aaef3aebe66e55f2b7cb6c0f1fc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-09-09 12:34:46.000000000","message":"Uploaded patch set 19: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":19},{"id":"1e77aeaf3a54eae8791da7b408cc84c7c9f3ae42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 14:53:08.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2efa677f628e46beab51d3ef74c4998f\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/8d1ba5f10348452fbc836a28cc4dcae6 : SUCCESS in 19m 59s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/89276f6858814c189e6b10b0a736b688 : SUCCESS in 56m 12s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4c6b6b1222dc4767ac8f4b42ce21ccc4 : SUCCESS in 10m 45s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/cf6945df53764adfbeed7507684a276d : SUCCESS in 14m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ff56492629b5469e95770e5dc015a931 : SUCCESS in 44m 02s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/955d438a416742b283a723f4ac145e83 : SUCCESS in 1h 06m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfde4dd7871f4d568562d55151d9edad : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/920bfd08168944928fc50e3ab5f0a8ef : SUCCESS in 3m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4f12d64b9b4e411bb4fcc33e49ae2e2d : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cd2f89b5e2ef4941ba35804e08bedff9 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c2bee4973294e39b32fc1d1bb95dc38 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/478c34bba8a847d6962c579b8c3ac05c : SUCCESS in 7m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c48e0c7ef7e48a5898137a1b5dc99f7 : SUCCESS in 4m 05s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d2d4f951ad5d4478ac55d1f508be9f19 : SUCCESS in 3m 15s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7ca9e73c9541461183e8384f0b997252 : SUCCESS in 21m 37s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6542aa2d45614fd3b9114e6400c39666 : SUCCESS in 27m 22s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8d028b7cf4074276b5d0bfa392eec73f : SUCCESS in 53m 11s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b62abfbfd0c24a0898ed27ec52c38e52 : SUCCESS in 1h 17m 14s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/959f5dc2275442f48d19a33a38212c17 : SUCCESS in 1h 36m 41s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/4725cc184ae24f819f986c6ad86c2d98 : SUCCESS in 2h 16m 19s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/cfe4baa910464b12af9632a134bfc35e : SUCCESS in 1h 17m 48s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/dfb7a7fc460f4a22b71238b7bab98b56 : SUCCESS in 23m 15s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/20591e37154544c1929e3ae05212d48c : SUCCESS in 1h 03m 07s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/7cda48ca9db5492db838fe30413ba1c7 : SUCCESS in 50m 22s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/c10ecf42710e42a998504c18a404c36d : SUCCESS in 1h 19m 48s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/610e19bca6b54afe9a3abed7b2b0c083 : SUCCESS in 32m 15s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"216375855926c9a39835bc6f41869a449c81fb28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-09-09 15:03:23.000000000","message":"Uploaded patch set 20.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"b54b5a75c703b1298778ac240b73039e684d3bf4","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-09-09 15:03:56.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"be3476b9e7e813da17d519e1f614ea5b98ad9d94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 17:23:59.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1bb228ad077452bbb1657ff0b238a33\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/423b5a8ba8344c419bd60d39eaf927d6 : SUCCESS in 16m 12s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/b970762e51ff4d16b4712f77c04f6321 : SUCCESS in 1h 03m 37s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/769eba7301324fa1882d16d75ef3b216 : SUCCESS in 19m 30s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/46c1b175e6e64986abc2a7ea46568ba3 : SUCCESS in 17m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6e0167dcb4ac4fb18251e7dc8fef1504 : SUCCESS in 45m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9830c56609d844789e852270afa3d9d5 : SUCCESS in 1h 07m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfd482f676574275954b39186edee0f3 : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eba4e158b82446b398ecfa489474b1cb : SUCCESS in 4m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e7e7ed30389e4ca08a2e8d7a6fa6e6d7 : SUCCESS in 2m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a4327a1620fe415fba8000c22ee5d218 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d45b6fce6a84ad2b3a46e38395bdb8c : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8911848eeb3248499d59578a732ece20 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/148b338d8d0e43799b1a9353ab20299d : SUCCESS in 4m 30s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/24285f018f30408383b8c0692031f57c : SUCCESS in 3m 17s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e874e0899fa140dc9e8a80e0e9e0ec2f : SUCCESS in 19m 54s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f15fb9ce9371423385348f4d3f1e16f4 : POST_FAILURE in 19m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8805792c5d144339a849dee0e5430f94 : SUCCESS in 48m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c70803315f2d4e35959b767026ef00eb : SUCCESS in 1h 11m 30s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/4abf0aec2a844cbabd55d4e01754f74d : SUCCESS in 1h 38m 17s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/58c0488b84f14e73a88f44f48d68fb0d : SUCCESS in 2h 17m 56s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/bcec8d8bb78f4760b39fbdc142ed83bf : SUCCESS in 1h 09m 26s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/ba0065523a4d4b86ad153eb479f09073 : SUCCESS in 23m 05s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/6f81f3f7983542beb18313404ce6b050 : SUCCESS in 57m 46s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/3127871e844e4fc4ac2576fbd415477e : SUCCESS in 53m 15s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/21a881b989f9444992ffa7869424f723 : FAILURE in 1h 05m 56s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d29959fd08104577b2ed89da0e16eb15 : SUCCESS in 32m 22s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"32308b47a60e491fde652b632ee02a507bbe6a35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 17:57:38.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/927bd55daf5e43a0ab946d434174328a\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/1d6d1c2958bd4feb83000cd808f238cb : SUCCESS in 55m 41s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/f322ff3abfcd423c8a6d5e98ebc3bd39 : TIMED_OUT in 2h 33m 39s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cb2456747e7a4d1dadcd31e763229385 : SUCCESS in 53m 17s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d9a84403840e40af94a74b538cdd5687 : SUCCESS in 1h 02m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/974db00b44c24bfda9b415d1834a5695 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5d3c0a0aaf4a444aadbf6fb52d0c2021 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d3ba798da623474ba105ba2e02f034d9 : SUCCESS in 11m 37s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8cd0850cb7b04d6cb8c598e75d0da722 : SUCCESS in 32m 25s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"3d2283465e1e5658c38a718e770dcbe6daa5f0d3","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-09-09 18:09:54.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"388791fff7ade93af19dca63c683ddc29694f476","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 20:23:05.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f90cbacadd34f9681cd0785160299dc\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/6f5b2826d3364c31be9ea537ca9122fb : SUCCESS in 13m 24s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/650f2e70a3a14f05adc1dc0f6e562a6b : SUCCESS in 1h 16m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ed38af3c60634d3f8297edf81408b4e8 : SUCCESS in 16m 52s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/75cc873063c243e68468ff4bf31bc8a3 : SUCCESS in 14m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ed3ceebf14134e4198f54127b5834904 : SUCCESS in 44m 02s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/cff3200cc7fd4bf1ac46f05f8f427468 : SUCCESS in 1h 05m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d71514157e14470afb4291adfe11b4e : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d56f42de0e064a30a34726a1b85fea4f : SUCCESS in 2m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bfaff0669aea4deaa6e52bfe2514db7d : SUCCESS in 2m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/debc5d269fb2419086662fa1b0c23dc0 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/997d46624c0a4496b9f6c7fb07db450b : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a7a3dfced264a52a56c94d82b426cdd : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/647c1d9a7cbc4836a8e5ea781e6267a6 : SUCCESS in 3m 43s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c51bc81f265d42439e481e8ecb5eaa55 : SUCCESS in 2m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3625ed144e6f41acaa706c29ba52443d : SUCCESS in 22m 12s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/872881c029104866b7869ff81c1bbf71 : SUCCESS in 32m 53s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/962a843bd64f48f3be07401c883b89f1 : SUCCESS in 1h 00m 35s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1e296167386b408eaf54aeaedf24ec2a : SUCCESS in 1h 26m 13s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/21b631249efd4b86bf331997d13ce859 : SUCCESS in 1h 47m 22s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/18cf7d803f054ee9980325652f408e78 : SUCCESS in 2h 12m 16s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/2fd268dc39954ba29a88bc5491fb423a : SUCCESS in 1h 20m 27s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/ad409516f3cc46d681f2a49e483d93a9 : SUCCESS in 27m 17s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/4f157c5c52474a2c8d3981117b3b8f4b : SUCCESS in 27m 47s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/8a6b117634c84a16abaffcf361e3cf40 : SUCCESS in 1h 03m 05s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/6aa7f2e2a6b94539a05b524494d8c9a5 : SUCCESS in 1h 26m 58s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/7384d3a24d5447b6a09f5e30857df051 : SUCCESS in 31m 32s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"952a7d9eff87693329b393f8ca2fa1fd9d929284","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 22:49:52.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29c251e9940c46a598e56ee31d738b0f\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/cfa6bc662c2549618466dc10f8cc53c0 : SUCCESS in 55m 22s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/1f44af9215bc464fa8470bb139bb3b0b : SUCCESS in 1h 04m 38s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c96cbb427b5640dda0eb7f1e44c17384 : SUCCESS in 26m 41s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9bb7d3972bb441e8bb81397f3d391880 : SUCCESS in 55m 16s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/90f95ed88a22474baecef342ac9f8cad : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bccbba9ecb424335ac8e40d62f584976 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3de40309ae1b47f9bad9d96cd084812f : SUCCESS in 9m 28s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0d5565f34b43473eaafa854c90229838 : SUCCESS in 1h 01m 09s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"16939384177369768fa3dbd41d032087cb2493c5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-17 14:29:55.000000000","message":"Patch Set 20: Review-Priority+2","accounts_in_message":[],"_revision_number":20},{"id":"ed10a216dcf99b4381d9978e40fcc3ae85857eaa","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-09-19 09:43:53.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"d257f69940e0fd70fed3214150123219b8a3fd75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 12:00:46.000000000","message":"Patch Set 20:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37cea7b7bd8b437a8ef2375a4efb90f5\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/bc4cffbe3a7943e292bf2f7ef56f675f : FAILURE in 12m 59s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/61a5ff1c1d6049f9816c5bd9812d1f37 : FAILURE in 14m 49s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/33fb7b4b7c0e4fc2a623ad491fdc32d2 : SUCCESS in 20m 03s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/374606f919b543dfbf778d9859873d5f : SUCCESS in 29m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/48ea77251cad4f1db23b25bd3c4d5cce : SUCCESS in 46m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5702a0e58c9f42699c264d2a94823c28 : SUCCESS in 1h 07m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77028d34d31a4d3fa7f05cf486b0d5ad : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5dba1eda9b5a4ecaa7720f7d0106d4bb : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/df844fe15c9d4a79aae7bf76c1c0c266 : SUCCESS in 4m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/04a7c81256644ac2b9f1a792a826d41f : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/75d3f0bf60b4474f93f33dea27764444 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4156bc52c6ae4d82b10796836acec211 : SUCCESS in 7m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/324dfa20d8c74bd682dc707f8fd200a1 : SUCCESS in 3m 21s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/9efaad17528b4566a22e584d8f0e9311 : SUCCESS in 2m 56s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9b7913c7637b4a19a37053eed232396b : SUCCESS in 21m 33s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1ad030b5a7fb459ea36ad2b5bd3ab865 : SUCCESS in 22m 50s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1a94e77baf334671bc5fcceff449554a : SUCCESS in 49m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c0961ba955d946c489c124c7fde4f347 : SUCCESS in 1h 15m 07s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/e30082df180a400b894f36e65e281937 : SUCCESS in 1h 31m 43s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/602b7c2c4a824039a8206a25f44b4779 : SUCCESS in 2h 15m 00s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/85df433ffafa42c1a2aa7f15ae3468bb : SUCCESS in 1h 09m 58s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/0142155c38e84f229abca4d4858e2c36 : SUCCESS in 22m 01s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/841262c8851f4ccb960642232e5975aa : SUCCESS in 1h 04m 19s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/b4903ce5facd4f6f85ee77b391e03230 : SUCCESS in 50m 42s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/37de86b418bf48e0843999e8282fde62 : SUCCESS in 1h 18m 43s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f0bfe188010f4c08b918e5c1061fba34 : SUCCESS in 29m 37s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"3ab4c2b1272a1ba9ffb4fec05fb165f9ae7b1b3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 12:12:03.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0f45cf0cd214fda8ffbd0b90ebe041a\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/9269ef4a0dc34544b810cf047bf4840a : FAILURE in 23m 56s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/eb13a441021148b188713ae64734b144 : FAILURE in 27m 27s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f24309c5bdd14b3aa3f0659f000f808a : SUCCESS in 38m 45s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7fa3d3e0db004a6c97c29d673b8d2baa : SUCCESS in 1h 11m 45s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9227f91ee0ac40288522c21d24479e1d : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/99b36846d9b9486ab9ec603e96545026 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8e8b645ce5954b31aaa2d5d73fd9c8e1 : SUCCESS in 5m 39s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/24f8cdb773d84ad98fcc3c232d313b77 : SUCCESS in 24m 38s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"aead6a187a809e1874a9a409ad44ff1b7c3463cc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-19 12:18:31.000000000","message":"Patch Set 20: Code-Review+1","accounts_in_message":[],"_revision_number":20},{"id":"818932d13b0d657e2b79f96b9d877f1fb478a8cf","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-09-19 12:46:38.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"ef17af9cf490126c14230de746c06cfc8c7f80c8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-19 12:49:18.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"dc4c453f9086df09079c5ce3fe7de8c33a3fcdd7","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-09-24 11:09:20.000000000","message":"Patch Set 20: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":20},{"id":"455736551291b1f9a5626876c675858476978b24","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-09-24 11:11:23.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"2f8769ec425bbd5198ad957006ed67be87b3add5","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-09-24 11:12:02.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"80cb7f57dd641e09fb000b8089ca1d86052a76c3","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-09-24 13:45:02.000000000","message":"Patch Set 20:\n\n(2 comments)","accounts_in_message":[],"_revision_number":20},{"id":"fd5b3902d28cf652d551137903aa7083c2de1db8","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-09-24 14:05:30.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"cff02c59091d7535eedf0e3120cf5c20f7b08221","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-09-24 14:06:17.000000000","message":"Uploaded patch set 21.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"a2da231ea0328da6346639b272d21dcd294d6f40","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-24 14:14:07.000000000","message":"Patch Set 21: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"270d4b8cee4e471bc6233f3d7429a5ef61daba25","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-09-24 14:33:19.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"68bfb8d5db8b046f351a96e64510ab00c064decb","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-09-24 14:40:33.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"4bd607e77229327dd18d9ed69f0103557a7ba842","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-24 14:46:18.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"b0255e898f24b33bb8383f4d300ea624f30ace9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-09-24 14:51:54.000000000","message":"Uploaded patch set 22.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":22},{"id":"e88388664bd4d1a189149a8d39e3cf07d8a1cf28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 15:27:42.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/280d138605ab4344b0f17f1e778dcf97\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/d6ff075afb9d4062a6401fceedc23950 : SUCCESS in 18m 41s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/faf86483dff2423da32d337200474f8b : SUCCESS in 34m 50s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c4ad9bd5c06444dd81be218058ee6ade : SUCCESS in 9m 56s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/12d76f28b28140ddaa7cdfc951a0943e : SUCCESS in 20m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f65fc0c5b5543ffb88e6ae322b57470 : FAILURE in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/adf25a7624734bc6a22fdf27a54e47ac : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4e7302051cba4e72bb24d8c8975d6efd : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/78b4c3f38a2d431aaf28ebc635e9feed : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d8148d56fd54ff4a69de3f9c03e2d32 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b14d40541f1477bb6bd13c05d433775 : SUCCESS in 3m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d2d4b8f584c4969beadd44357b79c30 : SUCCESS in 2m 21s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/9eae6c68b45b4b738cdab13fad82ecef : SUCCESS in 3m 07s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d813f9d47c1f4581ada0da496497248d : SUCCESS in 24m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6f0c69c0f8dd44e0927e90e06e540ef9 : SUCCESS in 29m 08s","accounts_in_message":[],"_revision_number":22},{"id":"73fdb1c85658166b7489f424d403f7cfbe0563dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-24 19:06:03.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c84401f7d84844629b2fd7218bd22529\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/7f0f0d9e892640408813efec8f47c976 : SUCCESS in 36m 56s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/b05c91cc684544c19f13ef2cd8da0c7f : SUCCESS in 2h 23m 04s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/68a25540a3b84aa18eb644b4e95dcdd5 : SUCCESS in 26m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c91271c6012d4074a7c1d5fb8ac1a1b8 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0d2990a72116425c8493c181417678b5 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e3905b85ff3740679c52586e97a65a05 : SUCCESS in 9m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/db404527fd0b446c855005abc1dee1f1 : SUCCESS in 50m 30s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"9659b2c0553427523c282099207ca19ecc8dec16","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-09-25 09:19:33.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"4201573acf30efb8457c8242c36b13ca61650115","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-09-26 13:34:08.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"8e1360afa68c5cf924bc8b10eb826ab7501d4ac4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-09-29 10:33:42.000000000","message":"Uploaded patch set 23.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"657b218d77ac93b7555651e5174a6051fb18dc6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-09-29 12:15:29.000000000","message":"Uploaded patch set 24: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n* Review-Priority+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":24},{"id":"c44ba5b4e07d0b8d7d07a9e7dd2c0318d63605f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 13:53:12.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4fd9db5fae5d46c79bda688595b04fc3\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/d78111f30edf442d91a79f99f3ae8cea : SUCCESS in 10m 45s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/c8c1a91360a0477f8ec603ac26be3cd7 : SUCCESS in 21m 24s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/e56013d3e13b4ae697679cc72e271f61 : SUCCESS in 29m 40s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16a90fa849494503a7d1fde32c919e26 : SUCCESS in 22m 49s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a61cdb9474c24c3c8cb3fe06b3ba6007 : SUCCESS in 29m 36s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/619ef7584f1b43368206349844cbb68d : FAILURE in 10m 04s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/2794c8c2cf4e4330889009def02e0b0f : FAILURE in 15m 46s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/806e9e1748d143ddb061798d735855d9 : FAILURE in 54m 05s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/339a8cea52814859b24433ec9330b1f9 : FAILURE in 27m 48s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/dd6b7beb0d624a7188839aed81fcffc2 : FAILURE in 57m 13s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/de386b50de9542e488345eb26db13447 : FAILURE in 25m 06s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/56f7d86cb4b44dd2b2d6c76eb746bf73 : FAILURE in 1h 31m 44s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/f3917f14540a4bebac4536979c1ebe65 : FAILURE in 36m 09s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/a5fb4d828e684e37b15b2cc1d551edf7 : FAILURE in 1h 35m 45s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/b459a21685234b9298a5271730e5fb91 : FAILURE in 21m 07s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/23dc6ce8e14944ef95c9d768a234cd6b : FAILURE in 20m 11s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/f781ac55734e420f882d9f4b45e43276 : FAILURE in 18m 41s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/4de3ec5e8d59440194c22a35d8d5f3fe : FAILURE in 15m 12s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/b297886984da4c82810908487a021b79 : FAILURE in 11m 56s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/93bf8bea305c4977814bc2f44d98a7fb : FAILURE in 19m 53s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/49af55f4e83f4c9082aa51d114bbdebc : FAILURE in 16m 33s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/2d11f2ec511a4ffa9a8590735cbfb707 : FAILURE in 15m 31s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/c02e3345d0a54aaf9fcdcf09d5f2ae21 : FAILURE in 23m 34s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/41a1c5ba00b442ca8536ecafb5c2a882 : FAILURE in 22m 40s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d330cb4df60b45ba864067cd41c3bcb0 : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/628152e0ae8847cf94379392a16ffdb6 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7f09e38f48584d3eaac687750b94db8b : SUCCESS in 4m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/98958e7cc9044ec7b407cab58183f001 : SUCCESS in 2m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/95210dcda1274236baf779407e92aabd : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc03aaeb24414c6790fae4cd175f144d : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8ab33c405aa40bfa1e88efb828502b1 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c455c4ef92e1422b90e06e5e4a753ae9 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":24},{"id":"0f4bff4ec8116f9928e08631bfe292be28c0924b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 13:54:39.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84b3a419ab494d959c0e10decac8d658\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/173d42ed19384fe6b5852a8918d0e58f : SUCCESS in 48m 51s (non-voting)\n- kolla-build-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/6afb392fbcca4879a2f8c72497078e77 : SUCCESS in 24m 48s (non-voting)\n- kolla-build-ubuntu-noble-aarch64 https://zuul.opendev.org/t/openstack/build/102f6edfdeda497e8e6adbd647cf34ef : SUCCESS in 1h 00m 13s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/0f26536bc6634ae783865f83eb890891 : FAILURE in 51m 06s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/a0467747686f4fbea42e9c15511e2926 : FAILURE in 50m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/85d7da8902704457848d4229158de2d2 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/42e2c7721c3947c992bd697cb8c6437e : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f97587ef65e14fefb8fb6cc4fba33d6a : SUCCESS in 6m 52s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"479950ce165c1e7323ea17852d778b2ede4990e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-09-30 14:19:19.000000000","message":"Uploaded patch set 25: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n* Review-Priority+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"b71a7c4aaf7de0e6a3abd23d4f1df6a01171c900","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 16:47:48.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06fa9c94c7ab40d296b3a0e6ee4cf231\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/1c61f2bad5934fe5bef9b4088c03782f : SUCCESS in 30m 53s (non-voting)\n- kolla-build-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/af30875e0a7443b78a44a1573e83b389 : SUCCESS in 1h 10m 35s (non-voting)\n- kolla-build-ubuntu-noble-aarch64 https://zuul.opendev.org/t/openstack/build/0c4ffcc953074579b513ba057c51ab89 : SUCCESS in 1h 23m 02s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/e11fef3dfa334140ac0cba0e3e061c86 : SUCCESS in 2h 22m 51s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/56a11e637c8f443aaf4b257b40f8fde7 : FAILURE in 1h 52m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1e2011c8daa14251b00044db1641051e : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/19d9dc4486bc4983bb3d54bf99d5b29c : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/80d3034658894793a7e062a90e6307e0 : SUCCESS in 10m 51s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"33ffce022fc01d379372ce221be295e3e6ebd95d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 16:58:12.000000000","message":"Patch Set 25: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f8e1937662c442698d1b68328f515d50\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/a071daad18ca4d0384a18548ff72fdd2 : SUCCESS in 17m 54s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/7c35aca0209042fcb8db61d119a4220f : SUCCESS in 21m 15s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/da2e423dbb404a5aaad3817b3a7fe120 : SUCCESS in 28m 56s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/741b40534d6340f38abd97c9dec27a0e : SUCCESS in 20m 06s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/269fc475aa2049bd9daf007fa8464f4b : SUCCESS in 23m 52s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/32124d3a0a804f7cb63132a3649df36c : SUCCESS in 33m 55s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/cded2449cc834cb993334b8acb2e6a62 : FAILURE in 50m 01s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/db03862348214d578cdf939eba7af3ae : FAILURE in 1h 03m 12s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/70753dd83c774f47bd70e74611e0a920 : FAILURE in 39m 31s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/bc6e250e780f44e38e9727b4967597fd : FAILURE in 1h 23m 51s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/ea1c9738e0194bdf9e981952ee7a699c : FAILURE in 1h 23m 13s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/10459611bc9247bca6285ed79decaab8 : FAILURE in 2h 35m 58s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/658320f489c045458464d7fb4e4a96d3 : FAILURE in 1h 32m 15s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/190329662b3549cda5dfced5dd114d91 : FAILURE in 2h 37m 34s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/fd887ec8406740b98109a9e2b7ba917d : FAILURE in 1h 02m 12s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/af04c5f4a9ba44f5a31668d61c3c2c71 : FAILURE in 1h 07m 12s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/9b9a4d735aeb4e6986949a15bf48ed0d : FAILURE in 21m 43s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/bddaab3f5b9d45dc950a9a6ae85d91f9 : FAILURE in 24m 29s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/29fea312a5294f40970117d7a756b580 : FAILURE in 32m 42s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/846f0529f24844149f3452e94b594b72 : FAILURE in 36m 11s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/482ab392c0c248f6ba9e49bf72e907d7 : FAILURE in 45m 13s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/313ec48e56b346aeb2e5599dac91425c : FAILURE in 41m 56s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/53a1d0e363d0472ba64e3985c4078aed : FAILURE in 1h 11m 31s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/4e78a515c3574e4fa9b51a9f1ed9c9c3 : FAILURE in 1h 11m 31s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/96d7039728cd4b929d34d7af06c5a9ac : SUCCESS in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5199450146964ce7b6feecff83f8f483 : SUCCESS in 2m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15569887a4254e95b2434b063610b915 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/41fa7995028a4b8397184870f7770162 : SUCCESS in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9021b304cbf74f7a81fbb22f20f195e5 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1de4e1870374ba1897bd438d1ea2c24 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf3835484c4a4b53b3b80350ce904d2e : SUCCESS in 3m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4dbb239883343619d924ce7ed343223 : SUCCESS in 3m 27s","accounts_in_message":[],"_revision_number":25},{"id":"e347bdba787201e9c646e0f1f9a6df95180fd139","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-10-01 09:12:34.000000000","message":"Patch Set 25:\n\n(2 comments)","accounts_in_message":[],"_revision_number":25},{"id":"c49cbc19b0c7298c7d72c7a6d3cf1928f5d9f4c5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-01 09:58:47.000000000","message":"Patch Set 25: Code-Review-1 Review-Priority+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":25},{"id":"8c2289d5c8f1f13830802dd6d053983f06d7c58c","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-10-01 10:15:30.000000000","message":"Patch Set 25:\n\n(2 comments)","accounts_in_message":[],"_revision_number":25},{"id":"d1f0ca489824510a30e19c2d1fa77db51febe61f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-10-01 10:26:47.000000000","message":"Uploaded patch set 26.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":26},{"id":"3ba6cf3f7c2f58b9bd3a045533005e893b144c70","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-10-01 10:30:06.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"ff2de2bb5424996e748534a8ae38d4e8a87d5936","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-10-01 11:15:18.000000000","message":"Patch Set 26:\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"e3b4e2bc75d823f45260d09e0018dcd38d755809","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 12:31:02.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41a2c5d7529b4acd84d3598b8b9ce667\n\n- kolla-build-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/02fe632fa6104e4b886c2c8e18096b03 : SUCCESS in 46m 34s (non-voting)\n- kolla-build-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/c72026210de346d0bd7a50370ac49387 : SUCCESS in 45m 53s (non-voting)\n- kolla-build-ubuntu-noble-aarch64 https://zuul.opendev.org/t/openstack/build/49c61e718fc24906a2832d88adc9ff14 : SUCCESS in 51m 36s (non-voting)\n- kolla-ansible-centos-10s-aarch64 https://zuul.opendev.org/t/openstack/build/01ead5c7f1b04fc4a0b482f0c4af7222 : SUCCESS in 1h 09m 28s (non-voting)\n- kolla-ansible-debian-bookworm-aarch64 https://zuul.opendev.org/t/openstack/build/f638098505ca4f9f8ce7a04370f51d58 : SUCCESS in 58m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/713205cdafa648d198776b0ec9539244 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/114b1f0c6517456ab9612649308c6d18 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/da8aaa9bbc224f16974674076c629f1d : SUCCESS in 10m 38s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"cb2eb96dc41e2d64cf2e32bf0c56e51910fae0b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-01 13:03:54.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6b6cb7e8ae242eaa85dfde6336b8b86\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/39a2649f95b84fd6a68cb76a513e4ccf : SUCCESS in 17m 38s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/f63c530ae4bf4a8d9bd2144ca0635af4 : SUCCESS in 21m 49s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/15ab85e0c4d14099b1ce506e313a6453 : SUCCESS in 15m 50s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c12f6b8ac5ef4185b2f4efbcecead051 : SUCCESS in 11m 28s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/cb44a15dc6944d4cbdb477c002ff1916 : SUCCESS in 21m 10s\n- kolla-ansible-centos-10s https://zuul.opendev.org/t/openstack/build/4b3d3467b830463bab34ee3356dd3c00 : SUCCESS in 35m 04s (non-voting)\n- kolla-ansible-debian-bookworm https://zuul.opendev.org/t/openstack/build/43bba32fdd0f4cfeaad7e04014b3a033 : SUCCESS in 51m 49s\n- kolla-ansible-debian-bookworm-upgrade https://zuul.opendev.org/t/openstack/build/35f6f95718ea487bb89760c1cf4de895 : SUCCESS in 1h 11m 02s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5bc4366447664743bf3dc6f82e00d5c6 : SUCCESS in 57m 41s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/49ed7448e6a94231bcd8d112ece74a93 : SUCCESS in 1h 17m 35s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/755c7ed00c0a4cb98f7fda9179cefc11 : SUCCESS in 1h 35m 28s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/4cd07d6d20814897b46b102f924fa61d : SUCCESS in 2h 25m 08s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/7b533f083f03477189378b7e1b12b89b : SUCCESS in 1h 49m 27s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/dfa04ca41607462486064e433a85ba32 : SUCCESS in 2h 35m 37s (non-voting)\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/f8ef75916a5f42c786a6956bb63d333b : SUCCESS in 56m 47s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/2278d115475d41a5bf9d4513bcb2d52c : FAILURE in 1h 14m 26s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/cfa0386a256444faa0697228678a0b1f : SUCCESS in 26m 48s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/430bcef3a8fe473f85dd2714759e6d0c : SUCCESS in 16m 29s\n- kolla-ansible-debian-bookworm-masakari https://zuul.opendev.org/t/openstack/build/7feac45d6ccf406fb6f2afecc2cad27d : SUCCESS in 51m 44s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/8895789704d84be884d1a235a3a1abcc : SUCCESS in 31m 22s (non-voting)\n- kolla-ansible-debian-bookworm-octavia https://zuul.opendev.org/t/openstack/build/be7fb12dd1604e1d93c12a389240332e : FAILURE in 6m 13s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/d3889388c0494334a70c7afc98fe3525 : SUCCESS in 59m 15s (non-voting)\n- kolla-ansible-debian-bookworm-ovn https://zuul.opendev.org/t/openstack/build/4504118e9d314d3aae2590d636b1a849 : FAILURE in 1h 08m 44s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/793c0d8410d04b18aa210f2628f06086 : SUCCESS in 1h 28m 16s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8783b7f6531b471093077eca4839d54e : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd6f21ee0a6e4bf3aafff8ee74c4b86b : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26fbd2893e4449d9bba3894703e593bc : SUCCESS in 3m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a67e136eb92a48f8b850a1b6099a59f6 : SUCCESS in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7c68efb5b9954bd2a47ce137a7f65137 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf51dff37f4840b69eddb857f5defa62 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fa49dad868244b7988ed66f4d890135 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/780944cc1b2e464389b3b79fb74d12bf : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":26},{"id":"2f95a60f75a87ef8e57c88d327fcf8027e7d051d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-10-06 08:45:55.000000000","message":"Uploaded patch set 27.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":27},{"id":"f71f76fb167bdd6849263f9bcfff35fcd87201eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 10:20:02.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e47d8131797488ab0044682d74fbc32\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/d196a2a0eb0540308cc7aa2dd647cca3 : SUCCESS in 29m 43s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/812bd42cc2e54f749abca909886e1024 : SUCCESS in 41m 47s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/57ff0f0e2c8a42f09e2e5a7660513862 : SUCCESS in 31m 04s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/e8d6b52f4cb6459aa6c5450e33435b25 : FAILURE in 44m 06s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2881b3663b3b480ba13fcb0ac456d8e6 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4cee8f6c2e6b4c928e6d556186e69349 : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e918095cf114477e8a4bda4ba63b67b4 : SUCCESS in 6m 00s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"f4916f0e48b0870ade5b456e45688012e1934f4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-06 11:17:04.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b993ac51cb3441a6a81118fdfa702e55\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/af970c9f58a5415bba70199a4079d9d8 : SUCCESS in 15m 24s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/d88b386e84da4c05a23865f36c8b8845 : SUCCESS in 10m 32s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/dc0f9cc302354aeea213f62645897bc6 : SUCCESS in 22m 53s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/7be9b1d464ee435db75883c1cb82953b : SUCCESS in 10m 44s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/b00ab9d0763841e5a0ed39e87eff977d : SUCCESS in 37m 07s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cb4933570fa7494abd985b6084541308 : SUCCESS in 20m 50s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/a313d9ebfeea4ea080283bae602d03d3 : SUCCESS in 25m 15s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/3358311bc9f44e4c8ac46983b48912c4 : SUCCESS in 1h 46m 02s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/44bfeeacf626482dae1414c346b23cf7 : FAILURE in 1h 24m 37s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/04854d516beb4f88894403f98cd1a33f : SUCCESS in 1h 53m 01s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/b69275573a014be38e3ee24c30db1cd9 : SUCCESS in 2h 29m 52s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/5199a9e0dc9f4fc7b780bfa8ab4dfa26 : SUCCESS in 28m 49s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/9706936ad65349489cc118a725aba8fb : FAILURE in 28m 56s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6e7f1db3db124951a3deee0b6bda31e6 : SUCCESS in 2m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40a4389548ac4281b6518be018d08c2c : SUCCESS in 2m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/837de8ca88414ef58d1b1b85ff96bf3e : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e68c1f515c5f4c92849d653f36795b1d : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9da9e68eefe84636a5846640a8e777bf : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e781d97993de4c93bb0acf72f1d4f4b1 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5895738dcdb947838363ca6c2bfd56bd : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed8d7fdb02214a898001ff98647241ec : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":27},{"id":"f884b094edc57284ab2a25a1889bd7a72d72196d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"date":"2025-10-14 15:08:28.000000000","message":"Uploaded patch set 28.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":28},{"id":"3b7148e41634ba32c34cdd5424c43f26b142f6ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 16:17:39.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e556e82378e641438aed4d6d67c988a6\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/36ff5e2f65ee4d87af0c0fb34bcdb79b : SUCCESS in 56m 02s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/8c14b3768d7e4012b6375162d6dd7c5a : SUCCESS in 23m 49s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/a3590ddc4265415c81a0c995e1e91155 : SUCCESS in 36m 43s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/08d6cc21068943a4bf2f7cfdc2c077a9 : SUCCESS in 31m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3531ee66aba34a79bbfbbf781a3575e3 : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f8431004ccf5402ea71e5f87b946a824 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2e570f6fabb547d98252c130143435a5 : SUCCESS in 12m 38s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"bb055311ea827ce73afc621beeef548a4dffbbc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 18:12:56.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4ffe0aafa1743cea13be04caeae5ee9\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/ba76fc560db642779e109f2dbcd0f4bd : SUCCESS in 16m 36s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/34b82e7ec9b848feb2aebcfd2ebf1e49 : SUCCESS in 17m 01s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/4757c2d250ba45589aa73e14b29446cc : SUCCESS in 15m 18s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/ebbd6e86723e4a4bb6080895ea87fa7f : SUCCESS in 15m 13s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/062da2cce2ce4213a08619af5732a417 : SUCCESS in 55m 57s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5f2959fa1e094a4b962bcb27268b9173 : SUCCESS in 20m 46s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/824f337654564b5b863a0cc85cd78bc1 : SUCCESS in 28m 43s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/3cf0630705424e259688dfcb250dcbad : SUCCESS in 1h 37m 13s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/55cf801f3e7e45b185326d04ff5d77c7 : FAILURE in 2h 28m 13s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/4128a1cfcd57489e8dbe93f922eddd96 : SUCCESS in 2h 14m 29s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/0ebd3092d2e5485b80f6ab4381d525c5 : TIMED_OUT in 3h 03m 10s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/bdd2dd21720c4f7bb8c1782dd2a56c1c : SUCCESS in 31m 33s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/43a56199ac2d4f849329b4bded04ce68 : SUCCESS in 26m 07s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/65bb4043f6eb48659e75e9c613d821dd : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86d0655644144939a0499e9bc558ed2b : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d612b557dd64454aa89d609fe41fae26 : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1fa771f39463465e8e1c87eb2e35f502 : SUCCESS in 3m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dc87369153a44abd818b72286075301b : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29764d10aff94dd1b44ef5871d617ce2 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57d55355feef4af998443108145ab1c2 : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a53e7974852c4cf9bfc1e73a95e9148d : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":28},{"id":"aaed1a46c5c17606684a00d6c19ccdc578129a5b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-23 08:15:33.000000000","message":"Patch Set 28: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"ec830da55afa6a06d7c07b3744eb3f75f2ab8cea","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-23 08:15:48.000000000","message":"Patch Set 28:\n\n(1 comment)","accounts_in_message":[],"_revision_number":28},{"id":"32d763aac52489ab26f37c6510b86d0acea91973","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-23 09:16:34.000000000","message":"Patch Set 28: Code-Review+2","accounts_in_message":[],"_revision_number":28},{"id":"a7750affc36dcf549bfd7a6d3099035276c24cb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 09:54:19.000000000","message":"Patch Set 28:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c887f35d1412482c82cfccb8194e0578\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/3afb1e33a9044b80a636e5c27a882087 : SUCCESS in 10m 16s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/4659fba1d90d46b482e2350ade0479ab : SUCCESS in 20m 12s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/e7fabc3ac2f54f49b5c7bdc843639324 : SUCCESS in 24m 17s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/e0f2e86f3003477f9971955a0471860e : SUCCESS in 10m 59s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/007e37c2f4bf4b378cb0238af4c5c0ee : SUCCESS in 1h 04m 01s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e5778aa393174ac6b28d61e2f5b88389 : SUCCESS in 17m 43s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/af39acdd0e2848be96a6a0d75b45b838 : SUCCESS in 14m 52s\n- kolla-ansible-debian-bookworm-cephadm https://zuul.opendev.org/t/openstack/build/2f7e943b66fb4f898b0ebe61ecd6864e : SUCCESS in 1h 28m 55s (non-voting)\n- kolla-ansible-debian-bookworm-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/3666626f6a9c405694b10814efe474b8 : FAILURE in 1h 14m 03s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/a8eb88f191db495491831707cf732288 : SUCCESS in 1h 37m 33s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/047dda6af0534b669c64de3ad3351894 : FAILURE in 1h 25m 36s (non-voting)\n- kolla-ansible-debian-bookworm-mariadb https://zuul.opendev.org/t/openstack/build/13483bd05fdb41f981cac65d15957e04 : SUCCESS in 22m 44s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/30fdf760bebe4324855778d4e7ffe947 : SUCCESS in 27m 51s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a42bb9208f114c8497572e7c493e006f : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ba9b16f08304ecb8f712b25cb615f39 : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/216529ef60064446b2f7c9e20e328c14 : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e04ce846a7924d0992596d51c7c7dcd8 : SUCCESS in 7m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/161a24b65125434bbeec6c60571f4e74 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87bd1d0f6aeb4092b446d6587c2a67c0 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2630fd14409740e7ae6e137cc6be7a2a : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":28},{"id":"b5ba5fccc168adf30d12a13a434be4958ff968fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 10:43:51.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/becf2426d7ef4af5b98899d7c1c95c11\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/d5d1c4c942084563a58b636c1808e298 : SUCCESS in 1h 00m 25s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/b941473908b24792856f7bb521cc08c9 : SUCCESS in 56m 10s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/5cbaa6c77bd44e099993db9dfa1e5f0d : SUCCESS in 1h 02m 25s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/c2f63f917ee542379af042ec32613f79 : SUCCESS in 1h 14m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/67832f3a3a3749148d4e9d6677f34514 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/338a0ef8ccf24c4c8b8d2396b83f84f7 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/846b6ee517b541519cc13a23b61654e2 : SUCCESS in 12m 09s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"fdc769fa924540a1313192e8f3cc5be3f82b80d2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-10-23 13:08:29.000000000","message":"Patch Set 28: Code-Review+1","accounts_in_message":[],"_revision_number":28},{"id":"2ad9c5c2c7c2617da8f86d8158eb0200d7e81cdc","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-24 15:47:40.000000000","message":"Patch Set 28: Code-Review+2","accounts_in_message":[],"_revision_number":28},{"id":"f57880c240a4239a2523d9dbab9a73868ad5e37d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 17:09:41.000000000","message":"Patch Set 28: Workflow+1","accounts_in_message":[],"_revision_number":28},{"id":"9590425061c1e10d7e0403070ab75615f0260d6f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 17:10:23.000000000","message":"Patch Set 28: Backport-Candidate-1","accounts_in_message":[],"_revision_number":28},{"id":"2ecebc760e2d8624c435ae5e82f62dc30a6a5883","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 17:11:06.000000000","message":"Patch Set 28: -Workflow","accounts_in_message":[],"_revision_number":28},{"id":"9fa757e928194aad8e395bd787a73009790d59f3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-27 17:11:10.000000000","message":"Patch Set 28: Workflow+1","accounts_in_message":[],"_revision_number":28},{"id":"d07ce526fe8e77f5ef9e8c484b9114c4d71ed4d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 17:12:34.000000000","message":"Patch Set 28: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":28},{"id":"5fcdad95cbc7f1d7cfa0243217bf144f8ee3e1d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 17:52:32.000000000","message":"Patch Set 28: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83edc5eda7fd40e58e3836b49b08a375\n\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/deca9aef9ac4499087208e55a3d033b1 : SUCCESS in 16m 13s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/3490a7345a864afc87293a28de097e20 : SUCCESS in 27m 42s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/45387ef925c247c9a11e5b0a1cf00802 : SUCCESS in 22m 38s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/d5655638862046f9bb1f37f3108a2720 : SUCCESS in 36m 39s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/249f9ca1f41a4109b4cea5283e06fbcb : SUCCESS in 22m 49s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/dd06093533284d5ba209e6d5bde7c9b4 : SUCCESS in 31m 17s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/39001c0bcf6e4d2eba18f857520de2ea : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4c4f82d53cb4d2abc8aff4ae41d5d53 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8283904c49b346959d358e79b62afa0c : SUCCESS in 2m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/522bfeb0bb2c4de7a016d8eb2557c882 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3fc6a716da64a6ea10b49b4e98d3394 : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/863e8f3bb2894fa194a12cbd9e6476d8 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":28},{"id":"2ec581e6315ef45240466b690aadfa3d5d8985af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 17:52:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":28},{"id":"b0df5d08d0305aa13e01d694c7d188b9293eefc9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-27 17:53:47.000000000","message":"Patch Set 28:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e082bb02dc3043849dc7f4cd5f6916cc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d41c4170d9141d6a9564bdd11b7fc7b : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c084262551ce4a63b348d45a3d039e93 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":28}],"current_revision_number":28,"current_revision":"c24528917661a9dc6f0fb1379cc74c544c690a42","revisions":{"39591be8754f96b1b53b109eece7b10279e6370b":{"kind":"REWORK","_number":1,"created":"2024-08-23 06:37:47.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/926981/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/1"}}},"commit":{"parents":[{"commit":"8f6bc332bfe84275836c9153ebab6902041ca83a","subject":"Merge \"Add exclude for permissions in kolla config.json file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f6bc332bfe84275836c9153ebab6902041ca83a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:45.000000000","tz":120},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\n[1]: https://mariadb.org/11-4-lts/\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39591be8754f96b1b53b109eece7b10279e6370b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/39591be8754f96b1b53b109eece7b10279e6370b"}]},"branch":"refs/heads/master"},"43faf4851d6e1e266277ef04a59b6ec34555ebaa":{"kind":"REWORK","_number":2,"created":"2024-08-23 06:39:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/926981/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/2"}}},"commit":{"parents":[{"commit":"8f6bc332bfe84275836c9153ebab6902041ca83a","subject":"Merge \"Add exclude for permissions in kolla config.json file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f6bc332bfe84275836c9153ebab6902041ca83a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:39:02.000000000","tz":120},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\n[1]: https://mariadb.org/11-4-lts/\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/43faf4851d6e1e266277ef04a59b6ec34555ebaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/43faf4851d6e1e266277ef04a59b6ec34555ebaa"}]},"branch":"refs/heads/master"},"9b9e2ec008e2ae0a0a5d306cf903e3284873f0b1":{"kind":"REWORK","_number":3,"created":"2024-08-23 07:13:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/926981/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/3"}}},"commit":{"parents":[{"commit":"8f6bc332bfe84275836c9153ebab6902041ca83a","subject":"Merge \"Add exclude for permissions in kolla config.json file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f6bc332bfe84275836c9153ebab6902041ca83a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 07:13:32.000000000","tz":0},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\n[1]: https://mariadb.org/11-4-lts/\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b9e2ec008e2ae0a0a5d306cf903e3284873f0b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b9e2ec008e2ae0a0a5d306cf903e3284873f0b1"}]},"branch":"refs/heads/master"},"9217d833ced66f0a56286ee6ae5531f03585d1b7":{"kind":"REWORK","_number":4,"created":"2024-08-23 08:16:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/926981/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/4"}}},"commit":{"parents":[{"commit":"8f6bc332bfe84275836c9153ebab6902041ca83a","subject":"Merge \"Add exclude for permissions in kolla config.json file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f6bc332bfe84275836c9153ebab6902041ca83a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 08:16:07.000000000","tz":0},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\n[1]: https://mariadb.org/11-4-lts/\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9217d833ced66f0a56286ee6ae5531f03585d1b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9217d833ced66f0a56286ee6ae5531f03585d1b7"}]},"branch":"refs/heads/master"},"397031ffcca2049956280346f7ac4f2b5c34d89c":{"kind":"REWORK","_number":5,"created":"2024-08-23 14:01:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/926981/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/5"}}},"commit":{"parents":[{"commit":"8f6bc332bfe84275836c9153ebab6902041ca83a","subject":"Merge \"Add exclude for permissions in kolla config.json file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f6bc332bfe84275836c9153ebab6902041ca83a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 14:01:39.000000000","tz":0},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\n[1]: https://mariadb.org/11-4-lts/\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/397031ffcca2049956280346f7ac4f2b5c34d89c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/397031ffcca2049956280346f7ac4f2b5c34d89c"}]},"branch":"refs/heads/master"},"7fe5e04513bfda824db86423589c409afedf9a74":{"kind":"REWORK","_number":6,"created":"2024-08-23 14:04:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/926981/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/6"}}},"commit":{"parents":[{"commit":"8f6bc332bfe84275836c9153ebab6902041ca83a","subject":"Merge \"Add exclude for permissions in kolla config.json file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f6bc332bfe84275836c9153ebab6902041ca83a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 14:04:01.000000000","tz":0},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\n[1]: https://mariadb.org/11-4-lts/\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7fe5e04513bfda824db86423589c409afedf9a74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7fe5e04513bfda824db86423589c409afedf9a74"}]},"branch":"refs/heads/master"},"e57ee2d60b5fc5916afd4dd11a8f4f0eaa9029e5":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2024-08-23 15:07:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/926981/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/7"}}},"commit":{"parents":[{"commit":"8f6bc332bfe84275836c9153ebab6902041ca83a","subject":"Merge \"Add exclude for permissions in kolla config.json file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f6bc332bfe84275836c9153ebab6902041ca83a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 15:07:42.000000000","tz":0},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\n[1]: https://mariadb.org/11-4-lts/\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/927039\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e57ee2d60b5fc5916afd4dd11a8f4f0eaa9029e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e57ee2d60b5fc5916afd4dd11a8f4f0eaa9029e5"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e5b9818451f2a4f4b3dbfd5bdc3e4edbf284e73c":{"kind":"REWORK","_number":8,"created":"2024-08-24 05:43:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/926981/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/8"}}},"commit":{"parents":[{"commit":"8f6bc332bfe84275836c9153ebab6902041ca83a","subject":"Merge \"Add exclude for permissions in kolla config.json file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f6bc332bfe84275836c9153ebab6902041ca83a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-24 05:43:27.000000000","tz":0},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\n[1]: https://mariadb.org/11-4-lts/\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/927039\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5b9818451f2a4f4b3dbfd5bdc3e4edbf284e73c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5b9818451f2a4f4b3dbfd5bdc3e4edbf284e73c"}]},"branch":"refs/heads/master"},"80e7d4eba4a1b9d53e7f676af07156fb9afb18cc":{"kind":"REWORK","_number":9,"created":"2024-08-25 17:32:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/926981/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/9"}}},"commit":{"parents":[{"commit":"8f6bc332bfe84275836c9153ebab6902041ca83a","subject":"Merge \"Add exclude for permissions in kolla config.json file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f6bc332bfe84275836c9153ebab6902041ca83a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-25 17:32:28.000000000","tz":0},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\n[1]: https://mariadb.org/11-4-lts/\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/927039\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/80e7d4eba4a1b9d53e7f676af07156fb9afb18cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/80e7d4eba4a1b9d53e7f676af07156fb9afb18cc"}]},"branch":"refs/heads/master"},"bbf1dd37ec03af2aae02bf8e60bcbb5bfcacc51a":{"kind":"NO_CODE_CHANGE","_number":10,"created":"2024-08-25 17:33:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/926981/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/10"}}},"commit":{"parents":[{"commit":"8f6bc332bfe84275836c9153ebab6902041ca83a","subject":"Merge \"Add exclude for permissions in kolla config.json file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f6bc332bfe84275836c9153ebab6902041ca83a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-25 17:33:19.000000000","tz":0},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\nclustercheck.sh has been imported from the old location, because of\nchanged command names (mysql* commands have been removed and only\navailable in -compat package which we are not installing)\n\n[1]: https://mariadb.org/11-4-lts/\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/927039\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bbf1dd37ec03af2aae02bf8e60bcbb5bfcacc51a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bbf1dd37ec03af2aae02bf8e60bcbb5bfcacc51a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"eabd54150be43dc089a553217ed2dacd3bc198bc":{"kind":"REWORK","_number":11,"created":"2024-08-25 18:33:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/926981/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/11"}}},"commit":{"parents":[{"commit":"8f6bc332bfe84275836c9153ebab6902041ca83a","subject":"Merge \"Add exclude for permissions in kolla config.json file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f6bc332bfe84275836c9153ebab6902041ca83a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-25 18:33:42.000000000","tz":0},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\nclustercheck.sh has been imported from the old location, because of\nchanged command names (mysql* commands have been removed and only\navailable in -compat package which we are not installing)\n\n[1]: https://mariadb.org/11-4-lts/\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/927039\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eabd54150be43dc089a553217ed2dacd3bc198bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eabd54150be43dc089a553217ed2dacd3bc198bc"}]},"branch":"refs/heads/master"},"6e6aaa179ab2c4b263bf8888954bf3c39474dc67":{"kind":"NO_CODE_CHANGE","_number":12,"created":"2024-08-26 12:14:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/926981/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/12"}}},"commit":{"parents":[{"commit":"8f6bc332bfe84275836c9153ebab6902041ca83a","subject":"Merge \"Add exclude for permissions in kolla config.json file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f6bc332bfe84275836c9153ebab6902041ca83a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 12:14:55.000000000","tz":0},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\nclustercheck.sh has been imported from the old location, because of\nchanged command names (mysql* commands have been removed and only\navailable in -compat package which we are not installing)\n\n[1]: https://mariadb.org/11-4-lts/\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/927096\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e6aaa179ab2c4b263bf8888954bf3c39474dc67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e6aaa179ab2c4b263bf8888954bf3c39474dc67"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3ff62bb8529060bc9063f5eba7bd9e38bba9ed24":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2024-08-26 15:13:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/926981/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/13"}}},"commit":{"parents":[{"commit":"98015e78aec7fa2753934f3db3325d30a8e7c617","subject":"mariadb: Add label before upgrade to 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98015e78aec7fa2753934f3db3325d30a8e7c617"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 15:13:09.000000000","tz":0},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\nclustercheck.sh has been imported from the old location, because of\nchanged command names (mysql* commands have been removed and only\navailable in -compat package which we are not installing)\n\n[1]: https://mariadb.org/11-4-lts/\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/927096\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ff62bb8529060bc9063f5eba7bd9e38bba9ed24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ff62bb8529060bc9063f5eba7bd9e38bba9ed24"}]},"branch":"refs/heads/master","description":"Rebase"},"ab86c55b19b4e62e7b78cc4af7e4f2f8c4f18395":{"kind":"REWORK","_number":14,"created":"2024-08-26 15:13:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/926981/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/14"}}},"commit":{"parents":[{"commit":"98015e78aec7fa2753934f3db3325d30a8e7c617","subject":"mariadb: Add label before upgrade to 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98015e78aec7fa2753934f3db3325d30a8e7c617"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 15:13:19.000000000","tz":0},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\nclustercheck.sh has been imported from the old location, because of\nchanged command names (mysql* commands have been removed and only\navailable in -compat package which we are not installing)\n\n[1]: https://mariadb.org/11-4-lts/\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/927096\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab86c55b19b4e62e7b78cc4af7e4f2f8c4f18395"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ab86c55b19b4e62e7b78cc4af7e4f2f8c4f18395"}]},"branch":"refs/heads/master"},"d71465adf7a104cd1aca9cf25c846874fd529286":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2024-09-08 08:10:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/926981/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/15"}}},"commit":{"parents":[{"commit":"98015e78aec7fa2753934f3db3325d30a8e7c617","subject":"mariadb: Add label before upgrade to 11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98015e78aec7fa2753934f3db3325d30a8e7c617"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-08 08:10:40.000000000","tz":0},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\nclustercheck.sh has been imported from the old location, because of\nchanged command names (mysql* commands have been removed and only\navailable in -compat package which we are not installing)\n\n[1]: https://mariadb.org/11-4-lts/\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/928487\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d71465adf7a104cd1aca9cf25c846874fd529286"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d71465adf7a104cd1aca9cf25c846874fd529286"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6fdc099df1ab0d766eead29c662b5f71633e308a":{"kind":"REWORK","_number":16,"created":"2024-11-27 16:23:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/926981/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/16"}}},"commit":{"parents":[{"commit":"dca8462bc491a93c6a6ae71e86d59a2dcb804ff5","subject":"Merge \"Add mechanism for patching files in containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dca8462bc491a93c6a6ae71e86d59a2dcb804ff5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-11-27 16:23:00.000000000","tz":0},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\nclustercheck.sh has been imported from the old location, because of\nchanged command names (mysql* commands have been removed and only\navailable in -compat package which we are not installing)\n\n[1]: https://mariadb.org/11-4-lts/\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/928487\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6fdc099df1ab0d766eead29c662b5f71633e308a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6fdc099df1ab0d766eead29c662b5f71633e308a"}]},"branch":"refs/heads/master","description":"Rebase"},"32f321b7df716a36e64d746f4e45d13cb56b4033":{"kind":"REWORK","_number":17,"created":"2024-11-27 16:24:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/81/926981/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/17"}}},"commit":{"parents":[{"commit":"dca8462bc491a93c6a6ae71e86d59a2dcb804ff5","subject":"Merge \"Add mechanism for patching files in containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dca8462bc491a93c6a6ae71e86d59a2dcb804ff5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-11-27 16:24:01.000000000","tz":0},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\nclustercheck.sh has been imported from the old location, because of\nchanged command names (mysql* commands have been removed and only\navailable in -compat package which we are not installing)\n\n[1]: https://mariadb.org/11-4-lts/\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/928487\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32f321b7df716a36e64d746f4e45d13cb56b4033"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/32f321b7df716a36e64d746f4e45d13cb56b4033"}]},"branch":"refs/heads/master"},"8bf838a6555c6523e161a67a703469e566570643":{"kind":"REWORK","_number":18,"created":"2025-09-09 10:53:15.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/81/926981/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/18"}}},"commit":{"parents":[{"commit":"dca8462bc491a93c6a6ae71e86d59a2dcb804ff5","subject":"Merge \"Add mechanism for patching files in containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dca8462bc491a93c6a6ae71e86d59a2dcb804ff5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-09-09 10:48:23.000000000","tz":60},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\nclustercheck.sh has been imported from the old location, because of\nchanged command names (mysql* commands have been removed and only\navailable in -compat package which we are not installing)\n\nNew variable ``VERIFY_DB_SERVER_CERT`` is added to mitigate the MariaDB\nclient variable ``ssl-verify-server-cert`` change from ``FALSE`` to\n``TRUE``. [2]\nThis variable can be overidden by assigning ``TRUE`` to environment\nvariable with the same name.\n\n[1]: https://mariadb.org/11-4-lts/\n[2]: https://jira.mariadb.org/browse/MDEV-31857\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/928487\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8bf838a6555c6523e161a67a703469e566570643"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8bf838a6555c6523e161a67a703469e566570643"}]},"branch":"refs/heads/master"},"bc5f0e4f053830258e8aa41ab0685129155a2ec4":{"kind":"NO_CODE_CHANGE","_number":19,"created":"2025-09-09 12:34:46.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/81/926981/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/19"}}},"commit":{"parents":[{"commit":"dca8462bc491a93c6a6ae71e86d59a2dcb804ff5","subject":"Merge \"Add mechanism for patching files in containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dca8462bc491a93c6a6ae71e86d59a2dcb804ff5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-09-09 12:34:16.000000000","tz":60},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\nclustercheck.sh has been imported from the old location, because of\nchanged command names (mysql* commands have been removed and only\navailable in -compat package which we are not installing)\n\nNew variable ``VERIFY_DB_SERVER_CERT`` is added to mitigate the MariaDB\nclient variable ``ssl-verify-server-cert`` change from ``FALSE`` to\n``TRUE``. [2]\nThis variable can be overidden by assigning ``TRUE`` to environment\nvariable with the same name.\n\n[1]: https://mariadb.org/11-4-lts/\n[2]: https://jira.mariadb.org/browse/MDEV-31857\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/928487\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/958281\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc5f0e4f053830258e8aa41ab0685129155a2ec4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bc5f0e4f053830258e8aa41ab0685129155a2ec4"}]},"branch":"refs/heads/master"},"8e2cf9b988cb1a71dbf1f4e17339e33740f6fe72":{"kind":"REWORK","_number":20,"created":"2025-09-09 15:03:23.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/81/926981/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/20"}}},"commit":{"parents":[{"commit":"dca8462bc491a93c6a6ae71e86d59a2dcb804ff5","subject":"Merge \"Add mechanism for patching files in containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dca8462bc491a93c6a6ae71e86d59a2dcb804ff5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-09-09 15:02:42.000000000","tz":60},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\nclustercheck.sh has been imported from the old location, because of\nchanged command names (mysql* commands have been removed and only\navailable in -compat package which we are not installing)\n\nNew variable ``VERIFY_DB_SERVER_CERT`` is added to mitigate the MariaDB\nclient variable ``ssl-verify-server-cert`` change from ``FALSE`` to\n``TRUE``. [2]\nThis variable can be overidden by assigning ``TRUE`` to environment\nvariable with the same name.\n\n[1]: https://mariadb.org/11-4-lts/\n[2]: https://jira.mariadb.org/browse/MDEV-31857\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/928487\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/958281\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e2cf9b988cb1a71dbf1f4e17339e33740f6fe72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8e2cf9b988cb1a71dbf1f4e17339e33740f6fe72"}]},"branch":"refs/heads/master"},"795ff28d38bcda2eb33b05f8c064db18639feff1":{"kind":"REWORK","_number":21,"created":"2025-09-24 14:06:17.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/81/926981/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/21"}}},"commit":{"parents":[{"commit":"dca8462bc491a93c6a6ae71e86d59a2dcb804ff5","subject":"Merge \"Add mechanism for patching files in containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dca8462bc491a93c6a6ae71e86d59a2dcb804ff5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-09-24 14:05:58.000000000","tz":60},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\nclustercheck.sh has been imported from the old location, because of\nchanged command names (mysql* commands have been removed and only\navailable in -compat package which we are not installing)\n\nNew variable ``VERIFY_DB_SERVER_CERT`` is added to mitigate the MariaDB\nclient variable ``ssl-verify-server-cert`` change from ``FALSE`` to\n``TRUE``. [2]\nThis variable can be overidden by assigning ``TRUE`` to environment\nvariable with the same name.\n\n[1]: https://mariadb.org/11-4-lts/\n[2]: https://jira.mariadb.org/browse/MDEV-31857\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/928487\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/958281\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/795ff28d38bcda2eb33b05f8c064db18639feff1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/795ff28d38bcda2eb33b05f8c064db18639feff1"}]},"branch":"refs/heads/master"},"5769a8a5db98377769112247c896c48ccdbdbb98":{"kind":"REWORK","_number":22,"created":"2025-09-24 14:51:54.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/81/926981/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/22"}}},"commit":{"parents":[{"commit":"dca8462bc491a93c6a6ae71e86d59a2dcb804ff5","subject":"Merge \"Add mechanism for patching files in containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dca8462bc491a93c6a6ae71e86d59a2dcb804ff5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-09-24 14:51:36.000000000","tz":60},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\nclustercheck.sh has been imported from the old location, because of\nchanged command names (mysql* commands have been removed and only\navailable in -compat package which we are not installing)\n\nNew variable ``VERIFY_DB_SERVER_CERT`` is added to mitigate the MariaDB\nclient variable ``ssl-verify-server-cert`` change from ``FALSE`` to\n``TRUE``. [2]\nThis variable can be overidden by assigning ``TRUE`` to environment\nvariable with the same name.\n\n[1]: https://mariadb.org/11-4-lts/\n[2]: https://jira.mariadb.org/browse/MDEV-31857\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/928487\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/958281\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5769a8a5db98377769112247c896c48ccdbdbb98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5769a8a5db98377769112247c896c48ccdbdbb98"}]},"branch":"refs/heads/master"},"eb1f2a59df9f93da27306897470159a8a4908f1a":{"kind":"REWORK","_number":23,"created":"2025-09-29 10:33:42.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/81/926981/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/23"}}},"commit":{"parents":[{"commit":"dca8462bc491a93c6a6ae71e86d59a2dcb804ff5","subject":"Merge \"Add mechanism for patching files in containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dca8462bc491a93c6a6ae71e86d59a2dcb804ff5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-09-29 10:33:35.000000000","tz":60},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\nReplaced clustercheck.sh with healthcheck.sh from official MariaDB\ndocker image, because of changed command names (mysql* commands have\nbeen removed and only available in -compat package which we are not\ninstalling)\n\nNew variable ``VERIFY_DB_SERVER_CERT`` is added to mitigate the MariaDB\nclient variable ``ssl-verify-server-cert`` change from ``FALSE`` to\n``TRUE``. [2]\nThis variable can be overidden by assigning ``TRUE`` to environment\nvariable with the same name.\n\n[1]: https://mariadb.org/11-4-lts/\n[2]: https://jira.mariadb.org/browse/MDEV-31857\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/928487\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/958281\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb1f2a59df9f93da27306897470159a8a4908f1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb1f2a59df9f93da27306897470159a8a4908f1a"}]},"branch":"refs/heads/master"},"7336e6d7212172f66d1a3029a517e9c77e0a0f6a":{"kind":"NO_CODE_CHANGE","_number":24,"created":"2025-09-29 12:15:29.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/81/926981/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/24"}}},"commit":{"parents":[{"commit":"dca8462bc491a93c6a6ae71e86d59a2dcb804ff5","subject":"Merge \"Add mechanism for patching files in containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dca8462bc491a93c6a6ae71e86d59a2dcb804ff5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-09-29 12:14:53.000000000","tz":60},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\nReplaced clustercheck.sh with healthcheck.sh from official MariaDB\ndocker image, because of changed command names (mysql* commands have\nbeen removed and only available in -compat package which we are not\ninstalling)\n\nNew variable ``VERIFY_DB_SERVER_CERT`` is added to mitigate the MariaDB\nclient variable ``ssl-verify-server-cert`` change from ``FALSE`` to\n``TRUE``. [2]\nThis variable can be overidden by assigning ``TRUE`` to environment\nvariable with the same name.\n\n[1]: https://mariadb.org/11-4-lts/\n[2]: https://jira.mariadb.org/browse/MDEV-31857\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/928487\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/958281\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/962479\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7336e6d7212172f66d1a3029a517e9c77e0a0f6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7336e6d7212172f66d1a3029a517e9c77e0a0f6a"}]},"branch":"refs/heads/master"},"816a21d7193c326ab75e73b3185517c3ec9642dd":{"kind":"NO_CODE_CHANGE","_number":25,"created":"2025-09-30 14:19:19.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/81/926981/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/25"}}},"commit":{"parents":[{"commit":"dca8462bc491a93c6a6ae71e86d59a2dcb804ff5","subject":"Merge \"Add mechanism for patching files in containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dca8462bc491a93c6a6ae71e86d59a2dcb804ff5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-09-30 14:18:50.000000000","tz":60},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\nReplaced clustercheck.sh with healthcheck.sh from official MariaDB\ndocker image, because of changed command names (mysql* commands have\nbeen removed and only available in -compat package which we are not\ninstalling)\n\nNew variable ``VERIFY_DB_SERVER_CERT`` is added to mitigate the MariaDB\nclient variable ``ssl-verify-server-cert`` change from ``FALSE`` to\n``TRUE``. [2]\nThis variable can be overidden by assigning ``TRUE`` to environment\nvariable with the same name.\n\n[1]: https://mariadb.org/11-4-lts/\n[2]: https://jira.mariadb.org/browse/MDEV-31857\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/928487\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/958281\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/816a21d7193c326ab75e73b3185517c3ec9642dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/816a21d7193c326ab75e73b3185517c3ec9642dd"}]},"branch":"refs/heads/master"},"97b1d42e545494f84cdda7f145c4c0cad306c564":{"kind":"REWORK","_number":26,"created":"2025-10-01 10:26:47.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/81/926981/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/26"}}},"commit":{"parents":[{"commit":"dca8462bc491a93c6a6ae71e86d59a2dcb804ff5","subject":"Merge \"Add mechanism for patching files in containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dca8462bc491a93c6a6ae71e86d59a2dcb804ff5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-10-01 10:26:11.000000000","tz":60},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\nclustercheck.sh has been imported from the old location, because of\nchanged command names (mysql* commands have been removed and only\navailable in -compat package which we are not installing)\n\nNew variable ``VERIFY_DB_SERVER_CERT`` is added to mitigate the MariaDB\nclient variable ``ssl-verify-server-cert`` change from ``FALSE`` to\n``TRUE``. [2]\nThis variable can be overidden by assigning ``TRUE`` to environment\nvariable with the same name.\n\n[1]: https://mariadb.org/11-4-lts/\n[2]: https://jira.mariadb.org/browse/MDEV-31857\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/928487\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/958281\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97b1d42e545494f84cdda7f145c4c0cad306c564"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/97b1d42e545494f84cdda7f145c4c0cad306c564"}]},"branch":"refs/heads/master"},"a18f721c14291925819e6c1dc0959a666cdfd918":{"kind":"REWORK","_number":27,"created":"2025-10-06 08:45:55.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/81/926981/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/27"}}},"commit":{"parents":[{"commit":"6c8f2b9042aeb6be020f1c74a73b8dc3ff65de8f","subject":"Merge \"Adds ironic-pxe-filter container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c8f2b9042aeb6be020f1c74a73b8dc3ff65de8f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-10-06 08:45:33.000000000","tz":60},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\nclustercheck.sh has been imported from the old location, because of\nchanged command names (mysql* commands have been removed and only\navailable in -compat package which we are not installing)\n\nNew variable ``VERIFY_DB_SERVER_CERT`` is added to mitigate the MariaDB\nclient variable ``ssl-verify-server-cert`` change from ``FALSE`` to\n``TRUE``. [2]\nThis variable can be overidden by assigning ``TRUE`` to environment\nvariable with the same name.\n\n[1]: https://mariadb.org/11-4-lts/\n[2]: https://jira.mariadb.org/browse/MDEV-31857\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/928487\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/958281\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a18f721c14291925819e6c1dc0959a666cdfd918"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a18f721c14291925819e6c1dc0959a666cdfd918"}]},"branch":"refs/heads/master"},"c24528917661a9dc6f0fb1379cc74c544c690a42":{"kind":"REWORK","_number":28,"created":"2025-10-14 15:08:28.000000000","uploader":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"ref":"refs/changes/81/926981/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/81/926981/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/81/926981/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/81/926981/28"}}},"commit":{"parents":[{"commit":"f71ab2be350a3366a5113ac04248bc8cb368dc52","subject":"Merge \"Drop support for building MariaDB clustercheck\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f71ab2be350a3366a5113ac04248bc8cb368dc52"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 06:37:13.000000000","tz":120},"committer":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2025-10-14 15:07:46.000000000","tz":60},"subject":"mariadb: Bump to 11.4","message":"mariadb: Bump to 11.4\n\n11.4 is the next LTS release (see [1])\n\nNew variable ``VERIFY_DB_SERVER_CERT`` is added to mitigate the MariaDB\nclient variable ``ssl-verify-server-cert`` change from ``FALSE`` to\n``TRUE``. [2]\nThis variable can be overidden by assigning ``TRUE`` to environment\nvariable with the same name.\n\n[1]: https://mariadb.org/11-4-lts/\n[2]: https://jira.mariadb.org/browse/MDEV-31857\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/928487\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/958281\n\nChange-Id: Ifb41c0c48c4743e70e95fa7e1d329e91362e966c\nSigned-off-by: Seunghun Lee \u003cseunghun@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c24528917661a9dc6f0fb1379cc74c544c690a42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c24528917661a9dc6f0fb1379cc74c544c690a42"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
