)]}'
{"id":"openstack%2Fkolla~926983","triplet_id":"openstack%2Fkolla~master~I78bffabfbb407b0a8ad5e8a5c70fec8d7181c8b4","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-11-21 10:57:28.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I78bffabfbb407b0a8ad5e8a5c70fec8d7181c8b4","subject":"ubuntu: switch to Dalmatian UCA","status":"ABANDONED","created":"2024-08-23 07:01:42.000000000","updated":"2024-11-21 10:57:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"63a76420058b4cf3fbd546ff4cbe1e3c7bff0944","_number":926983,"virtual_id_number":926983,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-23 09:35:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"635620cb2c9c9ed1f4b5d8635b9ec85238e91152","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-23 07:01:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff54313eed503c74e0f5829391024e97b4cf3e02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 09:35:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab6b6250e99d47e28af82208933a2044\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ab8944aa38344979b1365dc0d6e59b7d : SUCCESS in 31m 42s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0785bcc8b9e5430485209ce325021679 : SUCCESS in 59m 48s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a811a50303d042e1a0f38d9ec822a5e5 : SUCCESS in 17m 35s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d92ce3c4f3ae4901b170458966f07690 : SUCCESS in 24m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/36dc7961d44148dc92c04e0bd97cdc1e : SUCCESS in 57m 29s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e89cb757301746579783d63f5329cb5b : SUCCESS in 1h 14m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c43fec5d011442d7820c67cd37e833cb : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e04c803c2a74979a9981492fe3c4813 : SUCCESS in 2m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6d9da29aaa9d482c9f92a234c3580652 : SUCCESS in 2m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/014e24b41f0147b4950c7bb3814fcc93 : SUCCESS in 2m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21b3899d12264ccf92b7bd5beaf23df8 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b22a650187c4f538cbd2d22603f019e : SUCCESS in 3m 54s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/47717c56aae94d9abaf4b0f2fcdac4c3 : SUCCESS in 24m 32s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5db52384809b4d309e418e22854005be : SUCCESS in 35m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4b039e9c15d04cad8caba9b4bc11eada : SUCCESS in 1h 05m 16s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6f4ad3a695fe4109b0b5458669f06ff1 : SUCCESS in 1h 44m 54s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/e9800a960b95419f988d367cac19ba8a : SUCCESS in 1h 55m 31s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5d71643b5a764b5c9c0f24ed92e30e01 : FAILURE in 2h 33m 20s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/f725aae3a3444764ad17f7ce4a4763bf : SUCCESS in 1h 26m 43s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/00df6e42bd1a43ecb6217136e289da05 : SUCCESS in 1h 13m 03s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/3b81d7585a5947bb8e74292f0f50ff25 : SUCCESS in 31m 44s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/dd50c9fb189a4709b2d164edd336c1aa : SUCCESS in 52m 16s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/16cca214ab2e49de993dbeeaefcd7f3a : SUCCESS in 1h 12m 08s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/4e3758c3ee6e4776b9584080338c7835 : SUCCESS in 1h 28m 53s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/55b1c81d1c344fcea7b751c3f06e3904 : SUCCESS in 31m 33s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d01971f1e4004f3da9800b8e88b57b31 : FAILURE in 4m 58s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7b8d0f0fb17e4f6d92124680d01bade1 : FAILURE in 6m 25s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a1dd40d10a714912b6d91daa15af77ea : FAILURE in 7m 13s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/94957fd1e64847619815494548af3106 : FAILURE in 7m 40s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/2079b21df33c4c199fd253c6e3316fe3 : FAILURE in 9m 00s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a5865d60dfc7427aaf7f366773a6bd4a : FAILURE in 13m 30s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/aba191173edd4476913116e553d2455c : FAILURE in 8m 14s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/092d2cf76b094b508823f6f63d19abcc : FAILURE in 9m 01s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/68295f7728dc4a9d8e19308ee0b8b095 : FAILURE in 7m 42s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/54997b86bf7e43c4808a2161e126e773 : FAILURE in 12m 48s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/a2ab9d38688c4bdbadbc0efa20fe626a : FAILURE in 8m 35s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/fbed8b95d82f40f384d9951330cc9e2e : FAILURE in 8m 40s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0d09e120e94248eb83ff30d584f79b2b : FAILURE in 7m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d0be4a41751f9c1017f9ca76fd889ef2121a54c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-24 04:20:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67f57d352e3e42eeb0381645e5c39f3c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/fa45dcf752934c2d9a694585f4ce35dc : TIMED_OUT in 3h 03m 55s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6a416fedc6f94c919530e0b124fa84ee : SUCCESS in 2h 47m 44s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0e85c998b4f042de811c1e79446f05ee : SUCCESS in 1h 30m 43s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b0230311444c4123bec7881b4fd1d6bb : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a958f8e99b23499fb7896f6a97aab4a6 : SUCCESS in 3m 37s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c503766612584279a418ba67bded27ef : TIMED_OUT in 3h 05m 45s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3d621353ff7f4824a757bf23d03afc06 : FAILURE in 7m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"37d6bbd7e061d69dd88aa5035dbcb4a15d178df5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-21 10:57:28.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nThe following files contain Git conflicts:\n* docker/base/sources.list.ubuntu\n* docker/base/sources.list.ubuntu.aarch64\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"63a76420058b4cf3fbd546ff4cbe1e3c7bff0944","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-21 10:57:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"49f105f108130857a22a38816e5f3bd6989835ef","revisions":{"3d4e447e5a6782af8b3788461a8c601809472b2f":{"kind":"REWORK","_number":1,"created":"2024-08-23 07:01:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/83/926983/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/926983/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/926983/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/926983/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/926983/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/926983/1"}}},"commit":{"parents":[{"commit":"8f6bc332bfe84275836c9153ebab6902041ca83a","subject":"Merge \"Add exclude for permissions in kolla config.json file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f6bc332bfe84275836c9153ebab6902041ca83a"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 07:01:26.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 07:01:40.000000000","tz":120},"subject":"ubuntu: switch to Dalmatian UCA","message":"ubuntu: switch to Dalmatian UCA\n\nChange-Id: I78bffabfbb407b0a8ad5e8a5c70fec8d7181c8b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d4e447e5a6782af8b3788461a8c601809472b2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d4e447e5a6782af8b3788461a8c601809472b2f"}]},"branch":"refs/heads/master"},"49f105f108130857a22a38816e5f3bd6989835ef":{"kind":"REWORK","_number":2,"created":"2024-11-21 10:57:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/83/926983/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/83/926983/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/926983/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/926983/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/83/926983/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/83/926983/2"}}},"commit":{"parents":[{"commit":"dca8462bc491a93c6a6ae71e86d59a2dcb804ff5","subject":"Merge \"Add mechanism for patching files in containers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dca8462bc491a93c6a6ae71e86d59a2dcb804ff5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-23 07:01:26.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-11-21 10:57:28.000000000","tz":0},"subject":"ubuntu: switch to Dalmatian UCA","message":"ubuntu: switch to Dalmatian UCA\n\nChange-Id: I78bffabfbb407b0a8ad5e8a5c70fec8d7181c8b4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49f105f108130857a22a38816e5f3bd6989835ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/49f105f108130857a22a38816e5f3bd6989835ef"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
