)]}'
{"id":"openstack%2Fkolla~927000","triplet_id":"openstack%2Fkolla~master~I66421b14668722607b58865a2ad100204fd387f7","project":"openstack/kolla","branch":"master","topic":"drop-prometheus-msteams","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-09-02 16:38:10.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2024-09-02 20:56:03.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-09-02 20:56:03.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-09-02 20:56:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I66421b14668722607b58865a2ad100204fd387f7","subject":"Drop prometheus-msteams container images","status":"MERGED","created":"2024-08-23 09:10:31.000000000","updated":"2024-09-02 20:57:09.000000000","submitted":"2024-09-02 20:56:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927000-drop-prometheus-msteams","meta_rev_id":"8266a1e5fb48a61a1b3b0ee12e25952bce3ce869","_number":927000,"virtual_id_number":927000,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-02 20:56:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-09-02 16:38:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-08-27 10:24:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-09-02 16:38:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":-1,"date":"2024-08-27 10:24:43.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-23 09:12:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2024-08-23 10:50:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-27 10:24:43.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-08-29 09:02:35.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-08-29 09:02:35.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"e2ccc1e060cbde06a5d2a40eb26c58e3702904c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-08-23 09:10:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"96047f890600e95ec7892a4174f1671fff8cdfa2","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-08-23 09:11:17.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"88bf5599bf6e172e9f224fbfd5b2e589e567aa43","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-08-23 09:12:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5bdbe16635252c2f3474381019f76a2a799aac4f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-08-23 09:17:12.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0496c6383246cfb6624e8bd46734e6d35e39b97d","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-08-23 09:18:53.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"30571344440736eadb457ebbb4fda6bea6e7656a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-08-23 09:27:18.000000000","message":"Topic set to drop-prometheus-msteams","accounts_in_message":[],"_revision_number":2},{"id":"22c774e12fd5e84467c94752d97980d2558bf289","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 10:50:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/033a8730e773462aa813fb08285b4b06\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0e2a3a070bb8464e8220f3f4c83e7165 : SUCCESS in 23m 58s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/7f09e6e1cfb14396a8c23dc7f3aed205 : SUCCESS in 1h 04m 01s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/754fdd5f67244e3094aac9a51342e46a : SUCCESS in 16m 28s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8d6aebad0f2b45b093bd7e3d89cf4809 : SUCCESS in 26m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2b16bdca6ddc428f8bf08ec6507bf5a9 : SUCCESS in 55m 38s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fc03a45d86b047bebc23c849725328f8 : SUCCESS in 1h 25m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29ecd1602a584e2bb49ddc7502955e4f : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7dfb260df6d84ee9949e8a207c614df0 : SUCCESS in 2m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1daa599178154c42a5ed2924958053d7 : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3481a654b2b24134a2690552463c2a81 : SUCCESS in 3m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7aa22ace80bc40e6b554c905f3d79986 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/746bb2a4fdb646478dddabac52b0e376 : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0c9334ea71d406fa3440b9c05206d47 : SUCCESS in 3m 24s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b5d3155c4a2b4189933d138019a5f92b : SUCCESS in 23m 32s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/af9ba768b0374a37a7271382c9b3a42b : SUCCESS in 38m 21s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/703b691b1a3e4fb8906d5c92fe149218 : SUCCESS in 1h 03m 18s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/44c326c0f21241b1acb3960ba1dfaa19 : SUCCESS in 1h 27m 00s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/dbb593311a014e37ac8bf27bc0301848 : SUCCESS in 33m 53s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/782a6c4c7d374915add07f14f130628c : SUCCESS in 28m 13s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/78fff1e9b2b442b99a8f4f1b3c5e54af : SUCCESS in 1h 01m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8d4e40ee0df74ca2a7ef1b6b80e5a293 : SUCCESS in 56m 35s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/56a7f7f88c7a41caa382ebd34e6bbecb : SUCCESS in 1h 19m 10s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/cd8a96101de84a528b057204b49f0a2a : SUCCESS in 26m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d7f2d08f8eb3b20abde8209be65f2418b16ac4bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-24 13:56:54.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81e2d2fd2c9e44b5a21225af42a6595f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3c9b054e9e7248369c0a47098fec1085 : TIMED_OUT in 3h 06m 22s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/59339e9ff2354a9b91f924eca39158ff : TIMED_OUT in 3h 02m 36s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/214876ed0ef64453a4a836bf359ca34d : TIMED_OUT in 3h 03m 11s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f6edeac5bd9447cf919f38802e52cbeb : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8c6b1e047e6346b38a4a5e09a872db28 : SUCCESS in 3m 37s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d2d2fcd72ad54481a199b939fa821a9e : TIMED_OUT in 3h 03m 43s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a94a08fe52cb4c0ca636dd54cfc477ae : TIMED_OUT in 3h 02m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7a13142a5fe56454b152b879000651f153bf70fa","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-27 10:24:43.000000000","message":"Patch Set 2: Backport-Candidate-1 Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4edc7ead66a684f8b04126e5cb7b67129f13d76c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-09-02 16:38:10.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d131bcc1dd7fe222e300d3b94a1a394f38c4aa11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 16:38:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5428f7ae03ee386ae9d414ca8c11f7fb032958af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 18:02:40.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5be0966b0b62453790d82fd50e9079a5\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/00b9cfae1a6e4896840b3c9fdeacb306 : SUCCESS in 13m 01s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1fa5280208d441609388c70335e6a651 : SUCCESS in 20m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/38fc6bd71e27472589a8b92af4404966 : FAILURE in 6m 29s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e1589efe40444fe7a7aae1bdfa427b28 : SUCCESS in 1h 15m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/becc31ef1a07411f902d775f1a41754f : SUCCESS in 2m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f781c787f0984eca936503ef5b20050e : SUCCESS in 2m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c88f810b9f184e10808159f2305fd5de : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/175a7c5e0c444f9da0e215540ff781f9 : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40040314214f46fdadc784039117e532 : SUCCESS in 3m 17s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2bd9a4002867411eab926c6eaace0095 : SUCCESS in 26m 14s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/70cc4001143a4c8dbf5cccdfd92ec61e : SUCCESS in 33m 49s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/47555a6bca634c29ab680e0cdd13a51f : SUCCESS in 1h 02m 01s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/45bab4341eb64829a8a2fb3febc00d39 : SUCCESS in 1h 19m 41s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/789f600276384fbcbbb1f6906ed739ef : SUCCESS in 15m 50s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a76d31a19aba4d81abd5dfee31dc5eb4 : SUCCESS in 58m 09s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c924af318cb247d19af0e7c13e442469 : SUCCESS in 48m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/80dc9df805e54af7a8f5c3b5e4df10a2 : SUCCESS in 1h 04m 00s","accounts_in_message":[],"_revision_number":2},{"id":"9ed872f0112c0b8246a2d9749fe6e5719fcdc695","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-09-02 18:06:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8f59f12b6dd6bd2298379a94f197f20c3ee7fdc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 19:32:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b183b58b80dc416e965fdc9d708c25b8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b12daf4125864c8ebb7a725b389a765b : SUCCESS in 23m 42s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f7d823094b1045e98fc4f9013559873d : SUCCESS in 1h 00m 34s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/13f1035253ad4d2ebc35c65834a79d8e : SUCCESS in 17m 01s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3abbdf3968d44facaaf518e7e580cd69 : SUCCESS in 22m 46s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b290ef2bec2a4d00b9e5928f5bf62e1a : SUCCESS in 53m 16s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/602bb150dfbd4d2d8e9b8ebf38a20d09 : SUCCESS in 1h 09m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36163883732a45aabb1d97666e78c6bf : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d2d0b7a9322446f8c17b702eb5d1e96 : SUCCESS in 2m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9edd7fe0064245cf813b4892c66fa568 : SUCCESS in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/749f978e454147f29e989ed911f488da : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/727bb9f37428464796fc3c130fb55fd8 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acb18f4d556f4706895c71c959175111 : SUCCESS in 5m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c9f913fbe5e349f58c67ab47cf2b2b4c : SUCCESS in 3m 29s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a0ecbe414c4448a0a7d7ebc4b9903d65 : SUCCESS in 29m 03s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/21358d7be9af407997e111dc52e59227 : SUCCESS in 35m 51s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c0c487ef99d44aaab7b9b3faba057548 : SUCCESS in 1h 10m 07s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b8337d855f884fccb33445917bfd1067 : SUCCESS in 1h 20m 28s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/848cce9474074c02ba574a0287ed207f : SUCCESS in 34m 00s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/caabd635668940eaaf82097c0ed655a6 : SUCCESS in 21m 03s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/466178ba2ed74acdbf3ce16fc7cfa6c8 : SUCCESS in 48m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5a2c5ee43083499b9ab7139889b19ad9 : SUCCESS in 46m 41s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1a110f16186840a6b0fcea71a7028e66 : SUCCESS in 1h 14m 27s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/db3fdf200bbc435b8939cd773cbb65c6 : SUCCESS in 29m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0ce696efd5a11ce1b4a382a60813c6ca97f227f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 19:32:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"241db09bb63254200e23b93d7d7046705af80429","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 19:44:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ba7639d35d54605b8c5ff85db643eb8\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c025340516c6407d8707698a44e956a4 : SUCCESS in 1h 37m 09s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5f347c076528449dbddcea24ef0599bf : SUCCESS in 1h 32m 14s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/57e76ba9f85d4f5b9dd014af65df250a : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/783a00b656d449acb9a5a2256e64cee8 : SUCCESS in 3m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"46eebe686bacf00984493119d656e4e69d2c648a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 20:56:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02edf774283849f19dc11d52c9c62b04\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d00378e7ec494d36866c97d3145166d8 : SUCCESS in 14m 38s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f2ca0b6b97934a4e9785e52cfd366267 : SUCCESS in 28m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/81deba3b800041e7b438d30b597aa411 : SUCCESS in 45m 41s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/87badc4dd8b44fd2bc7c42e0fc4be25a : SUCCESS in 1h 21m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9074b331e1834c5e82f7f4e017627063 : SUCCESS in 2m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59b8647c8e9e4af1aa4358c1a8e25bcf : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a59dd8f19730405abf929a40650382d5 : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a91ec85ca8d6425e84f3b0af40e23b00 : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d2b703ec9914273a1bd4b57e61acfaa : SUCCESS in 3m 22s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d24167e9c3f44890b0a772f1247cd65e : SUCCESS in 26m 53s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0254df5f651c4034a4364df15d7c7838 : SUCCESS in 30m 18s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2f9490921f604cb7a9456e68540860df : SUCCESS in 1h 04m 06s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/19162ea8b009419baed33c58fafaf0cf : SUCCESS in 1h 06m 41s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/75db5caf2b284a81980207e794ed1a4c : SUCCESS in 15m 28s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9c0170440bde4b57b65fb817a4bb2799 : SUCCESS in 42m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/03cf80a2e2664c3c9e3e95f6d902f840 : SUCCESS in 52m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1f5f877b10ce4707a560e21dfc36a10d : SUCCESS in 1h 11m 30s","accounts_in_message":[],"_revision_number":2},{"id":"2b2b254f06651376ead650a3f09d2b64da256383","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 20:56:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8266a1e5fb48a61a1b3b0ee12e25952bce3ce869","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 20:57:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eaf4454743714f99b691bec09e8c24f8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8966b9e9995046a2b7c76f4b636521ac : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b43cfd7187e846e1ac9df90203a4dd8d : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a2c07b3c2bdfe16e4f2ebe52616b08189f6b5af7","revisions":{"08e8916fceecb71e89153f6a6594b70f17799e44":{"kind":"REWORK","_number":1,"created":"2024-08-23 09:10:31.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/00/927000/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/927000/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/927000/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/927000/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/927000/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/927000/1"}}},"commit":{"parents":[{"commit":"8f6bc332bfe84275836c9153ebab6902041ca83a","subject":"Merge \"Add exclude for permissions in kolla config.json file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f6bc332bfe84275836c9153ebab6902041ca83a"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-08-23 09:10:07.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-08-23 09:10:07.000000000","tz":120},"subject":"Drop prometheus-msteams container images","message":"Drop prometheus-msteams container images\n\nThe prometheus-msteams project is no longer maintained [1]. As a result, we\nare removing its associated container images from our repository. Users are\nencouraged to switch to the native Prometheus Alertmanager integration with\nMicrosoft Teams [2], which offers ongoing support and features.\n\n[1] https://github.com/prometheus-msteams/prometheus-msteams/issues/343\n[2] https://prometheus.io/docs/alerting/latest/configuration/#msteams_config\n\nChange-Id: I66421b14668722607b58865a2ad100204fd387f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08e8916fceecb71e89153f6a6594b70f17799e44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08e8916fceecb71e89153f6a6594b70f17799e44"}]},"branch":"refs/heads/master"},"a2c07b3c2bdfe16e4f2ebe52616b08189f6b5af7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-08-23 09:17:12.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/00/927000/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/927000/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/927000/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/927000/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/927000/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/927000/2"}}},"commit":{"parents":[{"commit":"8f6bc332bfe84275836c9153ebab6902041ca83a","subject":"Merge \"Add exclude for permissions in kolla config.json file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8f6bc332bfe84275836c9153ebab6902041ca83a"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-08-23 09:10:07.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-08-23 09:10:53.000000000","tz":120},"subject":"Drop prometheus-msteams container images","message":"Drop prometheus-msteams container images\n\nThe prometheus-msteams project is no longer maintained [1]. As a result\nwe are removing its associated container images from our repository.\nUsers are encouraged to switch to the native Prometheus Alertmanager\nintegration with Microsoft Teams [2], which offers ongoing support and\nfeatures.\n\n[1] https://github.com/prometheus-msteams/prometheus-msteams/issues/343\n[2] https://prometheus.io/docs/alerting/latest/configuration/#msteams_config\n\nChange-Id: I66421b14668722607b58865a2ad100204fd387f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2c07b3c2bdfe16e4f2ebe52616b08189f6b5af7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2c07b3c2bdfe16e4f2ebe52616b08189f6b5af7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
