)]}'
{"id":"openstack%2Fkolla~927149","triplet_id":"openstack%2Fkolla~master~Ic9e706694e564d230caacabf44f2c352991d54cc","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Ic9e706694e564d230caacabf44f2c352991d54cc","subject":"DNM: Add some debug to healthcheck_port","status":"ABANDONED","created":"2024-08-26 14:46:13.000000000","updated":"2025-02-21 17:11:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"73073b763afed8d0e514e8a7b026d7b17a88768c","_number":927149,"virtual_id_number":927149,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-08-26 17:43:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-26 17:43:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"86131df86563c4af6e09818eded4afd29367887c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-26 14:46:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01e9c648efdf3633aa2d989a7891ad0781fced85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 17:43:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1713d14c039a4fa19a66a99e7b7c747b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/65715e7966e443b8b922c29f9f6ce5fd : SUCCESS in 22m 55s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/137ee6ace42d4067af6d622f83894ded : SUCCESS in 1h 04m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/66f57e541a8449f28efd88e9623cff70 : SUCCESS in 19m 47s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d9ef0a3af73740a49717e2c42d64df3f : SUCCESS in 28m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ab0089224ca94758a92c590ec73dc9d9 : SUCCESS in 58m 37s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/64078887cb9f4c3e8797bc9c0c7c62cb : SUCCESS in 1h 11m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/261b409a03b447b8a8a2f65f5c691a82 : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ec7fae371a64695b2ff1366fd5f159b : SUCCESS in 2m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bd381c93c9c14cb685bfb51ab3a741b6 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/97a9ee164f1248b3b0b73bbb3608b262 : SUCCESS in 2m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ff7bbbeb4ac47279143a6634d3ff66f : SUCCESS in 2m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59e7d75feeb4423f91ef9be05ba2c828 : SUCCESS in 4m 26s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b7b2f17f22f24eeebc54019d9d0d564c : SUCCESS in 29m 45s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/50bd5fb91e094bce9c6307d105aaf37a : SUCCESS in 33m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e9836497eb644df4b13f54a780138294 : SUCCESS in 1h 05m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c0442c1fd03f434d86e2a2f9cb7894e2 : SUCCESS in 1h 17m 22s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/8458ffdb778d448c88645463b541a19d : SUCCESS in 1h 45m 32s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2b3e9dfb34f34d9987d095ddb8ded665 : SUCCESS in 2h 50m 55s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/3b90cffd33ad4b3d8862da0a635e1e0f : SUCCESS in 1h 07m 31s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/9279de5f5e004aada60169933eb14164 : SUCCESS in 1h 23m 18s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/5686e46608ac4c8c8b159d3b4b1d9a46 : SUCCESS in 28m 13s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/e940b1bee5dd4e66afad6ad325b2a35d : SUCCESS in 53m 15s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/a286d809ef4341818a4584fa7fc9d59f : SUCCESS in 1h 09m 38s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/c85772493c35400ebe0bab6965bf451c : SUCCESS in 1h 25m 49s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/14b098e9d0984f379ea61ff655f43d9d : SUCCESS in 29m 21s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4361d43887ef443e8e4d7934d5ff29ce : SUCCESS in 21m 13s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/fe003d7cf14c4d799300c2a20678ad22 : SUCCESS in 48m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2f0d5c125c9f43218514cf5830776754 : SUCCESS in 1h 00m 23s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/73defe7c73dd4c319dc0d9edfca0c19c : SUCCESS in 1h 11m 07s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/b12fbd83142e4eb28d19545a61e5e6ff : SUCCESS in 1h 47m 34s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/dba547cee00b4308b5fa947253dccec0 : SUCCESS in 2h 29m 32s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/d9ccbe6eff1f4372a20b74e0541e66f2 : SUCCESS in 1h 01m 23s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/94a4b7fb48b346a8b360be9f12760edc : SUCCESS in 1h 18m 10s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/f7a98564f448465a9d29e8d8a0e1a674 : SUCCESS in 20m 57s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/b775e92cd3a54344a3bdeffc4da80ed6 : SUCCESS in 49m 22s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/241c351e574f49308e51e891c02bdd75 : SUCCESS in 1h 00m 34s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/a12cc2bf800f4076909b2c3fea842006 : SUCCESS in 1h 04m 17s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/350086d494ba405caf78a4cd51538995 : SUCCESS in 25m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"649d675ad1158f6843e58e433606f9309d7958fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 22:38:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54842a50a1d94558b5403ea3c1b379d5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/95a6c07bf4a34c6b8920d173a40232e0 : SUCCESS in 3h 04m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a04acc5aaaaa46538a700ae35b6b32a2 : SUCCESS in 2h 47m 16s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a2481003b10a4636861f96e10818989b : SUCCESS in 1h 42m 54s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7ffaf0bc4d7b40859da7acf3807ce891 : SUCCESS in 9m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/06563b61a4104e3da3d1b92feb04f924 : SUCCESS in 3m 53s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/514a6fc4b59d422d87f7832b7d2678af : TIMED_OUT in 3h 05m 18s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/65e889f536284a449eabe93833af7d14 : FAILURE in 1h 48m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"73073b763afed8d0e514e8a7b026d7b17a88768c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-02-21 17:11:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"140feb170c69ccc187fa78cabd05ed0c52544814","revisions":{"140feb170c69ccc187fa78cabd05ed0c52544814":{"kind":"REWORK","_number":1,"created":"2024-08-26 14:46:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/49/927149/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/49/927149/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/927149/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/927149/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/49/927149/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/49/927149/1"}}},"commit":{"parents":[{"commit":"ccb13b672d4a178fefa624b8ab5b214a0bec7e45","subject":"Merge \"bifrost: bump Ansible version to 8 (rpm) and 9 (deb)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ccb13b672d4a178fefa624b8ab5b214a0bec7e45"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 14:45:58.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 14:46:11.000000000","tz":120},"subject":"DNM: Add some debug to healthcheck_port","message":"DNM: Add some debug to healthcheck_port\n\nChange-Id: Ic9e706694e564d230caacabf44f2c352991d54cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/140feb170c69ccc187fa78cabd05ed0c52544814"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/140feb170c69ccc187fa78cabd05ed0c52544814"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
