)]}'
{"id":"openstack%2Fkolla~927210","triplet_id":"openstack%2Fkolla~master~Iab9eaf234eeb15b8edfca454f7b3aacdc4ead544","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-10-15 19:16:34.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"Iab9eaf234eeb15b8edfca454f7b3aacdc4ead544","subject":"DNM: Run fio tests","status":"ABANDONED","created":"2024-08-26 20:06:34.000000000","updated":"2025-12-02 11:35:02.000000000","total_comment_count":2,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"1bbe308a81634cc8c8e504def637e0edb1a0579d","_number":927210,"virtual_id_number":927210,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-11-26 10:43:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2024-08-26 21:45:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4ebb6dc604661c8bb1c41d916f4c37a10a735834","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-26 20:06:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cef02729a60164107303feae384eab582c95715a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-26 20:09:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b3ee34b37b93fb4ddbaa5880a4cc6c4bfe03fe4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-26 20:14:19.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2c87c2a8c2b2fc657fbfb78cb850021055f278ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-26 20:18:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bf01e2a174a1126f8dce21a58789b0185101ccc6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-26 21:45:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fac1b3fbb00448f1bca0b7082acf804a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/bb07b678c89f4d998732434671ca27cb : SUCCESS in 5m 19s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0ac937fbdcf241bba983ca298affa6b9 : SUCCESS in 1h 06m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7ada8a527e014f619e60072ff9aec019 : SUCCESS in 4m 03s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c3ec00e9b95b45aeabae061140adc199 : SUCCESS in 4m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/30387551b3d54559a5484203bc854b80 : SUCCESS in 1h 06m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ed682fd14a104ca897b06c0d67dc2156 : SUCCESS in 1h 06m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cb78af521524d6fb1749a62bdafe596 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a303b3184b64e27a43e130bd3c78c4d : SUCCESS in 3m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0e52bcf5df2e4a3ba611c4da245ef5cb : SUCCESS in 3m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a89f3ea6a0544c9282b01f1b46e47579 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d79bb11e304b4f98afc7d50a0ca39727 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/391dd7e31c5840158616aa66b2af9d43 : SUCCESS in 4m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6b0b09804d164e78af283aa4120f991f : SUCCESS in 7m 40s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/12a6d408963c452f887017a687de97e0 : SUCCESS in 8m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9509f42d7e4642ac91c6c39d015305b8 : SUCCESS in 58m 30s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6890dc339d964234a43fda929e2119ed : SUCCESS in 1h 21m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/05674551a95941909d084729d69f56ed : SUCCESS in 4m 27s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f64da4f3467641f098c55d15b08f8f57 : SUCCESS in 4m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2951be6855d348d185ad64734bd05749 : SUCCESS in 50m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8b67c9da173346b0a69bcb739766bdaf : SUCCESS in 1h 02m 12s","accounts_in_message":[],"_revision_number":4},{"id":"e1e865de756233221e58bb20559ce49bb3f67796","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 12:22:45.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/558067647f174e549f504cacda44a2ba\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ad903f690e4847cea01dd46a8358bf34 : NODE_FAILURE Node(set) request 300-0025019672 failed in 0s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4048e52bc97a4f638c37168f2ff5a142 : NODE_FAILURE Node(set) request 300-0025019673 failed in 0s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fe1600d5c8e9480ab75c23d6fa6443c4 : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/119ba1b4b4e34427a15e7170bb4dac9a : SUCCESS in 3m 54s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b369dd0f243b16f331a3cd8a3742aacd33a54087","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-27 15:17:26.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c2ee7bd25b7e6c0a3ff539d909edee30ee5f3fb1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-27 15:25:46.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"5c6a795f677ca415ed26146e0c15f1a0cc6ea32e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 17:24:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97946a125765489bb6752ed30afe2a80\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/47e953e015a048e3ad6d15873ff3f49e : SUCCESS in 4m 16s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4fcb6437552f436594a4540dd227a758 : SUCCESS in 3m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b6c7487d4c544c18b6dd6eac9ac7cbc1 : SUCCESS in 1h 00m 06s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/78450771b7184a16988b766ce03125f8 : SUCCESS in 1h 19m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d0949cf67ec4c97a7765872cd7d19cf : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a743d8bf5c0a4ff29d6e0ca84ab35140 : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a1c91c60b02645b7a59454a238d06f9d : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/846a0b6818d4411183edfefe7e7b9748 : SUCCESS in 3m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eae227b86a0c495a8472f78ec171611f : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0333259c4435421a8e869ae27851e6e6 : SUCCESS in 4m 49s","accounts_in_message":[],"_revision_number":5},{"id":"5f68b55477e0e0ce6433403a33cf09e2abb55a8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-27 18:48:50.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"75e7a9a0bb460386f86037aa7862d23b95774326","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-27 18:49:53.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"7e0834c96e2ded8eafaa56b6d9949b412bb4be5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-27 18:50:17.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"9ee7d1eadc16a128ee0739efe18adbdaa7fb29bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 19:04:29.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db700121006347188c51bfa2d681ab6e\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5136fc3d0ad54b97b0a3165dd33c1254 : SUCCESS in 4m 58s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d7d7b4410c5b4298b6db715ebdbb3fd7 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ed7d4fa9bb24ad8af0e4b6a9cc3cc22 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":8},{"id":"bdf150afa9c3ef4aa5d659b458ecfc625734804b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 21:16:00.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79a7f74254b94e43bdc1bdd247c1cea5\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f1fbacd2785a4871a64320bfc3df65e7 : SUCCESS in 2h 17m 43s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"98562eece3958f1fe074c27d980dcb0d31131a9a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-15 19:16:03.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c937cfd7ed6745b5734ac8ece39d2bac251c22f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 19:16:18.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 927210,8","accounts_in_message":[],"_revision_number":8},{"id":"6030ef3a2dfbcc3872f78797ea75aa19d0cbb2cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 19:16:21.000000000","message":"Patch Set 8: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 927210,8","accounts_in_message":[],"_revision_number":8},{"id":"983068a1482607492ca4f97fbf86709c1361a878","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-15 19:16:34.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nThe following files contain Git conflicts:\n* .zuul.d/centos.yaml\n* .zuul.d/rocky.yaml\n* .zuul.d/ubuntu.yaml\n* tests/playbooks/run.yml\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"74b5b679aa2b49272d700e48088423b6fa36db64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 19:16:56.000000000","message":"Patch Set 9: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch master.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/kolla/.zuul.d/centos.yaml@master\", line 2, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/kolla/.zuul.d/centos.yaml@master\", line 3, column 1","accounts_in_message":[],"_revision_number":9},{"id":"69424b95536459474b74b34d5da018ff2d00acee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 19:17:03.000000000","message":"Patch Set 9:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch master.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/kolla/.zuul.d/centos.yaml@master\", line 2, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/kolla/.zuul.d/centos.yaml@master\", line 3, column 1","accounts_in_message":[],"_revision_number":9},{"id":"b8b776d72e389c90a452ceb0e8a27e667f334328","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-15 19:18:24.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"77e6a0e15d1a4f8294f84bd372258227d2cfd482","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 19:18:47.000000000","message":"Patch Set 10: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch master.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/kolla/.zuul.d/centos.yaml@master\", line 2, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/kolla/.zuul.d/centos.yaml@master\", line 3, column 1","accounts_in_message":[],"_revision_number":10},{"id":"75f27b179808534908610670716e0bb554f2e0ba","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-15 19:18:54.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"bff8c6165b91cf1a465286f4b6a33f63ecc9d9d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 19:18:55.000000000","message":"Patch Set 10:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch master.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/kolla/.zuul.d/centos.yaml@master\", line 2, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/kolla/.zuul.d/centos.yaml@master\", line 3, column 1","accounts_in_message":[],"_revision_number":10},{"id":"5b7d2d3e49719d514be6caf5ee2a8529a2734a8f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-15 19:19:14.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"4d205551cbffcecc7426d50c8b4b680117e41da2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 19:19:15.000000000","message":"Patch Set 11:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch master.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/kolla/.zuul.d/rocky.yaml@master\", line 2, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/kolla/.zuul.d/rocky.yaml@master\", line 3, column 1","accounts_in_message":[],"_revision_number":11},{"id":"e8176966baa81b40a6afebf233abec1f8734a646","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 19:19:25.000000000","message":"Patch Set 11: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch master.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/kolla/.zuul.d/rocky.yaml@master\", line 2, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/kolla/.zuul.d/rocky.yaml@master\", line 3, column 1","accounts_in_message":[],"_revision_number":11},{"id":"970de5daa90758f91fb165ef409e270c54a823c1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-15 19:19:29.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"e7cc5c284c536008730e15f46698ae47a4bf02c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 19:19:38.000000000","message":"Patch Set 12: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch master.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/kolla/.zuul.d/ubuntu.yaml@master\", line 2, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/kolla/.zuul.d/ubuntu.yaml@master\", line 3, column 1","accounts_in_message":[],"_revision_number":12},{"id":"b6fe62c29d4190f08de2cb7c6e4e0b6a7a90b45d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 19:19:46.000000000","message":"Patch Set 12:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch master.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/kolla/.zuul.d/ubuntu.yaml@master\", line 2, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/kolla/.zuul.d/ubuntu.yaml@master\", line 3, column 1","accounts_in_message":[],"_revision_number":12},{"id":"bc11ce34446ca49a0c931fc1cc1f441ed2a3431b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-15 19:19:48.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"5bf984b1824b5e6ca536e2f073ba2709975968f0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-10-15 19:20:46.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"cdeae96189ec8171c41c7d99560f8f1242ecbdd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 19:26:27.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06adafbb03d945e3a3fda95fbd6a4c00\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2bb93e3211ae4e8e9e53b2081103a545 : SUCCESS in 3m 41s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/694f116aacdb4883ad4f462d814409a2 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7016544cec84315b9671d32f0b3fa0e : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":15},{"id":"d69967057277eabd91f35ee639a288a719fe2178","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-15 21:31:11.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13c7255bb7c94c00a02a2b44a665a5a3\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8a3a7db44af84da98b68cebd28a5340e : SUCCESS in 2h 07m 35s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"0e34a0a9362d8ee91b374539aa48bc8eeb1535a2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-12-11 12:15:32.000000000","message":"Patch Set 16: Patch Set 15 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"748b1b4211374e008cd164a8e789bbc2c309fae6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 12:34:43.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cabd431d9e754728881a7faeccccf7cf\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e507f28c62294b4fa4fce520b1bc7d4e : SUCCESS in 12m 42s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"a694ba65a9a64c96ea2bb84733a7cff24a8d7adc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-11 12:41:36.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ec80bc152f54bb78a46502a060ad7a9\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a1d9b2b058cb46189ad4326af7c4f93a : SUCCESS in 4m 58s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/53b57c6a0c62431dbafaadb93b2ae3db : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71f7d82b0e9540788fbc9555f654228e : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":16},{"id":"b5a7b80d656ce5c6d7e46bb3c12a571d40a8b2b5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-02-21 17:11:50.000000000","message":"Patch Set 17: Patch Set 16 was rebased\n\nThe following files contain Git conflicts:\n* tests/playbooks/pre.yml\n* tests/playbooks/run.yml\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"a0a6f84d9e047b0222fe0387b085f2c6ed98d109","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-02-21 17:13:17.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"e0e883c6a0067d351e9486353d5ffe8ffa9f09d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 17:36:46.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca83f2c56d414f53814b40ade5db1ddf\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0e4a1e1c1eae43c2b42e49b5f5651f86 : SUCCESS in 5m 21s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ed449051dbb749198a6835140aacf077 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9abd1978c3143d9bf0ee2bebcc70044 : SUCCESS in 4m 36s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/189a159d74a548d783b33e2a0141618a : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":18},{"id":"58734e6fffe2a4c3105ac3c172f6496d895bbb9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 19:02:31.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c40415caae5d4f6281c8fbc69b5d21e0\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3906329c85ca42fba6783a36ed34bfe2 : SUCCESS in 17m 08s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"6f68ad66c9f441fc6d2384cd192076ba54bfd635","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-20 16:46:46.000000000","message":"Abandoned\n\nnot needed anymore","accounts_in_message":[],"_revision_number":18},{"id":"afa68bfbcf2c5875282566b2cd37721180e52ba8","tag":"autogenerated:gerrit:restore","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-26 10:24:01.000000000","message":"Restored","accounts_in_message":[],"_revision_number":18},{"id":"81e8714d5794ec8f9a8aa4951d3696a714ec5c76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 10:24:08.000000000","message":"Patch Set 18: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 927210,18","accounts_in_message":[],"_revision_number":18},{"id":"9941e9cb98cc909add487332b765aaf88578123c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 10:24:08.000000000","message":"Patch Set 18:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 927210,18","accounts_in_message":[],"_revision_number":18},{"id":"c9ed7a3977fa8931232e7c77ddf362a27790cd4c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-26 10:24:23.000000000","message":"Patch Set 19: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"3c68d80cb2aadf0b78666b0574e060b55247ba3f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-26 10:24:28.000000000","message":"Patch Set 20: Patch Set 19 was rebased\n\nThe following files contain Git conflicts:\n* .zuul.d/debian.yaml\n* .zuul.d/project.yaml\n","accounts_in_message":[],"_revision_number":20},{"id":"7bc6d203ec155d92040af1e8b064309b9209ebd0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 10:24:29.000000000","message":"Patch Set 19:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kolla for 927210,19","accounts_in_message":[],"_revision_number":19},{"id":"60ae2dcebb865c5ed90bcb7b76e106c03098feba","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-11-26 10:24:45.000000000","message":"Patch Set 21: Published edit on patch set 20.","accounts_in_message":[],"_revision_number":21},{"id":"7b7b06e0257edb87bded60298b8c87384e62b2fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 10:43:49.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/83f7b2dac7654a6b8dd94306b74bc52a\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/cddafd3c8cfb4a5a808ddecc9d0e2a1b : SUCCESS in 5m 19s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/4ced2ab9fa9d46a09665115eb64627c1 : SUCCESS in 5m 32s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/7cbfeebe75d4476fb7816cb8cebd671f : SUCCESS in 3m 43s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/9804d149d00a4a3b851b1c7f121945be : SUCCESS in 7m 00s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/6c4e1187a862487a99b3589e4fb96623 : SUCCESS in 7m 36s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5d768b1eb544ce5ad1762a0902453dd : SUCCESS in 5m 16s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/6ebd536a910c4530b6e375fe3596b6b2 : SUCCESS in 5m 33s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/cc45c3ef9a7b4836b04126e1f0042bf4 : SUCCESS in 2m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0cd865fbee34e1c9137a798ddde4c85 : FAILURE in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14a3408c2a3b4d1490012d2dec45fb2b : SUCCESS in 2m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/831b6418e1274e5d9b4dd5c0de2973e8 : SUCCESS in 3m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea79b0f0f62f489781c2e9134091a895 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e82c706632564be69b65db9119e1761d : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":21},{"id":"c1e9ebaa4e2e708b6a8dbbf537325c64db055e39","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 10:53:29.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fcca0d830874937b3f700bf5f8b7c8a\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/95b3200b71f34f8589706c9dce397155 : SUCCESS in 19m 39s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/69825972600e4123a36b48f7553faffb : SUCCESS in 4m 53s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/953b8f0c05884cc584941a1046976bba : SUCCESS in 23m 52s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/b557b070a42641479d0f1f8ac5d41bb6 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bb52b8947d8c49a6b235db7037142e22 : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ae7698e5e6094d94a98920306e9f15c7 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/541b73f3164c4f6f861281d9b0e707fb : SUCCESS in 5m 41s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"1bbe308a81634cc8c8e504def637e0edb1a0579d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-12-02 11:35:02.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":21}],"current_revision_number":21,"current_revision":"f9b6fb46d0b9c8e4e8aa025cdb2444c56e864f62","revisions":{"00631d7893c39374bee171bf40128c6351d39cc1":{"kind":"REWORK","_number":1,"created":"2024-08-26 20:06:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/927210/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/927210/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/927210/1"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 20:06:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 20:06:32.000000000","tz":120},"subject":"DNM: Run fio tests","message":"DNM: Run fio tests\n\nChange-Id: Iab9eaf234eeb15b8edfca454f7b3aacdc4ead544\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00631d7893c39374bee171bf40128c6351d39cc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00631d7893c39374bee171bf40128c6351d39cc1"}]},"branch":"refs/heads/master"},"16b2cf8053adac4e136b8d852a3eb5ca990994f3":{"kind":"REWORK","_number":2,"created":"2024-08-26 20:09:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/927210/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/927210/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/927210/2"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 20:06:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 20:09:48.000000000","tz":120},"subject":"DNM: Run fio tests","message":"DNM: Run fio tests\n\nChange-Id: Iab9eaf234eeb15b8edfca454f7b3aacdc4ead544\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/16b2cf8053adac4e136b8d852a3eb5ca990994f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/16b2cf8053adac4e136b8d852a3eb5ca990994f3"}]},"branch":"refs/heads/master"},"f417468dcda6bc99c21ce15e4e8350f8cc57871a":{"kind":"REWORK","_number":3,"created":"2024-08-26 20:14:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/927210/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/927210/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/927210/3"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 20:06:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 20:14:09.000000000","tz":120},"subject":"DNM: Run fio tests","message":"DNM: Run fio tests\n\nChange-Id: Iab9eaf234eeb15b8edfca454f7b3aacdc4ead544\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f417468dcda6bc99c21ce15e4e8350f8cc57871a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f417468dcda6bc99c21ce15e4e8350f8cc57871a"}]},"branch":"refs/heads/master"},"299b2125207aeec331d737457a22f886af393170":{"kind":"REWORK","_number":4,"created":"2024-08-26 20:18:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/927210/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/927210/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/927210/4"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 20:06:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 20:18:51.000000000","tz":120},"subject":"DNM: Run fio tests","message":"DNM: Run fio tests\n\nChange-Id: Iab9eaf234eeb15b8edfca454f7b3aacdc4ead544\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/299b2125207aeec331d737457a22f886af393170"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/299b2125207aeec331d737457a22f886af393170"}]},"branch":"refs/heads/master"},"2d579625c8d5fdc25dedffe332b2ed3ab69eb225":{"kind":"REWORK","_number":5,"created":"2024-08-27 15:25:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/927210/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/927210/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/927210/5"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 20:06:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-27 15:25:43.000000000","tz":0},"subject":"DNM: Run fio tests","message":"DNM: Run fio tests\n\nChange-Id: Iab9eaf234eeb15b8edfca454f7b3aacdc4ead544\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d579625c8d5fdc25dedffe332b2ed3ab69eb225"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d579625c8d5fdc25dedffe332b2ed3ab69eb225"}]},"branch":"refs/heads/master"},"b69da39903279af9f6c88ca87f8e0f0fb10e68b3":{"kind":"REWORK","_number":6,"created":"2024-08-27 18:48:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/927210/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/927210/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/927210/6"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 20:06:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-27 18:48:50.000000000","tz":0},"subject":"DNM: Run fio tests","message":"DNM: Run fio tests\n\nChange-Id: Iab9eaf234eeb15b8edfca454f7b3aacdc4ead544\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b69da39903279af9f6c88ca87f8e0f0fb10e68b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b69da39903279af9f6c88ca87f8e0f0fb10e68b3"}]},"branch":"refs/heads/master"},"aceed99b3cceade3956cac14e9dadf643ee26eb5":{"kind":"REWORK","_number":7,"created":"2024-08-27 18:49:53.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/927210/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/927210/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/927210/7"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 20:06:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-27 18:49:53.000000000","tz":0},"subject":"DNM: Run fio tests","message":"DNM: Run fio tests\n\nChange-Id: Iab9eaf234eeb15b8edfca454f7b3aacdc4ead544\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aceed99b3cceade3956cac14e9dadf643ee26eb5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aceed99b3cceade3956cac14e9dadf643ee26eb5"}]},"branch":"refs/heads/master"},"b0e5c09210111714f41a35373f0383072b8d997b":{"kind":"REWORK","_number":8,"created":"2024-08-27 18:50:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/927210/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/927210/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/927210/8"}}},"commit":{"parents":[{"commit":"fd4f1a8a82686bd0766526f007620ad8217b42e5","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fd4f1a8a82686bd0766526f007620ad8217b42e5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 20:06:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-27 18:50:17.000000000","tz":0},"subject":"DNM: Run fio tests","message":"DNM: Run fio tests\n\nChange-Id: Iab9eaf234eeb15b8edfca454f7b3aacdc4ead544\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0e5c09210111714f41a35373f0383072b8d997b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b0e5c09210111714f41a35373f0383072b8d997b"}]},"branch":"refs/heads/master"},"3f32dcb1cc573413a5a653ce28b4d5796b5a25b3":{"kind":"REWORK","_number":9,"created":"2024-10-15 19:16:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/927210/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/927210/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/927210/9"}}},"commit":{"parents":[{"commit":"ec4c4310dd870ff5bbdfd9762a0beb8389009b9f","subject":"Add way to backup from MariaDB replica server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec4c4310dd870ff5bbdfd9762a0beb8389009b9f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 20:06:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-15 19:16:34.000000000","tz":0},"subject":"DNM: Run fio tests","message":"DNM: Run fio tests\n\nChange-Id: Iab9eaf234eeb15b8edfca454f7b3aacdc4ead544\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3f32dcb1cc573413a5a653ce28b4d5796b5a25b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3f32dcb1cc573413a5a653ce28b4d5796b5a25b3"}]},"branch":"refs/heads/master","description":"Rebase"},"143fab4f467d43b06630c04cbfa43dde728905f1":{"kind":"REWORK","_number":10,"created":"2024-10-15 19:18:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/927210/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/927210/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/927210/10"}}},"commit":{"parents":[{"commit":"ec4c4310dd870ff5bbdfd9762a0beb8389009b9f","subject":"Add way to backup from MariaDB replica server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec4c4310dd870ff5bbdfd9762a0beb8389009b9f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 20:06:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-15 19:18:07.000000000","tz":0},"subject":"DNM: Run fio tests","message":"DNM: Run fio tests\n\nChange-Id: Iab9eaf234eeb15b8edfca454f7b3aacdc4ead544\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/143fab4f467d43b06630c04cbfa43dde728905f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/143fab4f467d43b06630c04cbfa43dde728905f1"}]},"branch":"refs/heads/master"},"66f0eec34bcb635e9418063b58cdc3d7d3eb73f0":{"kind":"REWORK","_number":11,"created":"2024-10-15 19:18:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/927210/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/927210/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/927210/11"}}},"commit":{"parents":[{"commit":"ec4c4310dd870ff5bbdfd9762a0beb8389009b9f","subject":"Add way to backup from MariaDB replica server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec4c4310dd870ff5bbdfd9762a0beb8389009b9f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 20:06:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-15 19:18:51.000000000","tz":0},"subject":"DNM: Run fio tests","message":"DNM: Run fio tests\n\nChange-Id: Iab9eaf234eeb15b8edfca454f7b3aacdc4ead544\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/66f0eec34bcb635e9418063b58cdc3d7d3eb73f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/66f0eec34bcb635e9418063b58cdc3d7d3eb73f0"}]},"branch":"refs/heads/master"},"ac4e257cf92ba24d5ca55a35a6cbc63146f6aec8":{"kind":"REWORK","_number":12,"created":"2024-10-15 19:19:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/927210/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/927210/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/927210/12"}}},"commit":{"parents":[{"commit":"ec4c4310dd870ff5bbdfd9762a0beb8389009b9f","subject":"Add way to backup from MariaDB replica server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec4c4310dd870ff5bbdfd9762a0beb8389009b9f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 20:06:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-15 19:19:14.000000000","tz":0},"subject":"DNM: Run fio tests","message":"DNM: Run fio tests\n\nChange-Id: Iab9eaf234eeb15b8edfca454f7b3aacdc4ead544\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac4e257cf92ba24d5ca55a35a6cbc63146f6aec8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac4e257cf92ba24d5ca55a35a6cbc63146f6aec8"}]},"branch":"refs/heads/master"},"5a4c84fcfe3fcc08f3d837cf184215a1aa558f73":{"kind":"REWORK","_number":13,"created":"2024-10-15 19:19:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/927210/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/927210/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/927210/13"}}},"commit":{"parents":[{"commit":"ec4c4310dd870ff5bbdfd9762a0beb8389009b9f","subject":"Add way to backup from MariaDB replica server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec4c4310dd870ff5bbdfd9762a0beb8389009b9f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 20:06:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-15 19:19:29.000000000","tz":0},"subject":"DNM: Run fio tests","message":"DNM: Run fio tests\n\nChange-Id: Iab9eaf234eeb15b8edfca454f7b3aacdc4ead544\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a4c84fcfe3fcc08f3d837cf184215a1aa558f73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5a4c84fcfe3fcc08f3d837cf184215a1aa558f73"}]},"branch":"refs/heads/master"},"c51641d295d70d9292ea98249deb55bbd623404b":{"kind":"REWORK","_number":14,"created":"2024-10-15 19:19:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/927210/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/927210/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/927210/14"}}},"commit":{"parents":[{"commit":"ec4c4310dd870ff5bbdfd9762a0beb8389009b9f","subject":"Add way to backup from MariaDB replica server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec4c4310dd870ff5bbdfd9762a0beb8389009b9f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 20:06:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-15 19:19:48.000000000","tz":0},"subject":"DNM: Run fio tests","message":"DNM: Run fio tests\n\nChange-Id: Iab9eaf234eeb15b8edfca454f7b3aacdc4ead544\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c51641d295d70d9292ea98249deb55bbd623404b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c51641d295d70d9292ea98249deb55bbd623404b"}]},"branch":"refs/heads/master"},"c75656e5efae0d5697f948ebbc644bf3a95bd679":{"kind":"REWORK","_number":15,"created":"2024-10-15 19:20:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/927210/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/927210/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/927210/15"}}},"commit":{"parents":[{"commit":"ec4c4310dd870ff5bbdfd9762a0beb8389009b9f","subject":"Add way to backup from MariaDB replica server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec4c4310dd870ff5bbdfd9762a0beb8389009b9f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 20:06:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-10-15 19:20:43.000000000","tz":0},"subject":"DNM: Run fio tests","message":"DNM: Run fio tests\n\nChange-Id: Iab9eaf234eeb15b8edfca454f7b3aacdc4ead544\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c75656e5efae0d5697f948ebbc644bf3a95bd679"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c75656e5efae0d5697f948ebbc644bf3a95bd679"}]},"branch":"refs/heads/master"},"c1f9c4fe221ad8c95fbbe16b13cef834536378db":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2024-12-11 12:15:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/927210/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/927210/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/927210/16"}}},"commit":{"parents":[{"commit":"da7562c73fb4139a6a32a2137850e371782a1390","subject":"Merge \"Revert \"[release] Use Dalmatian sources by default\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/da7562c73fb4139a6a32a2137850e371782a1390"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 20:06:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-12-11 12:15:32.000000000","tz":0},"subject":"DNM: Run fio tests","message":"DNM: Run fio tests\n\nChange-Id: Iab9eaf234eeb15b8edfca454f7b3aacdc4ead544\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c1f9c4fe221ad8c95fbbe16b13cef834536378db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c1f9c4fe221ad8c95fbbe16b13cef834536378db"}]},"branch":"refs/heads/master","description":"Rebase"},"6fada13125bde95358149f788f3114f20a0cb07b":{"kind":"REWORK","_number":17,"created":"2025-02-21 17:11:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/927210/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/927210/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/927210/17"}}},"commit":{"parents":[{"commit":"48423b908fcd64457776bfc5ae8f39d3ad69e172","subject":"Merge \"CI: Use ansible-lint for CI roles and playbooks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48423b908fcd64457776bfc5ae8f39d3ad69e172"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 20:06:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-02-21 17:11:50.000000000","tz":0},"subject":"DNM: Run fio tests","message":"DNM: Run fio tests\n\nChange-Id: Iab9eaf234eeb15b8edfca454f7b3aacdc4ead544\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6fada13125bde95358149f788f3114f20a0cb07b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6fada13125bde95358149f788f3114f20a0cb07b"}]},"branch":"refs/heads/master","description":"Rebase"},"08fb4187a42f307213c401ef6e1e45f8225a620d":{"kind":"REWORK","_number":18,"created":"2025-02-21 17:13:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/927210/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/927210/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/927210/18"}}},"commit":{"parents":[{"commit":"48423b908fcd64457776bfc5ae8f39d3ad69e172","subject":"Merge \"CI: Use ansible-lint for CI roles and playbooks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48423b908fcd64457776bfc5ae8f39d3ad69e172"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 20:06:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-02-21 17:13:17.000000000","tz":0},"subject":"DNM: Run fio tests","message":"DNM: Run fio tests\n\nChange-Id: Iab9eaf234eeb15b8edfca454f7b3aacdc4ead544\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08fb4187a42f307213c401ef6e1e45f8225a620d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08fb4187a42f307213c401ef6e1e45f8225a620d"}]},"branch":"refs/heads/master"},"bed29c18fdccea54ae901c426185c47e746815d0":{"kind":"NO_CODE_CHANGE","_number":19,"created":"2025-11-26 10:24:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/927210/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/927210/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/927210/19"}}},"commit":{"parents":[{"commit":"48423b908fcd64457776bfc5ae8f39d3ad69e172","subject":"Merge \"CI: Use ansible-lint for CI roles and playbooks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48423b908fcd64457776bfc5ae8f39d3ad69e172"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 20:06:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-26 10:24:23.000000000","tz":0},"subject":"DNM: Run fio tests","message":"DNM: Run fio tests\n\nChange-Id: Iab9eaf234eeb15b8edfca454f7b3aacdc4ead544\nSigned-Off-By: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bed29c18fdccea54ae901c426185c47e746815d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bed29c18fdccea54ae901c426185c47e746815d0"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ecf9e9786a9b138dab2e6beb8cd4462dded5207e":{"kind":"REWORK","_number":20,"created":"2025-11-26 10:24:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/927210/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/927210/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/927210/20"}}},"commit":{"parents":[{"commit":"0dedcb82e46b2ed05d8b53f7ddd8fb0ba19049f9","subject":"reno: Update master for unmaintained/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0dedcb82e46b2ed05d8b53f7ddd8fb0ba19049f9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 20:06:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-26 10:24:28.000000000","tz":0},"subject":"DNM: Run fio tests","message":"DNM: Run fio tests\n\nChange-Id: Iab9eaf234eeb15b8edfca454f7b3aacdc4ead544\nSigned-Off-By: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ecf9e9786a9b138dab2e6beb8cd4462dded5207e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ecf9e9786a9b138dab2e6beb8cd4462dded5207e"}]},"branch":"refs/heads/master","description":"Rebase"},"f9b6fb46d0b9c8e4e8aa025cdb2444c56e864f62":{"kind":"REWORK","_number":21,"created":"2025-11-26 10:24:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/10/927210/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/10/927210/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/10/927210/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/10/927210/21"}}},"commit":{"parents":[{"commit":"0dedcb82e46b2ed05d8b53f7ddd8fb0ba19049f9","subject":"reno: Update master for unmaintained/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0dedcb82e46b2ed05d8b53f7ddd8fb0ba19049f9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 20:06:16.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-11-26 10:24:40.000000000","tz":0},"subject":"DNM: Run fio tests","message":"DNM: Run fio tests\n\nChange-Id: Iab9eaf234eeb15b8edfca454f7b3aacdc4ead544\nSigned-Off-By: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9b6fb46d0b9c8e4e8aa025cdb2444c56e864f62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9b6fb46d0b9c8e4e8aa025cdb2444c56e864f62"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
