)]}'
{"id":"openstack%2Fkolla~927239","triplet_id":"openstack%2Fkolla~stable%2F2024.1~Ie133d6efc2e4c00002b8f0724e0183eaab35d1c5","project":"openstack/kolla","branch":"stable/2024.1","topic":"slurp-rabbitmq","attention_set":{},"removed_from_attention_set":{"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2024-08-28 11:05:32.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ie133d6efc2e4c00002b8f0724e0183eaab35d1c5","subject":"Add support for checking incoming RabbitMQ version","status":"ABANDONED","created":"2024-08-27 08:05:41.000000000","updated":"2024-08-28 11:05:32.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":926145,"cherry_pick_of_patch_set":1,"meta_rev_id":"c0ba4c2a548676cd21a22f7b7e0ea064baa7097f","_number":927239,"virtual_id_number":927239,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":0,"date":"2024-08-28 10:13:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-08-28 10:11:45.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2024-08-28 10:12:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-08-28 10:12:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-27 10:31:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-28 10:11:45.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-08-28 10:12:24.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"334e52b34d9ae98b9593e6b799ac3a737b9a3857","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-08-27 08:05:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"56e952008c99e8ff1ac274234a93847991176c20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 10:31:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91b2844c20be4748b8fce50d9e9cd24b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5bca62f1c55843a59ed8b1a4b5f253a4 : SUCCESS in 23m 05s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d2a0f6a2f0644d83843c11a5e7b5f314 : SUCCESS in 1h 09m 18s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7a5d7a9a5dd447099d9a267cad1d0044 : SUCCESS in 17m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/751cda7df2564d5e8a8438815c093364 : SUCCESS in 27m 39s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f179dd92ca044ab79759a316aefe3238 : SUCCESS in 49m 24s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5d54d4692eb244fc810beac28ba15c7f : SUCCESS in 1h 32m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c42d3efdfce4b989897500f64e7912e : SUCCESS in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1944e5a2ef1c4a4fa2544c75eb77ab70 : SUCCESS in 3m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/76496d091e4a4971bc7332d1f232d773 : SUCCESS in 2m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac5dd2fc6b954043872a5c413783f1b7 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd1de8eca2b44ac494717f726bfaee03 : SUCCESS in 3m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0ccc4ed7c2ae4a81b6d4f34885920ced : SUCCESS in 24m 02s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/01b35363cf6d4b41a14c38d49b6d8242 : SUCCESS in 27m 21s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/92de2b8e7292430094ae807c91862216 : SUCCESS in 1h 04m 58s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4c05fb9726ca47dba26e0983b551124e : SUCCESS in 1h 29m 27s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/fa576b4c82e9423d82e47c0753e9eee9 : SUCCESS in 1h 42m 37s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/953e39cf50d5469e90bbb93311b5d3d7 : SUCCESS in 2h 14m 12s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/094279e33ff04419a287757ebe90e580 : SUCCESS in 19m 41s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/549cfd5781de442dbf0d5afbf416535d : SUCCESS in 42m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6185ae2486044ed7920856172838dd2c : SUCCESS in 52m 21s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b360e51a66744b018d8691d2bf9cccad : SUCCESS in 1h 06m 59s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/4568e24129cc4582a5c7ac863de2c2e5 : SUCCESS in 1h 17m 29s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b03401227bf849fbb6936efb160c0fbb : SUCCESS in 2h 12m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8aec3c8180d051526ebece1d8c4c9dcedffce373","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 18:37:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0664d0b614c949f4b2208e63c0b94b0f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a027c8c101cc41cfa32ae53066a335e5 : TIMED_OUT in 3h 04m 42s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b599320e43514db29a93d15307f56085 : TIMED_OUT in 3h 03m 58s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d4cf245d80b74c9496aca96368aa44e6 : SUCCESS in 2h 51m 00s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c5809d56804d4bc4b68805007a3a6a13 : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/19238d8041f34aada14a2a7115225fd4 : SUCCESS in 5m 31s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a5a96136e836488eba35edfe92534ef4 : TIMED_OUT in 3h 05m 03s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d3c4067ef8c94245b505c388c7b926cb : TIMED_OUT in 3h 02m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1d598095f7a5182c42d7a8f99453efbbe0b509e6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 10:11:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8b457b602e5593c943c3efac2232eca8f335a22e","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-08-28 10:12:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a02bb8e9685a86075e897867125c7f392dee4a4d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-08-28 10:13:30.000000000","message":"Topic slurp-rabbitmq-stable/2024.1 removed","accounts_in_message":[],"_revision_number":1},{"id":"4a944abf828077369a8b40f3de10faa9ac632dfa","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-08-28 10:13:45.000000000","message":"Topic set to slurp-rabbitmq","accounts_in_message":[],"_revision_number":1},{"id":"2a6dfc404f21828f86119b8e9e7a3baa74ce3cde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 10:13:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c0ba4c2a548676cd21a22f7b7e0ea064baa7097f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-08-28 11:05:32.000000000","message":"Abandoned\n\nabandoned: feature not needed anymore, see here: https://review.opendev.org/c/openstack/kolla-ansible/+/926146/comment/e593d721_7c48143b/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6bd59f8c917092849b7c91616ac1060c23962ea0","revisions":{"6bd59f8c917092849b7c91616ac1060c23962ea0":{"kind":"REWORK","_number":1,"created":"2024-08-27 08:05:41.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/39/927239/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/39/927239/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/927239/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/927239/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/927239/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/39/927239/1"}}},"commit":{"parents":[{"commit":"7bc03b99ff4fd35fcece4aa4b840d3c5c0b7b45d","subject":"Configure Apache services to use correct locale","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7bc03b99ff4fd35fcece4aa4b840d3c5c0b7b45d"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-08-12 15:08:43.000000000","tz":60},"committer":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-08-27 08:05:41.000000000","tz":0},"subject":"Add support for checking incoming RabbitMQ version","message":"Add support for checking incoming RabbitMQ version\n\nWill be used to check if the current RabbitMQ version is high enough\nbefore upgrading to the new one.\n\nChange-Id: Ie133d6efc2e4c00002b8f0724e0183eaab35d1c5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6bd59f8c917092849b7c91616ac1060c23962ea0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6bd59f8c917092849b7c91616ac1060c23962ea0"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
