)]}'
{"id":"openstack%2Fkolla~927241","triplet_id":"openstack%2Fkolla~master~I876f3c88c5bcd3c0a7c776796e9870c8512f90e2","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-08-28 12:43:49.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I876f3c88c5bcd3c0a7c776796e9870c8512f90e2","subject":"Remove incorrect reno","status":"MERGED","created":"2024-08-27 08:24:47.000000000","updated":"2024-08-28 16:35:02.000000000","submitted":"2024-08-28 16:33:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927241","meta_rev_id":"4e950b2eee376aff9584099d9912f8a577355e1a","_number":927241,"virtual_id_number":927241,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-28 16:33:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-08-28 08:49:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-08-28 08:50:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2024-08-28 08:50:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-27 09:57:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-28 08:49:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"bceab70b0951897bb1a8125e40359d974235b1a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-08-27 08:24:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"383fa7f0b1ded2215817e26c6b825862dac69069","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 09:57:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77e886a16c0d4ef8a2124abac46b0047\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b691f1ac95ff4dbca49cf915843fefdd : SUCCESS in 25m 19s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/524e7386dc73405bb6dc2a1693456f2b : SUCCESS in 1h 06m 12s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b15e103c6b8e40639ff6df3ed458b33e : SUCCESS in 17m 33s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/91a265e393e442f98a76a0e25bbc771e : SUCCESS in 28m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3ccfc3c054884a57a64b8b02f815b0fb : SUCCESS in 1h 02m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c84b07f307fc4f4b899484ff1bf4d8c6 : SUCCESS in 1h 17m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d421ef906f24a76a3fa967edb5951fd : SUCCESS in 2m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d17e203b20d14aeea7b8f1fbcad7a5ea : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc0c693c5931461a88e8012308058c97 : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f806bdf5da740abb5eb6d97e25d4d5d : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c64d669f66914ce090138fdab3cf244d : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57101a3f4b9b44019326d0f03ad90402 : SUCCESS in 4m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b7537e58b4e047cab0ff4f7c75ab21d5 : SUCCESS in 26m 57s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/36d7abb72d9945e882fa4c73773539de : SUCCESS in 36m 00s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5adf2587d4dc44d988e18915b024bd10 : SUCCESS in 1h 06m 01s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/87a3c8506f0e484ba932535f2933cd95 : SUCCESS in 1h 21m 58s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3776a706ae83439eb46a47bb374d0abf : SUCCESS in 34m 45s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9fb1b99e58a1415c848199aed353340e : SUCCESS in 21m 02s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/990b16f2e6954f71b090dccc5f407f9e : SUCCESS in 1h 12m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1f9bb424c6cb4b08ad003b96af9d9d89 : SUCCESS in 51m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/309fe84418a24993b0924de106246067 : SUCCESS in 1h 13m 57s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/92e03267c159423a88f8be1471b99e6c : SUCCESS in 25m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a7d1fb1279b69bb4485a2419c3da5845bce468b5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 08:49:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0f8e93e4800946e13f5c8d68f8b96b562312f702","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 08:50:02.000000000","message":"Patch Set 1: -Workflow Backport-Candidate-1","accounts_in_message":[],"_revision_number":1},{"id":"994a086c76e357c0233d4304e1513790ee0fc07f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 08:50:06.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"87788515168f985c334f96acf216221925ae0f28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 08:50:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e636b5fc8d4cb7244f428afa70c5915a406460f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 10:31:49.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/30b8eff1b1f54cc5b4eddf82da7d802b\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8592a27e6a7b4c739a24066fb6adf67c : SUCCESS in 35m 15s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9ae0205dd8334a5895f836cc17211bf2 : SUCCESS in 40m 36s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ffc91f1c5de44efeaaef3d0e0eaa6272 : SUCCESS in 1h 19m 47s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/49d8ad4afc9d4cec95061fb9f4699ddf : SUCCESS in 1h 36m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0489b04e9c9440fd8a6a55bce10d8747 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8b5270dad114b769e224e95579b97cc : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e2df3b196c0e48d1b4c03f9b7db5ea75 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c062001caa64ea5984f638aec86b933 : SUCCESS in 4m 40s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/825978e79c1a4a9086b3a184aac75e4d : SUCCESS in 28m 47s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/541b8e90db534fc18a39296ffceca984 : SUCCESS in 37m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e0d06c6bcba04e64819dfc9d4b4047a9 : SUCCESS in 1h 07m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/67a0f70de4af410cb144491e1880d7e3 : SUCCESS in 1h 25m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b06ee914d8e3481c9d65f6ffe496442b : SUCCESS in 19m 02s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9e34ba22dd9b466dbee6dc5c10cbc5ac : SUCCESS in 1h 07m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/bb029a9aa4934b0c81e0a6284a2ce480 : SUCCESS in 53m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e589b623f7f447bb957d5b2d79242a2e : FAILURE in 6m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a1444e6eeac3b99cd04dfc0d83596e13155a9179","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-08-28 10:46:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2a52d4052675d11b10e83dc8debb62347a4a8e30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 12:24:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/909b38034bff43c9a48aa2fb8e41126b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ad494758089e49c4a6a5acfaee64eed7 : SUCCESS in 26m 44s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a845c0053bfe434fbd2ed87ad3e957d7 : SUCCESS in 58m 08s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/acea687287174e25a556f607eecff135 : SUCCESS in 16m 39s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f6d5a6fa69364144955193eb2d99ceb0 : SUCCESS in 27m 46s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6ac10b3fc3754b28b48d5575380b5759 : SUCCESS in 59m 33s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2b6b0d7eeefb4a85824f2e4cf1924f0b : SUCCESS in 1h 28m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebc451cf755b41cb96de737363b73004 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a8b3b60ff244d33b52360d3510d43ea : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/23e6d5af9f77437e85a8657a5fd08e4c : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bb28d0cd4e954aba83b7482f2fbcd4ef : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72de42ae8cc149319ac7b966f959e123 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa5038db546a47488e8d6adaa52cf20d : SUCCESS in 4m 22s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5ed14056a9434f1aaadfaf31cb33e46f : SUCCESS in 26m 31s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a459d7b90a9b4faea13b62232e028239 : SUCCESS in 45m 09s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/21e8fff8c8cf40c8911db31bd6bce935 : SUCCESS in 1h 03m 54s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b2314dc35b9945f3aee5c13e94c534db : SUCCESS in 1h 24m 37s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a4c0e436d69f4776a011f119faa91226 : SUCCESS in 38m 56s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/bd97a97e71b649af9c03561500e90b21 : SUCCESS in 19m 48s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c725c5199f8c48f4bc980cb8f594a05e : SUCCESS in 1h 00m 54s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/64faae91d2dc43818bcc3b45ff6ba376 : SUCCESS in 1h 02m 40s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e0d37443c9fd46f0a6d499464ac14342 : FAILURE in 45m 26s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e4bc7e450f63492fa8b7e451acf474dc : SUCCESS in 27m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1929a85c6b7f2e0296ec1ace6ba68e5745a09b9f","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-08-28 12:43:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"49d05d45ad9cd676a48cf3becce457818925dd0b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 13:51:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5e180aa46464a369fec0939923c5a57\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/20ddea55caec4e36819eec9a62ee28a3 : SUCCESS in 2h 28m 19s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/965475702f8544b1843afd7173cda226 : SUCCESS in 1h 43m 44s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ccf111d32ae4413984566c40e01975f9 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ecbb19db03894246a2c12f173cef5c71 : SUCCESS in 3m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"141abbd54d8c69e0ee4919ccb8eac677bb68863c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 14:57:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/824d7192b3da49fd9b83ae8953b377a8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f84eeaf32bc5498cb7d68cad36322150 : SUCCESS in 28m 43s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9ab43b171d5d42208a40ed65301131a8 : SUCCESS in 1h 05m 08s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d24d1bbc74254bc3a25b97cac928c77d : SUCCESS in 18m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b87483b56c324ad994b5cbbb22e348f0 : SUCCESS in 38m 11s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/59c8f368e14644eb90d8ab504162687a : SUCCESS in 57m 11s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/93ee6a45e464447caacff8c4deb88969 : SUCCESS in 1h 28m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa405898c16848f8ad2eea29326a75cd : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6778c9e38c154870899e926f0525528f : SUCCESS in 2m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/be5568ceec33481c87a7d459b3ae1dbd : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6dceebab0de94fea86881407d3d5052a : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6542e9cb543f45bd9586dc69a63d706e : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1315ea73d2f749f6bef718aad503b5af : SUCCESS in 4m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d3b4e351d646496fb8277212ac48c2ed : SUCCESS in 30m 04s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7e0f45c843a54250b0d2dd89ca89e973 : SUCCESS in 40m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/91b1a05504794b3fb2d3c0bed69462ec : SUCCESS in 1h 05m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7c42c56e600444c8ac1019db7ac1e126 : SUCCESS in 1h 17m 15s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5f849dfee46e406e9820d8f8c23a63f1 : SUCCESS in 33m 44s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/40a27df2cdb64e4eb78d0e12ad9211c9 : SUCCESS in 16m 36s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/45a0e74028df47d4ab5efab51eea785d : SUCCESS in 59m 13s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c5094a08c4fe440a8775931a4ac3aca8 : SUCCESS in 51m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0bc9d62011a94f95a5dc206c7871177b : SUCCESS in 1h 12m 48s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/282326cd32ac42db861d14a81bdf44b8 : SUCCESS in 29m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd1631cd4f6c8c4389b5ece555a2f40c355438ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 14:58:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4ed34954bcaa238d0dc8602cbaeb375450f5d93d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 16:33:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39b84e1ba36349b8b1880b32349c71ac\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/224298553af34e69a63c95b0ec89f1ce : SUCCESS in 18m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b5229481d7554c68a092a9b7796ccffd : SUCCESS in 27m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9bfba74324994ac395f8b83191094541 : SUCCESS in 55m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c9e32dea801c4014b6dbe47ca454e385 : SUCCESS in 1h 22m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a447b175383440448fc1bbfc0667812d : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a80e2835114649e8bca5908805f83261 : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/391d7f7743374f2d97b4dd3c3c1ef180 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49c4cd899320453db082e2d9c721addc : SUCCESS in 4m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0aecb56213114920a1509df2d7b0f194 : SUCCESS in 27m 22s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/97e55356a67842c9b84d4ac72abedd1d : SUCCESS in 35m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7f55fc928c3a4e46b8ae64547701c133 : SUCCESS in 1h 05m 22s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/20fc62661a844e08bd0f58b7b23d5de0 : SUCCESS in 1h 27m 19s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/393fa874668543daa121a413d712c490 : SUCCESS in 20m 04s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a8498af96fa3424dbc3c63541826f38c : SUCCESS in 58m 58s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f6d5461d0adf4c87a86e9916b1e51ad1 : SUCCESS in 58m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/15ce9618714d4ef99572b52b48d7ba19 : SUCCESS in 1h 19m 02s","accounts_in_message":[],"_revision_number":1},{"id":"840788ccda27bc3bdd63629104e05e18de7d9ad6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 16:33:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4e950b2eee376aff9584099d9912f8a577355e1a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 16:35:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6089b382d425468da9aed2669fffd020\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9309c9d1965e4e97a2c6824bd63dd3a3 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a8c7df6ea8b1dee2aa1199f1957384e3718c5317","revisions":{"a8c7df6ea8b1dee2aa1199f1957384e3718c5317":{"kind":"REWORK","_number":1,"created":"2024-08-27 08:24:47.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/41/927241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/927241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/927241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/927241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/927241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/927241/1"}}},"commit":{"parents":[{"commit":"ccb13b672d4a178fefa624b8ab5b214a0bec7e45","subject":"Merge \"bifrost: bump Ansible version to 8 (rpm) and 9 (deb)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ccb13b672d4a178fefa624b8ab5b214a0bec7e45"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-08-27 08:20:40.000000000","tz":120},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2024-08-27 08:20:40.000000000","tz":120},"subject":"Remove incorrect reno","message":"Remove incorrect reno\n\nThe cAdvisor update change [1] mistakenly added a reno to the wrong\nfolder. Moving it back to the correct folder would accidentally add it\nto the master branch. Therefore, the reno is being removed.\n\n[1] https://review.opendev.org/c/openstack/kolla/+/704018\n\nChange-Id: I876f3c88c5bcd3c0a7c776796e9870c8512f90e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8c7df6ea8b1dee2aa1199f1957384e3718c5317"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a8c7df6ea8b1dee2aa1199f1957384e3718c5317"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
