)]}'
{"id":"openstack%2Fkolla~927290","triplet_id":"openstack%2Fkolla~stable%2F2023.1~I7f0f723b37e552c4ea01bf51b6354209608d904a","project":"openstack/kolla","branch":"stable/2023.1","hashtags":[],"change_id":"I7f0f723b37e552c4ea01bf51b6354209608d904a","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","status":"MERGED","created":"2024-08-27 15:21:40.000000000","updated":"2024-08-27 17:26:22.000000000","submitted":"2024-08-27 16:57:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927290","cherry_pick_of_change":927289,"cherry_pick_of_patch_set":4,"meta_rev_id":"d0665d7167f4124e90a6bd3cabf161e7638b8166","_number":927290,"virtual_id_number":927290,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-27 16:57:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-08-27 15:22:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-08-27 15:22:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-27 15:22:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-27 16:40:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac67d638ae82627a30fe410f7ef500d2345e07e0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-27 15:21:40.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.2.\n\nThe following files contain Git conflicts:\n* .zuul.d/rocky.yaml\n* .zuul.d/ubuntu.yaml","accounts_in_message":[],"_revision_number":1},{"id":"7069c9c5834144124d3e55580419ae0f411905ca","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-27 15:22:08.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"27e91e540caa0ad8608a3947dca4ea05568b587d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-27 15:22:14.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"454c248361c72938e0883c43cb542c0304c69d60","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 15:22:22.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2023.1.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/kolla/.zuul.d/rocky.yaml@stable/2023.1\", line 48, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/kolla/.zuul.d/rocky.yaml@stable/2023.1\", line 49, column 1","accounts_in_message":[],"_revision_number":1},{"id":"af1848b783e511305301223665f225b8157bc8cd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-27 15:22:31.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d4540f0555de7eeccabe72a323da05f7eb33d214","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 15:22:34.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/kolla on branch stable/2023.1.  The\nproblem was:\n\n  while scanning a simple key\n    in \"openstack/kolla/.zuul.d/rocky.yaml@stable/2023.1\", line 48, column 1\n  could not find expected \u0027:\u0027\n    in \"openstack/kolla/.zuul.d/rocky.yaml@stable/2023.1\", line 49, column 1","accounts_in_message":[],"_revision_number":1},{"id":"1754b81f5be757e2266082ca8715b6de6086ab15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 16:40:06.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0808fd7b27a4bc3936febbd8d69d3e7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3c957e869a14ac8ae1fe0bfca48a8ba : SUCCESS in 2m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9789149263614f1b86648d8d5a7f5680 : SUCCESS in 2m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aec3d8812b7a43ed8c4fb876c491d587 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8be3ae8e1d0d4aebbc6bb0648a91d686 : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":2},{"id":"841029d03a0c3773119f75d37dc303c40b54139e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 16:40:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7ae2f28c36623599fd238eb020e7444485c3fbd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 16:57:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/def3deb1cb05452ba454ff49516fedfb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af8be354195043999576038b066af28a : SUCCESS in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/038a5dd1b24446bba8845ab8eb0cf84a : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cde44509255c4e748d48b616aa3feb4d : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":2},{"id":"2718e612ee27f4dd7344a6908432bbd6d734d588","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 16:57:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"39df95a39d4b18c6e8c546ba557bef34679b34d1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 17:01:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/278ee5350c5646ce8f71cf266ee98c86\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ac8e1a176cd415e8e6be64a0ce00f4e : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2},{"id":"d0665d7167f4124e90a6bd3cabf161e7638b8166","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 17:26:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fbe5fadd82cd443289a97b79330f1ab8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f06d755dcc1e481aa3d193d3da020b08 : SUCCESS in 4m 04s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0d55fe0f78eaa24742e4b8f8dc318f9fc3b95299","revisions":{"350f6a52d85d006cb130315d2003d39301af4bf2":{"kind":"REWORK","_number":1,"created":"2024-08-27 15:21:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/90/927290/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/927290/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/927290/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/927290/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/927290/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/927290/1"}}},"commit":{"parents":[{"commit":"a7a602b9bbd4d25780584534e882e7f33513506b","subject":"Configure Apache services to use correct locale","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7a602b9bbd4d25780584534e882e7f33513506b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 15:16:03.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-27 15:21:40.000000000","tz":0},"subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","message":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs\n\nSince Linaro shut down their nodepool provider - OpenDev infra\nis left with OSUOSL cloud with max 15 instances running - let\u0027s\nnot put any additional strain on CI (especially that it seems\nthere are some I/O or similar issues that cause our jobs to\ntime out).\n\nChange-Id: I7f0f723b37e552c4ea01bf51b6354209608d904a\n(cherry picked from commit fd4f1a8a82686bd0766526f007620ad8217b42e5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/350f6a52d85d006cb130315d2003d39301af4bf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/350f6a52d85d006cb130315d2003d39301af4bf2"}]},"branch":"refs/heads/stable/2023.1"},"0d55fe0f78eaa24742e4b8f8dc318f9fc3b95299":{"kind":"REWORK","_number":2,"created":"2024-08-27 15:22:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/90/927290/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/927290/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/927290/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/927290/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/927290/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/927290/2"}}},"commit":{"parents":[{"commit":"a7a602b9bbd4d25780584534e882e7f33513506b","subject":"Configure Apache services to use correct locale","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a7a602b9bbd4d25780584534e882e7f33513506b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-26 15:16:03.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-27 15:22:08.000000000","tz":0},"subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","message":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs\n\nSince Linaro shut down their nodepool provider - OpenDev infra\nis left with OSUOSL cloud with max 15 instances running - let\u0027s\nnot put any additional strain on CI (especially that it seems\nthere are some I/O or similar issues that cause our jobs to\ntime out).\n\nChange-Id: I7f0f723b37e552c4ea01bf51b6354209608d904a\n(cherry picked from commit fd4f1a8a82686bd0766526f007620ad8217b42e5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d55fe0f78eaa24742e4b8f8dc318f9fc3b95299"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d55fe0f78eaa24742e4b8f8dc318f9fc3b95299"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
