)]}'
{"id":"openstack%2Fkolla~927352","triplet_id":"openstack%2Fkolla~master~Iad25ccde7cec22183e6d1fde4e399ffd88918bf9","project":"openstack/kolla","branch":"master","hashtags":[],"change_id":"Iad25ccde7cec22183e6d1fde4e399ffd88918bf9","subject":"DNM: Reproduce bug 2074075","status":"ABANDONED","created":"2024-08-28 07:38:44.000000000","updated":"2024-11-21 10:58:04.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"4e1fe115c50af2354e778da2128995766a42db86","_number":927352,"virtual_id_number":927352,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-08-30 10:10:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"reviewer_updates":[{"updated":"2024-08-28 08:15:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-29 07:03:20.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"}],"messages":[{"id":"0dc8e73edc97b350b11c4b2befb9c6ed57d6ce08","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 07:38:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe55a976b448fc1282ea48b26155be68eecd0853","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 07:49:24.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"57969859ec43f318d1f1d7ef212972fe6892d87f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 08:15:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2e88cb8e52348d585bc1497f9cd355f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/aeeb82b22d074cedaf1bc3c28f737e31 : FAILURE in 15m 26s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ea6fb381e4914b258516e74db0bf6f68 : FAILURE in 16m 38s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d21053ab09e94175898370f3b6b87cc4 : FAILURE in 15m 21s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/40fc1caaa1b842acb6223d51bbe23480 : FAILURE in 14m 08s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/15e5153b880143c0b3071a991b980072 : FAILURE in 10m 14s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d7f874ed3afb4b47ae59610aeb3222be : FAILURE in 9m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed58d3cfd0d345acb994a819cd11cf1a : SUCCESS in 2m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/15912d518d474069a7166cc9c229d97a : SUCCESS in 2m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/03731867fef242828a328467c5537406 : SUCCESS in 2m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dfd28fb5844d48188bd89bc18a70ae8d : SUCCESS in 2m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8960079f70f343daab8cde945ca77a43 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c187003c14347a280880923f5792d0b : SUCCESS in 6m 02s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/afc591169fda444f8d18ede3eabb1c59 : FAILURE in 20m 12s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/79dab2ec070b4b0b9cbded37001c9408 : FAILURE in 21m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/16030863e7044e9db4e44440097142c2 : FAILURE in 21m 01s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2ecea6293ab3431aa70a8d3864ebf614 : FAILURE in 16m 17s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3fec998243c54918ae20255aec52d92b : FAILURE in 11m 07s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/20212cf45cb740cebbb9b9cc6fc4ef66 : FAILURE in 19m 07s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ad1c2609bfda4a5bb2b27c29de8297ab : FAILURE in 10m 42s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6ecafd8e5a0a493d91457210570cb46e : FAILURE in 11m 44s","accounts_in_message":[],"_revision_number":2},{"id":"75a2022e17e87dc261985bd07e32ae1479b57ece","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 08:32:54.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"29bc81b4a6c9099f7ffc8ea1db321ee804498b54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 09:52:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce3b7da2b1bf47e887800a8e706e743c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/56b0cfd4af00461ab63c8bd9228cdfd6 : FAILURE in 51m 14s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/cc639d7638b444e5971929fe1bd00c55 : FAILURE in 40m 20s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1ee448f0cb2d4f42a7ed5907b7ffae47 : FAILURE in 23m 18s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5be7588ba65d4c7aa1dc2baa0c03433a : FAILURE in 24m 39s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/889d0402e5b2436dbfad59979235458f : FAILURE in 21m 48s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7499e55502664490aa26fe86f8f8a0f8 : FAILURE in 27m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d3cd2b0fdb14d10a2120f6a111eea6f : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1285dc462ff44816b16e54bf01dfea83 : SUCCESS in 5m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e68b792ebf7541b4912b27eada2badbd : SUCCESS in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/91f137dd8bec429997a372e82004ba95 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1224bce749bd41bd9784806c79c6bac1 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/643ced61be0d4ad3b0c88bb8c66e25b5 : SUCCESS in 4m 29s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a0d55eabfbcb4c56b66e90bf411dc4d6 : FAILURE in 39m 04s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/fa26bebc2c2c4a74bd8dd5d304892f8d : FAILURE in 44m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/58fd271ef25d49eda9292779d9708990 : FAILURE in 30m 18s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cf99648c80c54207b2bc2120ea41d371 : FAILURE in 32m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f95a35c559a847738429f3ca7fc69dda : FAILURE in 17m 55s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/026643288aa7455f8313e638f1fd62e2 : FAILURE in 55m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/bcb4474469e64c5da3867e9d7f9622a2 : FAILURE in 36m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c48188f0ffaf4f24a7e08d820422fdee : FAILURE in 14m 41s","accounts_in_message":[],"_revision_number":3},{"id":"39eaa1d5d1288cdf22731a6e89311b4eadd62a42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 11:06:08.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae26eea128674ffe941e04c7248d6d32\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9c7ca3118cdc484c8128e91a9a165e0f : FAILURE in 1h 56m 02s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cd35ed5d4673429f8dc371919fe299b4 : FAILURE in 30m 46s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a5f488347024c64b17f0d0b85e93441 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a3c269d4ed2e475fb7d864199c68a215 : SUCCESS in 4m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dad5d3e146f5e57ebc7749e2ebf3342d127fe9f4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 11:34:01.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"05505c2ca729bec3a91f168fe30c3605bf38aafd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 12:39:42.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"632c717b22b9477f7957de35be368c528c9856fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 15:08:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f3735d075184f61825ba8606d4257b1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f987fe647c1e429485e20e6f5761d34e : SUCCESS in 25m 53s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6db492ff66784079bcc21a9d4ba73363 : SUCCESS in 1h 11m 31s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/72801d6b4cd54b3bac804e27cff56a90 : SUCCESS in 17m 38s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fd53020382c64d57babf8abd9b56ce99 : SUCCESS in 29m 34s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0490170f92ef4d0db127b77aa0a4c85f : SUCCESS in 1h 08m 37s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/028cf639a0ee4157ab61b9cc18789663 : SUCCESS in 1h 34m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de7fae3718424f178da28bae86327718 : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62c761557ba4408e82a1d8e4beb77ab6 : SUCCESS in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36d56a4820c544599fcc707524f2b84b : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8356e830b8424ee5977ba9c9d9f8c5f0 : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09176ad32faf4c3b821bef950080c027 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5a1083ed67944018bd919b4edaa4d64 : SUCCESS in 4m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/185ef38f9c804a278c150b96d1b47f0a : SUCCESS in 25m 37s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/4f072c3aa69944069a200fcaf2f60571 : SUCCESS in 32m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fd3b57f7867443bc965950f891e61450 : SUCCESS in 1h 04m 17s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e2dde3c4b6b746f6951d827e589202c8 : SUCCESS in 1h 29m 36s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/874fee2ae5214d0ab58b4eeda139b066 : SUCCESS in 18m 16s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9d0a002ed29e41ca84862528807668d7 : SUCCESS in 1h 10m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d5b78c969af840459fd669b92e05a48a : SUCCESS in 1h 01m 29s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5266c1967f2149f39f4f4859777062e4 : SUCCESS in 1h 22m 45s","accounts_in_message":[],"_revision_number":5},{"id":"3b487d66a3a538b9200b520a1d2db285899c195e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 16:07:35.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"58084065c607b8123300f302b63b7a7a7fc29117","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 17:45:07.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d465d66e405e47da9e37c945986884e6\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/65ea6cec8ba8454dab302851e6b4890f : SUCCESS in 25m 53s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1833517b831b482cb77269571e2b7ef4 : SUCCESS in 1h 00m 22s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c25579d228a34b50967de33aaf0883cd : SUCCESS in 22m 25s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c700f9fc9f93450eb465d3f83bff3a24 : SUCCESS in 30m 50s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/23311a63616c49c8b28df9b2d48362c4 : SUCCESS in 58m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/23aab431d5ee41a8bfedb99917762670 : SUCCESS in 1h 34m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0114ab877d4b437d82763eaacc57f37f : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2b0fb66347e4779bea232ce16f7149f : SUCCESS in 2m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/92712dcbf3ca458cbb405e6d6f427b1c : SUCCESS in 4m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/34b91bf612034d738430ae6b59e5787e : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c36fc2aaa064ad89180612ba46a7dde : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbc2b09225bc4284a8c82c7018f0a769 : SUCCESS in 5m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b23a01d00a9f447bbfd9ecd74563fb11 : SUCCESS in 26m 50s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9e7e38c184e14c25a2c3bf965ce97ed4 : SUCCESS in 39m 43s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f89edc02999a4ae4bc88e468f7bc5416 : SUCCESS in 1h 13m 03s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6891c5302bea4e279c318ca7cbb9fc1d : SUCCESS in 1h 33m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a5cfafd6510744a593a2c67a28848705 : SUCCESS in 19m 14s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/10542d0d327d4eeab56abc584d87832d : SUCCESS in 56m 41s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/955ff41cf4db4df3b0c184596c9d2df7 : SUCCESS in 46m 19s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8874d03d869a400aa0766c80337b0551 : SUCCESS in 1h 16m 59s","accounts_in_message":[],"_revision_number":6},{"id":"0ec06f8faad8a0dbf8c394cfb3d80148d835eebb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 20:48:00.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cea15671d806468a8c9a48413b4b1de2\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1adf672d1acd4d148d402945f14de2e7 : TIMED_OUT in 3h 11m 20s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7a9ac666d06c407ebbe6c13c94b3b3a3 : SUCCESS in 2h 20m 11s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7b0cd8375c7042b2887fdf3155d9815c : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3a104dbc4d1e47759a4fd6b9cf1aad2a : SUCCESS in 4m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9b39bbc6d66415b54946257391ba8ffe989eddbe","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-29 06:22:04.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"f5fab02b0e72f04ef0773d4a26710d0bd914f68a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 07:56:38.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7285ce47ccbc4f80bf081cda3f49a60b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c7e4fa865b0a459c8a4026682f7d7906 : SUCCESS in 24m 55s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/66e05df661b7466db59de52d7c769a86 : SUCCESS in 1h 09m 33s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b130bea65fcf4394bcc3ee7e3d1b9060 : SUCCESS in 20m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0f9c8e2863054801beeeda376a326f6a : SUCCESS in 29m 11s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/34343922a7134444ad5ef6dfa82cd5c6 : SUCCESS in 1h 07m 17s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/65214e6bab584d6a81179bd574cf5ccb : SUCCESS in 1h 27m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd09453975554015a6a90eb5cfbd745b : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4cadf5baebbe448291af15cce0bd879e : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/94cc70aaca204dbeb0d826cb101a2b70 : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e1e1fc8ba9c842708ab0e92b300f4dd3 : SUCCESS in 2m 49s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1aca0440d82048349d6af8cebfad976d : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/855e1ddcce294a9cb487726383cd1112 : SUCCESS in 4m 15s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cc6855f98dda4574895854e4b3192b3e : SUCCESS in 30m 15s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/623a2f9b45ae46cab1249b36a647bf94 : SUCCESS in 36m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/889da52918ce4f89a2968533415d0936 : SUCCESS in 58m 39s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/17aaeea20f274589b8a03c4893ace175 : SUCCESS in 1h 22m 05s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/292c83e24d0d40519b515b01fb423750 : SUCCESS in 18m 18s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d93c6167bcd34026989efdd811e26920 : SUCCESS in 1h 00m 07s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1e22984ae93040bdafd38756f76c922f : SUCCESS in 49m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/305ed2613218416fad9ef4fa739bbec7 : SUCCESS in 1h 26m 08s","accounts_in_message":[],"_revision_number":7},{"id":"288604e92e0ad6bfdf5fa0149709dfef1e51dbf9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 10:42:56.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7850e66e87bc441b80ff49ff30c61d5f\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fcef3733d12e45878d08d275782e2ab3 : TIMED_OUT in 3h 03m 23s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b7d7ca6c37ab427ba3ecb5a872283216 : SUCCESS in 2h 50m 17s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/33d5d8f595b7441f9f5928d521c72eb0 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ea880597928e47cca789cce4d528aee5 : SUCCESS in 6m 53s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e83c389cdad5db4a05d39a02d8200de9b9fa328d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-30 08:36:16.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"11dbe64541dbd04ce07d29bd94944aa3aa9a0e59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 10:10:09.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45c0ff37a5104534973388adf181e334\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/664d7c71867043d1bcc102688c56934b : SUCCESS in 29m 04s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9d6658fa56424322bd551f566cfab928 : SUCCESS in 1h 03m 45s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1f02102abae14661980cd54a9753ad7c : SUCCESS in 21m 23s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/54385661c05040278ce5fa63c8adb27e : SUCCESS in 32m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4477809de3684934be99460eaa77985e : SUCCESS in 1h 09m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9364f44eacb54af7bfe8700704d3521f : SUCCESS in 1h 22m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1d490dcc9ac4f1d8b59556b4dff6bdd : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25802aca4fa742d2b12c1f756f2146f8 : SUCCESS in 3m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7fcb89402a874cf4b2af853c8aa0210e : SUCCESS in 2m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/030eb07ad03d4625a8cc20796581897e : SUCCESS in 2m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bb5523a89054329adf05aec3adf95d7 : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d22e93da00ee4ffb94ea54b64612a43a : SUCCESS in 4m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f1e59db6dca2433d90678a8611821885 : SUCCESS in 27m 03s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/107caaf3ecf34cb2b042c74ac8769723 : SUCCESS in 34m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/82b961016e8543b99cf81ea102fd594c : SUCCESS in 1h 06m 45s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e66b7257857d4d6e97be964782116d16 : SUCCESS in 1h 27m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c6ac8dfe84e44da4a6cab0948c8d2e2d : SUCCESS in 15m 36s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/00a2fdaa08ca4e05b518a334f7102b5e : SUCCESS in 1h 09m 04s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c2352bfd485e4bcc8b725c705e092e71 : SUCCESS in 46m 38s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ecc72867d34c4c129d22d72246a2d53c : SUCCESS in 1h 12m 54s","accounts_in_message":[],"_revision_number":8},{"id":"332a94f1f8d7f76b1ed571d325df0fcdab9efcfe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 11:11:40.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/567f6ed923524fd4bca8ecbcebab18c0\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e683305406a445a5bd341efbed008c75 : SUCCESS in 2h 33m 21s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dbbfb6265ad545fb8e1c823df5d3e5c1 : SUCCESS in 2h 21m 16s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bb5fc66965c24d24bc434b3b87b0b880 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/efbe2d6225fb408aa73d1165e93b36f9 : SUCCESS in 4m 18s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4e1fe115c50af2354e778da2128995766a42db86","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-11-21 10:58:04.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"6c5a9da115dab9525cc2a0f6ad5111d447d922f5","revisions":{"c3182e7e75b3b49a2907f7a70e5d5dbe7fa2c426":{"kind":"REWORK","_number":1,"created":"2024-08-28 07:38:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/927352/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/927352/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/927352/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/927352/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/927352/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/927352/1"}}},"commit":{"parents":[{"commit":"ece67cd2aec82d661ba355ce6bc6e41bb88001b6","subject":"Check if archive is tarfile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ece67cd2aec82d661ba355ce6bc6e41bb88001b6"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 07:38:23.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 07:38:42.000000000","tz":120},"subject":"DNM: Reproduce bug 2074075","message":"DNM: Reproduce bug 2074075\n\nChange-Id: Iad25ccde7cec22183e6d1fde4e399ffd88918bf9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3182e7e75b3b49a2907f7a70e5d5dbe7fa2c426"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c3182e7e75b3b49a2907f7a70e5d5dbe7fa2c426"}]},"branch":"refs/heads/master"},"c59c1f84531a803f764beff445982ba7dbfdc5c6":{"kind":"REWORK","_number":2,"created":"2024-08-28 07:49:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/927352/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/927352/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/927352/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/927352/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/927352/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/927352/2"}}},"commit":{"parents":[{"commit":"ece67cd2aec82d661ba355ce6bc6e41bb88001b6","subject":"Check if archive is tarfile","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ece67cd2aec82d661ba355ce6bc6e41bb88001b6"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 07:38:23.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 07:49:24.000000000","tz":0},"subject":"DNM: Reproduce bug 2074075","message":"DNM: Reproduce bug 2074075\n\nChange-Id: Iad25ccde7cec22183e6d1fde4e399ffd88918bf9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c59c1f84531a803f764beff445982ba7dbfdc5c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c59c1f84531a803f764beff445982ba7dbfdc5c6"}]},"branch":"refs/heads/master"},"26bc3611bd37eb6754d6eee2a58f7623202aa870":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-08-28 08:32:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/927352/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/927352/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/927352/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/927352/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/927352/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/927352/3"}}},"commit":{"parents":[{"commit":"515d49fca876452d1ae70f70c198fbcce7f01860","subject":"Fix symlinking of openstack-base source","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/515d49fca876452d1ae70f70c198fbcce7f01860"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 07:38:23.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 08:32:54.000000000","tz":0},"subject":"DNM: Reproduce bug 2074075","message":"DNM: Reproduce bug 2074075\n\nChange-Id: Iad25ccde7cec22183e6d1fde4e399ffd88918bf9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/26bc3611bd37eb6754d6eee2a58f7623202aa870"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/26bc3611bd37eb6754d6eee2a58f7623202aa870"}]},"branch":"refs/heads/master","description":"Rebase"},"a15d89f1fb1a47244d217df925339aa2f8f97371":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-08-28 11:34:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/927352/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/927352/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/927352/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/927352/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/927352/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/927352/4"}}},"commit":{"parents":[{"commit":"9aa3813940413a6f19039fc66ef8c62faa128217","subject":"Fix symlinking of openstack-base source","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9aa3813940413a6f19039fc66ef8c62faa128217"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 07:38:23.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 11:34:01.000000000","tz":0},"subject":"DNM: Reproduce bug 2074075","message":"DNM: Reproduce bug 2074075\n\nChange-Id: Iad25ccde7cec22183e6d1fde4e399ffd88918bf9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a15d89f1fb1a47244d217df925339aa2f8f97371"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a15d89f1fb1a47244d217df925339aa2f8f97371"}]},"branch":"refs/heads/master","description":"Rebase"},"f5429f9ac4239e8e7b21a36778a19e8e27d8df94":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-08-28 12:39:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/927352/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/927352/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/927352/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/927352/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/927352/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/927352/5"}}},"commit":{"parents":[{"commit":"65f6597358b75a005d76743a8e61033b3f3f651f","subject":"Fix symlinking of openstack-base source","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/65f6597358b75a005d76743a8e61033b3f3f651f"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 07:38:23.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 12:39:42.000000000","tz":0},"subject":"DNM: Reproduce bug 2074075","message":"DNM: Reproduce bug 2074075\n\nChange-Id: Iad25ccde7cec22183e6d1fde4e399ffd88918bf9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f5429f9ac4239e8e7b21a36778a19e8e27d8df94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f5429f9ac4239e8e7b21a36778a19e8e27d8df94"}]},"branch":"refs/heads/master","description":"Rebase"},"dd9367e08b679bf3f3766cb1cb350a31656c10c8":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-08-28 16:07:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/927352/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/927352/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/927352/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/927352/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/927352/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/927352/6"}}},"commit":{"parents":[{"commit":"2c4ddc5f12e6043a69c72fe32589593e812a94dc","subject":"Fix symlinking of openstack-base source","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c4ddc5f12e6043a69c72fe32589593e812a94dc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 07:38:23.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 16:07:35.000000000","tz":0},"subject":"DNM: Reproduce bug 2074075","message":"DNM: Reproduce bug 2074075\n\nChange-Id: Iad25ccde7cec22183e6d1fde4e399ffd88918bf9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd9367e08b679bf3f3766cb1cb350a31656c10c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dd9367e08b679bf3f3766cb1cb350a31656c10c8"}]},"branch":"refs/heads/master","description":"Rebase"},"a67eafda697f29b4c7836fdb1eb31d1d6a071042":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-08-29 06:22:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/927352/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/927352/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/927352/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/927352/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/927352/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/927352/7"}}},"commit":{"parents":[{"commit":"67f314b5563892743b23f3e48b0689ce39f5d015","subject":"Fix symlinking of openstack-base source","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/67f314b5563892743b23f3e48b0689ce39f5d015"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 07:38:23.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-29 06:22:04.000000000","tz":0},"subject":"DNM: Reproduce bug 2074075","message":"DNM: Reproduce bug 2074075\n\nChange-Id: Iad25ccde7cec22183e6d1fde4e399ffd88918bf9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a67eafda697f29b4c7836fdb1eb31d1d6a071042"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a67eafda697f29b4c7836fdb1eb31d1d6a071042"}]},"branch":"refs/heads/master","description":"Rebase"},"6c5a9da115dab9525cc2a0f6ad5111d447d922f5":{"kind":"NO_CHANGE","_number":8,"created":"2024-08-30 08:36:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/52/927352/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/52/927352/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/927352/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/927352/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/52/927352/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/52/927352/8"}}},"commit":{"parents":[{"commit":"978a2864f923fa2e65afa13808571ea7738729e6","subject":"Fix symlinking of openstack-base source","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/978a2864f923fa2e65afa13808571ea7738729e6"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-28 07:38:23.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-30 08:36:16.000000000","tz":0},"subject":"DNM: Reproduce bug 2074075","message":"DNM: Reproduce bug 2074075\n\nChange-Id: Iad25ccde7cec22183e6d1fde4e399ffd88918bf9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c5a9da115dab9525cc2a0f6ad5111d447d922f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c5a9da115dab9525cc2a0f6ad5111d447d922f5"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
