)]}'
{"id":"openstack%2Fkolla~927378","triplet_id":"openstack%2Fkolla~master~Ic0f9521c66240f13ff5b7fc94480dbbb76498339","project":"openstack/kolla","branch":"master","topic":"slurp-rabbitmq","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-08-28 11:10:52.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-08-28 15:50:58.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-08-28 15:50:58.000000000","reason":"Change was submitted"},"35263":{"account":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"last_update":"2024-08-28 15:50:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic0f9521c66240f13ff5b7fc94480dbbb76498339","subject":"Revert \"Add support for checking incoming RabbitMQ version\"","status":"MERGED","created":"2024-08-28 11:05:50.000000000","updated":"2024-08-28 20:02:15.000000000","submitted":"2024-08-28 15:50:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"revert_of":926145,"submission_id":"927378-slurp-rabbitmq","meta_rev_id":"d1c66964d82aedbde231769b2a78acf00d3ff9bf","_number":927378,"virtual_id_number":927378,"owner":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-28 15:50:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-08-28 11:10:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-08-28 20:02:15.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-08-28 11:10:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2024-08-28 11:10:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-28 11:05:50.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-08-28 11:05:50.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-08-28 11:05:50.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-08-28 11:05:50.000000000","updated_by":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8d05f8ec510186222a605f18d5d857357da22916","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"date":"2024-08-28 11:05:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ecf89ccac493700b1ad6f9747aa56e273354dcce","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-08-28 11:10:52.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"28ff410aa488927c92b59e65decbaca705547672","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 13:44:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23b6e91788244e428b5461ba87de32e9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d40b3871d1d34f45806a1ca060050cac : SUCCESS in 40m 55s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e6de507abf934a6eb63c49d2781de321 : SUCCESS in 1h 14m 49s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c86f160b5ad04a23b385d25032e59efc : SUCCESS in 15m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/27c4ef1e476a4729a9d79bf37fb253e7 : SUCCESS in 29m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c364276327564670a695c523f9d5df09 : SUCCESS in 1h 04m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c8540a8cef59430ea8c70e1d8ac8af95 : SUCCESS in 1h 10m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f4b6d0e0aae43898751a2eb679d53d6 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fccc80effca64fc7bbc3271757b35088 : SUCCESS in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b6352187bf6341f7a91ad42ba2bcf399 : SUCCESS in 3m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/06e7f3eb932b42b0b11e7aca5782c269 : SUCCESS in 3m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df6446f15f874f8f887ed1492fca120f : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46882cca42684b84b12b2801035841f2 : SUCCESS in 4m 52s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ac509ad26f844ec5b799c62c6d1578c6 : SUCCESS in 27m 35s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c0c1a518753b4f478b5f0e29322c1eaf : SUCCESS in 35m 20s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/49ba3ada74f74f4a8fa8d3b198a3d1c4 : SUCCESS in 1h 09m 32s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e626498a2d9a41eaaa3c41dd298d37d7 : SUCCESS in 1h 28m 25s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/e061e138330748a9b247f00be1721f87 : SUCCESS in 1h 34m 15s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/45553eb900b348a097a9601a395409e0 : SUCCESS in 2h 29m 30s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6ef901b300ba43ae952f77aa94739e6f : SUCCESS in 14m 42s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/161b3b677a884fe49203a741a23b3512 : SUCCESS in 1h 01m 22s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a7b3b93faab243058432bf2ebd5c41c0 : SUCCESS in 50m 52s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/625adea1e89b4e3cb745b82b1c1f46fc : SUCCESS in 1h 03m 08s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/19e8985f0c8541edaf2ef9f0432b3aa9 : SUCCESS in 1h 39m 50s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/88c77634e87f40cf84c656d08f3bfa6e : SUCCESS in 2h 13m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f86cfa9d30b2aee5bf105faf94b04cf72b01257b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 13:45:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"20d3c5413d4c3bb796fa966236c084403c7b6bfb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 14:03:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfc56dd51f0b4fc091c6bb1262755275\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f96091c451cd459ab1a56d4dc7491f2c : SUCCESS in 2h 37m 47s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f991762c94d346a88dc3293545081e84 : SUCCESS in 1h 38m 19s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/569ba42147f04162967b343361773bb4 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/fb1bacab0fd84e8f9ec0ce3987087e6e : SUCCESS in 3m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a2674bdd7160913a08de064ed3fb72b23e441fa9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 15:50:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9140c0bdc8f4ebfaecc3fe9446da7c5\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7c9c5d59b16349c08754f86c7157d822 : SUCCESS in 18m 10s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/16c9916da9aa4a46831057dbcdf8f6f3 : SUCCESS in 27m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ce149eaa578f4f15991bdb1b51e9f799 : SUCCESS in 56m 47s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/717b744e760542ea8410da652e7f5d83 : SUCCESS in 1h 18m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fe132faba6a483f8a576741a8dc127a : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f51f3e3c76b14eaeaa47ed869ee1d0fe : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d74170c956d443c290c88261800cabb1 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30d2700c50f84b4e8ac7b61b59eb6bd6 : SUCCESS in 4m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/bfd131bf22164d1199791bfed3ff1f3f : SUCCESS in 29m 06s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/39869529632d4ccfaa00771a5e3e2fc8 : SUCCESS in 39m 13s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/79ce437fc43040a88adcef7d1bb8409d : SUCCESS in 1h 09m 38s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/93f9ba79ad2743938fce647af646a1f6 : SUCCESS in 1h 26m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/78837f38e842477a9407da2d37c13563 : SUCCESS in 19m 36s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2038fbf2de5b4fc39ece8fa94a387be7 : SUCCESS in 1h 08m 27s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/12f8f5ba85024ad491a7fd45287979db : SUCCESS in 54m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4f28544dab114752be29940b2ac56ead : SUCCESS in 1h 28m 59s","accounts_in_message":[],"_revision_number":1},{"id":"0571bf12a9e8af1d2d13b9c9a635a8b90ede606d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 15:50:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f5d9294b602c63d5cb3200f13612d386ff42afd7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 15:52:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07b1f09ccdfd4ee489e5faf411d8ca72\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14d09e747e51478da27133bedc9be546 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1},{"id":"d1c66964d82aedbde231769b2a78acf00d3ff9bf","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-08-28 20:02:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"423a73d9d1bf2d2b9a4c1a71848506cad4ff1319","revisions":{"423a73d9d1bf2d2b9a4c1a71848506cad4ff1319":{"kind":"REWORK","_number":1,"created":"2024-08-28 11:05:50.000000000","uploader":{"_account_id":35263,"name":"Matt Crees","email":"mattc@stackhpc.com","username":"mattcrees"},"ref":"refs/changes/78/927378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/78/927378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/927378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/927378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/78/927378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/78/927378/1"}}},"commit":{"parents":[{"commit":"36a7cd053d5b13ecc9fad1866f244a733fdd58e5","subject":"Add support for checking incoming RabbitMQ version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/36a7cd053d5b13ecc9fad1866f244a733fdd58e5"}]}],"author":{"name":"Matt Crees","email":"mattc@stackhpc.com","date":"2024-08-28 11:05:50.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2024-08-28 11:05:50.000000000","tz":0},"subject":"Revert \"Add support for checking incoming RabbitMQ version\"","message":"Revert \"Add support for checking incoming RabbitMQ version\"\n\nThis reverts commit 36a7cd053d5b13ecc9fad1866f244a733fdd58e5.\n\nReason for revert: feature not needed anymore, see here:\nhttps://review.opendev.org/c/openstack/kolla-ansible/+/926146/comment/e593d721_7c48143b/\n\nChange-Id: Ic0f9521c66240f13ff5b7fc94480dbbb76498339\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/423a73d9d1bf2d2b9a4c1a71848506cad4ff1319"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/423a73d9d1bf2d2b9a4c1a71848506cad4ff1319"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
