)]}'
{"id":"openstack%2Fkolla~927722","triplet_id":"openstack%2Fkolla~master~I8b2677465686f77126fd74114d944936269214bd","project":"openstack/kolla","branch":"master","topic":"remove-wsgi_scripts","attention_set":{},"removed_from_attention_set":{"36624":{"account":{"_account_id":36624,"name":"Matúš Jenča","email":"matus.jenca@dnation.cloud","username":"matusjenca"},"last_update":"2024-09-04 07:27:37.000000000","reason":"\u003cGERRIT_ACCOUNT_36624\u003e replied on the change","reason_account":{"_account_id":36624,"name":"Matúš Jenča","email":"matus.jenca@dnation.cloud","username":"matusjenca"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-09-09 17:16:05.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8b2677465686f77126fd74114d944936269214bd","subject":"Change Manila container user to root","status":"ABANDONED","created":"2024-09-02 09:17:21.000000000","updated":"2025-09-09 17:16:05.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"48594d8bfc5fc0885f1091535ec3d0443afc8abf","_number":927722,"virtual_id_number":927722,"owner":{"_account_id":36624,"name":"Matúš Jenča","email":"matus.jenca@dnation.cloud","username":"matusjenca"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-13 12:48:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2025-05-13 11:02:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-02 10:48:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-04 01:13:09.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"44332591bbf8a6544260015594f53c78be39c624","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36624,"name":"Matúš Jenča","email":"matus.jenca@dnation.cloud","username":"matusjenca"},"date":"2024-09-02 09:17:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"06fa088c7efbaa5cd4a8728e4e1e7a46224072bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 10:48:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04dc998351e640baa3ab4661f6eccb19\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e847a6be7ae14c8cbd903277a2db8c30 : SUCCESS in 25m 40s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/efe77ecfc29d42d288b3bc687aeafdac : SUCCESS in 1h 02m 34s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/19dfaa9ffda6453aae7587147d8cb9b9 : SUCCESS in 16m 19s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d51cfce80fd74af2b9bbdef429bb03a5 : SUCCESS in 28m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d0ce5fc132d849a39ae2861417f17bf9 : SUCCESS in 49m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/166fff9e6fe343368aee2023a19ef2d2 : SUCCESS in 1h 20m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1ba10e047744e13ac2ace1f456d144a : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d675c9ce333476c8cc786af689e1a68 : SUCCESS in 2m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1a3b291ac9234c09aad028448a877991 : SUCCESS in 2m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ff5b41afb061409789fd45e9c48c14d1 : SUCCESS in 3m 05s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0f9c047225340118849584096dede7b : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/585cad1e2fbb439aa5062f3a50cdabd8 : SUCCESS in 4m 16s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d3c071dc0a084178b779e225017f6c10 : SUCCESS in 29m 05s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2c287f173ed9400189686b9fe4591910 : SUCCESS in 33m 53s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3fc6467449fa4c3fbeb53cb00be6fc99 : SUCCESS in 1h 04m 14s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f0dd4aa7bf2c4c3faf84115dc0a56923 : SUCCESS in 1h 18m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2df4894eb9554e4f9787971306b8ce8d : SUCCESS in 20m 26s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/aace80d045d041088bb6e4bdc52a7533 : SUCCESS in 53m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2795c75234ae4477bad10145f68f6f9c : SUCCESS in 53m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f044f155b54d4adb8c71b83286850031 : SUCCESS in 1h 07m 42s","accounts_in_message":[],"_revision_number":1},{"id":"9fd88720b423976b13d35c1efa7d32b3cd9db5a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 11:38:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c69fa03c06eb4788a1286970bc706aaf\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b825a74adece43a79499045736f4e93b : SUCCESS in 2h 19m 15s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bdb3c2aa6bc749b1906195f111b74461 : SUCCESS in 1h 20m 00s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/63913fab03e74770889d0103347bbfd7 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/66a9f944f862485199e9bd2e6bb9ec0b : SUCCESS in 3m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"81212ba7c8b473870536d6f743456b6aa6a57748","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-04 01:13:09.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a413db572b6bb33461763e0fdc75a615381e2a8f","author":{"_account_id":36624,"name":"Matúš Jenča","email":"matus.jenca@dnation.cloud","username":"matusjenca"},"date":"2024-09-04 07:27:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"23b751b47a2a4e6fc3f4c94aebe507d1cea7fe53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-13 11:02:35.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"49892981a1a4c5958a4f8d72ab7c199d097af653","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 12:46:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2c846b8234e4c8dbea4fe28ab2b0d19\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/815c3f88102e49159eb861c3d432f4a9 : SUCCESS in 41m 07s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/313083a763754c61970206fe782f6c55 : SUCCESS in 38m 30s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/36fd6f5d0b764b38855915abe047e66b : SUCCESS in 1h 42m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7c73e88ac77f48e39b881ee993185942 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/07a159de35ea4328a9088d8a4a177f73 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/54d0329b2ecd400eb71946692e44ece9 : FAILURE in 5m 52s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d3f993e8694c4ade9c43e37ab29dc567 : SUCCESS in 30m 11s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/03c21fb8c3eb489aa55893d127d520ff : SUCCESS in 47m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"117afbabf73aef8af037f87abb456a6d05548f71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 12:48:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/797be1ce16ef4200b8a46d79ebc6f68a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/41bc5283b8cf448d9a5372d998c16e10 : SUCCESS in 25m 54s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/94bf3c04ccf64edbbd3417c019b957ba : SUCCESS in 1h 05m 36s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/622f31d6cfb445fc8bc5d0586f1d0caa : SUCCESS in 14m 47s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/cec9a091ff8642fe9522f1f39d5500db : SUCCESS in 29m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0db1ba3b80dd43efa6b96c72db0f7b30 : SUCCESS in 1h 05m 49s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3daab32902d3481b8ab78aa329749949 : SUCCESS in 1h 29m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2e3b3b07a4f4ef89e0e1cadc5cd827c : SUCCESS in 2m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/91902af504ad467d8362cf1e1d421fc9 : SUCCESS in 3m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3310526939bd461aace619bf2e18d030 : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d0e161eb009141bfbad782e4d7a67f8b : FAILURE in 6m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9412d276d1441e5b07eddd70f3fde6b : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eed07caf89f8462498c364924c5d8292 : SUCCESS in 5m 46s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/80bd99763e784c08aa53bd66d4fe95cc : SUCCESS in 2m 01s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2185fecf809a4525947a93244153446b : SUCCESS in 23m 02s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c8542183b39a477a8026b0cc786b22f6 : SUCCESS in 1h 16m 26s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/079383819f3848ad8fcfb0bf2bb3d2e1 : SUCCESS in 1h 03m 16s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f11f47036d034204a779784f552d8537 : SUCCESS in 1h 36m 48s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/830d41ba50fc4bcd8b6384a0249bfc3e : SUCCESS in 13m 09s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3c723ebf5747494380400aeb292ecd72 : SUCCESS in 18m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7ac1d19b4d7845e8b8d27db4a45b6be6 : SUCCESS in 1h 00m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/57fe14579b4c4e3499c55e434c64f783 : SUCCESS in 1h 44m 00s","accounts_in_message":[],"_revision_number":2},{"id":"48594d8bfc5fc0885f1091535ec3d0443afc8abf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-09 17:16:05.000000000","message":"Abandoned\n\nManila has been reworked to use uWSGI, it\u0027s not needed anymore.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a65fe152012508f8bcea754e21f3fb968dd8b6fb","revisions":{"09654759a1a3e39053e8c160523bf90f3bb57e73":{"kind":"REWORK","_number":1,"created":"2024-09-02 09:17:21.000000000","uploader":{"_account_id":36624,"name":"Matúš Jenča","email":"matus.jenca@dnation.cloud","username":"matusjenca"},"ref":"refs/changes/22/927722/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/927722/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/927722/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/927722/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/927722/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/927722/1"}}},"commit":{"parents":[{"commit":"779a9a33f4c1f26b81b8e3e9c00b73318435b1fd","subject":"Merge \"Revert \"[release] Change RDO\u0027s DELOREAN_DEPS repo to Caracal\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/779a9a33f4c1f26b81b8e3e9c00b73318435b1fd"}]}],"author":{"name":"Matus Jenca","email":"matus.jenca@dnation.cloud","date":"2024-09-02 09:04:27.000000000","tz":120},"committer":{"name":"Matus Jenca","email":"matus.jenca@dnation.cloud","date":"2024-09-02 09:17:19.000000000","tz":120},"subject":"Change Manila container user to root","message":"Change Manila container user to root\n\nCurrently, Manila has container USER directive is \u0027manila\u0027.\nThis makes it impossible to use Apache Mod WSGI, because that\nrequires root access. This patch changes manila user to\n\u0027root\u0027 by removing USER directive, similar to other containers\nthat use Apache Mod WSGI such as Nova.\n\nChange-Id: I8b2677465686f77126fd74114d944936269214bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09654759a1a3e39053e8c160523bf90f3bb57e73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/09654759a1a3e39053e8c160523bf90f3bb57e73"}]},"branch":"refs/heads/master"},"a65fe152012508f8bcea754e21f3fb968dd8b6fb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-13 11:02:35.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/22/927722/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/927722/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/927722/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/927722/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/927722/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/927722/2"}}},"commit":{"parents":[{"commit":"a01e339edc7de6c1682a5b8fb19b053ec397210e","subject":"CI: Set upgrade jobs back to voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a01e339edc7de6c1682a5b8fb19b053ec397210e"}]}],"author":{"name":"Matus Jenca","email":"matus.jenca@dnation.cloud","date":"2024-09-02 09:04:27.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-13 10:58:18.000000000","tz":60},"subject":"Change Manila container user to root","message":"Change Manila container user to root\n\nCurrently, Manila has container USER directive is \u0027manila\u0027.\nThis makes it impossible to use Apache Mod WSGI, because that\nrequires root access. This patch changes manila user to\n\u0027root\u0027 by removing USER directive, similar to other containers\nthat use Apache Mod WSGI such as Nova.\n\nChange-Id: I8b2677465686f77126fd74114d944936269214bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a65fe152012508f8bcea754e21f3fb968dd8b6fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a65fe152012508f8bcea754e21f3fb968dd8b6fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
