)]}'
{"id":"openstack%2Fkolla~927741","triplet_id":"openstack%2Fkolla~master~Icbd001f1f487b25c4bfd6f6a78106a7f0ed1c8c8","project":"openstack/kolla","branch":"master","topic":"gnocchi-4-6-4","attention_set":{},"removed_from_attention_set":{"36624":{"account":{"_account_id":36624,"name":"Matúš Jenča","email":"matus.jenca@dnation.cloud","username":"matusjenca"},"last_update":"2024-09-16 21:44:33.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-09-14 08:16:59.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2024-09-16 21:44:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Icbd001f1f487b25c4bfd6f6a78106a7f0ed1c8c8","subject":"Gnocchi: Bump to 4.6.4","status":"MERGED","created":"2024-09-02 14:53:20.000000000","updated":"2024-09-16 21:45:41.000000000","submitted":"2024-09-16 21:44:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927741-gnocchi-4-6-4","meta_rev_id":"38985eaf0ab143f3914cfe366292589ccff246c3","_number":927741,"virtual_id_number":927741,"owner":{"_account_id":36624,"name":"Matúš Jenča","email":"matus.jenca@dnation.cloud","username":"matusjenca"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-16 21:44:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-09-15 11:24:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-09-14 08:16:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-09-16 19:44:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-09-16 19:44:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-02 14:54:49.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2024-09-02 16:39:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-13 19:23:55.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2024-09-13 23:03:12.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-09-15 11:24:53.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"ad14302570b64d75d595fed6884815cd3e37d16f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36624,"name":"Matúš Jenča","email":"matus.jenca@dnation.cloud","username":"matusjenca"},"date":"2024-09-02 14:53:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f49f87d2ba89a5ec394202e6229ccf02ff340edf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-02 14:54:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"10d19f077b07ebf4d28f3a81a9c91a7512dee70e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36624,"name":"Matúš Jenča","email":"matus.jenca@dnation.cloud","username":"matusjenca"},"date":"2024-09-02 15:04:21.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"93f7fbaa79a80fe90123dd5b3a9eb4fe06317eca","author":{"_account_id":36624,"name":"Matúš Jenča","email":"matus.jenca@dnation.cloud","username":"matusjenca"},"date":"2024-09-02 15:17:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f505ff031807eea4f718d5a2423e9d362dfb35ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 16:39:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0501d7d259644cd3b61a7b106ea0eaed\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1feee16de24d47149221d1d8f7851d42 : SUCCESS in 22m 45s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/360854f0baa04535a1fdf9265e69d3a8 : SUCCESS in 50m 32s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c3cd00b9fa9e4a89932602b2e1820466 : SUCCESS in 15m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/efd4656f65c34142a17a8212935b4217 : FAILURE in 5m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ef8a538f407f4eba9f0e0fc6c0c46aee : SUCCESS in 52m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f070a0443e3e4bd4879bb9061d1e2cfa : FAILURE in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed6f5c0bdaf74159af88f54003ac79e7 : SUCCESS in 2m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e922b56e726f4ccba0459cdfc86bb687 : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/deaed93071a24c7680d176432bfa4405 : SUCCESS in 2m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a5ca29ec50d14782a70305f4fc8ad25b : SUCCESS in 2m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1c1eaa754f947a6bee1f693d0a52e05 : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1efe58778bf54fd38efff936ab29689d : SUCCESS in 6m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3919ca9b355b4302a314fa30b320d159 : SUCCESS in 3m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3ee261ad2fc2436c9fbde03a1b1b3d88 : SUCCESS in 32m 59s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/57f262321655409faeba8a67a3982450 : SUCCESS in 43m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/08b074a662134ff691eeeb0014f0604e : SUCCESS in 1h 04m 42s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6eb73b5fb08f4645bf1edadc00377ab0 : SUCCESS in 1h 27m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/88430cdabc124a13a46222965c989657 : SUCCESS in 15m 21s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9f2c9e3604214e4dbce2b8aa63078147 : SUCCESS in 56m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3d472af467da4ff186fedaaa9eaa7f6e : SUCCESS in 47m 20s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/89071126ee97483baade76fd931a76dd : SUCCESS in 1h 10m 06s","accounts_in_message":[],"_revision_number":2},{"id":"33815f850ddfb2c7b2bdb607463f648a48e35358","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 17:06:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4ceca0e598a4499a458bad982760f24\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7ed285351ca74968b8227a1c329d9ee8 : SUCCESS in 2h 00m 54s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f747d5952e024ce49e7fb0b5df98718f : SUCCESS in 1h 33m 36s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1ff9705b3fde4424944a9d4793badfcc : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b963f384162c41f38c8d9454fbb8c1c9 : SUCCESS in 3m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c77fe17381932cd5edd4243a55a0c865485b9bc7","author":{"_account_id":36624,"name":"Matúš Jenča","email":"matus.jenca@dnation.cloud","username":"matusjenca"},"date":"2024-09-03 06:43:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2fe8e427303ca254ca35266fa292471f2e7262e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 08:20:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31ee12c46829477b9164639d66c8ba07\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/13d01bc844ca4c51a257df5933d7e3f2 : SUCCESS in 24m 37s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/28b68c8bf29645e3bbb45a25639dd498 : SUCCESS in 59m 31s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c070a4c2e1254e8a888d74dd3f1febf1 : SUCCESS in 16m 08s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/73a3329a299243ccbc4fd419c4e2cce7 : SUCCESS in 30m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/548017b9d358495984ccbb4dc790182c : SUCCESS in 1h 00m 39s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/90572e41841845ddb956f5c813b84c32 : SUCCESS in 1h 31m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed4d0eb907104b0abf18ddd4c2a6aada : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75ad97e082104c61ae3a1204be752d08 : SUCCESS in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/395cba53419745fd911d519e38ac9933 : SUCCESS in 2m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/df9839bc41b2481faee0dc90709c9c4e : SUCCESS in 2m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23d39027cf684c15a47c2e1340c8aef7 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbb1c1a99d4d40e491bb71a25e92356a : SUCCESS in 4m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eddd9dc234774e2eb854c2d7d8d9446c : SUCCESS in 5m 01s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/729bc2ef70d94c0eb61eeea40cc7c8de : SUCCESS in 29m 57s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e798bb551e3f4e3aad8e3ca306ab825c : SUCCESS in 34m 43s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4393b8ea55a6464d9f97bb03d6e570f9 : SUCCESS in 1h 05m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/22b26b6371ef4d70856e092f5a93664c : SUCCESS in 1h 27m 50s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/72e79b8ff3f54c7f8a207055bdb237ab : SUCCESS in 17m 57s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/094fec608def420d8c8f0a7a63f6bbf2 : SUCCESS in 57m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/71bb57a5c6b0416bbe20c3e3b875fdd3 : SUCCESS in 49m 41s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/fa06c9657535481a92110c291cf352f0 : SUCCESS in 1h 05m 38s","accounts_in_message":[],"_revision_number":2},{"id":"c66348a364f97a8822aae7e8cb64b581e08a8863","author":{"_account_id":36624,"name":"Matúš Jenča","email":"matus.jenca@dnation.cloud","username":"matusjenca"},"date":"2024-09-03 08:37:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1f9f040c1e4a395f4b476bbb9628ec2db9d261d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 09:00:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23b6a343767c4be5bbf9e1913e326eb4\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0e847e31fd324e53b5f70c014a71fc9c : SUCCESS in 2h 14m 29s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8c9e76feae13473399840cc9c143489b : SUCCESS in 1h 47m 42s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5f5bd40c2c80445ebf27684f129a3c92 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8037ca3466154f8ea5d360df3bbcf72c : SUCCESS in 5m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cf3b54e00cfc82e4a56050402c8002e067d4babe","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-13 19:23:55.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a10a92ba8ab16330445d5448013a69d7bec5cc42","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-13 23:04:42.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9974fde9da91a73f20aea33ca1969de0f08f055e","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-13 23:05:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d403cc322225063687aca773ab41ec746d2bd60b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-14 07:33:55.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"1ee24c347a68e12d82cbddcf628fa7d8194db581","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-14 07:35:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b2bba2370e87d81f166e78268c1b896ba1310847","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-14 07:35:25.000000000","message":"Topic gnocchi-4-6-3 removed","accounts_in_message":[],"_revision_number":3},{"id":"c2a28dc2a0b448f7a1d0844f8dcb88b93f84c118","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-14 07:35:34.000000000","message":"Topic set to gnocchi-4-6-4","accounts_in_message":[],"_revision_number":3},{"id":"4a96dda77d26a340087c4cea6bc1f1d47b148ff2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-14 08:16:59.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"54e61e53317b24f05390621bc5d50e7aa853e26d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-14 09:07:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbebf76d237f4ab8a51ffa9e8af72546\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4f1a2fdb1d0243469e2a32b8ee2e0340 : SUCCESS in 25m 51s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f82fbeb7cb2f4dba864fcb9d9162e95d : SUCCESS in 54m 29s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9bcd025408f8449da1ec5f711fce3e3e : SUCCESS in 20m 53s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e059754be2b047748731f73568f51471 : SUCCESS in 27m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a0131244d1484554ac68bcca45d8f148 : SUCCESS in 1h 05m 32s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f597618642ef4b9fb2f51611edeca52c : SUCCESS in 1h 18m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a9b36ebd1e5485a8ce508a38687d64d : SUCCESS in 2m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff12619bba49499586517a7d92a8f742 : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/54c4f583a4914968984c4052fcd6e424 : SUCCESS in 2m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b4259c9495a6458484f0229fb65a4d84 : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fce4804201d5465b98d333c2a179e3ee : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff88ff79b08a47d2b1f8a02067d15788 : SUCCESS in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/284fd34751724835b82851383ca7c3f5 : SUCCESS in 3m 33s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/65513f5dc23644b9829a797c2baf35fd : SUCCESS in 29m 13s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/62b91f53f64c4e55997c8ecc4b3bb621 : SUCCESS in 29m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/dd2307be5c6a49f3abba6e9ea7b4d92a : SUCCESS in 1h 04m 41s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9e32ef9d641542ca81715f5ed30e64f1 : SUCCESS in 1h 22m 00s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/74add8fa227c4c94a8ff9a3f47865ff7 : SUCCESS in 16m 42s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/391684f79f0e41738987b26ad3c4f08e : SUCCESS in 48m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/799499f75d174aeca5db3a0212569027 : SUCCESS in 46m 29s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1b4bc1a890c8425bb91a129a5f783d02 : SUCCESS in 1h 04m 09s","accounts_in_message":[],"_revision_number":3},{"id":"bd8950009ec9f19e7a23411b55a27d85fd3e4e38","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-14 10:43:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea86fc6dacb7493c9456c893aa312930\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ed063c4603c24bd9b13ffdaa12ca72cb : TIMED_OUT in 3h 07m 35s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cc24b4b4196a44d89b5630826f77e20b : SUCCESS in 2h 09m 42s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6b3ca83ce15243daaba4c46099e21b4b : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a2865ec2f5744c109122d436cda37ecc : SUCCESS in 5m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3de8258d6d621af4a640907caf3783a4161651dd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-15 11:24:53.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5d90a149385a4e31e4329b39361c76807237c51f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-16 19:42:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d36351de554504fb102980f09f1b84cb46cfbf11","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-16 19:44:25.000000000","message":"Patch Set 3: -Workflow Backport-Candidate+2","accounts_in_message":[],"_revision_number":3},{"id":"23a8ab8890e499ac3dca0cbd99b97eab81880a4b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-16 19:44:29.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8e7e30c9852c6daff987613fa3ec9e3daf95ad35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 19:44:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4ca0d99b8905f9d67913af1337835456817b0ef9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-16 19:52:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7b643d2a334f6e9e4201a264485f15be09dbd7c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 21:44:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d483b7fd44d4aedaf8e0f9cc14df4eb\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fe9900c7f1494a04bb352a4209ef660e : SUCCESS in 15m 49s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/df164742d3e847718637a8ece045304a : SUCCESS in 22m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1fd8dc95e2804a9facf54baefd8eaa34 : SUCCESS in 50m 36s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d6161b184b7a42c6a47988bf7679ce17 : SUCCESS in 1h 19m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f36fa72dd31d4e0e890bb52dffedecc7 : SUCCESS in 2m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5d0fd338523416dbe1111446ddac31b : SUCCESS in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6d7c015605f64272a3e959df69070402 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6ac66c8c3d5427c9bf4b329554b8c9f : SUCCESS in 4m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fbe5295149d548298de3061deff70a04 : SUCCESS in 3m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/40e575448133457db68ef867c475267b : SUCCESS in 28m 03s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/93f16c24836c40d4949003b0bf523597 : SUCCESS in 33m 30s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/882018620908455f94a50d716a9932cb : SUCCESS in 1h 05m 50s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/dfec19b922184c1696bfcb80ffcebf99 : SUCCESS in 1h 11m 50s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8de0a25d4e64457ca9fa6372caf640e0 : SUCCESS in 17m 43s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/65a265825b92412e90621284d2aece09 : SUCCESS in 1h 03m 57s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1a3bb4dfd06a4150aec3ffa0e5b27725 : SUCCESS in 47m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ae9851805e8c46ff87b3a06ede84a6fa : SUCCESS in 1h 15m 18s","accounts_in_message":[],"_revision_number":3},{"id":"12f7aa144d7d392ea9c4d9b78704dd6b5c75948c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 21:44:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"38985eaf0ab143f3914cfe366292589ccff246c3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-16 21:45:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e7e1e5ce6d742fc98e5b8d7e009f6c2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b600786e57c14bc3aeb1f15f7fb87432 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e8ebcc95c2440e89d7becc6788883d2 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"88b7ee1c2ae9bcb1506ddecd6199f8dd12d3b7ad","revisions":{"aa76fce8ea134ed47014e73e1c226804ef93c263":{"kind":"REWORK","_number":1,"created":"2024-09-02 14:53:20.000000000","uploader":{"_account_id":36624,"name":"Matúš Jenča","email":"matus.jenca@dnation.cloud","username":"matusjenca"},"ref":"refs/changes/41/927741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/927741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/927741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/927741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/927741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/927741/1"}}},"commit":{"parents":[{"commit":"779a9a33f4c1f26b81b8e3e9c00b73318435b1fd","subject":"Merge \"Revert \"[release] Change RDO\u0027s DELOREAN_DEPS repo to Caracal\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/779a9a33f4c1f26b81b8e3e9c00b73318435b1fd"}]}],"author":{"name":"Matus Jenca","email":"matus.jenca@dnation.cloud","date":"2024-09-02 14:44:31.000000000","tz":120},"committer":{"name":"Matus Jenca","email":"matus.jenca@dnation.cloud","date":"2024-09-02 14:47:23.000000000","tz":120},"subject":"Gnocchi: Bump to 4.6.3","message":"Gnocchi: Bump to 4.6.3\n\nThis makes Gnocchi work with Reduis Sentinel + TLS\nSee: https://github.com/gnocchixyz/gnocchi/releases/tag/4.6.3\n\nChange-Id: Icbd001f1f487b25c4bfd6f6a78106a7f0ed1c8c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa76fce8ea134ed47014e73e1c226804ef93c263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa76fce8ea134ed47014e73e1c226804ef93c263"}]},"branch":"refs/heads/master"},"5d5b62607cafced2e5a7084f86e4fe0ac6555040":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-09-02 15:04:21.000000000","uploader":{"_account_id":36624,"name":"Matúš Jenča","email":"matus.jenca@dnation.cloud","username":"matusjenca"},"ref":"refs/changes/41/927741/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/927741/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/927741/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/927741/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/927741/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/927741/2"}}},"commit":{"parents":[{"commit":"779a9a33f4c1f26b81b8e3e9c00b73318435b1fd","subject":"Merge \"Revert \"[release] Change RDO\u0027s DELOREAN_DEPS repo to Caracal\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/779a9a33f4c1f26b81b8e3e9c00b73318435b1fd"}]}],"author":{"name":"Matus Jenca","email":"matus.jenca@dnation.cloud","date":"2024-09-02 14:44:31.000000000","tz":120},"committer":{"name":"Matus Jenca","email":"matus.jenca@dnation.cloud","date":"2024-09-02 15:04:00.000000000","tz":120},"subject":"Gnocchi: Bump to 4.6.3","message":"Gnocchi: Bump to 4.6.3\n\nThis makes Gnocchi work with Redis Sentinel + TLS\nSee: https://github.com/gnocchixyz/gnocchi/releases/tag/4.6.3\n\nChange-Id: Icbd001f1f487b25c4bfd6f6a78106a7f0ed1c8c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d5b62607cafced2e5a7084f86e4fe0ac6555040"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5d5b62607cafced2e5a7084f86e4fe0ac6555040"}]},"branch":"refs/heads/master"},"88b7ee1c2ae9bcb1506ddecd6199f8dd12d3b7ad":{"kind":"REWORK","_number":3,"created":"2024-09-14 07:33:55.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/41/927741/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/927741/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/927741/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/927741/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/927741/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/927741/3"}}},"commit":{"parents":[{"commit":"6c5a443efe4e267639913e827918b83551318cab","subject":"Merge \"rabbitmq: Switch to Cloudsmith mirrors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6c5a443efe4e267639913e827918b83551318cab"}]}],"author":{"name":"Matus Jenca","email":"matus.jenca@dnation.cloud","date":"2024-09-02 14:44:31.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2024-09-14 07:31:42.000000000","tz":120},"subject":"Gnocchi: Bump to 4.6.4","message":"Gnocchi: Bump to 4.6.4\n\nThis makes Gnocchi work with Redis Sentinel + TLS [1][2].\n\n[1] https://github.com/gnocchixyz/gnocchi/releases/tag/4.6.3\n[2] https://github.com/gnocchixyz/gnocchi/releases/tag/4.6.4\n\nChange-Id: Icbd001f1f487b25c4bfd6f6a78106a7f0ed1c8c8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88b7ee1c2ae9bcb1506ddecd6199f8dd12d3b7ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/88b7ee1c2ae9bcb1506ddecd6199f8dd12d3b7ad"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
