)]}'
{"id":"openstack%2Fkolla~927773","triplet_id":"openstack%2Fkolla~stable%2F2024.1~I51c69d0673280e351daa410d943f769d33a10d6c","project":"openstack/kolla","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-09-03 10:11:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I51c69d0673280e351daa410d943f769d33a10d6c","subject":"keystone: Fix requests-kerberos installation","status":"MERGED","created":"2024-09-03 06:02:38.000000000","updated":"2024-09-03 18:23:40.000000000","submitted":"2024-09-03 10:11:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927773","cherry_pick_of_change":927461,"cherry_pick_of_patch_set":1,"meta_rev_id":"a058658a58e12a03800ea4919edcf904caca8097","_number":927773,"virtual_id_number":927773,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-03 10:11:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-09-03 07:50:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-09-03 18:23:40.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-09-03 07:14:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2024-09-03 08:28:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2024-09-03 08:28:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-03 07:14:11.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-09-03 07:48:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-03 08:28:45.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-09-03 18:23:40.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"119359aa6c17b9cc53d0fbd7fb7cff0cbc31fc45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-03 06:02:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7a33d916d944001136c06161e675fa95fbe42964","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-09-03 07:14:11.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"a1511ab3e8b52c403ec04ea720e02a63abb39b18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 07:48:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2195b5895d3f45e480ac22562de5da22\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/63c05de640e34df19c264f4bab1ec494 : SUCCESS in 21m 45s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/27aca2351c4046e88bcc75736b222f5f : SUCCESS in 1h 05m 58s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8a439491954a42d2bf50f6ba71020722 : SUCCESS in 19m 12s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c4808c63707644e89a4ce1c456a2f641 : SUCCESS in 30m 59s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/154ea608146e4bfa801f1e279df21b54 : SUCCESS in 1h 02m 21s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/62b028c858fb4f7ba5123a941b3479c6 : SUCCESS in 1h 39m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a18fe662dffb47d88edb9dd9a2d042cb : SUCCESS in 2m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69b96c38e132465ba539b93a4b61c283 : SUCCESS in 2m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/83d81b0286fb4b12a98ec6e3f9f28dbe : SUCCESS in 2m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30fe68e3a8404e0c96232f5d06ee9362 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f77c3eb90524f798d8ea186af3150e0 : SUCCESS in 4m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e8500a05e9264ef1a496eb305c159cc3 : SUCCESS in 21m 28s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c0481815a36b41109ad1553d7ebe9042 : SUCCESS in 28m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ab8fa549ae564444af4def30d9a78d34 : SUCCESS in 50m 42s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8e45d4eaa70048d19cb9c419a8ca3b86 : SUCCESS in 1h 18m 40s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b1cf3fa74f6f483a87f8048d914323ca : SUCCESS in 15m 49s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/708446c484564f7a8fb6ab3f4ccdfeb1 : SUCCESS in 49m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/76bcbb3ae6fe45b1871e51ac65ba3c6d : SUCCESS in 42m 32s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/53a3d5c82915429f88537877858e620b : SUCCESS in 1h 07m 58s","accounts_in_message":[],"_revision_number":1},{"id":"f68ff114bd078184d122757c3c51de51fe1e70dd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-03 07:50:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4a625f62c28954ee18067895b7b98d3e4672e574","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-09-03 08:28:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4425c38d47afeceffb3012c517803c3c1c3f73ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 08:29:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"71e5aab18e4638ada8219103b45796845eb12ee7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 08:32:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e94f4cab883e4055b9316047ac86bc9e\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/492f0dbac33d4e1493b60ef84c93fd63 : SUCCESS in 2h 28m 24s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a0a11c6ed607425e947ffb00ae397e5f : SUCCESS in 1h 40m 10s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/225a7705d59c4495b5b45b035d9e5376 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0ff3192d4fb948dd923d7c33c046e9d2 : SUCCESS in 4m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"92dc3170a82b5a81aa381075df6f222e4add1ad1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 10:11:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27a3eccf9d684ab5a8a5bf67aa1e388f\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3bc95932d5d84606a435a512598adc77 : SUCCESS in 23m 13s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f2b61785a2794e10ae73b81aa286a023 : SUCCESS in 30m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a369726901b14574a497e837ab330caa : SUCCESS in 1h 06m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/89420fbddc0d4a6998d4c8e2e7d0e6bc : SUCCESS in 1h 30m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56d8dabcd8274cc1a142e4898bcd78ba : SUCCESS in 2m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23f191fc27a74cd6889161899ee171c7 : SUCCESS in 2m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e9954c0bb969414f8d196a96b3d4bd6c : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/955ff8dfe3d640d990f09ed5c07ffef5 : SUCCESS in 4m 09s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9eddddd8e9a44771b60e2230bad4abe3 : SUCCESS in 28m 42s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/93a6c2d69b9745b18512698cd55bda0c : SUCCESS in 34m 28s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4c1a267ae7df432ea71449dfd5925fa3 : SUCCESS in 1h 01m 46s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f414a5e142a34ecca47e059dc8f0bd98 : SUCCESS in 1h 31m 02s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4dcde1581d3d46bca6c60fbf09878ad7 : SUCCESS in 25m 12s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ae3a9c5c22b54be7bb73832721bd3296 : SUCCESS in 56m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/27190efb09854b5aa2044619ea23d488 : SUCCESS in 50m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4ac5396cfbc842a097470e6c69d50cf2 : SUCCESS in 1h 20m 44s","accounts_in_message":[],"_revision_number":1},{"id":"2154e3a35c4b4df48848e5682413a3d8365fcb59","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 10:11:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e1e84583416272b139d67d0260638389c029ee80","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 10:12:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59cbf611277b4e6b9c141225ae553002\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ba15c645d8b47458ec710d41afe953b : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1},{"id":"a058658a58e12a03800ea4919edcf904caca8097","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-03 18:23:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9c54f236bcfcfb51818279b4ef4df4684f098bb8","revisions":{"9c54f236bcfcfb51818279b4ef4df4684f098bb8":{"kind":"REWORK","_number":1,"created":"2024-09-03 06:02:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/73/927773/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/927773/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/927773/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/927773/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/927773/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/927773/1"}}},"commit":{"parents":[{"commit":"08f6b89cc3a7150b17c41160dc7d85490bba3bd9","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08f6b89cc3a7150b17c41160dc7d85490bba3bd9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-29 06:19:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-03 06:02:38.000000000","tz":0},"subject":"keystone: Fix requests-kerberos installation","message":"keystone: Fix requests-kerberos installation\n\nSince it was bundled with plugins installation, and we did not\nhave any plugins in keystone container - it was not installed\nat all.\n\nAdding in libkrb5 dependencies for requests-kerberos.\n\nCloses-Bug: #2078305\nChange-Id: I51c69d0673280e351daa410d943f769d33a10d6c\n(cherry picked from commit 082d90cca02b0e3935363c7ce2e7ae718bcbb789)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c54f236bcfcfb51818279b4ef4df4684f098bb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c54f236bcfcfb51818279b4ef4df4684f098bb8"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
