)]}'
{"id":"openstack%2Fkolla~927775","triplet_id":"openstack%2Fkolla~stable%2F2023.2~I51c69d0673280e351daa410d943f769d33a10d6c","project":"openstack/kolla","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-09-03 10:11:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I51c69d0673280e351daa410d943f769d33a10d6c","subject":"keystone: Fix requests-kerberos installation","status":"MERGED","created":"2024-09-03 06:03:12.000000000","updated":"2024-09-03 18:23:34.000000000","submitted":"2024-09-03 10:11:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927775","cherry_pick_of_change":927461,"cherry_pick_of_patch_set":1,"meta_rev_id":"8880a6be010c3db03638eae611aba1f18ba7de27","_number":927775,"virtual_id_number":927775,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-03 10:11:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-09-03 07:49:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-09-03 18:23:34.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-09-03 07:14:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":2,"date":"2024-09-03 08:28:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":1,"date":"2024-09-03 08:28:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-03 07:14:22.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-09-03 07:45:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-03 08:28:38.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-09-03 18:23:34.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"94d4b32e445d0d410175beeb7a193441b29d5779","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-03 06:03:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6387723873a5eed05a04fffbe81e8c2c20a22627","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2024-09-03 07:14:22.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"36a62aff0c2dfbfe10095a19cf23f162be085aec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 07:45:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/177536b8c4e242aeab8527110f4a59ad\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9b8370336d8c413badc77c8037e9b427 : SUCCESS in 28m 03s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/77d3a0efbc7f4b28affafa5fb016aa04 : SUCCESS in 1h 09m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b850fc70c56f4957bafe8d30ae0096d9 : SUCCESS in 22m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/84ac5812e6674557abe753d1c6f22a1e : SUCCESS in 31m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7f6ba2f0391240f19c9c6972dfc215c7 : SUCCESS in 1h 06m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/46b894d80bf44ebf9e04f9f248e41a85 : SUCCESS in 1h 36m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2069cd93913a4f9f946e2096b736d0d0 : SUCCESS in 2m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aa935dfb565244ee9b21e8af3e0f81f3 : SUCCESS in 2m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6652bca496c340e69b4dc4dc8e0eee65 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7201037ae82b4667ae99546ed73be119 : SUCCESS in 2m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f2fbd1c45b044d99de3ab8b2ee081b2 : SUCCESS in 3m 09s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cda6fb87f8684484a873c0ac84669b7a : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b166d705e8b044ccb0969167a3e1dd52 : SUCCESS in 4m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e21c7dbb7fb5427eb850697b0315dda9 : SUCCESS in 24m 57s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/714e67492a3b41dc9c89bcb0f086fc60 : SUCCESS in 36m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e2aafafcae9c4dc19c855d3fa80fbd25 : SUCCESS in 1h 06m 48s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0128faa5f1224da2b22aafb936fec70d : SUCCESS in 1h 36m 58s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/156682fd448f455992c21434d6678dfd : SUCCESS in 18m 50s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e77ee3ac796e46da94939fdee434a9c5 : SUCCESS in 1h 05m 20s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5de8332e6cac478da04b02cc2f50c480 : SUCCESS in 55m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/018cb877d47e4b628e64c66ecd5394d5 : SUCCESS in 1h 09m 20s","accounts_in_message":[],"_revision_number":1},{"id":"569a04f07e6da85b7937a1939f067e40dcbb2b56","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-03 07:49:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e557cb649e04ea0d4bf27ccec9c1a0f06329ddcf","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-09-03 08:28:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4eaf74e2a4a0f170438328ae00f003986d4646fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 08:29:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"73b541f21c37076d04ec0577e22230446b439f7a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 08:40:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2af8183cfab4de29f27c16bcf2d12b5\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/41dc4f8128334323b3f5c1e068d8fc3c : SUCCESS in 2h 35m 28s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ea9fba6b25a541229d1083020d851a7a : SUCCESS in 2h 04m 40s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4bb8db5718f14a029e4cd3609e586f4a : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/68459a02166a4f9181b127a1cc637fb0 : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/19c250f12c734c75a51fae1fcf92a791 : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/309bedbd847f41afab5bd6356915ddf0 : SUCCESS in 4m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5f0c65cae137e1f6977da647e7f9751495c16c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 10:11:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32220a4863df44bba14ef50eaa26b3bb\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/bbeb649c311b4ee69281c94877928a5f : SUCCESS in 18m 22s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c20a5493494b4aa4a6213dd81defa68b : SUCCESS in 27m 53s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/aa2034efba704b93acdb2aa82adad515 : SUCCESS in 1h 08m 53s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5e208f5532e6432d9c23d23577bad104 : SUCCESS in 1h 29m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/219b7fbb740442e1a4162a1eccc60d19 : SUCCESS in 2m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e34eda658d9f4b02b6614a43e64557dc : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43e583b047bc403daca7fc962deaeb0f : SUCCESS in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02290d7c44f64b0bae4eea41f82c0b23 : SUCCESS in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4daa6543a5b84e3f9eed9753627654a9 : SUCCESS in 4m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/db5977b85f894430a70c740248f72575 : SUCCESS in 22m 43s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7c636e542a0749409d63b19da783003f : SUCCESS in 31m 22s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2d2d3d1a748d46d995d7d150f9220454 : SUCCESS in 1h 06m 22s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2c35e7b02c0e476a978840dece0de3a1 : SUCCESS in 1h 29m 22s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/71300dc0228c4244b3ea4700178a8685 : SUCCESS in 24m 13s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c81eee2fb22f467d85756486a15275ac : SUCCESS in 1h 00m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a0cd955beafe4ff9a0cedfbe103a59f6 : SUCCESS in 52m 32s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/46cac0eafedb4fb69a8d8f3c291064c2 : SUCCESS in 1h 25m 28s","accounts_in_message":[],"_revision_number":1},{"id":"fa0e35f2dcde56f13c58b9272330348bd61efd75","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 10:11:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b508ea3add2ca21873b13c7d176f5e81ee3e28e8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 10:12:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2184bd65b755402e889b1d95ba6b4fd5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bf3d05a2e7143998235aa59f5dcf2fd : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1},{"id":"8880a6be010c3db03638eae611aba1f18ba7de27","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-03 18:23:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f156fb394370b3dbb2c26a1655d830f9720f6a0","revisions":{"1f156fb394370b3dbb2c26a1655d830f9720f6a0":{"kind":"REWORK","_number":1,"created":"2024-09-03 06:03:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/75/927775/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/75/927775/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/927775/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/927775/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/75/927775/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/75/927775/1"}}},"commit":{"parents":[{"commit":"ae20362c62c08c191f4a94ce9e39e6dad5640cf6","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ae20362c62c08c191f4a94ce9e39e6dad5640cf6"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-08-29 06:19:04.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-03 06:03:12.000000000","tz":0},"subject":"keystone: Fix requests-kerberos installation","message":"keystone: Fix requests-kerberos installation\n\nSince it was bundled with plugins installation, and we did not\nhave any plugins in keystone container - it was not installed\nat all.\n\nAdding in libkrb5 dependencies for requests-kerberos.\n\nCloses-Bug: #2078305\nChange-Id: I51c69d0673280e351daa410d943f769d33a10d6c\n(cherry picked from commit 082d90cca02b0e3935363c7ce2e7ae718bcbb789)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f156fb394370b3dbb2c26a1655d830f9720f6a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f156fb394370b3dbb2c26a1655d830f9720f6a0"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
