)]}'
{"id":"openstack%2Fkolla~927798","triplet_id":"openstack%2Fkolla~stable%2F2024.1~I5a9162a30b23cfabb64dabe5c12c226fa8627abd","project":"openstack/kolla","branch":"stable/2024.1","topic":"fix-openstack-base-symlink-stable/2024.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-09-03 12:59:33.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-09-03 17:35:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5a9162a30b23cfabb64dabe5c12c226fa8627abd","subject":"Fix symlinking of openstack-base source","status":"MERGED","created":"2024-09-03 09:43:53.000000000","updated":"2024-09-03 18:21:05.000000000","submitted":"2024-09-03 17:35:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927798-fix-openstack-base-symlink-stable/2024.1","cherry_pick_of_change":924759,"cherry_pick_of_patch_set":15,"meta_rev_id":"05033fe089a83208bbe0c2d3bf4570d4c9da958d","_number":927798,"virtual_id_number":927798,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-03 17:35:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2024-09-03 12:59:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-09-03 18:21:05.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-03 12:59:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2024-09-03 12:59:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"CC":[{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-03 09:43:53.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"state":"CC"},{"updated":"2024-09-03 09:44:47.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-09-03 11:52:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-03 12:00:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-03 18:21:05.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"6ef591a152d947bb4ba9198fc4c1eb1ead4535fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-03 09:43:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5806d43117e21da964a91d2f2172553af3aaa447","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-03 09:44:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7af2ddb12dfa556be8e24879caf7f786592d1a94","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2024-09-03 09:44:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2f71692f3a50d35f4c36426d17396ecccd3f23e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 11:52:45.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0776ffb89740462c95a5c5ca4cb2871c\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6217cd5d61844f51976d9826a402ff2d : FAILURE in 1h 49m 17s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5afc29e0e59e4ff5b421f262729d00d8 : FAILURE in 31m 20s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e66bce1715b84ff7b896c7e5a36d0927 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9084f550d5974ac8bcd69c9577984d1d : SUCCESS in 4m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bbeb65012339f8a740c25fada0096ac801793357","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 12:00:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f68ab5732c444c1a386094966b36f85\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6e360416e8414eb79ab5d96fae852114 : SUCCESS in 30m 25s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/7e346f6bad8c4c35bfb79c3a4de3b4ac : SUCCESS in 1h 11m 35s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/dccd9f099c214b11a85fbd4eccfb29f4 : FAILURE in 23m 46s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ac6d4b423fa54b628b7a4ceaf199e2f1 : FAILURE in 29m 38s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a02a613ad71b4b1cb3a48a67430b4b59 : FAILURE in 7m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1e9d4afe16014d13b5d32aad770895f4 : FAILURE in 22m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8aa0913133940dbbbacaf4434e3b48e : SUCCESS in 3m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78790fe8527b482bbb455c6650998baf : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bda1e7892d7c40d49deb2d83dc7c47e0 : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87cd55e40b98428287dd5d29ceed79c5 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01576cc74d594577aee45b2e3164e98f : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3436b01174374603a859b8ead0f9914e : SUCCESS in 3m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/40ecdb98d24946c0bc469a04399eeacb : FAILURE in 24m 57s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2d01ea28d7d147e8a917a5fcbb326e16 : FAILURE in 35m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e9274defed8143fc87b1e44fb78dcfc1 : FAILURE in 22m 51s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/74e028c540b34329ab31c82b882f098b : FAILURE in 21m 13s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/39d03b105724498b8d621c67774fc08b : FAILURE in 24m 35s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/79d653e7fb0c401d9da02e388e4c0e73 : FAILURE in 23m 10s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/2e07234d16924304a79c29dcb776fdb8 : FAILURE in 22m 38s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/b7a90a59c79a421b8f6c26224f4e17e6 : FAILURE in 19m 59s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/ff70b20a2441433b8825caceb6f065f4 : FAILURE in 21m 24s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/7c92650695ac41a698049f19ad251e18 : FAILURE in 20m 00s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/3dcea7e63772441fab65749010267319 : FAILURE in 18m 56s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/00eb80e081934cb892d71ff371bf05f7 : SUCCESS in 25m 22s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/88c12f0cce044f6d88d8c3c152fd94cd : SUCCESS in 58m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e58ab54824a04335b67af21d1e70b5c6 : SUCCESS in 51m 55s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/552d1912710648c59b3bddecbad059a9 : SUCCESS in 1h 20m 01s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/ccd1923ea9cb4d539cf3a49b25653643 : SUCCESS in 1h 29m 16s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2ea43b1662f24922b7ef9d306feab1e5 : SUCCESS in 2h 03m 50s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/2264ce3e54624defa5339135e8d3fcb3 : SUCCESS in 1h 07m 06s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/6e8c0960d788428ead194db8a297bcfa : SUCCESS in 1h 09m 49s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/f84d81c00c9f4c1c9faa7ac44a0f10c9 : SUCCESS in 49m 51s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/f2ac4a87cd06427b8acde4ba90a5e298 : SUCCESS in 58m 33s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/87ef83d4ae594908bb5137f8ff168103 : SUCCESS in 1h 05m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a27cbddf5e8a3c350205a3676af3b71cdfb74bde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-03 12:59:24.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d7524d9dc430ef3efac315e529cdb376d3391136","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-03 12:59:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7958d0c8760021c6aab33d1ff936da2df503ea03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 15:52:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21c7f11ab13440789371a3e6d42b92b9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e72946c05e6545449b729a1dfb28fe12 : SUCCESS in 27m 04s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/676681754d8b45d68ac818e4457d35c9 : SUCCESS in 1h 05m 37s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ec814918edd348d6ad873e2840889ec4 : SUCCESS in 19m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ef5b2f9508d646bab3c9dc6ae2593cf7 : SUCCESS in 28m 36s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d80b7807fffb49f5bda1b9eb6a8a74d9 : SUCCESS in 1h 09m 11s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8ce52c67220e4422987bcb34c1594433 : SUCCESS in 1h 34m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dacc2198c964b98b5230058681b0566 : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0373bc264dde419aae55e46656d3348a : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c420fef02f941a1b71cadf460efd4db : SUCCESS in 2m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/50b3f90158fc4b1a9146479e28c3bf41 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a23c3f63f784443f858edad03613cb5a : SUCCESS in 4m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d405382903f4d1990c9aaa04080afb5 : SUCCESS in 3m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d36f79ecfed841ff9e96a300f30e375d : SUCCESS in 25m 45s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/f1919f4d2fce4dec831941b58b6b7cc4 : SUCCESS in 33m 29s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2e09cca7acc446ef8fe6db8a025c9259 : SUCCESS in 1h 08m 26s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ac98fe27195b4294acd6fe181b5858b0 : SUCCESS in 1h 29m 24s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/87fd9250636b43edac765d2290003b74 : SUCCESS in 2h 00m 35s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/fef28d83264142349357ad5fa42f9230 : SUCCESS in 2h 34m 49s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/f40d7eae56ec4af784cb2977e5067d50 : SUCCESS in 1h 17m 50s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/8e47a4af8a82450fbea5b24378927d1b : SUCCESS in 1h 29m 04s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/c5f6fbc347e541ee8f5ca9a0261cd3e9 : SUCCESS in 1h 07m 17s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/9f82925fd95c447f95d96b74847ed6f6 : SUCCESS in 1h 07m 17s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/085866c6522a4927819188a720e3581b : SUCCESS in 1h 29m 52s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/862196e79aa145069517fa496b886eb0 : SUCCESS in 25m 23s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a7a73a3550564e50833ff2af750a92ee : SUCCESS in 49m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/87ec5d60d1c14359a92d1f772c59207f : SUCCESS in 55m 36s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2477838bdb3a4617b1f97bc8f36cae9a : SUCCESS in 1h 14m 16s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/6add79c1e0334217aaa1a3d0ffd2a81d : SUCCESS in 1h 26m 51s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/66480c62d376492e96d94bf73617198b : SUCCESS in 2h 22m 07s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/27d5759108fd47b58a9d8a00223aa38b : SUCCESS in 1h 07m 10s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/e9e0045c59c2443a951bdbffdfc07b87 : SUCCESS in 1h 11m 19s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/e3c89dbe8647428594f23c1589c16aec : SUCCESS in 45m 32s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/2723722def1b443181f08faf24458038 : SUCCESS in 1h 00m 56s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/e58ccb3b8ae54264ad6a4b9cc86995a5 : SUCCESS in 1h 14m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0ac83cab7eda8da19a7c4ecfa2aad4002a6a9865","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 15:52:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9e36651a0f452b7a59e4e6fee9fdcd65fd1d1095","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 15:55:03.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9107f5f2477c422dbf4dd445628c77c0\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0071b9cc12a64a53928550d3d8b6fbe2 : SUCCESS in 2h 54m 00s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b47c99726f3c4d51a0e0a20e9af6f78a : SUCCESS in 2h 13m 03s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3abc774220f44d668c0683d288a7c074 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6c47f9f28dee4e429b8668e21709ce84 : SUCCESS in 4m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a5d03938fa721d7e2f8783dfc8565530932960f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 17:35:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0da6cb886dfa4d74b6b25f520fcee9fc\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6866e550c8f0411ca1023054b0e682eb : SUCCESS in 20m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a6d1428ddabf4bb5ae1b36de4e62e356 : SUCCESS in 28m 38s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f07d32150f8d4a5cb9c296c2e22d42bc : SUCCESS in 56m 39s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7f0821d4194e47869c6b9230120dcfa6 : SUCCESS in 1h 32m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adef009dad8249cba5d2351c1cecbd33 : SUCCESS in 2m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6701375c6ed54ac9ac83abe402abb0a4 : SUCCESS in 2m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/db5ff55205374d33aaf41c19098c8d9d : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6093e1c137964ba2bd277ab1172e2218 : SUCCESS in 4m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22949bb6f2634524b3429dfd2a571a48 : SUCCESS in 3m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1e553a01ef424a0fab46b52d6e36d856 : SUCCESS in 24m 20s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c63b5b283a8c44e8baac26f1475db90b : SUCCESS in 31m 48s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/522c37c80f7e4d0e9c83ec43b3b9426e : SUCCESS in 1h 04m 51s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1141fba7a22a44a9b96b8b672a669540 : SUCCESS in 1h 21m 42s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f46da2e162d34a5ba7b35e49f74bad03 : SUCCESS in 25m 46s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9b3b5582a1a04a2d8ff0508c4cfbda45 : SUCCESS in 54m 16s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a70b191f5ee248088e8947a54fedd129 : SUCCESS in 50m 04s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f7cda9c30dce409694384991d43bd0aa : SUCCESS in 1h 13m 10s","accounts_in_message":[],"_revision_number":2},{"id":"52a1392b0eff3f76a9f8ad938c61c61c75ee3aef","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 17:35:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b0b1551be0906861eabcee43d65263e864856961","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 17:36:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2062b3d3625d4e4097cf9e555051c11a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d84d00b39c78435894b0be404bb6a3e7 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/296bcd360f5140b3b45f8183f197cd12 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2},{"id":"05033fe089a83208bbe0c2d3bf4570d4c9da958d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-03 18:21:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4df43abd70c6045f0f8295fbc46f993fbf9290f4","revisions":{"6f48c9aa5963539c60becd032a9d1b9bbb1a4330":{"kind":"REWORK","_number":1,"created":"2024-09-03 09:43:53.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/98/927798/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/927798/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/927798/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/927798/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/927798/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/927798/1"}}},"commit":{"parents":[{"commit":"08f6b89cc3a7150b17c41160dc7d85490bba3bd9","subject":"CI: Disable CentOS/Rocky/Ubuntu aarch64 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08f6b89cc3a7150b17c41160dc7d85490bba3bd9"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-07-23 15:58:09.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-03 09:43:53.000000000","tz":0},"subject":"Fix symlinking of openstack-base source","message":"Fix symlinking of openstack-base source\n\nOriginal symlinking works fine with a tarball source for the\nrequirements repo.\n\nHowever, when user supplies git type source of it, we cannot\nguarantee that the source direcotry name contains ``requirements``.\n\nThis patch fixes it by moving plugins directory outside of\nopenstack-base directory, which ultimately allow us to use simpler\nsymlinking as openstack-base-archive would be the only directory\ninside.\n\nCloses-Bug: #2074075\nChange-Id: I5a9162a30b23cfabb64dabe5c12c226fa8627abd\n(cherry picked from commit 978a2864f923fa2e65afa13808571ea7738729e6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f48c9aa5963539c60becd032a9d1b9bbb1a4330"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f48c9aa5963539c60becd032a9d1b9bbb1a4330"}]},"branch":"refs/heads/stable/2024.1"},"4df43abd70c6045f0f8295fbc46f993fbf9290f4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-09-03 12:59:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/98/927798/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/98/927798/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/927798/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/927798/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/98/927798/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/98/927798/2"}}},"commit":{"parents":[{"commit":"9c54f236bcfcfb51818279b4ef4df4684f098bb8","subject":"keystone: Fix requests-kerberos installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c54f236bcfcfb51818279b4ef4df4684f098bb8"}]}],"author":{"name":"Seunghun Lee","email":"seunghun@stackhpc.com","date":"2024-07-23 15:58:09.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2024-09-03 12:59:24.000000000","tz":0},"subject":"Fix symlinking of openstack-base source","message":"Fix symlinking of openstack-base source\n\nOriginal symlinking works fine with a tarball source for the\nrequirements repo.\n\nHowever, when user supplies git type source of it, we cannot\nguarantee that the source direcotry name contains ``requirements``.\n\nThis patch fixes it by moving plugins directory outside of\nopenstack-base directory, which ultimately allow us to use simpler\nsymlinking as openstack-base-archive would be the only directory\ninside.\n\nCloses-Bug: #2074075\nChange-Id: I5a9162a30b23cfabb64dabe5c12c226fa8627abd\n(cherry picked from commit 978a2864f923fa2e65afa13808571ea7738729e6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4df43abd70c6045f0f8295fbc46f993fbf9290f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4df43abd70c6045f0f8295fbc46f993fbf9290f4"}]},"branch":"refs/heads/stable/2024.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
