)]}'
{"id":"openstack%2Fkolla~927886","triplet_id":"openstack%2Fkolla~master~Ia678484c001c66a9db675035dbb4505b0448bf58","project":"openstack/kolla","branch":"master","topic":"bug/2078886","attention_set":{},"removed_from_attention_set":{"35793":{"account":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"last_update":"2024-09-17 12:14:48.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-09-17 10:48:48.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-09-17 08:36:45.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"Ia678484c001c66a9db675035dbb4505b0448bf58","subject":"Fix cinder-backup: add missing sysfsutils,lsscsi package","status":"MERGED","created":"2024-09-04 03:43:09.000000000","updated":"2024-09-17 12:15:50.000000000","submitted":"2024-09-17 12:14:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927886-bug/2078886","meta_rev_id":"88d2cb88ed924fc2699f854857b9a5366d702a47","_number":927886,"virtual_id_number":927886,"owner":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-17 12:14:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},{"value":2,"date":"2024-09-17 10:48:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-09-17 08:36:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},{"value":1,"date":"2024-09-17 10:48:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-09-17 08:36:45.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},{"value":1,"date":"2024-09-17 10:48:48.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-04 04:22:13.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-09-04 06:43:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-04 09:22:06.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"592e27087ced48d9f0527ce30d16191d21095ce3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"date":"2024-09-04 03:43:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5affdb388e677961ccf2b0d0ebe69cca9a046669","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"date":"2024-09-04 03:45:40.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"11d980c02ed0a720b65ecc39379e5df2342965e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"date":"2024-09-04 04:11:22.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"1630285404dbb05aa779f400ea5e293d185315cc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-04 04:22:13.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"251aff133147193e23682cd3ec3145a0bf251492","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"date":"2024-09-04 04:37:14.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"eef253eca8e3de80069cf6d9fe1554c3cf407b51","author":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"date":"2024-09-04 04:40:23.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"134562e6df7ef66f42f679636239baf322ada0e5","author":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"date":"2024-09-04 04:40:45.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9dbafa9c915a5a4f389e6f51b9d1ff1ad21ddb66","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-04 04:45:22.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6bd945f38c0dd370cf7e3c32f7242bcb57224a48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"date":"2024-09-04 04:51:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6a9010c6f074d785db7af714c765d7aca5cfae7d","author":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"date":"2024-09-04 04:51:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"bff0e50b98a80ff59522169d42cc0736de4d720b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-04 04:54:25.000000000","message":"Patch Set 5: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"cb278baf059835cdfcfd0aaff97f1e4f21cbc436","author":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"date":"2024-09-04 04:55:46.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"2bd6f4a1e96f6bd792f899120694b74ecf0f5880","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 06:43:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/259bf8b416aa4a8f8c78533d35fa29f3\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8c37adce583a4557bc06d17695e54bd2 : SUCCESS in 25m 58s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6c09e68ebcf14c5fba48161e0db0ed7c : SUCCESS in 1h 08m 49s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b81c57e2412b48fab76cadc688910d69 : SUCCESS in 20m 01s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/59751713898a467da4059034b4be5e05 : SUCCESS in 28m 26s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/36c368a12d834e238567b0ecf9236252 : SUCCESS in 1h 01m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/837366de1d6f44df9e38d94dbc1b5d3b : SUCCESS in 1h 27m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70d5a27ce25648d28d1fe0617ad4d61c : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f2d734b7f0a4435af3db64454b23e6c : SUCCESS in 3m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4b1c9a98ebfe44de8426cc5bd6f02d7b : SUCCESS in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52e381ccaff842d69b996e24fb70c492 : SUCCESS in 2m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da9eb3954aea4e7c94cbd42956a76790 : SUCCESS in 3m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eadb8bab96a741e983dce6bba049e815 : SUCCESS in 4m 03s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/235d17d4c2b2487fb7fe7e9b506030c1 : SUCCESS in 27m 46s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/563acead3d934c1e85642b76e4502e58 : SUCCESS in 36m 33s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ae7fc2a2abe34cd8ae6b6163bc2c9452 : SUCCESS in 1h 04m 33s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/11308f98a3f24197bdb908eb04d26f50 : SUCCESS in 1h 27m 39s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/2c57cb06cbcd4f0ea0b37147f56e3e5f : FAILURE in 1h 34m 34s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/409d12fb825048fcaad0067d7df1a690 : FAILURE in 1h 40m 30s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/72eeab6a3ebd4136b187f0ddb3496b45 : SUCCESS in 28m 21s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d60ab8e98c154879b2114433c39a0c25 : SUCCESS in 45m 04s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/74a88427317041a0b5d280864e300061 : SUCCESS in 46m 39s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ae007d9933ff449abf38fe62a5d28612 : SUCCESS in 1h 21m 13s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/24ff2fc4cb314d4fb533704c524f8923 : SUCCESS in 1h 15m 56s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ed58d4becdd94eeea86fbb60f93ed341 : SUCCESS in 1h 49m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"12570b9d0e42c86aa13163ed489be54a748592ba","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-04 09:22:06.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"3129b21eadbe2b37964bb7a517b392f539c0eb6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 10:02:52.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6032f8c74766479b939a01f540518a38\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d4030764c17347f19b6419c11448abc7 : TIMED_OUT in 3h 04m 01s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9439ea30ad914daab4f56c124594ebb6 : FAILURE in 2h 21m 30s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a60d3bfe49c4a7f90fd1f3aa742a21c : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c52ca83d20df46088effe86c86b7edd5 : SUCCESS in 4m 53s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"e39b25ea8d617602ee6338dde5e2e349db6282cc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-12 14:26:39.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"05743fe35c4164ede0533c0351ce180a48f489bd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-12 14:28:17.000000000","message":"Patch Set 5: Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"5f467b8f396d65ced1539935a0581ffc0429ab9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"date":"2024-09-17 01:55:11.000000000","message":"Patch Set 6: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":6},{"id":"baabf4f1f4069e667bd906881e1da3fdfaee5593","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 04:24:12.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5278194ab74847a5994fb85866cbe88a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/10a819ce65674b33ba44d936aa62b340 : SUCCESS in 30m 55s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/4b7a9d6aac8443428883496154db569d : SUCCESS in 1h 00m 22s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e56d06edca5247869f3a5dc06651e6c9 : SUCCESS in 36m 49s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/592090b007e54068b6d1f1845b43f1a0 : SUCCESS in 44m 42s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4655c62a9b6d4462b12278094d9cb29a : SUCCESS in 56m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ebe69564683a4903a8785ee2dc725c15 : SUCCESS in 1h 25m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62cbe0b29fb44135b4368a9363978c35 : SUCCESS in 2m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b138772595264522b2a2e3cc2b367b08 : SUCCESS in 2m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1097266ac378416b818a178c4bd1f85a : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e97a412b166c4706b5d5ed08941e9f80 : SUCCESS in 3m 17s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8a60ae0132643819d3581f1096b96f1 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb3d214724de4db1b1cc60dd97ff1e9f : SUCCESS in 4m 48s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/592fed8e2a5f4aec8ee36920c5baf43a : SUCCESS in 31m 36s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/4cf9f1a75e5e4375a4957a901e652868 : SUCCESS in 36m 16s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7262639fb0ad417f912c87425bebb07d : SUCCESS in 1h 17m 13s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8abece78a10c40148b4a049efbacaf08 : SUCCESS in 1h 23m 13s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/89748dd85d784d6da1fa8182c8389674 : SUCCESS in 1h 53m 17s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/91e3ca8ef0b6400da80a5f54bf83e4c9 : SUCCESS in 2h 28m 09s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a42a2e0de3014ec4bb29009b7b72e831 : SUCCESS in 27m 40s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f8eccb53a29b4c738d3b1a188046af33 : SUCCESS in 53m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3b21549d3ede4794ad9696b245d777a5 : SUCCESS in 54m 54s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c00a446db50e4055a0152a6c00d33704 : SUCCESS in 1h 16m 08s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/e81e2596db6c410fa78eefa4014cb557 : SUCCESS in 1h 54m 56s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5be0fdf36d144c3a8a3b04b74e83ca1f : SUCCESS in 2h 09m 56s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"038be088abaf4ca71c5b81d231839b44f8619fbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 05:00:10.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50dd34d7dab14721a749f7cb8d18db81\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bcbf9f21a5644fa48b731ab18e3927a6 : TIMED_OUT in 3h 03m 14s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/69903ae69743474c990ce228084bb77d : FAILURE in 2h 10m 38s\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/80264c1915394e02957e838a0a9dba4e : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e8e53d2db16641538c6161e96d8458ab : SUCCESS in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8d92b1bf6051b497bafe8e073a0a1df431a54539","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-17 08:36:45.000000000","message":"Patch Set 6: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"2aab62d5a715441e64c96b8d4689f057d60c56dd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-17 10:48:48.000000000","message":"Patch Set 6: Code-Review+2 Review-Priority+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ac6eb4e49d0ba31b963ebf766355e40ff9db0982","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 10:49:12.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"2e92531421e065409e516bc297781a1765bec396","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 12:14:47.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9bab670c01845f9a24c5578a8aa7378\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c7ce171f3b3443e08de450937529f0e1 : SUCCESS in 18m 50s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7d72e8e1f7b3427785a79ab7782cca16 : SUCCESS in 24m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9ca4975c78d54b66b0bd8261d566943d : SUCCESS in 1h 01m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9f272a03e3944a7fb909bfc69442a041 : SUCCESS in 1h 24m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b52bd81fe9084a9abff08f699da8f29d : SUCCESS in 2m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c7ab27609024adfbe527af38466303b : SUCCESS in 2m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2cdda0c7881c4bc0b1c5c01af2e16bde : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef1c93519a3b40ad864467c8139651fe : SUCCESS in 4m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/28af70b4e818410ca6eb2214a897cf9c : SUCCESS in 24m 24s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a19ca448564048ae82e6828e283fc22e : SUCCESS in 35m 25s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/be37d6fc819249f5ba257d726ae3992e : SUCCESS in 1h 08m 00s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/97fdeb5e7e2f4872aba9aed5efbf4b07 : SUCCESS in 1h 22m 38s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d8c64e61c0dc45369759828de5c7e080 : SUCCESS in 19m 37s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/35b24f6511eb450c941ac53fa8dee0ff : SUCCESS in 1h 09m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d5304079198a49b287a55b972943c4eb : SUCCESS in 48m 49s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/af213f29e3f245cb903ac64a7dfeacca : SUCCESS in 1h 09m 31s","accounts_in_message":[],"_revision_number":6},{"id":"462b7ded13c586e9d826931946e022352d9dadd2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 12:14:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"88d2cb88ed924fc2699f854857b9a5366d702a47","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 12:15:50.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fa5e84953e74be988ca6a349ecd4c5a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef70d4ab92e7485bb76bbdca6f4eb7ff : SUCCESS in 45s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"546084b0a8ac866a34e67b558a2e210e471bec8b","revisions":{"224619db32d14a383074d1265ef5874819f1aeb9":{"kind":"REWORK","_number":1,"created":"2024-09-04 03:43:09.000000000","uploader":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"ref":"refs/changes/86/927886/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/927886/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/927886/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/927886/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/927886/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/927886/1"}}},"commit":{"parents":[{"commit":"c63961fdbe4b00707dfc744c57b5b7b0a3257a64","subject":"Merge \"Fix symlinking of openstack-base source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c63961fdbe4b00707dfc744c57b5b7b0a3257a64"}]}],"author":{"name":"Bui Doan Dang","email":"doandang27052000@gmail.com","date":"2024-09-04 03:39:31.000000000","tz":420},"committer":{"name":"Bui Doan Dang","email":"doandang27052000@gmail.com","date":"2024-09-04 03:39:31.000000000","tz":420},"subject":"Fix cinder-backup: add missing sysfsutils,lsscsi package Closes-Bug: https://bugs.launchpad.net/kolla/+bug/2078886","message":"Fix cinder-backup: add missing sysfsutils,lsscsi package\nCloses-Bug: https://bugs.launchpad.net/kolla/+bug/2078886\n\nChange-Id: Ia678484c001c66a9db675035dbb4505b0448bf58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/224619db32d14a383074d1265ef5874819f1aeb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/224619db32d14a383074d1265ef5874819f1aeb9"}]},"branch":"refs/heads/master"},"f3c25b2306924cc91c7657f1c7db18ab8ef4d4bb":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-09-04 03:45:40.000000000","uploader":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"ref":"refs/changes/86/927886/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/927886/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/927886/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/927886/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/927886/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/927886/2"}}},"commit":{"parents":[{"commit":"c63961fdbe4b00707dfc744c57b5b7b0a3257a64","subject":"Merge \"Fix symlinking of openstack-base source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c63961fdbe4b00707dfc744c57b5b7b0a3257a64"}]}],"author":{"name":"Bui Doan Dang","email":"doandang27052000@gmail.com","date":"2024-09-04 03:39:31.000000000","tz":420},"committer":{"name":"Bui Doan Dang","email":"doandang27052000@gmail.com","date":"2024-09-04 03:45:10.000000000","tz":420},"subject":"Fix cinder-backup: add missing sysfsutils,lsscsi package Closes-Bug: #2078886","message":"Fix cinder-backup: add missing sysfsutils,lsscsi package\nCloses-Bug: #2078886\n\nChange-Id: Ia678484c001c66a9db675035dbb4505b0448bf58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3c25b2306924cc91c7657f1c7db18ab8ef4d4bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f3c25b2306924cc91c7657f1c7db18ab8ef4d4bb"}]},"branch":"refs/heads/master"},"594b63721e75657845477d471835f4f2b154e497":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-09-04 04:11:22.000000000","uploader":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"ref":"refs/changes/86/927886/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/927886/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/927886/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/927886/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/927886/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/927886/3"}}},"commit":{"parents":[{"commit":"c63961fdbe4b00707dfc744c57b5b7b0a3257a64","subject":"Merge \"Fix symlinking of openstack-base source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c63961fdbe4b00707dfc744c57b5b7b0a3257a64"}]}],"author":{"name":"Bui Doan Dang","email":"doandang27052000@gmail.com","date":"2024-09-04 03:39:31.000000000","tz":420},"committer":{"name":"Bui Doan Dang","email":"doandang27052000@gmail.com","date":"2024-09-04 04:09:54.000000000","tz":420},"subject":"Fix cinder-backup: add missing sysfsutils,lsscsi package","message":"Fix cinder-backup: add missing sysfsutils,lsscsi package\n\nCloses-Bug: #2078886\n\nSigned-off-by: Bui Doan Dang \u003cdoandang27052000@gmail.com\u003e\n\nChange-Id: Ia678484c001c66a9db675035dbb4505b0448bf58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/594b63721e75657845477d471835f4f2b154e497"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/594b63721e75657845477d471835f4f2b154e497"}]},"branch":"refs/heads/master"},"bf8a8d96b48a2bc6b0838eaa570348287a749ca9":{"kind":"REWORK","_number":4,"created":"2024-09-04 04:37:14.000000000","uploader":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"ref":"refs/changes/86/927886/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/927886/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/927886/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/927886/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/927886/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/927886/4"}}},"commit":{"parents":[{"commit":"c63961fdbe4b00707dfc744c57b5b7b0a3257a64","subject":"Merge \"Fix symlinking of openstack-base source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c63961fdbe4b00707dfc744c57b5b7b0a3257a64"}]}],"author":{"name":"Bui Doan Dang","email":"doandang27052000@gmail.com","date":"2024-09-04 03:39:31.000000000","tz":420},"committer":{"name":"Bui Doan Dang","email":"doandang27052000@gmail.com","date":"2024-09-04 04:36:55.000000000","tz":420},"subject":"Fix cinder-backup: add missing sysfsutils,lsscsi package","message":"Fix cinder-backup: add missing sysfsutils,lsscsi package\n\nCloses-Bug: #2078886\n\nSigned-off-by: Bui Doan Dang \u003cdoandang27052000@gmail.com\u003e\n\nChange-Id: Ia678484c001c66a9db675035dbb4505b0448bf58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf8a8d96b48a2bc6b0838eaa570348287a749ca9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf8a8d96b48a2bc6b0838eaa570348287a749ca9"}]},"branch":"refs/heads/master"},"914d047cbdb8fc9f1bdef157b9910172aa3b9f2f":{"kind":"REWORK","_number":5,"created":"2024-09-04 04:51:00.000000000","uploader":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"ref":"refs/changes/86/927886/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/927886/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/927886/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/927886/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/927886/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/927886/5"}}},"commit":{"parents":[{"commit":"c63961fdbe4b00707dfc744c57b5b7b0a3257a64","subject":"Merge \"Fix symlinking of openstack-base source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c63961fdbe4b00707dfc744c57b5b7b0a3257a64"}]}],"author":{"name":"Bui Doan Dang","email":"doandang27052000@gmail.com","date":"2024-09-04 03:39:31.000000000","tz":420},"committer":{"name":"Bui Doan Dang","email":"doandang27052000@gmail.com","date":"2024-09-04 04:50:43.000000000","tz":420},"subject":"Fix cinder-backup: add missing sysfsutils,lsscsi package","message":"Fix cinder-backup: add missing sysfsutils,lsscsi package\n\nCloses-Bug: #2078886\n\nSigned-off-by: Bui Doan Dang \u003cdoandang27052000@gmail.com\u003e\n\nChange-Id: Ia678484c001c66a9db675035dbb4505b0448bf58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/914d047cbdb8fc9f1bdef157b9910172aa3b9f2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/914d047cbdb8fc9f1bdef157b9910172aa3b9f2f"}]},"branch":"refs/heads/master"},"546084b0a8ac866a34e67b558a2e210e471bec8b":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2024-09-17 01:55:11.000000000","uploader":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"ref":"refs/changes/86/927886/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/86/927886/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/927886/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/927886/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/86/927886/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/86/927886/6"}}},"commit":{"parents":[{"commit":"c63961fdbe4b00707dfc744c57b5b7b0a3257a64","subject":"Merge \"Fix symlinking of openstack-base source\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c63961fdbe4b00707dfc744c57b5b7b0a3257a64"}]}],"author":{"name":"Bui Doan Dang","email":"doandang27052000@gmail.com","date":"2024-09-04 03:39:31.000000000","tz":420},"committer":{"name":"Bui Doan Dang","email":"doandang27052000@gmail.com","date":"2024-09-17 01:55:11.000000000","tz":0},"subject":"Fix cinder-backup: add missing sysfsutils,lsscsi package","message":"Fix cinder-backup: add missing sysfsutils,lsscsi package\n\nCloses-Bug: #2078886\nChange-Id: Ia678484c001c66a9db675035dbb4505b0448bf58\nSigned-off-by: Bui Doan Dang \u003cdoandang27052000@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/546084b0a8ac866a34e67b558a2e210e471bec8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/546084b0a8ac866a34e67b558a2e210e471bec8b"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
