)]}'
{"id":"openstack%2Fkolla~927890","triplet_id":"openstack%2Fkolla~stable%2F2023.1~Ia678484c001c66a9db675035dbb4505b0448bf58","project":"openstack/kolla","branch":"stable/2023.1","topic":"bug/2078886-stable/2023.1","attention_set":{},"removed_from_attention_set":{"35793":{"account":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"last_update":"2024-09-22 02:43:29.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2024-09-20 09:31:53.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-09-22 02:43:29.000000000","reason":"Change was submitted"},"32553":{"account":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"last_update":"2024-09-22 02:43:29.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2024-09-22 02:43:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia678484c001c66a9db675035dbb4505b0448bf58","subject":"Fix cinder-backup: add missing sysfsutils,lsscsi package","status":"MERGED","created":"2024-09-04 04:46:36.000000000","updated":"2024-09-22 02:44:40.000000000","submitted":"2024-09-22 02:43:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927890-bug/2078886-stable/2023.1","cherry_pick_of_change":927886,"cherry_pick_of_patch_set":4,"meta_rev_id":"80bf62609b78e47093ecb8fb8b04bff0015886f0","_number":927890,"virtual_id_number":927890,"owner":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-22 02:43:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"all":[{"value":1,"date":"2024-09-19 03:09:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},{"value":2,"date":"2024-09-20 09:31:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2024-09-18 08:46:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2024-09-21 22:11:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2024-09-21 22:11:54.000000000","_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-04 06:32:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-04 07:42:18.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2024-09-04 09:23:19.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2024-09-12 14:07:38.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2024-09-18 15:18:29.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2024-09-18 15:18:29.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"},{"updated":"2024-09-21 22:11:54.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"}],"messages":[{"id":"c4507f9b94615436b6c6425f545f04a3e0259914","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"date":"2024-09-04 04:46:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a1480076a0e1eb58ffec0ea447614d5b198f5996","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"date":"2024-09-04 04:59:00.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"fc710c6eb712b5442febad3bc4ef5e6c0c66c4d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 06:32:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7109480b5e2d404eaeaee158d98d96a3\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3cc5ddabe4f247fabe2acb27c274e967 : SUCCESS in 24m 21s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f6a3b04123ef472287a66bb59eefe2cf : SUCCESS in 54m 33s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/62e86a982999404182abdb73ab74bffc : SUCCESS in 26m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2de03a4c884046258a3188806b4d4d6b : SUCCESS in 2m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8385770dcb554795ad495418c645185d : SUCCESS in 2m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/130a880083bd49a89e3068abc13655f0 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73086b4c03164d0a805cd785e64c9527 : SUCCESS in 3m 55s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/427951b3245f480ab42621f956be0b5b : SUCCESS in 25m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/49b4e2f13b11411d8869615396efb538 : SUCCESS in 54m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/dae83c4668024c9a82856698292ec3e5 : SUCCESS in 1h 27m 33s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/71bfc45ad13b438ea88c8b6c097a2ff4 : SUCCESS in 23m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/43a58601394c4203b76221a88aa22745 : SUCCESS in 49m 34s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/dfc638256dc94f79b8d4c17413591c2b : SUCCESS in 1h 21m 46s","accounts_in_message":[],"_revision_number":2},{"id":"f8ae30003ab517c71e52b15bfff948b4c4e85018","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2024-09-04 07:42:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5cc20fd0bb1f6c5efa288c921accee18d3c366e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 09:57:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17613457570847548f7e190f81a43aa7\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3ae44d6dc4584c8aa397808cfe290008 : TIMED_OUT in 3h 03m 14s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/17ac10feada941cfb490869847fe1eae : SUCCESS in 3m 42s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0ec117ab9e1a4cd9a2287077a5415786 : TIMED_OUT in 3h 04m 48s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ee4938b2130749519895852f2d6ccd74 : FAILURE in 2h 20m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c7923f990615655101eafd3118aa7c8c25076190","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-12 14:07:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ab3762a94c071ae3b57151c1110c55ba785c4aa5","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-12 14:27:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ca1623210e531c226eb85b6c693d84d2b8dd04db","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-12 14:30:46.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"532c0bf9683bb5c34ef8206fa96e89564dd1bd6c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-17 08:39:06.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0943d92304acf8d8938edb0e763d02834c2bd816","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"date":"2024-09-18 08:16:28.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"5b0bf541f2d58307c4a3dc8c370c4dac0b3f534d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2024-09-18 08:46:31.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"136e86d344727105f681f0b275a92042812bb0e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 09:59:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/634529d75d0b4ed08f2dfffc38b6432c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/56cfc6f485de40d4b882f9bb2e1f288d : SUCCESS in 21m 36s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f286f2c12a6b42fd9d901898a74a4c8c : SUCCESS in 56m 08s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f48a7a04a9f0428ea1a2b2a767eacb0a : SUCCESS in 25m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/502fbff76b4f4346afe3b746e6b9a106 : SUCCESS in 2m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d362482dda9c4ff49934fa658d23b64a : SUCCESS in 2m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e2a6aeb9b31f455fab13d157ec2f7c56 : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0f8b01441dd4c3b90bf77f7cebe9705 : SUCCESS in 3m 09s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4fa9caa5b8264434b5e5cf2c358be0cf : SUCCESS in 32m 00s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8c6e007e41864bc19bfeef987556d3ae : SUCCESS in 1h 03m 13s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/70e57ab925ae49168c24bc418e4d8fc2 : SUCCESS in 1h 31m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f6561178945f40708032aec7dcbb6188 : SUCCESS in 24m 50s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/829322d5211042339ada237a25143e19 : SUCCESS in 48m 49s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6314e3955cd24d48b198b71c4f7e810c : SUCCESS in 1h 11m 35s","accounts_in_message":[],"_revision_number":3},{"id":"1922d4f10fd89acd36b9f8b5321fb02011f87867","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 11:22:56.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f467b3d91cf4ccba657609fe3050a9d\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0f88fc0e452b4a0bb170118010cb8a01 : TIMED_OUT in 3h 03m 16s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a60ce112721b4880b452c30e9e61b48d : SUCCESS in 4m 21s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c8bf2b249ec648c48be1059d5a21b4f5 : TIMED_OUT in 3h 04m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/52507e508296488aa28bd1f9238f85a6 : FAILURE in 7m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"94cea025e9a7954431adb65522d195016ef28735","author":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"date":"2024-09-19 03:09:42.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"de96898e91c421bedefab8353570758e2322ff06","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2024-09-20 09:31:53.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"dbdfd29ff40fc6a3f09bf7f1f1dac79af8a49596","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2024-09-21 22:11:54.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3f6ac5f5dc6c28bd4366bd484504206adf9924e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-21 22:12:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8e22285d6dc63063bebc5fb0a155e3f988c952a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-22 02:43:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef0cf5c5d0a443fcab217b4ca0d5be91\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/de1a2e8daf2e448b9384d878f5d1238b : SUCCESS in 21m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9dcc3b8cabb417e9334be7386388397 : SUCCESS in 2m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c68c560193fa455b88875ba92c6e85b6 : SUCCESS in 2m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/031f9215689641b99fd05e67547a4ea1 : SUCCESS in 4m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b875d1507db4446d97135f9274d140a3 : SUCCESS in 27m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2edf60848eaf464caee982c103462ca5 : SUCCESS in 1h 02m 23s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/fce6b66396434b8388c41d27af0ae6e6 : SUCCESS in 1h 26m 06s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/758e99ef251547029d438cec6a9ec4aa : SUCCESS in 17m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ef5ff482429949f79219c697d5298435 : SUCCESS in 44m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/df8746a6056445ca82ce4fad9171e715 : SUCCESS in 1h 10m 17s","accounts_in_message":[],"_revision_number":3},{"id":"ab012a128cf840b1aad138a2e6335d383235868a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-22 02:43:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"80bf62609b78e47093ecb8fb8b04bff0015886f0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-22 02:44:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c578a7adf2274cbba770357b1515ec1b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b96bee1ec9db493b8fe951be68231896 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d0fd3e6006b839f51605840d8bc09406c26b8f9c","revisions":{"fbceba223532310a4b3730cf1fc5f79a58fb7454":{"kind":"REWORK","_number":1,"created":"2024-09-04 04:46:36.000000000","uploader":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"ref":"refs/changes/90/927890/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/927890/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/927890/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/927890/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/927890/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/927890/1"}}},"commit":{"parents":[{"commit":"3a39a79eed39fb844aecadd524cf4cc54d812cdf","subject":"keystone: Fix requests-kerberos installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a39a79eed39fb844aecadd524cf4cc54d812cdf"}]}],"author":{"name":"Bui Doan Dang","email":"doandang27052000@gmail.com","date":"2024-09-04 03:39:31.000000000","tz":420},"committer":{"name":"Bui Doan Dang","email":"doandang27052000@gmail.com","date":"2024-09-04 04:46:36.000000000","tz":0},"subject":"Fix cinder-backup: add missing sysfsutils,lsscsi package","message":"Fix cinder-backup: add missing sysfsutils,lsscsi package\n\nCloses-Bug: #2078886\n\nSigned-off-by: Bui Doan Dang \u003cdoandang27052000@gmail.com\u003e\n\nChange-Id: Ia678484c001c66a9db675035dbb4505b0448bf58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fbceba223532310a4b3730cf1fc5f79a58fb7454"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fbceba223532310a4b3730cf1fc5f79a58fb7454"}]},"branch":"refs/heads/stable/2023.1"},"26ad10e18e2604f24c8dcfc829a52e8be96c3403":{"kind":"REWORK","_number":2,"created":"2024-09-04 04:59:00.000000000","uploader":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"ref":"refs/changes/90/927890/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/927890/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/927890/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/927890/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/927890/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/927890/2"}}},"commit":{"parents":[{"commit":"3a39a79eed39fb844aecadd524cf4cc54d812cdf","subject":"keystone: Fix requests-kerberos installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a39a79eed39fb844aecadd524cf4cc54d812cdf"}]}],"author":{"name":"Bui Doan Dang","email":"doandang27052000@gmail.com","date":"2024-09-04 03:39:31.000000000","tz":420},"committer":{"name":"Bui Doan Dang","email":"doandang27052000@gmail.com","date":"2024-09-04 04:59:00.000000000","tz":0},"subject":"Fix cinder-backup: add missing sysfsutils,lsscsi package","message":"Fix cinder-backup: add missing sysfsutils,lsscsi package\n\nCloses-Bug: #2078886\n\nSigned-off-by: Bui Doan Dang \u003cdoandang27052000@gmail.com\u003e\n\nChange-Id: Ia678484c001c66a9db675035dbb4505b0448bf58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/26ad10e18e2604f24c8dcfc829a52e8be96c3403"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/26ad10e18e2604f24c8dcfc829a52e8be96c3403"}]},"branch":"refs/heads/stable/2023.1"},"d0fd3e6006b839f51605840d8bc09406c26b8f9c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-09-18 08:16:28.000000000","uploader":{"_account_id":35793,"name":"Bui Doan Dang","display_name":"Dang Bui Doan","email":"doandang27052000@gmail.com","username":"dangbd2705"},"ref":"refs/changes/90/927890/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/90/927890/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/927890/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/927890/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/90/927890/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/90/927890/3"}}},"commit":{"parents":[{"commit":"3a39a79eed39fb844aecadd524cf4cc54d812cdf","subject":"keystone: Fix requests-kerberos installation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a39a79eed39fb844aecadd524cf4cc54d812cdf"}]}],"author":{"name":"Bui Doan Dang","email":"doandang27052000@gmail.com","date":"2024-09-04 03:39:31.000000000","tz":420},"committer":{"name":"Bui Doan Dang","email":"doandang27052000@gmail.com","date":"2024-09-18 08:16:28.000000000","tz":0},"subject":"Fix cinder-backup: add missing sysfsutils,lsscsi package","message":"Fix cinder-backup: add missing sysfsutils,lsscsi package\n\nCloses-Bug: #2078886\nChange-Id: Ia678484c001c66a9db675035dbb4505b0448bf58\nSigned-off-by: Bui Doan Dang \u003cdoandang27052000@gmail.com\u003e\n(cherry picked from commit 546084b0a8ac866a34e67b558a2e210e471bec8b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d0fd3e6006b839f51605840d8bc09406c26b8f9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d0fd3e6006b839f51605840d8bc09406c26b8f9c"}]},"branch":"refs/heads/stable/2023.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
